)]}'
{"id":"openstack%2Fglance~969017","triplet_id":"openstack%2Fglance~stable%2F2025.2~I77f08847d36329f7b90b061f33dacb176afbe8f4","project":"openstack/glance","branch":"stable/2025.2","topic":"bug/2133170-stable/2025.2","attention_set":{},"removed_from_attention_set":{"9303":{"account":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"last_update":"2025-12-05 19:23:59.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I77f08847d36329f7b90b061f33dacb176afbe8f4","subject":"Fix FileNotFoundError when registering task operations","status":"MERGED","created":"2025-12-01 07:17:33.000000000","updated":"2025-12-05 19:24:52.000000000","submitted":"2025-12-05 19:23:59.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"969017-bug/2133170-stable/2025.2","cherry_pick_of_change":968711,"cherry_pick_of_patch_set":1,"meta_rev_id":"0c860ed5ab1cd99a8d17aec9cea813efb7b3c129","_number":969017,"virtual_id_number":969017,"owner":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-12-05 19:23:59.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"all":[{"value":2,"date":"2025-12-04 13:48:30.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2025-12-05 10:59:27.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"all":[{"value":1,"date":"2025-12-04 13:48:30.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Top Priority / Holds Gate"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-12-01 09:39:06.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-12-04 13:48:30.000000000","updated_by":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"reviewer":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"state":"REVIEWER"},{"updated":"2025-12-05 10:59:27.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"state":"REVIEWER"}],"messages":[{"id":"df5070e658d21373c31326f1884243ff69b414d5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2025-12-01 07:17:33.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"a288151069d2c607ef8c6ade11c8ac70ef697c1d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-01 09:39:06.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e2e6fbe09951499eb81de7df64c22e88\n\n- grenade https://zuul.opendev.org/t/openstack/build/17afd806c25747f9874ecac5d59af0e4 : SUCCESS in 29m 10s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/944b430e71864f6b94afffd7f31fc45f : SUCCESS in 33m 32s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/cb45b67558a3476d8bad66a126cdd586 : SUCCESS in 1h 19m 15s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/8757092dd6ef4e1aa1819434108e1661 : SUCCESS in 1h 01m 27s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6d518b1ca3764a4db1c671e873a09f41 : SUCCESS in 6m 31s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/fd833918a2f542ed991493217fa0943a : SUCCESS in 2m 58s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/33b001cc0b04432e91a7c1965f27bb01 : SUCCESS in 5m 25s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/483d6cee36c644c3ad72e0c1c612d0cc : SUCCESS in 8m 24s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b23a8fb5b18548d7ba4e7a84b799dc01 : SUCCESS in 6m 01s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/b3441ddd35ee4b809c539dd823b0b38b : SUCCESS in 4m 54s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/6f46c9e099b9461ea9c5d1161b269d20 : SUCCESS in 9m 48s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/b272e8a60989418db4f2e9cc8f731d1e : SUCCESS in 2h 11m 49s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/88b7216445be4a0fa6125bffad63cc65 : SUCCESS in 2h 15m 15s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/a3671009fd6e4abdb6aba96f892d5b2f : SUCCESS in 1h 35m 53s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/b697d1653ac3410cb095f6a1fdc63c9a : SUCCESS in 2h 13m 51s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/0393cb7b4692464aa0b1749235f8ddbe : SUCCESS in 40m 30s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/4a1042e06c9244ba92ebd65b68d1fa91 : SUCCESS in 1h 56m 17s\n- glance-grenade-centralized-cache https://zuul.opendev.org/t/openstack/build/98733f3aab9849d5998ce3c2887bcd17 : FAILURE in 16m 24s (non-voting)\n- glance-s3-multistore https://zuul.opendev.org/t/openstack/build/6a127b1c144b4d9590c11289b679a631 : FAILURE in 26m 24s (non-voting)\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/ef2c3d8913304b6e87d27f1a82930e66 : SUCCESS in 16m 57s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/2ed62bbd969d49d0a1668172b6267525 : SUCCESS in 29m 08s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"32fcd0cc19de65bfa9f4c0fcf8112d802063185c","author":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"date":"2025-12-04 13:48:30.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"8f9d57957ce8588baaa12cb7867f22e07c09d68a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-04 13:48:58.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"a85c8b50b76ca7f2a0b4691fb14b9341e0c18bdf","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-04 15:40:35.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/ac0e1158b2644338be8839efe8cb1f5d\n\n- grenade https://zuul.opendev.org/t/openstack/build/bf6cd7e34a7a4fbf9200221b6d74c306 : SUCCESS in 31m 05s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/536b97a7007d46328ecd3d654787d4d0 : SUCCESS in 32m 22s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/c37c7efb04bf40cdb682c50c721c532e : SUCCESS in 1h 23m 41s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/25dadac643fd43b69385bc2cea185fe7 : SUCCESS in 54m 59s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/721f55059ba3461fb6a7616dbece1afa : SUCCESS in 5m 06s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/a6424574d8354440becdc59ccc1f8e36 : SUCCESS in 5m 09s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/dfc7b42e3ff543a28dfc441b39da1a66 : SUCCESS in 5m 32s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/79d9365e186a494e92fbbb313837f6e3 : SUCCESS in 6m 29s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/227f9ce833614bf498a2327b11a4d14c : SUCCESS in 7m 02s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/87d639792941475d93eca2068a1b496a : SUCCESS in 10m 35s\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/e96f8f34a5d94a70b8b26cfcdccae982 : SUCCESS in 1h 48m 33s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/d0552390bed64659874b81dfd721fd49 : FAILURE in 1h 50m 14s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/a135d852ea024551ace5bb63cdd8c454 : SUCCESS in 43m 58s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/dccb7335283943468331d131b606b6df : SUCCESS in 1h 43m 52s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/af1a3df5493c4f0f8ebe0ae2e1a4071b : SUCCESS in 16m 49s","accounts_in_message":[],"_revision_number":1},{"id":"afa8b7b5aeae546971381aa1fed1ede6f140a390","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2025-12-05 10:59:27.000000000","message":"Patch Set 1: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"86cf2693681489751ad050617964eeea90fe69f4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-05 13:45:16.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a0cc86cd07374a56859e4c471719b2b4\n\n- grenade https://zuul.opendev.org/t/openstack/build/87649539340b4d0d97c2f4d05029ae49 : SUCCESS in 1h 07m 30s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/19d6806a0782413fa575e6e02594e251 : SUCCESS in 1h 06m 13s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/a929848146624e7eb2a1da38b249971f : SUCCESS in 1h 09m 12s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/13857391aefc47a0a80705decc56e511 : SUCCESS in 46m 11s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c901bab2390948fd9e8c39ca7412d604 : SUCCESS in 5m 26s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/ef509c073d0b443a93b1e82afd18d058 : SUCCESS in 5m 09s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/03693dd1b140490e92f90b3ee6a77728 : SUCCESS in 5m 53s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/ac37f343eda64a2fac45664b626002d5 : SUCCESS in 7m 51s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fdbc7cbea72748f2a95a4bea47a111ff : SUCCESS in 4m 54s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/3b5cc70c58fa4ddc83a59fe9b24b4659 : SUCCESS in 6m 50s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/2d1f5df244c044c48953b20de952ad57 : SUCCESS in 9m 28s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/d03eb203bf804a858005d79a3393033d : SUCCESS in 1h 54m 23s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/16c24caf31db4f9eaf63a54ddc4bae1b : SUCCESS in 2h 08m 00s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/a4a8b29c085f4e00ac0163ae4ab787da : SUCCESS in 1h 58m 51s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/1a4c56b5b4984a17b6f18b9204870c92 : SUCCESS in 2h 31m 21s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/3609fc4dc8d14a3f8b4eedcbc2aba21a : SUCCESS in 55m 55s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/e1ab1519582f40d185a0b7f87f097658 : SUCCESS in 1h 55m 05s\n- glance-grenade-centralized-cache https://zuul.opendev.org/t/openstack/build/5fec218eaa844981ad84a2b0299b1efa : FAILURE in 30m 48s (non-voting)\n- glance-s3-multistore https://zuul.opendev.org/t/openstack/build/1e58787649c64979bfa72bb62b25dba7 : FAILURE in 12m 42s (non-voting)\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/fd743bdda6584f7dbcde9d13b1f18f2f : SUCCESS in 18m 00s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/837c9eff921f47f1978f35b7337355c4 : SUCCESS in 32m 22s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"773ed6326f31f3e641953e2d6f23f1b55fa0b618","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-05 13:45:43.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"9847f8c00cb8ac307e877dc918ee91d2fa02facd","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-05 15:45:49.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/f868b7a1bc204f478a1bddc8f58c3bc2\n\n- grenade https://zuul.opendev.org/t/openstack/build/8572b45a67b9427da9ed402671917fb5 : SUCCESS in 1h 00m 27s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/4b04ca9390a746089f0c7930278cb3ba : SUCCESS in 1h 04m 07s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/281460eb53f04464b805b648a89feef7 : SUCCESS in 1h 29m 13s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/c53ef3ed1fdb4629acb8700d352d2bac : SUCCESS in 35m 58s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a855b87e20d54cdb86d393b49b169f34 : SUCCESS in 4m 58s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/1974ec91dcc2400e9d11a762d00a7ef0 : SUCCESS in 4m 22s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/ae7cd9272cb841258b3ebc0cc60ab49f : SUCCESS in 5m 44s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7d7270ec31784b2892961779c01b84ef : SUCCESS in 5m 55s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/77eda1096dea4f5090f0be8dbbdc6ce9 : SUCCESS in 7m 58s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/b48ab5f145ee4f8c99be62c9ab055ada : SUCCESS in 12m 57s\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/e3568ace14c04691a9cf05fcd38eda97 : SUCCESS in 1h 51m 44s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/a0a991b1ca354fed91a69f0cf799400a : SUCCESS in 1h 58m 57s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/2b609754bb434f719fda1e63d86bf8a4 : SUCCESS in 57m 26s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/a14dbb0be08e4663945cb021a022e054 : SUCCESS in 1h 25m 55s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/c5296b7bd9ea4b42b2ec5d1db3e2c44c : FAILURE in 35m 27s","accounts_in_message":[],"_revision_number":1},{"id":"8cf704df2193df110a0a3e9560687cb5c680b6ef","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2025-12-05 15:48:10.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"8309fd872f6fc2d90ffc5e91e7674b93caf60943","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-05 17:35:19.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f423ed77848344b4a9acb6f365b95de1\n\n- grenade https://zuul.opendev.org/t/openstack/build/df9091cadfb14c8eacea9f25d91c0940 : SUCCESS in 1h 02m 04s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/92e194d2790a4d87a61ab0fb55d9751e : SUCCESS in 1h 03m 40s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/3bce77e56f3f4e44a54a40b8b9ca788c : SUCCESS in 1h 45m 14s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/46db2bf104634ed68f0ce908b0c2ae23 : SUCCESS in 1h 05m 38s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f6e0f53c4004442791b1a6f29f5aec40 : SUCCESS in 5m 10s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/a586056937c94952b8ce34dc67dbd2b7 : SUCCESS in 3m 23s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/b4e813863e8f49369f4def6db6716e5e : SUCCESS in 3m 18s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/d4a34f8c602a4c7597c53d4c049bae59 : SUCCESS in 9m 18s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4fe91b000ae44b6cb20689ed61010a29 : SUCCESS in 7m 04s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/6932b5dc23c3415e8fa9a2b9f579bb6f : SUCCESS in 8m 26s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/3e8af697c8444a4ebc47a2c62ab3e09a : SUCCESS in 6m 12s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/99ee1a6688484538af12e27384b04cfc : SUCCESS in 1h 16m 05s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/c92ef3d145fe49f2a7938aceaaebf28b : SUCCESS in 1h 14m 59s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/7032469decc844ee86cc3e20ca930f30 : SUCCESS in 1h 15m 06s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/ad0f645f96984cf7b888769fec8204e9 : SUCCESS in 1h 14m 55s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/ef84e1d37d7d4f5ca672c9b49e7ab933 : SUCCESS in 53m 17s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/cc40ace46f7d4fb3b7fc4aff1ddb796a : SUCCESS in 1h 01m 52s\n- glance-grenade-centralized-cache https://zuul.opendev.org/t/openstack/build/a89b0f4b3f424931bb5e81135563f576 : FAILURE in 29m 35s (non-voting)\n- glance-s3-multistore https://zuul.opendev.org/t/openstack/build/5314c1fbb19e4ffc8905ae3ce53ef972 : FAILURE in 21m 56s (non-voting)\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/1d34c50ca40f41859ccd4f79cf72dc5f : SUCCESS in 27m 44s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/5ca20920c43f4c2d9656040e9b2a8bee : SUCCESS in 28m 53s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"fdcf4acf296b80e0635562c82702fb6efeca0441","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-05 17:35:59.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"ef1129064d2d4bea7039cb71fbcae4e8d6d9631b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-05 19:23:59.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/84a4cbdbd679454f897a030328e6855e\n\n- grenade https://zuul.opendev.org/t/openstack/build/f3f67154addd44d79b2f465b94ffb9e2 : SUCCESS in 29m 43s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/f7d18dc5550f4088a0757c0b14d73f64 : SUCCESS in 36m 13s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/c34ca7d472934cc79cdc9f940183069d : SUCCESS in 1h 40m 02s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/e9a8761e9509481b9644544353eb5bc5 : SUCCESS in 35m 11s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e09b719ec9db4f32b0ca88508b714cce : SUCCESS in 4m 26s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/2f2e11b32ab74e899581adca0f25e1f7 : SUCCESS in 4m 24s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/ea73093ef74d4ea785d70bcc4cc59933 : SUCCESS in 6m 26s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/be96d35979e3403faa5f1bcda5d7f853 : SUCCESS in 6m 52s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/dd319ea96a1c4cf98c6332909898ce51 : SUCCESS in 7m 07s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/680e46678a994b3589adc1ae0fe1abfc : SUCCESS in 8m 45s\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/d9e2e28ae1f243a4b51dcec6ba2af071 : SUCCESS in 1h 09m 27s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/15f2f7386a8749fba1d155e7f4239241 : SUCCESS in 1h 42m 28s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/a9ac2855a6104799a2143fa9f0390b11 : SUCCESS in 53m 17s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/e573107ae81a45fa8d70c633a76d7f4e : SUCCESS in 1h 43m 57s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/29c36110472c4efb8dd69442e74c781e : SUCCESS in 36m 51s","accounts_in_message":[],"_revision_number":1},{"id":"ab2ebbaead71cea54d06a9c7ecab17eb870e2781","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-05 19:23:59.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"0c860ed5ab1cd99a8d17aec9cea813efb7b3c129","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-05 19:24:52.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7332669f48384c1fbdb8a4af45f6c166\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/be008a6c0b014b0d8d6ccf495c1f87e5 : SUCCESS in 42s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"0489ff393d728dcea70e22778684fff6f8990f26","revisions":{"0489ff393d728dcea70e22778684fff6f8990f26":{"kind":"REWORK","_number":1,"created":"2025-12-01 07:17:33.000000000","uploader":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"ref":"refs/changes/17/969017/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/17/969017/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/17/969017/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/17/969017/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/17/969017/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/17/969017/1"}}},"commit":{"parents":[{"commit":"81a6e1acfa50da1977dd1b39d031baaa4b9f799e","subject":"Merge \"s3: Do not log access keys\" into stable/2025.2","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/81a6e1acfa50da1977dd1b39d031baaa4b9f799e"}]}],"author":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2025-11-27 14:50:15.000000000","tz":0},"committer":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2025-12-01 07:17:33.000000000","tz":0},"subject":"Fix FileNotFoundError when registering task operations","message":"Fix FileNotFoundError when registering task operations\n\nEnsure parent directory exists before creating operation lock files\nto fix failures in legacy single-store deployments where staging\ndirectory is created during import operations.\n\nCloses-Bug: #2133170\nChange-Id: I77f08847d36329f7b90b061f33dacb176afbe8f4\nSigned-off-by: Abhishek Kekane \u003cakekane@redhat.com\u003e\n(cherry picked from commit 8a5aa7a98b12de24b5732eb69c5525c70fbc8fb2)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/0489ff393d728dcea70e22778684fff6f8990f26"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/0489ff393d728dcea70e22778684fff6f8990f26"}]},"branch":"refs/heads/stable/2025.2"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
