)]}'
{"id":"openstack%2Fglance~970048","triplet_id":"openstack%2Fglance~master~I1dd1d2f31203d69be93bafe9222009f58e78500a","project":"openstack/glance","branch":"master","attention_set":{},"removed_from_attention_set":{"38656":{"account":{"_account_id":38656,"name":"Vishnu C","display_name":"Vishnu Chandrasenan","email":"vischan2@cisco.com","username":"vischan2"},"last_update":"2025-12-12 02:10:05.000000000","reason":"Change was submitted"},"9303":{"account":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"last_update":"2025-12-09 07:15:36.000000000","reason":"\u003cGERRIT_ACCOUNT_9303\u003e replied on the change","reason_account":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"}}},"hashtags":[],"change_id":"I1dd1d2f31203d69be93bafe9222009f58e78500a","subject":"Fix S3 URL corruption on secret rotation","status":"MERGED","created":"2025-12-07 10:08:14.000000000","updated":"2025-12-12 02:13:47.000000000","submitted":"2025-12-12 02:10:05.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":12,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"970048","meta_rev_id":"4a60d82249771770590ae73c30b7b9ac5836d024","_number":970048,"virtual_id_number":970048,"owner":{"_account_id":38656,"name":"Vishnu C","display_name":"Vishnu Chandrasenan","email":"vischan2@cisco.com","username":"vischan2"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":21113,"name":"Min Sun","email":"minsu@cisco.com","username":"MinSun"},{"value":0,"_account_id":25402,"name":"Francesco Pantano","email":"fpantano@redhat.com","username":"fmount"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-12-12 02:10:05.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"all":[{"value":2,"date":"2025-12-12 00:05:44.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":2,"date":"2025-12-09 07:15:36.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":1,"date":"2025-12-08 23:48:29.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":21113,"name":"Min Sun","email":"minsu@cisco.com","username":"MinSun"},{"value":1,"date":"2025-12-10 07:38:10.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":25402,"name":"Francesco Pantano","email":"fpantano@redhat.com","username":"fmount"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"all":[{"value":1,"date":"2025-12-12 00:05:44.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":21113,"name":"Min Sun","email":"minsu@cisco.com","username":"MinSun"},{"value":0,"_account_id":25402,"name":"Francesco Pantano","email":"fpantano@redhat.com","username":"fmount"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":21113,"name":"Min Sun","email":"minsu@cisco.com","username":"MinSun"},{"value":0,"_account_id":25402,"name":"Francesco Pantano","email":"fpantano@redhat.com","username":"fmount"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Top Priority / Holds Gate"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"}],"reviewers":{"REVIEWER":[{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":21113,"name":"Min Sun","email":"minsu@cisco.com","username":"MinSun"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25402,"name":"Francesco Pantano","email":"fpantano@redhat.com","username":"fmount"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-12-07 10:16:04.000000000","updated_by":{"_account_id":38656,"name":"Vishnu C","display_name":"Vishnu Chandrasenan","email":"vischan2@cisco.com","username":"vischan2"},"reviewer":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"state":"REVIEWER"},{"updated":"2025-12-07 10:16:04.000000000","updated_by":{"_account_id":38656,"name":"Vishnu C","display_name":"Vishnu Chandrasenan","email":"vischan2@cisco.com","username":"vischan2"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2025-12-07 10:16:04.000000000","updated_by":{"_account_id":38656,"name":"Vishnu C","display_name":"Vishnu Chandrasenan","email":"vischan2@cisco.com","username":"vischan2"},"reviewer":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"state":"REVIEWER"},{"updated":"2025-12-07 10:16:04.000000000","updated_by":{"_account_id":38656,"name":"Vishnu C","display_name":"Vishnu Chandrasenan","email":"vischan2@cisco.com","username":"vischan2"},"reviewer":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"state":"REVIEWER"},{"updated":"2025-12-07 10:16:04.000000000","updated_by":{"_account_id":38656,"name":"Vishnu C","display_name":"Vishnu Chandrasenan","email":"vischan2@cisco.com","username":"vischan2"},"reviewer":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"state":"REVIEWER"},{"updated":"2025-12-07 10:16:04.000000000","updated_by":{"_account_id":38656,"name":"Vishnu C","display_name":"Vishnu Chandrasenan","email":"vischan2@cisco.com","username":"vischan2"},"reviewer":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"state":"REVIEWER"},{"updated":"2025-12-07 10:19:30.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2025-12-07 13:17:32.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-12-08 23:48:29.000000000","updated_by":{"_account_id":21113,"name":"Min Sun","email":"minsu@cisco.com","username":"MinSun"},"reviewer":{"_account_id":21113,"name":"Min Sun","email":"minsu@cisco.com","username":"MinSun"},"state":"REVIEWER"},{"updated":"2025-12-10 07:38:10.000000000","updated_by":{"_account_id":25402,"name":"Francesco Pantano","email":"fpantano@redhat.com","username":"fmount"},"reviewer":{"_account_id":25402,"name":"Francesco Pantano","email":"fpantano@redhat.com","username":"fmount"},"state":"REVIEWER"}],"messages":[{"id":"0c7ddcf6d3a9989214bfa9f9461a090c4c8db1e7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":38656,"name":"Vishnu C","display_name":"Vishnu Chandrasenan","email":"vischan2@cisco.com","username":"vischan2"},"date":"2025-12-07 10:08:14.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"b52d400b4d786bb1a270e651c6d0e39304e5d592","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-07 10:19:30.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0e4a1d1e904744b587f7ddc42a53ba68\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/3840fb56e205401daef4d88705831018 : FAILURE in 8m 46s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/cbe13e681d924fbe8e46fa8993cd108b : FAILURE in 9m 26s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/ce388a1d1fcc4528a97c50d8aa5b4d6f : FAILURE in 9m 18s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"771f62fb46f4dac713f01f2b88c62714640ba444","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":38656,"name":"Vishnu C","display_name":"Vishnu Chandrasenan","email":"vischan2@cisco.com","username":"vischan2"},"date":"2025-12-07 10:58:30.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"673d744b9331f07d6825528a3fb3f77f612a7973","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-07 11:12:04.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a2ac78334cc44982929b026de123ff74\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/662138afdb8f455ea939fdef7314deed : SUCCESS in 5m 36s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/96c22fcdfb9249b0ba0c39a3f3514552 : SUCCESS in 8m 07s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/2b07e06f60ef45dd8bbad747439a9132 : SUCCESS in 11m 40s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"28feda52675654055776495df8b0c243e1ef9c7b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-07 13:17:32.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4117da54dedd4f209be509d9bad2513c\n\n- grenade https://zuul.opendev.org/t/openstack/build/de597537d82342158d23ab66fbb4a887 : SUCCESS in 1h 02m 48s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/51b32cfa8cea42c4819f05b7ae53adeb : SUCCESS in 1h 06m 06s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/b5bc51796ec84ed6a2f6b3771120b789 : SUCCESS in 1h 42m 30s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/4569a4c072ba42c68af8c75a4e0b85ac : SUCCESS in 35m 39s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e06da3f9ae9443fe9a2febf6bf62bc41 : SUCCESS in 4m 18s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/67067f37bf1f460394b88890442e84fb : SUCCESS in 2m 33s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/6a34a7ef22754d67b3afe76c4f5f9a5b : SUCCESS in 4m 05s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bf4741f85b0f408e903fab56b92ecac8 : SUCCESS in 9m 03s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/76e144d535504d159fbb80a7153852f8 : SUCCESS in 6m 26s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/38dc81d7fc284c3a8ebd64a8c6737c6d : SUCCESS in 9m 56s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/07f3c61357bf420ab0906a581247bafe : SUCCESS in 2h 05m 57s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/23da22852fa64640886f8974a1172002 : SUCCESS in 2h 13m 41s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/cc4e05497bf14ccea11bc7e6043990c2 : SUCCESS in 1h 30m 07s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/205e9ef28824400592e9c9d09c8c5f6c : SUCCESS in 1h 12m 01s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/1b37bfb7b53a40dca55f22b8eac08bd0 : SUCCESS in 1h 01m 21s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/17d7ad8a93184c4d847a5bcf20a7ec5a : SUCCESS in 1h 59m 01s\n- glance-grenade-centralized-cache https://zuul.opendev.org/t/openstack/build/584b5e3030104784aca8fee75aa203ed : FAILURE in 10m 03s (non-voting)\n- glance-s3-multistore https://zuul.opendev.org/t/openstack/build/c9021684a468451fb4ebe1ce38992746 : FAILURE in 25m 03s (non-voting)\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/c69e371b2a974601b40c808d8082b40b : SUCCESS in 34m 51s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/ee0b0d29f94c459ba2a5b7ec95abf381 : SUCCESS in 1h 16m 31s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"f70616d70d6a73e47163075bbf7d6ddb44498cc0","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2025-12-08 10:22:15.000000000","message":"Patch Set 2: Code-Review-1\n\n(3 comments)","accounts_in_message":[],"_revision_number":2},{"id":"10ec5c2ba3869ccc2002828823217804f8c10fd9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":38656,"name":"Vishnu C","display_name":"Vishnu Chandrasenan","email":"vischan2@cisco.com","username":"vischan2"},"date":"2025-12-08 15:53:32.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":3},{"id":"3106a13d2574c3eaef25126f836a935765d69e1d","author":{"_account_id":38656,"name":"Vishnu C","display_name":"Vishnu Chandrasenan","email":"vischan2@cisco.com","username":"vischan2"},"date":"2025-12-08 16:05:26.000000000","message":"Patch Set 3:\n\n(3 comments)","accounts_in_message":[],"_revision_number":3},{"id":"d64d796fe3c3ba874d4d0202f0407522a6af5597","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-08 16:09:16.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a5db05a8ef164d2885cb88916ea4e001\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/f642dedb85294f029293a48116571d39 : SUCCESS in 8m 09s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/401f6627117147da97303a02485cb887 : SUCCESS in 8m 58s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/85009a3e5bca4aa9ad549f13feaeb8c3 : SUCCESS in 13m 17s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"5815279c3e61d89efccab31f115de3bea9fcae55","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-08 18:03:04.000000000","message":"Patch Set 3: Verified-1\n\n(2 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/73270839a94c40c790144acd76d879a7\n\n- grenade https://zuul.opendev.org/t/openstack/build/ffc170633a354637882f5fb857ead542 : SUCCESS in 59m 29s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/b192476ee0c14676a3e80211320de6f2 : SUCCESS in 40m 46s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/c2177c26324c4756896f2f8472fb846f : SUCCESS in 1h 44m 56s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/7fdf975740414a2cb0a2699728294196 : SUCCESS in 1h 00m 38s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/588ad8b175c649b09a37b9ad7d50870a : FAILURE in 4m 38s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/6ea931fdf8914ff8b066c665f3cd3095 : SUCCESS in 4m 42s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/c4e06d90c35e4c468ad23b2d9b568099 : SUCCESS in 6m 40s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f0f018827f2c4b128cd3c5a6b90f1ead : SUCCESS in 3m 14s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/34be9c0f08d741919fd24f8be5e78f25 : SUCCESS in 6m 54s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/6e0c96766aac454593b111fb0884240c : SUCCESS in 9m 54s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/d38c30a9d8aa4649971a2efeb45b270c : SUCCESS in 2h 04m 10s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/3f4271266dc74531a66e55923f49e3bb : SUCCESS in 1h 27m 53s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/fdbd61bbabff47c0ac747179e7d51192 : SUCCESS in 1h 06m 21s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/2f8e2f066fc144f6832a4ef8cc251112 : SUCCESS in 1h 13m 51s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/54efc9485b624fe4b57fb3c36a725ed7 : SUCCESS in 56m 38s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/06f20e04dbd34b209a9d7fd96a5a7f35 : SUCCESS in 2h 00m 19s\n- glance-grenade-centralized-cache https://zuul.opendev.org/t/openstack/build/373353d8cc4041edbcb3427aeab89fab : FAILURE in 20m 54s (non-voting)\n- glance-s3-multistore https://zuul.opendev.org/t/openstack/build/03eeac2fa66b4675a29c72f99c9b6b66 : FAILURE in 26m 45s (non-voting)\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/e298db1001a64f9eaa1a1120823bcd19 : FAILURE in 35m 15s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/831b63601dc64401b0b4b5116ac0ed2c : SUCCESS in 32m 46s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"43284f3d23f1672310b51183077ea591fed88ee3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":38656,"name":"Vishnu C","display_name":"Vishnu Chandrasenan","email":"vischan2@cisco.com","username":"vischan2"},"date":"2025-12-08 18:49:29.000000000","message":"Uploaded patch set 4.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":4},{"id":"88532126be06e0072539bbb3e2ee64728578ac4d","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-08 19:01:00.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/165766dc37204498b60412c706678c0e\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/13f35527a94f4da882b676bafbc1666e : SUCCESS in 9m 14s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/268023b714704a2d9afe5209296c24f4 : SUCCESS in 7m 05s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/d07290883c46487499db1a5d8e867ec8 : SUCCESS in 7m 55s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"4cd1ec60e832d22a7d83564b9ae222dc0524aad8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-08 20:48:43.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/a7ce7ba7d26f45d494a45913d5cf3c0c\n\n- grenade https://zuul.opendev.org/t/openstack/build/f40c614e91194234938710faec6ba38a : SUCCESS in 1h 09m 10s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/14e77b4100914e53ad45204d27f8b9ab : SUCCESS in 1h 02m 49s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/1421667cbe874eb181a65bbae6c0f1bc : SUCCESS in 1h 30m 26s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/048f2cfd3b1e4b5c838e835a56ba2323 : SUCCESS in 35m 09s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c67192c63bb84d38aafc0e5240f0a5d8 : SUCCESS in 3m 02s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/3f69cd731ada43ddbb632b03f37d0775 : POST_FAILURE in 4m 25s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/2e80d472c1124330b322c46ac872376d : SUCCESS in 9m 51s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b716fbec80714742923188be8ecac007 : SUCCESS in 5m 31s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/5d27478ae46a404a8cd935b595372a82 : SUCCESS in 5m 54s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/319f2d236483439f8117ec580799f59b : SUCCESS in 9m 15s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/b1432907fbce4d9997616bfcb44bd89e : SUCCESS in 1h 49m 10s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/b50cf54ed9bb41788fad31244e342f71 : SUCCESS in 1h 57m 35s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/72ed3be1a7ca4be2a63a83dbab8a87ca : SUCCESS in 1h 01m 49s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/e56bd3bb563349c5bbcbfc6a3209f49a : SUCCESS in 1h 48m 20s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/10125a87e7ee41408d223fb6c66169a1 : SUCCESS in 28m 41s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/1c08d6e24f0c438a9170ff750a912aa9 : SUCCESS in 1h 50m 50s\n- glance-grenade-centralized-cache https://zuul.opendev.org/t/openstack/build/f0cb681ddecc46579555ba999cd54785 : FAILURE in 15m 27s (non-voting)\n- glance-s3-multistore https://zuul.opendev.org/t/openstack/build/d326b7147d4948cc80e1eaa3812432aa : FAILURE in 24m 47s (non-voting)\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/f7d137d5cf684dbbb080dd9b8726896a : SUCCESS in 33m 45s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/ba0661cb510444528b99be63ed0a986a : SUCCESS in 1h 23m 24s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"89d67a156c528d158f8dcb5596c169461c35f8eb","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":38656,"name":"Vishnu C","display_name":"Vishnu Chandrasenan","email":"vischan2@cisco.com","username":"vischan2"},"date":"2025-12-08 22:40:58.000000000","message":"Uploaded patch set 5: Commit message was updated.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":5},{"id":"bdcfbc5e54352df61671228c4d67de1fbbd74f06","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-08 22:55:37.000000000","message":"Patch Set 5:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/256722ff4d0048cf86fa7fc5f84d9cd9\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/6635c44542ce4c3ba219b0060958247b : SUCCESS in 6m 04s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/c709888228004d82a8343e96f346c6f3 : SUCCESS in 8m 41s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/bb606e20449e45cca6d4f82977b2bb31 : SUCCESS in 7m 42s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"1af440345e6b18de20dc2d569b911240136c1c26","author":{"_account_id":21113,"name":"Min Sun","email":"minsu@cisco.com","username":"MinSun"},"date":"2025-12-08 23:48:29.000000000","message":"Patch Set 5: Code-Review+1","accounts_in_message":[],"_revision_number":5},{"id":"dbe924adb3711cbe1159d36a017e9a459c8abc46","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-09 01:05:27.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/57a890a559bb446b97f2e4e8685be852\n\n- grenade https://zuul.opendev.org/t/openstack/build/15df467ebfe74eb69256bfa48e53137d : SUCCESS in 1h 00m 23s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/fab3e8f7509941c9b93e11772c547c9b : SUCCESS in 43m 32s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/14538bdcd1f740e5b2a59f5100412d30 : SUCCESS in 1h 23m 14s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/141ece0787e4431c988dd1ba0415bae8 : SUCCESS in 58m 31s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/61eb1ce2baa24317955ff3c3684c3eb6 : SUCCESS in 3m 54s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/ca54f10a12ef4f1b85566d5a7723ac90 : SUCCESS in 6m 14s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/3b2b093be5f54c57973e12c47bbd8297 : SUCCESS in 7m 24s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1ac106e5065d40ecb50a77a6b44f5c7a : SUCCESS in 5m 15s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/eb1f0979a307404e9d04d7f8bca5bfc4 : SUCCESS in 6m 26s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/0772c311f03e4a2f84f1f99fa35c0137 : SUCCESS in 9m 12s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/1e5e3ce235ee406a9010ff41fff40e44 : FAILURE in 1h 15m 30s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/11603dbd39ad45a19fb45308a2a55c3f : SUCCESS in 1h 22m 31s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/b81836c7a6af41c4906e65e54880bc4b : SUCCESS in 1h 59m 33s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/1d80d2b2b7e245f39eaa1c16314a35e6 : SUCCESS in 2h 19m 31s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/22c2088a6be146b598f99f23b79624f0 : SUCCESS in 33m 59s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/a8381856e4eb4dcbb4a79df01765e92a : SUCCESS in 56m 17s\n- glance-grenade-centralized-cache https://zuul.opendev.org/t/openstack/build/c7fe4d7438cd417393f20ee426014cac : FAILURE in 20m 57s (non-voting)\n- glance-s3-multistore https://zuul.opendev.org/t/openstack/build/b7f0124fd9fb473da45cdcd6bb9de576 : FAILURE in 12m 55s (non-voting)\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/881375729bcb4352b6fb053a082ca112 : SUCCESS in 23m 44s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/ed3008ab3f04466d9218867ad457b872 : SUCCESS in 1h 03m 51s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"a8adc3708fb79371f8ad969d63d2edbb3f3e4656","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2025-12-09 07:15:36.000000000","message":"Patch Set 5: Code-Review+2\n\n(3 comments)","accounts_in_message":[],"_revision_number":5},{"id":"705c877729f08dfba2fbbe3f7d34359a6b38ee04","author":{"_account_id":38656,"name":"Vishnu C","display_name":"Vishnu Chandrasenan","email":"vischan2@cisco.com","username":"vischan2"},"date":"2025-12-09 21:03:58.000000000","message":"Patch Set 5:\n\n(2 comments)","accounts_in_message":[],"_revision_number":5},{"id":"0b4232f72178098b7da9dfe1ccd69776177a6b2c","author":{"_account_id":25402,"name":"Francesco Pantano","email":"fpantano@redhat.com","username":"fmount"},"date":"2025-12-10 07:38:10.000000000","message":"Patch Set 5: Code-Review+1","accounts_in_message":[],"_revision_number":5},{"id":"36d4c3c51ebd66b90c4e3f946d439563aab56a1b","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2025-12-12 00:05:44.000000000","message":"Patch Set 5: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"5fd3d56b5b0f7d9b150e9ef392ff7d36d46fb5e8","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-12 00:09:55.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"a832e00d8b68de4927da4a16090512426c8237d7","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-12 02:10:05.000000000","message":"Patch Set 5: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ed4ed00878694c5f914344cbc24f57e4\n\n- grenade https://zuul.opendev.org/t/openstack/build/e0d296fa9e494f0c8fec815b98cb2692 : SUCCESS in 55m 36s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/13e54ffa710c40ce83a9ef68a2a4c638 : SUCCESS in 1h 01m 50s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/5d5d71b9f2bc42b0ad2c26819f9e6210 : SUCCESS in 1h 33m 45s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/4a9506ec7ce848ebb213715b15b3f2fd : SUCCESS in 51m 37s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d44480b89f664798b74f65e9383a49b7 : SUCCESS in 4m 00s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/03cd8614c7a94df9861279376b137e5c : SUCCESS in 3m 12s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/0b08ccfc4ef54ffab89f5752228c89a2 : SUCCESS in 7m 47s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5ddc1e981dd34ce5afd26ffa3e040fad : SUCCESS in 6m 05s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/60e78e1da76242c39a2e15a4eb27b147 : SUCCESS in 8m 04s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/d0386a48768440ce90bc40826be73226 : SUCCESS in 6m 25s\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/cf61ae50aaa547c188c13363a5d4d369 : SUCCESS in 1h 03m 07s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/b24dc4cd5eac481791f02f8358786a27 : SUCCESS in 1h 55m 33s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/0c9959f79bdc43a1af19f6f8e0e24a6d : SUCCESS in 42m 05s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/fe1206966eb24922af92caf9e5a04b20 : SUCCESS in 1h 37m 24s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/39f90c16509140269e6a03084d88855a : SUCCESS in 16m 00s","accounts_in_message":[],"_revision_number":5},{"id":"a8f8e38211e4f7e162235594cf5f302c42ee3124","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-12 02:10:05.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":5},{"id":"4a60d82249771770590ae73c30b7b9ac5836d024","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-12 02:13:47.000000000","message":"Patch Set 5:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ad44644991bb4284963092deea6cb985\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bca2db86c9a7414e9af427dded5cdc22 : SUCCESS in 46s","accounts_in_message":[],"_revision_number":5}],"current_revision_number":5,"current_revision":"cbc4d9f5b1942308be32544825d0fcae746b4059","revisions":{"e92597b4ee002da0ee102a8b82f3cf4bc0e1d91a":{"kind":"REWORK","_number":1,"created":"2025-12-07 10:08:14.000000000","uploader":{"_account_id":38656,"name":"Vishnu C","display_name":"Vishnu Chandrasenan","email":"vischan2@cisco.com","username":"vischan2"},"ref":"refs/changes/48/970048/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/48/970048/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/48/970048/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/48/970048/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/48/970048/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/48/970048/1"}}},"commit":{"parents":[{"commit":"955d086873b9368090134aa01d5226f9b9fa19df","subject":"Merge \"Correct read function in _UWSGIChunkFile not to exceed length\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/955d086873b9368090134aa01d5226f9b9fa19df"}]}],"author":{"name":"vischan2","email":"vischan2@cisco.com","date":"2025-12-07 09:56:17.000000000","tz":-480},"committer":{"name":"vischan2","email":"vischan2@cisco.com","date":"2025-12-07 10:07:27.000000000","tz":-480},"subject":"Fix S3 image download errors from credential rotation where forward slash in credentials corrupts URLs in DB.","message":"Fix S3 image download errors from credential rotation where forward slash in credentials corrupts URLs in DB.\n\nThe _update_s3_url() function uses urlparse which incorrectly parses\nURLs when the secret key contains a forward slash (/). urlparse\ninterprets the slash as the start of the path, corrupting the netloc.\n\nThis causes the credential rotation to produce corrupted URLs in database by\nconcatenating old and new credentials, resulting in image download\nfailures after S3 credential rotation as AWS S3 secret access keys can legitimately\ncontain forward slash characters.\n\nThis fix use rfind(\u0027@\u0027) on the raw URL string instead to correctly locate the\nlast \u0027@\u0027 which always separates credentials from the host, regardless\nof special characters in the secret.\n\nCloses-Bug: #2134325\nChange-Id: I1dd1d2f31203d69be93bafe9222009f58e78500a\nSigned-off-by: vischan2 \u003cvischan2@cisco.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/e92597b4ee002da0ee102a8b82f3cf4bc0e1d91a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/e92597b4ee002da0ee102a8b82f3cf4bc0e1d91a"}]},"branch":"refs/heads/master"},"fc1a3f1f754ffe3abf935c066250e9a7f739f034":{"kind":"REWORK","_number":2,"created":"2025-12-07 10:58:30.000000000","uploader":{"_account_id":38656,"name":"Vishnu C","display_name":"Vishnu Chandrasenan","email":"vischan2@cisco.com","username":"vischan2"},"ref":"refs/changes/48/970048/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/48/970048/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/48/970048/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/48/970048/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/48/970048/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/48/970048/2"}}},"commit":{"parents":[{"commit":"955d086873b9368090134aa01d5226f9b9fa19df","subject":"Merge \"Correct read function in _UWSGIChunkFile not to exceed length\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/955d086873b9368090134aa01d5226f9b9fa19df"}]}],"author":{"name":"vischan2","email":"vischan2@cisco.com","date":"2025-12-07 09:56:17.000000000","tz":-480},"committer":{"name":"vischan2","email":"vischan2@cisco.com","date":"2025-12-07 10:57:43.000000000","tz":-480},"subject":"Fix S3 image download errors from credential rotation where forward slash in credentials corrupts URLs in DB.","message":"Fix S3 image download errors from credential rotation where forward slash in credentials corrupts URLs in DB.\n\nThe _update_s3_url() function uses urlparse which incorrectly parses\nURLs when the secret key contains a forward slash (/). urlparse\ninterprets the slash as the start of the path, corrupting the netloc.\n\nThis causes the credential rotation to produce corrupted URLs in database by\nconcatenating old and new credentials, resulting in image download\nfailures after S3 credential rotation as AWS S3 secret access keys can legitimately\ncontain forward slash characters.\n\nThis fix use rfind(\u0027@\u0027) on the raw URL string instead to correctly locate the\nlast \u0027@\u0027 which always separates credentials from the host, regardless\nof special characters in the secret.\n\nCloses-Bug: #2134325\nChange-Id: I1dd1d2f31203d69be93bafe9222009f58e78500a\nSigned-off-by: vischan2 \u003cvischan2@cisco.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/fc1a3f1f754ffe3abf935c066250e9a7f739f034"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/fc1a3f1f754ffe3abf935c066250e9a7f739f034"}]},"branch":"refs/heads/master"},"9a11e6c10cc1ddeb5951e583a02b56120f3da8bd":{"kind":"REWORK","_number":3,"created":"2025-12-08 15:53:32.000000000","uploader":{"_account_id":38656,"name":"Vishnu C","display_name":"Vishnu Chandrasenan","email":"vischan2@cisco.com","username":"vischan2"},"ref":"refs/changes/48/970048/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/48/970048/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/48/970048/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/48/970048/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/48/970048/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/48/970048/3"}}},"commit":{"parents":[{"commit":"955d086873b9368090134aa01d5226f9b9fa19df","subject":"Merge \"Correct read function in _UWSGIChunkFile not to exceed length\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/955d086873b9368090134aa01d5226f9b9fa19df"}]}],"author":{"name":"vischan2","email":"vischan2@cisco.com","date":"2025-12-07 09:56:17.000000000","tz":-480},"committer":{"name":"vischan2","email":"vischan2@cisco.com","date":"2025-12-08 15:53:20.000000000","tz":-480},"subject":"Fix S3 image download errors from credential rotation where forward slash in credentials corrupts URLs in DB.","message":"Fix S3 image download errors from credential rotation where forward slash in credentials corrupts URLs in DB.\n\nThe _update_s3_url() function uses urlparse which incorrectly parses\nURLs when the secret key contains a forward slash (/). urlparse\ninterprets the slash as the start of the path, corrupting the netloc.\n\nThis causes the credential rotation to produce corrupted URLs in database by\nconcatenating old and new credentials, resulting in image download\nfailures after S3 credential rotation as AWS S3 secret access keys can legitimately\ncontain forward slash characters.\n\nThis fix use rfind(\u0027@\u0027) on the raw URL string instead to correctly locate the\nlast \u0027@\u0027 which always separates credentials from the host, regardless\nof special characters in the secret.\n\nCloses-Bug: #2134325\nChange-Id: I1dd1d2f31203d69be93bafe9222009f58e78500a\nSigned-off-by: vischan2 \u003cvischan2@cisco.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/9a11e6c10cc1ddeb5951e583a02b56120f3da8bd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/9a11e6c10cc1ddeb5951e583a02b56120f3da8bd"}]},"branch":"refs/heads/master"},"6f1b536f8e3ff536f1615f6aee79507e2f28b636":{"kind":"REWORK","_number":4,"created":"2025-12-08 18:49:29.000000000","uploader":{"_account_id":38656,"name":"Vishnu C","display_name":"Vishnu Chandrasenan","email":"vischan2@cisco.com","username":"vischan2"},"ref":"refs/changes/48/970048/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/48/970048/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/48/970048/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/48/970048/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/48/970048/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/48/970048/4"}}},"commit":{"parents":[{"commit":"955d086873b9368090134aa01d5226f9b9fa19df","subject":"Merge \"Correct read function in _UWSGIChunkFile not to exceed length\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/955d086873b9368090134aa01d5226f9b9fa19df"}]}],"author":{"name":"vischan2","email":"vischan2@cisco.com","date":"2025-12-07 09:56:17.000000000","tz":-480},"committer":{"name":"vischan2","email":"vischan2@cisco.com","date":"2025-12-08 18:49:05.000000000","tz":-480},"subject":"Fix S3 image download errors from credential rotation where forward slash in credentials corrupts URLs in DB.","message":"Fix S3 image download errors from credential rotation where forward slash in credentials corrupts URLs in DB.\n\nThe _update_s3_url() function uses urlparse which incorrectly parses\nURLs when the secret key contains a forward slash (/). urlparse\ninterprets the slash as the start of the path, corrupting the netloc.\n\nThis causes the credential rotation to produce corrupted URLs in database by\nconcatenating old and new credentials, resulting in image download\nfailures after S3 credential rotation as AWS S3 secret access keys can legitimately\ncontain forward slash characters.\n\nThis fix use rfind(\u0027@\u0027) on the raw URL string instead to correctly locate the\nlast \u0027@\u0027 which always separates credentials from the host, regardless\nof special characters in the secret.\n\nCloses-Bug: #2134325\nChange-Id: I1dd1d2f31203d69be93bafe9222009f58e78500a\nSigned-off-by: vischan2 \u003cvischan2@cisco.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/6f1b536f8e3ff536f1615f6aee79507e2f28b636"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/6f1b536f8e3ff536f1615f6aee79507e2f28b636"}]},"branch":"refs/heads/master"},"cbc4d9f5b1942308be32544825d0fcae746b4059":{"kind":"NO_CODE_CHANGE","_number":5,"created":"2025-12-08 22:40:58.000000000","uploader":{"_account_id":38656,"name":"Vishnu C","display_name":"Vishnu Chandrasenan","email":"vischan2@cisco.com","username":"vischan2"},"ref":"refs/changes/48/970048/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/48/970048/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/48/970048/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/48/970048/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/48/970048/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/48/970048/5"}}},"commit":{"parents":[{"commit":"955d086873b9368090134aa01d5226f9b9fa19df","subject":"Merge \"Correct read function in _UWSGIChunkFile not to exceed length\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/955d086873b9368090134aa01d5226f9b9fa19df"}]}],"author":{"name":"vischan2","email":"vischan2@cisco.com","date":"2025-12-07 09:56:17.000000000","tz":-480},"committer":{"name":"vischan2","email":"vischan2@cisco.com","date":"2025-12-08 22:40:11.000000000","tz":-480},"subject":"Fix S3 URL corruption on secret rotation","message":"Fix S3 URL corruption on secret rotation\n\nThe _update_s3_url() function uses urlparse which incorrectly parses\nURLs when the secret key contains a forward slash (/). urlparse\ninterprets the slash as the start of the path, corrupting the netloc.\n\nThis causes credential rotation to produce corrupted URLs in the\ndatabase by concatenating old and new credentials, resulting in image\ndownload failures. AWS S3 secret access keys can legitimately contain\nforward slash characters.\n\nThe fix uses a hybrid approach:\n- urlparse for query/fragment extraction\n- rfind(\u0027@\u0027) on the raw URL to locate credentials (to handle \u0027/\u0027\n  in secrets correctly)\n\nUnit test updates:\n- test_update_s3_url_helper_function\n- test_update_s3_url_with_slash_in_secret\n- test_update_s3_url_preserves_query_and_fragment\n- test_update_s3_url_with_port\n\nCloses-Bug: #2134325\nChange-Id: I1dd1d2f31203d69be93bafe9222009f58e78500a\nSigned-off-by: vischan2 \u003cvischan2@cisco.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/cbc4d9f5b1942308be32544825d0fcae746b4059"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/cbc4d9f5b1942308be32544825d0fcae746b4059"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":25402,"name":"Francesco Pantano","email":"fpantano@redhat.com","username":"fmount"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
