)]}'
{"id":"openstack%2Fglance~970053","triplet_id":"openstack%2Fglance~master~Ie7a7eb4eedb2963e289ae475dc27a6602fce61f9","project":"openstack/glance","branch":"master","attention_set":{},"removed_from_attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2026-01-12 17:18:05.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ie7a7eb4eedb2963e289ae475dc27a6602fce61f9","subject":"Use Ubuntu Noble to run py312 job","status":"MERGED","created":"2025-12-07 14:52:34.000000000","updated":"2026-01-12 17:22:57.000000000","submitted":"2026-01-12 17:18:05.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"970053","meta_rev_id":"002a7a91eb4e55d17e16e95c46dcaa8054790f44","_number":970053,"virtual_id_number":970053,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-01-12 17:18:04.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"all":[{"value":2,"date":"2026-01-06 14:58:34.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"all":[{"value":1,"date":"2026-01-12 15:01:18.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Top Priority / Holds Gate"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-12-07 15:06:07.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2025-12-07 17:23:30.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-01-06 14:58:34.000000000","updated_by":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"reviewer":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"state":"REVIEWER"}],"messages":[{"id":"1377b7daf414f0718dd780f68bd239c2fffe3a38","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2025-12-07 14:52:34.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"cfe601ea343ceb97188d542b0e1027f7edd9bdec","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-07 15:06:07.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7c1ce399b9da4fce9f74922cd53ed6cd\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/a9ce886e5e1f46abb95c80fab3c0f188 : SUCCESS in 8m 40s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/014b80adb9524355a1d0fefa85720961 : SUCCESS in 9m 11s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/4fe78d7c306348c39bf33ece1c8c0a91 : SUCCESS in 7m 45s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"b0bade819f637bc734878f0e5ba4648318bd78cc","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2025-12-07 15:09:17.000000000","message":"Patch Set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"bae9e79b1fae872d0a3d8d53cd3baea13929b949","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-07 15:31:21.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/bf790a5688c848e0b1291b33d36972bf\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/34c48e3323b740f2967353b4845b406f : SUCCESS in 7m 20s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/ad6c46ede81849cea3eef85d3cfd052b : SUCCESS in 9m 46s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/8f875673f3e24ca994ffa4a66896c7b5 : SUCCESS in 14m 15s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"7595b353e967c893e0dcd6de8dc169a20db93b94","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-07 17:23:30.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7951dd4b91ab44698f9329fa4829a04a\n\n- grenade https://zuul.opendev.org/t/openstack/build/a4e35fd160434a398c7779f13aa2b09b : SUCCESS in 1h 01m 29s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/a59b16c3171945dca14cfb87ca1a2790 : SUCCESS in 31m 03s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/8ee108c9262e480cbd160fa09eb531f0 : SUCCESS in 1h 32m 09s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/40439a073e444b14af603d36bc3ffcb9 : SUCCESS in 1h 20m 42s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d8d6016021c04882be1faef3b3eaea30 : SUCCESS in 4m 04s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/a826f5504fce4601b6d41eca9b441f7d : SUCCESS in 4m 36s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/1aa035b9d107437f91841ca2cad52ece : SUCCESS in 7m 46s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5d647455e68949199f7b2858f494de25 : SUCCESS in 4m 30s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/8ab05a8a018a455284d4b3d3161553f9 : SUCCESS in 4m 34s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/245269b909ee4abc8956dfd217e143fe : SUCCESS in 6m 22s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/371d276aa136421ca5fc1f864a65455e : SUCCESS in 1h 11m 51s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/eaaaec22d483476896829b7ec6f8b14a : SUCCESS in 59m 08s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/94eced0e57bd4b4382f31f53762af290 : SUCCESS in 1h 51m 59s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/171e728a5ddb4b538aabf0d81dfb3f04 : SUCCESS in 2h 05m 48s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/f21d179a629043f3b40e7978398efecd : SUCCESS in 30m 54s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/e3d6dcd6c20f444ea8e646a42fa33811 : SUCCESS in 1h 47m 41s\n- glance-grenade-centralized-cache https://zuul.opendev.org/t/openstack/build/9e3ebf7eb8424190b739d61caf8e4eb1 : FAILURE in 11m 48s (non-voting)\n- glance-s3-multistore https://zuul.opendev.org/t/openstack/build/713e4465445f47929cbba84e67fe63bf : FAILURE in 25m 10s (non-voting)\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/825c713fb7124eb482af22f3753b94c0 : SUCCESS in 25m 06s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/fe523c241bc6448ab18d8361ca7f34b2 : SUCCESS in 1h 12m 50s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"a87445a7db4a33bc8ebfa731aa1710288de9674f","author":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"date":"2026-01-06 14:58:34.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"6957b110797b813e65a267e75fd12db14a0db273","author":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"date":"2026-01-12 15:01:18.000000000","message":"Patch Set 2: Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"68bfe1a1297b210136817c3ed1d9b2e7a8a2bd2f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-12 15:03:17.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"123b634fa7c15d914e53ae871e5c2aeea89a4383","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-12 17:18:04.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/aeb1209f3439434386ab869925aacdef\n\n- grenade https://zuul.opendev.org/t/openstack/build/0fd88ddea61c4d968cf4376805a8526b : SUCCESS in 46m 06s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/ed2a45706f4742cb8bd89872d6b0bc80 : SUCCESS in 54m 46s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/f17af260c7c848688bb63d79ce4bed73 : SUCCESS in 1h 24m 28s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/983d82c6e37148b7978bc0b09f47d57f : SUCCESS in 1h 00m 56s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/25a80d460df9433b945d1a12a4968dbf : SUCCESS in 4m 22s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/9c1551bdd0e04874a97944bc54ce9abd : SUCCESS in 5m 48s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/8ce0e5367e6847f2adf856949e44b91a : SUCCESS in 10m 06s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a251585d9f284e0aad60d19df5557a71 : SUCCESS in 5m 07s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/65c0962f71404ac78f3570ba02abbaba : SUCCESS in 6m 15s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/8550920f523b4651a5bb465f69faaffe : SUCCESS in 7m 42s\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/0dd3258136e74778a08ccdb3d67e9cd0 : SUCCESS in 1h 07m 54s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/d6d9463cbb7645368e758ef5c6e5edea : SUCCESS in 2h 06m 19s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/9f4e990e09584ba7bc49fdb58432f506 : SUCCESS in 58m 14s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/9f18726ea0bf446f8e0caa12ee69da92 : SUCCESS in 1h 46m 26s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/ec03258efbc047fa9ec2b91f1b686b36 : SUCCESS in 15m 46s","accounts_in_message":[],"_revision_number":2},{"id":"1e55c04cadd29ea724df2fbf5362d501ea58f5f1","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-12 17:18:05.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"002a7a91eb4e55d17e16e95c46dcaa8054790f44","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-12 17:22:57.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/58c3f575a62a4e48ad5c1a4cf44cb7d5\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/039d7db12c314f61a440a165dce503e5 : SUCCESS in 1m 00s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"d6d49f10602676f2dfca79c36a6fd619c2b996dd","revisions":{"bb67c231c34061c68512e690f7787e3c97e3f7fc":{"kind":"REWORK","_number":1,"created":"2025-12-07 14:52:34.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/53/970053/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/53/970053/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/53/970053/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/53/970053/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/53/970053/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/53/970053/1"}}},"commit":{"parents":[{"commit":"8a5aa7a98b12de24b5732eb69c5525c70fbc8fb2","subject":"Fix FileNotFoundError when registering task operations","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/8a5aa7a98b12de24b5732eb69c5525c70fbc8fb2"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-12-07 14:49:47.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-12-07 14:50:21.000000000","tz":540},"subject":"Use Ubuntu Noble to run py312 job","message":"Use Ubuntu Noble to run py312 job\n\nPython 3.12 is the default python in Ubuntu Noble. Using Noble allows\nus to use the native python package which is more efficient than using\npyenv which builds the target python.\n\nChange-Id: Ie7a7eb4eedb2963e289ae475dc27a6602fce61f9\nSigned-off-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/bb67c231c34061c68512e690f7787e3c97e3f7fc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/bb67c231c34061c68512e690f7787e3c97e3f7fc"}]},"branch":"refs/heads/master"},"d6d49f10602676f2dfca79c36a6fd619c2b996dd":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2025-12-07 15:09:17.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/53/970053/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/53/970053/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/53/970053/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/53/970053/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/53/970053/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/53/970053/2"}}},"commit":{"parents":[{"commit":"8a5aa7a98b12de24b5732eb69c5525c70fbc8fb2","subject":"Fix FileNotFoundError when registering task operations","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/8a5aa7a98b12de24b5732eb69c5525c70fbc8fb2"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-12-07 14:49:47.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-12-07 15:09:17.000000000","tz":0},"subject":"Use Ubuntu Noble to run py312 job","message":"Use Ubuntu Noble to run py312 job\n\nPython 3.12 is the default python in Ubuntu Noble. Using Noble allows\nus to use the native python package which is more efficient than using\npyenv.\n\nChange-Id: Ie7a7eb4eedb2963e289ae475dc27a6602fce61f9\nSigned-off-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/d6d49f10602676f2dfca79c36a6fd619c2b996dd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/d6d49f10602676f2dfca79c36a6fd619c2b996dd"}]},"branch":"refs/heads/master","description":"Edit commit message"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
