)]}'
{"id":"openstack%2Fglance~970054","triplet_id":"openstack%2Fglance~master~I9d7a48055a7ebb7729267ffda92a7dd61ef64e3d","project":"openstack/glance","branch":"master","topic":"py313-trixie","attention_set":{},"removed_from_attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2026-01-12 17:18:08.000000000","reason":"Change was submitted"},"8122":{"account":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"last_update":"2026-01-07 12:31:10.000000000","reason":"\u003cGERRIT_ACCOUNT_8122\u003e replied on the change","reason_account":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"}}},"hashtags":[],"change_id":"I9d7a48055a7ebb7729267ffda92a7dd61ef64e3d","subject":"Migrate py313 job to Debian Trixie","status":"MERGED","created":"2025-12-07 14:52:34.000000000","updated":"2026-01-12 17:25:07.000000000","submitted":"2026-01-12 17:18:08.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":4,"unresolved_comment_count":1,"has_review_started":true,"submission_id":"970054-py313-trixie","meta_rev_id":"315c755a40d19c418bc03812531a3a5cb4e17b47","_number":970054,"virtual_id_number":970054,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-01-12 17:18:08.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"all":[{"value":2,"date":"2026-01-07 12:31:10.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"all":[{"value":1,"date":"2026-01-12 15:01:11.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Top Priority / Holds Gate"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-12-07 14:53:27.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-01-07 09:42:09.000000000","updated_by":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"reviewer":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"state":"CC"},{"updated":"2026-01-07 12:31:10.000000000","updated_by":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"reviewer":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"state":"REVIEWER"}],"messages":[{"id":"43262b2f51e48a796eadb310d56effc74abb3696","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2025-12-07 14:52:34.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"ac893f05aa105512940c9657dd2d37a4b355bcf0","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2025-12-07 14:53:02.000000000","message":"Topic set to py313-trixie","accounts_in_message":[],"_revision_number":1},{"id":"170c653795f71338dc43d85721d8f8b0893643ba","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-07 14:53:27.000000000","message":"Patch Set 1: Verified-1\n\n(7 comments)\n\nZuul encountered a syntax error while parsing its\nconfiguration in the repo openstack/glance on branch master.  The\nproblem was:\n\n  extra keys not allowed @ data[\u0027node\u0027]\n\nThe problem appears in the the \"glance-tox-functional-py313-oslo-tips\" job stanza:\n\n  job:\n      name: glance-tox-functional-py313-oslo-tips\n      parent: glance-tox-oslo-tips-base\n      node: debian-trixie-8GB\n      description: |\n        Glance py313 functional tests vs. oslo libraries masters\n      vars:\n        python_version: 3.13\n        tox_envlist: functional-py313\n  ...\n\n  in \"openstack/glance/.zuul.yaml@master\", line 41","accounts_in_message":[],"_revision_number":1},{"id":"584914732f32b35f294d912c4c575385cdbe4660","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-07 14:54:06.000000000","message":"Patch Set 1:\n\n(7 comments)\n\nZuul encountered a syntax error while parsing its\nconfiguration in the repo openstack/glance on branch master.  The\nproblem was:\n\n  extra keys not allowed @ data[\u0027node\u0027]\n\nThe problem appears in the the \"glance-tox-functional-py313-oslo-tips\" job stanza:\n\n  job:\n      name: glance-tox-functional-py313-oslo-tips\n      parent: glance-tox-oslo-tips-base\n      node: debian-trixie-8GB\n      description: |\n        Glance py313 functional tests vs. oslo libraries masters\n      vars:\n        python_version: 3.13\n        tox_envlist: functional-py313\n  ...\n\n  in \"openstack/glance/.zuul.yaml@master\", line 41","accounts_in_message":[],"_revision_number":1},{"id":"aecc8b8fd5d8a49ce23d05f745a44dcba61c992f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2025-12-07 14:54:38.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"c8c010ab1dcc230913677cee017f03d1f7aa1616","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2025-12-07 15:09:38.000000000","message":"Patch Set 3: Commit message was updated.","accounts_in_message":[],"_revision_number":3},{"id":"6dc21a2884bffe1e91f6e39c6281b833d5ab49e8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2025-12-07 15:09:42.000000000","message":"Patch Set 4: Patch Set 3 was rebased","accounts_in_message":[],"_revision_number":4},{"id":"db4ab6375654150a3e04595cfcb54ec2d691570c","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-07 15:31:23.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/cc88011385ce4a86a8178c83bb264a32\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/17d0dcaf2b584a90be4c2bd5a201127d : SUCCESS in 6m 27s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/62a597cf3cf64ba7b8738a58d248f93e : SUCCESS in 9m 18s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/8fdd79c8e5a945adbd4ca2c96d3ace51 : SUCCESS in 14m 16s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"42a7ddd844a0a29268ae13b11807ab755276d9a5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2025-12-07 16:47:54.000000000","message":"Patch Set 5: Published edit on patch set 4.","accounts_in_message":[],"_revision_number":5},{"id":"a0207fbc463c434a47f5d43b4f7a3b5b7ec78ec3","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-07 17:08:47.000000000","message":"Patch Set 5:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b52a561472d549da9603753caf097990\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/bf432f92519f482e9bb42a3d23ed2812 : SUCCESS in 8m 37s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/2c1d133de05d4c8aa6b595c23bad276b : SUCCESS in 7m 37s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/bd9faacdabf7457e803c8076f7932ebd : SUCCESS in 13m 59s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"7143c1c1fad37d8bcb5d2b4b1a0971f0d846494f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-07 19:58:27.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/bf2c08ea2d6543f689f92f0fcf779244\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/df1797475fc849b49c6a89251940ac76 : SUCCESS in 3m 59s\n- grenade https://zuul.opendev.org/t/openstack/build/0eff9194b1964767b22db57b63be2bea : SUCCESS in 55m 47s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/9a33179a46be49deabd00df24443c493 : SUCCESS in 57m 31s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/043b5b6adcb54845b2bb26d4493cc300 : SUCCESS in 1h 46m 52s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/21051b962b914919a0b311830758a53c : SUCCESS in 1h 09m 27s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9af1325673264e8fa0b9a82edc0d4066 : SUCCESS in 4m 41s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/95befe7da9474cc880aa85549bd240da : SUCCESS in 3m 04s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/16e790172609473b834339be88ee6dba : SUCCESS in 7m 52s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e038a412ce0e4eed89ee70c15a03d93f : SUCCESS in 5m 17s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b7da6b7945ab466984bfa53a599f9bd8 : SUCCESS in 7m 46s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/8e16bbfc3a774f6eb257f4cee7bfd646 : SUCCESS in 6m 15s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/355268bb586c4221b7d84011501d67ed : SUCCESS in 6m 36s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/efa5e8323ccf4826b1ea770d63d153ab : SUCCESS in 2h 03m 35s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/89ba3114735b4913b144e772b3ec2a49 : SUCCESS in 1h 08m 12s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/353e9432b02c46b5a2cbe7268ee0a1da : SUCCESS in 1h 56m 13s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/a6ca49bb5a124748a151299eec79fc16 : SUCCESS in 1h 49m 22s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/5e3e70498e3a4ad7a6ce5729de230136 : SUCCESS in 55m 51s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/6d245dfd2e824d548d3d98c65b7d8706 : SUCCESS in 2h 02m 02s\n- glance-grenade-centralized-cache https://zuul.opendev.org/t/openstack/build/b3be10cdc418497db53589453c42c398 : FAILURE in 19m 50s (non-voting)\n- glance-s3-multistore https://zuul.opendev.org/t/openstack/build/1b48ea28bea04c2898b291dfb0fca2bc : FAILURE in 11m 02s (non-voting)\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/15d54c7cbc10453d804fbb6b2113466d : SUCCESS in 32m 56s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/dddd0556502540fdbb3683c1aa20d464 : SUCCESS in 1h 16m 30s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"31b4a1a29fa3ce58931de42d2f9de6dd9e34c57f","author":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"date":"2026-01-07 09:42:09.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"ce1f8c9387b93fa32ee03ceeecf455d0d5cd9a35","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2026-01-07 10:22:29.000000000","message":"Uploaded patch set 6: Commit message was updated.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":6},{"id":"9b93af624d8221bfd268121774010ebc24c179fd","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2026-01-07 10:23:01.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"ff9b4d28ca178ec43aeb18a4cdc20fe2d081ff86","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-07 11:24:01.000000000","message":"Patch Set 6:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/25cda1e7af9047b9960b1b2d66a380fe\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/3e09f2be22e84037b3e24c6599a5a156 : SUCCESS in 5m 04s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/e911ac4ee3e2450e9dfa16c573b44b4a : SUCCESS in 5m 56s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/0970fca511e5490abe183a82943663f2 : SUCCESS in 8m 09s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"bbcad0dd060dade370a50a6eb8c3ff2f894e3a44","author":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"date":"2026-01-07 12:31:10.000000000","message":"Patch Set 6: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"79eb0117db5756079d9f5bcce49c547f49f07329","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-07 12:54:07.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/876ee9bd8f924f28b66162a27cdd02d3\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/3fb5a4934648449a9abe6fa36028c2aa : SUCCESS in 4m 15s\n- grenade https://zuul.opendev.org/t/openstack/build/19d0964f03b84d2db6b967aa16f712a1 : SUCCESS in 29m 10s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/63a8d888b6784f2a9b738c3eb51405f0 : SUCCESS in 44m 48s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/71f8cd2c1ff44f29a11695c0fafd63b7 : SUCCESS in 1h 30m 21s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/e2cedbdb0d6a482d9cd7e40d7fd4ab5d : SUCCESS in 35m 24s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f77fef456fe14cc1a27b40f3d8df99e9 : SUCCESS in 4m 47s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/3f92f5f4729a4d41b86d1d2df06a0863 : SUCCESS in 4m 49s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/2a96455b261b427cab1cde6bd31759bd : SUCCESS in 6m 34s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0c0520be33c845b98a50c72996f79254 : SUCCESS in 5m 19s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/95f1720829c84fc88b892b703ccc7dc9 : SUCCESS in 3m 52s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/180c8c7462394338a22b1687f4b8dabc : SUCCESS in 4m 25s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/66e41cdd8f72466ca02b5cf20c3b5bc3 : SUCCESS in 5m 19s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/cee646724ceb409a8ea9e06fd042d26c : SUCCESS in 1h 56m 01s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/53b4ef9dddfa4ae5997e8e62cb2f651c : SUCCESS in 1h 28m 33s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/5b39ec05acf5424cb6c5e4a94604fd30 : SUCCESS in 1h 56m 41s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/aab408db17bf455b8aede82c482aa052 : SUCCESS in 2h 21m 46s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/4874b5b8036e4088a9388ed75afe1abd : SUCCESS in 30m 29s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/02e2adcf7eab4db19164002dc035186a : SUCCESS in 1h 44m 03s\n- glance-grenade-centralized-cache https://zuul.opendev.org/t/openstack/build/a74e3271884e4793a8e02be310ecd438 : FAILURE in 21m 30s (non-voting)\n- glance-s3-multistore https://zuul.opendev.org/t/openstack/build/aab4b4fdf758495d82a576fb0ea00b41 : FAILURE in 30m 24s (non-voting)\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/5706d5bc7794485ca1a0dfbb83e3a760 : SUCCESS in 30m 12s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/c51576adc19149a4a5312539d5468650 : SUCCESS in 1h 07m 24s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"a854c6df97291d9721759d91fbed4f9a181e662f","author":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"date":"2026-01-12 15:01:11.000000000","message":"Patch Set 6: Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"8df1b289dc121ed946edaab698d0dc44092580c3","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-12 15:03:18.000000000","message":"Patch Set 6: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":6},{"id":"7e7e67c26117dbf887a343d751b5c9bb4e5905aa","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-12 17:18:08.000000000","message":"Patch Set 6: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/255f62db49c14af0b3c6939a2f0daefe\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/dee3e9ee1437426084041babd9ac973c : SUCCESS in 5m 42s\n- grenade https://zuul.opendev.org/t/openstack/build/15cb81b70599428cb6d4e10823052cce : SUCCESS in 56m 33s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/f0dc2bab2dd64bcfafaff260bccd59f7 : SUCCESS in 56m 06s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/dee60411eb5a46f0b44529a0c83f6a55 : SUCCESS in 1h 23m 20s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/b07a5a908f8f4306a838e7228179e2fc : SUCCESS in 37m 06s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/376d6b938fde4eac865638c939c6cf4c : SUCCESS in 6m 22s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/2b1a391492f543c29b82dcb4a340121c : SUCCESS in 5m 14s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/9b6e249c63184969b5740d76d672e152 : SUCCESS in 8m 28s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0712452d7b65463297c8799c9792f492 : SUCCESS in 6m 38s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ded6499640db4bde9c7871ccd271f6ad : SUCCESS in 7m 26s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/fbb262ba77354dd29f3e8f6ca8cc72af : SUCCESS in 7m 08s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/b8940666448b4d0bb036d8e37d1fd86f : SUCCESS in 7m 35s\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/36501703259e4576a054057c5db9d3b6 : SUCCESS in 1h 57m 29s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/6ff85d0775c948238422ce0afb1d1758 : SUCCESS in 2h 04m 40s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/4a59fa1f91a445b6950d408ed21a8804 : SUCCESS in 55m 13s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/ba67bb1c198b45978125182503bfce98 : SUCCESS in 1h 47m 54s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/ee9dbc0709334a1197effbf7930b0578 : SUCCESS in 32m 26s","accounts_in_message":[],"_revision_number":6},{"id":"8b04bd571d07a09d2330c8cfd37439d1909d5ae5","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-12 17:18:08.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":6},{"id":"315c755a40d19c418bc03812531a3a5cb4e17b47","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-12 17:25:07.000000000","message":"Patch Set 6:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8f3e470a2ff44d86a7b1f1861df4e983\n\n- promote-api-ref https://zuul.opendev.org/t/openstack/build/b9ed8992e0c14cd6b1b8f603e32d30d0 : SUCCESS in 1m 22s\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e6ed7e74ea3a4c4397e5a1630a5938d2 : SUCCESS in 1m 20s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/610d67d5b87b4475800d32149ea44536 : SUCCESS in 1m 01s","accounts_in_message":[],"_revision_number":6}],"current_revision_number":6,"current_revision":"3e148ba2f1f3a1d81f60423099a436b29575a15e","revisions":{"5b8c99a261be498fc50040203919f87568439f96":{"kind":"REWORK","_number":1,"created":"2025-12-07 14:52:34.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/54/970054/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/54/970054/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/54/970054/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/54/970054/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/54/970054/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/54/970054/1"}}},"commit":{"parents":[{"commit":"bb67c231c34061c68512e690f7787e3c97e3f7fc","subject":"Use Ubuntu Noble to run py312 job","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/bb67c231c34061c68512e690f7787e3c97e3f7fc"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-12-07 14:51:38.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-12-07 14:52:25.000000000","tz":540},"subject":"Migrate py313 job to Debian Trixie","message":"Migrate py313 job to Debian Trixie\n\nPython 3.13 is the default version in Debian Trixie so we can use\nmore efficient distro build.\n\nPin py313 job in stable/2025.2 because Trixie is not part of its\nsupported platforms.\n\nChange-Id: I9d7a48055a7ebb7729267ffda92a7dd61ef64e3d\nSigned-off-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/5b8c99a261be498fc50040203919f87568439f96"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/5b8c99a261be498fc50040203919f87568439f96"}]},"branch":"refs/heads/master"},"f21472b005f06b845c0f6623058f532efbe54288":{"kind":"REWORK","_number":2,"created":"2025-12-07 14:54:38.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/54/970054/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/54/970054/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/54/970054/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/54/970054/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/54/970054/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/54/970054/2"}}},"commit":{"parents":[{"commit":"bb67c231c34061c68512e690f7787e3c97e3f7fc","subject":"Use Ubuntu Noble to run py312 job","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/bb67c231c34061c68512e690f7787e3c97e3f7fc"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-12-07 14:51:38.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-12-07 14:54:29.000000000","tz":540},"subject":"Migrate py313 job to Debian Trixie","message":"Migrate py313 job to Debian Trixie\n\nPython 3.13 is the default version in Debian Trixie so we can use\nmore efficient distro build.\n\nPin py313 job in stable/2025.2 because Trixie is not part of its\nsupported platforms.\n\nChange-Id: I9d7a48055a7ebb7729267ffda92a7dd61ef64e3d\nSigned-off-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/f21472b005f06b845c0f6623058f532efbe54288"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/f21472b005f06b845c0f6623058f532efbe54288"}]},"branch":"refs/heads/master"},"ac3c1ebe8578b0ec74d7b0f432ff43be33f04673":{"kind":"NO_CODE_CHANGE","_number":3,"created":"2025-12-07 15:09:38.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/54/970054/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/54/970054/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/54/970054/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/54/970054/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/54/970054/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/54/970054/3"}}},"commit":{"parents":[{"commit":"bb67c231c34061c68512e690f7787e3c97e3f7fc","subject":"Use Ubuntu Noble to run py312 job","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/bb67c231c34061c68512e690f7787e3c97e3f7fc"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-12-07 14:51:38.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-12-07 15:09:38.000000000","tz":0},"subject":"Migrate py313 job to Debian Trixie","message":"Migrate py313 job to Debian Trixie\n\nPython 3.13 is the default version in Debian Trixie so we can use\nmore efficient package provided by the distribution.\n\nChange-Id: I9d7a48055a7ebb7729267ffda92a7dd61ef64e3d\nSigned-off-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/ac3c1ebe8578b0ec74d7b0f432ff43be33f04673"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/ac3c1ebe8578b0ec74d7b0f432ff43be33f04673"}]},"branch":"refs/heads/master","description":"Edit commit message"},"658b2e9dcedf40673efcf6137ade4de2ad92868d":{"kind":"NO_CHANGE","_number":4,"created":"2025-12-07 15:09:42.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/54/970054/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/54/970054/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/54/970054/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/54/970054/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/54/970054/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/54/970054/4"}}},"commit":{"parents":[{"commit":"d6d49f10602676f2dfca79c36a6fd619c2b996dd","subject":"Use Ubuntu Noble to run py312 job","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/d6d49f10602676f2dfca79c36a6fd619c2b996dd"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-12-07 14:51:38.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-12-07 15:09:42.000000000","tz":0},"subject":"Migrate py313 job to Debian Trixie","message":"Migrate py313 job to Debian Trixie\n\nPython 3.13 is the default version in Debian Trixie so we can use\nmore efficient package provided by the distribution.\n\nChange-Id: I9d7a48055a7ebb7729267ffda92a7dd61ef64e3d\nSigned-off-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/658b2e9dcedf40673efcf6137ade4de2ad92868d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/658b2e9dcedf40673efcf6137ade4de2ad92868d"}]},"branch":"refs/heads/master","description":"Rebase"},"b3d5b636c30ee27200fb1570f7dd49421de342f3":{"kind":"REWORK","_number":5,"created":"2025-12-07 16:47:54.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/54/970054/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/54/970054/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/54/970054/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/54/970054/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/54/970054/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/54/970054/5"}}},"commit":{"parents":[{"commit":"d6d49f10602676f2dfca79c36a6fd619c2b996dd","subject":"Use Ubuntu Noble to run py312 job","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/d6d49f10602676f2dfca79c36a6fd619c2b996dd"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-12-07 14:51:38.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-12-07 16:47:54.000000000","tz":0},"subject":"Migrate py313 job to Debian Trixie","message":"Migrate py313 job to Debian Trixie\n\nPython 3.13 is the default version in Debian Trixie so we can use\nmore efficient package provided by the distribution.\n\nChange-Id: I9d7a48055a7ebb7729267ffda92a7dd61ef64e3d\nSigned-off-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/b3d5b636c30ee27200fb1570f7dd49421de342f3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/b3d5b636c30ee27200fb1570f7dd49421de342f3"}]},"branch":"refs/heads/master"},"3e148ba2f1f3a1d81f60423099a436b29575a15e":{"kind":"NO_CODE_CHANGE","_number":6,"created":"2026-01-07 10:22:29.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/54/970054/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/54/970054/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/54/970054/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/54/970054/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/54/970054/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/54/970054/6"}}},"commit":{"parents":[{"commit":"d6d49f10602676f2dfca79c36a6fd619c2b996dd","subject":"Use Ubuntu Noble to run py312 job","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/d6d49f10602676f2dfca79c36a6fd619c2b996dd"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-12-07 14:51:38.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-01-07 10:21:15.000000000","tz":540},"subject":"Migrate py313 job to Debian Trixie","message":"Migrate py313 job to Debian Trixie\n\nPython 3.13 is the default version in Debian Trixie so we can use\nmore efficient package provided by the distribution.\n\nNote that this also limits installation of PCRE(3) in doc target\nbecause it is required by whereto, which is only called in the doc\ntarget. We need this change because PCRE3 is no longer available in\nDebian Trixie[1].\n\n[1] https://lists.openstack.org/archives/list/openstack-discuss@lists.openstack.org/message/KVXZIET5ZK3C6I7S45O4WWYPGBVFDIWC/\n\nChange-Id: I9d7a48055a7ebb7729267ffda92a7dd61ef64e3d\nSigned-off-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/3e148ba2f1f3a1d81f60423099a436b29575a15e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/3e148ba2f1f3a1d81f60423099a436b29575a15e"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
