)]}'
{"id":"openstack%2Fglance~970059","triplet_id":"openstack%2Fglance~master~I6e2e30e52f0dd8c72c90d15406d940bd766a3d03","project":"openstack/glance","branch":"master","topic":"py313","attention_set":{},"removed_from_attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2026-01-13 15:23:19.000000000","reason":"\u003cGERRIT_ACCOUNT_9816\u003e replied on the change","reason_account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"}}},"hashtags":[],"change_id":"I6e2e30e52f0dd8c72c90d15406d940bd766a3d03","subject":"Migrate upper functional job to Python 3.13","status":"MERGED","created":"2025-12-07 15:06:51.000000000","updated":"2026-01-13 20:01:18.000000000","submitted":"2026-01-13 19:58:14.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"970059-py313","meta_rev_id":"53c4bc87aac95415dfb2f1678acfa7ce96490412","_number":970059,"virtual_id_number":970059,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-01-13 19:58:14.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"all":[{"value":2,"date":"2026-01-07 12:31:31.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"all":[{"value":1,"date":"2026-01-12 18:43:53.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Top Priority / Holds Gate"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-12-07 15:33:57.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2025-12-07 17:02:53.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-12-08 13:20:55.000000000","updated_by":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"reviewer":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"state":"REVIEWER"}],"messages":[{"id":"ab5fb83ed5fe1a220e2569d0656a205e75ab8b72","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2025-12-07 15:06:51.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"f06df7c9921817346d22226e806af5c82fecd002","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2025-12-07 15:09:48.000000000","message":"Patch Set 2: Patch Set 1 was rebased","accounts_in_message":[],"_revision_number":2},{"id":"e139c78548f474ab5cce9bafdc17445219116d1c","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-07 15:33:57.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6ae978dadb984567bec8b33b7a677be4\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/5a02af143d98402abf916ca4b27a638b : SUCCESS in 8m 17s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/d6444ad7f69047a3b3325ff4f4c45963 : SUCCESS in 9m 20s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/e6770c213b444e38995f06f722d29fee : SUCCESS in 7m 59s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"8d04ceed8b8340b9f15f74672de88aca99e88087","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2025-12-07 16:48:00.000000000","message":"Patch Set 3: Patch Set 2 was rebased","accounts_in_message":[],"_revision_number":3},{"id":"d16df5d8fd09f4eb955d12c9b0923453b1f1d4e6","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-07 17:02:50.000000000","message":"Patch Set 3:\n\nThis change depends on a change that failed to merge.\n\nChange https://review.opendev.org/c/openstack/nova/+/970011 is needed.","accounts_in_message":[],"_revision_number":3},{"id":"ed47f50a8a9978fa2378ce0023890d9f78c5c2df","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-07 17:02:53.000000000","message":"Patch Set 3: Verified-1\n\nThis change depends on a change that failed to merge.\n\nChange https://review.opendev.org/c/openstack/nova/+/970011 is needed.","accounts_in_message":[],"_revision_number":3},{"id":"0abc4ca8485cb682e0a3ce13b7559ad4b12030de","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2025-12-07 17:24:43.000000000","message":"Patch Set 4: Commit message was updated.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":4},{"id":"7cbf6dda68b832e1d55526829ddbbfe1b357a39b","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-07 17:41:54.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9e5bc0759a27415ab846e677fbe1ff1c\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/519a8f543b924e0f8ee166008fac9fca : SUCCESS in 9m 17s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/e1a9778cb7bf4db69803909559e9735d : SUCCESS in 9m 46s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/a96523540f904b71a40c4c28948879e5 : SUCCESS in 13m 02s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"9992ab3ebb1c0ad9a5ece8db703f3a99c2cea839","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-07 19:40:38.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7d3e52ae54d149a5abf5626b30ca3b1b\n\n- grenade https://zuul.opendev.org/t/openstack/build/60be65ba6cfa4e0ab6bdd990c4d1830c : SUCCESS in 58m 43s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/286b6c1b394c42c2af96bf25525175b6 : SUCCESS in 47m 30s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/87ebcc2097fe4c2d9b5541c705f5459b : SUCCESS in 1h 25m 56s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/90b692a20ae34324bba9cc6d99342af4 : SUCCESS in 35m 08s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/52a60b58d9e94a5786eb7da0263c3305 : SUCCESS in 4m 42s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/999d0c6a06924a4fa20265aeb3860835 : SUCCESS in 4m 49s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/3977edc9bb454066a1d31ada9a8dcee5 : SUCCESS in 7m 55s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e8d33d4fae7940e7aae065caeb9ec843 : SUCCESS in 7m 00s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/025eb2fbf777410e9d683082146abb97 : SUCCESS in 7m 11s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/9f108f442e444e529b8e26436000562d : SUCCESS in 5m 56s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/6da402a8e50141aebf219218624f91c6 : SUCCESS in 1h 46m 43s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/fd92c8a2001147aaa2574a5a721bae88 : SUCCESS in 2h 03m 41s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/5250660d314946d39cb601a7e2010d80 : SUCCESS in 2h 00m 11s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/31489462d3f14e1285a8492fe7656ac6 : SUCCESS in 2h 07m 35s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/0dca2daf3e784d72a66fcc8a36a22d77 : SUCCESS in 44m 05s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/060d920835aa4e49a3a78f132882fa2c : SUCCESS in 1h 00m 23s\n- glance-grenade-centralized-cache https://zuul.opendev.org/t/openstack/build/be1f698706a047f0b187918ab60854d9 : FAILURE in 21m 13s (non-voting)\n- glance-s3-multistore https://zuul.opendev.org/t/openstack/build/55cd8859e60f4343a02eb63c34a532dc : FAILURE in 25m 35s (non-voting)\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/8c89fdb18b2b4d1eb3ff3005fbdba743 : SUCCESS in 17m 05s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/4e4bde5b67874693a7389f1cd02eddf5 : SUCCESS in 34m 20s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"d23b3ef0f875563211ef3081af42f745b9149434","author":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"date":"2025-12-08 13:20:55.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":4},{"id":"f306385068db023bbf2b1e1f5b07777a7cc2276d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-08 13:20:59.000000000","message":"Patch Set 4: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/f3f276ab3d43422f9e7383aa8e08b455\n\n\nWarning:\n  Change 960639,2 in project openstack/openstack-zuul-jobs does not share a change queue with 970059 in project openstack/glance","accounts_in_message":[],"_revision_number":4},{"id":"a1bea24694163a9685be16a8ce5461cb136cf964","author":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"date":"2025-12-08 13:27:36.000000000","message":"Patch Set 4: -Code-Review -Workflow","accounts_in_message":[],"_revision_number":4},{"id":"2941483b93761fa6ecadcbe622543b4109956ebd","author":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"date":"2026-01-06 14:59:17.000000000","message":"Patch Set 4:\n\nrecheck","accounts_in_message":[],"_revision_number":4},{"id":"908ce0c130e268fae31e5f1641eecfe9896fafea","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-06 15:16:10.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/47c88907f62849e59cb11773161ddfa9\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/99e8b7c7df2048f7b8933d5d4f275e49 : SUCCESS in 5m 57s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/aa56a84678494d0b9693b98c132d7e90 : SUCCESS in 6m 15s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/cc8ec4c8d99941fd8fb46d8b168dd216 : SUCCESS in 12m 37s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"e8c8ca0f65222ab4b78225700ff5e1e752f8b5d9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-06 17:15:32.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0e5c1716e6584002a8f7be67c34827c1\n\n- grenade https://zuul.opendev.org/t/openstack/build/c6f3ffbcbd414a21a26262a6e6a698e4 : SUCCESS in 57m 32s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/d65fc2e476ac42d8871c48b58e190b24 : SUCCESS in 55m 33s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/026f3ca2dfa74bd5b2603f44ee936fb2 : SUCCESS in 1h 42m 47s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/11266741568d4804a4cdb5076779a58b : SUCCESS in 1h 01m 33s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0829235df4d14511a704492fffe1723f : SUCCESS in 4m 40s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/4cb5ad7f133048fc9b7b0f3cf17d27a0 : SUCCESS in 4m 36s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/24697a731a264e6b9e6e05d1c53c5273 : SUCCESS in 4m 32s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1b1300bfd7b94128bc0347d4d84b122d : SUCCESS in 4m 49s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/f0be2b7c6d0c4a8fab3e9d2552968e3b : SUCCESS in 7m 40s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/445ea570a5b34da6a0b86a72f95d7fbf : SUCCESS in 6m 04s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/a4387a29239c4c28b9c9da205f745f2d : SUCCESS in 1h 37m 13s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/9148b844f21843be94f7472d0f5a7f62 : SUCCESS in 1h 38m 11s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/28591c8c50d8495e8a263ccbc4d69219 : SUCCESS in 2h 08m 20s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/7e600cbaeafe4d92b56bea1e96935ccb : SUCCESS in 1h 54m 26s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/685c21160d11485e95b833a8b343fe36 : SUCCESS in 1h 18m 05s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/846acc2feba048b2972463db1c5eb1c5 : SUCCESS in 1h 39m 19s\n- glance-grenade-centralized-cache https://zuul.opendev.org/t/openstack/build/049877d51ce3445f9c4c49718bc368bd : FAILURE in 10m 31s (non-voting)\n- glance-s3-multistore https://zuul.opendev.org/t/openstack/build/2d750950c1cd4f3e99d13b427ca63ec8 : FAILURE in 17m 09s (non-voting)\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/fe7cea7b9b4843719ac007994ebd4abb : SUCCESS in 30m 38s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/4a03027c7c664da093c6c02c5cb52add : SUCCESS in 35m 50s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"ce6599c609ed74446aeb65e25f3f75e196f2efe7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2026-01-07 10:23:05.000000000","message":"Patch Set 5: Patch Set 4 was rebased\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":5},{"id":"6d280cba8267c8ff74bf37b1717f9de6c87c9ea2","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-07 11:30:00.000000000","message":"Patch Set 5:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0aba488a5d2e4e0ebade2e5ad7ab0bf7\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/103323d8b3934d60a6be7d5e38631e92 : SUCCESS in 5m 23s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/3c6becf95ba0476e8d57da35a75610e0 : SUCCESS in 7m 19s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/0b3ac7e1dcf24f00b7e67d641c0471c1 : SUCCESS in 8m 56s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"370388a5636501cc75133f6fc22fe92f8ba7a8b4","author":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"date":"2026-01-07 12:31:31.000000000","message":"Patch Set 5: Code-Review+2","accounts_in_message":[],"_revision_number":5},{"id":"ba53633d9dd63cafaa6e57d34938d0c2e8b14bde","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-07 12:34:32.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/3563a58b019c41faa360651aa0725518\n\n- grenade https://zuul.opendev.org/t/openstack/build/62dceabbc6014b559123aa005cf26e10 : SUCCESS in 55m 59s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/15ad775e8a174f85bb47d7fd4d8c97c3 : SUCCESS in 28m 44s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/943ea9e7fea242a987b9ab037494be79 : SUCCESS in 1h 28m 49s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/d9ad898d589d4e85a628e04fffb0bfb6 : SUCCESS in 58m 33s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0a6e89bc2e444f93883e3b2c0c569c94 : SUCCESS in 6m 00s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/6dcdaa2fb49c46cd8d4ec88b58a531a8 : SUCCESS in 4m 40s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/6a1a3c06034f41a3aba5dc244331ed1f : SUCCESS in 8m 28s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f78ade55d99e4985be1ac5072e928c88 : SUCCESS in 5m 37s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/33fb0357a5d6484b98ca4311cc03bf4f : SUCCESS in 6m 51s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/da00ea466ea84e41b82d32b332e413bd : SUCCESS in 4m 16s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/736f2651a7b54376a0f58ab52a601220 : SUCCESS in 1h 48m 38s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/3f2daefb5482415ba39ba9cc7328ee21 : SUCCESS in 1h 48m 06s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/76af8362c79a49d595ec21cc0ac79a9f : SUCCESS in 1h 25m 09s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/5bc9533efce540a5b0109d85d5c070e1 : FAILURE in 1h 42m 03s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/68b9f18a5ce44e04ab144cebf7caae06 : SUCCESS in 59m 08s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/e1df450898884c36a7804c5616a01267 : SUCCESS in 1h 46m 18s\n- glance-grenade-centralized-cache https://zuul.opendev.org/t/openstack/build/36a4f248b54e437eadbfcffaaf2e4041 : FAILURE in 15m 11s (non-voting)\n- glance-s3-multistore https://zuul.opendev.org/t/openstack/build/3cfc7bb5de504826ac986d798a5850ac : FAILURE in 25m 01s (non-voting)\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/d14d74bf74a7461d834a663c3d38a313 : FAILURE in 35m 18s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/62900ef88f8c4e17871c883ee9648b15 : SUCCESS in 1h 12m 10s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"bc75d9d01d608fe9418d0693e71cf70180540116","author":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"date":"2026-01-12 15:01:32.000000000","message":"Patch Set 5:\n\nrecheck","accounts_in_message":[],"_revision_number":5},{"id":"4c6ab658f8e1dd882d16a0d56962772cbc23c329","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-12 15:21:04.000000000","message":"Patch Set 5:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/938b11167ed743389d5e3ceb0c51efc8\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/2abb2a0b5b6e4b3ab6eaa67f2ad6d1fe : SUCCESS in 9m 23s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/67bf7b5f96ca4b5ba613c488ab371528 : SUCCESS in 10m 02s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/30837f6a8836468187a33dd3378b8ac3 : SUCCESS in 11m 18s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"52a28f12d5060076e1cd50263796fba7fca8c45d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-12 17:37:35.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/bc955110d0ad47c9ae5b828bb4aea4a1\n\n- grenade https://zuul.opendev.org/t/openstack/build/67972e908dbd4f07b8ed8126b8256966 : SUCCESS in 58m 03s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/e37d6b6ad6f646a984444a53e3c889f7 : SUCCESS in 35m 26s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/e7838bdd1c46479e95c19439a14ea77b : SUCCESS in 1h 30m 33s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/c79399cd7e244928b53890e0b1562954 : SUCCESS in 1h 02m 19s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a6f1ee4543d44625a25781fa5573fb76 : SUCCESS in 6m 28s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/a052ed55a8bc4f299122f56dac66f990 : SUCCESS in 6m 21s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/b74728b4d4d3497cb5bfa7b0921ace55 : SUCCESS in 8m 10s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ca200f7e8bfd4f7096d6a4be3d1f33ff : SUCCESS in 6m 04s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/e9071e8c41b14871bb49d021fb4cec4e : SUCCESS in 6m 58s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/0a2c4c74f0c841b483d0717815851861 : SUCCESS in 6m 34s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/7cef97ddb45a4f87b72777b2e1a8bda8 : SUCCESS in 1h 54m 17s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/fd7708a965c4457382151b503d700939 : SUCCESS in 2h 04m 00s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/d0e4b6d138454f1799c1ecd4d4481c17 : SUCCESS in 2h 06m 42s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/adff2f4a0f664e48a93d25e1cf3e75b2 : SUCCESS in 2h 28m 07s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/4a4c92993f93453abb26d5e797054e63 : SUCCESS in 59m 43s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/2c51f04b8d4d48559613dcb31db329a5 : SUCCESS in 1h 38m 33s\n- glance-grenade-centralized-cache https://zuul.opendev.org/t/openstack/build/853f99ee09f949e7ac428386f2edb4d7 : FAILURE in 19m 17s (non-voting)\n- glance-s3-multistore https://zuul.opendev.org/t/openstack/build/8c7e9251ffd048e1b538c4b1bcae0b88 : FAILURE in 26m 03s (non-voting)\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/6665c79d693849a5868caf99baa3167a : SUCCESS in 36m 53s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/c2a2bab02910418586a09aef8ece7856 : SUCCESS in 1h 27m 25s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"2b09192f324e547d99a41ba0de7771fbdbb54b25","author":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"date":"2026-01-12 18:43:53.000000000","message":"Patch Set 5: Workflow+1","accounts_in_message":[],"_revision_number":5},{"id":"51f6480e5da302691a7d3d826e2a1fbd662cefdc","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-12 18:44:41.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"3536ddefceca716e7c3de8f572b6a0f5f617d08b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-12 20:56:27.000000000","message":"Patch Set 5: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/f5466d15f92044bc8d841a31278f1e58\n\n- grenade https://zuul.opendev.org/t/openstack/build/baedce607b1d48ba8fec89a4c6c15849 : SUCCESS in 1h 01m 54s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/9ae6ff7d03fa47c38bd2b7b31b0bfd47 : SUCCESS in 53m 12s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/0014b67e189e4f4ba605acca626e074e : SUCCESS in 1h 30m 26s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/220f59881ca94d5192e1d2f00f100c90 : SUCCESS in 1h 02m 56s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e2fa7f94060c45b1876544d39bc434bc : SUCCESS in 5m 03s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/a9e5a9ea87484161aea0f79ec7dcc4ff : SUCCESS in 3m 12s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/122c456287b249e6a919fc1bd733cf0b : SUCCESS in 7m 54s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ce0a82e292804ad291caad91c50ca55b : SUCCESS in 6m 26s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/ea88d7670b8c42a08e735ee99571c224 : SUCCESS in 5m 32s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/7fd46cadf65f4a2a9700001151b5a596 : SUCCESS in 7m 01s\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/a85b8c9db8e445468eb56c314f68d90b : SUCCESS in 2h 10m 24s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/95dea8e3673449d7bdc481f309392808 : SUCCESS in 1h 06m 13s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/641369fb1fc74d08adcbd02105f78bff : SUCCESS in 30m 25s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/f402370c92ff408498f2a842b83a984e : FAILURE in 1h 05m 46s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/d17365f0a3564351a73d9b33d5b828e3 : SUCCESS in 36m 25s","accounts_in_message":[],"_revision_number":5},{"id":"38784fff3b7a5d058a458391a48a27ce0a6a399f","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2026-01-13 02:13:58.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"8782f8c2c04ea706534203b8742a15a4ba970bce","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-13 03:19:59.000000000","message":"Patch Set 5:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8fc13090bb5c43d19c7a99ff93d61c83\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/2aefb4b1d140450e9530b386ad77bea8 : SUCCESS in 6m 32s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/d4f0e004868f434f9d1c12724c3aff0c : SUCCESS in 10m 47s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/ccd5ab47b4464b01a83b185c2a2d7626 : SUCCESS in 11m 31s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"7cd5b5710cd32923e3b8e02f6f97b986959234cb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-13 04:40:08.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8932751738cc4929a9edd5862ceb0c27\n\n- grenade https://zuul.opendev.org/t/openstack/build/b7165718afc74bc58ae5a2dc9b40487c : SUCCESS in 35m 11s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/7904486ce1d644d9beacd54b806a2968 : SUCCESS in 49m 29s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/bc1a97f8b7b04735859644e2fe5f7bc1 : SUCCESS in 1h 40m 24s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/ecfa8ce7207847b48a486e79afb1473a : SUCCESS in 1h 02m 12s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3f907ab41f1a4dafbdfde06ab0bbad05 : SUCCESS in 4m 23s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/d716da86b3e44a058c1a9e7d38b71bc2 : SUCCESS in 5m 01s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/efd0995891a143ae8b939725ce281168 : SUCCESS in 5m 37s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8d3dce52d1464c18851152e58bc2f033 : SUCCESS in 7m 32s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/820d177bddcf4bbd94b88fdbc6364835 : SUCCESS in 5m 59s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/f097f90cad124847ae540c7c0cc73d04 : SUCCESS in 7m 43s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/cfb0c8925c584ca4b20d100a65f0c381 : SUCCESS in 1h 04m 50s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/b53baa62e0b44896b2499b309f022042 : SUCCESS in 1h 10m 19s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/2956a58d9da848d584ac85a53d45e59f : SUCCESS in 2h 08m 21s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/868fad53a85347059aca6f289d2c9550 : SUCCESS in 1h 13m 26s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/651552d3bd664d459bbd81b76be18531 : SUCCESS in 58m 06s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/6b972a24c2b44a37a6257db9faee82d6 : SUCCESS in 1h 46m 59s\n- glance-grenade-centralized-cache https://zuul.opendev.org/t/openstack/build/dd25e1ce79404b29821a6a75a7d69100 : FAILURE in 23m 24s (non-voting)\n- glance-s3-multistore https://zuul.opendev.org/t/openstack/build/a2977151f3c14c149d88aaec3c5de27e : FAILURE in 18m 18s (non-voting)\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/83a0f5e9190c454a85547d080de671e9 : SUCCESS in 33m 48s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/de80dfd234f0440b96f1259b2378eb1a : SUCCESS in 36m 25s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"79f794d1334bd60c45e59cf1ff605f1233ac90ad","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-13 04:41:04.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"9d504b1e499ee91e2c5428bf52e05ff0bfe7f530","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-13 07:38:07.000000000","message":"Patch Set 5: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/38576e47835e4e5f93701099afa6e1cf\n\n- grenade https://zuul.opendev.org/t/openstack/build/15626193e73d4e2c897aa69145845655 : SUCCESS in 42m 31s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/38ed4570164d4281a73b252cb13a6492 : SUCCESS in 56m 39s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/20bd9f22671f47788f10081e2838b8f7 : SUCCESS in 1h 15m 27s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/046334de8f074100a90dd4c40872f636 : SUCCESS in 57m 26s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/64e262156fe545eeb43812aeec07bffb : SUCCESS in 2m 35s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/c6ccd05111f8478b855430fcc03a8136 : SUCCESS in 3m 09s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/594624635bf047a9b379946f6aa8f01a : SUCCESS in 7m 40s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e53808d557754c0aadc2038aded48328 : SUCCESS in 5m 27s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/530d5846e31643f490014eccae784157 : SUCCESS in 6m 04s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/de48ac2c856b4ff7bbab91a2dd92e06a : SUCCESS in 4m 13s\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/80b652fd97ff4c4782bfa4bd0af92115 : SUCCESS in 2h 03m 12s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/4983f9fc73624d4fb00ecc8bfa07827f : SUCCESS in 1h 58m 03s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/e0fb97f847c14e88a0e4ea1e7b55e7c0 : SUCCESS in 32m 01s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/830fdb82d05d41c8996692de013a108e : FAILURE in 1h 12m 27s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/b0decec8cabe49c198e7326815ee20a7 : SUCCESS in 23m 26s","accounts_in_message":[],"_revision_number":5},{"id":"cacc64d65aba54c4d24e26c2d8b2a728d89cca83","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2026-01-13 11:05:52.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"f99ee273f8f8e200fe1eab8289ca9768b13d21ed","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-13 11:17:57.000000000","message":"Patch Set 5:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/cf462c70535d4c8790c752fccd696676\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/73e45102637f4c46a37418072301e7a8 : SUCCESS in 6m 22s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/7b3e149787f345e095487f8dcc2efbc2 : SUCCESS in 6m 28s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/c3ec57dfa8cc4d1a8361975bf7d9efb6 : SUCCESS in 9m 31s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"5e22409cdde6749e42d8346114752f4a53371159","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-13 13:23:37.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/e1bf0f025da640ff82a6412b909764c7\n\n- grenade https://zuul.opendev.org/t/openstack/build/7fe97ece13414ff98ee062c7c4619156 : SUCCESS in 1h 02m 56s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/0345f3c9f8864437ae7831f1283168a2 : SUCCESS in 55m 48s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/eea6429ad9aa47729fb87efed4e1aedf : SUCCESS in 1h 26m 15s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/9a7c3c34817d4c0a8e9ae24e882915b3 : SUCCESS in 45m 13s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/359ca891783343fb8fe69c26c2d4b879 : SUCCESS in 4m 42s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/30197d3ed2284803a84d3860c7c48c31 : SUCCESS in 3m 15s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/766d8c1627084b3981d6b710355fb2f3 : SUCCESS in 7m 45s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ca5b324d18aa4c6981b8cc3198b3fb47 : SUCCESS in 7m 36s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/3c0d2a64edc843d59de951a5d4c3536f : SUCCESS in 6m 41s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/640c7368f6204ae0b706bad24d48d042 : SUCCESS in 6m 22s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/eb312d08a5b341018777c494b97fc620 : SUCCESS in 1h 56m 29s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/212713beae5c4c7c9511a2eaf978da36 : SUCCESS in 1h 47m 41s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/266dbdf381e74432b69f5695383e04c6 : SUCCESS in 2h 03m 01s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/3d7b3045592e4b10838df7c71da52395 : SUCCESS in 2h 10m 18s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/bb62786bf7b04d389ba252b949a94166 : SUCCESS in 1h 02m 21s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/98a0e438ebea40eba2b6c819ef700393 : FAILURE in 1h 01m 04s\n- glance-grenade-centralized-cache https://zuul.opendev.org/t/openstack/build/13896e4008874b36a0db5c67bc5f39cf : FAILURE in 11m 15s (non-voting)\n- glance-s3-multistore https://zuul.opendev.org/t/openstack/build/7434114a03bb4c89827585a1c5b1d19c : FAILURE in 13m 32s (non-voting)\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/54fe0e92c9f1461a8e4d084f89f4701b : SUCCESS in 18m 10s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/531397360c2d4060b89ed946f2847b0e : SUCCESS in 1h 12m 23s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"3ce27a0dee6f66c49015c19737251efb116d6cce","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2026-01-13 15:23:19.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"9442a334f9dfaff8c5fd6f7073f4899bd1b26126","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-13 15:36:13.000000000","message":"Patch Set 5:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/950f7f301c2e4127916f74bb79012581\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/350753e34f334cdf8c68e73fb0b671aa : SUCCESS in 9m 48s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/75cbba4ae24d4c79919d15ea3860592c : SUCCESS in 6m 48s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/a0915f0991f54dac973976975ef2974a : SUCCESS in 8m 01s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"677725bb03afb2dd382580cfcb83818f6d1ef245","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-13 17:41:36.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/35711f46448647cf8f1c8969a9e85c20\n\n- grenade https://zuul.opendev.org/t/openstack/build/afd506c8e0e74a8ea2e17ab94fd531db : SUCCESS in 36m 27s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/3423b2431bd549768560312060e5fb6b : SUCCESS in 1h 02m 14s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/bf8278529ebb4deb92863990fa8c6e5b : SUCCESS in 1h 40m 11s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/5f26f2516a074b7fa3601e1de146d245 : SUCCESS in 37m 36s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6afb22524ddc461fa3e2496061ece083 : SUCCESS in 4m 26s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/e78106937ca54497895f8329f969d74a : SUCCESS in 4m 42s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/0762219cccf749768f9ebeef962b15ac : SUCCESS in 7m 48s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ea468fab4a3d4d9c993932827ba4acec : SUCCESS in 5m 46s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/c0164fb9a4ac427bb43dd63558416978 : SUCCESS in 7m 14s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/45dd3c01cfbb46148999f07385d8330d : SUCCESS in 6m 55s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/4e0d8047a64347fb92eb7de516dda051 : SUCCESS in 1h 58m 00s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/3e463e4911964fe9883062e09120a345 : SUCCESS in 2h 07m 17s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/0a55eb8b51ef4e788fb88143a246b9b1 : SUCCESS in 1h 58m 37s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/0b497f4406c34fbb8d64d97d3aafb0f7 : SUCCESS in 2h 11m 44s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/da91b4ca3128469da971b83be7d14e18 : SUCCESS in 54m 12s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/9813074d50fb4940b42a179a91e4803f : SUCCESS in 1h 49m 52s\n- glance-grenade-centralized-cache https://zuul.opendev.org/t/openstack/build/599dec358674461ab47f6cab9799bdda : FAILURE in 22m 27s (non-voting)\n- glance-s3-multistore https://zuul.opendev.org/t/openstack/build/b7d263f22f2445b2887ddc9d570e6540 : FAILURE in 25m 27s (non-voting)\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/9641eafcc5e7428aa3de73af579de675 : SUCCESS in 16m 53s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/340e1777d6bd446e87002a18eb25e876 : SUCCESS in 1h 20m 47s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"162a2827c2cfc04cd707ba5f75d3c49e95db01cd","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-13 17:42:32.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"0e4f2599c7b48fe99a5561d370b33a4194b9d8fd","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-13 19:58:14.000000000","message":"Patch Set 5: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9e9a958106f849ac8849aeb37d73d9cf\n\n- grenade https://zuul.opendev.org/t/openstack/build/f61776c73bad4da3a728d5bf1f456508 : SUCCESS in 29m 53s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/d0cb12e3aac14bb4bc8e826fc8eb822f : SUCCESS in 51m 47s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/c778f2059030484a91cf449fbfffe0cb : SUCCESS in 1h 28m 11s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/b1daee9258e44c879d3cfdd40b5547ab : SUCCESS in 54m 52s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/04ec27a280574255aa765c2586bb643a : SUCCESS in 2m 44s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/0b97005f00b94f15b3258e9fe7d28d6b : SUCCESS in 4m 36s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/a5bfe41a2d0e4bbba7a3a468bebd7cf3 : SUCCESS in 7m 49s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fc79b38ce72f4ada9d173e65e7525973 : SUCCESS in 3m 37s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/93f7435fcd2f41fe955335fa20958ae4 : SUCCESS in 5m 00s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/42231e7c93f943199b3004aedcd1d5d3 : SUCCESS in 6m 01s\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/7e07c8ca811b4add9b2fe1b94af10e76 : SUCCESS in 1h 48m 55s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/152e43823bc04adf83f947ecbc7ac4fd : SUCCESS in 1h 09m 02s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/5fef12dfa7b64dc3aecef7a8b32a78d3 : SUCCESS in 58m 19s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/e4b0fb255a9f41d5b8fbc1c627d76fae : SUCCESS in 2h 11m 08s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/ff7a2e4d21f946908ea72fed0e21fe44 : SUCCESS in 18m 59s","accounts_in_message":[],"_revision_number":5},{"id":"b8d594c6c004eb6a8b9c61a12597e7c1779d72a9","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-13 19:58:14.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":5},{"id":"53c4bc87aac95415dfb2f1678acfa7ce96490412","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-13 20:01:18.000000000","message":"Patch Set 5:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/06505c49eb01457f90bb09338a4d291a\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/981f1b8693ad4caca3fbb8b1dbe31e2f : SUCCESS in 58s","accounts_in_message":[],"_revision_number":5}],"current_revision_number":5,"current_revision":"ebaa4fae40de73451485d793332561fec5e40c68","revisions":{"67ef234ce9887c5ff4293607c4f07f9af971d9c2":{"kind":"REWORK","_number":1,"created":"2025-12-07 15:06:51.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/59/970059/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/59/970059/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/59/970059/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/59/970059/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/59/970059/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/59/970059/1"}}},"commit":{"parents":[{"commit":"f21472b005f06b845c0f6623058f532efbe54288","subject":"Migrate py313 job to Debian Trixie","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/f21472b005f06b845c0f6623058f532efbe54288"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-12-07 15:06:35.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-12-07 15:06:35.000000000","tz":540},"subject":"Migrate upper functional job to Python 3.13","message":"Migrate upper functional job to Python 3.13\n\nDepends-on: https://review.opendev.org/970011\nChange-Id: I6e2e30e52f0dd8c72c90d15406d940bd766a3d03\nSigned-off-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/67ef234ce9887c5ff4293607c4f07f9af971d9c2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/67ef234ce9887c5ff4293607c4f07f9af971d9c2"}]},"branch":"refs/heads/master"},"7a6007376f151425b91781a8f41cbd5aaacb20d2":{"kind":"NO_CHANGE","_number":2,"created":"2025-12-07 15:09:48.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/59/970059/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/59/970059/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/59/970059/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/59/970059/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/59/970059/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/59/970059/2"}}},"commit":{"parents":[{"commit":"658b2e9dcedf40673efcf6137ade4de2ad92868d","subject":"Migrate py313 job to Debian Trixie","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/658b2e9dcedf40673efcf6137ade4de2ad92868d"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-12-07 15:06:35.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-12-07 15:09:48.000000000","tz":0},"subject":"Migrate upper functional job to Python 3.13","message":"Migrate upper functional job to Python 3.13\n\nDepends-on: https://review.opendev.org/970011\nChange-Id: I6e2e30e52f0dd8c72c90d15406d940bd766a3d03\nSigned-off-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/7a6007376f151425b91781a8f41cbd5aaacb20d2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/7a6007376f151425b91781a8f41cbd5aaacb20d2"}]},"branch":"refs/heads/master","description":"Rebase"},"97eae00d61ac49e981fdfcdd1728cd61bdae4829":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2025-12-07 16:48:00.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/59/970059/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/59/970059/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/59/970059/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/59/970059/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/59/970059/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/59/970059/3"}}},"commit":{"parents":[{"commit":"b3d5b636c30ee27200fb1570f7dd49421de342f3","subject":"Migrate py313 job to Debian Trixie","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/b3d5b636c30ee27200fb1570f7dd49421de342f3"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-12-07 15:06:35.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-12-07 16:48:00.000000000","tz":0},"subject":"Migrate upper functional job to Python 3.13","message":"Migrate upper functional job to Python 3.13\n\nDepends-on: https://review.opendev.org/970011\nChange-Id: I6e2e30e52f0dd8c72c90d15406d940bd766a3d03\nSigned-off-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/97eae00d61ac49e981fdfcdd1728cd61bdae4829"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/97eae00d61ac49e981fdfcdd1728cd61bdae4829"}]},"branch":"refs/heads/master","description":"Rebase"},"69cff4e700878c3ab88269ad2c08577a23689eec":{"kind":"NO_CODE_CHANGE","_number":4,"created":"2025-12-07 17:24:43.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/59/970059/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/59/970059/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/59/970059/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/59/970059/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/59/970059/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/59/970059/4"}}},"commit":{"parents":[{"commit":"b3d5b636c30ee27200fb1570f7dd49421de342f3","subject":"Migrate py313 job to Debian Trixie","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/b3d5b636c30ee27200fb1570f7dd49421de342f3"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-12-07 15:06:35.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-12-07 17:24:43.000000000","tz":0},"subject":"Migrate upper functional job to Python 3.13","message":"Migrate upper functional job to Python 3.13\n\nDepends-on: https://review.opendev.org/960639\nChange-Id: I6e2e30e52f0dd8c72c90d15406d940bd766a3d03\nSigned-off-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/69cff4e700878c3ab88269ad2c08577a23689eec"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/69cff4e700878c3ab88269ad2c08577a23689eec"}]},"branch":"refs/heads/master","description":"Edit commit message"},"ebaa4fae40de73451485d793332561fec5e40c68":{"kind":"NO_CHANGE","_number":5,"created":"2026-01-07 10:23:05.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/59/970059/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/59/970059/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/59/970059/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/59/970059/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/59/970059/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/59/970059/5"}}},"commit":{"parents":[{"commit":"3e148ba2f1f3a1d81f60423099a436b29575a15e","subject":"Migrate py313 job to Debian Trixie","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/3e148ba2f1f3a1d81f60423099a436b29575a15e"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-12-07 15:06:35.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-01-07 10:23:05.000000000","tz":0},"subject":"Migrate upper functional job to Python 3.13","message":"Migrate upper functional job to Python 3.13\n\nDepends-on: https://review.opendev.org/960639\nChange-Id: I6e2e30e52f0dd8c72c90d15406d940bd766a3d03\nSigned-off-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/ebaa4fae40de73451485d793332561fec5e40c68"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/ebaa4fae40de73451485d793332561fec5e40c68"}]},"branch":"refs/heads/master","description":"Rebase"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
