)]}'
{"id":"openstack%2Fglance~970060","triplet_id":"openstack%2Fglance~master~I10fdde5854f64a75bae8173677cdc40f75ece7c8","project":"openstack/glance","branch":"master","topic":"py313","attention_set":{},"removed_from_attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2026-01-15 01:28:45.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I10fdde5854f64a75bae8173677cdc40f75ece7c8","subject":"Remove py312 tips jobs","status":"MERGED","created":"2025-12-07 15:08:41.000000000","updated":"2026-01-15 01:30:28.000000000","submitted":"2026-01-15 01:28:45.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"970060-py313","meta_rev_id":"87972883fa0fcd85af4dce1ab865753abbc7a3fd","_number":970060,"virtual_id_number":970060,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-01-15 01:28:45.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"all":[{"value":2,"date":"2026-01-13 20:29:52.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"all":[{"value":1,"date":"2026-01-13 20:29:52.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Top Priority / Holds Gate"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-12-07 15:35:04.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2025-12-07 17:02:55.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-12-08 13:20:26.000000000","updated_by":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"reviewer":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"state":"REVIEWER"}],"messages":[{"id":"aadff8384cb4a98accb676520edb6946d6b4ccc9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2025-12-07 15:08:41.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"d1a55430019670dc029982175573f1f7eced2bfa","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2025-12-07 15:09:53.000000000","message":"Patch Set 2: Patch Set 1 was rebased","accounts_in_message":[],"_revision_number":2},{"id":"cc5ce5e2e85c3b14bdd97a911370914a2b58f211","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-07 15:35:04.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2ff41e3fef344f91b70e633ee84e315a\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/53d93fd5001746e7a1abf6a957bb6bf4 : SUCCESS in 7m 48s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/982daa4a761e4b498c2922ec3a4826e1 : SUCCESS in 9m 23s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/dda3b33ffcda4e17b32d114c567fa02a : SUCCESS in 13m 23s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"eabe61e3f264dd4152d6e2428ac08220bcedd469","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2025-12-07 16:48:05.000000000","message":"Patch Set 3: Patch Set 2 was rebased","accounts_in_message":[],"_revision_number":3},{"id":"af37feec9c58ad4462ddf46ff94b98c3d622fa5c","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-07 17:02:51.000000000","message":"Patch Set 3:\n\nThis change depends on a change that failed to merge.\n\nChange https://review.opendev.org/c/openstack/glance/+/970059 is needed.","accounts_in_message":[],"_revision_number":3},{"id":"8c4fe7a72fe4a08fcc984a929c72e2de15856ccc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-07 17:02:55.000000000","message":"Patch Set 3: Verified-1\n\nThis change depends on a change that failed to merge.\n\nChange https://review.opendev.org/c/openstack/glance/+/970059 is needed.","accounts_in_message":[],"_revision_number":3},{"id":"8bb20b3530f7e7cf84fcfb0a0de99f83d0813796","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2025-12-07 17:24:50.000000000","message":"Patch Set 4: Patch Set 3 was rebased\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":4},{"id":"6d0ae5d84b6b1ade5a5942a736bbcb89f6a9f5dd","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-07 17:43:44.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/de35226e73d04d229d1449a33cf6ee3f\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/05a4e6afe57e4d038273aff89274f432 : SUCCESS in 8m 26s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/b5a52ac7ef9f471d92d64d49f3090676 : SUCCESS in 9m 12s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/f68de63e844d4daf9b643b725a0c0b7b : SUCCESS in 12m 49s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"5b19ef7fd4e258480e49b0c319dc04cd93455334","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-07 19:29:04.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/bdebcbaf213c4785a4936d8cbec0af60\n\n- grenade https://zuul.opendev.org/t/openstack/build/ecaef4bb56fd4788b8344e5fedf12c46 : SUCCESS in 34m 06s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/b192de95acf0498ba7ea3261d24aec23 : SUCCESS in 39m 56s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/fd2befdbfa7a4a59afc9402ba79b5608 : SUCCESS in 1h 22m 59s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/b47b6b7b06ff4c228d0d426cb2acd3be : SUCCESS in 44m 25s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2b23366603224066ab9ab1285c5dba07 : SUCCESS in 4m 44s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/8decaeeb7edb42cd9aa14af2a4c7adc6 : SUCCESS in 6m 06s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/e8fdc3a086bb42d09be373564cad8330 : SUCCESS in 7m 40s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a9ca3a2a14c847599dadf804023b95bf : SUCCESS in 5m 20s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/c80ffb986bf04777959ea3ec2fc9134f : SUCCESS in 7m 49s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/be4cd96404924751a84361f047df9b36 : SUCCESS in 6m 47s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/84b3fe06ce854b0c96f82b2a4fc67e24 : SUCCESS in 1h 54m 31s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/02f651ecbceb459396ea1fa230509f0d : SUCCESS in 1h 09m 55s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/fa07f0ccffcf49148351e4d49da75465 : SUCCESS in 1h 45m 54s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/4b89bd8cb4d348559e5a4440b2bda421 : SUCCESS in 1h 20m 35s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/970bc2fbdb2b4afb910806ba8eb057f3 : SUCCESS in 1h 04m 15s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/f8188de2f8ea4e7597678c71a064ee71 : SUCCESS in 1h 50m 07s\n- glance-grenade-centralized-cache https://zuul.opendev.org/t/openstack/build/cedfafebd92742bf8cb3d81fd708f26c : FAILURE in 21m 26s (non-voting)\n- glance-s3-multistore https://zuul.opendev.org/t/openstack/build/2aaa62e91d264d7e80ffa0ae77455c12 : FAILURE in 26m 02s (non-voting)\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/c0b4c7a1c3d7421da63aa842fc3a618c : SUCCESS in 30m 07s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/220983d9d9c443a3a5f0fedfde4e81cc : SUCCESS in 57m 21s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"a7d1563a7b1c4e4dedb9ebe11864092dfe812f3a","author":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"date":"2025-12-08 13:20:26.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"09735becb0f735d42ead856f41b7e601bcf97ef3","author":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"date":"2025-12-08 13:27:51.000000000","message":"Patch Set 4: -Code-Review -Workflow","accounts_in_message":[],"_revision_number":4},{"id":"a1f1b003c3bb1df499e9f6ed1a468cfd2abc0eed","author":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"date":"2026-01-06 14:59:50.000000000","message":"Patch Set 4: Code-Review+2","accounts_in_message":[],"_revision_number":4},{"id":"2027fcc37b4f87381aa0b9e14a6f31f266c4719e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2026-01-07 10:23:10.000000000","message":"Patch Set 5: Patch Set 4 was rebased\n\nCopied Votes:\n* Code-Review+2 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":5},{"id":"fd1e9396a9a53bfe78aa7e13b0d08a384b3c6dcf","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-07 11:16:15.000000000","message":"Patch Set 5:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/096727afcd8942d788a380ea5d3d7df2\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/05dafa4610ac4617ace6a12e729b7282 : SUCCESS in 8m 19s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/16afd1665f6a428b97aa1f73266ef104 : SUCCESS in 6m 03s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/f9bb2dfda2c54d059d4861d08baa0d2c : SUCCESS in 9m 01s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"8079e14bb5c8149eda88d818fabbae0488030bfb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-07 12:35:37.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c7fa423d75d346b3a4b884d82ef737a7\n\n- grenade https://zuul.opendev.org/t/openstack/build/82e9b98c91b343b88be2382c4dc1bcaf : SUCCESS in 56m 29s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/82dfcb073d924856a6a3a74df377a985 : SUCCESS in 59m 06s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/08035b78e48845579e3e5d3cecc45f58 : SUCCESS in 1h 36m 21s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/ccaee00c3a1942f2b2370bf627023631 : SUCCESS in 59m 46s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0a87d4b2f5b042778f2c4c5bb2cbc9c0 : SUCCESS in 4m 41s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/d2906f4c892444b19bbfdedbe181f77b : SUCCESS in 4m 41s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/3b93dd4363234238ab2fe7697ad41461 : SUCCESS in 8m 10s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2ce42eb750654afdacf1b7c7b8e31aba : SUCCESS in 3m 00s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/f3b8ad93f4554ca1a4e65b9e7b10f76c : SUCCESS in 6m 38s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/9cd931ea52c34f168631fd3b3ff40382 : SUCCESS in 6m 59s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/0682e619d7114bc389fc8fd8571a9439 : SUCCESS in 1h 46m 49s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/3347705b055547fbbb09f121246a4916 : SUCCESS in 1h 59m 39s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/f8a8c13ea4fc4abaae86bf5d37674a43 : SUCCESS in 59m 02s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/fc6aaca6be5e4e4e93d2ccb0f97f9091 : SUCCESS in 2h 02m 06s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/038d7c1dac7f4572af5ec9041a9ffc22 : SUCCESS in 1h 08m 57s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/8ab11f445fd44a45b6f49998a487f1de : SUCCESS in 1h 00m 39s\n- glance-grenade-centralized-cache https://zuul.opendev.org/t/openstack/build/8ce9b3addb7e4f3fa5226726019f2847 : FAILURE in 10m 26s (non-voting)\n- glance-s3-multistore https://zuul.opendev.org/t/openstack/build/f1234d6dd0e243319bbb213e90a6e805 : FAILURE in 13m 32s (non-voting)\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/6085589d3b7e48299c126df4e7c31c0f : SUCCESS in 26m 13s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/68de935898504c448f255393f9277eb1 : SUCCESS in 52m 08s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"2dbeca0231ddff0074f1a4f70162aa317e796bc1","author":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"date":"2026-01-13 20:29:52.000000000","message":"Patch Set 5: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":5},{"id":"e84c6ed6e06728d4fed0584cb8462061ff3982ba","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-13 20:30:37.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"6e16b00ed52e418803ae9eb5ca0fe5c5ca130e88","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-13 20:50:32.000000000","message":"Patch Set 5: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/0ceee362d17c43d29a1990f41bc0733d\n\n- grenade https://zuul.opendev.org/t/openstack/build/30c586ad9db44842b22337feb35411d3 : RETRY_LIMIT in 2m 10s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/df7e61e5ae33424eb5ed3ee2e66eb8a3 : RETRY_LIMIT in 1m 25s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/0693feb91b494dfea1f1839c2dd4efb4 : RETRY_LIMIT in 1m 37s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/bd23dd25c7374e8f90afb7322d1fd8d8 : RETRY_LIMIT in 1m 30s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a72ab7f05619461eb7c92221bed25931 : RETRY_LIMIT in 1m 04s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/2ae244de3d2c421f87fc1b935239bbe8 : RETRY_LIMIT in 58s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/d4042a23a5cd488f8b0c2c3e0fc9dee0 : RETRY_LIMIT in 1m 06s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/576f841267e84c9cada8ecad13bcc29c : RETRY_LIMIT in 56s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/6580fe30de7b4b2985e8f565bbd447ba : RETRY_LIMIT in 1m 07s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/356a75343ffd44388ecaee297dcd6e66 : RETRY_LIMIT in 1m 06s\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/0eb6f8daa5da4324a22a1bb4c2536825 : RETRY_LIMIT in 1m 31s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/39541feca1fa4b3bb4d0261fc95a9e85 : RETRY_LIMIT in 1m 37s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/6b9dd0f00b8440aaad155621e169912d : RETRY_LIMIT in 1m 32s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/dca996e417e64d0ea064557fbb608cd9 : RETRY_LIMIT in 1m 33s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/64b4427ee5e34a9fb7cae905d91be66f : RETRY_LIMIT in 1m 28s","accounts_in_message":[],"_revision_number":5},{"id":"3ede1afd0917f772b5551c6d6a5fb6b173918408","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2026-01-13 23:49:08.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"916f55ec948df6e4ca91fa364f63d3d42d14f058","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-14 00:00:32.000000000","message":"Patch Set 5:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7852c19d4846491b9e9299c01520f914\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/40bc73dd56f54fc7a28681974c5a97ac : SUCCESS in 5m 34s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/65db686b468e49ecb0ddf5b148a884c2 : SUCCESS in 7m 12s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/55d4ea812bb148b88a492418a9064834 : SUCCESS in 8m 27s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"db1284dabfefc4de17a00c3bad9a4059f8061927","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-14 01:35:22.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/1064b5e368bb477c87778b74d9434fa7\n\n- grenade https://zuul.opendev.org/t/openstack/build/73d1ca7927c941f7a593c0c88f7ac045 : SUCCESS in 59m 59s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/37e4f42cb7654cae94e90f3e6912b5dd : SUCCESS in 58m 45s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/089bbb9d10b347fbb02432d3c3c9bff7 : SUCCESS in 1h 26m 16s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/55da6af5571842db801b9e205fddbb1d : SUCCESS in 52m 34s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1b529b4e469744669a2377a682b8edf6 : SUCCESS in 4m 40s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/a6389014bcc54ce0988781c328f1edfd : SUCCESS in 4m 27s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/73e73c094d344ef1a9284f2c0043fa5e : SUCCESS in 9m 07s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/31886ebd8bff47249f02d1a7509c2db0 : SUCCESS in 5m 08s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/89eadfe5d0274d268cbe855726b5d7e0 : SUCCESS in 6m 16s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/16e32966ceef4709b2c925fdf1b2ff39 : SUCCESS in 5m 43s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/764a21f89e1b43e4922c24335fbc55b0 : FAILURE in 1h 15m 49s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/9cea9adb412442a5b601b109733d50ad : SUCCESS in 1h 01m 38s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/c5bf07cb7cbd45ea9b6499eb5a65ea9e : SUCCESS in 1h 04m 26s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/8cd13e47c756405a98455f6335fecf5d : SUCCESS in 1h 43m 53s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/d86fd0827d434192a69a411f5b349da4 : SUCCESS in 48m 11s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/7a207be1753b47d5911fbb0643dd909d : FAILURE in 19m 51s\n- glance-grenade-centralized-cache https://zuul.opendev.org/t/openstack/build/54718c1c5b6d48e1942ca56ce455d825 : FAILURE in 10m 42s (non-voting)\n- glance-s3-multistore https://zuul.opendev.org/t/openstack/build/da80659ba591410a9a41cef100e9cdf5 : FAILURE in 26m 24s (non-voting)\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/4f76285eaf934d608a0741c47e032ef9 : SUCCESS in 16m 02s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/dc123ad91d6444b4b1b14601ac76b1bd : SUCCESS in 1h 16m 38s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"51696ab139a37ba4424bbeea621cd06a4b1fb803","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2026-01-14 14:48:12.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"c2dfa1dba3659aac98ed737254aecd72b58c2f01","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-14 15:36:59.000000000","message":"Patch Set 5:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/191efc1e0fb94762ad92b06b40cfb1e1\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/1634cfd2421c4168ada8047af0eb18c1 : SUCCESS in 7m 12s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/d52049260ff8465b8a41d27fdef45ecf : SUCCESS in 8m 16s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/47a51d63f8e24bed8e9010b5452d7f67 : SUCCESS in 12m 55s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"e5a52cf9f1ad58f1e9364738eb8ecbe27f45266b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-14 17:29:29.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e84db9e3ec3f4bd89cde08cbe26e016b\n\n- grenade https://zuul.opendev.org/t/openstack/build/ffaede4a2bfb447f9adadf4aee5bca5b : SUCCESS in 39m 33s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/aa21dccb1498465aa99c09929a4fcdb7 : SUCCESS in 1h 04m 22s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/191cd98139ac49d385efe2bbbda9a00d : SUCCESS in 1h 17m 04s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/f386009739654225b3de549b79496695 : SUCCESS in 55m 30s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d7b9af7544bc4877a82d679049cbffa6 : SUCCESS in 2m 41s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/3fe52cbbe1ab4d9c8dd6dc496472a80e : SUCCESS in 4m 41s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/1509a33d568f4e7ab69be8cc80899246 : SUCCESS in 7m 31s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5c845a5cc3ad4f4a81b6826d8b4e469c : SUCCESS in 5m 07s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/500e44c55f0040498bec25a50a5864d8 : SUCCESS in 7m 47s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/d59a76a24e4942aa87b3a6b591dd5406 : SUCCESS in 5m 05s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/7911288eeaf54e20ab6c7d2b70f72ade : SUCCESS in 1h 52m 03s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/439b5e093e1943f5babf2c9861d08837 : SUCCESS in 2h 06m 58s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/e339ff42d80247d69c3ab356997f0cc0 : SUCCESS in 1h 02m 33s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/d1355d803d61462daacf02a56b965e10 : SUCCESS in 2h 34m 26s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/f5ed2e5e2f0e4fc1be49da3644d310e9 : SUCCESS in 1h 00m 03s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/c33e2b83d94744ce9f8a3cae5521a577 : SUCCESS in 1h 13m 49s\n- glance-grenade-centralized-cache https://zuul.opendev.org/t/openstack/build/ead4e0a964d74a3383ed24c7c6b06afc : FAILURE in 15m 58s (non-voting)\n- glance-s3-multistore https://zuul.opendev.org/t/openstack/build/0f767fea281c46b7ae83c3a66a85c028 : FAILURE in 25m 32s (non-voting)\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/07adf2b3fec94a56803d063b92d2bfe2 : SUCCESS in 34m 09s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/28616de2c4ba4755bcb965d1d9be518b : SUCCESS in 1h 02m 06s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"177efb1162503c155a630a84fc9949fbcf7d2796","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-14 17:31:49.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"cc26bc0e8a8149dfef08f8aeee717ec9b62ffe92","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-14 20:20:48.000000000","message":"Patch Set 5: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/41b05c6ba01941c085795d7fb7ada407\n\n- grenade https://zuul.opendev.org/t/openstack/build/8c540fc0b7ba46a496d977ae33e4db46 : SUCCESS in 40m 58s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/af457b2d6a10426fb242381fabdb6946 : SUCCESS in 55m 52s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/dee09ed710ed4cd1a38e397f73d83256 : SUCCESS in 1h 26m 05s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/53b038ee7ef14595a30eb5078aece867 : SUCCESS in 47m 13s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/836e7f5249ab4983ac5628383445f243 : SUCCESS in 4m 25s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/954bca7801b94094beb802004c9d2716 : SUCCESS in 2m 42s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/eee1133da52948c798a7e1903f1ea7a6 : SUCCESS in 8m 37s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4f3b3d7773a94c6ba720195cdb5326fa : SUCCESS in 5m 56s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/8c90c50379a34299a36a3d87202fcce7 : SUCCESS in 6m 46s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/b003562e985e407088be901d3d986e10 : SUCCESS in 4m 22s\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/e1ab6ac390ab4727b9d6ee726c0e0b23 : SUCCESS in 1h 27m 50s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/74bdaa292b644d0ea491f5ec91b9f48c : SUCCESS in 1h 31m 23s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/e22254533d984504a20735b4b33372b9 : SUCCESS in 56m 42s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/a6390749d60f40589bc1aafc154e0c40 : SUCCESS in 1h 51m 26s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/a6c18dfcd62e4d83be47e1db362bf339 : FAILURE in 33m 41s","accounts_in_message":[],"_revision_number":5},{"id":"9bac466b281d7cf59077d4b61814f448cc1958a7","author":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"date":"2026-01-14 20:38:34.000000000","message":"Patch Set 5:\n\nrecheck","accounts_in_message":[],"_revision_number":5},{"id":"6524d4423d33c71005fbac0183b4d9b0ec8e881a","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-14 20:49:29.000000000","message":"Patch Set 5:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/fd09bdf18d5f47eba9b2ae3689f993dc\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/1af0d2e25aaa4762a918c5a8f91e5c40 : SUCCESS in 6m 42s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/2056ed016a8d4a62b88513e992ab2b2a : SUCCESS in 6m 54s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/862f19f78ef34c66be47cf3ebf4bfb0c : SUCCESS in 8m 28s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"1b1d4077e78cfbf956343e38a64e68a672d34ef4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-14 23:05:20.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/466bf31c93da462abbd3885b93514587\n\n- grenade https://zuul.opendev.org/t/openstack/build/30e8375addcb44cf80c3746ea66fd802 : SUCCESS in 1h 00m 18s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/2ed727cb56614f4d91901da339824016 : SUCCESS in 44m 27s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/46c44ed466ad4053bad1bdcb3d97aa82 : SUCCESS in 1h 24m 17s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/04d04e21806d4009ae5a76478727f49d : SUCCESS in 49m 16s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cd89b008b0e641c49b860908fc149a7e : SUCCESS in 4m 42s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/27869bf1d78446a286ae8e155e7a440a : SUCCESS in 4m 40s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/73b668e110ae4d7bb3d62332f376090b : SUCCESS in 4m 11s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/33ea4c0b59574830a6cbabb934e769fc : SUCCESS in 5m 22s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/8d55e2035c3f4de48444b3f2621daaf9 : SUCCESS in 6m 11s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/51852f1ab3a649729aaddbfb3ebe94e9 : SUCCESS in 6m 58s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/770b6d2bbc2443b0a2e9ba837562644e : SUCCESS in 1h 26m 53s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/9298fde16aff4474bd2813cdcdea568e : SUCCESS in 1h 52m 41s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/cb4c71a3cad8436c8557129da6a5a2da : SUCCESS in 1h 39m 06s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/e2c1b3d358f1446a898c6becd9b54e01 : SUCCESS in 2h 20m 20s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/40be1e30bf624832b15e75dbb2e1ef28 : SUCCESS in 55m 16s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/1351d3c80bf34e709ea6c46947e8b4f8 : SUCCESS in 2h 07m 02s\n- glance-grenade-centralized-cache https://zuul.opendev.org/t/openstack/build/d896f78e2c9c437c8574e3f7eeb408a4 : FAILURE in 21m 33s (non-voting)\n- glance-s3-multistore https://zuul.opendev.org/t/openstack/build/314b296bfc6c4b6c8f011427722c7aa5 : FAILURE in 16m 38s (non-voting)\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/233619a47d4c4f8aaac121fdce71b7c9 : SUCCESS in 23m 38s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/32bf78d15d2643d59bdfad1823a352df : SUCCESS in 49m 30s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"6143330cb551ff6798e2c474bf2fe42dd2cb0ff7","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-14 23:06:02.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"03ae620219953f25d4ab51780d58a74808ed321d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-15 01:28:45.000000000","message":"Patch Set 5: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8b8204de17f14294be66563da53297da\n\n- grenade https://zuul.opendev.org/t/openstack/build/1d2267324f0e48c6b5a275ad347d9ae2 : SUCCESS in 1h 00m 19s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/6218552351b54ad7820625aea03372bd : SUCCESS in 1h 03m 13s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/96fe2003f39242e9a568ee1e0fa1f97b : SUCCESS in 1h 41m 38s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/745873c885ab4ad3a4a542c41372f42c : SUCCESS in 58m 59s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3b612a261a3d4bb7b83437c95826f10e : SUCCESS in 5m 05s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/3085920911844dc8b32f8a4cedb2ae9b : SUCCESS in 5m 02s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/be1e728c1b034696b1ff35073578daaa : SUCCESS in 9m 01s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8ff84c0efa4d4828a74e08aa45f233e0 : SUCCESS in 5m 47s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/cb22a836839744629a3d0d6e6b214c30 : SUCCESS in 6m 26s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/6e6654beab67494a839882a35c1bd1fa : SUCCESS in 6m 41s\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/f65ca017ccf24d56a3a45b58b3cdabd8 : SUCCESS in 2h 10m 38s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/0b908aeb64fa4dc0832dd09d14938ab6 : SUCCESS in 2h 10m 43s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/5fbf5237559d4905bd452e8ec0ee2a60 : SUCCESS in 59m 54s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/e2572c109dd0412cbdf767926fb09e5b : SUCCESS in 1h 50m 18s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/d3c543624a624fc78a9949b45f3eccf0 : SUCCESS in 28m 49s","accounts_in_message":[],"_revision_number":5},{"id":"4f03128bf8e866e4eb83ee99dd405a1d1273042d","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-15 01:28:45.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":5},{"id":"87972883fa0fcd85af4dce1ab865753abbc7a3fd","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-15 01:30:28.000000000","message":"Patch Set 5:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/869b3e8d435d4a4d8c08aed15af22fc6\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bc481c5dc60b4492a530769e13876f37 : SUCCESS in 43s","accounts_in_message":[],"_revision_number":5}],"current_revision_number":5,"current_revision":"f77264e9e1e6b345de22fce137816a3794770560","revisions":{"19fb681876fcf8428516763b583b1d8d6508c1bd":{"kind":"REWORK","_number":1,"created":"2025-12-07 15:08:41.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/60/970060/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/60/970060/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/60/970060/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/60/970060/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/60/970060/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/60/970060/1"}}},"commit":{"parents":[{"commit":"67ef234ce9887c5ff4293607c4f07f9af971d9c2","subject":"Migrate upper functional job to Python 3.13","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/67ef234ce9887c5ff4293607c4f07f9af971d9c2"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-12-07 15:07:47.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-12-07 15:08:25.000000000","tz":540},"subject":"Remove py312 tips jobs","message":"Remove py312 tips jobs\n\nNow Python 3.13 is officially supported and all tests are voting.\n\nChange-Id: I10fdde5854f64a75bae8173677cdc40f75ece7c8\nSigned-off-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/19fb681876fcf8428516763b583b1d8d6508c1bd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/19fb681876fcf8428516763b583b1d8d6508c1bd"}]},"branch":"refs/heads/master"},"ab8f6b9367d68d60902b07faa2cd70c97ed4389b":{"kind":"NO_CHANGE","_number":2,"created":"2025-12-07 15:09:53.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/60/970060/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/60/970060/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/60/970060/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/60/970060/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/60/970060/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/60/970060/2"}}},"commit":{"parents":[{"commit":"7a6007376f151425b91781a8f41cbd5aaacb20d2","subject":"Migrate upper functional job to Python 3.13","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/7a6007376f151425b91781a8f41cbd5aaacb20d2"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-12-07 15:07:47.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-12-07 15:09:53.000000000","tz":0},"subject":"Remove py312 tips jobs","message":"Remove py312 tips jobs\n\nNow Python 3.13 is officially supported and all tests are voting.\n\nChange-Id: I10fdde5854f64a75bae8173677cdc40f75ece7c8\nSigned-off-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/ab8f6b9367d68d60902b07faa2cd70c97ed4389b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/ab8f6b9367d68d60902b07faa2cd70c97ed4389b"}]},"branch":"refs/heads/master","description":"Rebase"},"19b724d3271ca5023c65cd96e2cfba859ee74490":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2025-12-07 16:48:05.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/60/970060/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/60/970060/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/60/970060/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/60/970060/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/60/970060/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/60/970060/3"}}},"commit":{"parents":[{"commit":"97eae00d61ac49e981fdfcdd1728cd61bdae4829","subject":"Migrate upper functional job to Python 3.13","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/97eae00d61ac49e981fdfcdd1728cd61bdae4829"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-12-07 15:07:47.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-12-07 16:48:05.000000000","tz":0},"subject":"Remove py312 tips jobs","message":"Remove py312 tips jobs\n\nNow Python 3.13 is officially supported and all tests are voting.\n\nChange-Id: I10fdde5854f64a75bae8173677cdc40f75ece7c8\nSigned-off-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/19b724d3271ca5023c65cd96e2cfba859ee74490"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/19b724d3271ca5023c65cd96e2cfba859ee74490"}]},"branch":"refs/heads/master","description":"Rebase"},"552b3b5898fb13f6aee801dae5ae2a17054ced31":{"kind":"NO_CHANGE","_number":4,"created":"2025-12-07 17:24:50.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/60/970060/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/60/970060/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/60/970060/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/60/970060/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/60/970060/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/60/970060/4"}}},"commit":{"parents":[{"commit":"69cff4e700878c3ab88269ad2c08577a23689eec","subject":"Migrate upper functional job to Python 3.13","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/69cff4e700878c3ab88269ad2c08577a23689eec"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-12-07 15:07:47.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-12-07 17:24:50.000000000","tz":0},"subject":"Remove py312 tips jobs","message":"Remove py312 tips jobs\n\nNow Python 3.13 is officially supported and all tests are voting.\n\nChange-Id: I10fdde5854f64a75bae8173677cdc40f75ece7c8\nSigned-off-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/552b3b5898fb13f6aee801dae5ae2a17054ced31"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/552b3b5898fb13f6aee801dae5ae2a17054ced31"}]},"branch":"refs/heads/master","description":"Rebase"},"f77264e9e1e6b345de22fce137816a3794770560":{"kind":"NO_CHANGE","_number":5,"created":"2026-01-07 10:23:10.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/60/970060/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/60/970060/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/60/970060/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/60/970060/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/60/970060/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/60/970060/5"}}},"commit":{"parents":[{"commit":"ebaa4fae40de73451485d793332561fec5e40c68","subject":"Migrate upper functional job to Python 3.13","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/ebaa4fae40de73451485d793332561fec5e40c68"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-12-07 15:07:47.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-01-07 10:23:10.000000000","tz":0},"subject":"Remove py312 tips jobs","message":"Remove py312 tips jobs\n\nNow Python 3.13 is officially supported and all tests are voting.\n\nChange-Id: I10fdde5854f64a75bae8173677cdc40f75ece7c8\nSigned-off-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/f77264e9e1e6b345de22fce137816a3794770560"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/f77264e9e1e6b345de22fce137816a3794770560"}]},"branch":"refs/heads/master","description":"Rebase"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
