)]}'
{"id":"openstack%2Fglance~971516","triplet_id":"openstack%2Fglance~master~Ia7552d2e3ef84a8e258bf918b632ae8d309ccd0a","project":"openstack/glance","branch":"master","attention_set":{},"removed_from_attention_set":{"9303":{"account":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"last_update":"2026-01-05 16:50:43.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ia7552d2e3ef84a8e258bf918b632ae8d309ccd0a","subject":"Handle glance_store.Invalid exception in upload method","status":"MERGED","created":"2025-12-19 09:24:08.000000000","updated":"2026-01-05 16:51:56.000000000","submitted":"2026-01-05 16:50:43.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"971516","meta_rev_id":"e925bf3a94d3b279b6b94d5896f6218560491125","_number":971516,"virtual_id_number":971516,"owner":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-01-05 16:50:42.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"all":[{"value":2,"date":"2026-01-05 14:16:36.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"all":[{"value":1,"date":"2026-01-05 14:16:36.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Top Priority / Holds Gate"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-12-19 09:40:30.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2025-12-19 11:35:40.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-01-05 14:16:36.000000000","updated_by":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"reviewer":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"state":"REVIEWER"}],"messages":[{"id":"18d8f0180c8ffef2740601385f764d8bdf346d31","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2025-12-19 09:24:08.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"c7f26e281aed0ce02008e7745e73c69347864959","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-19 09:40:30.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/fd0da491b1b94995a8476be37678c993\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/139f06944dd644f686ee04f3dc763f9f : SUCCESS in 6m 07s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/6d1b8883ce2b496d8629ba27fb9eff6c : SUCCESS in 7m 00s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/73d6e97b1aa0474599a9ce3c5f2f1b5e : SUCCESS in 13m 44s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"8362138f21980a0b4e97112d1ca8f28d1c151afe","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-19 11:35:40.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/a5ecd8885e1944a7bdae472d05fb9af3\n\n- grenade https://zuul.opendev.org/t/openstack/build/7d5e670794f24bca94014031c6d995bf : SUCCESS in 43m 24s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/358cd165f61f456ba915567e8b7f66d5 : SUCCESS in 30m 02s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/77d39c6cb8e4475bbe0ac567eeacd169 : SUCCESS in 1h 35m 19s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/deb09e26d6be4bc0ab674790b566b38d : SUCCESS in 1h 06m 47s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4fae68a1b3764ced8feab4f68ad5acb0 : SUCCESS in 2m 34s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/ca6a339e5bbd49eab9dc9b80098cccea : SUCCESS in 4m 42s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/df058eecea304afe85dfa9bdfd4d7abe : SUCCESS in 4m 13s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2205f05eb09a484db794b4707aa1923f : SUCCESS in 5m 19s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/6974289dc5e341ca8fdb8e7b7111e641 : SUCCESS in 7m 24s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/8fc5cc4dd30a4e09aa62f3bf6adc59c3 : SUCCESS in 6m 02s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/c591a341a96a43689f76a398d0c1148e : FAILURE in 1h 02m 52s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/fbacd5cf21264e60be118839176a1f47 : SUCCESS in 2h 06m 20s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/5055b6ec5be0447186320ab8cec11aab : SUCCESS in 1h 51m 09s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/d9917172e20b4106b0cc2a5386c7842a : FAILURE in 1h 57m 41s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/3f2d0a1b7e914e55ab8407d4521ed9ad : SUCCESS in 57m 06s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/844399004798483f87f1bc9304bc7910 : SUCCESS in 1h 53m 34s\n- glance-grenade-centralized-cache https://zuul.opendev.org/t/openstack/build/a0ffa0a5094f41dfa50875bae9c2252c : FAILURE in 19m 38s (non-voting)\n- glance-s3-multistore https://zuul.opendev.org/t/openstack/build/ad1a807834664a71b379ed12f8a9dc39 : FAILURE in 23m 18s (non-voting)\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/c59b4c1247dc4765bfd27d4cdef99a49 : SUCCESS in 31m 27s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/5a8f5127722d4f948a61c8ca23e484ea : SUCCESS in 1h 09m 40s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"d2ee7275cd5795feda12f8366307fc6afc00fbea","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2025-12-19 13:04:57.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"4e4f4e4af6f11f8fff215d57b2e1f2b32da05089","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-19 13:29:45.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7a2687ea34f84cd79aa785e8a0e69e30\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/fefac17e04204aa6b2d8ed7c9a880bfa : SUCCESS in 8m 41s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/41d570467e304fc1bdaaaea8f6d61593 : SUCCESS in 9m 51s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/326f8d5f541c4dfaa92e1173cdc4b1b2 : SUCCESS in 8m 01s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"931163a7ac601e78e701f56adbdd9278c303b455","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-19 14:56:23.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e4927cc3377d43e3818d15e0d04f2d60\n\n- grenade https://zuul.opendev.org/t/openstack/build/1fe806e1fd0445bb914061137768d697 : SUCCESS in 47m 28s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/7aeaa1eda423419db72e2c71f68c4087 : SUCCESS in 1h 00m 23s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/d73c812307b4454f9a48e60d6621b30f : SUCCESS in 1h 22m 31s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/33ac7b6bc03440179e72c8a1fda89845 : SUCCESS in 44m 54s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4655382ab87b4731bbb032e956791679 : SUCCESS in 2m 58s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/4298c2f1e8d441ffba120e39a543be1d : SUCCESS in 2m 41s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/7d881a30409040eca028831de803a6aa : SUCCESS in 7m 53s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ebce88fd6a574cca96d2dca6a52d2864 : SUCCESS in 5m 02s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/28c8dc97e48e481ba9a77142c4bb26db : SUCCESS in 6m 53s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/e8c1f7bc676f4495883ba5f4b94ee2a4 : SUCCESS in 8m 22s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/679b65397c2f428ba81e0b271acc9a5e : SUCCESS in 1h 09m 02s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/18ff9c5a154649d8a0f079950a98bbd3 : SUCCESS in 1h 41m 24s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/82562f7ac19b47908b38511f2f6eb7a5 : SUCCESS in 1h 14m 48s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/0bd50658b99a4217aba827c90bdf4a9e : SUCCESS in 1h 16m 28s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/0494bd820866485884af8472c84a1fc2 : SUCCESS in 42m 35s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/b3c44788518b4ac08dbc3368bf279ad4 : SUCCESS in 1h 45m 38s\n- glance-grenade-centralized-cache https://zuul.opendev.org/t/openstack/build/17b923da3e87447595a011efc1ba151f : FAILURE in 21m 34s (non-voting)\n- glance-s3-multistore https://zuul.opendev.org/t/openstack/build/90887cfb2fd34c5ba672aabeb30b93cf : FAILURE in 12m 28s (non-voting)\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/8c3b97aa61b44fe193019f4724ec91ed : SUCCESS in 34m 16s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/52efea1711e14c2788fbd6d728ece39b : SUCCESS in 35m 14s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"c2b6e093b52ae984cad0b0d06074171f54a049cb","author":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"date":"2026-01-05 14:16:36.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"0af2b44447ad6ddd373bddf9e93b6a90cbb9685a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-05 14:17:19.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"3cf4a94e6a3fe6167318f43621a3be99d47a4ea9","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-05 16:50:42.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f14fbabbb6d8419faa82bfd5eaf80490\n\n- grenade https://zuul.opendev.org/t/openstack/build/45146444c67d402c866d301cbb1e817d : SUCCESS in 57m 15s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/da541acebc3d406cbbf1c501d4d16ad6 : SUCCESS in 54m 10s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/30ff2bffdbe04f0a95c22fd730ffa4a6 : SUCCESS in 1h 17m 49s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/bb81e925ab5b4a8ea58f4f1d7aa50ceb : SUCCESS in 1h 01m 25s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/28f16134200b469faf819c5c4046aeb9 : SUCCESS in 5m 23s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/d5f0dccdee4646eab7c69bdbfc53dd0c : SUCCESS in 6m 15s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/4e5f5e2499b34f91b4b7d37b42542f99 : SUCCESS in 7m 55s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1af603ab396b41198992d54df7929cc9 : SUCCESS in 6m 03s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/2613c89d1f744476960b137ac5b00d18 : SUCCESS in 7m 45s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/a9bbed804d834512aa2a2939795c1992 : SUCCESS in 10m 10s\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/5f6064a0be2143a6b42ca9ed81798226 : SUCCESS in 1h 56m 54s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/4fa6109607b54eeb9156ab31c6d414ae : SUCCESS in 1h 49m 45s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/a92c836f770244a79dcfcbd28b944742 : SUCCESS in 1h 01m 15s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/0c06683c3cf14bf38c93c8f96b5a5917 : SUCCESS in 1h 55m 24s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/ed40fc4553b949e7b9ecf2d3fbdd7ac0 : SUCCESS in 34m 16s","accounts_in_message":[],"_revision_number":1},{"id":"f3889aa84c4e0b7ecb2bc6a668bc776a894d2167","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-05 16:50:43.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"e925bf3a94d3b279b6b94d5896f6218560491125","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-05 16:51:56.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c23d14d47fb24bcab930ad13d15b9cae\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cf17c6a84f974971b33b5df2be8bc9a2 : SUCCESS in 59s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"3c55c87e19a3afbcb033494fd92883413adfb8e9","revisions":{"3c55c87e19a3afbcb033494fd92883413adfb8e9":{"kind":"REWORK","_number":1,"created":"2025-12-19 09:24:08.000000000","uploader":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"ref":"refs/changes/16/971516/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/16/971516/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/16/971516/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/16/971516/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/16/971516/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/16/971516/1"}}},"commit":{"parents":[{"commit":"04dffb2a203029988bc97d8a277e24e304789706","subject":"Merge \"Remove eventlet-specific functional tests\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/04dffb2a203029988bc97d8a277e24e304789706"}]}],"author":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2025-12-19 09:22:41.000000000","tz":0},"committer":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2025-12-19 09:23:36.000000000","tz":0},"subject":"Handle glance_store.Invalid exception in upload method","message":"Handle glance_store.Invalid exception in upload method\n\nAdd exception handler for glance_store.Invalid in the upload method\nto properly convert size validation errors to HTTPBadRequest responses.\nExisting unit tests verify the fix.\n\nCloses-Bug: #2136859\nChange-Id: Ia7552d2e3ef84a8e258bf918b632ae8d309ccd0a\nSigned-off-by: Abhishek Kekane \u003cakekane@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/3c55c87e19a3afbcb033494fd92883413adfb8e9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/3c55c87e19a3afbcb033494fd92883413adfb8e9"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
