)]}'
{"id":"openstack%2Fglance~972643","triplet_id":"openstack%2Fglance~master~I6e1e9e591908ee8bf6e154ef74d3656be954ddc9","project":"openstack/glance","branch":"master","topic":"eventlet-removal","attention_set":{},"removed_from_attention_set":{"9303":{"account":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"last_update":"2026-01-12 19:13:56.000000000","reason":"Change was submitted"},"8122":{"account":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"last_update":"2026-01-12 17:03:50.000000000","reason":"\u003cGERRIT_ACCOUNT_8122\u003e replied on the change","reason_account":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"}}},"hashtags":[],"change_id":"I6e1e9e591908ee8bf6e154ef74d3656be954ddc9","subject":"Migrate functional tests to SynchronousAPIBase","status":"MERGED","created":"2026-01-08 09:07:29.000000000","updated":"2026-01-12 19:15:38.000000000","submitted":"2026-01-12 19:13:56.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":8,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"972643-eventlet-removal","meta_rev_id":"dc62a064dd1a61a7044ed0a581d0f215b319b56e","_number":972643,"virtual_id_number":972643,"owner":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-01-12 19:13:55.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"all":[{"value":2,"date":"2026-01-12 17:03:50.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"all":[{"value":1,"date":"2026-01-12 17:03:50.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Top Priority / Holds Gate"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-01-08 09:08:07.000000000","updated_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"reviewer":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"state":"REVIEWER"},{"updated":"2026-01-08 09:17:09.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2026-01-08 09:52:54.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"0de59491b7e1e68e87baa43ef8170db49ee2d330","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2026-01-08 09:07:29.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"d4b98439236b4d82b57ed0031fad1d148f93fd1c","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2026-01-08 09:07:48.000000000","message":"Topic set to evnetlet-removal","accounts_in_message":[],"_revision_number":1},{"id":"8ad914fe3656d4dab416da56027bdf3fa867e0d8","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-08 09:17:09.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/77dbb1edd2e14709a310479efe91df50\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/c677276677434d9f89ed9d432cb5e6bc : SUCCESS in 6m 44s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/0e659bc37e094980969781b15050a8b6 : SUCCESS in 6m 14s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/419f78adf1004869934c32da6eca8c6f : SUCCESS in 7m 49s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"bdc0cc8286ae79a5508cc01470fc820c63949aaf","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-08 09:52:54.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c722cdfefea54521b3098159ffc7e1bf\n\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/98345a812cb94f7c993981328c940cc5 : SUCCESS in 28m 29s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/110f343f2e6e4c13870e38f0ffdd59ff : SUCCESS in 44m 12s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/47153d83569f4b93975664508358fa9f : SUCCESS in 4m 30s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/10910ed5c6304a8e8c9bf54a58ad629a : SUCCESS in 4m 13s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/46f301861e674ae4a0704f2d0a01ddba : SUCCESS in 8m 33s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/285f0e3af1d24db5b2d81bc45aa3be41 : SUCCESS in 5m 32s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/75a27842d95b4cce97005524c07ca3f4 : SUCCESS in 6m 25s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/3a24756a2be546f8b43da393b62f293e : SUCCESS in 8m 24s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/af4e2d6ca8144c72a59088bf0fa7dcb1 : SUCCESS in 32m 39s","accounts_in_message":[],"_revision_number":1},{"id":"b912a1529fb0c618490fd7f427f34c8d90ea4a1b","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"date":"2026-01-08 21:44:16.000000000","message":"Topic evnetlet-removal removed","accounts_in_message":[],"_revision_number":1},{"id":"a7d6c64a51daae65bf17adf5fc4dd941711c29cb","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"date":"2026-01-08 21:44:52.000000000","message":"Topic set to eventlet-removal","accounts_in_message":[],"_revision_number":1},{"id":"c475852d1c2975ca243fd56adc0bd12391834684","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2026-01-09 05:49:17.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":2},{"id":"47e2a75f2efdbc1f788acc75ee00528510b1d264","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-09 06:03:10.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/77a938dfc5074a1d94264a61f5965e1d\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/ff4de236fbfa439193433f0caa273022 : SUCCESS in 5m 57s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/e251ef7ecfcd4ba4bac893c4f05dc46e : SUCCESS in 7m 56s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/365f1f37db2c4f45b97e1ce7419243c1 : SUCCESS in 7m 54s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"e89c39aa5224a3f3e4f8b4d2f2acabff362ed14f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-09 06:20:13.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/c384dc18a35141789f7faabfaa5775cc\n\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/97fc09780f6047689ceaaf0af0f42a7a : SUCCESS in 29m 08s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/cf266c6b0b5f4098a68392639dedabdb : RETRY_LIMIT in 26s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ba204404f8584e8795995c03e13fe46b : SUCCESS in 4m 34s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/ac8fff3ee8c04a6a8f5587005c6f2039 : SUCCESS in 4m 17s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/080f2deb9c9446d6ad85e1d934d7a3cb : SUCCESS in 7m 45s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3efcaff7f0a54367b73314ece0b0360f : SUCCESS in 3m 23s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/316b6c77a3984429a29b9b5d66d36966 : SUCCESS in 6m 12s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/d9b5d1e680754e18a1d4ee1584784e18 : SUCCESS in 7m 07s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/35381095001342eaa01bfcb1be3051b4 : RETRY_LIMIT in 24s","accounts_in_message":[],"_revision_number":2},{"id":"42e79be3477affaf963ddf1e365eb78fbaebbb2e","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2026-01-09 06:23:57.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"d0f7c5ba9396f9699920adc61eb16bd13c267f4b","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-09 06:59:13.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/befdc43f14714825afba06090f69514c\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/83884d9322e34279a17fc5dc4b3bab9b : SUCCESS in 8m 36s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/64e7668eff7541779518040b4b9ea64a : SUCCESS in 8m 33s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/18a6db78392247988ccc694a38431f5f : SUCCESS in 8m 26s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"ae4405e57e91a394b4f58c0c4a69058784a2cbbe","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-09 07:13:59.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/7cc2b5185c3c478aaa1e90942f55eb4c\n\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/57f38c4d065c46b2af71955ce499e39c : SUCCESS in 39m 57s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/43ad0d689b754543ada7b14f99faf052 : RETRY_LIMIT in 35s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6ffdac95510b4bbeb1e797d098fb04b8 : SUCCESS in 4m 53s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/2bcacd42a6ac4220ba3e39681946b777 : SUCCESS in 3m 18s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/5c17d79b64f04e51996a0813d32a7386 : SUCCESS in 7m 58s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c370d80ecef5460f83c11413b6977b71 : SUCCESS in 3m 16s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/0fb2ea5cbfd14abba94676034069039b : SUCCESS in 6m 22s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/3b43ecfba10746dc9dbc8ff60ca62f5a : SUCCESS in 7m 12s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/ea4a9ac8b74c4bfd83c9e9931198fb0b : RETRY_LIMIT in 37s","accounts_in_message":[],"_revision_number":2},{"id":"88a5964aff65ddf5dae1ed9c4e8c0b594d311edc","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2026-01-09 08:13:36.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"eb7d408513835ef7900134b0a08fb6740426e0a7","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-09 08:39:50.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/034d29dc1f754d67aae4bb1b54fb1354\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/153cd3f7cf744735acdbef8ff213d085 : SUCCESS in 7m 13s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/16ded275155d40a9bfec3c073bc85224 : SUCCESS in 6m 57s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/1816e5c1ccc34aed808a2c9b5e7142da : SUCCESS in 8m 13s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"0232ab6e08de545377f86d8e5fc237438b862f16","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-09 09:20:47.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/b6b480862f6d4ed7963e916719c0429d\n\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/9f804d8229e8491f89fcfb92b8e14731 : SUCCESS in 57m 40s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/ea08b43a32604883acd4ff2d7381220c : RETRY_LIMIT in 41s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/000b0dfeed174085b5711f83f328b800 : SUCCESS in 7m 33s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/58b3206b1f9f4396a3fb35eba9ab1586 : SUCCESS in 5m 53s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/2724a5dab34140c49d486a673695c7f9 : SUCCESS in 7m 45s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c0e7d787bbca4eb0b37ac73dc75e2fdc : SUCCESS in 5m 19s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/cdd33c40652c4bdb8d687950cd2badf4 : SUCCESS in 6m 55s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/11e995c505f6408fab05a8250cfd2cc3 : SUCCESS in 7m 41s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/fe196e02733047a4884ae4e30735b9a1 : RETRY_LIMIT in 47s","accounts_in_message":[],"_revision_number":2},{"id":"1df4e14acf4a9aa19b6316eca9e752e7d4ceed17","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2026-01-09 13:12:53.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"a63bc03b4aded909ba1ca036070fced7be4ece4d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-09 14:27:44.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/380c959d4d904432a33b1e6e1f3b7481\n\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/039547079e1f4261844a6d22002db1b2 : SUCCESS in 54m 56s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/628140ccc6dc48df95ce742822623849 : SUCCESS in 1h 05m 22s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/63afb596c7004ae283689b74384bf317 : SUCCESS in 5m 08s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/964b0a8505bb44a4ba6ea85b7fff52b3 : SUCCESS in 5m 30s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/7cb1b58db83e4bd8a715596c1bde249e : SUCCESS in 5m 06s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7f2e789c3e25499d980033e3e2349776 : SUCCESS in 5m 34s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/3e48f15ba40d4986bfdcc05641e6bdd7 : SUCCESS in 6m 32s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/456fc174f10a430584cced4e8191d3fc : SUCCESS in 5m 23s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/3b4b4e8a17ff4737a3f48b1613fa256a : SUCCESS in 35m 53s","accounts_in_message":[],"_revision_number":2},{"id":"d130c6f9a25d2eb2ce2774add54aae475ef45cd1","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-09 18:12:32.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e64066ac38434e3c9dbba8ba6d8e92f1\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/057bd08a658f4f5b82e54e49e41f870d : SUCCESS in 8m 41s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/60babe19b1974501bcee7cfa5bca33c7 : SUCCESS in 10m 24s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/108329ad743c4fd2b4d940fccbe0d6b0 : SUCCESS in 14m 12s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"6104a52a6c2be178c73843c5a4db069153e29e9b","author":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"date":"2026-01-12 16:52:34.000000000","message":"Patch Set 2:\n\n(2 comments)","accounts_in_message":[],"_revision_number":2},{"id":"75acd71ebb9bec0c173c19c84c7c25c65559bb9a","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2026-01-12 17:01:28.000000000","message":"Patch Set 2:\n\n(2 comments)","accounts_in_message":[],"_revision_number":2},{"id":"6706cf525f1be36461e52d10309925c82ec82edf","author":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"date":"2026-01-12 17:03:50.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"ad34fbbae5a2ce3964a0e7302e74fe85852b3d90","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-12 17:04:25.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"3639726fa28709cac18fbacdf53ecb75c38717db","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-12 19:13:55.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b4bd35f16c2d429b8c2245d9666f8d4c\n\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/b698315665784a85850b34f9079e5518 : SUCCESS in 1h 01m 05s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/9b6490ef7b0b4afca21c1647676263bc : SUCCESS in 1h 01m 42s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d4bd76e815a44893bda2906cb252b3f4 : SUCCESS in 3m 19s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/119bfad89ec149f7b2dcad49dca0c313 : SUCCESS in 7m 20s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/eb40887c6bac43b38b866d6fc83d26fd : SUCCESS in 9m 17s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d76f3db03c2d45899b6923b52799f857 : SUCCESS in 5m 49s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/e833335e1c8b412998cc17381dd03971 : SUCCESS in 10m 00s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/ab5b0113a6c04d2b8d8e65ac7e80f547 : SUCCESS in 9m 25s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/1941b6b5fbc647bf938685eba18f675b : SUCCESS in 30m 58s","accounts_in_message":[],"_revision_number":2},{"id":"77e57fccbfd6f547f83a2d2d7ba523b6ee6b121b","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-12 19:13:56.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"dc62a064dd1a61a7044ed0a581d0f215b319b56e","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-12 19:15:38.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5689980565c64727adc95297a910d4e8\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0d9ef0f594604d6e829318f156faddbb : SUCCESS in 44s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"851fcaf5358c27f1e72f22b72f72b36d5a1314e6","revisions":{"8c3c2aeca8247d490b530446166373154133858e":{"kind":"REWORK","_number":1,"created":"2026-01-08 09:07:29.000000000","uploader":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"ref":"refs/changes/43/972643/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/43/972643/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/43/972643/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/43/972643/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/43/972643/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/43/972643/1"}}},"commit":{"parents":[{"commit":"01ead31b80b2973b2b196aad2fcde0a3947287d6","subject":"Merge \"Cache PropertyRules instances in Gateway\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/01ead31b80b2973b2b196aad2fcde0a3947287d6"}]}],"author":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2026-01-08 09:06:35.000000000","tz":0},"committer":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2026-01-08 09:06:35.000000000","tz":0},"subject":"Migrate functional tests to SynchronousAPIBase","message":"Migrate functional tests to SynchronousAPIBase\n\nMigrate test_api, test_logging, and test_images to SynchronousAPIBase\ninstead of eventlet-based FunctionalTest. Add MinimalIPv6TestBase for\nIPv6 tests. Centralize Controller response handling in _call_api().\n\nChange-Id: I6e1e9e591908ee8bf6e154ef74d3656be954ddc9\nSigned-off-by: Abhishek Kekane \u003cakekane@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/8c3c2aeca8247d490b530446166373154133858e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/8c3c2aeca8247d490b530446166373154133858e"}]},"branch":"refs/heads/master"},"851fcaf5358c27f1e72f22b72f72b36d5a1314e6":{"kind":"REWORK","_number":2,"created":"2026-01-09 05:49:17.000000000","uploader":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"ref":"refs/changes/43/972643/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/43/972643/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/43/972643/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/43/972643/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/43/972643/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/43/972643/2"}}},"commit":{"parents":[{"commit":"6009363d2a474c251f3953734b935e84c9c96c2d","subject":"Merge \"Avoid asserting internal processutils.execute call\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/6009363d2a474c251f3953734b935e84c9c96c2d"}]}],"author":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2026-01-08 09:06:35.000000000","tz":0},"committer":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2026-01-09 05:46:38.000000000","tz":0},"subject":"Migrate functional tests to SynchronousAPIBase","message":"Migrate functional tests to SynchronousAPIBase\n\nMigrate test_api, test_logging, and test_images to SynchronousAPIBase\ninstead of eventlet-based FunctionalTest. Add MinimalIPv6TestBase for\nIPv6 tests. Centralize Controller response handling in _call_api().\n\nChange-Id: I6e1e9e591908ee8bf6e154ef74d3656be954ddc9\nSigned-off-by: Abhishek Kekane \u003cakekane@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/851fcaf5358c27f1e72f22b72f72b36d5a1314e6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/851fcaf5358c27f1e72f22b72f72b36d5a1314e6"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
