)]}'
{"id":"openstack%2Fglance~972826","triplet_id":"openstack%2Fglance~master~Ie66a5c367bf825fe988e15889a2c17da3c97de24","project":"openstack/glance","branch":"master","topic":"eventlet-removal","attention_set":{},"removed_from_attention_set":{"9303":{"account":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"last_update":"2026-02-12 02:38:55.000000000","reason":"Change was submitted"},"8122":{"account":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"last_update":"2026-02-11 22:02:32.000000000","reason":"removed on reply"}},"hashtags":[],"change_id":"Ie66a5c367bf825fe988e15889a2c17da3c97de24","subject":"Remove CONF.workers check from simple backend","status":"MERGED","created":"2026-01-09 07:22:58.000000000","updated":"2026-02-12 02:40:36.000000000","submitted":"2026-02-12 02:38:55.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":5,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"972826-eventlet-removal","meta_rev_id":"998d1f5cdcc0a58a582704151a7898a72d1f3176","_number":972826,"virtual_id_number":972826,"owner":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},{"value":0,"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-02-12 02:38:55.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"recommended":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"all":[{"value":1,"date":"2026-02-11 22:02:23.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},{"value":2,"date":"2026-02-12 00:28:45.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"all":[{"value":0,"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},{"value":1,"date":"2026-02-12 00:28:45.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},{"value":0,"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Top Priority / Holds Gate"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-01-09 07:41:10.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2026-01-09 08:35:34.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-01-19 11:30:08.000000000","updated_by":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"reviewer":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"state":"REVIEWER"},{"updated":"2026-02-05 16:45:41.000000000","updated_by":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"reviewer":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"state":"REVIEWER"}],"messages":[{"id":"7e3b00b13db58e28b21b7ce26a2255bb33b63d51","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2026-01-09 07:22:58.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"58867b99f398e185ad22ca17997e95b8fd0f7da1","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2026-01-09 07:23:20.000000000","message":"Topic set to eventlet-removal","accounts_in_message":[],"_revision_number":1},{"id":"6d0a6958653d9b8ffbfecfd7f6e3493fb2b4c507","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-09 07:41:10.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9406d641ae4b49cebd2ab0cfa660f967\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/5d9075275b5647dc8196dd355d191375 : SUCCESS in 7m 57s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/02987f0fef114624a0270778f6aa91b0 : SUCCESS in 8m 39s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/42fb2823446a483492f37c6970626d3e : SUCCESS in 7m 31s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"c77784b49a0221877de7bc7f160b7e6b0dab2413","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-09 08:35:34.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/83b4a652f0b642f8a64382cfe73cbd22\n\n- grenade https://zuul.opendev.org/t/openstack/build/a3eb16dd173041fcab36fd9c80facc3d : SUCCESS in 58m 43s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/4e08d3f8d58a437194874105d9649518 : SUCCESS in 53m 00s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/f60284e345694b2491efc234a0bacfdb : RETRY_LIMIT in 34s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/f76c71f24d46473583a9210e2c395e2b : RETRY_LIMIT in 28s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/eaa0693c3beb46b4be598afccc919fac : SUCCESS in 6m 26s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/e2e1b7fc62744de093697fc17f7aa039 : SUCCESS in 4m 48s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/f82649acacb14d2291c899de4e39e76d : SUCCESS in 6m 16s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/462f3864d450459a8f1806004600e20b : SUCCESS in 7m 06s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/d353a8c10636425ebb7d0b7727af1bae : SUCCESS in 6m 36s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/384854d4ea6146cfaa264f61487c3981 : SUCCESS in 6m 43s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/ddcbaf2263fe4190ac25d589b393a880 : RETRY_LIMIT in 50s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/a270db8af0924ae38d6e18f8174335b4 : RETRY_LIMIT in 23s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/d418bee45992466c9849658538314d11 : RETRY_LIMIT in 51s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/a67f04840f774661aca4015fbf43ffbb : RETRY_LIMIT in 38s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/263c141c7c9042418001cd31226a136b : RETRY_LIMIT in 37s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/f4329fdedeef4855878c05139e826473 : RETRY_LIMIT in 27s\n- glance-grenade-centralized-cache https://zuul.opendev.org/t/openstack/build/699d53ce34054669bebc082f2fb8a980 : FAILURE in 20m 03s (non-voting)\n- glance-s3-multistore https://zuul.opendev.org/t/openstack/build/282b9dd1ee59429aae9daa8eef603bd7 : RETRY_LIMIT in 1m 09s (non-voting)\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/89294ddbf2934119b7a76f9e0f92c80e : RETRY_LIMIT in 42s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/0db403deafdb443fb0e8cb8ef962d755 : RETRY_LIMIT in 33s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"2851ffc7c4065afbdcea0df7930ef055c688f4f6","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2026-01-09 13:12:07.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"bba3bb18219aa315a1ad532d912bc62ec07edc39","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-09 15:26:44.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/19a54fdf5bf445f89a5dcc269ae12935\n\n- grenade https://zuul.opendev.org/t/openstack/build/53299cb56700478f8619284d322028f4 : SUCCESS in 50m 59s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/3e6ee8b6599044a1864a4d758a21ad05 : SUCCESS in 1h 03m 43s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/62102fb4d2ac46a4b6922ded41871567 : SUCCESS in 1h 35m 56s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/0a16f94a83dc42868b33fbe8c103f75e : SUCCESS in 33m 30s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/992d2f7ec3b0456ea3e5f98019094929 : SUCCESS in 4m 51s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/6baf65daee65467092774f7253b326a9 : SUCCESS in 5m 06s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/027c2d1e5b324120a882c48101bb7cca : SUCCESS in 9m 28s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/48222813bdca4b0ebf27f98726cef8ae : SUCCESS in 3m 58s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/d837b96a9287497db0634a2e11676504 : FAILURE in 8m 36s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/a39a334a643b4e17bba94c3b08747506 : SUCCESS in 7m 01s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/9b00522277d344d9998b4cdc7b265fa8 : SUCCESS in 1h 57m 36s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/628d1621c5d84ea8b8555ce7eb269255 : SUCCESS in 1h 36m 29s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/e1a87c8308024246ad0ab63caf867acc : SUCCESS in 1h 39m 30s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/01d974f9af324e4d84094466efcfdf61 : SUCCESS in 1h 21m 11s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/982c561b2a44457d8e346720ab26216c : SUCCESS in 1h 43m 36s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/e36d8bc061f2400dbb61c4676cdaf08f : SUCCESS in 1h 51m 01s\n- glance-grenade-centralized-cache https://zuul.opendev.org/t/openstack/build/9aac45171252458caf8c779e85e8d6ad : FAILURE in 11m 59s (non-voting)\n- glance-s3-multistore https://zuul.opendev.org/t/openstack/build/36e00d4a077f4b849133e2a26857926a : FAILURE in 12m 05s (non-voting)\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/49e8d3ba6d434e668488ade2a79ec7ed : SUCCESS in 17m 57s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/3f42ce81bce64a2385a2be6486fc3c90 : SUCCESS in 1h 10m 21s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"bf779e9d5c50e258995150b915b0244b0629962c","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2026-01-09 15:28:58.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"5bbf0005c626acc73c0709cb818cdf4684d6052e","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-09 17:22:49.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9d5c576f95114682ba7972abf2c7d7d3\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/6cfbd84879b14bfd9ee773c1f8f2b62d : SUCCESS in 7m 35s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/0906624cf14e498a848c81eeff768945 : SUCCESS in 7m 39s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/86480638762c4299b15ccdc639463d69 : SUCCESS in 14m 39s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"834d773b9dfeebc80ed6f63e71500edb66004acc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-09 18:12:10.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/bfc7bdf3f643461fb39952003cb8797f\n\n- grenade https://zuul.opendev.org/t/openstack/build/8309aa62ab884b0a853062844a72cd78 : SUCCESS in 1h 00m 16s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/7aa94f3103914dd7862f007db8b28b7d : SUCCESS in 1h 02m 07s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/0d6ef950e20142e18396c7ec5628bbdd : SUCCESS in 1h 43m 57s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/cc22535ca6624bf29bda75d6e25efe44 : SUCCESS in 56m 33s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d3dbad40c51241b6ac24c928215a9793 : SUCCESS in 4m 43s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/3377f9d9f37149a09eb49be81ac1eb58 : SUCCESS in 6m 44s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/b404014070564c68a474478fba016f5a : SUCCESS in 7m 55s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1ac311c7d48c471cbeddff0f5f8a95f2 : SUCCESS in 7m 21s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/ebc4e201b63b4a8aa9904eb9aa98b4ed : SUCCESS in 5m 56s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/75bbad50a133448fbf9dec73e7592219 : SUCCESS in 7m 38s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/89c7e951c2c44cbd92709cf0b954c0d9 : SUCCESS in 1h 18m 03s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/ae4f694a61c7440385ba94423583150f : SUCCESS in 1h 18m 03s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/afbe9bd3f80b4619b2823cecffff2630 : SUCCESS in 2h 14m 44s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/f498c16c339f4870a169038c57dbfefd : SUCCESS in 2h 15m 59s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/5b7a4f2128db493aacf80e641747fcf1 : SUCCESS in 57m 22s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/101bfc7c22e6478a80932e2e27fcb5c0 : SUCCESS in 2h 07m 28s\n- glance-grenade-centralized-cache https://zuul.opendev.org/t/openstack/build/ac6f9f0ca9324509b9216223c9478caa : FAILURE in 22m 39s (non-voting)\n- glance-s3-multistore https://zuul.opendev.org/t/openstack/build/6356424d30bd4ebb8414e00d755504ed : FAILURE in 18m 48s (non-voting)\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/052d07f525e74e44854b17bf33afd3a3 : SUCCESS in 34m 27s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/5c778f28763945e7a7e11d17a6744a3c : SUCCESS in 41m 54s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"5d9405ac38bac957a3596e6ca64c91f09d1b194d","author":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"date":"2026-01-19 11:30:08.000000000","message":"Patch Set 1: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"5c15d3cc064f24fa8912d2b11bdb9e104f8da9e9","author":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"date":"2026-02-05 16:45:41.000000000","message":"Patch Set 1: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"fa08c38c97b8cec2c2480ab7e6b09a86baaa7768","author":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"date":"2026-02-09 14:57:35.000000000","message":"Patch Set 1: Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"314ea09e8e3a03086be81e3387b52ab09b5d09ce","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-09 14:58:23.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"1138342127a4d1a145e47ccd00be473dbb41e42b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-09 17:18:38.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/0a2ae808ce13416690b9226f15989852\n\n- grenade https://zuul.opendev.org/t/openstack/build/dd4f7610bd5848c09b7f080c2a56ffee : FAILURE in 13m 09s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/7de64893c5514e889a2396bf0ca82afb : FAILURE in 12m 46s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/a87500eca7c34bbbacac355c493ae937 : SUCCESS in 1h 35m 03s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/bb22011321c54119a39ef8f84cb18e39 : FAILURE in 1h 01m 56s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c35cceeaf4454fb98143f362b4275f7e : SUCCESS in 3m 17s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/c14df6141c674dd3b92b49d2d267c584 : SUCCESS in 4m 27s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/5e9c641da0be48d793cc01da63af3cea : SUCCESS in 5m 46s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/36e85a44188e4fd4b2c3b4b3c7cd4123 : SUCCESS in 5m 48s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/73aff46e630a4ad98f083b3989b567b3 : SUCCESS in 6m 02s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/f045e253f9f7418c9b191c2f69706f71 : SUCCESS in 6m 36s\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/8326039212d940739c6102230632e05a : SUCCESS in 1h 38m 39s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/d7a8279c8706490cafe013b29b466038 : SUCCESS in 2h 15m 36s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/2a4bfecbf29a47c3a46092a60a65ffc1 : SUCCESS in 1h 03m 03s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/45c31cbcf4bb47e58dff7c4fae554544 : SUCCESS in 1h 44m 14s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/84ffcef723934af19e2abf10c7d66205 : SUCCESS in 22m 05s","accounts_in_message":[],"_revision_number":1},{"id":"e38b17bae5c0c98ad46b80d710b330b47db9e614","author":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"date":"2026-02-11 20:07:29.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"95fb7b2ba947b0c7fa41a4e8e5934f7ffd739abd","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-11 20:29:36.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/305a013b374a4800b4ab5763fd67e4ea\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/174a06d7c97c409d9802de7e3806f158 : SUCCESS in 9m 43s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/dfa3570d03da4fa791f24e9330b578ca : SUCCESS in 8m 07s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/5548d6fe73444d00b7c18d723d4628fc : RETRY_LIMIT in 4m 01s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"3f5b1202770a58829edfc41609c8a6390f9c54fb","author":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"date":"2026-02-11 22:02:20.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"0757992637c4258cfd577f6e1152f406d5673289","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"date":"2026-02-11 22:02:23.000000000","message":"Patch Set 2: Patch Set 1 was rebased\n\nCopied Votes:\n* Code-Review+1, Code-Review+2 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n\nOutdated Votes:\n* Verified-2\n* Workflow+1\n","accounts_in_message":[],"_revision_number":2},{"id":"c019248b4cd6a54658006ac344cc74150e61487b","author":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"date":"2026-02-11 22:02:32.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"c8ceae9071c597d3ad156f0a19c2918536803836","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-11 22:18:01.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1d80851231aa48ef8a2220e3705b79dd\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/bc7e5825235943969438bbfdfe4a77ed : SUCCESS in 8m 00s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/05651605cdd44afb9323b16c75d87498 : SUCCESS in 8m 35s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/841802120a8440c3b5710dd4a6efbef0 : RETRY_LIMIT in 3m 10s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"7c12e6f5c6301efbc5ff7ce5d39169d45a87c012","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-12 00:27:48.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/78ba5f863cdd4333aa9a9f97c2b9eb44\n\n- grenade https://zuul.opendev.org/t/openstack/build/e2f29ac813c7437ca925ee92cdda537c : SUCCESS in 42m 25s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/40b0ded7b0a244e5b2c2c25f4340cf77 : SUCCESS in 1h 02m 49s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/c9dfbd65d2034d6ebfa5fccc3d882df5 : SUCCESS in 1h 34m 08s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/d2c380ca3510431aa1bd17c4d2345f64 : SUCCESS in 1h 10m 33s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/850846f0b00d4d77915ba151eeee3a9d : SUCCESS in 4m 32s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/40fad358760547db9de2816b5e64f927 : SUCCESS in 2m 43s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/c77df6bbc430482ca9b93de52684d8d8 : SUCCESS in 2m 44s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/20e98d43de2f4e09817ea321ad742c6d : SUCCESS in 2m 59s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/8e1cf636fadd409c8e9f0fee7906fa33 : SUCCESS in 6m 09s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/69523463aebb4a8fbaed00fbfd179b4c : SUCCESS in 6m 26s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/a18328d0381146a097e5b8a688e11d67 : SUCCESS in 1h 52m 31s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/980286f7ef3c4b2baf82eae973776b62 : SUCCESS in 1h 53m 42s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/cc2742c5a5c2479687ac4bded3e2f4f0 : SUCCESS in 2h 09m 36s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/dabf3237ea3b437a9dd5059076995f47 : SUCCESS in 2h 19m 58s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/6ee351a980fe4492bac0ccad01289606 : SUCCESS in 43m 31s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/234d6ea0a7fc4344add14fb633cf67fa : SUCCESS in 1h 32m 25s\n- glance-grenade-centralized-cache https://zuul.opendev.org/t/openstack/build/d51b39ffbae8449f88aa11ebb894d642 : FAILURE in 21m 06s (non-voting)\n- glance-s3-multistore https://zuul.opendev.org/t/openstack/build/86fdfe1026214fa4978e1aa075f68b94 : SUCCESS in 1h 56m 15s (non-voting)\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/a1c045d45ecd408a87efb27a1b7970d9 : SUCCESS in 15m 33s\n- ironic-tempest-bios-ipmi-direct https://zuul.opendev.org/t/openstack/build/b024c613f46c451bb6538eda44bae1e0 : SUCCESS in 1h 03m 20s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"3372a5a64f493c2414e203c3044abaa345619513","author":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"date":"2026-02-12 00:28:45.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"097314795116dd3834a9a27c99bb4dd5a317d79e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-12 00:29:10.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"1b437d90e81188b8304b347b477e18117c78dc73","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-12 02:38:55.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1495fe028bc8469086ce32b7d6eaa515\n\n- grenade https://zuul.opendev.org/t/openstack/build/a580d0a1b85b4b008fc9052fe87c221b : SUCCESS in 33m 17s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/c7daf609543c487aa4800bad3a3b910a : SUCCESS in 29m 50s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/a86a8c8ae70f44ebb1cf1ef9b328be05 : SUCCESS in 1h 32m 55s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/bc3b42dcf792423a863dbd2ba1b937c3 : SUCCESS in 34m 22s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bd18816134f74b809b6601fc86d80812 : SUCCESS in 4m 36s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/73d31e045341464695efb0c364995558 : SUCCESS in 4m 43s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/c358518b7a6d4a148392cebff0415785 : SUCCESS in 7m 59s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/812d51999fcc40ac91a9fd59a62959de : SUCCESS in 3m 02s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/869b663624544737b539d02c4048b85e : SUCCESS in 6m 27s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/14916b4517f84fb5b7d0e60662f2efca : SUCCESS in 6m 39s\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/12b03618cf294dca95a9d4fef3ffc35d : SUCCESS in 1h 13m 12s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/79668ddbf8c446e196c22c25f220e449 : SUCCESS in 1h 05m 21s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/8ade8cefdffb449fba72fad203b85dd5 : SUCCESS in 58m 03s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/5b2fd06e410f425b8727f6ff26cf82e5 : SUCCESS in 2h 05m 37s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/9a90ae9cb0dc4907ab88c41cbec79097 : SUCCESS in 33m 22s","accounts_in_message":[],"_revision_number":2},{"id":"c69728df861d75b55cc3fe3b858c61ed96dc7971","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-12 02:38:55.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"998d1f5cdcc0a58a582704151a7898a72d1f3176","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-12 02:40:36.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ea9db92691d444379665793da0001287\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f2f369e8cf364e17b960ccbfbd821643 : SUCCESS in 1m 16s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"e2b0fd70d2468b9fb311ae83ee6f3ecd521e4f78","revisions":{"f90ae5a7f54cc1c2fc8a242ad88d7133ce730806":{"kind":"REWORK","_number":1,"created":"2026-01-09 07:22:58.000000000","uploader":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"ref":"refs/changes/26/972826/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/26/972826/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/26/972826/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/26/972826/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/26/972826/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/26/972826/1"}}},"commit":{"parents":[{"commit":"6009363d2a474c251f3953734b935e84c9c96c2d","subject":"Merge \"Avoid asserting internal processutils.execute call\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/6009363d2a474c251f3953734b935e84c9c96c2d"}]}],"author":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2026-01-09 07:21:55.000000000","tz":0},"committer":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2026-01-09 07:21:55.000000000","tz":0},"subject":"Remove CONF.workers check from simple backend","message":"Remove CONF.workers check from simple backend\n\nThe standalone server that used CONF.workers is being deprecated.\nMake configure() a no-op since workers are managed by uWSGI.\n\nChange-Id: Ie66a5c367bf825fe988e15889a2c17da3c97de24\nSigned-off-by: Abhishek Kekane \u003cakekane@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/f90ae5a7f54cc1c2fc8a242ad88d7133ce730806"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/f90ae5a7f54cc1c2fc8a242ad88d7133ce730806"}]},"branch":"refs/heads/master"},"e2b0fd70d2468b9fb311ae83ee6f3ecd521e4f78":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2026-02-11 22:02:23.000000000","uploader":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"ref":"refs/changes/26/972826/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/26/972826/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/26/972826/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/26/972826/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/26/972826/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/26/972826/2"}}},"commit":{"parents":[{"commit":"96040c00a65c3c73ca11e771663b0d7f01415fe0","subject":"Merge \"Deprecate the glance-cache-{cleaner,prefetcher,pruner} commands\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/96040c00a65c3c73ca11e771663b0d7f01415fe0"}]}],"author":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2026-01-09 07:21:55.000000000","tz":0},"committer":{"name":"Cyril Roelandt","email":"cyril@redhat.com","date":"2026-02-11 22:02:23.000000000","tz":0},"subject":"Remove CONF.workers check from simple backend","message":"Remove CONF.workers check from simple backend\n\nThe standalone server that used CONF.workers is being deprecated.\nMake configure() a no-op since workers are managed by uWSGI.\n\nChange-Id: Ie66a5c367bf825fe988e15889a2c17da3c97de24\nSigned-off-by: Abhishek Kekane \u003cakekane@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/e2b0fd70d2468b9fb311ae83ee6f3ecd521e4f78"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/e2b0fd70d2468b9fb311ae83ee6f3ecd521e4f78"}]},"branch":"refs/heads/master","description":"Rebase"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
