)]}'
{"id":"openstack%2Fglance~973195","triplet_id":"openstack%2Fglance~master~Ic37985dee753b823b1f1c29908f47c4463e3b0fc","project":"openstack/glance","branch":"master","topic":"keystoneauth-cleanup","attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2026-08-27 17:48:06.000000000","reason":"Someone else replied on the change"},"8122":{"account":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"last_update":"2026-08-27 17:41:51.000000000","reason":"\u003cGERRIT_ACCOUNT_9816\u003e replied on the change","reason_account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"}}},"removed_from_attention_set":{},"hashtags":[],"change_id":"Ic37985dee753b823b1f1c29908f47c4463e3b0fc","subject":"Use keystoneauth to get token","status":"NEW","created":"2026-01-13 04:20:34.000000000","updated":"2026-09-04 07:45:44.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":true,"submittable":false,"total_comment_count":11,"unresolved_comment_count":2,"has_review_started":true,"meta_rev_id":"e903c8ef6a3cb10c54560d7254671de437de8ec8","_number":973195,"virtual_id_number":973195,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"tag":"autogenerated:zuul:check","value":1,"date":"2026-08-27 20:08:57.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"recommended":{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},"all":[{"value":2,"date":"2026-09-04 07:45:44.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":1,"date":"2026-08-27 17:48:06.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Top Priority / Holds Gate"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-01-13 04:35:57.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2026-01-13 06:28:18.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-08-27 13:48:17.000000000","updated_by":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"reviewer":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"state":"REVIEWER"},{"updated":"2026-08-27 17:48:06.000000000","updated_by":{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},"reviewer":{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},"state":"REVIEWER"},{"updated":"2026-09-04 07:45:44.000000000","updated_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"reviewer":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"state":"REVIEWER"}],"messages":[{"id":"6ba7a7448846eb33c60cfbecc619d9776708d545","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2026-01-13 04:20:34.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"15a7f005d269bc743ec062a4f5b475f19512f802","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-13 04:35:57.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8c6bd5dbf9c14560a8ec5b5596a46165\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/f490c2ca8a034ed39839864d74e8c85f : SUCCESS in 8m 53s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/aa2dc875a6a44560b988bb839b15e872 : SUCCESS in 9m 15s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/b1421a3c90f44f1786b3de9bd351e476 : SUCCESS in 13m 23s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"417f0d3feafff7e32972554fe28d9da5fbc496ea","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-13 06:28:18.000000000","message":"Patch Set 1: Verified-1\n\n(6 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/5ec25aec358c4ff7932c63b0a75646c1\n\n- grenade https://zuul.opendev.org/t/openstack/build/6b186ec0edad45bca66e65a76e5c06d3 : SUCCESS in 59m 14s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/4e94e00885054edf90f88bf13521222e : SUCCESS in 59m 00s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/ccca7d8e291542c8be015b8e4347027e : SUCCESS in 1h 33m 37s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/a65ab0450a9d4aecb20b059771172b39 : SUCCESS in 33m 31s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9e3e5de9ed7544f390e56d489bbfe906 : FAILURE in 3m 04s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/9919dedc549c4ab48a7cc0f8896968b9 : SUCCESS in 4m 39s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/5f09a1f5bfcc492aaf9519d8263b2f70 : SUCCESS in 7m 44s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a88d7b47466a4a5fb4329d858bcf0d40 : SUCCESS in 6m 39s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/1c59eef0bfbd419793c0daf02e8d9177 : SUCCESS in 5m 11s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/aed791cafa0c4b1f9c9163ccddd4ffbe : SUCCESS in 7m 25s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/383aa2bed38443609a670e8f4d20554b : SUCCESS in 1h 50m 53s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/b86d46dc8fed4e64b456f17ff2a98606 : SUCCESS in 1h 16m 31s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/50e88ab9e7e449faa6728a47f5d91cca : SUCCESS in 2h 02m 38s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/60def29fd04d42948b596138f8403992 : SUCCESS in 1h 36m 48s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/79a1e465795a43608d86405e88008efc : SUCCESS in 1h 05m 11s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/523130e9137d443fa2cde9b1332e5eda : SUCCESS in 1h 11m 35s\n- glance-grenade-centralized-cache https://zuul.opendev.org/t/openstack/build/fe262b8b9b3344c39c61f1d05ff09299 : FAILURE in 12m 42s (non-voting)\n- glance-s3-multistore https://zuul.opendev.org/t/openstack/build/f0130a433b074cdbb75cecea1ff620ce : FAILURE in 19m 14s (non-voting)\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/2f5183904d5c43e589328571c85ff67a : SUCCESS in 33m 53s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/97542e9a89e2496da9bb824031d402cf : SUCCESS in 1h 11m 09s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"69c4031d98d013d3b7384815c2f308b3966af75f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2026-01-13 09:10:25.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"ab4fdc5c2f4bb50da922ed26fd6c4a121d33df3f","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-13 09:23:47.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b85b44556b094f6ab62e425709b0ee4a\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/bd638f6113774612b8137fd3f81e9091 : SUCCESS in 6m 38s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/0c81d0c647654a2f915c1f569cae2801 : SUCCESS in 6m 47s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/f973de8865074d36bbaef3430b07f534 : SUCCESS in 10m 23s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"fef2587c36eb7302395898927a156efc14b1361e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-13 11:22:55.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8f7c7e0d32c042edab4d00377de3b9a8\n\n- grenade https://zuul.opendev.org/t/openstack/build/7a3a9d9a423546508adc1db6371de625 : SUCCESS in 56m 47s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/76c5d1b7bffb4c768817c1407f0f5d67 : SUCCESS in 1h 04m 57s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/b68f10d15f784205a93b0e16a27700a2 : SUCCESS in 1h 36m 35s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/df9123135d334efd9e0a9508e79fbca1 : SUCCESS in 54m 59s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/53345eaecef4468395750b931e7b1166 : SUCCESS in 4m 45s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/41c7ebcd32bb4304a59ee46b68dd1e0b : SUCCESS in 4m 48s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/e649696dbcda41ccaca389de0950f16c : SUCCESS in 7m 59s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5ce5419b97194d88b52b34d0e66ef706 : SUCCESS in 5m 35s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/804a5cbf4e674091b63e479def9547fd : SUCCESS in 6m 12s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/50e7e5fdeecc4623b3d137c874df8b49 : SUCCESS in 5m 06s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/6820ad66454b4c94a98d15c935df1e56 : SUCCESS in 1h 51m 33s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/8bba6337d9a54ef79a04bbd2b624b259 : SUCCESS in 1h 57m 33s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/d578640a9fb14db4a1ebaaf3309dcfd2 : SUCCESS in 2h 01m 18s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/7432a206e4f64e45b9614f5b62675007 : SUCCESS in 1h 49m 22s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/2fc9b295325b44b99b7b227e7a300e71 : SUCCESS in 1h 00m 38s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/47c0571968154127a3b84b8bb0a1fb37 : SUCCESS in 2h 07m 10s\n- glance-grenade-centralized-cache https://zuul.opendev.org/t/openstack/build/9fcca232d10e45b0b825d0c24b920898 : FAILURE in 11m 02s (non-voting)\n- glance-s3-multistore https://zuul.opendev.org/t/openstack/build/91928eeb4fe04d1e9ddb6a1293279908 : FAILURE in 19m 43s (non-voting)\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/fcbd613ef39a4106b7d1a5c715601ada : SUCCESS in 29m 47s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/8e5d199359c7483d9ad5b4a0052a7a30 : SUCCESS in 1h 05m 45s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"6e5ae2004cb0c172557300b0260f0e386b6ccb3a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2026-01-13 15:03:28.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":3},{"id":"f1397b52154a2671ce29c7b91db48fb4c3e00f87","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-13 15:21:59.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2d70631d1f824e958d37697da252e9bb\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/bc4c7a2a417944d29fe512b3f90dad5a : SUCCESS in 5m 31s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/9320e144b19240c4acbefb5ce06d4550 : SUCCESS in 7m 38s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/a204c702ada24995a73e1d323fcbcf9e : SUCCESS in 9m 19s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"f60b2fd8e172be4b7e07ec2dd3e180b71bef58a1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2026-01-13 16:17:43.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"3ec9a11140e5dc9add3a98507a6197a4645d14d0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2026-01-13 16:23:19.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"c0f26bfd3789d1544e80b1779efab0d72aec52a4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2026-01-13 16:25:24.000000000","message":"Uploaded patch set 6: Patch Set 5 was rebased.","accounts_in_message":[],"_revision_number":6},{"id":"d4b088e386a90fce1313db4d818e4dfb12f255bd","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2026-01-13 16:27:31.000000000","message":"Topic set to keystoneauth-cleanup","accounts_in_message":[],"_revision_number":6},{"id":"708d910bd46c60fd2d588e258ee0fcf6bc9d0b4e","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-13 16:38:31.000000000","message":"Patch Set 6:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/64bd6c4a073e4c04b0d247f9ae512b51\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/de94fc0246294542b6d2d4d147d86ce9 : SUCCESS in 6m 19s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/73c3aea5c8da463da7d92feced52df1f : SUCCESS in 7m 23s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/9a0ab6af0abb426c887e085a3cab0c69 : SUCCESS in 9m 31s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"048abb027ffaa0c559636d1685fc41aefbd22a21","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2026-01-13 17:36:36.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"e095cd3db73ea00e83b8b7217c853b4efdcc3f31","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-13 18:09:28.000000000","message":"Patch Set 7:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/abbca9b532ab48d5baafb4cd47fd0090\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/d2d58cf8c0d747e285307370a3ac263e : SUCCESS in 9m 56s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/2362c86dcefb4a9f85a46ce8818bb297 : SUCCESS in 9m 32s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/719e48ae17254a8db0f6ed5a58761e45 : SUCCESS in 10m 20s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"6ce06ad3602e6af09f1b471911aa7e7fad974f71","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-13 19:57:05.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/57c711b0427e4a7db89a3a818f53feb6\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/0346ba588fa5411ebdb8b42003984be9 : SUCCESS in 4m 27s\n- requirements-check https://zuul.opendev.org/t/openstack/build/a08b54b63941481fb6abeaa5569afaab : SUCCESS in 2m 56s\n- grenade https://zuul.opendev.org/t/openstack/build/afb53ab327bc442f94f790aa11417bdd : SUCCESS in 32m 07s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/2169ecea2c5948bfb5a284b03d20508e : SUCCESS in 1h 03m 31s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/bd440e8b763249789fde8006a3ed1cf4 : SUCCESS in 1h 38m 43s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/68310dcf1cfd44588d95c642308f1549 : SUCCESS in 1h 05m 11s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2a90fcac1401476fbe15baef1b8c45f6 : SUCCESS in 6m 26s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/11ad8c1d74e247c4945fb8f4a8812a82 : SUCCESS in 2m 53s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/90fe2c18155f441999637572154137c5 : SUCCESS in 4m 33s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/986d3a18014346ceb01fc71d8593954c : SUCCESS in 3m 22s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/0d3a969d8c034e4abb08dcf12f8b05ed : SUCCESS in 7m 36s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/df452f8c01bd45b49f983336e0511ce1 : SUCCESS in 4m 39s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/e3d6b91c2f954129b28c62e1ccfe7a16 : SUCCESS in 5m 14s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/57b1dd51b2924c9ca4a55f62c01fbffd : SUCCESS in 2h 05m 12s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/8879dcb3ebfa4a7a9644d92ec5133d73 : SUCCESS in 2h 02m 35s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/f615d353fc9b43b3a758092ced4df7c9 : SUCCESS in 2h 11m 27s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/24847db09c3047469c8f72c1d8faa236 : SUCCESS in 2h 12m 15s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/a475da4889ec442a86fd74dc087f6eb7 : SUCCESS in 59m 47s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/f0dcaa68a7b04af59318e7c5d91351ed : SUCCESS in 2h 00m 02s\n- glance-grenade-centralized-cache https://zuul.opendev.org/t/openstack/build/570b05d7ebc84373bdf19e5726568d36 : FAILURE in 20m 50s (non-voting)\n- glance-s3-multistore https://zuul.opendev.org/t/openstack/build/65ec83f844804cb0b3b0e0a8da7ec075 : FAILURE in 24m 41s (non-voting)\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/7698a469027d4c5d8cd39fa400041970 : SUCCESS in 19m 05s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/19a2ab350cd646c790f94b94a529a0b4 : SUCCESS in 34m 20s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"f8896c9b6b9babe3b36c5257ef3105f64a892b58","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2026-01-13 23:48:04.000000000","message":"Uploaded patch set 8.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":8},{"id":"1139d65539489ed0500369ec3da6a7adb88fdb2e","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-14 00:04:55.000000000","message":"Patch Set 8:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d461e820da284242a405213d89d760eb\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/0457d664990a413b996d280fa355a5bb : SUCCESS in 9m 01s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/695d5c2bcc8243dab19de1b235018acc : SUCCESS in 10m 10s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/2780ae4ddf6149b3a825f999b600392b : SUCCESS in 14m 30s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"d1557adae734a481160efd8b240ba4b76749e788","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-14 02:20:29.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b8c66688d218472cb4181c6509f27eb4\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/916b00be251e47eba74efdf73cef24ce : SUCCESS in 3m 44s\n- requirements-check https://zuul.opendev.org/t/openstack/build/0916ae567aa04b21b040a30bbf33fd9c : SUCCESS in 2m 06s\n- grenade https://zuul.opendev.org/t/openstack/build/be771b50f6fd46dfa7029f2fa1bc9790 : SUCCESS in 1h 03m 08s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/f17d74ad34c149f3a4e9e97de281ae7c : SUCCESS in 58m 05s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/8f0f92a949cc4ae5a47505e047b06fad : SUCCESS in 1h 18m 08s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/7082e2488ab349e4937d4c09a78a211a : SUCCESS in 44m 29s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5d34aababab34397a816e11e55a56815 : SUCCESS in 4m 32s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/591fb60ac1f54142952391de49d30ecc : SUCCESS in 4m 42s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/dc5e9ff1953947068701ef5f42e55e73 : SUCCESS in 4m 13s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/29ed9bdf755b464b88c59e77473d8a65 : SUCCESS in 5m 28s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d1fb959ef98e4a51bfb6ca4a75aa363a : SUCCESS in 5m 03s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/6c99a4e81ade4780beeed7d3cb16e7a9 : SUCCESS in 5m 07s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/96274f44f22d402bbab35017f2d61cc2 : SUCCESS in 6m 21s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/671fe8167eee42c9aacf2c2b0fba7b1a : SUCCESS in 1h 40m 27s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/e7557b935c8a4ad187f303d73933a715 : SUCCESS in 1h 09m 24s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/1cedb041f3ba476eb156e21ebdcc21ed : SUCCESS in 2h 05m 20s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/c8519c4922b749bc90c40b97ddbf3782 : SUCCESS in 2h 21m 15s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/c631b0055c824eb1875d44a268d79d22 : SUCCESS in 59m 02s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/244650e6299c4bc4a7b6c43c69c8944d : SUCCESS in 1h 57m 16s\n- glance-grenade-centralized-cache https://zuul.opendev.org/t/openstack/build/a1dfd33588c54491aed0694a082448e1 : FAILURE in 12m 33s (non-voting)\n- glance-s3-multistore https://zuul.opendev.org/t/openstack/build/a12cee08edeb45b6a8deb99141d89293 : FAILURE in 13m 30s (non-voting)\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/4279f715fd4f47a18441ab24ddea7569 : SUCCESS in 18m 36s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/4a53bb98bfae4a93b443a47fdb921d4a : SUCCESS in 1h 20m 08s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"93df3b3e5ce78f2ae393011eac294dc73b2062db","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2026-06-25 08:41:12.000000000","message":"Uploaded patch set 9: Patch Set 8 was rebased.\n\nOutdated Votes:\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":9},{"id":"6518d8a2ef9fab2302ad734dec286f949d2e196d","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-25 08:54:58.000000000","message":"Patch Set 9:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d146b332402a49bfa9f85df19738389d\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/a148058283da4970ae0d937c1a104ba4 : SUCCESS in 6m 12s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/e2768d01f8a84a22ab1c9d13a3559969 : SUCCESS in 8m 45s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/724c866b08f94765ad5a93d58154657e : SUCCESS in 11m 16s (non-voting)","accounts_in_message":[],"_revision_number":9},{"id":"c5fc11836fe7c28abab653aa6eab94d49c3e6865","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-25 11:19:40.000000000","message":"Patch Set 9: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/83ee20acccff4a59be5c19a4037df031\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/c8e61cdc077244ba8aa7e63ff61777aa : SUCCESS in 4m 26s\n- requirements-check https://zuul.opendev.org/t/openstack/build/785522f4ecf546cd81b31fa37446ccd7 : SUCCESS in 3m 39s\n- grenade https://zuul.opendev.org/t/openstack/build/adc91525e5af4adfb400f738f2307334 : SUCCESS in 1h 10m 15s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/ef2f1df3c4c144c6bf6e53dd2a432f3e : SUCCESS in 1h 03m 38s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/5bd20715b6684406bb4654859a66870b : SUCCESS in 1h 22m 26s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/3e3d67c8c7824845987538c0b99c0059 : SUCCESS in 57m 48s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1187847d60234ef892be5f52918a8c12 : SUCCESS in 2m 57s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/5add0ced7186407faba3d8df94e03608 : SUCCESS in 3m 43s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/342cba09b2ff4700bddb43e8639139b7 : SUCCESS in 4m 00s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/87b2c1a4423b4772b0a288e20aae40aa : SUCCESS in 7m 44s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/82c80d7feba348f6a31a7c547118e596 : SUCCESS in 5m 47s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/1d731db3bcfe4b17bf9cc1e1a6d2111b : SUCCESS in 8m 21s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/e659e63df0854ebebe3de8747928790a : SUCCESS in 8m 26s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/797722e6efd64d7bb295b58f8e900346 : SUCCESS in 4m 26s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/fca4d9d2069647109e9b7fea6c6a0622 : SUCCESS in 1h 43m 29s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/78dc65e0331f476da25e28e743e61ffd : SUCCESS in 2h 06m 50s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/e677bc790676491b909903f3fa9c69e8 : SUCCESS in 1h 11m 09s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/dcb43ad87e3741e9aa6327331b5d9de0 : SUCCESS in 2h 35m 35s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/f1c27d6bcd5c471099b4466c3d9e8c01 : SUCCESS in 1h 08m 39s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/ee6cd84ab16047a891a84749a643c682 : SUCCESS in 1h 34m 43s\n- glance-grenade-centralized-cache https://zuul.opendev.org/t/openstack/build/3b9c9193b8cf4161a19e88319ca3a88c : FAILURE in 9m 19s (non-voting)\n- glance-s3-multistore https://zuul.opendev.org/t/openstack/build/42486ed533db45c696f70325349833b5 : SUCCESS in 1h 08m 58s (non-voting)\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/5f011677ce38455a93c04e914dcaeca8 : SUCCESS in 17m 36s\n- glance-rally-task-uwsgi https://zuul.opendev.org/t/openstack/build/04a7d7d593ed451291a83a0b5db6ed61 : SUCCESS in 53m 09s (non-voting)","accounts_in_message":[],"_revision_number":9},{"id":"c9c9be7a1a53c0ce5cedc0f342f11a2f510befc0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2026-06-25 12:46:54.000000000","message":"Uploaded patch set 10.\n\nOutdated Votes:\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":10},{"id":"bd4dd4063fabb84728bbff4b818220c9563ac79a","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-25 13:00:22.000000000","message":"Patch Set 10:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c11548ef5d9b4ee7bdb3550d251d57d7\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/4adb58903d8c491c867871543c3ff656 : SUCCESS in 6m 51s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/24450147de7649b3a06acb2f256d5f31 : FAILURE in 6m 12s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/55a347a7bb594639b6fbba6ebf387a06 : SUCCESS in 11m 33s (non-voting)","accounts_in_message":[],"_revision_number":10},{"id":"9becc888530659745872f6d88be14d164516c1a4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-25 14:51:32.000000000","message":"Patch Set 10: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/98b1847f5f644bff85bcc963513878fe\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/66abd5b99854474cac7b29582c84851e : SUCCESS in 6m 12s\n- requirements-check https://zuul.opendev.org/t/openstack/build/23efd2a7d0da4d4ba2f4735777e2ec5e : SUCCESS in 2m 29s\n- grenade https://zuul.opendev.org/t/openstack/build/3b604c2366294690ad821fc0d3273a09 : SUCCESS in 35m 26s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/8debc217b8ba4e189b46773517759408 : SUCCESS in 1h 02m 56s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/194bb7162cfd494e9bb803a8db4d5574 : SUCCESS in 1h 39m 08s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/fc488638487f45e7a336d2449edd2707 : SUCCESS in 1h 12m 11s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3ff169df45b947f49d23ef047c60f2a3 : SUCCESS in 3m 43s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/1532ffd332f540639e50279239f6d638 : SUCCESS in 5m 01s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/1f451e572a0a4dbf861fbc1fb5f2d589 : SUCCESS in 5m 05s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/41969bc1977b4ddfb013d2544b098ab4 : SUCCESS in 4m 49s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/de8ae27e19c649579c641b60051aa95a : SUCCESS in 3m 43s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/7af9d81390c84354a528db7048c13e68 : SUCCESS in 8m 28s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/11892b707f424af7a0da68ab7065a01c : SUCCESS in 6m 52s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/36b15f4a0e1c4240814b112f737b0dd1 : SUCCESS in 5m 10s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/a98c26dee25e46cca8c438cd55c7afb3 : SUCCESS in 1h 13m 27s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/93b79ab6d89f4f23b54353010182b041 : SUCCESS in 1h 58m 50s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/03f8c3c6a2d24820bec4b2cccf56ee12 : SUCCESS in 1h 09m 59s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/e4c2e98e88a74ff7923e77af225b68c9 : SUCCESS in 1h 24m 49s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/7fbcb7418dc54b7e98b18e63b58364ad : SUCCESS in 1h 20m 35s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/d4765577a98a461b9ac58acf5f931fb0 : SUCCESS in 1h 50m 50s\n- glance-grenade-centralized-cache https://zuul.opendev.org/t/openstack/build/10cef10e6a4344b894eee31ed5ac8f43 : FAILURE in 8m 33s (non-voting)\n- glance-s3-multistore https://zuul.opendev.org/t/openstack/build/e51dcd1f2aa64d1395f12c2f1272cc21 : SUCCESS in 1h 10m 10s (non-voting)\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/52986b6705944f37bdb15793fc79e05f : SUCCESS in 24m 54s\n- glance-rally-task-uwsgi https://zuul.opendev.org/t/openstack/build/28339a4d490b44d189806206471b6c89 : SUCCESS in 51m 25s (non-voting)","accounts_in_message":[],"_revision_number":10},{"id":"dc41a6f9935a91b7f46ba44370ae75f68b48effa","author":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"date":"2026-08-27 13:48:17.000000000","message":"Patch Set 10: Code-Review-1\n\n(6 comments)","accounts_in_message":[],"_revision_number":10},{"id":"25cc37c3c2fca9e3792ec017a9c5572de205da73","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2026-08-27 17:41:44.000000000","message":"Uploaded patch set 11.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":11},{"id":"115fe978df11da98d60e0ab63881f34cb5d5d151","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2026-08-27 17:41:51.000000000","message":"Patch Set 10:\n\n(4 comments)","accounts_in_message":[],"_revision_number":10},{"id":"4d8e512657b120688cf74238e1f4a6affc1d8c30","author":{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},"date":"2026-08-27 17:48:06.000000000","message":"Patch Set 11: Code-Review+1\n\nBuild succeeded (check pipeline).\nhttps://gateway-cloud-softwarefactory.apps.ocp.cloud.ci.centos.org/zuul/t/rdoproject.org/buildset/631e03ac356d4728ab7fc865144b27b4\n\n- s2i-openstack-container-content-provider https://gateway-cloud-softwarefactory.apps.ocp.cloud.ci.centos.org/zuul/t/rdoproject.org/build/4e50ad1394a64ce7a54e275b58af88f9 : SUCCESS in 4m 57s (non-voting)","accounts_in_message":[],"_revision_number":11},{"id":"919d33288a8938ca846d16f18ce38c0134e49de9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-27 20:08:57.000000000","message":"Patch Set 11: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f0167e1fc6eb41e39457c6aae34cfdd5\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/1f2f42eaa0174a26abfd954eb6893aac : SUCCESS in 3m 13s\n- grenade https://zuul.opendev.org/t/openstack/build/1cf90f5ce7184c9cb16b87cd6bf9b6eb : SUCCESS in 34m 35s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/0e662a29d7324f12a3bf1953b9bc6e30 : SUCCESS in 1h 15m 38s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/a0d9770526e342cdbc41714c436ef7aa : SUCCESS in 1h 47m 03s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/9290712db5ca4ec9ab53e58c0c3302fc : SUCCESS in 57m 44s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/77da21ea515e46c0a2dde523fc5bc2b4 : SUCCESS in 5m 54s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/2fd4a74244de4f28b36ce06b2b61da43 : SUCCESS in 4m 31s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/ebb4924cabb7468a82b3e306f5d96084 : SUCCESS in 3m 48s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/f2db53aabab24b0fafb90d2557e1090d : SUCCESS in 4m 50s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b11369e1c32444b6b1370b77633bfff2 : SUCCESS in 7m 05s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/9442be3901cc48c0bcac1a13a948ed8d : SUCCESS in 8m 06s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/752a3babd11b4b549607fa0173f634d1 : SUCCESS in 7m 01s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/08b5f2f8e1f144149dee510a18e559cd : SUCCESS in 6m 03s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/fe741214b9d24a76a383f3bc1ac4a382 : SUCCESS in 1h 42m 49s (non-voting)\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/2531d05b66624b0693d71d9b6f40fcd1 : SUCCESS in 1h 47m 35s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/63ffae5974be486a873063b9f1479f50 : SUCCESS in 2h 19m 45s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/a0c6a8bb83534acda95b48ea482e0e09 : SUCCESS in 1h 03m 20s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/d5ac158db93a4a81973c86c3ae28e228 : SUCCESS in 1h 25m 05s\n- glance-s3-multistore https://zuul.opendev.org/t/openstack/build/001cd4beaaee450c93c5826fedb94d12 : FAILURE in 20m 50s (non-voting)\n- glance-multistore-add-location-do-secure-hash-false https://zuul.opendev.org/t/openstack/build/e1c2c133b7114b749d9e3ea44d4c34a7 : FAILURE in 19m 26s (non-voting)\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/385349f2a9944c96865475a6a0e75091 : SUCCESS in 40m 06s\n- glance-rally-task-uwsgi https://zuul.opendev.org/t/openstack/build/ab1b0c3ff93c41dba003d85dd96630f5 : SUCCESS in 43m 41s (non-voting)","accounts_in_message":[],"_revision_number":11},{"id":"4f91feca6eac95ef3e83bc74ab0977c12094afe6","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-27 20:42:08.000000000","message":"Patch Set 11:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/989ccd54ef3e4d8da07347bf6c2fcafe\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/742ebe5e1cb1449ca7e1dd23afca1a3f : SUCCESS in 8m 14s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/4943bf54e48f41d8b808282d1614a8ba : SUCCESS in 8m 37s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/56fbcc3c5bf740a685d1398e6d01f39a : SUCCESS in 16m 16s (non-voting)","accounts_in_message":[],"_revision_number":11},{"id":"e903c8ef6a3cb10c54560d7254671de437de8ec8","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2026-09-04 07:45:44.000000000","message":"Patch Set 11: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":11}],"current_revision_number":11,"current_revision":"89b54462861077989f89f26af08ca5b4a2218371","revisions":{"29e226fbb2016224d0eda5c49af920bdeed11e4b":{"kind":"REWORK","_number":1,"created":"2026-01-13 04:20:34.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/95/973195/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/95/973195/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/95/973195/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/95/973195/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/95/973195/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/95/973195/1"}}},"commit":{"parents":[{"commit":"62d51017ab4add955b1b6b108a3ba1618f4bbe4c","subject":"Drop unused logic to detect management_url","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/62d51017ab4add955b1b6b108a3ba1618f4bbe4c"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-01-13 03:53:26.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-01-13 04:19:44.000000000","tz":540},"subject":"Use keystoneauth to get token","message":"Use keystoneauth to get token\n\n... to replace own implementation to call Keystone API.\n\nChange-Id: Ic37985dee753b823b1f1c29908f47c4463e3b0fc\nSigned-off-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/29e226fbb2016224d0eda5c49af920bdeed11e4b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/29e226fbb2016224d0eda5c49af920bdeed11e4b"}]},"branch":"refs/heads/master"},"839cff6edaf0ffde0ec64b0d3e909f01b751e4cc":{"kind":"REWORK","_number":2,"created":"2026-01-13 09:10:25.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/95/973195/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/95/973195/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/95/973195/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/95/973195/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/95/973195/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/95/973195/2"}}},"commit":{"parents":[{"commit":"62d51017ab4add955b1b6b108a3ba1618f4bbe4c","subject":"Drop unused logic to detect management_url","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/62d51017ab4add955b1b6b108a3ba1618f4bbe4c"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-01-13 03:53:26.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-01-13 04:30:19.000000000","tz":540},"subject":"Use keystoneauth to get token","message":"Use keystoneauth to get token\n\n... to replace own implementation to call Keystone API.\n\nChange-Id: Ic37985dee753b823b1f1c29908f47c4463e3b0fc\nSigned-off-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/839cff6edaf0ffde0ec64b0d3e909f01b751e4cc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/839cff6edaf0ffde0ec64b0d3e909f01b751e4cc"}]},"branch":"refs/heads/master"},"6b6bce978214f5a34dc1c8c9d2bd19f944cee807":{"kind":"REWORK","_number":3,"created":"2026-01-13 15:03:28.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/95/973195/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/95/973195/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/95/973195/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/95/973195/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/95/973195/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/95/973195/3"}}},"commit":{"parents":[{"commit":"62d51017ab4add955b1b6b108a3ba1618f4bbe4c","subject":"Drop unused logic to detect management_url","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/62d51017ab4add955b1b6b108a3ba1618f4bbe4c"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-01-13 03:53:26.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-01-13 15:03:11.000000000","tz":540},"subject":"Use keystoneauth to get token","message":"Use keystoneauth to get token\n\n... to replace own implementation to call Keystone API.\n\nChange-Id: Ic37985dee753b823b1f1c29908f47c4463e3b0fc\nSigned-off-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/6b6bce978214f5a34dc1c8c9d2bd19f944cee807"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/6b6bce978214f5a34dc1c8c9d2bd19f944cee807"}]},"branch":"refs/heads/master"},"fd5ce57a3267af37f87e35379b3b3826ae810e1b":{"kind":"REWORK","_number":4,"created":"2026-01-13 16:17:43.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/95/973195/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/95/973195/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/95/973195/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/95/973195/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/95/973195/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/95/973195/4"}}},"commit":{"parents":[{"commit":"89b1b7d29c338299e309703caf987cc3369b3a6a","subject":"Remove remaining logic for identity v1/v2 API","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/89b1b7d29c338299e309703caf987cc3369b3a6a"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-01-13 03:53:26.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-01-13 16:16:17.000000000","tz":540},"subject":"Use keystoneauth to get token","message":"Use keystoneauth to get token\n\n... to replace own implementation to call Keystone API.\n\nChange-Id: Ic37985dee753b823b1f1c29908f47c4463e3b0fc\nSigned-off-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/fd5ce57a3267af37f87e35379b3b3826ae810e1b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/fd5ce57a3267af37f87e35379b3b3826ae810e1b"}]},"branch":"refs/heads/master"},"5e228749ace828c36a83a61cc08c9d9e8f739e7a":{"kind":"REWORK","_number":5,"created":"2026-01-13 16:23:19.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/95/973195/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/95/973195/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/95/973195/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/95/973195/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/95/973195/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/95/973195/5"}}},"commit":{"parents":[{"commit":"3ade090be5561f8dcffae43ca4e9622636d4919d","subject":"Remove remaining logic for identity v1/v2 API","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/3ade090be5561f8dcffae43ca4e9622636d4919d"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-01-13 03:53:26.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-01-13 16:23:07.000000000","tz":540},"subject":"Use keystoneauth to get token","message":"Use keystoneauth to get token\n\n... to replace own implementation to call Keystone API.\n\nChange-Id: Ic37985dee753b823b1f1c29908f47c4463e3b0fc\nSigned-off-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/5e228749ace828c36a83a61cc08c9d9e8f739e7a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/5e228749ace828c36a83a61cc08c9d9e8f739e7a"}]},"branch":"refs/heads/master"},"c93e5d210b45ea4898db55e6470e6c1243d7188c":{"kind":"TRIVIAL_REBASE","_number":6,"created":"2026-01-13 16:25:24.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/95/973195/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/95/973195/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/95/973195/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/95/973195/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/95/973195/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/95/973195/6"}}},"commit":{"parents":[{"commit":"ae206dddb3048522841e71f76e6583e01f292898","subject":"Remove remaining logic for identity v1/v2 API","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/ae206dddb3048522841e71f76e6583e01f292898"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-01-13 03:53:26.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-01-13 16:25:14.000000000","tz":540},"subject":"Use keystoneauth to get token","message":"Use keystoneauth to get token\n\n... to replace own implementation to call Keystone API.\n\nChange-Id: Ic37985dee753b823b1f1c29908f47c4463e3b0fc\nSigned-off-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/c93e5d210b45ea4898db55e6470e6c1243d7188c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/c93e5d210b45ea4898db55e6470e6c1243d7188c"}]},"branch":"refs/heads/master"},"85c56388ba47b429725fba82f796c3ac3e164559":{"kind":"REWORK","_number":7,"created":"2026-01-13 17:36:36.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/95/973195/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/95/973195/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/95/973195/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/95/973195/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/95/973195/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/95/973195/7"}}},"commit":{"parents":[{"commit":"22618484ea3feb26b85cae690cf9fbb6744ed470","subject":"Remove remaining logic for identity v1/v2 API","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/22618484ea3feb26b85cae690cf9fbb6744ed470"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-01-13 03:53:26.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-01-13 17:32:45.000000000","tz":540},"subject":"Use keystoneauth to get token","message":"Use keystoneauth to get token\n\n... to replace own implementation to call Keystone API.\n\nThis change technically restores compatibility with identity v2 API,\nbecause generic.Password is theoretically compatible with both v2 and\nv3 API (though v2 API has been untested really for long time).\n\nChange-Id: Ic37985dee753b823b1f1c29908f47c4463e3b0fc\nSigned-off-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/85c56388ba47b429725fba82f796c3ac3e164559"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/85c56388ba47b429725fba82f796c3ac3e164559"}]},"branch":"refs/heads/master"},"8a3685b9c22563f820a5c0a739633994759623d1":{"kind":"REWORK","_number":8,"created":"2026-01-13 23:48:04.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/95/973195/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/95/973195/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/95/973195/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/95/973195/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/95/973195/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/95/973195/8"}}},"commit":{"parents":[{"commit":"3582e28cdb0519765d3a7ce369610fcf0b65c0df","subject":"Remove remaining logic for identity v1/v2 API","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/3582e28cdb0519765d3a7ce369610fcf0b65c0df"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-01-13 03:53:26.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-01-13 23:47:51.000000000","tz":540},"subject":"Use keystoneauth to get token","message":"Use keystoneauth to get token\n\n... to replace own implementation to call Keystone API.\n\nThis change technically restores compatibility with identity v2 API,\nbecause generic.Password is theoretically compatible with both v2 and\nv3 API (though v2 API has been untested really for long time).\n\nChange-Id: Ic37985dee753b823b1f1c29908f47c4463e3b0fc\nSigned-off-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/8a3685b9c22563f820a5c0a739633994759623d1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/8a3685b9c22563f820a5c0a739633994759623d1"}]},"branch":"refs/heads/master"},"fee5cddcfd503f69fa5a0e9d4925c6cf3f4edccb":{"kind":"TRIVIAL_REBASE","_number":9,"created":"2026-06-25 08:41:12.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/95/973195/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/95/973195/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/95/973195/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/95/973195/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/95/973195/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/95/973195/9"}}},"commit":{"parents":[{"commit":"85ccc01ca0a56f6d116f3e9da108438af3e62b2a","subject":"Remove remaining logic for identity v1/v2 API","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/85ccc01ca0a56f6d116f3e9da108438af3e62b2a"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-01-13 03:53:26.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-06-25 08:40:52.000000000","tz":540},"subject":"Use keystoneauth to get token","message":"Use keystoneauth to get token\n\n... to replace own implementation to call Keystone API.\n\nThis change technically restores compatibility with identity v2 API,\nbecause generic.Password is theoretically compatible with both v2 and\nv3 API (though v2 API has been untested really for long time).\n\nChange-Id: Ic37985dee753b823b1f1c29908f47c4463e3b0fc\nSigned-off-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/fee5cddcfd503f69fa5a0e9d4925c6cf3f4edccb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/fee5cddcfd503f69fa5a0e9d4925c6cf3f4edccb"}]},"branch":"refs/heads/master"},"56dde733bc2fa3fe0d46067b93a6462b07858de6":{"kind":"REWORK","_number":10,"created":"2026-06-25 12:46:54.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/95/973195/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/95/973195/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/95/973195/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/95/973195/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/95/973195/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/95/973195/10"}}},"commit":{"parents":[{"commit":"0b7ca4e9b412a9503f97335c45c8460058ddb930","subject":"Remove remaining logic for identity v1/v2 API","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/0b7ca4e9b412a9503f97335c45c8460058ddb930"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-01-13 03:53:26.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-06-25 12:46:40.000000000","tz":540},"subject":"Use keystoneauth to get token","message":"Use keystoneauth to get token\n\n... to replace own implementation to call Keystone API.\n\nThis change technically restores compatibility with identity v2 API,\nbecause generic.Password is theoretically compatible with both v2 and\nv3 API (though v2 API has been untested really for long time).\n\nChange-Id: Ic37985dee753b823b1f1c29908f47c4463e3b0fc\nSigned-off-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/56dde733bc2fa3fe0d46067b93a6462b07858de6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/56dde733bc2fa3fe0d46067b93a6462b07858de6"}]},"branch":"refs/heads/master"},"89b54462861077989f89f26af08ca5b4a2218371":{"kind":"REWORK","_number":11,"created":"2026-08-27 17:41:44.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/95/973195/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/95/973195/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/95/973195/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/95/973195/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/95/973195/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/95/973195/11"}}},"commit":{"parents":[{"commit":"a0291c1477602e6572ae9f9d0aeca9f1e0f1a28d","subject":"Remove remaining logic for identity v1/v2 API","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/a0291c1477602e6572ae9f9d0aeca9f1e0f1a28d"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-01-13 03:53:26.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-08-27 17:41:32.000000000","tz":540},"subject":"Use keystoneauth to get token","message":"Use keystoneauth to get token\n\n... to replace own implementation to call Keystone API.\n\nThis change technically restores compatibility with identity v2 API,\nbecause generic.Password is theoretically compatible with both v2 and\nv3 API (though v2 API has been untested really for long time).\n\nChange-Id: Ic37985dee753b823b1f1c29908f47c4463e3b0fc\nSigned-off-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/89b54462861077989f89f26af08ca5b4a2218371"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/89b54462861077989f89f26af08ca5b4a2218371"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"}},{"label":"Workflow","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{"label:Review-Priority\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
