)]}'
{"id":"openstack%2Fglance~974027","triplet_id":"openstack%2Fglance~master~Ia874d350a9de6b6d24b289bbfc9e6c7c8182fdb5","project":"openstack/glance","branch":"master","topic":"eventlet-removal","attention_set":{},"removed_from_attention_set":{"9303":{"account":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"last_update":"2026-02-12 16:23:19.000000000","reason":"\u003cGERRIT_ACCOUNT_9303\u003e replied on the change","reason_account":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"}},"8122":{"account":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"last_update":"2026-02-12 21:04:12.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ia874d350a9de6b6d24b289bbfc9e6c7c8182fdb5","subject":"Move enabled_backends config option to glance.common.config","status":"MERGED","created":"2026-01-21 07:28:19.000000000","updated":"2026-02-12 23:35:18.000000000","submitted":"2026-02-12 21:04:12.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":5,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"974027-eventlet-removal","meta_rev_id":"fb0752401a69449fc60051771b5540423174642b","_number":974027,"virtual_id_number":974027,"owner":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},{"value":0,"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-02-12 21:04:11.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"recommended":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"all":[{"value":1,"date":"2026-02-12 09:28:40.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},{"value":2,"date":"2026-02-12 15:20:52.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"all":[{"value":0,"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},{"value":1,"date":"2026-02-12 15:20:52.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},{"value":0,"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Top Priority / Holds Gate"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-01-21 07:48:17.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2026-01-21 09:36:34.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-01-21 15:31:38.000000000","updated_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"reviewer":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"state":"REVIEWER"},{"updated":"2026-01-27 11:40:06.000000000","updated_by":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"reviewer":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"state":"REVIEWER"}],"messages":[{"id":"deb9fd47757d2aa7f690fcf5229a4a4c0fd2c6fd","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2026-01-21 07:28:19.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"95ddc8b56a9a7d47f24902cfffb08deb05400e90","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2026-01-21 07:32:26.000000000","message":"Topic set to eventlet-removal","accounts_in_message":[],"_revision_number":1},{"id":"e66a11c45820d60163fae90ac6b160ace0cd48bc","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-21 07:48:17.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0431d9312a594bb380534bd66a82a272\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/71bfe0cf6a3147da9542d270f7187dd0 : SUCCESS in 8m 39s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/f84f363e49fd40e1a73c20c9dd00e150 : SUCCESS in 9m 02s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/8c4f13b865c9481f9b79ac43dc2ef918 : RETRY_LIMIT in 4m 53s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"891f2e4cd13197cd7b8ea135021806def0b6563a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-21 09:36:34.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/721a9b2190e245828357148425a1404a\n\n- grenade https://zuul.opendev.org/t/openstack/build/422bb5ac576949f7b6b28019712f8e00 : SUCCESS in 58m 44s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/95caaf7cf7424b53b8a186d1987b0bd2 : SUCCESS in 33m 14s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/b91a05bee09b466d98b4485b3c2a772f : SUCCESS in 1h 42m 51s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/c06bc5ffc0034030a1a11bd97cf68df0 : SUCCESS in 44m 29s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/145528e47e9946988d0e20f103eb2071 : SUCCESS in 4m 39s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/2dcabd01c2054faf9024360b0c2810b7 : SUCCESS in 3m 04s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/9dee15ee0a9547f2a73e189b185cc6e8 : SUCCESS in 4m 26s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/df04f1c6c9fe48e09026ff63f127d938 : SUCCESS in 3m 26s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/3ddc8065881940c386fb80b6d617b44d : SUCCESS in 6m 01s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/13cd2e700b6b477c92011ee228c8fb70 : SUCCESS in 6m 45s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/c2b8980f16b246158f06be1c540bcd3e : SUCCESS in 1h 04m 43s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/a2f896591d494549bf9b2f2d2d6f6b5f : SUCCESS in 2h 03m 15s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/0a1aec8c9e55422fb4b2fdb84de77718 : SUCCESS in 1h 40m 06s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/604733a992c645f3af4200df309c98a4 : SUCCESS in 1h 40m 24s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/42be054ab8e34ce583dbcfbfaa00c79e : SUCCESS in 55m 52s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/c77de0dd02ef4f7bb84a0991a4a6043c : SUCCESS in 1h 43m 11s\n- glance-grenade-centralized-cache https://zuul.opendev.org/t/openstack/build/b53f84ee018d420889e015c461803ede : FAILURE in 19m 45s (non-voting)\n- glance-s3-multistore https://zuul.opendev.org/t/openstack/build/366c5d0cc23b4ac48fd36dbbeaa3194f : FAILURE in 28m 24s (non-voting)\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/e91692f2f613417499122b21815846dc : SUCCESS in 36m 21s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/176af598d5d0453fa27f3661ee4a1696 : SUCCESS in 1h 17m 56s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"822e540b3ed27775e3203b73e5348336d35b2d6a","author":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"date":"2026-01-27 11:40:06.000000000","message":"Patch Set 1: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"5e911266c6355510c7c58eba24b6023e69eec281","author":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"date":"2026-02-05 17:01:18.000000000","message":"Patch Set 1: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"001f0a06739e8cfc473c7801d3f8dbad36cf0868","author":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"date":"2026-02-09 14:57:56.000000000","message":"Patch Set 1: Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"87e7ca76c228b1a893c657996bedb646acc91390","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-09 14:58:33.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"047c044f1937f6db3842efd5a88936dc4f0073f4","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-09 17:21:45.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/1378184a387d498e8b4b3b7db508a450\n\n- grenade https://zuul.opendev.org/t/openstack/build/ac5824f6363a44c6a82c1badd04db013 : FAILURE in 13m 38s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/e66c1125df3f46d3949c9cd4553cee28 : FAILURE in 6m 43s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/a146169788674d0bbbbbbc51944968b1 : SUCCESS in 1h 06m 43s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/359bbec1206d483990a9811f44c71eea : FAILURE in 1h 07m 52s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b929300b3f964eaa99e673d2553b2d74 : SUCCESS in 4m 43s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/ce391783b0be4a4aa5b0c80d783ca391 : SUCCESS in 4m 49s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/8f38706be6964fa2a5b42249840bf359 : SUCCESS in 7m 49s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/658c868a3610409eb57b5baa137aad30 : SUCCESS in 5m 23s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/091159d42981474c8e2dd287ff23ec0d : SUCCESS in 6m 58s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/7a3051b3ef8941c6b5fea8b30b89bb2a : SUCCESS in 6m 59s\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/93d28d1086bc40c9bfed66fffe7edd6f : SUCCESS in 1h 08m 13s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/4866585d61104659b14ae23334b86bd9 : SUCCESS in 1h 09m 31s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/7eb3455fe70d46f889cc0c8bf8fe7691 : SUCCESS in 59m 07s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/fa1d1de4ede740af95fa021680f88380 : SUCCESS in 1h 59m 19s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/0d0615f1168345e090d08ca57ae19737 : SUCCESS in 24m 14s","accounts_in_message":[],"_revision_number":1},{"id":"b5453bb4e63080a1664b836e9d2c71382dc4f8e3","author":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"date":"2026-02-11 20:07:28.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"f18110783da6bc9b2af2e32ed507bfe62d0bfc3b","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-11 20:26:43.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2eff9ac0ebe54c26ba7a8e24d0206b80\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/d3cfee44098f408f900e016c0d2dbbee : SUCCESS in 9m 38s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/dde36cb1c9df40bca2880a64fe76e8b7 : SUCCESS in 10m 06s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/a2d53d5983a8468c87e4c8575fe5e359 : RETRY_LIMIT in 4m 14s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"036147eec7a7b0b18a1e195468b449383586c753","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-11 22:44:19.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4557bfa452f749b48aab2067c9c929c5\n\n- grenade https://zuul.opendev.org/t/openstack/build/9be730a05c2a444880e5c5e9fa7d17d2 : SUCCESS in 48m 25s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/44c8a769374f4f3191a2b3bd4c0e5f79 : SUCCESS in 1h 04m 31s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/acc53899c3064853bfe52cec66609ca8 : SUCCESS in 1h 35m 12s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/e4b30535b0594d869103570584c7d1fb : SUCCESS in 46m 13s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b4b90564e5554cf4821374bbabccb8b7 : SUCCESS in 4m 42s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/09f001b7746b4cc890a5231b87f80ec5 : SUCCESS in 7m 00s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/42ed6e0a9d5443e3885f9a4062e374ae : SUCCESS in 4m 51s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9ee4f8dd7e3d4e2194ab3807dc9c1b2c : SUCCESS in 3m 02s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/eaa56dcbaba8470ba2197cb0f9eabfca : SUCCESS in 6m 09s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/516d3c554da64f4cbfb33102912e2bf8 : SUCCESS in 6m 14s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/b63e0be29c424c6f9f93a9ace8d01daf : SUCCESS in 1h 12m 01s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/d7d4ea2c537748dea16e45bf22c02cbb : SUCCESS in 1h 34m 47s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/0b99ce00f18849da95a53fee1225357e : SUCCESS in 2h 04m 19s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/efa39524f2554c4aa81ac1cb1b4becbb : SUCCESS in 2h 28m 20s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/5af0a2690478484187b41ea4bb26c557 : SUCCESS in 40m 41s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/1f34bd134ae54e3faf9cf456f596b868 : SUCCESS in 1h 07m 56s\n- glance-grenade-centralized-cache https://zuul.opendev.org/t/openstack/build/d3ac3840718b4ec58fa513a472237830 : FAILURE in 14m 38s (non-voting)\n- glance-s3-multistore https://zuul.opendev.org/t/openstack/build/e4fdf24878a0408387a9515061ec7769 : SUCCESS in 2h 28m 44s (non-voting)\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/7a1c499d1e404e59a9eab8f09e8957fb : SUCCESS in 19m 48s\n- ironic-tempest-bios-ipmi-direct https://zuul.opendev.org/t/openstack/build/b4675323f9174de0a5d8f992da6d8a9c : SUCCESS in 36m 50s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"e3f1a43e2886183c52f3ab89e9d2870dc7e47041","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-11 22:44:44.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"52b77bfa974eabc4bcdb71e4e4c0c930724bffa0","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-12 01:02:28.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/0a0b6309544e4da9a40e42818348a3c5\n\n- grenade https://zuul.opendev.org/t/openstack/build/e62e7d4ef7d149808e97a4ce77ca1ffd : SUCCESS in 32m 53s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/2ff47286735d4e6c81184b84013b907c : SUCCESS in 30m 21s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/bfa98b02d021497e93c7398b22df43ee : FAILURE in 1h 23m 39s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/9dff3bf958d44ff1bfad756d1d117278 : SUCCESS in 1h 02m 51s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7843e49b84e3460d9f4abf1f7b2e846f : SUCCESS in 4m 34s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/b5a6c102abad49be96536c668d23356d : SUCCESS in 4m 11s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/93a3e2e22659459798e7bcea127ede93 : SUCCESS in 4m 44s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/558bc2ee5eca4bf8987f4503d5aaa1aa : SUCCESS in 5m 37s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/814cbc7e5898423c95acbc918c91a84e : SUCCESS in 6m 11s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/e52383f29895433f8d07ef28eb3bb138 : SUCCESS in 6m 42s\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/773501cd909c4373a4e22f0ae630473c : SUCCESS in 2h 08m 20s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/311fa553fafe41ce995d1b0871aa72b0 : SUCCESS in 2h 13m 00s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/5ff818b9cee0463bba5010b057d1a884 : SUCCESS in 30m 39s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/9cbac22f8c8744dfbe8476983abc6007 : SUCCESS in 1h 23m 41s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/4f94d55d402148729f5d9ea6e545cad5 : SUCCESS in 33m 10s","accounts_in_message":[],"_revision_number":1},{"id":"640af0531ded35e2b68f6f254396719b7a0de038","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2026-02-12 05:10:20.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"fdf74ce9f70a3ce924dc4164761f87159773fff2","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-12 05:44:19.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9e5d78ac2d1349f6a4341c3083dc61bc\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/3df731223e4e4a9e9cc1b97eec9dc47e : SUCCESS in 11m 42s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/b03ef8dca97a42babf2ee1dbe8f747fa : SUCCESS in 11m 53s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/ea1415294c084aaaa86daeefd8734769 : RETRY_LIMIT in 5m 07s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"cfcb542c1d77dadb0001240f2b2d805ea8506e64","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-12 07:39:17.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/c6eed54967c34cfbaeceae07a7cf2f74\n\n- grenade https://zuul.opendev.org/t/openstack/build/34080d5570634a1296ee4cc647ae8aed : SUCCESS in 29m 53s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/da3c90f3b4cb440caa350b93d3f232cd : SUCCESS in 28m 34s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/7f484c26512c4e60b63918e7bc81b521 : FAILURE in 1h 24m 56s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/14189d2b665d466d811b09d3e193e567 : SUCCESS in 1h 11m 26s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4a248e9d6ee7422184ac0f2bd7bda36c : SUCCESS in 4m 34s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/24b2a4ac1d3248f8b7da8f4c053d42c7 : SUCCESS in 5m 26s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/f6e525b095f748eda5ce9e8f97cd0c53 : SUCCESS in 5m 11s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/402498011ec04976be5e950d2386cc4c : SUCCESS in 6m 12s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/32bc8219db0846feb24e6478b3b134ed : SUCCESS in 10m 34s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/12c5a1df677c4eca8432dbb091467fde : SUCCESS in 4m 20s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/c02c818db0db44df9a6240d99e5225ca : SUCCESS in 1h 46m 23s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/59550b68a9f74e04b6dfe7ed967960c4 : SUCCESS in 1h 59m 52s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/0a0b76e11b0e48c8996f3e200ebe6c7f : SUCCESS in 1h 56m 06s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/0984b18e11d34480b612ac9cd8c6b716 : SUCCESS in 2h 27m 12s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/7e812357d14640ccb9c4ed18da460e2d : SUCCESS in 1h 04m 31s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/7952a7d55b8948ffb01d3c2bfbd26b2a : SUCCESS in 1h 15m 12s\n- glance-grenade-centralized-cache https://zuul.opendev.org/t/openstack/build/e8b03336ad26403598edc3690106316b : FAILURE in 22m 17s (non-voting)\n- glance-s3-multistore https://zuul.opendev.org/t/openstack/build/eee936dafe374d8a9d502ba87658b7d2 : SUCCESS in 1h 54m 27s (non-voting)\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/c7d418edf85c4b18be1c98796d4168fd : SUCCESS in 36m 38s\n- ironic-tempest-bios-ipmi-direct https://zuul.opendev.org/t/openstack/build/6867fdb36ddb4940a5a70a031de30c93 : SUCCESS in 1h 14m 21s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"cd2d804d8df12eb02878f502c911971ce5e64f12","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2026-02-12 07:41:21.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"bc51b1405ce12b94ef8a82fdb5a006fa6a326e0c","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-12 09:17:39.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/00072199497c4e239f6c4e51ef8eb81d\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/0687e3704a17437a9d6267e18f2ba64a : SUCCESS in 11m 05s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/792a828a8c964504af0c1a1e8fcf1a53 : SUCCESS in 5m 56s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/0f8baddd62e449389de215a997ddab88 : RETRY_LIMIT in 4m 09s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"5809c3b103cd8a6c0fe85d79ac82a905cca10d77","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"date":"2026-02-12 09:28:40.000000000","message":"Patch Set 2: Patch Set 1 was rebased\n\nCopied Votes:\n* Code-Review+1, Code-Review+2 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n\nOutdated Votes:\n* Verified-1\n* Workflow+1\n","accounts_in_message":[],"_revision_number":2},{"id":"4ad7bf074f7a92de7101c2767223bc60fa09aa84","author":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"date":"2026-02-12 09:28:43.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"eb5a3446cf15c0c62e25f4972a781bdfb1219b25","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-12 10:44:12.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ebafba679c9f4119a82bd0ca04f4408a\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/a462c9e9b68f47568dd1b477f43623c4 : SUCCESS in 9m 37s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/52f08bb21fd043f8a290ac09f5545b09 : SUCCESS in 10m 19s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/1c533c73d3f54fd8b85c3c0663043a80 : RETRY_LIMIT in 3m 53s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"4bf14650759bbe77981e7779fe148116efdffe96","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-12 12:23:00.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/ac35f805d4824410b83bea47d34f15f0\n\n- grenade https://zuul.opendev.org/t/openstack/build/4d5f320295b2471ba6265ec2f521cd63 : SUCCESS in 57m 01s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/059eac9687134e89b34851852817fa04 : SUCCESS in 31m 14s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/3ef32c74053c459ea3706347d63f59d5 : SUCCESS in 1h 14m 52s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/a8b103d0c2e2430f82baea3204241f7d : SUCCESS in 48m 50s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b500ffd4f8974695a1c72bcfc2a952b9 : SUCCESS in 4m 31s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/2b5f77828eab44c9b209b5a72b470295 : SUCCESS in 3m 08s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/32120eb60e7f4a539d3f6ac1a8db28ac : SUCCESS in 5m 00s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/226559f2424b42bfbd4c61546c75ca78 : SUCCESS in 4m 17s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/87e83904aa9641d28e9a0bc1c64d1950 : SUCCESS in 5m 39s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/1afb7920149a4f208e369d970f742012 : SUCCESS in 7m 07s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/3bdef43289db4a6d8a35cbff436bc5ac : SUCCESS in 1h 51m 30s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/d4fdb5bb9551429f977460f89865e81f : SUCCESS in 2h 10m 55s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/3c4ee8f92b854b238c028d27591bea12 : SUCCESS in 1h 33m 38s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/7b5b29271740437d8643b085938d6266 : SUCCESS in 2h 44m 06s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/f3999642a00b4017a48f30729d3c87aa : SUCCESS in 35m 42s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/8cc4562e7f894d5ebe037d127873f662 : SUCCESS in 1h 22m 46s\n- glance-grenade-centralized-cache https://zuul.opendev.org/t/openstack/build/7e5899e65da141db9d41457d327d0006 : FAILURE in 13m 37s (non-voting)\n- glance-s3-multistore https://zuul.opendev.org/t/openstack/build/0e670e8a2df34bfd97a629443d07bae8 : SUCCESS in 1h 47m 27s (non-voting)\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/c9051b9963504a2497a25a399f9dd2bd : FAILURE in 23m 33s\n- ironic-tempest-bios-ipmi-direct https://zuul.opendev.org/t/openstack/build/279e08a0c47e4d188c8b258f65b4e621 : SUCCESS in 44m 47s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"43c003daa82738851a59efbdcae3ccfe75dc315c","author":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"date":"2026-02-12 13:29:27.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"26c774d725d7379fd3943bbf124367d50d28a425","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-12 14:35:32.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/dbddc68bcc3a4af993098bcb327aa546\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/e5ae4c31f6294752baa415e8e09ce57d : SUCCESS in 8m 33s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/32a80da878aa4d8ea5d98cc4100c8978 : SUCCESS in 6m 49s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/4c69757d3c7142f4995795349294388a : RETRY_LIMIT in 4m 30s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"7af4479907c834ae89234373b41ce027cba2d29f","author":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"date":"2026-02-12 15:20:52.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"ec0fc2c582afdfb73b633b1f07bbaf0ed8717bb6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-12 15:52:40.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/c595c86084ac45c1916dbf4116141795\n\n- grenade https://zuul.opendev.org/t/openstack/build/fd9c8c6d5794468da05eb7b5756a1357 : SUCCESS in 28m 57s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/b20fc7a7c86b4af487df85572144575f : SUCCESS in 1h 04m 53s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/93dfc68eb6d54753a6eb958be99e0b2e : SUCCESS in 1h 39m 55s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/e0f3227fffb542d0a094d3afbfa9adb5 : FAILURE in 1h 07m 19s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1bc68a10a7624009bf7a5a26ac53213b : SUCCESS in 6m 28s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/e1f6d6d3ee624926ad0d7b784f890c5a : SUCCESS in 7m 45s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/6e05f6a3cdec42f09c504c94d0edfaca : SUCCESS in 3m 25s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2f446ba25b61457bb5350920a7d420f5 : SUCCESS in 3m 10s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/d71c089662c64284bbba54307bfed95e : SUCCESS in 10m 05s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/6f72483caa034b77ad20d10945697c1b : SUCCESS in 6m 33s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/4404aad32eef47f587cbdafdcb07648e : SUCCESS in 1h 44m 48s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/a33f3e4d8bf34e4c853e28772d1da412 : SUCCESS in 2h 05m 54s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/e5d9554012714d2b97f31a98f4e7dad3 : FAILURE in 1h 55m 43s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/8f44b69cc7da4b2fb8e60ffaa74645f9 : SUCCESS in 2h 17m 55s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/f528cff0aa2b4053be6cf6c9a9a82b22 : SUCCESS in 56m 42s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/169883d704404968aa317e4e9e411ddd : SUCCESS in 2h 00m 18s\n- glance-grenade-centralized-cache https://zuul.opendev.org/t/openstack/build/acb1083436874ce4adfedd5d22a55fed : FAILURE in 13m 03s (non-voting)\n- glance-s3-multistore https://zuul.opendev.org/t/openstack/build/38832b3ab6854b50883553606707cd44 : SUCCESS in 2h 02m 36s (non-voting)\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/55c00eb7327c4a39939389eca21f2520 : SUCCESS in 33m 37s\n- ironic-tempest-bios-ipmi-direct https://zuul.opendev.org/t/openstack/build/e0d2cc6670df43fab4a4068a1dcfd3fb : SUCCESS in 49m 24s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"97cea44da9d7fc113a2d44796f389eea0b2327e9","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2026-02-12 16:23:19.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"f8b4c5179e1e8634f14256ef8df11941348598f2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-12 18:33:59.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c1dc5693f1d4408b839f1c72714407a4\n\n- grenade https://zuul.opendev.org/t/openstack/build/0ff4561347e24bcd8373f64f2fbe4fa9 : SUCCESS in 1h 01m 49s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/d828146b4f3e4972ab2da0aa1d9ee5fd : SUCCESS in 1h 02m 27s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/ef1187bf718945609c94f875c3f13b1a : SUCCESS in 1h 36m 14s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/60cf37293b79423ead6c87b31473ef32 : SUCCESS in 1h 01m 55s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8208cd6320cb4d0992fd64b7a103c892 : SUCCESS in 5m 58s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/ff1cc50f61b54d68b898d663a3a82c14 : SUCCESS in 3m 10s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/110adb213e354c0d98ad09d52015f2b9 : SUCCESS in 5m 41s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e990d9ec2e4f4f12b11408d345635c44 : SUCCESS in 7m 49s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/6e3f701ab8fb42a58c7e5898609b2201 : SUCCESS in 6m 34s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/2522d164cd664e09b2998c147c8136a8 : SUCCESS in 7m 09s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/f87e9f19f9914a018f6c87e0ea0968e3 : SUCCESS in 1h 07m 54s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/14a35290445047038af6ccfa37774bd5 : SUCCESS in 1h 53m 47s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/aec5a7ca76994b75a06a284a39cd0ad8 : SUCCESS in 1h 12m 44s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/dc7c5b0aaa0343918528e2999557e710 : SUCCESS in 1h 52m 40s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/768fd8fd6cb1465ea9a5ca1f02b2d1b3 : SUCCESS in 35m 21s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/f8b4cd5c1fde4681a1c9d04ca484ffd9 : SUCCESS in 2h 05m 10s\n- glance-grenade-centralized-cache https://zuul.opendev.org/t/openstack/build/e5806cd9149e4072b8c527b85fea9220 : FAILURE in 11m 53s (non-voting)\n- glance-s3-multistore https://zuul.opendev.org/t/openstack/build/af17c6ab857c406e9f93b87ff223ce0c : SUCCESS in 1h 35m 24s (non-voting)\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/ce3fe3efbe1d464489d0278f908a700e : SUCCESS in 36m 49s\n- ironic-tempest-bios-ipmi-direct https://zuul.opendev.org/t/openstack/build/b9143aba5aad42d5b87156079e354d71 : SUCCESS in 1h 24m 18s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"c8b9826df80a11cae7c97e14b767bfbeb9ae8024","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-12 18:35:00.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"a2b5d605263eb5cbf37a6fba704aa857a1533591","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-12 21:04:11.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/aedaf3da98b54f4ea27470ef772a3d8c\n\n- grenade https://zuul.opendev.org/t/openstack/build/3a896b2a405d4d2c86de86530d865847 : SUCCESS in 59m 06s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/d55c6a83961640d5863dc15a29b30236 : SUCCESS in 57m 34s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/6187aed958924b04ba51874feaa5c0f7 : SUCCESS in 1h 39m 51s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/d5b99871df8742768df7dd6a6f7a3d00 : SUCCESS in 1h 00m 36s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3430ac83c3c341b4a24cfc9c9407b3b9 : SUCCESS in 5m 10s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/d79a8608b15249c3acdd114d8d4fbc82 : SUCCESS in 5m 30s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/8e3b56019edc45d88307a4aa6eba7d44 : SUCCESS in 5m 42s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0da3392cf3a44d859179f576205a8e19 : SUCCESS in 6m 59s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/12a4fdfbee624c9280d31299d5b4300b : SUCCESS in 7m 03s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/c0b0b7ee2b454dccb2b861f41be1c7c7 : SUCCESS in 6m 52s\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/8630d3162d304667aa66096d5aca5046 : SUCCESS in 2h 20m 27s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/06b038b6b3104ddf8f1814926a42fa85 : SUCCESS in 2h 02m 57s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/aa50b090e0e84e808131e7b03fb313bb : SUCCESS in 55m 45s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/3389159391624742b9507b370863318b : SUCCESS in 1h 04m 26s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/24d83934e044459682442dc2b23e972e : SUCCESS in 34m 20s","accounts_in_message":[],"_revision_number":2},{"id":"f971c6624da4c5366d282c2895fdb87c5d5d5172","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-12 21:04:12.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"176c10853ed3f86fa56b6d510bcca0e6f8bd0c17","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-12 21:05:32.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/cbc083e557de41e5be631882dd4356a6\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d78190cd84b14443a47cb084d6306afa : SUCCESS in 1m 06s","accounts_in_message":[],"_revision_number":2},{"id":"fb0752401a69449fc60051771b5540423174642b","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-12 23:35:18.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/eaa19adc7f57469babbc14355f8aae79\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/d80c28a54ffa4cd795d218698ba9997f : SUCCESS in 7m 34s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/b78f81ddfc274413ac419bc4ef952d7c : SUCCESS in 11m 59s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/38789f20f2a54664acdf39feef0efdcd : RETRY_LIMIT in 3m 41s (non-voting)","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"6afc6695d13c1f7f77a346f9f2059189612d69dd","revisions":{"4591a3b90fc63fcd45cc6160a339c73f1f471c06":{"kind":"REWORK","_number":1,"created":"2026-01-21 07:28:19.000000000","uploader":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"ref":"refs/changes/27/974027/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/27/974027/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/27/974027/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/27/974027/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/27/974027/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/27/974027/1"}}},"commit":{"parents":[{"commit":"9f4165f7738d48a5081261d1991d4821d1c33eb0","subject":"Merge \"Simplify import of sqlalchemy\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/9f4165f7738d48a5081261d1991d4821d1c33eb0"}]}],"author":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2026-01-21 07:06:30.000000000","tz":0},"committer":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2026-01-21 07:24:21.000000000","tz":0},"subject":"Move enabled_backends config option to glance.common.config","message":"Move enabled_backends config option to glance.common.config\n\nMove store_opts (containing enabled_backends) from glance.common.wsgi\nto glance.common.config. This is a preparatory change for eventlet\nremoval since enabled_backends is not WSGI-specific and is used by\nmultiple services.\n\nUpdates import_opt calls in wsgi_app.py, scrubber.py, and\ncache_prefetcher.py to reference the new location.\n\nChange-Id: Ia874d350a9de6b6d24b289bbfc9e6c7c8182fdb5\nSigned-off-by: Abhishek Kekane \u003cakekane@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/4591a3b90fc63fcd45cc6160a339c73f1f471c06"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/4591a3b90fc63fcd45cc6160a339c73f1f471c06"}]},"branch":"refs/heads/master"},"6afc6695d13c1f7f77a346f9f2059189612d69dd":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2026-02-12 09:28:40.000000000","uploader":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"ref":"refs/changes/27/974027/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/27/974027/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/27/974027/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/27/974027/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/27/974027/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/27/974027/2"}}},"commit":{"parents":[{"commit":"e2b0fd70d2468b9fb311ae83ee6f3ecd521e4f78","subject":"Remove CONF.workers check from simple backend","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/e2b0fd70d2468b9fb311ae83ee6f3ecd521e4f78"}]}],"author":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2026-01-21 07:06:30.000000000","tz":0},"committer":{"name":"Cyril Roelandt","email":"cyril@redhat.com","date":"2026-02-12 09:28:40.000000000","tz":0},"subject":"Move enabled_backends config option to glance.common.config","message":"Move enabled_backends config option to glance.common.config\n\nMove store_opts (containing enabled_backends) from glance.common.wsgi\nto glance.common.config. This is a preparatory change for eventlet\nremoval since enabled_backends is not WSGI-specific and is used by\nmultiple services.\n\nUpdates import_opt calls in wsgi_app.py, scrubber.py, and\ncache_prefetcher.py to reference the new location.\n\nChange-Id: Ia874d350a9de6b6d24b289bbfc9e6c7c8182fdb5\nSigned-off-by: Abhishek Kekane \u003cakekane@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/6afc6695d13c1f7f77a346f9f2059189612d69dd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/6afc6695d13c1f7f77a346f9f2059189612d69dd"}]},"branch":"refs/heads/master","description":"Rebase"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
