)]}'
{"id":"openstack%2Fglance~974684","triplet_id":"openstack%2Fglance~master~I41ee38b146cfaf6713d392b832c9f48390c90efc","project":"openstack/glance","branch":"master","attention_set":{},"removed_from_attention_set":{"9303":{"account":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"last_update":"2026-01-28 22:29:53.000000000","reason":"Change was submitted"},"8122":{"account":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"last_update":"2026-01-28 20:24:32.000000000","reason":"\u003cGERRIT_ACCOUNT_8122\u003e replied on the change","reason_account":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"}},"37598":{"account":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"last_update":"2026-01-28 22:29:53.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I41ee38b146cfaf6713d392b832c9f48390c90efc","subject":"Remove outdated comment in setup.py","status":"MERGED","created":"2026-01-26 09:29:16.000000000","updated":"2026-01-29 00:12:41.000000000","submitted":"2026-01-28 22:29:53.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"974684","meta_rev_id":"38720a204665c84434336ffb56b1dd7a63f2ccab","_number":974684,"virtual_id_number":974684,"owner":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-01-28 22:29:53.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"all":[{"value":2,"date":"2026-01-28 20:24:32.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"all":[{"value":1,"date":"2026-01-28 20:24:32.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Top Priority / Holds Gate"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"}],"reviewers":{"REVIEWER":[{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-01-26 09:29:55.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"state":"REVIEWER"},{"updated":"2026-01-26 09:48:05.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2026-01-26 11:46:51.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-01-28 14:09:54.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"state":"REVIEWER"}],"messages":[{"id":"d1267e24ea39ef5a2f6c4ca12942dde6cd60ea5c","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-01-26 09:29:16.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"0101dab7fedea7d33bf80b47d509cea37d9cddd1","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-01-26 09:29:50.000000000","message":"Patch Set 2: Published edit on patch set 1.","accounts_in_message":[],"_revision_number":2},{"id":"1637ec77db651fd551607a0df638862af36d9825","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-01-26 09:29:55.000000000","message":"Patch Set 2:\n\nThis change is ready for review.","accounts_in_message":[],"_revision_number":2},{"id":"40dfec910fde66d149a2cc3a5e9bd35ec815eb65","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-26 09:48:05.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/eabe87e3dbdf4268979e139a116a5100\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/d2e73a2e2d3f498cb518cb6b03eecf9b : SUCCESS in 8m 34s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/09d3485f4f4442f4b56e29cf8da9c65f : SUCCESS in 9m 12s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/ee6345ee3da14c739ba19fca3d1a82d7 : RETRY_LIMIT in 4m 53s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"21329442809c4cdff09a206cd1d2828d2b7ef86f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-26 11:46:51.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/2e23856ceea545479932d5c21829af57\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/9a6b85b6139344398eda60a4154c1449 : SUCCESS in 1m 35s\n- grenade https://zuul.opendev.org/t/openstack/build/756d9657ac964cfb8ebe7bd3447c31c3 : SUCCESS in 29m 45s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/67f80e9c9a8744948eef4056e3d5c0a8 : SUCCESS in 47m 19s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/93dd507eddce4ee2bf58738bd3c2cb0d : SUCCESS in 1h 37m 47s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/a55cfc979221425e86cace5a523f696b : FAILURE in 55m 35s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e57d4f1cd97744bf896818a316d3d0b7 : SUCCESS in 2m 36s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/a9b31b0cf42c45828fc9370edf15277f : SUCCESS in 3m 14s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/1ce30bb1bbed46a9956d8b9cd0968967 : SUCCESS in 5m 02s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0d78dec7ae69494086fd9de085087372 : SUCCESS in 5m 14s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/1b4de5e2743b431cbfb4a484d69e3eae : SUCCESS in 7m 27s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/8550c89b489f440db23dd74c36c50cff : SUCCESS in 6m 31s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/49030bf357404c6f88f475e94fbb066a : SUCCESS in 1h 56m 54s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/c8acd27f0b3a4c15966e3d544316d1f9 : SUCCESS in 2h 07m 54s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/5cee644ed78840d0bf34597a40dcf014 : SUCCESS in 2h 01m 25s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/927243d018594b72999181488d0cff31 : SUCCESS in 2h 15m 14s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/d67b3b2e95944e5e96a9fff96a119e29 : SUCCESS in 1h 00m 54s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/b40a7c1afb564ec98114ad6230b73c43 : SUCCESS in 1h 44m 38s\n- glance-grenade-centralized-cache https://zuul.opendev.org/t/openstack/build/9f78396a5fa841528648b661b282594a : FAILURE in 12m 08s (non-voting)\n- glance-s3-multistore https://zuul.opendev.org/t/openstack/build/c1636507451c41f0a5631a0cb9b1a833 : FAILURE in 12m 42s (non-voting)\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/1d0d052f000a4ffdbdde1757ad218fa3 : SUCCESS in 39m 22s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/3fe63557102b4743b095206c5c4ee0f9 : SUCCESS in 1h 16m 27s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"b0dfd4edde761fefbbd370525548bb64647618c3","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-01-26 11:53:02.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"e2a21a26f9e01984c6e17d92892ee47f3704cc18","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-26 12:40:35.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8e5a8fa4b1d3467c8676503c743db16f\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/2f092be219c14a139b366008eadb98de : SUCCESS in 8m 32s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/b8964870314a481d9411d5256555cfa8 : SUCCESS in 10m 09s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/999440b3becd482e8f4172a18df897b2 : RETRY_LIMIT in 4m 56s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"562c3709122c15548b8d2a4fd4af3eb15397e562","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-26 14:26:45.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b12d25d09f524bb4bf4e8b6582d89c99\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/7a9baa83e937481fbccbd520bdf1e3a8 : SUCCESS in 3m 08s\n- grenade https://zuul.opendev.org/t/openstack/build/8d92cdc8fdeb487dbbe951f773e6d589 : SUCCESS in 32m 00s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/a287f0aa13b34e52a38750b29dca3dea : SUCCESS in 29m 57s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/f78e6c63a27a4a71b66d1b4a3aa679f3 : SUCCESS in 1h 20m 20s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/14b9cb38d53e4752b36c826c374ed781 : SUCCESS in 59m 07s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b7b686783c2648e081d661ea1aa8c270 : SUCCESS in 4m 40s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/341ac7303c8c451490be50390a56b9a3 : SUCCESS in 5m 06s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/35bf256bdae04e57afc018fdc7eb7e8e : SUCCESS in 5m 36s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/403bd9d8d40d415c97085494171a0e19 : SUCCESS in 5m 37s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/d2b40d1fa1d245b690fb40c299f96df5 : SUCCESS in 5m 16s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/f7a2eadb5fe149e7879b7e51b04d802d : SUCCESS in 9m 18s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/0062e4de64eb458199f3e3bdb1ed07e8 : SUCCESS in 1h 35m 01s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/9bff481f18124f2c94c1336a625cd4b7 : SUCCESS in 1h 35m 22s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/74707599d2504532aa9fa890290ca41b : SUCCESS in 2h 01m 14s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/cad112cec15b46239ff3734a8e5ef2cd : SUCCESS in 2h 28m 12s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/86adc7c8306946af9ea248df09347945 : SUCCESS in 1h 01m 44s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/152ba170ae624dccaa93c3c9338a7094 : SUCCESS in 2h 21m 33s\n- glance-grenade-centralized-cache https://zuul.opendev.org/t/openstack/build/e2e3fa13563e4b97b110a99d08759f7e : FAILURE in 21m 52s (non-voting)\n- glance-s3-multistore https://zuul.opendev.org/t/openstack/build/acd84d00db244278ae488fbc54dbf2a0 : FAILURE in 29m 15s (non-voting)\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/21f44a2bbc96455ca011fc87e3155148 : SUCCESS in 18m 38s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/70f562edd2054a0a98fc53949dd8f8c3 : SUCCESS in 56m 36s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"89ec5bde9db2c9a7f3d22df9ff85569562f4d6a5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-01-28 14:09:44.000000000","message":"Patch Set 3: Commit message was updated.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":3},{"id":"03fd1d69df049ec3db8ddbfea25520fa23e491e6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-28 16:29:10.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/adb8cc71d92f4c3a97042a86ea4e2620\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/a41db28109ef4537b1cf511ef2e0871d : SUCCESS in 2m 22s\n- grenade https://zuul.opendev.org/t/openstack/build/c86b8dd468194b4a9687c6056e9ab245 : SUCCESS in 54m 41s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/1898a197b1c7402b98618bfa955c43e6 : SUCCESS in 1h 01m 40s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/541d648d78294a3d8eb06ab2f24e1744 : SUCCESS in 1h 35m 28s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/46161bec6e02440a9c2196b0b5abcf33 : SUCCESS in 57m 04s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fb87b2e9d8a944908af62d487fe0533f : SUCCESS in 4m 57s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/5105fc227a834afd8ad7c90c0948c995 : SUCCESS in 4m 35s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/d81393ef01c040a1b2ae860a15f605e3 : SUCCESS in 5m 30s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f216f4e09aac4655908c65f0f48a033c : SUCCESS in 7m 26s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/5fcd7b62180c4cfcad8195cd3c25f74f : SUCCESS in 5m 55s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/2881a1d4e97d409b98b2029a13ceceed : SUCCESS in 8m 06s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/5651ca8899c94a93a0fb943979e7edaf : SUCCESS in 1h 54m 15s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/33fa9c5ad6ec4fd8836e433bbc2f2e53 : SUCCESS in 2h 01m 06s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/e6f2b99108724e1ba4a97b51856bd532 : SUCCESS in 1h 10m 45s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/6f231dcd1bee4b07b51265572a7a96d2 : SUCCESS in 2h 05m 14s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/7a4381c22e07428ab83411caf5942488 : SUCCESS in 57m 11s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/5c81e3c16751461588ebd1637aa90e74 : SUCCESS in 2h 12m 09s\n- glance-grenade-centralized-cache https://zuul.opendev.org/t/openstack/build/f6ba7a1872aa441ab044e9e871f38ba9 : FAILURE in 20m 05s (non-voting)\n- glance-s3-multistore https://zuul.opendev.org/t/openstack/build/876837105406413b8703cf933a81a285 : FAILURE in 11m 42s (non-voting)\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/2162bcebaa214429b039a7a6dfb29c39 : SUCCESS in 48m 17s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/9fe9fdaebcb9403b847110a4f4801c56 : SUCCESS in 1h 25m 33s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"60b17988c73bc6045f3a69604276a13c98216c2d","author":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"date":"2026-01-28 20:24:32.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"f8a08b8d287396bcb04a353baba92c269a4e71ef","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-28 20:25:07.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"ea63b0ce306e11029e3d48152bf6b0a57cc9b2f6","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-28 22:29:53.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/cc30580f42004e7ca7de0fcaea9fcd25\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/27b4363dba414004a5d5c0eeb8cc3ce5 : SUCCESS in 2m 53s\n- grenade https://zuul.opendev.org/t/openstack/build/55ee6db1bb1c4eaaa8b359f3f9a231cd : SUCCESS in 1h 01m 17s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/0b8b4523c9f74d53b85a6bdf7c3dc92e : SUCCESS in 49m 20s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/7ba361324e854c32b73e200fe08a7d6e : SUCCESS in 1h 36m 35s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/ae18be12af0545e2b34cb0a78c5312b9 : SUCCESS in 36m 35s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b88dbdf5499c4e48827c498c479c0847 : SUCCESS in 4m 40s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/859da4b7a6044b64a2e01f4c59093016 : SUCCESS in 5m 33s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/5560a7d57856493c806c31e3d65dbbd4 : SUCCESS in 5m 24s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3c1dac0b8f854067a5eb5f71360f25a8 : SUCCESS in 5m 19s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/016c009b0fff4a809ecb2ecac3fb3868 : SUCCESS in 6m 21s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/2f8d8fd41050460db88fe795b2100544 : SUCCESS in 5m 33s\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/7f8c1d272dbf4522ba0f6862278db46d : SUCCESS in 1h 59m 43s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/9b2964e597b64388b50108cd5cac69cf : SUCCESS in 1h 31m 50s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/69b65b48b0414fc5b9e07aff9874f90c : SUCCESS in 1h 01m 53s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/2e5fec1b950e4807926a2a5d6ca635af : SUCCESS in 1h 32m 37s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/32cee4a3467b4f7aa2ade3cdb81267a5 : SUCCESS in 23m 51s","accounts_in_message":[],"_revision_number":3},{"id":"1b3893395e6963bb1b6ac43cddbd3fbbcf304708","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-28 22:29:53.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3},{"id":"8e38ba96ae29fec3bfa6e897090ecca781258cf0","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-28 22:31:24.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/be12a5a6e4c443cdb0cde0bc7be42a92\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e459ebfcb1bd49d292ce638f8869ba8c : SUCCESS in 49s","accounts_in_message":[],"_revision_number":3},{"id":"38720a204665c84434336ffb56b1dd7a63f2ccab","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-29 00:12:41.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/20d7fe0d80fc4543b09c24e380ca36f6\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/32589ce1fe3c4ab68e6367f0dedf082c : SUCCESS in 9m 50s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/a601bac75b2c4f40820ff8c9dbd055f2 : SUCCESS in 10m 01s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/1c649cbaaf904679a32b5c30cfbcc013 : RETRY_LIMIT in 4m 21s (non-voting)","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"601fcfcfb74a7e4b56b9fda03a1b5de1ecf43c0e","revisions":{"a1f2ae6e9c2b747ef7ef11d2711a11416e4e87ea":{"kind":"REWORK","_number":1,"created":"2026-01-26 09:29:16.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/84/974684/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/84/974684/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/84/974684/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/84/974684/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/84/974684/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/84/974684/1"}}},"commit":{"parents":[{"commit":"9f4165f7738d48a5081261d1991d4821d1c33eb0","subject":"Merge \"Simplify import of sqlalchemy\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/9f4165f7738d48a5081261d1991d4821d1c33eb0"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-01-26 09:29:16.000000000","tz":0},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-01-26 09:29:16.000000000","tz":0},"subject":"Remove outdated comment in setup.py","message":"Remove outdated comment in setup.py\n\nChange-Id: I41ee38b146cfaf6713d392b832c9f48390c90efc\nSigned-off-by: Ivan Anfimov \u003clazekteam@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/a1f2ae6e9c2b747ef7ef11d2711a11416e4e87ea"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/a1f2ae6e9c2b747ef7ef11d2711a11416e4e87ea"}]},"branch":"refs/heads/master"},"a259fcaee45369c88c3fa20f397ca037cc53e24b":{"kind":"REWORK","_number":2,"created":"2026-01-26 09:29:50.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/84/974684/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/84/974684/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/84/974684/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/84/974684/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/84/974684/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/84/974684/2"}}},"commit":{"parents":[{"commit":"9f4165f7738d48a5081261d1991d4821d1c33eb0","subject":"Merge \"Simplify import of sqlalchemy\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/9f4165f7738d48a5081261d1991d4821d1c33eb0"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-01-26 09:29:16.000000000","tz":0},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-01-26 09:29:49.000000000","tz":0},"subject":"Remove outdated comment in setup.py","message":"Remove outdated comment in setup.py\n\nChange-Id: I41ee38b146cfaf6713d392b832c9f48390c90efc\nSigned-off-by: Ivan Anfimov \u003clazekteam@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/a259fcaee45369c88c3fa20f397ca037cc53e24b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/a259fcaee45369c88c3fa20f397ca037cc53e24b"}]},"branch":"refs/heads/master"},"601fcfcfb74a7e4b56b9fda03a1b5de1ecf43c0e":{"kind":"NO_CODE_CHANGE","_number":3,"created":"2026-01-28 14:09:44.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/84/974684/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/84/974684/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/84/974684/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/84/974684/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/84/974684/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/84/974684/3"}}},"commit":{"parents":[{"commit":"9f4165f7738d48a5081261d1991d4821d1c33eb0","subject":"Merge \"Simplify import of sqlalchemy\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/9f4165f7738d48a5081261d1991d4821d1c33eb0"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-01-26 09:29:16.000000000","tz":0},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-01-28 14:09:44.000000000","tz":0},"subject":"Remove outdated comment in setup.py","message":"Remove outdated comment in setup.py\n\nsetup.py is no longer managed by the global tool.\n\nChange-Id: I41ee38b146cfaf6713d392b832c9f48390c90efc\nSigned-off-by: Ivan Anfimov \u003clazekteam@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/601fcfcfb74a7e4b56b9fda03a1b5de1ecf43c0e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/601fcfcfb74a7e4b56b9fda03a1b5de1ecf43c0e"}]},"branch":"refs/heads/master","description":"Edit commit message"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
