)]}'
{"id":"openstack%2Fglance~974981","triplet_id":"openstack%2Fglance~master~I71e190371236faefcc4615ef347fb350fe138c26","project":"openstack/glance","branch":"master","topic":"eventlet-removal","attention_set":{},"removed_from_attention_set":{"19138":{"account":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"last_update":"2026-02-10 05:31:54.000000000","reason":"\u003cGERRIT_ACCOUNT_19138\u003e replied on the change","reason_account":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"}},"9303":{"account":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"last_update":"2026-02-12 22:21:21.000000000","reason":"Change was submitted"},"8122":{"account":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"last_update":"2026-02-11 20:07:25.000000000","reason":"removed on reply"}},"hashtags":[],"change_id":"I71e190371236faefcc4615ef347fb350fe138c26","subject":"Update unit tests to use native threading","status":"MERGED","created":"2026-01-28 05:41:26.000000000","updated":"2026-02-12 22:22:27.000000000","submitted":"2026-02-12 22:21:21.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":8,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"974981-eventlet-removal","meta_rev_id":"f22ea38ca7a20f1c3162432f869a9f5a6f1cefc9","_number":974981,"virtual_id_number":974981,"owner":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},{"value":0,"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-02-12 22:21:21.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"recommended":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"all":[{"value":1,"date":"2026-02-10 05:31:54.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},{"value":2,"date":"2026-02-12 21:46:43.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"all":[{"value":0,"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},{"value":1,"date":"2026-02-12 21:46:43.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},{"value":0,"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Top Priority / Holds Gate"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-01-28 07:03:40.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-02-05 16:44:24.000000000","updated_by":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"reviewer":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"state":"CC"},{"updated":"2026-02-06 15:43:59.000000000","updated_by":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"reviewer":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"state":"REVIEWER"},{"updated":"2026-02-10 04:45:54.000000000","updated_by":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"reviewer":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"state":"REVIEWER"}],"messages":[{"id":"a00384d48282c76f43cea9f84458e01dfb02942e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2026-01-28 05:41:26.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"d50e6340c94b4ef716e0559272b11c7190f5fa0b","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2026-01-28 06:17:10.000000000","message":"Topic set to eventlet-removal","accounts_in_message":[],"_revision_number":1},{"id":"dfee7ce9abbb787d52f3e171bee8eb8cbef084dc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-28 07:03:40.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9d7c2a0c700b43f9b056c436838f301b\n\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/4828e37f7b284aa995116f9b36dcb033 : SUCCESS in 1h 12m 07s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/ede3f1b7a1704fab960c6a485973c3c0 : SUCCESS in 32m 27s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ff5f64af5e1f4b459b5b0ddf0b6d8634 : SUCCESS in 5m 56s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/2f20a530b21f4c05ad9124d9622bf51f : SUCCESS in 4m 47s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/5f86b8aa1fdc440ca31b1b202275b516 : SUCCESS in 3m 20s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6b319349adfc4ade8c54097e70f41954 : SUCCESS in 5m 45s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/27d779f65dc143b9a6345dcea483330b : SUCCESS in 5m 12s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/6eee8ee16406412781ad0d2f6aa76d9f : SUCCESS in 6m 22s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/fdb74b9a7bed4c53b656dab3af400bd1 : SUCCESS in 15m 22s","accounts_in_message":[],"_revision_number":1},{"id":"5d874db03a122e145bd031aeab31c59e6cf5b827","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-28 11:19:41.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0a895638325747e3ab0ccd11580fc9e2\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/12d9d6dea1914f00822029ca335b5b5c : SUCCESS in 9m 02s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/f02093d1d7014b59a20e99ee8b1e8783 : SUCCESS in 8m 39s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/4fff1efc12bd4d73a058da11ead28c68 : RETRY_LIMIT in 5m 14s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"912448d3bbb7b006dd3a5aea232108cb41e4fab7","author":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"date":"2026-02-05 16:44:24.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"1d3f44ab1d71aa413387e815b73f3b4976b41540","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2026-02-06 06:16:36.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"4e4242ba0687cee2330c5410d0a70a6a04c132e5","author":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"date":"2026-02-06 15:43:59.000000000","message":"Patch Set 1: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"4f0f5b0978ecaf57df3524f73deda8e0493e6de4","author":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"date":"2026-02-10 04:45:54.000000000","message":"Patch Set 1: Code-Review+1\n\n(2 comments)","accounts_in_message":[],"_revision_number":1},{"id":"60159a1fa1c3bc9f408795712608794817d30854","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2026-02-10 05:24:11.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"000a1400170c372dfe29ab649fdf1e893d0c262c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2026-02-10 05:24:35.000000000","message":"Patch Set 2: Published edit on patch set 1.\n\nOutdated Votes:\n* Code-Review+1, Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":2},{"id":"5372ceb21795b365894c2b826abdc229a2da62a3","author":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"date":"2026-02-10 05:31:54.000000000","message":"Patch Set 2: Code-Review+1\n\n(2 comments)","accounts_in_message":[],"_revision_number":2},{"id":"b07317987761918f36e553235b01db3630033622","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-10 05:41:59.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/068ebc274a5d48e1b768290fef089a39\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/8e51e28e87dc4177af4f62aba1c5150a : SUCCESS in 9m 20s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/32790cc112c04246a180d268e9c82c69 : SUCCESS in 8m 36s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/3eb6356d5b7b4360a588fbeea733c3f1 : RETRY_LIMIT in 3m 38s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"3de35118b9fcc9a28b8e40dd5a8a994a5d76d5b0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-10 06:35:04.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/6b427a07ae2c473cb383d2c34a66df27\n\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/9c85ee49f2a34184a653763887fd1f4d : FAILURE in 14m 04s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/01550963c0d44e2db025f18c572bca75 : FAILURE in 1h 08m 57s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/547a93f51ca842a4a69501d43871e93b : SUCCESS in 4m 39s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/9b939a5e629e44c6b9a95144730857d4 : SUCCESS in 6m 49s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/413f43a39f5a4e89b54345dcad0cef73 : SUCCESS in 4m 17s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a7dc3597638b4186864328629002f589 : SUCCESS in 5m 23s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/ed6ae9e8945f40ca979493a54ae8fbc8 : SUCCESS in 4m 10s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/ca49f002e989457eb5b9fabf71aedabc : SUCCESS in 6m 21s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/8782ef89be364c6d9f3c08149e3c4987 : SUCCESS in 36m 30s","accounts_in_message":[],"_revision_number":2},{"id":"d8bc70b7905341373b9a86350e3c71021b26be0a","author":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"date":"2026-02-11 20:07:25.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"f6c0006bea109bf87f582fa46133f09837df7b4a","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-11 20:25:24.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9c649a7f231f4a16a73f010d999c85e6\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/3493a6bbc9b1497f821e2c910ff7d419 : SUCCESS in 9m 02s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/9631c116be0343aab9ee2518edfff5b8 : SUCCESS in 7m 48s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/5fd5f708c9284659a04e8f86c534c5e1 : RETRY_LIMIT in 3m 58s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"0ed15a5ce1e2a24c686573e0c637fcdb63aa67f1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-11 21:21:08.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c3dee26caa5747729527b049c861bf9d\n\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/7ffd184870d34860a8c44166acae49fd : SUCCESS in 1h 07m 19s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/b1ea552ccc1944a9946a50f9053c0d86 : SUCCESS in 1h 03m 16s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f3601394bd494c72bd451edc9e406d5b : SUCCESS in 3m 17s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/60ebe0cae7c342038a09056befbbfe56 : SUCCESS in 4m 39s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/dacbbabb7d5a4f948266c99d1c95e856 : SUCCESS in 3m 21s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a6011f7bb34648efb386782e8ebc710f : SUCCESS in 5m 41s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/b8c157393cc740a1a9315449fe5f1e27 : SUCCESS in 6m 30s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/39a281844dc743a6aed034ef98c253af : SUCCESS in 4m 51s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/b02182c7d8844714bf4030c63e5cbc1b : SUCCESS in 16m 54s","accounts_in_message":[],"_revision_number":2},{"id":"66e3113020f3a039732ba69995ce32de743442e2","author":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"date":"2026-02-12 21:46:43.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"e677646f188f11da7461229fed60dece452be3b8","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-12 21:47:13.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"206630f86da55e547ad3f9b0c14dc649ac5dfcaf","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-12 22:21:21.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/cd8c0b6035b748e3af149a4e5c6d048e\n\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/17f6132776ea4369a5497e0618eabda8 : SUCCESS in 33m 43s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/4120a9cf977d488888e5b512108481b2 : SUCCESS in 32m 11s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e6de8d9c921b4bccab1be9e89a361f5e : SUCCESS in 4m 34s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/6ee7498a737d4dd8b81e018ab28a5b81 : SUCCESS in 4m 51s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/371258525998461bb3c376d999721530 : SUCCESS in 7m 25s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/69e1830e34364831adf02e326f7b00d8 : SUCCESS in 4m 50s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/cd267e94bec54d2aa226e6f63cfa0718 : SUCCESS in 4m 56s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/d0c5ff9c5c714f5db35333895b23aa30 : SUCCESS in 6m 17s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/25718693de4049b5b9fafcd0744cd096 : SUCCESS in 23m 06s","accounts_in_message":[],"_revision_number":2},{"id":"c45d7261bcc45440c640b16d7081b0ca4a2d0bec","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-12 22:21:21.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"f22ea38ca7a20f1c3162432f869a9f5a6f1cefc9","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-12 22:22:27.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3a30faa569954d5c91cfa1d838269e01\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7b7c27bd574b4ffb87f5aac84fe2502d : SUCCESS in 55s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"220135776d1edf97c991882d795a4c9c72bd9eac","revisions":{"959e7142d4f4fe9c585414179ef71840486cb140":{"kind":"REWORK","_number":1,"created":"2026-01-28 05:41:26.000000000","uploader":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"ref":"refs/changes/81/974981/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/81/974981/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/81/974981/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/81/974981/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/81/974981/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/81/974981/1"}}},"commit":{"parents":[{"commit":"9f4165f7738d48a5081261d1991d4821d1c33eb0","subject":"Merge \"Simplify import of sqlalchemy\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/9f4165f7738d48a5081261d1991d4821d1c33eb0"}]}],"author":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2026-01-27 19:08:14.000000000","tz":0},"committer":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2026-01-28 05:41:14.000000000","tz":0},"subject":"Update unit tests to use native threading","message":"Update unit tests to use native threading\n\n- Skipping eventlet-specific tests with @testtools.skip decorator\n- Updating test setUp() methods to use \u0027native\u0027 threadpool model\n- Changing test assertions to expect ThreadPoolExecutor instead of\n  GreenThreadPoolExecutor\n- Updating test_model_map to use NativeThreadPoolModel\n\nChange-Id: I71e190371236faefcc4615ef347fb350fe138c26\nSigned-off-by: Abhishek Kekane \u003cakekane@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/959e7142d4f4fe9c585414179ef71840486cb140"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/959e7142d4f4fe9c585414179ef71840486cb140"}]},"branch":"refs/heads/master"},"220135776d1edf97c991882d795a4c9c72bd9eac":{"kind":"REWORK","_number":2,"created":"2026-02-10 05:24:35.000000000","uploader":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"ref":"refs/changes/81/974981/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/81/974981/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/81/974981/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/81/974981/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/81/974981/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/81/974981/2"}}},"commit":{"parents":[{"commit":"9f4165f7738d48a5081261d1991d4821d1c33eb0","subject":"Merge \"Simplify import of sqlalchemy\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/9f4165f7738d48a5081261d1991d4821d1c33eb0"}]}],"author":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2026-01-27 19:08:14.000000000","tz":0},"committer":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2026-02-10 05:24:35.000000000","tz":0},"subject":"Update unit tests to use native threading","message":"Update unit tests to use native threading\n\n- Skipping eventlet-specific tests with @testtools.skip decorator\n- Updating test setUp() methods to use \u0027native\u0027 threadpool model\n- Changing test assertions to expect ThreadPoolExecutor instead of\n  GreenThreadPoolExecutor\n- Updating test_model_map to use NativeThreadPoolModel\n\nChange-Id: I71e190371236faefcc4615ef347fb350fe138c26\nSigned-off-by: Abhishek Kekane \u003cakekane@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/220135776d1edf97c991882d795a4c9c72bd9eac"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/220135776d1edf97c991882d795a4c9c72bd9eac"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
