)]}'
{"id":"openstack%2Fglance~975118","triplet_id":"openstack%2Fglance~master~Ic3f2c78d36f011fd11bb52b1255f75835df60e2d","project":"openstack/glance","branch":"master","topic":"eventlet-removal","attention_set":{},"removed_from_attention_set":{"19138":{"account":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"last_update":"2026-06-19 06:49:13.000000000","reason":"Change was submitted"},"9303":{"account":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"last_update":"2026-06-19 01:45:31.000000000","reason":"\u003cGERRIT_ACCOUNT_9303\u003e replied on the change","reason_account":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"}},"4393":{"account":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"last_update":"2026-06-19 06:49:13.000000000","reason":"Change was submitted"},"8122":{"account":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"last_update":"2026-06-19 06:49:13.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ic3f2c78d36f011fd11bb52b1255f75835df60e2d","subject":"Switch tests to native threading","status":"MERGED","created":"2026-01-29 06:59:10.000000000","updated":"2026-06-19 06:50:12.000000000","submitted":"2026-06-19 06:49:13.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":66,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"975118-eventlet-removal","meta_rev_id":"ac7cf72175f3f738d513057d4063653cee461992","_number":975118,"virtual_id_number":975118,"owner":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"value":0,"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-06-19 06:49:13.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"all":[{"value":2,"date":"2026-06-18 14:24:20.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"value":0,"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"all":[{"value":1,"date":"2026-06-18 14:24:20.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"value":0,"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"value":0,"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Top Priority / Holds Gate"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"}],"REVIEWER":[{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-01-29 09:19:32.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-05-12 15:10:58.000000000","updated_by":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"reviewer":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"state":"CC"},{"updated":"2026-05-12 15:10:58.000000000","updated_by":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"reviewer":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"state":"REVIEWER"},{"updated":"2026-06-04 14:58:36.000000000","updated_by":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"reviewer":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"state":"CC"},{"updated":"2026-06-10 17:45:14.000000000","updated_by":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"reviewer":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"state":"REVIEWER"}],"messages":[{"id":"a7c9293d3b3b726c4655ec966a9ac92e544b9489","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2026-01-29 06:59:10.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"77b2643f4e8700da4b4682a80a047c1abf2929c2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-29 09:19:32.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/2804796e403a4101b7ae4d8ba9d87291\n\n- grenade https://zuul.opendev.org/t/openstack/build/97218f9f8f934534b779018af5031e60 : SUCCESS in 1h 00m 54s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/641306c386614f999c9c75a76764e625 : SUCCESS in 59m 33s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/fa70417a93fb446daebd90d7a4117506 : SUCCESS in 1h 16m 43s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/04c37f0e042e45c5b5262959bcd0e586 : SUCCESS in 1h 06m 01s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a21431d41dff409385b1ed187962e3ae : SUCCESS in 2m 52s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/bc8358f848884b5199421455ee540bc1 : SUCCESS in 5m 04s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/33add028bfcc4b359d7cea3981d2e943 : SUCCESS in 2m 48s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/68e6ba60fdcb4875b3644b987eebb2d2 : SUCCESS in 4m 04s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/21ec8c8dcda648b089f6e86543dd8cdb : FAILURE in 4m 23s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/2bf8aa9c3eac442e81a7b8384a859b5e : SUCCESS in 6m 47s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/128d977acbbb4865bfefb35742e099fa : SUCCESS in 2h 09m 10s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/ea24a1920b3c4c3e8870b9a1e3eafd57 : SUCCESS in 1h 44m 33s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/0eb1fc4d69b543ccbf14ae004818dcd9 : SUCCESS in 2h 09m 55s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/328be9f3c4824a859ff4e7b520b5efca : SUCCESS in 2h 15m 02s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/1455d8bc9256466dae5f2e0774879e4c : SUCCESS in 47m 25s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/1fa7d86d91144b7e9e746d2bf00696e2 : SUCCESS in 1h 57m 51s\n- glance-grenade-centralized-cache https://zuul.opendev.org/t/openstack/build/00b1f2aabac841eab3e497a271640724 : FAILURE in 12m 37s (non-voting)\n- glance-s3-multistore https://zuul.opendev.org/t/openstack/build/35c9c99376c443189552a352dd212619 : FAILURE in 27m 18s (non-voting)\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/8db2e339ebb749818ae07ad1bdba6dd5 : SUCCESS in 28m 14s\n- ironic-tempest-bios-ipmi-direct https://zuul.opendev.org/t/openstack/build/567fc0e323a743579eb03a490f1db16b : SUCCESS in 35m 54s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"db72005b965648136a24f030dce20b5db5e47685","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2026-01-29 09:34:10.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"a1d3cfa1a1fe77c03a95be6b5e01caf2f2f7805f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-29 11:58:51.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/7bf7b985a03c47e8b197f8dfe59d81af\n\n- grenade https://zuul.opendev.org/t/openstack/build/81ad0ddd102f4790876271c127d6f093 : SUCCESS in 58m 00s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/4313ae0fad294cd8b74def7177884625 : SUCCESS in 59m 59s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/a563b4a53cf94817bb74f17ed2b0821e : SUCCESS in 1h 54m 19s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/0f4dc2bd9984405a8a4a469d1fa20798 : SUCCESS in 1h 07m 28s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6aac2cd6dcdc4dcd8aedfbd5ede2ac95 : SUCCESS in 4m 30s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/4d0738bdbb654f378453619a23cf3b18 : SUCCESS in 4m 32s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/481bcd0354b14ce69cf818a8b9cd29ea : SUCCESS in 3m 27s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f7f4d9bf77b14fe0b07d09a7687ae7b6 : SUCCESS in 7m 38s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/2734895722164b14b56b0471b9f28514 : SUCCESS in 6m 08s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/bdaa2e9d37a140869cbb27bd7a7aba76 : SUCCESS in 6m 57s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/117e1df582ae4d36ab2df98c8a99ddd8 : SUCCESS in 2h 03m 59s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/64962b4b0a274a31af897164ad93564f : FAILURE in 1h 14m 14s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/50fc4c999cd34f53b739b2eff8bf4e8a : SUCCESS in 1h 51m 27s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/c063f8f593c14744a7538053b3ebf420 : SUCCESS in 2h 22m 19s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/47b9fb58f0bc49ec87201cd7eca40ef8 : SUCCESS in 41m 59s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/8d66ed97f6af4e2ab735424980cdeb7a : SUCCESS in 1h 41m 43s\n- glance-grenade-centralized-cache https://zuul.opendev.org/t/openstack/build/ba1143e9ba914387b286fb79e918bc9b : FAILURE in 10m 34s (non-voting)\n- glance-s3-multistore https://zuul.opendev.org/t/openstack/build/1cb28577c4944d05a51d75ea96626a5a : FAILURE in 22m 27s (non-voting)\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/6d73a91d182d4bbd93e8fbb33c4cafc4 : SUCCESS in 23m 16s\n- ironic-tempest-bios-ipmi-direct https://zuul.opendev.org/t/openstack/build/65e7b221a8484c0d91b251b787dcbc03 : SUCCESS in 55m 42s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"8dcf8ba6ff14f002f5ec876cb158e5ff3965f322","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2026-01-29 14:47:54.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"19522495c345e950f24b69c875b03d1991271369","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-29 16:49:05.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a91a465d4f7249399e036590c81aa3f1\n\n- grenade https://zuul.opendev.org/t/openstack/build/11d4d44559e24a1cba91fb24194432cc : SUCCESS in 1h 01m 13s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/2af6c3805e5349dbb2a8286cf7d94a25 : SUCCESS in 43m 09s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/a8840cf6cbbe4d78b6cdb94ca4921da6 : SUCCESS in 1h 41m 07s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/6e86907c0eb84904817204409a16055b : SUCCESS in 47m 55s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0b20f438cc2c494ba90fcec56685e315 : SUCCESS in 5m 23s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/a9a53621bf6b47a1b81734392c486f19 : SUCCESS in 4m 42s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/2311e8aa4afd4f7f872ff77be4036530 : SUCCESS in 3m 49s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0cfcdf30182a47b1bb671ba7dca52b20 : SUCCESS in 6m 34s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/67925c4d2b79467491201b87e039606e : SUCCESS in 6m 00s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/bd698b2b247d407393b62616bcbcc8e0 : SUCCESS in 6m 26s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/fabe154e80444f0bb4ed8dd130c23658 : SUCCESS in 1h 55m 35s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/449e3205753d4df4947116d3898f618a : SUCCESS in 1h 00m 35s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/9f3cd7d473e2444f9eade9f88cd11f1c : SUCCESS in 1h 55m 11s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/d0a810d106664d76bbd2e57ea9425293 : SUCCESS in 1h 30m 40s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/fa4b7c789aeb4be180c1cad9c0ee66f3 : SUCCESS in 1h 00m 17s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/a4234b234be14e5e9dfb275b42504342 : SUCCESS in 1h 52m 57s\n- glance-grenade-centralized-cache https://zuul.opendev.org/t/openstack/build/1c4f97652b1345cd9632a5669c5bfd71 : FAILURE in 22m 05s (non-voting)\n- glance-s3-multistore https://zuul.opendev.org/t/openstack/build/f50e568ce9e2445f8e980e657fc14875 : FAILURE in 17m 42s (non-voting)\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/9d83f09995194b5ba6e572c93b957453 : SUCCESS in 22m 32s\n- ironic-tempest-bios-ipmi-direct https://zuul.opendev.org/t/openstack/build/bd5d935f060440a7a470bd769c14a113 : SUCCESS in 36m 36s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"0bf265e138dda23134a89090d72521528ed20af4","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-29 17:56:06.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e998b7e6293a41c884917fbf175b1707\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/39085a67a6074b7ea4d8d3aee5c2a4b0 : SUCCESS in 9m 50s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/fa4de96c98f34e809e765bde00db0997 : SUCCESS in 9m 47s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/0f06af4d0a5c4a7abdf1bdc25e2538e7 : RETRY_LIMIT in 4m 49s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"0c6aa9ea1cfcfee6f8583f05340ca29d084a5ce3","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2026-05-07 14:16:18.000000000","message":"Topic set to eventlet-removal","accounts_in_message":[],"_revision_number":2},{"id":"35fb1dfe541611d355c4bec51f28e4de7225bf86","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"date":"2026-05-08 09:37:24.000000000","message":"Uploaded patch set 3: Patch Set 2 was rebased.\n\nOutdated Votes:\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":3},{"id":"f266a562d716b78f0dc311ade2a70876029ff698","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-08 10:24:54.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/fdaf488cf6f34ab587b14d7c90293c51\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/1e114e0d1f6147d295a35e899cb07ba5 : FAILURE in 5m 18s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/b00bcbb890fe46039a8cd77d6960b4dc : SUCCESS in 5m 50s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/5ccca77515ab4ec2a4acf9b09a7a8a78 : SUCCESS in 8m 37s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"a87a59a72ed1249366ce8131441a290961d02112","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-08 11:58:59.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5551f4b232324881b2c42e66e239bbc0\n\n- grenade https://zuul.opendev.org/t/openstack/build/8facd47485fc4a289fe4f49389ee6470 : SUCCESS in 57m 31s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/e89bf01f616e4ca293d5af3ed34c0f78 : SUCCESS in 1h 02m 26s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/5607d936d9624d93bc8cc464024ff3a3 : SUCCESS in 1h 39m 53s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/b405971622974a78aa06e217612462ac : SUCCESS in 1h 01m 59s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/374bc4d136934942b8c72db60584f9b3 : SUCCESS in 4m 29s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/83cb356c047444c8a8d8aafe9897736a : SUCCESS in 4m 31s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/e3c5facdf6a44052ae8e523c02ee1468 : SUCCESS in 4m 52s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/60bcec93bb92498a82b82e819a7e1bc2 : SUCCESS in 4m 21s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2c55faca58b240dfb49cefdbe6ce1474 : SUCCESS in 5m 44s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/c6310186bd0744dc9fbe0f2a70a7a8fc : SUCCESS in 5m 43s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/e767656a1ea44d99bb69d37993c1a6f7 : SUCCESS in 4m 33s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/8090cfb534f04a6a83b52605d03f5e8f : SUCCESS in 1h 49m 29s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/914678ae5beb4a5e825af0e36405cf32 : SUCCESS in 1h 26m 36s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/d11a4e59ec424e4598aeb204bcc67536 : SUCCESS in 2h 02m 22s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/f303ef0278b0494b8d323b881623eda3 : SUCCESS in 2h 16m 29s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/b74807aa51814efcb8fb18af5b033b58 : SUCCESS in 47m 51s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/06e5f6c9edff4095bc7e444a336449fa : SUCCESS in 1h 08m 18s\n- glance-grenade-centralized-cache https://zuul.opendev.org/t/openstack/build/2bb1a314244447caaa21df1675f65574 : FAILURE in 7m 34s (non-voting)\n- glance-s3-multistore https://zuul.opendev.org/t/openstack/build/d4a261729d934da9a24927042cd47c53 : SUCCESS in 1h 11m 25s (non-voting)\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/15b6550ca4ab4b4cbf6c29179c32b9ea : SUCCESS in 18m 20s","accounts_in_message":[],"_revision_number":3},{"id":"18423831540346ba738a850255089a9257816701","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"date":"2026-05-19 06:30:16.000000000","message":"Uploaded patch set 4: Patch Set 3 was rebased.\n\nOutdated Votes:\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":4},{"id":"8c49a174fdaeb9567b9f92fd55a7942d6876746b","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-19 06:47:16.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/96b897444ae84d0a988bc60238fb159f\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/e03d8dbda6b94fb0bb85fd642f54362d : SUCCESS in 8m 14s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/641b6e260a31490db450be010a75d8b5 : SUCCESS in 7m 33s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/1d3d43a7ceec4be6b078a5cc5db847a0 : SUCCESS in 13m 11s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"f45bb55e4833003a623744ba19589ca35073ea47","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-19 08:30:26.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5ea788601be34186a4f3fd47f3eeeb4e\n\n- grenade https://zuul.opendev.org/t/openstack/build/603bebfc17f747c3ad5ecd0901388f5e : SUCCESS in 1h 00m 34s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/5f499fbd6e014149b4a73477849f011f : SUCCESS in 31m 53s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/bff5c1c7bd144c6e950243a295afe2a1 : SUCCESS in 1h 30m 12s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/f3a9cd708eb544268a94e6117cfb6045 : SUCCESS in 52m 27s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/79e8bca67f1c46628878ca1833fe9bec : SUCCESS in 5m 40s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/57a7ac48f554469f9500622295b24208 : SUCCESS in 6m 03s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/3fe60e5dfe2b4a0aa8fafd2ab5880c4b : SUCCESS in 3m 39s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/ae4f0e184ebb43c59690097b95411d64 : SUCCESS in 4m 07s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ace08c63e3294805bf385baeedf7f602 : SUCCESS in 7m 07s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/8f0ba8c998c0486a98421426032ab7e2 : SUCCESS in 3m 26s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/4fff6573a7174327b65ab5e0b87c4250 : SUCCESS in 5m 37s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/dfbb784f5fa54060b0b7e88ce98c43b5 : SUCCESS in 5m 08s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/bb194b7d7df542ddbd3a003e14c98dc1 : SUCCESS in 1h 12m 28s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/6c8c356ad39e4ff0ae8969b72c709de4 : SUCCESS in 1h 36m 48s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/3e03af1499e04616aab96c76e37f86d6 : SUCCESS in 1h 58m 20s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/6b22836dc1de4fd2a28201b58cf998fd : SUCCESS in 1h 26m 19s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/776b27d0a9b349079832e3dde1c9e01c : SUCCESS in 43m 20s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/4ab388f2d83240e09fd5ca58459c0b7b : SUCCESS in 1h 31m 44s\n- glance-grenade-centralized-cache https://zuul.opendev.org/t/openstack/build/c9c19816d67540a8ae1238eb91a31f24 : FAILURE in 6m 58s (non-voting)\n- glance-s3-multistore https://zuul.opendev.org/t/openstack/build/cf8dc76b1fb64ca288c74f29fed949c5 : SUCCESS in 1h 35m 27s (non-voting)\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/afe36cc935ca47aab6a9d33cebe621cf : SUCCESS in 24m 02s","accounts_in_message":[],"_revision_number":4},{"id":"dcbd3f80bb8217736fbbe335ee48483ee66849db","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"date":"2026-05-29 10:40:07.000000000","message":"Uploaded patch set 5: Patch Set 4 was rebased.\n\nOutdated Votes:\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":5},{"id":"f8880005f305aaa8b03380d77d2a828a8630d9bf","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-29 13:04:02.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/235c81e625cb47bb94d60c2684891b40\n\n- grenade https://zuul.opendev.org/t/openstack/build/9a028ea43d5e4bf88f52c074b4f35897 : SUCCESS in 1h 08m 27s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/5b85da2ea93e46e3a10e9346e6c15d8c : SUCCESS in 1h 00m 50s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/45d593b2d35c4c31913f2d2934fc6ef1 : SUCCESS in 1h 35m 22s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/a22e9464f80c4cbb84b11f59438b39fb : SUCCESS in 1h 09m 42s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/abf82b285a3342b481017b2cc6b4e222 : SUCCESS in 5m 13s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/543b417afc594108ae585168cf488b8c : SUCCESS in 7m 12s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/603057f5c7e8400ca269cbe7bbb2858c : SUCCESS in 6m 21s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/687b4b0cb6a14fecbe2bb35f94fc8a2f : SUCCESS in 8m 34s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/656c2182e5354d0d96cffb8079912d30 : SUCCESS in 7m 11s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/dbc8e0691095418596c4dcb32711c02f : SUCCESS in 7m 22s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/c0879aeebd16434c81ebcbe515aa8458 : FAILURE in 6m 33s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/ac4c86a854804f93bc763e3e3540b4c4 : FAILURE in 6m 19s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/43bb9ab2c1bc46dea6fd292bd0faf70b : SUCCESS in 1h 37m 18s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/5badffa8e14e4a5cb8bf93119555e7d4 : SUCCESS in 2h 21m 23s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/45cf5a11325b4f5e97b028d456a0baff : SUCCESS in 2h 02m 07s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/77c1319651704fde953640807671aa0e : SUCCESS in 1h 27m 20s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/07153c770e5e444fb9f929aa5ddbf9a7 : SUCCESS in 58m 44s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/f2bbac76573f41ba85d9bd5f643fa36d : SUCCESS in 1h 02m 01s\n- glance-grenade-centralized-cache https://zuul.opendev.org/t/openstack/build/75759bc5383f4c048cf516f40b05876b : FAILURE in 12m 00s (non-voting)\n- glance-s3-multistore https://zuul.opendev.org/t/openstack/build/19b50d79ed354ef6a798ddf3fa7bcd34 : FAILURE in 1h 38m 09s (non-voting)\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/f47f5bb1a4744791ac78cc4643910330 : SUCCESS in 33m 54s","accounts_in_message":[],"_revision_number":5},{"id":"80171a596f592de5e85dbfbb47f2597e32fd5d60","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-29 14:54:13.000000000","message":"Patch Set 5:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/be01d03c0fd544a38d62b5915c519525\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/84015e2252b149b9a49aa16cdf202399 : SUCCESS in 7m 11s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/52508f283d5d48188147f4193e2bcd51 : SUCCESS in 7m 07s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/44b4c87aa7b0420e8f0074d4ec1f1bdc : SUCCESS in 11m 51s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"c37c630b589984ecfceae62ef2abb01c4a45eed7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"date":"2026-06-02 07:15:30.000000000","message":"Uploaded patch set 6: Patch Set 5 was rebased.\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":6},{"id":"75e7757ca68d56055dec21b3837e81c1f0b41122","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-02 07:36:22.000000000","message":"Patch Set 6:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/caf123698fd04f15b9e52de6a0cff585\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/24227db9b59141e3ae2a309886c247d1 : SUCCESS in 5m 06s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/3ab9257b233f4de4b46ed062a9fffdfe : SUCCESS in 5m 16s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/7da4154f15ea48c4b16e9125f32473c0 : SUCCESS in 8m 14s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"95b8f306d48e5e5b52a623edeeaea8add3e8df59","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-02 09:31:58.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/9cee9f62ee8a45738d0217b08922b492\n\n- grenade https://zuul.opendev.org/t/openstack/build/b9f24e39fcfd45c4a496d37ea190f86a : SUCCESS in 1h 16m 00s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/719a9bbffcf448bebd91c80c477c2c83 : SUCCESS in 34m 24s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/5f1cd38d76e54411a8839e9f5f876c63 : SUCCESS in 1h 45m 16s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/66dc2e9de9684ee7af9c83965de8fc4c : SUCCESS in 46m 32s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/835ee466f9c44b79be51050bec306f54 : SUCCESS in 3m 54s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/61a09e702412472eb21b1fe5d83f8dca : SUCCESS in 4m 31s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/e667fc1560b24b11980a01c91b5a7704 : SUCCESS in 4m 25s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/7f62aaa656f442149b0baaa9651dd5b2 : SUCCESS in 10m 09s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b5db6592da544cacbf32141f384e61c0 : SUCCESS in 7m 16s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ea281786032f41ad9e85bd66f7d3f094 : SUCCESS in 7m 46s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/b23b7ce22bda41a6ac2af03ce4b8231a : SUCCESS in 6m 26s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/a5e0eb0d361b4804aecac24a7a0d7528 : FAILURE in 5m 19s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/4c3ae2e67fd541928a0ec0dfbde88d74 : SUCCESS in 1h 37m 56s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/1f3a43f8fee34bb483940bc47dd122a3 : SUCCESS in 2h 10m 00s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/fe6b6fef69f7475cb095c4d21b49eaa5 : SUCCESS in 1h 53m 02s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/c6aa72d2aab549abb4aad4e46f97bb69 : SUCCESS in 1h 14m 38s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/c471b75fc6b847a1a40af337a1a4047b : SUCCESS in 1h 01m 10s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/098c707880b7497ab168f48766225ada : SUCCESS in 1h 01m 18s\n- glance-grenade-centralized-cache https://zuul.opendev.org/t/openstack/build/387469b8c36b4904af6e2c6a4de2b215 : FAILURE in 16m 25s (non-voting)\n- glance-s3-multistore https://zuul.opendev.org/t/openstack/build/bfc3a984784f41dfba499836659eb753 : SUCCESS in 1h 55m 22s (non-voting)\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/c474f2da8c4d4c52909cb6415c335fd2 : SUCCESS in 45m 32s","accounts_in_message":[],"_revision_number":6},{"id":"5973f19a86eedd3b56263414b92f98fa729e9c09","author":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"date":"2026-06-02 15:20:02.000000000","message":"Patch Set 6:\n\n(6 comments)","accounts_in_message":[],"_revision_number":6},{"id":"4736fd8e72e8ed881162033aecca5afa3344c9f8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2026-06-03 05:25:25.000000000","message":"Uploaded patch set 7.\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":7},{"id":"50d6c2491059b74fe5a2717cade7bf679b3aeff6","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2026-06-03 05:26:01.000000000","message":"Patch Set 6:\n\n(7 comments)","accounts_in_message":[],"_revision_number":6},{"id":"fed4fbe99a5da364e8b67e473c87c9c59e858575","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-03 07:43:34.000000000","message":"Patch Set 7: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/51ae44519b6f4e5e8cfff4dce92dd951\n\n- grenade https://zuul.opendev.org/t/openstack/build/0468417203ca49d2a594dffc1bf39af0 : SUCCESS in 1h 06m 11s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/d90de166efb14076913e3cf7a9c0b1f3 : SUCCESS in 46m 49s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/7488e96b644144c5b55d72ac36c52964 : SUCCESS in 1h 39m 06s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/e2a03775bc0e4109a48095653b19d221 : SUCCESS in 38m 56s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/13382a5e22fb41cba80f1536c80e888f : SUCCESS in 6m 16s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/3b1d0ac591914db2bf601f31070b4229 : SUCCESS in 3m 16s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/bb6f69fd332d4318a937034befc3700c : SUCCESS in 6m 47s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/4c1df4362b5b47a29fac2419caa07ed6 : SUCCESS in 10m 04s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e4cb02ff8e224c8f8b1f59b541e949f0 : SUCCESS in 7m 02s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/79d2150425374814bd0b0fbd1be4edf6 : SUCCESS in 6m 26s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/a0531f5ca0084e348dfad40b5ecf8713 : SUCCESS in 6m 41s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/850dd83992ac4eba80ffbcc356c97ae1 : SUCCESS in 8m 15s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/617dd1bc3eba4b1dbacb1bc6f7710a2d : SUCCESS in 1h 21m 52s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/eaa171911fea416dbfe8b8a9a9197688 : SUCCESS in 2h 05m 21s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/1d4e733c36a546bb81c0353046d60833 : FAILURE in 2h 00m 10s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/ed79c2a9ad2c4ad39ae2807209240df4 : SUCCESS in 2h 16m 10s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/7fcdd3a3f91d441a86f4f66aa3f53357 : SUCCESS in 49m 50s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/26ba5a22b84e42119f19cc10d345fb38 : SUCCESS in 1h 59m 23s\n- glance-grenade-centralized-cache https://zuul.opendev.org/t/openstack/build/9337262f08bb406696b8b7ba8d525a15 : FAILURE in 15m 24s (non-voting)\n- glance-s3-multistore https://zuul.opendev.org/t/openstack/build/f0d0bfa433fe4aeb8c0986b2f1cda087 : SUCCESS in 2h 00m 45s (non-voting)\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/162c8229d4954149b718f8ec46e9dade : SUCCESS in 33m 15s","accounts_in_message":[],"_revision_number":7},{"id":"f71adb6f4badf19e51b47257a59c47d7783669bc","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2026-06-03 08:14:19.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"101793de0461bc6dcbab472bcbb266469335fb82","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-03 10:43:08.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/050ef11a8f0e43b19b5bec9299dfc3b9\n\n- grenade https://zuul.opendev.org/t/openstack/build/f803201e9ea54a09b576eb74772125f5 : SUCCESS in 1h 11m 09s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/1c95427c8b9346b0af5483a467b5bff6 : SUCCESS in 32m 00s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/4c6729e628244696aed575062f5dfd58 : SUCCESS in 1h 17m 55s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/b4678bab36ba40fe92430c86d2b10795 : SUCCESS in 1h 06m 56s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/59d420736bf74f30a3ea1100d1f20100 : SUCCESS in 7m 00s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/f85780ae9a604e84890e7052ffdf4026 : SUCCESS in 5m 49s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/1e4c37315a24404fbdfe733cbb5def92 : SUCCESS in 5m 34s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/62078cc3414a4ed0b02840ed62908737 : SUCCESS in 8m 45s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6aa249a3499a4453bd60e9f8f3c37546 : SUCCESS in 6m 19s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/deae8a3f38e244128e080625dd064d97 : SUCCESS in 4m 23s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/c8116bc747cb41bd915c8a77342549da : SUCCESS in 6m 04s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/e27bca8db36d479384556c1a7058a30e : SUCCESS in 7m 22s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/cb4c47a02f134c6bbacfdd1d6219fee1 : SUCCESS in 1h 40m 05s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/7dc6bb3242fd437f81aa72861eef7cef : SUCCESS in 1h 58m 54s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/abd8c917baff478eb18ae612fd967062 : SUCCESS in 2h 00m 41s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/5b37d56cc6d24c5b84a20d6392492ad1 : SUCCESS in 1h 51m 29s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/2293f377c8a34dae9c7a91235f2bee1d : SUCCESS in 57m 49s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/09ad5d7f5f4c4898a3404f398cfadfce : SUCCESS in 1h 08m 40s\n- glance-grenade-centralized-cache https://zuul.opendev.org/t/openstack/build/0e61dd8f897b4d9c9ac88ba8c66d6018 : FAILURE in 8m 26s (non-voting)\n- glance-s3-multistore https://zuul.opendev.org/t/openstack/build/bdc078b896e34ad9b2320014ac79299f : SUCCESS in 1h 36m 05s (non-voting)\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/298a2dd164684d72a4048afc9c88b798 : SUCCESS in 34m 40s","accounts_in_message":[],"_revision_number":7},{"id":"5fa99f0fb986f3a72513acc262d1e81e5cd64ca6","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-03 10:43:15.000000000","message":"Patch Set 7:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2f0a2233cff048cfbcf64318659c8686\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/fdf00c72041748c384803e20bf3ae2ad : SUCCESS in 9m 43s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/e2eb5e67201e4c7dacaa31f386fa15b8 : SUCCESS in 9m 12s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/2afc26a52d77487fadec1f73937e1fce : SUCCESS in 13m 51s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"6b144adcd5ae2e4fddd3d9f8f5ea08163c3a4925","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"date":"2026-06-04 08:49:58.000000000","message":"Uploaded patch set 8: New patch set was added with same tree, parent tree, and commit message as Patch Set 7.\n\nOutdated Votes:\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":8},{"id":"1ffadce5d1cbdfdde0b8185a0ed482328de27b6f","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-04 09:10:19.000000000","message":"Patch Set 8:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/db45fcd02dc5477ca612b0b03bb3bff2\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/8e938930c00d4521a9788dfa07d67056 : SUCCESS in 8m 00s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/c4baf162f8d74354988708a19745e556 : SUCCESS in 5m 48s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/857ee04acd834674b781a2fcca4c2b88 : SUCCESS in 10m 41s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"46b1aeaf8efd3403e5f4980453d45eab596e44a8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-04 11:35:15.000000000","message":"Patch Set 8: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/96e312ddbdef4f1eb6f589f407529e00\n\n- grenade https://zuul.opendev.org/t/openstack/build/7b8e4b09806c423eb01c182eb12f86a1 : SUCCESS in 31m 54s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/71f8541a29874f87a85f30fd30471e3b : SUCCESS in 1h 15m 07s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/eca00a33f2a04b6b9caa7ea0f6c13b25 : SUCCESS in 1h 19m 55s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/6fdc6289da26439db719cb476eb7ee26 : SUCCESS in 45m 15s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d1d584c8b07849a4a607a1d4eda5f337 : SUCCESS in 3m 38s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/b1baee072ff24f42893c7fe1689b9026 : SUCCESS in 5m 05s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/80b99e5318e64f79a270211e0543d79c : SUCCESS in 5m 00s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/421c6c36451d4ce2a41889ed6bd30e2a : SUCCESS in 4m 31s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/11257d431a6444cbbe56d0dcc8ab5025 : SUCCESS in 5m 24s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/332cb6132893469391ae460e46fe7ae3 : SUCCESS in 9m 11s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/1f3e1273719849ef8457213e85522678 : SUCCESS in 5m 44s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/49b87284a70f423eacc98e172ca10c0b : FAILURE in 6m 19s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/a3b52590e5114562bf03056cafd0fae1 : FAILURE in 1h 19m 43s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/0faf93a8c2d94185824b559a9d56a7a1 : SUCCESS in 2h 16m 54s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/46bdc0b3bfc145b28e1b7b5f6daf5dfb : SUCCESS in 2h 14m 35s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/677eccc6e2b2430f9917e18332240cd3 : SUCCESS in 2h 37m 02s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/13f474d941f142999ce35e74f8feab5f : SUCCESS in 59m 29s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/ce89df8a719a499a98c869d7d8bbc7ec : SUCCESS in 1h 42m 50s\n- glance-grenade-centralized-cache https://zuul.opendev.org/t/openstack/build/87b6dbe43b0d46ee886991de246f5a72 : FAILURE in 7m 34s (non-voting)\n- glance-s3-multistore https://zuul.opendev.org/t/openstack/build/24b06b4482b549e0bca373dbbf29b8d7 : SUCCESS in 1h 53m 25s (non-voting)\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/ebef1dfc4e3241cdb62e0207932ac670 : SUCCESS in 39m 50s","accounts_in_message":[],"_revision_number":8},{"id":"44428589e1c5f3b89010b066533b5cdfdd11248b","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2026-06-04 14:58:36.000000000","message":"Patch Set 8:\n\n(7 comments)","accounts_in_message":[],"_revision_number":8},{"id":"1b4e79ff4db9af04a8c4cc13eed00df6f3395e82","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"date":"2026-06-05 11:26:30.000000000","message":"Uploaded patch set 9.\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":9},{"id":"4342653fc6d156150306cc5e1c1cd62105f1c84f","author":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"date":"2026-06-05 11:27:44.000000000","message":"Patch Set 9:\n\n(8 comments)","accounts_in_message":[],"_revision_number":9},{"id":"da95dc4e8e1a1cd85b0306a996233daa8d5ea379","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-05 11:43:24.000000000","message":"Patch Set 9:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8ea64a06cf8545318827d86b04260627\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/6311fb46126b426cac429893e3e078ad : SUCCESS in 7m 17s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/10a0abb23fb3473c89c3b8964a1261c6 : SUCCESS in 9m 09s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/94db163c9dd34c3c80c7c3a33b177e5a : SUCCESS in 14m 00s (non-voting)","accounts_in_message":[],"_revision_number":9},{"id":"b8ee7aa4ab323cf6d9ed7c01188593096c55282b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-05 14:01:08.000000000","message":"Patch Set 9: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/ca22602956be4f18931560b42fd6fe99\n\n- grenade https://zuul.opendev.org/t/openstack/build/b3648f593d6540878123a6b7cca13efb : SUCCESS in 1h 10m 01s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/b3ad409eb68246e4b81dcb5ce97a5ccd : SUCCESS in 1h 12m 02s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/29b01d9ee26d4e2fa2bfb15d972c54db : SUCCESS in 1h 37m 05s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/689113a028a243e88dca24200d63f604 : SUCCESS in 1h 02m 00s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6007f8ba6e98429ebccc9a66ae61985b : SUCCESS in 4m 21s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/bdf4e35e99334712924986f52ed712af : SUCCESS in 4m 44s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/99b7d4b2afee4d01a0e53c6e71ea7000 : SUCCESS in 4m 48s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/5677aaa2e6634234992147f8679e115e : SUCCESS in 7m 28s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/68141a3af311410ba47de47279d47fd9 : SUCCESS in 8m 24s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/094ccca281a347c185632b1091b01d99 : SUCCESS in 5m 12s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/c89687773f474a938df80ae82462ed30 : FAILURE in 6m 13s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/45c1716515a1415399f26717e4e81804 : FAILURE in 9m 03s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/46d27f92595a4f478302845ded7c6489 : SUCCESS in 1h 45m 25s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/461ed4c46afa40359480acadd5784fb9 : SUCCESS in 2h 07m 44s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/06c13df7f1a945199e3eb1bef06f251f : SUCCESS in 1h 10m 50s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/50c439074573485aa9adde87a0d4ac77 : SUCCESS in 2h 28m 51s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/bbdf92602c9d46e6ad5b7a27dcc864a6 : SUCCESS in 1h 05m 39s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/3931a4e5d2314e74afa6060b1f03f3e3 : SUCCESS in 1h 53m 25s\n- glance-grenade-centralized-cache https://zuul.opendev.org/t/openstack/build/992e5e751deb4c099e56bc1d3ff16491 : FAILURE in 16m 02s (non-voting)\n- glance-s3-multistore https://zuul.opendev.org/t/openstack/build/41010d0e552e40b79a7aaaf5087e9d4b : SUCCESS in 1h 33m 22s (non-voting)\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/7e75a2cfa45e4ac58faa9045fdcecac9 : SUCCESS in 16m 31s","accounts_in_message":[],"_revision_number":9},{"id":"01cfed2a617c83659f56b0645553535f644cdf1d","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2026-06-05 15:52:49.000000000","message":"Patch Set 9:\n\n(5 comments)","accounts_in_message":[],"_revision_number":9},{"id":"24fb67b074ee5471ecf0d8b8c29bab4b4ed0ed77","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"date":"2026-06-08 07:17:32.000000000","message":"Uploaded patch set 10.\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":10},{"id":"ba95cbbcfb05f039c822e7b7bd356877748a79b2","author":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"date":"2026-06-08 07:18:10.000000000","message":"Patch Set 9:\n\n(2 comments)","accounts_in_message":[],"_revision_number":9},{"id":"7b4db6a91e3e22c9673737b7b162a52629327938","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-08 08:59:05.000000000","message":"Patch Set 10:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f47d270aa73c427db1a4041d04ece7db\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/e5698eaf319e4229a1272e01e60db180 : SUCCESS in 5m 46s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/e144f1c67ef94267b1ba9fa4e85fe822 : SUCCESS in 9m 14s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/5ac4e549930141e799799633284ea79e : SUCCESS in 7m 56s (non-voting)","accounts_in_message":[],"_revision_number":10},{"id":"690d4041143e6596755fdb975208d96aeeab6083","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-08 09:49:12.000000000","message":"Patch Set 10: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/01924128a05746498567c0721de79caa\n\n- grenade https://zuul.opendev.org/t/openstack/build/0f34f7a4ab9f4068b75919140651a7fe : SUCCESS in 31m 22s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/d2c6c3e34b9f47c2958f1aa66a6c8ef8 : SUCCESS in 1h 06m 27s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/230da50fd9a3428ca371d0f5903a02eb : FAILURE in 8m 25s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/79c466bda3c749aea2830e8b12e77fa8 : SUCCESS in 1h 07m 19s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e166a96ac81a451f9e3a47c872d62104 : SUCCESS in 5m 17s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/b85775f8b4b6476ab35aebe21cd4938c : SUCCESS in 5m 17s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/f5b08242403e4104817f9937e995ed11 : SUCCESS in 4m 16s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/a74a82cd27374f2e86a99977db819c12 : SUCCESS in 6m 17s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b196368abb8140f9b61a26e3fdcc0848 : SUCCESS in 6m 28s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/2c76aa6879b04519862ea44a7e2bb4bc : SUCCESS in 4m 50s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/5c725e8f37c44d82a50cc5108880f9ae : FAILURE in 7m 27s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/ff7a4e1b35e64a2d9fa37414b8a8d930 : SUCCESS in 6m 24s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/b2518f5a240c426a91884608e1b332b4 : SUCCESS in 1h 54m 38s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/9ad554dc9be947c2abb0863a43d43ba5 : SUCCESS in 1h 54m 58s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/0e9b7d2e375e4db6be3569a6d6f20e7c : SUCCESS in 2h 18m 51s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/16c77a770f3546958dd63ff08c24bd9f : SUCCESS in 2h 25m 50s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/dd5bd273f05a4c60a9b802849d68f89b : SUCCESS in 58m 35s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/2cc4257fa3784d639db2d56dcb31e896 : FAILURE in 8m 02s\n- glance-grenade-centralized-cache https://zuul.opendev.org/t/openstack/build/2e4d5e9b6af445348d7f47fc1eaa89d4 : FAILURE in 15m 18s (non-voting)\n- glance-s3-multistore https://zuul.opendev.org/t/openstack/build/d9094fca62a5455b93fa13fc0c735414 : SUCCESS in 1h 04m 44s (non-voting)\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/6516d4ba3af14f8ca712bb168c185be2 : SUCCESS in 34m 32s","accounts_in_message":[],"_revision_number":10},{"id":"0d83d085e505ba6a1a353bc24bdc8b0acfa40115","author":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"date":"2026-06-08 10:18:44.000000000","message":"Patch Set 10:\n\n(1 comment)","accounts_in_message":[],"_revision_number":10},{"id":"30c59084a4b1a2401fe2764367902e3012c95dbb","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-08 11:40:10.000000000","message":"Patch Set 10:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/cc332b636632488a9c37ee03640b499d\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/090bf2df1c024a8b92f2159a6bef747c : SUCCESS in 7m 17s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/8a95fd3f56364858bdde94e91634caa6 : SUCCESS in 6m 58s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/e2598cc34f794a478f5e8778f49a2268 : SUCCESS in 13m 07s (non-voting)","accounts_in_message":[],"_revision_number":10},{"id":"6bef55c62413e37359366f61b6d5b803347c9b90","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-08 12:31:34.000000000","message":"Patch Set 10: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/688cad417b1d425a96a1387e61af9be7\n\n- grenade https://zuul.opendev.org/t/openstack/build/5ac5e8e15d924caaa6089c6dff15a389 : SUCCESS in 1h 04m 37s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/1def8805140c4a5d83e4e854e4abab86 : SUCCESS in 1h 07m 54s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/3883e6a8c9dc48d1b8a411da06f602f2 : SUCCESS in 1h 25m 33s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/c99cdfbb0ed642aea5977e9ee6f39dae : SUCCESS in 1h 03m 30s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/08bc1ce11752467aa973ed49e4ed9447 : SUCCESS in 6m 18s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/f644bfaf542a4f3388b93ff1c583aebd : SUCCESS in 2m 58s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/1c9c05f7941249378b88fa2bb52ea8b0 : SUCCESS in 5m 06s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/66b8c377df654537b266ac6c30288e37 : SUCCESS in 4m 29s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/651785049a774d46bf961e480d85c136 : SUCCESS in 3m 12s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/512b818947d34ec4a0ff0da4b7f8b8e9 : SUCCESS in 6m 03s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/9754375b91f04a289cf4d4b909c055a4 : SUCCESS in 4m 28s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/57870703585b40e796dc2b59360ddd81 : SUCCESS in 6m 24s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/16913c193c4e4ed48f3eae2866a7d673 : SUCCESS in 1h 48m 45s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/be31edd7b7e54032bd3a6ee64428557b : SUCCESS in 2h 04m 45s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/647e231e363e459c84de056387303d31 : SUCCESS in 2h 08m 21s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/afac12d02c284254beffc8bbee3ae294 : SUCCESS in 2h 00m 06s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/847d0dbe68c347a1a583d276eb2a5907 : SUCCESS in 1h 05m 25s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/dbf93535a6cb4b29833bad5c24bd3cdc : SUCCESS in 2h 06m 05s\n- glance-grenade-centralized-cache https://zuul.opendev.org/t/openstack/build/cc33f23500404908827c1ede1d2b6477 : FAILURE in 8m 30s (non-voting)\n- glance-s3-multistore https://zuul.opendev.org/t/openstack/build/37216c5376424fa8890e46f79c8d0f52 : FAILURE in 1h 34m 40s (non-voting)\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/ce90e3903d704f79902ef7626240acec : SUCCESS in 17m 18s","accounts_in_message":[],"_revision_number":10},{"id":"54f62926c198d66ecfddf6358b83b18693debd7c","author":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"date":"2026-06-08 14:48:25.000000000","message":"Patch Set 10: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":10},{"id":"694916c885d4d3ecf3bbae9fa6972bc4589d76be","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2026-06-08 15:13:03.000000000","message":"Patch Set 10:\n\n(4 comments)","accounts_in_message":[],"_revision_number":10},{"id":"a0856eda5cc4a2e33fe73f3a6d112bdfaeb76c0f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"date":"2026-06-10 12:30:51.000000000","message":"Uploaded patch set 11: Patch Set 10 was rebased.\n\nCopied Votes:\n* Code-Review+2 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n\nOutdated Votes:\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":11},{"id":"00bb135bb436ad69a9d0a404fce788abadf0eee0","author":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"date":"2026-06-10 12:35:56.000000000","message":"Patch Set 11:\n\n(1 comment)","accounts_in_message":[],"_revision_number":11},{"id":"c860dae640e3073f384372db8ae5db55101b4786","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-10 13:14:03.000000000","message":"Patch Set 11:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e906a275d7e94c8eae2b46a8d0b051bf\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/a7c428e542ab4ce9929ffe0d201d75c8 : SUCCESS in 10m 02s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/79f28456febe46bbad0f6bb76213b274 : SUCCESS in 8m 48s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/d3eb1e80f6a14d3ab605d1b60b217ff9 : SUCCESS in 8m 20s (non-voting)","accounts_in_message":[],"_revision_number":11},{"id":"d259986a292ff572708b0df7cec0b1707955b4bc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-10 15:15:43.000000000","message":"Patch Set 11: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/27c48adc5cf34ae7a551df8ff8ff8c76\n\n- grenade https://zuul.opendev.org/t/openstack/build/f06d3c9ece8c4d2487a99d520485a69e : SUCCESS in 1h 01m 22s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/390bd2c2c88941349f6b116f99a6aee9 : SUCCESS in 1h 11m 45s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/7512b94849734e70a9ccd9b8d96dd49a : FAILURE in 1h 23m 43s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/db1b43233f2f45eebd0a6c69f498d78c : SUCCESS in 52m 50s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/faa3eb56b12848808a9bde37e36b5b9f : SUCCESS in 5m 11s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/15d8083bfac3474095bc33f0974e545a : SUCCESS in 5m 07s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/dafe0e578c994255a584b7089dfca966 : SUCCESS in 3m 50s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/abb4195c980242879d01cef445e85e05 : SUCCESS in 8m 23s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/410f4ff14fe6488792239cd309d00587 : SUCCESS in 6m 19s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/06dc6e926a5a461db0081c1af48a3e67 : SUCCESS in 9m 14s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/9928dc9277df4f9d93c2a983c302638c : FAILURE in 7m 55s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/2630caa559544ab8938c6117d732e7a3 : FAILURE in 6m 53s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/5ea220e7fbab4f15a8411678c2b40b90 : SUCCESS in 2h 26m 39s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/54df33b8d7f44d769f5ca6410c86b6a5 : SUCCESS in 1h 15m 10s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/e2e76d8ef5c7417f845aa6eb3dd11bdf : SUCCESS in 2h 33m 32s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/5d3001a8f26f436280d4dcbd36741596 : SUCCESS in 2h 19m 11s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/d1d6c9e185164c359499c8900a6ffc46 : SUCCESS in 1h 08m 49s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/6150e13f1bc741fba9ddb2d852f2d894 : SUCCESS in 1h 41m 29s\n- glance-grenade-centralized-cache https://zuul.opendev.org/t/openstack/build/384b30664b1c44d8a3a9c69299cbfc56 : FAILURE in 21m 32s (non-voting)\n- glance-s3-multistore https://zuul.opendev.org/t/openstack/build/27b4e6fefabf407892431ed8df30e261 : SUCCESS in 2h 07m 40s (non-voting)\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/a2ea320bf26b431b83e1ab217756d9b4 : SUCCESS in 42m 24s\n- glance-rally-task-uwsgi https://zuul.opendev.org/t/openstack/build/bbe725bb281f4a198e1e1af99e2f07cf : SUCCESS in 56m 42s (non-voting)","accounts_in_message":[],"_revision_number":11},{"id":"cb72a23649f6e338a788d525b261e2d7acd1cc33","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"date":"2026-06-10 15:23:27.000000000","message":"Uploaded patch set 12: Patch Set 11 was rebased.\n\nCopied Votes:\n* Code-Review+2 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":12},{"id":"221bbfc80dbf4b5a8738a9585e640466391e1642","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-10 17:03:35.000000000","message":"Patch Set 12:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6790d7d4ea024242a167eda8ee80d859\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/f7650c39b68f42fdb5379500d8703657 : SUCCESS in 6m 49s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/a1fa707a58c84d0ba29d8b3a352a7943 : SUCCESS in 8m 34s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/505a121596484b5ca9b12736904c64cd : SUCCESS in 15m 34s (non-voting)","accounts_in_message":[],"_revision_number":12},{"id":"85a30e04a50941231b1cf591b9433d4467d55dc7","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2026-06-10 17:45:14.000000000","message":"Patch Set 12: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":12},{"id":"5bc6ab4a2794e050b77bcee1cb8d151e8fc60bec","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-10 17:59:00.000000000","message":"Patch Set 12: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8a43d22987014bba9ef673046471a3b4\n\n- grenade https://zuul.opendev.org/t/openstack/build/f23f19ce295b4f83ba7af82053b19bfd : SUCCESS in 1h 06m 49s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/3702e692ed5049719d954329b84f070e : SUCCESS in 1h 11m 17s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/17b641112c144e8e9dfae4a8450f6288 : SUCCESS in 1h 59m 55s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/6c38a17c1c934ad5a607ddad02a1e9a8 : SUCCESS in 1h 09m 41s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/65e6e73807844f5cb1e1f097a1105801 : SUCCESS in 5m 10s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/4f61ae07be964656af8d288d0ab9d3fa : SUCCESS in 5m 50s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/2fd5823e32d94181903fbd5cfdba898e : SUCCESS in 6m 33s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/76efab2bc84f41f394d135158c86295b : SUCCESS in 9m 26s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5ce9d0042eda4f8cba349c85c14be128 : SUCCESS in 7m 41s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/4a141ba0923c48a8aa26c0dd65a1ee1b : SUCCESS in 9m 06s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/b66389b136a149c7838d837c75a9b4b5 : SUCCESS in 9m 23s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/67fc539314f4475e888ce1c635f58289 : SUCCESS in 7m 07s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/54a2719397664905a990c3433cf352ed : SUCCESS in 2h 00m 30s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/2b2180cf656a477e9b46694019d4c67f : SUCCESS in 2h 18m 13s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/b4ea59dd17dd426db3f2e6658d5b91ef : SUCCESS in 2h 05m 51s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/15e400c422444881b361540cda257e78 : SUCCESS in 2h 22m 47s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/548ad15a35a4403cb3c8feb6ebc72310 : SUCCESS in 1h 03m 57s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/d391bc34084442d8819a51b3f1f31908 : SUCCESS in 1h 59m 33s\n- glance-grenade-centralized-cache https://zuul.opendev.org/t/openstack/build/522ee40c53374ad2a080f72e986aaf31 : FAILURE in 16m 56s (non-voting)\n- glance-s3-multistore https://zuul.opendev.org/t/openstack/build/55032bba81f245f78cd021baa8fdb6b8 : SUCCESS in 2h 06m 26s (non-voting)\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/380899c4d7064fcf85899b60b419fe24 : SUCCESS in 38m 40s\n- glance-rally-task-uwsgi https://zuul.opendev.org/t/openstack/build/5ddfb66244e14afbafe0a583b3ddad71 : SUCCESS in 58m 07s (non-voting)","accounts_in_message":[],"_revision_number":12},{"id":"e3498f1d894d0a2bf625a35dbd3386d00e18532e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-10 17:59:36.000000000","message":"Patch Set 12: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":12},{"id":"250178839597cc01e843e579a3a8d972c8924f65","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-10 19:48:17.000000000","message":"Patch Set 12: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/48ff32b950a04a05b9fdc69fe11ca43d\n\n- grenade https://zuul.opendev.org/t/openstack/build/aeb26c963af54a42a1d298b4ad7d6e29 : SUCCESS in 1h 03m 53s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/89e9171aeaf7437fa797acd301c661e6 : SUCCESS in 1h 07m 23s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/1541b22dc5984466a009156f27924de7 : SUCCESS in 1h 12m 53s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/bf85a5c2a93c4b8e869e501e75e245a2 : SUCCESS in 1h 00m 52s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c35f00e7ac0e478c9c0e216e6e91a117 : SUCCESS in 4m 52s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/8f2b7b33164a4d7a980d10d2e497a84a : SUCCESS in 6m 43s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/682d2ed1eeff42bd8fc8b7b709d7b267 : SUCCESS in 5m 11s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bfcf2878f35547d48798df4052a05c57 : SUCCESS in 6m 04s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/0e9e3f024c8f4560a27bf5c40370229c : SUCCESS in 5m 00s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/64c63059799a43d8b2fcaa5fccc6ee15 : FAILURE in 7m 37s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/3172021686524da4bbe7e06ae944745c : FAILURE in 6m 50s\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/e0f902d91b924ea0b6b5fca947fd1e0b : SUCCESS in 1h 13m 43s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/f0b464f2cd8a4f6ebbfb85b2818251dc : SUCCESS in 1h 36m 07s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/955ec36dd7a74f4288ce5d2600c2271a : SUCCESS in 34m 58s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/4c60411438b2483291a61a24ef708ca9 : SUCCESS in 1h 44m 47s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/fb1e37ad1b5646308894b6c11c6197d4 : SUCCESS in 35m 23s","accounts_in_message":[],"_revision_number":12},{"id":"1e508262a84bffe7530cfa471a06a5f3ecde5cb0","author":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"date":"2026-06-11 04:54:50.000000000","message":"Patch Set 12:\n\n(1 comment)","accounts_in_message":[],"_revision_number":12},{"id":"f3a29fb7c4e99a6b5d6d71100bd82afa77f1313e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-11 06:52:14.000000000","message":"Patch Set 12: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/1f711bb8d754484eaaf3eceac1b3593c\n\n- grenade https://zuul.opendev.org/t/openstack/build/d2f67a00fb504250b5499fecb271b406 : SUCCESS in 31m 50s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/a8c9bfa1598d42f1b46a47fd595ff229 : SUCCESS in 1h 02m 54s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/4892aee20878441d979c16ab6647874a : SUCCESS in 1h 35m 12s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/614a6d295aaf41689a71c2440b30d334 : SUCCESS in 1h 04m 48s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8015889246254ed192c98d527a2f7ec1 : SUCCESS in 4m 20s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/9a40d925eef944a684ca76e2db0b96fe : SUCCESS in 5m 29s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/08b55196f84541f4b80d90f950f29885 : SUCCESS in 6m 47s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/dcbd89c3530449ba91ac5c2261b2f440 : SUCCESS in 9m 19s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a0f0743477e44750baf79bde24d08161 : SUCCESS in 5m 14s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a0785c06a34044b695b4e9dfed6033f0 : SUCCESS in 6m 08s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/a6e9efad7b6b4adfb23cc5aaf8bd7008 : FAILURE in 6m 44s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/86afa8cdbd644849aebd63437a27562e : FAILURE in 7m 39s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/77ad473eb52346d8903407f8350e2d08 : SUCCESS in 1h 31m 39s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/2901d2633ca84e569bdbb5096e0878de : SUCCESS in 1h 11m 36s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/1b448151d40f4c4a9acd04458aa37be5 : SUCCESS in 1h 15m 14s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/bafaf8cde92b48ee8181f37fb2561db8 : SUCCESS in 1h 49m 22s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/7d0fd4ffb7d04ee897c89fb48ffd2b56 : SUCCESS in 56m 21s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/bc32cbe685474b9e8523c5888791e5e6 : SUCCESS in 1h 39m 49s\n- glance-grenade-centralized-cache https://zuul.opendev.org/t/openstack/build/add41acfa18c4c358b9861e4c948be02 : FAILURE in 7m 18s (non-voting)\n- glance-s3-multistore https://zuul.opendev.org/t/openstack/build/071fa957f94c434fa326d5a23f257ded : SUCCESS in 1h 06m 11s (non-voting)\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/5e1ce44463714fab936e9178ad8c3750 : SUCCESS in 19m 30s\n- glance-rally-task-uwsgi https://zuul.opendev.org/t/openstack/build/15602cad4fcf4a8487d8cc22309b4eef : SUCCESS in 37m 16s (non-voting)","accounts_in_message":[],"_revision_number":12},{"id":"3c9b9ad65d7816ed0bd3d90855d7f04d181368b5","author":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"date":"2026-06-11 08:13:35.000000000","message":"Patch Set 12:\n\n(1 comment)","accounts_in_message":[],"_revision_number":12},{"id":"d92bb61e9eb4eb66a335f3c40ed8cb9108fe4a47","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-11 10:58:46.000000000","message":"Patch Set 12:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/3596877a5159454891c152e673ca6f7e\n\n- grenade https://zuul.opendev.org/t/openstack/build/712ca05f83ba43aa9a42e64838696398 : SUCCESS in 32m 16s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/651fd5882adc463eb1c417168afca08c : SUCCESS in 1h 04m 37s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/a74367b659e143cc8f303529c8b2e05e : SUCCESS in 1h 36m 37s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/5369ee844b7e447b98e0205e9b9a00f0 : SUCCESS in 1h 07m 18s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bef49362dde44006b88ceabf3edcff0e : SUCCESS in 4m 24s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/906022a7eeb749a0964fc317d974affa : SUCCESS in 3m 02s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/fc281694a918412dae59c7322ad09161 : SUCCESS in 5m 10s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/f005b7fc5e154a3485aea5ecfff6c085 : SUCCESS in 7m 19s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/59db78e74f9541eab133306e055a0838 : SUCCESS in 6m 41s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/8150f646bab146d6b5b3e68bde4e650b : SUCCESS in 8m 00s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/bff1344fd7894bdb94ca372b35e7741c : FAILURE in 6m 33s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/e8172eb55b164293acaa741c0daa1d5c : SUCCESS in 5m 57s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/1fa23103ad854695af9d86b07e11e25f : SUCCESS in 1h 57m 06s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/0489b8433ba644aeb0d3d075ae788236 : SUCCESS in 1h 38m 00s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/8cdb0f4b219e429ba86a4389acc9b4e6 : SUCCESS in 2h 12m 06s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/7fe36727405c422c9d401328e1715adc : SUCCESS in 2h 39m 16s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/b2f373ff7d32438c985d5d7ef29495c6 : SUCCESS in 1h 02m 16s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/48301321a6e749e1b9891f77d0fc03b1 : SUCCESS in 1h 49m 13s\n- glance-grenade-centralized-cache https://zuul.opendev.org/t/openstack/build/4d15f6ec012944a6964621eac0d974b1 : FAILURE in 15m 52s (non-voting)\n- glance-s3-multistore https://zuul.opendev.org/t/openstack/build/5ec1b7c39db14f638a5cf402f04b1ca0 : SUCCESS in 1h 53m 20s (non-voting)\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/72e37538991c41c49f363d49dd1f0323 : SUCCESS in 33m 41s\n- glance-rally-task-uwsgi https://zuul.opendev.org/t/openstack/build/246a7df8b09f4aefb0e1035c323eaf83 : SUCCESS in 53m 30s (non-voting)","accounts_in_message":[],"_revision_number":12},{"id":"af4d7f1718c37ebc213cb5f9a86785a87d2418c4","author":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"date":"2026-06-11 12:41:24.000000000","message":"Patch Set 12:\n\n(1 comment)","accounts_in_message":[],"_revision_number":12},{"id":"88e211ebb9d2f776ca3ee54f2ef90c619933a455","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-11 15:17:04.000000000","message":"Patch Set 12:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/ae092c7c697b4ef1927664d413076584\n\n- grenade https://zuul.opendev.org/t/openstack/build/6698dde1a59440d7a7198a32c6e4b69b : SUCCESS in 1h 01m 07s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/f52f3d8f74054161bb0a3756cd16672a : SUCCESS in 57m 10s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/30e8f5ceccb740cc866502e7249770ea : SUCCESS in 1h 38m 24s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/8be18920b49d4e4697dc088687eca12a : SUCCESS in 37m 32s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5a6d564990ad4b27b7dd94454a368fe9 : SUCCESS in 5m 37s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/e198645f2e064666b8be61778627566c : SUCCESS in 5m 12s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/6f62b70da6ed490bbd52ab28e0e55894 : SUCCESS in 5m 41s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/2e8d2ffb118c40f7bf9710a8a1dc12c5 : SUCCESS in 5m 38s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b9dd986d85e94e268b0200d971b75e45 : SUCCESS in 5m 00s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/293a8541c2754c21969eea72ceeec678 : SUCCESS in 7m 35s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/0e164a3a06e94623a8c5d942843341a6 : SUCCESS in 5m 53s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/fdba3bb8e338445382865dd8b8f5fea2 : FAILURE in 6m 28s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/988e5029490e477d9cd61c24e56e48e2 : SUCCESS in 2h 03m 19s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/62f037a70673438a81340ede912b2353 : SUCCESS in 2h 25m 29s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/7ccd2180e46544eab781ef584cc11ba7 : SUCCESS in 2h 04m 05s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/481ffcc8598b45c8b15c5c4aed09bdf8 : SUCCESS in 1h 54m 08s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/28d73848f7a0439981df0019c1e43e6c : SUCCESS in 44m 33s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/1b302b9c1bca48e681238a8ed1ea8437 : SUCCESS in 1h 02m 01s\n- glance-grenade-centralized-cache https://zuul.opendev.org/t/openstack/build/502b2e2df88540399b9768f1acc956ca : FAILURE in 14m 43s (non-voting)\n- glance-s3-multistore https://zuul.opendev.org/t/openstack/build/1f8fd319965a4472aded91b62b211992 : SUCCESS in 1h 43m 13s (non-voting)\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/54e696fc9fd04f4db1ee52e7dc82402f : SUCCESS in 25m 54s\n- glance-rally-task-uwsgi https://zuul.opendev.org/t/openstack/build/7acec45a128a4025a9d2ae2fb356057a : SUCCESS in 45m 02s (non-voting)","accounts_in_message":[],"_revision_number":12},{"id":"00702224ce8126f52fd16ac0d0fb8a736237b552","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-11 18:12:41.000000000","message":"Patch Set 12:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b594df2f922f48fba8fafac4c31b921c\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/35acbf35f4744aa28f6eb8dd954ea804 : SUCCESS in 7m 50s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/ebc09285eb4448899dabd10306641764 : SUCCESS in 7m 56s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/921ea17106f04bad90b7a86036eaa31b : SUCCESS in 13m 33s (non-voting)","accounts_in_message":[],"_revision_number":12},{"id":"ea9b20b400886bfe13d26d2cf7b91ac8eda75c36","author":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"date":"2026-06-12 06:23:08.000000000","message":"Patch Set 12:\n\n(1 comment)","accounts_in_message":[],"_revision_number":12},{"id":"8df35b7e8a4ff9d15324f5566c35333eb404911a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-12 08:51:49.000000000","message":"Patch Set 12:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/a0777817438643908982a3dffcf69915\n\n- grenade https://zuul.opendev.org/t/openstack/build/daeda268023c4b589b7a51edeed2f699 : SUCCESS in 45m 25s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/066dab67a1ac4980b57e7f7b785cfd1e : SUCCESS in 1h 04m 03s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/7aa7e6ff81ac4d9b955152bce8a5cd60 : TIMED_OUT in 2h 05m 20s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/bdaf2b76436c4cbe8579dd4a254eaca3 : SUCCESS in 1h 01m 58s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fb3423f65065417e99a1b5a753069fdc : SUCCESS in 4m 33s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/529ce1eba4284f6ab2526ecbab3c36d6 : SUCCESS in 4m 16s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/66ca2e39e8404baab450841ec39cec57 : SUCCESS in 4m 55s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/260dc7ed5def4f348923526ad5475ef5 : SUCCESS in 7m 55s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/793a3c9dd80c43b1aa965d90e1212b48 : SUCCESS in 5m 31s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/949d4419a394443b814bb1e41b0b8663 : SUCCESS in 6m 29s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/41a67e1eaab645b4bef4b4c572fe85b7 : SUCCESS in 5m 59s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/dbd20a01416b45e48fd0007b11c64d3f : SUCCESS in 5m 38s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/d21b63223b8341a7a62bb7d30a7b12c0 : SUCCESS in 1h 51m 15s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/dcd5a14d5cb24a91825a84d83e71d9a5 : SUCCESS in 2h 04m 17s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/e8dae0298ca241a993814d4608fa860f : SUCCESS in 2h 22m 22s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/35abac4bfbb44e81aa0d0e792609aa8d : SUCCESS in 1h 17m 34s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/bc00877a1bbf4239895a82be5f258496 : SUCCESS in 1h 04m 48s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/939a3191c6f94bc88dfb453a2e84b5dd : SUCCESS in 1h 08m 19s\n- glance-grenade-centralized-cache https://zuul.opendev.org/t/openstack/build/9422a069b34c4585b9ab76add3b2bcf6 : FAILURE in 7m 05s (non-voting)\n- glance-s3-multistore https://zuul.opendev.org/t/openstack/build/346c1bb20a6242a39aac93fbaa2489fc : SUCCESS in 1h 33m 22s (non-voting)\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/69e501e924cc4645a37c33125e202742 : SUCCESS in 26m 29s\n- glance-rally-task-uwsgi https://zuul.opendev.org/t/openstack/build/3bba75fb808b417caf75f01d27167290 : SUCCESS in 57m 36s (non-voting)","accounts_in_message":[],"_revision_number":12},{"id":"1677ac7f4907aa78afc60bf9865425ec8c0e1f8a","author":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"date":"2026-06-12 10:08:19.000000000","message":"Patch Set 12:\n\n(1 comment)","accounts_in_message":[],"_revision_number":12},{"id":"c2f9763f04f751efd0f9c6f3ae537c9142e5a8b0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-12 12:33:29.000000000","message":"Patch Set 12: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f880ab08f30b43cd930ff7bbe9a80421\n\n- grenade https://zuul.opendev.org/t/openstack/build/37ab3a6053b74d0cbb3121ea00998552 : SUCCESS in 1h 10m 13s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/bee9b41abf444037b00de8c57fc9c745 : SUCCESS in 59m 57s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/0233a072c2784d39ae0644a0d042318d : SUCCESS in 1h 37m 18s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/1eedc0cafe914933a2cee8f6db5c9862 : SUCCESS in 1h 08m 41s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cb45748b635447ba8b130d30f204d82b : SUCCESS in 6m 13s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/83935072382d4770bad66233e62bdc0d : SUCCESS in 4m 59s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/ad8b992235734b1382b96be1cc0f36d5 : SUCCESS in 4m 58s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/c57ac6e488a6445f9b9aef6464690b41 : SUCCESS in 9m 56s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bb54799c5d4f4f0c93c2542c609844cc : SUCCESS in 6m 53s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/40da77e2ca534eb4aa316f7b92af4218 : SUCCESS in 8m 15s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/5128f157a19143bb809aa1aed70e1588 : SUCCESS in 7m 08s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/8b19d07515794eb590eafe4224cb81f5 : SUCCESS in 7m 18s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/f8802bdc73db40829416d491672fd0cc : SUCCESS in 1h 02m 11s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/407a0fde70cd4eeeb4586e537979cf8b : SUCCESS in 2h 14m 37s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/4d14aeba393141fb85fac41a313a753a : SUCCESS in 2h 12m 27s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/aa8c30f0077049489b92c297322e7d3d : SUCCESS in 2h 18m 56s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/1497f356a4cc4b7ca5f672e1f553e41c : SUCCESS in 58m 23s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/d7e11bd4684a4883bc80699b38a1b187 : SUCCESS in 2h 03m 20s\n- glance-grenade-centralized-cache https://zuul.opendev.org/t/openstack/build/c80f292d952144b283647b6da80fdc6a : FAILURE in 16m 51s (non-voting)\n- glance-s3-multistore https://zuul.opendev.org/t/openstack/build/3f454c265f6842b3a795965886894f74 : FAILURE in 1h 40m 37s (non-voting)\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/c65f5793a5584bacb39eceb12d55665d : SUCCESS in 35m 45s\n- glance-rally-task-uwsgi https://zuul.opendev.org/t/openstack/build/10bd72d954fc48f19a56587fcab3efe9 : SUCCESS in 51m 09s (non-voting)","accounts_in_message":[],"_revision_number":12},{"id":"98a40f591851af0ecc7f1b31b1e74d25af86d8d8","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-12 12:34:05.000000000","message":"Patch Set 12: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":12},{"id":"df53cbd4340069a268b1e69dfd82621c8aaf4555","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-12 14:44:50.000000000","message":"Patch Set 12: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/f07e492f9aac45cc871160d40055b415\n\n- grenade https://zuul.opendev.org/t/openstack/build/23f24db78e6f4220a2e42e2f7277ac9a : SUCCESS in 1h 03m 18s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/d3db6f7f66d34b7882fee7af78d54dab : SUCCESS in 1h 07m 17s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/b88d69b64d7743ed8dac30f717df926f : SUCCESS in 1h 38m 18s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/e996bb6212454ca2b64f00ffb1142f77 : SUCCESS in 34m 36s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f4d77cef83ea422ab0f9bbc6a072117a : SUCCESS in 6m 21s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/25bf57c126804105beb39c292c171f48 : SUCCESS in 5m 07s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/5106b9feaba54e0e9515aea212c9d466 : SUCCESS in 4m 58s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/672994482dc14944b0b5f284d397f59c : SUCCESS in 8m 06s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/87c031cb04314eea879a01dde0104883 : SUCCESS in 8m 34s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/a15724a5c7ab47bb8150e3a114c9d320 : SUCCESS in 6m 04s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/06b68a7512814c1fb3fe1cdb5f7faeb2 : FAILURE in 6m 34s\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/6df534ab9bda46d7b2f744e3c8821d6b : SUCCESS in 2h 07m 02s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/423ffdb4005142549162d847171768cc : SUCCESS in 2h 06m 15s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/e3fc54a371844ce6aa771abb539f3443 : SUCCESS in 36m 11s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/f0a82c4ea04a413a80b90fd9bf0c382c : SUCCESS in 1h 42m 43s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/a1b54fc5d4f64b599124c4ff90e3254c : SUCCESS in 32m 49s","accounts_in_message":[],"_revision_number":12},{"id":"5546df4f6d884a7cb74a62a6be877888ee159a9f","author":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"date":"2026-06-13 16:53:09.000000000","message":"Patch Set 12:\n\n(1 comment)","accounts_in_message":[],"_revision_number":12},{"id":"cc6952033f1f858cf8cdbaab98fb08ed75fcdf1a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-13 19:14:25.000000000","message":"Patch Set 12: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/527a337a242f4a6c825de8de091945d4\n\n- grenade https://zuul.opendev.org/t/openstack/build/3f88f741177e471990f17bfa60e2cea5 : SUCCESS in 1h 09m 26s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/702db3edbb1e429f99f054b6158f3dcd : SUCCESS in 45m 15s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/04a4d53d7ce54d6facb121245470d603 : SUCCESS in 1h 12m 34s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/c51c784f8b9944c9bf9571e4ff7339c4 : SUCCESS in 1h 05m 04s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f1a9c04d24aa44c9a0e8cba98a17f290 : SUCCESS in 4m 09s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/a84d5bc0800b4ab38e0f7fbfcfe75550 : SUCCESS in 5m 16s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/5001743f8aa5431594fcfd95fc57ede8 : SUCCESS in 2m 51s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/b23c489e52c341a09773e87be63d8edf : SUCCESS in 6m 43s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ab9d883a8c124e81ae7f6079fe338464 : SUCCESS in 5m 40s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/dba9e47e06c94c6cb44a6b593e460068 : SUCCESS in 7m 43s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/c309b84056754923814dcbd24fdce0b5 : SUCCESS in 6m 32s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/377049d50dc349889c5928fc3d649548 : SUCCESS in 5m 28s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/3cf837330bce487f85c67f4b84de816e : SUCCESS in 1h 23m 17s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/47cf91ad1d1c4bbfb53a61a85040fc8a : SUCCESS in 2h 15m 03s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/f47265b38bae47d79410d1158c11c0e0 : SUCCESS in 2h 07m 24s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/8ad25709b5af4cdebc37533bd7842f8a : SUCCESS in 2h 18m 43s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/04297e92616544cf9f5bbe7cecb7440d : SUCCESS in 43m 05s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/32c358594ca44441aef6241ad2ae6e8a : SUCCESS in 1h 43m 36s\n- glance-grenade-centralized-cache https://zuul.opendev.org/t/openstack/build/51b20ad41e89464ba2348f1cd5c253c6 : FAILURE in 15m 58s (non-voting)\n- glance-s3-multistore https://zuul.opendev.org/t/openstack/build/9c185e807ece47e983b8b08511b2afbc : FAILURE in 1h 14m 34s (non-voting)\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/8ef6303e8c0142fbac80b9f402d44491 : SUCCESS in 16m 45s\n- glance-rally-task-uwsgi https://zuul.opendev.org/t/openstack/build/8690bf21fcd04cb28729ddf16f2be7e2 : SUCCESS in 53m 32s (non-voting)","accounts_in_message":[],"_revision_number":12},{"id":"d9033ab8824186633bd57ec153c0bf47cffa2a31","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-13 19:14:58.000000000","message":"Patch Set 12: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":12},{"id":"2cf33103c362eae29bd98a77a4894d732668439d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-13 21:19:38.000000000","message":"Patch Set 12: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/5fa24587f5a940b583a7f0a22f1633fa\n\n- grenade https://zuul.opendev.org/t/openstack/build/747a28f5525040e097f211fc9d4eaed1 : SUCCESS in 1h 03m 41s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/9b7a2741198f4928ba78fad9b448c467 : SUCCESS in 1h 04m 02s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/3bb9c44ac44747eda4e6a9e52bd3f7a1 : SUCCESS in 1h 42m 42s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/e7ff36b769174c94b83cf880a42a3550 : SUCCESS in 1h 06m 43s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/32004e9bb1024c5c86b1fd2cb9918ed8 : SUCCESS in 3m 42s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/38d3d90c46eb467589e1f060f6587692 : SUCCESS in 4m 52s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/94e9fd208f3841c99f65840becc2ed72 : SUCCESS in 4m 27s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8094e9bc826845a1a755a82239d04084 : SUCCESS in 5m 23s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/5d847c5ccfec4745b393c7ac72d29ae1 : SUCCESS in 8m 00s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/b68086eef90240109bf0034d47cc4c69 : SUCCESS in 6m 21s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/1e9a07ee48ef453186037e3350e741c9 : FAILURE in 6m 38s\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/5a1daa2df4f04b2cb948994691a1d0fa : SUCCESS in 1h 55m 29s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/fe5628cf2ca343cba514598686abdee2 : SUCCESS in 2h 00m 05s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/bf963dcebefd43b6ba8af47036fed953 : SUCCESS in 1h 02m 09s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/83b8bea6421144dd81b354c8a303a32d : SUCCESS in 1h 42m 36s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/205ff888bde849a380030bc62ad6c568 : SUCCESS in 23m 12s","accounts_in_message":[],"_revision_number":12},{"id":"f6995d52c5109fe9b68f84015ecee0a9d5522a79","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2026-06-14 05:34:11.000000000","message":"Patch Set 12:\n\n(1 comment)","accounts_in_message":[],"_revision_number":12},{"id":"b35494584ef7c45770c5e7cc7862ff7d35041e81","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-14 07:41:22.000000000","message":"Patch Set 12: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/988a8b12325a4566a7277159b545f476\n\n- grenade https://zuul.opendev.org/t/openstack/build/dcc55432b79144c1a9321e65cc5187bf : SUCCESS in 43m 11s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/644c42026374423d826c6da0533f7452 : SUCCESS in 1h 05m 06s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/ed0aa2ce749d404b83521ef3a64db24e : SUCCESS in 1h 46m 49s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/f4fdd6a9af6f496ea99cd9df1813a0e8 : SUCCESS in 1h 08m 17s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9d42bf5445854d20902f26f37ff9a721 : SUCCESS in 4m 34s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/b0558a11a86c416aa06b138560d3fc74 : SUCCESS in 4m 23s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/620e4969935942c88901876e32617126 : SUCCESS in 4m 31s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/c19cf11ecc61418a82f14220509fa5cd : SUCCESS in 7m 02s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a287dabcccf04207bfcc4509260e662d : SUCCESS in 5m 24s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/46866d9c99b24e33acd510affed9b40c : SUCCESS in 7m 12s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/a78c15965ec141a09be5bcd02e6a22ca : SUCCESS in 6m 15s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/c889bc0a91104595b6ebdb3b6a0c3626 : SUCCESS in 6m 21s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/c8528fad81f04cf593bd76fe5daa29f0 : SUCCESS in 1h 47m 45s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/d4fca81c5b814fbebe4508a2cf34ab23 : SUCCESS in 2h 02m 04s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/d09508e152004a9699cec0b3923fb80d : SUCCESS in 1h 06m 34s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/cbf49792f902419a85b976cbe7553db5 : FAILURE in 1h 53m 06s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/8dde9367c35348f983d3e8ae1e28b56b : SUCCESS in 1h 04m 00s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/d9999b1d3e244f5dbb03d1489cf219e3 : SUCCESS in 1h 17m 35s\n- glance-grenade-centralized-cache https://zuul.opendev.org/t/openstack/build/b8601a349268400887a3b95d91724c6d : FAILURE in 9m 14s (non-voting)\n- glance-s3-multistore https://zuul.opendev.org/t/openstack/build/1d7921d8f99340339ba9ad9bf9ccbd8c : SUCCESS in 1h 03m 37s (non-voting)\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/c52a3f725383443aa9193f1de11b86df : SUCCESS in 36m 11s\n- glance-rally-task-uwsgi https://zuul.opendev.org/t/openstack/build/060f6e31e98d4e738835d82853d58f96 : SUCCESS in 54m 34s (non-voting)","accounts_in_message":[],"_revision_number":12},{"id":"80d4fca93aaf41753264236f58172fec31f53052","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2026-06-14 08:06:15.000000000","message":"Patch Set 12:\n\n(1 comment)","accounts_in_message":[],"_revision_number":12},{"id":"c43efbdd7cc3b5a5c139f9d9d098366b6349fca2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-14 09:53:52.000000000","message":"Patch Set 12:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/e8255e14d1f240ce9d8bc91e8e9b76e8\n\n- grenade https://zuul.opendev.org/t/openstack/build/fafc950d7cf14cddb93c2a85a2d92477 : SUCCESS in 31m 50s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/845ac0b140a8439ca18f235e3bcde0da : SUCCESS in 30m 53s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/8a3b94e35f4340c5aab6ecf08f9fb3dc : SUCCESS in 1h 15m 54s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/52a01adb4ce2457d833f471181f802fb : SUCCESS in 45m 18s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d717d973433d454e837ac23ae483b16a : SUCCESS in 3m 00s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/ac178d6cff2e468d8cf9b9d29e91c147 : SUCCESS in 3m 07s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/40d2210cb32140fda1fc8e6801b25679 : SUCCESS in 4m 36s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/b68324b8b66640cc8548dd85f6419eca : SUCCESS in 4m 14s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ebfee7cf5fcc4c99b6d30ca21876904b : SUCCESS in 3m 38s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/cc7c752b1d7049bb841a6fee6db8f7dc : SUCCESS in 4m 12s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/2bd999f808184cda93173d5077849def : SUCCESS in 4m 47s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/e0097ed8207d40ac8e0a8d7a43c156e6 : FAILURE in 6m 17s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/46dd4d1de55241578666282597c5afa6 : SUCCESS in 1h 08m 15s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/3379bb1a5fa9470c941eaea699d9cdc7 : SUCCESS in 1h 12m 37s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/0ce658a7bbe046a995b05460fa9856ab : SUCCESS in 1h 07m 22s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/68f486320f004e168fbfeb0b37c232a0 : SUCCESS in 1h 46m 24s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/1e7889ac8c204b7cb0bc185e376093a8 : SUCCESS in 41m 39s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/6e4fae223d654047902030a252ec1cb2 : SUCCESS in 1h 01m 36s\n- glance-grenade-centralized-cache https://zuul.opendev.org/t/openstack/build/4e2b15dab15a4655a8d3653e375d4e94 : FAILURE in 7m 55s (non-voting)\n- glance-s3-multistore https://zuul.opendev.org/t/openstack/build/b1cabed0f0c148c9be887cf42abc5330 : SUCCESS in 1h 02m 30s (non-voting)\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/b4f7f6513f774175ae9eb5dd49cc1f2d : SUCCESS in 18m 41s\n- glance-rally-task-uwsgi https://zuul.opendev.org/t/openstack/build/dc7b959b7b5b4177a46daed20002393a : SUCCESS in 36m 29s (non-voting)","accounts_in_message":[],"_revision_number":12},{"id":"1bafe325554e865501786a8ecda8675c0ca8bc09","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2026-06-14 13:05:14.000000000","message":"Patch Set 12:\n\n(1 comment)","accounts_in_message":[],"_revision_number":12},{"id":"295cdfe21b91a2e0f35de381a5166d64b15a60fb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-14 15:55:31.000000000","message":"Patch Set 12:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/c4791e168c2c4f15b08f0bbe3ebfcfae\n\n- grenade https://zuul.opendev.org/t/openstack/build/4ea3d52444f04379b0887902eb33c967 : SUCCESS in 1h 02m 51s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/1bb203f6c14e4fcf8186d4bc04f57110 : SUCCESS in 1h 07m 28s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/b3afa3a67d034d2da12f08a0020b9ba5 : SUCCESS in 1h 37m 52s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/8d060cb2b5fd44469e60a8e5aa672014 : SUCCESS in 45m 32s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3fd0265359544043b00ff30616bda06b : SUCCESS in 4m 29s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/f5220a8b08ff43df94bd4b1066dd2782 : SUCCESS in 4m 38s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/cc9071a49caa411eb8719c4ed330b5ce : SUCCESS in 5m 21s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/3a8ec71b49714570a31c33a846939ffe : SUCCESS in 4m 42s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7a365a143c6643d6ae288a4a205923e0 : SUCCESS in 6m 31s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ad37d1a6c86b4c37831701516b2e709b : SUCCESS in 7m 35s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/82004de77a134366adf1f0880927557c : FAILURE in 6m 15s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/1b7b5d7518544074a072ddb25f5acf1e : SUCCESS in 6m 02s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/22c85a418eed42fd91638ea291c2ba5a : SUCCESS in 57m 38s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/377a3c1ccb094ec8a88d7b22a9f38a5e : SUCCESS in 2h 20m 26s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/2bccf6aaccd04747a70a23de4b595888 : SUCCESS in 2h 42m 05s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/d7dac14e67c9482c85752ae5f83616f4 : SUCCESS in 2h 12m 31s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/3d288507c9db42cf8625f6128d736651 : SUCCESS in 1h 02m 51s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/fb97cfd3bd87468c917b59b95b01503a : SUCCESS in 1h 39m 15s\n- glance-grenade-centralized-cache https://zuul.opendev.org/t/openstack/build/53027ebf11eb434385209aa461d59475 : FAILURE in 17m 11s (non-voting)\n- glance-s3-multistore https://zuul.opendev.org/t/openstack/build/2499bd74491a4340bfe42ec5d3ec5f04 : SUCCESS in 1h 49m 40s (non-voting)\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/8960f8cfeb374b1397d26f3630072e2f : SUCCESS in 35m 48s\n- glance-rally-task-uwsgi https://zuul.opendev.org/t/openstack/build/17c8a9b2edb249b39dac25d796d980ab : SUCCESS in 37m 15s (non-voting)","accounts_in_message":[],"_revision_number":12},{"id":"335cec783497104ca8e09783974349dc8756835f","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2026-06-14 17:42:39.000000000","message":"Patch Set 12:\n\n(1 comment)","accounts_in_message":[],"_revision_number":12},{"id":"a2c78a27978da517aae86392535597e919cb7180","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-14 19:59:42.000000000","message":"Patch Set 12:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/394105d0395446438c21e9142eea4de2\n\n- grenade https://zuul.opendev.org/t/openstack/build/ca87e40804844594b6f85b8a6eec195d : SUCCESS in 32m 34s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/46f74e68029146d88cc27ccdb791d289 : SUCCESS in 42m 57s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/ff8d3b356c914c199f1979dc78c224dc : SUCCESS in 1h 34m 55s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/f7540cc9f1424732a13e59f72ab71161 : SUCCESS in 33m 54s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8fee7f520c574d8784e09d2e4340ee36 : SUCCESS in 4m 25s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/e978c6947bc845d2ae590b7b4dd84291 : SUCCESS in 3m 24s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/ec312419944b4846b057693945d6eb20 : SUCCESS in 2m 48s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/1dfb35ec51ba4dda8b7456a5889afa4e : SUCCESS in 9m 16s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/eb7a706aa1974492b921bdb79946c30d : SUCCESS in 5m 36s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/9b1e8fa302cf4105be19f14e57390f8c : SUCCESS in 7m 55s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/ff773ea564454344b222e91a1da813c2 : SUCCESS in 6m 07s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/8a9ef48e47b14c32b64f10621b8887d7 : SUCCESS in 9m 18s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/130151439f804ef2b20a9bb33e7f4d20 : SUCCESS in 59m 58s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/96e7783c42dd4321b0ae61d9b5cacb93 : SUCCESS in 2h 01m 49s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/75c5142c1e5d4b3b9b97e02a2a6932eb : SUCCESS in 2h 09m 44s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/43204a6ccb8d48f180e2562d380a50fb : FAILURE in 1h 59m 40s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/77e2a08072bb4d8d9889f73ecbf9cff8 : SUCCESS in 59m 15s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/b7dfa2832c79480aa9baf51541cd8c6c : SUCCESS in 1h 19m 40s\n- glance-grenade-centralized-cache https://zuul.opendev.org/t/openstack/build/1f461773b54b474b82ca4aac15ed6d6c : FAILURE in 15m 51s (non-voting)\n- glance-s3-multistore https://zuul.opendev.org/t/openstack/build/0661827a7b5147ddba7520edf20b452c : SUCCESS in 1h 57m 12s (non-voting)\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/de25982c867a43f8a6bcc44528d3f414 : SUCCESS in 16m 01s\n- glance-rally-task-uwsgi https://zuul.opendev.org/t/openstack/build/60a01b182256400da7d67766c75fe3d5 : SUCCESS in 26m 02s (non-voting)","accounts_in_message":[],"_revision_number":12},{"id":"fa4eeb6cdc0863c66755a99cad27ee131abd2ef8","author":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"date":"2026-06-15 00:27:22.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"ac2a34c5cfc18cdec38b86772ae5ecffa95228bb","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2026-06-16 05:46:19.000000000","message":"Patch Set 12:\n\n(1 comment)","accounts_in_message":[],"_revision_number":12},{"id":"7c0a3f3c35734b268d2b0ac586a44fa4f16c3112","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-16 05:46:24.000000000","message":"Patch Set 12:\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.\nWarning:\n  Error merging gerrit/openstack/glance for 975118,12","accounts_in_message":[],"_revision_number":12},{"id":"a511afb8b241a55426a73e20c8dd8036a07631dd","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"date":"2026-06-16 07:18:32.000000000","message":"Uploaded patch set 13.\n\nOutdated Votes:\n* Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified-1 (copy condition: \"NEVER\")\n* Workflow+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":13},{"id":"5f157fe0b019200c8453c4f15173cf8a32ffe859","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"date":"2026-06-16 09:22:13.000000000","message":"Uploaded patch set 14: Patch Set 13 was rebased.","accounts_in_message":[],"_revision_number":14},{"id":"bb992889431a4cb12aab01e7dd3c3483909ceca4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-16 11:39:17.000000000","message":"Patch Set 14: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/fa8c5e060a0a416cb94e6c8f5c32ddce\n\n- grenade https://zuul.opendev.org/t/openstack/build/fda68100cb3e4cf8abea5f6fc99990bd : SUCCESS in 1h 11m 34s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/65fb55684d4b4780970a87f5f29a52b0 : SUCCESS in 1h 06m 38s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/0361c56175ac48eab870c3535b80e2fa : SUCCESS in 1h 37m 11s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/ec379111f64c479b9df9f9f48e7a422e : SUCCESS in 1h 07m 23s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c10bc3f8dc23422983ac2e75de783abe : SUCCESS in 3m 04s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/c0330205e3a94517a3316802cbe14e91 : SUCCESS in 5m 33s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/7c34e3dbb9c6450eba1327807b154bed : SUCCESS in 4m 43s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/7cb3b0dcca2d413096d7efac1d087ab3 : SUCCESS in 4m 26s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/45d6f1de51ec48baa1da29e2acf97b54 : SUCCESS in 5m 37s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/051f123a78164932831621b03e2308b8 : SUCCESS in 4m 24s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/892270794cd74aec802f875038d8784c : FAILURE in 9m 01s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/fe0e10072482427492894fed693db4aa : SUCCESS in 6m 08s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/390e9b126aa94df582ddfb378b947066 : SUCCESS in 1h 51m 29s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/93a130b8d5304f2cb4fb96a05bae810d : SUCCESS in 1h 19m 40s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/71b5c86213fe4adea392d6e820f9f258 : SUCCESS in 1h 59m 21s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/80dbd9f269e64e7a9630ae6514df6d7e : SUCCESS in 2h 09m 33s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/184ecf7907c847009f7fb6f17dca69f7 : SUCCESS in 48m 18s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/f22ce6588acc4b5bbb030903683f5f24 : SUCCESS in 1h 44m 55s\n- glance-grenade-centralized-cache https://zuul.opendev.org/t/openstack/build/459120dde7044f7b8ceea53acf55a6a1 : FAILURE in 8m 19s (non-voting)\n- glance-s3-multistore https://zuul.opendev.org/t/openstack/build/78d9959cf50b431db6c34e227e758407 : FAILURE in 1h 55m 42s (non-voting)\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/39ca283271024bbb9862b415bc41eeec : SUCCESS in 40m 04s\n- glance-rally-task-uwsgi https://zuul.opendev.org/t/openstack/build/672f825ea8f4432d93061e6e20bcfad9 : SUCCESS in 56m 34s (non-voting)","accounts_in_message":[],"_revision_number":14},{"id":"44a16206478549ccce3151fd1ea54e720f9fb7e4","author":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"date":"2026-06-16 11:40:23.000000000","message":"Patch Set 14:\n\n(1 comment)","accounts_in_message":[],"_revision_number":14},{"id":"8bb71d8e85a85935e39c107d0d1229486fb4094a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-16 16:44:07.000000000","message":"Patch Set 14:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/d671bee0fa404db3b49aedd78a97df02\n\n- grenade https://zuul.opendev.org/t/openstack/build/f338df4c9acb4b90960006aadb4f018c : SUCCESS in 1h 33m 52s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/cbf36cec97d44b95bba2d55313db605e : FAILURE in 16m 24s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/b70bdbacdeb9458fa62fa711c2b69d5d : SUCCESS in 1h 44m 23s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/77c112bd7e8143078505cd11b72e6e66 : SUCCESS in 55m 07s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/29fff3f6c7a845d9838111da93ec25dc : SUCCESS in 4m 55s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/2667e21e48a84ae1849bf24648b43310 : SUCCESS in 6m 18s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/05db3c0748a3442193f068fd1e5efd32 : SUCCESS in 4m 49s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/11404b0002954279a906bb5add606fa1 : SUCCESS in 8m 11s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d1e558832caa43b4b3335d945b2ea13f : SUCCESS in 3m 39s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/242c16f2077e448a999562219b024c64 : SUCCESS in 8m 30s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/9418afc0e94f4b3986a071aecce3dbf3 : SUCCESS in 8m 12s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/9ea28ebdfe1c409585896b03d6917e73 : FAILURE in 7m 26s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/75a14772bfe14abf9700b9f1ea25ad55 : SUCCESS in 1h 05m 22s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/3271bb64fc644e5c9657164e717a5d31 : SUCCESS in 2h 09m 22s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/f82aa38205694d9b95a3f89659c51382 : FAILURE in 1h 45m 45s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/4f9d9a43c03e498bbca8d25a00cff3d0 : SUCCESS in 2h 34m 14s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/6ea75e36215643438c9f09f3475957b7 : SUCCESS in 48m 36s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/26f6e3cc4235453784afb20d47f21fea : SUCCESS in 1h 45m 57s\n- glance-grenade-centralized-cache https://zuul.opendev.org/t/openstack/build/e6ecaca8108644d6ab78b85c4e7e06b4 : FAILURE in 15m 44s (non-voting)\n- glance-s3-multistore https://zuul.opendev.org/t/openstack/build/0d75612cedaf4bae88d7885f9b2d720e : SUCCESS in 2h 00m 55s (non-voting)\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/9c52dc3c417f4b3bb61ec4259f337491 : SUCCESS in 35m 22s\n- glance-rally-task-uwsgi https://zuul.opendev.org/t/openstack/build/6fde6f6131f84fedb98614bdb24d8ca3 : SUCCESS in 1h 02m 16s (non-voting)","accounts_in_message":[],"_revision_number":14},{"id":"3030564c6ca80149d75a2938ec582bd93bf88918","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2026-06-16 16:59:50.000000000","message":"Patch Set 14:\n\n(1 comment)","accounts_in_message":[],"_revision_number":14},{"id":"30c39a6a3241f559805e7b932946531dfc941c99","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-16 19:13:13.000000000","message":"Patch Set 14:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/18ba534f35ac450187552907c2493059\n\n- grenade https://zuul.opendev.org/t/openstack/build/3f98b5cb4f5c4a798e967995b2efecc1 : SUCCESS in 1h 03m 49s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/69a43d1630e94a15add71558273587c5 : SUCCESS in 1h 21m 15s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/a68e84f6598d45958e2b441e54689f59 : SUCCESS in 1h 58m 37s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/2335217c65ca44d8b741462b88dbccd2 : SUCCESS in 1h 30m 47s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/db8084ad680146ffad13791b0f034f37 : SUCCESS in 3m 30s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/503a2160b8da47658716734a8507f853 : SUCCESS in 5m 09s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/cf4034edcdd54ba984f6f209ea2c837b : SUCCESS in 4m 03s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/7a19bd5960df46308fa7359f09987742 : SUCCESS in 5m 16s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4f19222710aa41f789a8048f4d8801f1 : SUCCESS in 5m 36s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/4c752291f14749b897825fac8b59a9f0 : SUCCESS in 8m 45s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/5cd42750f2374e4f975b89ca55f0ea54 : FAILURE in 6m 11s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/6af95cc76680420196a3cd7fd2c16654 : SUCCESS in 4m 30s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/862a4080d83e4435957a7c696b158cd1 : SUCCESS in 2h 05m 08s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/50159ecac0324ba8b0db0097fe89f145 : SUCCESS in 1h 58m 35s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/d60ee4cb071340c49e4deae023c7c6ae : SUCCESS in 1h 50m 07s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/247812f1108a4443afffafab36ad0614 : SUCCESS in 1h 14m 01s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/7df90d1ed2b84886a67d57b499ffc7a4 : SUCCESS in 1h 20m 31s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/ea00f4dd675f476ea64beb7d396d7ac9 : SUCCESS in 1h 28m 54s\n- glance-grenade-centralized-cache https://zuul.opendev.org/t/openstack/build/de945fb6c8cc4d2587e4adf971f8bb0b : FAILURE in 18m 00s (non-voting)\n- glance-s3-multistore https://zuul.opendev.org/t/openstack/build/9b71eadae71f442992ef07c1081a14bf : SUCCESS in 1h 57m 57s (non-voting)\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/3d7ce304819c4a7eb1e743fe974bfcc6 : SUCCESS in 24m 41s\n- glance-rally-task-uwsgi https://zuul.opendev.org/t/openstack/build/0522331eee68420c936249424b509820 : SUCCESS in 53m 51s (non-voting)","accounts_in_message":[],"_revision_number":14},{"id":"7c329cebe7aca1f473f341f7e2e4d873bec41030","author":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"date":"2026-06-17 03:33:36.000000000","message":"Patch Set 14:\n\n(1 comment)","accounts_in_message":[],"_revision_number":14},{"id":"ae41e06b3c1cac59af3970882da3e29be54324f8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-17 05:47:09.000000000","message":"Patch Set 14:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/9a50bc43da1748c788efdb69a75776f6\n\n- grenade https://zuul.opendev.org/t/openstack/build/233843c4e390466b875ee84ebf82307c : SUCCESS in 37m 57s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/ffa80f1140a642f9b199900e9db7d4d0 : SUCCESS in 32m 19s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/18a281b5998d4171aa437fa57c64e1e9 : SUCCESS in 1h 47m 48s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/1335e8ff77194e5bb2200506330799a7 : SUCCESS in 38m 17s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/925300d726c94579bf4d8170acb43136 : SUCCESS in 7m 13s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/acf45d7fd8c44e1ebc6c4d0afd06cf37 : SUCCESS in 4m 40s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/1d003727c56a4b028c7a49fcd3cbff7d : SUCCESS in 3m 21s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/9b5a601de34e43ca930bc34684a98e1c : SUCCESS in 9m 30s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2f5e539958cd47eea10330f12da87e9a : SUCCESS in 3m 17s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/18206fc6d23842c3af86836abd6d3583 : SUCCESS in 7m 54s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/825b33122bae4cfb8f0b944e9eec0879 : FAILURE in 5m 02s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/3b34544bda2445679e38f6623a77f657 : FAILURE in 7m 53s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/5508f07d1ea74150b0063d73cb79f8ad : SUCCESS in 1h 44m 04s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/151b1b8ba83e44f8ba97854978f8bddc : SUCCESS in 1h 04m 17s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/e3744583720343fba90ab32ea67bc42f : SUCCESS in 2h 08m 21s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/c7dcddb8c6084d8c8c04f60f534a6d49 : SUCCESS in 1h 17m 55s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/970d4ebcb1214bb18cf34fd88e315a7a : SUCCESS in 30m 14s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/71f46603f44948768e8cff078c07ca38 : SUCCESS in 1h 44m 53s\n- glance-grenade-centralized-cache https://zuul.opendev.org/t/openstack/build/6607ec9a35f84d12a6e2b1a97585e2d2 : FAILURE in 12m 11s (non-voting)\n- glance-s3-multistore https://zuul.opendev.org/t/openstack/build/c94eae4ed7944a3b8ea4cb936f3ba49d : SUCCESS in 1h 00m 44s (non-voting)\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/2dfbb69ffd5341c3a88af3312557e7f6 : SUCCESS in 34m 30s\n- glance-rally-task-uwsgi https://zuul.opendev.org/t/openstack/build/65dea75f97084004978f05324cc9ef79 : SUCCESS in 55m 08s (non-voting)","accounts_in_message":[],"_revision_number":14},{"id":"2a8501a759c90762d770a528657ce03e35d36539","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-17 11:09:25.000000000","message":"Patch Set 14:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/87bbd8a8b738478ca2d40af49e702709\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/66797e2c408b41c5ac504b19c09f1273 : SUCCESS in 6m 24s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/e2337f6908be44cfba06193f9de99c96 : SUCCESS in 6m 59s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/b2da05d94e1142bc816a5d783036f73f : SUCCESS in 13m 23s (non-voting)","accounts_in_message":[],"_revision_number":14},{"id":"090af347ad66c841346df3663648b01fc788ad02","author":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"date":"2026-06-17 13:40:54.000000000","message":"Patch Set 14:\n\n(1 comment)","accounts_in_message":[],"_revision_number":14},{"id":"a290d69ec2eb962db9af0c6a7f521c043b52e991","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-17 15:50:24.000000000","message":"Patch Set 14:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/2b6ea5b762bc4ad4805ddd80fb0ec46d\n\n- grenade https://zuul.opendev.org/t/openstack/build/252fdc6c0ee4415bb5af31fb61741c7e : SUCCESS in 1h 07m 55s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/5a4123403c9f4ac1bcc7085cf6359565 : SUCCESS in 1h 10m 01s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/3b012681eef8436297b0e08a69bcf306 : SUCCESS in 1h 39m 46s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/6e4973b1dc094de9963863eeecf1567a : SUCCESS in 1h 07m 40s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2478ec47f17d45e59fa5c21b9b8d9a49 : SUCCESS in 5m 27s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/f239d871f387496c963fdad32ca84cb5 : SUCCESS in 5m 06s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/bd1aca53ff8c475d83083ae7b748d7db : SUCCESS in 5m 06s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/89aa53168d564b068720e34e4e215768 : SUCCESS in 9m 10s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b0b40b95ba2c403dab156d2ca7ce3a40 : SUCCESS in 5m 48s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/5d6efb85120b4e51aa47802a8f2b0df8 : SUCCESS in 7m 35s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/8052d2f82d1c46189b97534110cb78f6 : FAILURE in 6m 26s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/e5ebbb27d38148dab3d68ec4cc6fb3a1 : FAILURE in 4m 19s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/f2d5025bd23446b09987bcdbb0c3789b : SUCCESS in 1h 13m 57s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/b696125ccf724f21953fc096292e5cb2 : SUCCESS in 1h 58m 45s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/f5d63281317e4b4ebfd3c78c6305ea01 : SUCCESS in 1h 12m 47s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/085e22b031104733bf1656c49b52dca9 : SUCCESS in 1h 21m 21s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/770efa81ab7342518601578b43f075d5 : SUCCESS in 1h 08m 28s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/689dd6117697409786997312f7fce2c6 : SUCCESS in 1h 41m 27s\n- glance-grenade-centralized-cache https://zuul.opendev.org/t/openstack/build/b885df68f1c44fa386635b728d68239a : FAILURE in 16m 29s (non-voting)\n- glance-s3-multistore https://zuul.opendev.org/t/openstack/build/eb969be9cb5e4ebbbf438c4145377639 : FAILURE in 1h 44m 58s (non-voting)\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/4411d1bbd1724363ac9dff89048fd5c6 : SUCCESS in 17m 26s\n- glance-rally-task-uwsgi https://zuul.opendev.org/t/openstack/build/c1640c45aaa54c47914cb5b6087c5f1a : SUCCESS in 1h 09m 54s (non-voting)","accounts_in_message":[],"_revision_number":14},{"id":"7652947222c7aaf438eaab5e61bf9ac5d2965e7c","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2026-06-17 17:17:25.000000000","message":"Patch Set 14:\n\n(1 comment)","accounts_in_message":[],"_revision_number":14},{"id":"2f79774ddb3c2630047ad2cbda174dd409f134cf","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-17 19:47:20.000000000","message":"Patch Set 14: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0b8d6e8767c4419d97560181aa94c4e6\n\n- grenade https://zuul.opendev.org/t/openstack/build/8882adf46d6e4c8ebcb31ff62f312dc0 : SUCCESS in 1h 04m 17s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/65aee13ab25e453a89c9026b9401d4eb : SUCCESS in 31m 29s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/9cd82a1efa94425bb258b4e5e9d54317 : SUCCESS in 1h 25m 45s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/dc3da772a7154be4bd1711c73f7c2588 : SUCCESS in 1h 07m 46s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9c1a5fb85a284228907e7dbf3fc30aca : SUCCESS in 5m 29s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/ac61f9634ba14931a8d1bdfc15a7770a : SUCCESS in 4m 52s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/3335452d1ec7450d87ef6c1d52ac9ceb : SUCCESS in 6m 33s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/d306de98bdd9443e9d304f796d4a9210 : SUCCESS in 6m 45s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bf00e0cd693541b18f1f6aecb940578f : SUCCESS in 3m 47s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ac77b9f6c1844b698bf0a4a8a03c3391 : SUCCESS in 4m 36s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/f003614a6a99494abbda3db6240d2630 : SUCCESS in 6m 06s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/5c5f12e94b964e7aa590de9af74ab68b : SUCCESS in 6m 54s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/d7427974356e4118be348ac0a4cb7146 : SUCCESS in 1h 54m 54s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/1e6ca47b6fd14376a639f736b7d297af : SUCCESS in 2h 20m 40s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/0fa8970bbd644f1c9749999cdd74e93e : SUCCESS in 2h 13m 32s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/4533a7ae39f74e4fa69156ad8845012b : SUCCESS in 2h 24m 23s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/51310e5375e748448edcec003beb51e9 : SUCCESS in 57m 54s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/a001bd754cdc403abfdbbd001ba704f9 : SUCCESS in 1h 39m 02s\n- glance-grenade-centralized-cache https://zuul.opendev.org/t/openstack/build/b3e157533c5142b8a3614b2c9f7977bc : FAILURE in 15m 17s (non-voting)\n- glance-s3-multistore https://zuul.opendev.org/t/openstack/build/61194a499c2741e5967dc1a8d90764cb : FAILURE in 1h 27m 06s (non-voting)\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/49f574cb05794b6389e048415cccf770 : SUCCESS in 36m 26s\n- glance-rally-task-uwsgi https://zuul.opendev.org/t/openstack/build/d1d53d6ff40e4d1381f196477592e557 : SUCCESS in 51m 43s (non-voting)","accounts_in_message":[],"_revision_number":14},{"id":"b37e9023fbd0423fa75bda58a1fa9baca5834472","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-18 00:03:05.000000000","message":"Patch Set 14:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3586f0b007f848b7a5eee27919f2280a\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/8f0f6602794345c9812f3fddbe953a4f : SUCCESS in 10m 12s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/df81d300b8d241a9a36100ca3008d31b : SUCCESS in 9m 48s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/bd7f7b3094ba465f9879d958021696b3 : SUCCESS in 16m 38s (non-voting)","accounts_in_message":[],"_revision_number":14},{"id":"2fcca7d126a02dbcba5c0e8e88fceabba676beb8","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2026-06-18 05:14:06.000000000","message":"Patch Set 14: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":14},{"id":"fa27f0cdf88cd5b3dedebce6b7e4cbf7b21b397c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-18 05:14:44.000000000","message":"Patch Set 14: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":14},{"id":"20542de0502537332407c46713a80c0b37b30138","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-18 07:23:29.000000000","message":"Patch Set 14: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/bdcc5ec1046542c68895b97186f4f650\n\n- grenade https://zuul.opendev.org/t/openstack/build/3e9f337bdd564f20b1eacbe947c2a890 : SUCCESS in 32m 59s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/a62d9c42875f4d9a9d0c441621827e38 : SUCCESS in 50m 12s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/7affc71825d44fb6b84916e0aa8ee3b5 : SUCCESS in 1h 48m 53s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/f03881b5ab604883bd0562648477d80d : SUCCESS in 45m 19s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0f44d1078f784625b1cdc5dcaa3826d2 : SUCCESS in 5m 14s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/8fb14c04374f4d11807a7934ff12fcac : SUCCESS in 4m 35s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/594c4e7606c646cdaced1fbf241b2bcb : SUCCESS in 3m 19s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9391406982eb46ceaf59804049014fc2 : SUCCESS in 5m 43s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/5e817174e3bd44489c0379d827c694b9 : SUCCESS in 7m 29s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/d764db85aa8446f7a5e6279b44f39b47 : FAILURE in 6m 02s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/9d26e8308a8547b084d3a461631d7ba9 : FAILURE in 6m 42s\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/478a3f53c51d48ab91bfee3036dde0a6 : SUCCESS in 1h 40m 34s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/29bf455cd39f4d7a826d181a91a1857f : SUCCESS in 2h 03m 52s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/31da914b184241cda772be7aa9926c47 : SUCCESS in 1h 06m 40s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/62d51a6a9f114d9dbe369369c5f745b5 : SUCCESS in 1h 39m 47s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/b27add60870341f892380ba1a00d328a : SUCCESS in 36m 08s","accounts_in_message":[],"_revision_number":14},{"id":"3457bbed5888f31b09363f22debab58ea13cff4a","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2026-06-18 07:30:14.000000000","message":"Patch Set 14:\n\n(1 comment)","accounts_in_message":[],"_revision_number":14},{"id":"dc3caf8931aacbe3f3da0cbe39c6c3cfc5722589","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-18 09:26:57.000000000","message":"Patch Set 14: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/37fa85723f12456e86437a7c65d9555b\n\n- grenade https://zuul.opendev.org/t/openstack/build/3ba9fbc84486464d8b0eda91334a5ab4 : MERGE_CONFLICT in 1m 03s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/3b79c3c37d494662a8f47bbbeee86026 : SUCCESS in 1h 08m 58s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/bf8a064e90154d27821cc43fcd2f7574 : MERGE_CONFLICT in 1m 04s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/3dc7c7b3a6b940dea995f609b46ab52f : SUCCESS in 1h 08m 57s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/73916be9b12546fda26befe6f4527b8e : MERGE_CONFLICT in 1m 01s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/f5185cbb4cbb4956bd05cfa3fba72ab3 : SUCCESS in 6m 05s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/e47aaace19634d599ce0a2a72e546a11 : SUCCESS in 4m 18s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/49861e4a58c5432fb83be44a28c778e2 : SUCCESS in 4m 49s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/08f484af29db41ff86bee7358745d1f1 : MERGE_CONFLICT in 1m 01s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/456138d6faa54b6c927ca86dd166c7d0 : SUCCESS in 4m 38s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/0ad28353888e45fc94a8d5c72781e2e9 : MERGE_CONFLICT in 1m 01s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/4f16f01af8844a1b974bca022fc788ad : SUCCESS in 7m 47s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/904f92be2c7d42798243efbac993417f : MERGE_CONFLICT in 1m 05s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/689c1a8b8786409faffcf61a61252707 : MERGE_CONFLICT in 1m 05s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/7671efb6e93847a8a7fd8f9928c5d3bd : SUCCESS in 1h 15m 30s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/ab158e137aaf4b698d0828977de3766f : SUCCESS in 1h 21m 37s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/b046c776aef54567ae13cb0e6b65cebf : SUCCESS in 31m 46s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/13071c6908b64b4ca1d04f98d616b79d : FAILURE in 1h 25m 24s\n- glance-grenade-centralized-cache https://zuul.opendev.org/t/openstack/build/da35072c9c614fc1a97f8cfb4eb79f56 : FAILURE in 7m 31s (non-voting)\n- glance-s3-multistore https://zuul.opendev.org/t/openstack/build/83f7a4f1cb7d49d996f10a1784db89d2 : SUCCESS in 1h 46m 45s (non-voting)\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/d478e1db65134a2d9dcf89b09a48cf31 : SUCCESS in 35m 58s\n- glance-rally-task-uwsgi https://zuul.opendev.org/t/openstack/build/753d4d6a3ee0449291d500354656ba5c : MERGE_CONFLICT in 1m 05s (non-voting)","accounts_in_message":[],"_revision_number":14},{"id":"6ca47d411dd60284f006d6161d90f572002cba29","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-18 09:40:33.000000000","message":"Patch Set 14:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4b75f92936b2482e9f9372dd64681c31\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/548d88f6f43843d3abbfbb0386bc20e0 : SUCCESS in 9m 50s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/8f62cd865e34409f9559b20ed27e50f7 : SUCCESS in 8m 44s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/b5f628f15805479caabb74da2319ace7 : SUCCESS in 11m 44s (non-voting)","accounts_in_message":[],"_revision_number":14},{"id":"e7b97b03a08df5ff6109646dac02b9944350eac3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"date":"2026-06-18 09:45:27.000000000","message":"Patch Set 15: Patch Set 14 was rebased\n\nCopied Votes:\n* Code-Review+2 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n* Workflow+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":15},{"id":"b33b1b4defd32ce6c5438fef04802679c21caea7","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-18 11:53:02.000000000","message":"Patch Set 15:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c34f6b114fc34724b2ccf3c1dcb70577\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/a7f23a4fbe264b3fa124867e1143673a : SUCCESS in 8m 27s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/c8ca8dd326ff49ee9c333ece27821b55 : SUCCESS in 6m 46s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/ffdef3075f8147c0a2326268a8b8e3f7 : SUCCESS in 12m 44s (non-voting)","accounts_in_message":[],"_revision_number":15},{"id":"247f0831d4e3c218ac907b833bae53bb0475c9e2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-18 12:19:23.000000000","message":"Patch Set 15: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a836f3ed73ff42f3bf0cfe3005801ba5\n\n- grenade https://zuul.opendev.org/t/openstack/build/4edf41cfdb6e4cb1b5d3db78b30d87f4 : SUCCESS in 34m 39s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/2d0ae57e756f41e3b22391ef948ced74 : SUCCESS in 30m 50s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/ec8dac81515e43caa52ef54fd157a986 : SUCCESS in 1h 22m 36s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/ad36a6c845424018a5437aa206ab22fb : SUCCESS in 33m 28s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/93c498188e5b42629cb2924b7708ac27 : SUCCESS in 2m 49s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/d0a78090df3b46ed9290d71dac35274e : SUCCESS in 2m 44s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/d56dfd402e154cb38ba2a3c56272c782 : SUCCESS in 3m 22s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/527ba9fc81114808bb6627a0703f3f6c : SUCCESS in 4m 56s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4647b8fe97314dd0abfd5c66629085f3 : SUCCESS in 3m 27s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b9bf7853e0d74fd9b8ee377833e0438d : SUCCESS in 4m 18s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/087c70fcd43e4a1b93ec3b1477004e41 : SUCCESS in 4m 41s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/6f7910fe25a64bd891ad802126dc5d25 : SUCCESS in 4m 50s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/26847295813a49c4aa7360593495d380 : SUCCESS in 1h 10m 40s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/79547718d1234faea3d33634fe5c71c1 : SUCCESS in 2h 02m 15s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/02914251af1d481186a6c1acaece8249 : SUCCESS in 2h 28m 26s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/6b7528755bc44c7baf879c35c4f14827 : SUCCESS in 1h 21m 10s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/fff2048f55ca4aca959558f3b5f9a8ec : SUCCESS in 1h 06m 05s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/767019cd65b44fb2b9c727116a4f2a26 : SUCCESS in 1h 02m 04s\n- glance-grenade-centralized-cache https://zuul.opendev.org/t/openstack/build/f05c86159f364a61b01e81d7b2a32daf : FAILURE in 9m 24s (non-voting)\n- glance-s3-multistore https://zuul.opendev.org/t/openstack/build/f20dfe6ed7204e5e81b92d7d5600811b : SUCCESS in 1h 06m 32s (non-voting)\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/5102b5bfa6cd423f8bda01be4a3c359b : SUCCESS in 18m 55s\n- glance-rally-task-uwsgi https://zuul.opendev.org/t/openstack/build/8f0851cf22eb4b5dae4c51f94fb9230d : SUCCESS in 23m 00s (non-voting)","accounts_in_message":[],"_revision_number":15},{"id":"9b4e0b8ccfdf0d26927f0a604237b5e71c08d39d","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2026-06-18 14:04:54.000000000","message":"Patch Set 15: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":15},{"id":"af77c46452d7f6c5fe9d66e9aa37a34f5079d4a9","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-18 14:06:34.000000000","message":"Patch Set 15: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":15},{"id":"e962ddb2fc8dd9e9b845ed1ac90a63454820598a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"date":"2026-06-18 14:23:30.000000000","message":"Patch Set 16: Patch Set 15 was rebased\n\nCopied Votes:\n* Code-Review+2 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n\nOutdated Votes:\n* Workflow+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":16},{"id":"98057d7de4c638fd9c4df5e4a6d5609a2c1d2b76","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2026-06-18 14:24:20.000000000","message":"Patch Set 16: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":16},{"id":"e6c7434d8a3b771333235ebfe8f2b9e49ed94687","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-18 17:11:25.000000000","message":"Patch Set 16: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/4a093453bc0249219f20b6885258c754\n\n- grenade https://zuul.opendev.org/t/openstack/build/ba8009228a0540b99ade81ddde8e5ca4 : SUCCESS in 1h 18m 38s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/26724073dca14844a1f306e228010d6e : SUCCESS in 1h 38m 35s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/21fb01887d104a02a3ce413d14370f24 : SUCCESS in 1h 53m 56s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/5ea8c365321e49078e07b74d82ff71c0 : FAILURE in 14m 07s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/60390c99593e41a4a9dfa8656999f4c1 : SUCCESS in 6m 42s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/b2ed11e83eff478eb8e9892a81ea080c : SUCCESS in 3m 12s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/775c0e9ea35e400eb1ceb8fcb6b87fe7 : SUCCESS in 2m 44s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/e53e56e1e881423da90011f2dd179a80 : SUCCESS in 5m 00s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3fdd46233cfc481d9aec659286cc2596 : SUCCESS in 7m 01s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/4a58ac4c409b4a2da6e1240c85a65de6 : SUCCESS in 4m 20s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/d4d2723a1a764538829a83f21b3a1deb : FAILURE in 7m 42s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/10a93cc5f21d44ee9a4856b309d43d9a : SUCCESS in 6m 22s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/377b22b933f7474780ec4d854178056c : SUCCESS in 1h 10m 59s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/ae1abf06d0464db594247eee25bb5546 : SUCCESS in 1h 57m 38s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/8cce14917a4a4408ba2c77a044231e23 : SUCCESS in 2h 01m 49s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/f6edf6fab5a44f70b174a2cd3caa7166 : SUCCESS in 2h 42m 03s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/71561b9828594525b117a90bf1cd1d5c : SUCCESS in 1h 05m 47s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/37e91794b3f9447c837702b83f656eb6 : SUCCESS in 2h 07m 06s\n- glance-grenade-centralized-cache https://zuul.opendev.org/t/openstack/build/93ac2d19f50e4f4e85fa248af8b4665f : FAILURE in 9m 09s (non-voting)\n- glance-s3-multistore https://zuul.opendev.org/t/openstack/build/8d062a21d0634d37920b135608e43a96 : SUCCESS in 1h 04m 46s (non-voting)\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/9dee89ceba8d4d0ba03536e7c97e6740 : SUCCESS in 19m 52s\n- glance-rally-task-uwsgi https://zuul.opendev.org/t/openstack/build/39e9afa91bb54eefa484bb0ee0960450 : SUCCESS in 52m 13s (non-voting)","accounts_in_message":[],"_revision_number":16},{"id":"6a0aa96c14364e6718dead679b4f2c4e6e81c7e4","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-18 17:12:20.000000000","message":"Patch Set 16:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0a49f25449ef4543bbd48848e5773aa0\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/6ae112ce495748c8b275268c61129765 : SUCCESS in 8m 04s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/59c0c1862b7649ef825874a2de9077df : SUCCESS in 9m 10s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/b57a00c51f2c48a1be4f444b833ee969 : SUCCESS in 17m 32s (non-voting)","accounts_in_message":[],"_revision_number":16},{"id":"b67f83845b4435d0969cb0da6b7837a175d6a8a6","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2026-06-18 17:43:26.000000000","message":"Patch Set 16:\n\n(1 comment)","accounts_in_message":[],"_revision_number":16},{"id":"c4dc66a26b02d0aea27d50a0dcc681d193bdab67","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-18 19:23:37.000000000","message":"Patch Set 16:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1e03efda4e7b4b169443f8aa904bf9b1\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/f8dba427b0364061b59c939cfd949c58 : SUCCESS in 6m 31s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/4475f5f1504341f5a6d2b9bdc51f028f : SUCCESS in 9m 45s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/84e36307958c4e2e8fcaaebdafe53a90 : SUCCESS in 16m 48s (non-voting)","accounts_in_message":[],"_revision_number":16},{"id":"a0d6edc9bb16ca1cd876cdfa62310b4e1c5e70aa","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-18 20:28:59.000000000","message":"Patch Set 16:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/80ee2696bd924517abe9532869463df3\n\n- grenade https://zuul.opendev.org/t/openstack/build/8bbe8eb4d6f7424fb040aedb26275f67 : SUCCESS in 1h 14m 38s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/b23c80df51fd4e31b91f8963c4dc5bda : SUCCESS in 34m 31s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/9ea58ab4a55d48339079a187e77e4cd1 : SUCCESS in 1h 39m 06s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/dbdcbd27caca4a5b87958b96c3a5888c : SUCCESS in 34m 29s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d935a2e70a484ce1bd10e727a3081151 : SUCCESS in 4m 43s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/f8aeb83f3bd64f0ca76f6090ce758665 : SUCCESS in 3m 00s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/5b8a45c9946840a6a43803ceb41cedff : SUCCESS in 5m 10s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/5c8f5b35e4474940954b6fb8844b0432 : SUCCESS in 7m 38s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2deffa2e2c304c7db41554ed5902aee5 : SUCCESS in 6m 37s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/bfad5e6aeb934ed6aa7a189d01235abc : SUCCESS in 7m 32s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/d116d86594af46e99f2a30c5317b30ce : SUCCESS in 6m 50s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/d846bf7b55b24620bc429b359195c12b : FAILURE in 6m 52s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/3e2754b9d56e4186b0e44c686a4e41cc : SUCCESS in 1h 30m 25s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/015c759d77bc456ca5db8f4805b892cf : SUCCESS in 2h 18m 13s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/740f7161a85a421786da55c5fb316ebd : SUCCESS in 2h 38m 53s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/e1129fa5da9a49f29afd3a5bf1e23c1e : SUCCESS in 2h 15m 36s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/024ae86ee7ca4cd4ae77aa7842afd7df : SUCCESS in 30m 37s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/5a24b73c97cc404eb50a7cd25f1e05b0 : SUCCESS in 1h 54m 38s\n- glance-grenade-centralized-cache https://zuul.opendev.org/t/openstack/build/df7cff516f4147fb952b2f0276566914 : FAILURE in 7m 43s (non-voting)\n- glance-s3-multistore https://zuul.opendev.org/t/openstack/build/d8e5dc4bd7b34a9187d4dec90e91862a : SUCCESS in 1h 56m 00s (non-voting)\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/48ccd6c1ee914cd3b9ddd046fcfa985e : SUCCESS in 38m 00s\n- glance-rally-task-uwsgi https://zuul.opendev.org/t/openstack/build/ff87a9b026be48c58d6581872c6e91bd : SUCCESS in 35m 52s (non-voting)","accounts_in_message":[],"_revision_number":16},{"id":"e8c3cb65b07230df4edbfa331e2d6d6e8a837c2f","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2026-06-19 01:45:31.000000000","message":"Patch Set 16:\n\n(1 comment)","accounts_in_message":[],"_revision_number":16},{"id":"b9a6354532cc0158204c013bad89ff50ec3ddf77","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-19 04:27:06.000000000","message":"Patch Set 16: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f7acb4f2e92e4506905038772d73e7bb\n\n- grenade https://zuul.opendev.org/t/openstack/build/d9dffdabe2d94003a985f3b26ceaaa00 : SUCCESS in 31m 13s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/98cd4f821d2b4e3faf06e9ec0d88cc5c : SUCCESS in 1h 07m 42s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/926d78759e0e4002a7ae46c64cf43925 : SUCCESS in 1h 44m 17s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/c45aa450704a4ab687a69483585a3d7f : SUCCESS in 37m 09s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e308802f216c4889aedf2939f5e4de44 : SUCCESS in 2m 36s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/af56e499082544a7930fa9b2199c9f50 : SUCCESS in 4m 34s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/288c9cf4cde04b7189ce11427224eea2 : SUCCESS in 2m 48s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/832a4cf556f8459b9f8c87295020c885 : SUCCESS in 7m 53s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e9e814ebe02d4551a4722d678e604aed : SUCCESS in 6m 35s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/92fe1df1f3fe4adb9348e85709f86884 : SUCCESS in 6m 34s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/ccd1c14c0a1c4e2e9747d33677c2c6a7 : SUCCESS in 4m 46s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/6a771a0bca744fb9801f1e6bacdf87a4 : SUCCESS in 7m 57s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/e7ff59178cd74eb7bd9c371e8e7c77c1 : SUCCESS in 1h 52m 24s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/6200e6f5d55f480db0415afc8ea64526 : SUCCESS in 2h 00m 01s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/fe4ab160f1f54dd0a7290a6ea0b3823c : SUCCESS in 2h 20m 51s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/9102b6b83ec34de98f3623ca47de6719 : SUCCESS in 2h 09m 13s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/45b7453ad3e74a8ca4e79b4c45f5b10d : SUCCESS in 33m 15s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/4362056ef790424ab60eec6fd185d32b : SUCCESS in 2h 34m 24s\n- glance-grenade-centralized-cache https://zuul.opendev.org/t/openstack/build/7553846853344909b8ad3b383250c9cd : FAILURE in 7m 11s (non-voting)\n- glance-s3-multistore https://zuul.opendev.org/t/openstack/build/c27abb5627f94d34b659f7dfe179e621 : SUCCESS in 1h 33m 02s (non-voting)\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/47e574040e8d43b2a8ab330963b3c190 : SUCCESS in 38m 27s\n- glance-rally-task-uwsgi https://zuul.opendev.org/t/openstack/build/33a518a100074ab993596eca81e4ea16 : SUCCESS in 53m 27s (non-voting)","accounts_in_message":[],"_revision_number":16},{"id":"5c3c15fe43dda304c19d1a3b53245b646b944f25","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-19 04:27:37.000000000","message":"Patch Set 16: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":16},{"id":"7cb99380aa438b9babb621585d129041f136751d","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-19 05:06:35.000000000","message":"Patch Set 16:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a3ac375a3cf74e5186528b5f8dd3be8f\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/5532dce50c624fbcbec5c114a1c9e9fc : SUCCESS in 6m 11s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/aa0579ce420d449898b8b3b576b6246c : SUCCESS in 7m 43s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/f9397c180f57464a9771d39047f95237 : SUCCESS in 10m 21s (non-voting)","accounts_in_message":[],"_revision_number":16},{"id":"3e547f154fd0419d1fd04d79d9b6b9c93d4f64de","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-19 06:49:13.000000000","message":"Patch Set 16: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5f71b57ad3fd42d1ae48d4fdb2e7be97\n\n- grenade https://zuul.opendev.org/t/openstack/build/b17e3641286e4dd09dfca51debc88065 : SUCCESS in 37m 51s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/8188b29612af43cdb240bb486b16d913 : SUCCESS in 1h 02m 23s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/4a06e9c29bac43ecb5cca4f5727298b7 : SUCCESS in 1h 26m 31s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/0887b47dca1c4386a924ec2cf0715301 : SUCCESS in 37m 56s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/afdd06f4505c434585f05e08898bb42a : SUCCESS in 4m 44s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/165cbfb331b54444b8af42034eac6d96 : SUCCESS in 4m 26s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/d74f1bde5f0143cbb772fcea11fc9764 : SUCCESS in 2m 53s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e6eaec79befb4e69a6ac515fafb751b2 : SUCCESS in 4m 58s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/1b82af40cd01473d82402077305a54dd : SUCCESS in 3m 34s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/731b86aeac6449e2ae98824505b23c8d : SUCCESS in 6m 13s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/c5a3ae87f2674ca68a2846a2602055fe : SUCCESS in 4m 44s\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/7eb2c65e684b4623b67a6b350d62123e : SUCCESS in 2h 16m 43s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/a838e7cce5be4fcca270d45ec840e4a3 : SUCCESS in 1h 59m 59s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/4f07133ddf794f61948454746334c8d4 : SUCCESS in 33m 59s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/3b950051040146e59b2ecef57e27b88e : SUCCESS in 1h 05m 36s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/f6d4d1cd70bf48fab071bf236bdcae63 : SUCCESS in 34m 30s","accounts_in_message":[],"_revision_number":16},{"id":"a1b23d45688cd64f8ef8903afd27085a92de31e2","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-19 06:49:13.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":16},{"id":"ac7cf72175f3f738d513057d4063653cee461992","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-19 06:50:12.000000000","message":"Patch Set 16:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2dfa15a81ec84055bfbb4e6b1ade2406\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/943d1fbeb6ff43c8a95223181715c1d7 : SUCCESS in 49s","accounts_in_message":[],"_revision_number":16}],"current_revision_number":16,"current_revision":"c5ad2013898f2deb151ef0070a60004fadee2b6d","revisions":{"bde8f2776601ab2716be452b033ab5529cb54ab3":{"kind":"REWORK","_number":1,"created":"2026-01-29 06:59:10.000000000","uploader":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"ref":"refs/changes/18/975118/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/18/975118/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/18/975118/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/18/975118/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/18/975118/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/18/975118/1"}}},"commit":{"parents":[{"commit":"9f4165f7738d48a5081261d1991d4821d1c33eb0","subject":"Merge \"Simplify import of sqlalchemy\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/9f4165f7738d48a5081261d1991d4821d1c33eb0"}]}],"author":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2026-01-27 18:19:29.000000000","tz":0},"committer":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2026-01-29 06:58:48.000000000","tz":0},"subject":"Switch tests to native threading","message":"Switch tests to native threading\n\nReplace eventlet with native threading in test initialization. This\nchange requires patching futurist\u0027s atexit handler to use timeouts\nwhen joining threads to prevent indefinite hangs during test cleanup,\nas native threads behave differently than eventlet green threads.\n\nAlso fix functional tests to handle race conditions that became more\napparent with native threading, where the CacheWorker processes cache\noperations asynchronously and tests need to account for timing\nvariations.\n\nChange-Id: Ic3f2c78d36f011fd11bb52b1255f75835df60e2d\nSigned-off-by: Abhishek Kekane \u003cakekane@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/bde8f2776601ab2716be452b033ab5529cb54ab3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/bde8f2776601ab2716be452b033ab5529cb54ab3"}]},"branch":"refs/heads/master"},"685bc766b7a1edcf9a9847b110fbbb5c95397323":{"kind":"REWORK","_number":2,"created":"2026-01-29 09:34:10.000000000","uploader":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"ref":"refs/changes/18/975118/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/18/975118/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/18/975118/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/18/975118/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/18/975118/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/18/975118/2"}}},"commit":{"parents":[{"commit":"9f4165f7738d48a5081261d1991d4821d1c33eb0","subject":"Merge \"Simplify import of sqlalchemy\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/9f4165f7738d48a5081261d1991d4821d1c33eb0"}]}],"author":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2026-01-27 18:19:29.000000000","tz":0},"committer":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2026-01-29 09:33:37.000000000","tz":0},"subject":"Switch tests to native threading","message":"Switch tests to native threading\n\nReplace eventlet with native threading in test initialization. This\nchange requires patching futurist\u0027s atexit handler to use timeouts\nwhen joining threads to prevent indefinite hangs during test cleanup,\nas native threads behave differently than eventlet green threads.\n\nAlso fix functional tests to handle race conditions that became more\napparent with native threading, where the CacheWorker processes cache\noperations asynchronously and tests need to account for timing\nvariations.\n\nChange-Id: Ic3f2c78d36f011fd11bb52b1255f75835df60e2d\nSigned-off-by: Abhishek Kekane \u003cakekane@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/685bc766b7a1edcf9a9847b110fbbb5c95397323"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/685bc766b7a1edcf9a9847b110fbbb5c95397323"}]},"branch":"refs/heads/master"},"9ac116eac0b73ddbeafa3f1a3293f8906d9a91f7":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2026-05-08 09:37:24.000000000","uploader":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"ref":"refs/changes/18/975118/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/18/975118/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/18/975118/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/18/975118/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/18/975118/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/18/975118/3"}}},"commit":{"parents":[{"commit":"dc5b35d66dd04a88092cdd1ee310ef76d2df47f1","subject":"Merge \"Change DB migration constant to 2026_2\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/dc5b35d66dd04a88092cdd1ee310ef76d2df47f1"}]}],"author":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2026-01-27 18:19:29.000000000","tz":0},"committer":{"name":"Pranali Deore","email":"pdeore@redhat.com","date":"2026-05-08 09:23:23.000000000","tz":0},"subject":"Switch tests to native threading","message":"Switch tests to native threading\n\nReplace eventlet with native threading in test initialization. This\nchange requires patching futurist\u0027s atexit handler to use timeouts\nwhen joining threads to prevent indefinite hangs during test cleanup,\nas native threads behave differently than eventlet green threads.\n\nAlso fix functional tests to handle race conditions that became more\napparent with native threading, where the CacheWorker processes cache\noperations asynchronously and tests need to account for timing\nvariations.\n\nChange-Id: Ic3f2c78d36f011fd11bb52b1255f75835df60e2d\nSigned-off-by: Abhishek Kekane \u003cakekane@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/9ac116eac0b73ddbeafa3f1a3293f8906d9a91f7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/9ac116eac0b73ddbeafa3f1a3293f8906d9a91f7"}]},"branch":"refs/heads/master"},"338e195274715b64afbf35ab9d5c3ec47922b917":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2026-05-19 06:30:16.000000000","uploader":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"ref":"refs/changes/18/975118/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/18/975118/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/18/975118/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/18/975118/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/18/975118/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/18/975118/4"}}},"commit":{"parents":[{"commit":"eac2fa47f26da3515c7a1e8c91226750517c52d4","subject":"Merge \"Remove usage of eventlet for image cache SQLite timeouts\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/eac2fa47f26da3515c7a1e8c91226750517c52d4"}]}],"author":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2026-01-27 18:19:29.000000000","tz":0},"committer":{"name":"Pranali Deore","email":"pdeore@redhat.com","date":"2026-05-19 05:43:08.000000000","tz":0},"subject":"Switch tests to native threading","message":"Switch tests to native threading\n\nReplace eventlet with native threading in test initialization. This\nchange requires patching futurist\u0027s atexit handler to use timeouts\nwhen joining threads to prevent indefinite hangs during test cleanup,\nas native threads behave differently than eventlet green threads.\n\nAlso fix functional tests to handle race conditions that became more\napparent with native threading, where the CacheWorker processes cache\noperations asynchronously and tests need to account for timing\nvariations.\n\nChange-Id: Ic3f2c78d36f011fd11bb52b1255f75835df60e2d\nSigned-off-by: Abhishek Kekane \u003cakekane@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/338e195274715b64afbf35ab9d5c3ec47922b917"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/338e195274715b64afbf35ab9d5c3ec47922b917"}]},"branch":"refs/heads/master"},"b3733e2ab2d7f55e040e73d1a08ffa0881ce184f":{"kind":"TRIVIAL_REBASE","_number":5,"created":"2026-05-29 10:40:07.000000000","uploader":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"ref":"refs/changes/18/975118/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/18/975118/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/18/975118/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/18/975118/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/18/975118/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/18/975118/5"}}},"commit":{"parents":[{"commit":"c986cf4dbc854cf3305d525757d5e01dc8c10ef4","subject":"Fix get_ksa_client() after identity_uri removal","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/c986cf4dbc854cf3305d525757d5e01dc8c10ef4"}]}],"author":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2026-01-27 18:19:29.000000000","tz":0},"committer":{"name":"Pranali Deore","email":"pdeore@redhat.com","date":"2026-05-29 07:19:03.000000000","tz":0},"subject":"Switch tests to native threading","message":"Switch tests to native threading\n\nReplace eventlet with native threading in test initialization. This\nchange requires patching futurist\u0027s atexit handler to use timeouts\nwhen joining threads to prevent indefinite hangs during test cleanup,\nas native threads behave differently than eventlet green threads.\n\nAlso fix functional tests to handle race conditions that became more\napparent with native threading, where the CacheWorker processes cache\noperations asynchronously and tests need to account for timing\nvariations.\n\nChange-Id: Ic3f2c78d36f011fd11bb52b1255f75835df60e2d\nSigned-off-by: Abhishek Kekane \u003cakekane@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/b3733e2ab2d7f55e040e73d1a08ffa0881ce184f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/b3733e2ab2d7f55e040e73d1a08ffa0881ce184f"}]},"branch":"refs/heads/master"},"628c0df9154dfdbfedfeef8ffcffe5d0d63c8e6a":{"kind":"TRIVIAL_REBASE","_number":6,"created":"2026-06-02 07:15:30.000000000","uploader":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"ref":"refs/changes/18/975118/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/18/975118/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/18/975118/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/18/975118/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/18/975118/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/18/975118/6"}}},"commit":{"parents":[{"commit":"1240b0bf2453c94590a6883645e86ef24f5d1ada","subject":"Merge \"Fix flaky web-download redirect validation test\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/1240b0bf2453c94590a6883645e86ef24f5d1ada"}]}],"author":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2026-01-27 18:19:29.000000000","tz":0},"committer":{"name":"Pranali Deore","email":"pdeore@redhat.com","date":"2026-06-02 07:14:36.000000000","tz":0},"subject":"Switch tests to native threading","message":"Switch tests to native threading\n\nReplace eventlet with native threading in test initialization. This\nchange requires patching futurist\u0027s atexit handler to use timeouts\nwhen joining threads to prevent indefinite hangs during test cleanup,\nas native threads behave differently than eventlet green threads.\n\nAlso fix functional tests to handle race conditions that became more\napparent with native threading, where the CacheWorker processes cache\noperations asynchronously and tests need to account for timing\nvariations.\n\nChange-Id: Ic3f2c78d36f011fd11bb52b1255f75835df60e2d\nSigned-off-by: Abhishek Kekane \u003cakekane@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/628c0df9154dfdbfedfeef8ffcffe5d0d63c8e6a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/628c0df9154dfdbfedfeef8ffcffe5d0d63c8e6a"}]},"branch":"refs/heads/master"},"52805acd31293fa82c25bd8c4a23b49a801fd693":{"kind":"REWORK","_number":7,"created":"2026-06-03 05:25:25.000000000","uploader":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"ref":"refs/changes/18/975118/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/18/975118/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/18/975118/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/18/975118/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/18/975118/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/18/975118/7"}}},"commit":{"parents":[{"commit":"1240b0bf2453c94590a6883645e86ef24f5d1ada","subject":"Merge \"Fix flaky web-download redirect validation test\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/1240b0bf2453c94590a6883645e86ef24f5d1ada"}]}],"author":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2026-01-27 18:19:29.000000000","tz":0},"committer":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2026-06-03 05:22:32.000000000","tz":0},"subject":"Switch tests to native threading","message":"Switch tests to native threading\n\nReplace eventlet with native threading in test initialization. This\nchange requires patching futurist\u0027s atexit handler to use timeouts\nwhen joining threads to prevent indefinite hangs during test cleanup,\nas native threads behave differently than eventlet green threads.\n\nAlso fix functional tests to handle race conditions that became more\napparent with native threading, where the CacheWorker processes cache\noperations asynchronously and tests need to account for timing\nvariations.\n\nChange-Id: Ic3f2c78d36f011fd11bb52b1255f75835df60e2d\nSigned-off-by: Abhishek Kekane \u003cakekane@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/52805acd31293fa82c25bd8c4a23b49a801fd693"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/52805acd31293fa82c25bd8c4a23b49a801fd693"}]},"branch":"refs/heads/master"},"f3e0479a8f29487da3db28716b203f77d9166351":{"kind":"NO_CHANGE","_number":8,"created":"2026-06-04 08:49:58.000000000","uploader":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"ref":"refs/changes/18/975118/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/18/975118/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/18/975118/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/18/975118/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/18/975118/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/18/975118/8"}}},"commit":{"parents":[{"commit":"1240b0bf2453c94590a6883645e86ef24f5d1ada","subject":"Merge \"Fix flaky web-download redirect validation test\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/1240b0bf2453c94590a6883645e86ef24f5d1ada"}]}],"author":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2026-01-27 18:19:29.000000000","tz":0},"committer":{"name":"Pranali Deore","email":"pdeore@redhat.com","date":"2026-06-04 05:31:16.000000000","tz":0},"subject":"Switch tests to native threading","message":"Switch tests to native threading\n\nReplace eventlet with native threading in test initialization. This\nchange requires patching futurist\u0027s atexit handler to use timeouts\nwhen joining threads to prevent indefinite hangs during test cleanup,\nas native threads behave differently than eventlet green threads.\n\nAlso fix functional tests to handle race conditions that became more\napparent with native threading, where the CacheWorker processes cache\noperations asynchronously and tests need to account for timing\nvariations.\n\nChange-Id: Ic3f2c78d36f011fd11bb52b1255f75835df60e2d\nSigned-off-by: Abhishek Kekane \u003cakekane@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/f3e0479a8f29487da3db28716b203f77d9166351"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/f3e0479a8f29487da3db28716b203f77d9166351"}]},"branch":"refs/heads/master"},"ea6036ff509051d61ae63d9c39120ee329a7e6dc":{"kind":"REWORK","_number":9,"created":"2026-06-05 11:26:30.000000000","uploader":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"ref":"refs/changes/18/975118/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/18/975118/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/18/975118/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/18/975118/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/18/975118/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/18/975118/9"}}},"commit":{"parents":[{"commit":"1240b0bf2453c94590a6883645e86ef24f5d1ada","subject":"Merge \"Fix flaky web-download redirect validation test\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/1240b0bf2453c94590a6883645e86ef24f5d1ada"}]}],"author":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2026-01-27 18:19:29.000000000","tz":0},"committer":{"name":"Pranali Deore","email":"pdeore@redhat.com","date":"2026-06-05 11:25:44.000000000","tz":0},"subject":"Switch tests to native threading","message":"Switch tests to native threading\n\nReplace eventlet with native threading in test initialization. This\nchange requires patching futurist\u0027s atexit handler to use timeouts\nwhen joining threads to prevent indefinite hangs during test cleanup,\nas native threads behave differently than eventlet green threads.\n\nAlso fix functional tests to handle race conditions that became more\napparent with native threading, where the CacheWorker processes cache\noperations asynchronously and tests need to account for timing\nvariations.\n\nChange-Id: Ic3f2c78d36f011fd11bb52b1255f75835df60e2d\nSigned-off-by: Abhishek Kekane \u003cakekane@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/ea6036ff509051d61ae63d9c39120ee329a7e6dc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/ea6036ff509051d61ae63d9c39120ee329a7e6dc"}]},"branch":"refs/heads/master"},"c4998d6ffeb79fb66bca67e690d4ab849ca59674":{"kind":"REWORK","_number":10,"created":"2026-06-08 07:17:32.000000000","uploader":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"ref":"refs/changes/18/975118/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/18/975118/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/18/975118/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/18/975118/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/18/975118/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/18/975118/10"}}},"commit":{"parents":[{"commit":"1240b0bf2453c94590a6883645e86ef24f5d1ada","subject":"Merge \"Fix flaky web-download redirect validation test\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/1240b0bf2453c94590a6883645e86ef24f5d1ada"}]}],"author":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2026-01-27 18:19:29.000000000","tz":0},"committer":{"name":"Pranali Deore","email":"pdeore@redhat.com","date":"2026-06-08 06:14:43.000000000","tz":0},"subject":"Switch tests to native threading","message":"Switch tests to native threading\n\nReplace eventlet with native threading in test initialization. This\nchange requires patching futurist\u0027s atexit handler to use timeouts\nwhen joining threads to prevent indefinite hangs during test cleanup,\nas native threads behave differently than eventlet green threads.\n\nAlso fix functional tests to handle race conditions that became more\napparent with native threading, where the CacheWorker processes cache\noperations asynchronously and tests need to account for timing\nvariations.\n\nChange-Id: Ic3f2c78d36f011fd11bb52b1255f75835df60e2d\nSigned-off-by: Abhishek Kekane \u003cakekane@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/c4998d6ffeb79fb66bca67e690d4ab849ca59674"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/c4998d6ffeb79fb66bca67e690d4ab849ca59674"}]},"branch":"refs/heads/master"},"aea34705f5e69ee645a0e90596d544b54b4e0505":{"kind":"TRIVIAL_REBASE","_number":11,"created":"2026-06-10 12:30:51.000000000","uploader":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"ref":"refs/changes/18/975118/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/18/975118/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/18/975118/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/18/975118/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/18/975118/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/18/975118/11"}}},"commit":{"parents":[{"commit":"0cd329e8898fbf59502befe52f5b8800ab43cf97","subject":"Merge \"Rally: use raw disk_format for synthetic fake.img\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/0cd329e8898fbf59502befe52f5b8800ab43cf97"}]}],"author":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2026-01-27 18:19:29.000000000","tz":0},"committer":{"name":"Pranali Deore","email":"pdeore@redhat.com","date":"2026-06-10 06:05:40.000000000","tz":0},"subject":"Switch tests to native threading","message":"Switch tests to native threading\n\nReplace eventlet with native threading in test initialization. This\nchange requires patching futurist\u0027s atexit handler to use timeouts\nwhen joining threads to prevent indefinite hangs during test cleanup,\nas native threads behave differently than eventlet green threads.\n\nAlso fix functional tests to handle race conditions that became more\napparent with native threading, where the CacheWorker processes cache\noperations asynchronously and tests need to account for timing\nvariations.\n\nChange-Id: Ic3f2c78d36f011fd11bb52b1255f75835df60e2d\nSigned-off-by: Abhishek Kekane \u003cakekane@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/aea34705f5e69ee645a0e90596d544b54b4e0505"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/aea34705f5e69ee645a0e90596d544b54b4e0505"}]},"branch":"refs/heads/master"},"6ea226a2cd214e96006a54f9af4567fa0c37a2a0":{"kind":"TRIVIAL_REBASE","_number":12,"created":"2026-06-10 15:23:27.000000000","uploader":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"ref":"refs/changes/18/975118/12","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/18/975118/12","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/18/975118/12 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/18/975118/12 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/18/975118/12 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/18/975118/12"}}},"commit":{"parents":[{"commit":"c7a5f6e00037725846ab30df32339fe3fb68df4b","subject":"Merge \"Add non-voting Zuul Rally job for Glance on uWSGI\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/c7a5f6e00037725846ab30df32339fe3fb68df4b"}]}],"author":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2026-01-27 18:19:29.000000000","tz":0},"committer":{"name":"Pranali Deore","email":"pdeore@redhat.com","date":"2026-06-10 15:20:24.000000000","tz":0},"subject":"Switch tests to native threading","message":"Switch tests to native threading\n\nReplace eventlet with native threading in test initialization. This\nchange requires patching futurist\u0027s atexit handler to use timeouts\nwhen joining threads to prevent indefinite hangs during test cleanup,\nas native threads behave differently than eventlet green threads.\n\nAlso fix functional tests to handle race conditions that became more\napparent with native threading, where the CacheWorker processes cache\noperations asynchronously and tests need to account for timing\nvariations.\n\nChange-Id: Ic3f2c78d36f011fd11bb52b1255f75835df60e2d\nSigned-off-by: Abhishek Kekane \u003cakekane@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/6ea226a2cd214e96006a54f9af4567fa0c37a2a0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/6ea226a2cd214e96006a54f9af4567fa0c37a2a0"}]},"branch":"refs/heads/master"},"6077c831c0280f230bcbf52626078d1b67f8ea26":{"kind":"REWORK","_number":13,"created":"2026-06-16 07:18:32.000000000","uploader":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"ref":"refs/changes/18/975118/13","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/18/975118/13","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/18/975118/13 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/18/975118/13 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/18/975118/13 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/18/975118/13"}}},"commit":{"parents":[{"commit":"1107c4614fcaceff5af997e5e347863506c16f04","subject":"Skip flaky web download redirect validation test","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/1107c4614fcaceff5af997e5e347863506c16f04"}]}],"author":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2026-01-27 18:19:29.000000000","tz":0},"committer":{"name":"Pranali Deore","email":"pdeore@redhat.com","date":"2026-06-16 07:16:59.000000000","tz":0},"subject":"Switch tests to native threading","message":"Switch tests to native threading\n\nReplace eventlet with native threading in test initialization. This\nchange requires patching futurist\u0027s atexit handler to use timeouts\nwhen joining threads to prevent indefinite hangs during test cleanup,\nas native threads behave differently than eventlet green threads.\n\nAlso fix functional tests to handle race conditions that became more\napparent with native threading, where the CacheWorker processes cache\noperations asynchronously and tests need to account for timing\nvariations.\n\nChange-Id: Ic3f2c78d36f011fd11bb52b1255f75835df60e2d\nSigned-off-by: Abhishek Kekane \u003cakekane@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/6077c831c0280f230bcbf52626078d1b67f8ea26"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/6077c831c0280f230bcbf52626078d1b67f8ea26"}]},"branch":"refs/heads/master"},"bd03794f1090e483233538f707a24b533b7a68d5":{"kind":"TRIVIAL_REBASE","_number":14,"created":"2026-06-16 09:22:13.000000000","uploader":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"ref":"refs/changes/18/975118/14","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/18/975118/14","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/18/975118/14 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/18/975118/14 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/18/975118/14 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/18/975118/14"}}},"commit":{"parents":[{"commit":"edf97a1b5b6790f7e79315a11c3f70cc92553ef8","subject":"Merge \"Fix flaky import task waits in functional tests\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/edf97a1b5b6790f7e79315a11c3f70cc92553ef8"}]}],"author":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2026-01-27 18:19:29.000000000","tz":0},"committer":{"name":"Pranali Deore","email":"pdeore@redhat.com","date":"2026-06-16 09:03:51.000000000","tz":0},"subject":"Switch tests to native threading","message":"Switch tests to native threading\n\nReplace eventlet with native threading in test initialization. This\nchange requires patching futurist\u0027s atexit handler to use timeouts\nwhen joining threads to prevent indefinite hangs during test cleanup,\nas native threads behave differently than eventlet green threads.\n\nAlso fix functional tests to handle race conditions that became more\napparent with native threading, where the CacheWorker processes cache\noperations asynchronously and tests need to account for timing\nvariations.\n\nChange-Id: Ic3f2c78d36f011fd11bb52b1255f75835df60e2d\nSigned-off-by: Abhishek Kekane \u003cakekane@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/bd03794f1090e483233538f707a24b533b7a68d5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/bd03794f1090e483233538f707a24b533b7a68d5"}]},"branch":"refs/heads/master"},"2f3393b8362aae64363ddb82b1f7d7f7a7385ec2":{"kind":"TRIVIAL_REBASE","_number":15,"created":"2026-06-18 09:45:27.000000000","uploader":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"ref":"refs/changes/18/975118/15","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/18/975118/15","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/18/975118/15 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/18/975118/15 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/18/975118/15 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/18/975118/15"}}},"commit":{"parents":[{"commit":"6dd6e4dd05304963bc5c06219fcce439fb463e48","subject":"Merge \"Add support for downloading images from preferred stores\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/6dd6e4dd05304963bc5c06219fcce439fb463e48"}]}],"author":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2026-01-27 18:19:29.000000000","tz":0},"committer":{"name":"Pranali Deore","email":"pdeore@redhat.com","date":"2026-06-18 09:45:27.000000000","tz":0},"subject":"Switch tests to native threading","message":"Switch tests to native threading\n\nReplace eventlet with native threading in test initialization. This\nchange requires patching futurist\u0027s atexit handler to use timeouts\nwhen joining threads to prevent indefinite hangs during test cleanup,\nas native threads behave differently than eventlet green threads.\n\nAlso fix functional tests to handle race conditions that became more\napparent with native threading, where the CacheWorker processes cache\noperations asynchronously and tests need to account for timing\nvariations.\n\nChange-Id: Ic3f2c78d36f011fd11bb52b1255f75835df60e2d\nSigned-off-by: Abhishek Kekane \u003cakekane@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/2f3393b8362aae64363ddb82b1f7d7f7a7385ec2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/2f3393b8362aae64363ddb82b1f7d7f7a7385ec2"}]},"branch":"refs/heads/master","description":"Rebase","conflicts":{"base":"edf97a1b5b6790f7e79315a11c3f70cc92553ef8","ours":"bd03794f1090e483233538f707a24b533b7a68d5","theirs":"6dd6e4dd05304963bc5c06219fcce439fb463e48","merge_strategy":"recursive","contains_conflicts":false}},"c5ad2013898f2deb151ef0070a60004fadee2b6d":{"kind":"TRIVIAL_REBASE","_number":16,"created":"2026-06-18 14:23:30.000000000","uploader":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"ref":"refs/changes/18/975118/16","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/18/975118/16","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/18/975118/16 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/18/975118/16 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/18/975118/16 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/18/975118/16"}}},"commit":{"parents":[{"commit":"9734292f7c4da958b4e29afa0a7f40b8a2003efc","subject":"Merge \"Fix flaky test_import_copy_bust_lock functional test\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/9734292f7c4da958b4e29afa0a7f40b8a2003efc"}]}],"author":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2026-01-27 18:19:29.000000000","tz":0},"committer":{"name":"Pranali Deore","email":"pdeore@redhat.com","date":"2026-06-18 14:23:30.000000000","tz":0},"subject":"Switch tests to native threading","message":"Switch tests to native threading\n\nReplace eventlet with native threading in test initialization. This\nchange requires patching futurist\u0027s atexit handler to use timeouts\nwhen joining threads to prevent indefinite hangs during test cleanup,\nas native threads behave differently than eventlet green threads.\n\nAlso fix functional tests to handle race conditions that became more\napparent with native threading, where the CacheWorker processes cache\noperations asynchronously and tests need to account for timing\nvariations.\n\nChange-Id: Ic3f2c78d36f011fd11bb52b1255f75835df60e2d\nSigned-off-by: Abhishek Kekane \u003cakekane@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/c5ad2013898f2deb151ef0070a60004fadee2b6d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/c5ad2013898f2deb151ef0070a60004fadee2b6d"}]},"branch":"refs/heads/master","description":"Rebase","conflicts":{"base":"6dd6e4dd05304963bc5c06219fcce439fb463e48","ours":"2f3393b8362aae64363ddb82b1f7d7f7a7385ec2","theirs":"9734292f7c4da958b4e29afa0a7f40b8a2003efc","merge_strategy":"recursive","contains_conflicts":false}}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{"label:Review-Priority\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
