)]}'
{"id":"openstack%2Fglance~975590","triplet_id":"openstack%2Fglance~stable%2F2025.2~I4770a86152607e0fa030e3acfb4dddbf18dc5abd","project":"openstack/glance","branch":"stable/2025.2","topic":"bug/2138919-stable/2025.2","attention_set":{},"removed_from_attention_set":{"38762":{"account":{"_account_id":38762,"name":"Sahil Kumbhar","display_name":"sakumbha","email":"sakumbha@redhat.com","username":"sakumbha"},"last_update":"2026-04-01 16:44:30.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I4770a86152607e0fa030e3acfb4dddbf18dc5abd","subject":"Raised NotImplementedError instead of returning","status":"MERGED","created":"2026-02-04 05:00:10.000000000","updated":"2026-04-01 16:46:25.000000000","submitted":"2026-04-01 16:44:30.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"975590-bug/2138919-stable/2025.2","cherry_pick_of_change":974392,"cherry_pick_of_patch_set":1,"meta_rev_id":"7ef048eb7c41e46f5a430e8b72227c4450493bdc","_number":975590,"virtual_id_number":975590,"owner":{"_account_id":38762,"name":"Sahil Kumbhar","display_name":"sakumbha","email":"sakumbha@redhat.com","username":"sakumbha"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-04-01 16:44:30.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"recommended":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"all":[{"value":1,"date":"2026-02-24 08:19:07.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},{"value":2,"date":"2026-04-01 14:17:55.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"all":[{"value":0,"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},{"value":1,"date":"2026-04-01 14:17:55.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Top Priority / Holds Gate"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-02-04 07:21:45.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-02-13 17:11:04.000000000","updated_by":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"reviewer":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"state":"CC"},{"updated":"2026-02-24 08:19:07.000000000","updated_by":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"reviewer":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"state":"REVIEWER"},{"updated":"2026-04-01 14:17:55.000000000","updated_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"reviewer":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"state":"REVIEWER"}],"messages":[{"id":"480b3b4fc21849392eec6c34270831060ff71309","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":38762,"name":"Sahil Kumbhar","display_name":"sakumbha","email":"sakumbha@redhat.com","username":"sakumbha"},"date":"2026-02-04 05:00:10.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"9c887ddfcf97d6bbe79cef3ad1a650832d33a3f5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-04 07:21:45.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/5a90ab7039664a72bab13a6bdc94df86\n\n- grenade https://zuul.opendev.org/t/openstack/build/9fedebf232074fb8bbc39294c9beba08 : SUCCESS in 31m 44s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/bb61f21f78cb4a6bb60b86acc7445885 : SUCCESS in 47m 12s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/451f923137324604959623ab55bd1de1 : SUCCESS in 1h 26m 48s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/b8e280c6322b42529558296045e9cca9 : SUCCESS in 34m 39s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e296bf271d91449eaeeae899668c3b61 : SUCCESS in 3m 07s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/1c08b9c0ae8d44fb89a0936ebe830b45 : SUCCESS in 4m 57s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/6c0110cc720f41f7954e6a3510336787 : SUCCESS in 5m 28s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/c071206de830410eb9aab05bf6087a28 : SUCCESS in 6m 55s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bff25a9120ee4351830ac286e70ee8f6 : SUCCESS in 5m 32s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/226d4d38493b419197323fc0661ee1be : SUCCESS in 6m 53s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/ca9ed2cf30bd4eab914d8ce1373ab786 : SUCCESS in 7m 36s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/87e3b764960e4f239e7f0283ed6189e4 : SUCCESS in 1h 54m 38s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/86d182ce23c84acbbeddbe2c6d441591 : SUCCESS in 1h 45m 55s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/6b3a61788ae64065b8bee616dd2832a2 : SUCCESS in 1h 59m 36s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/a5369a1f9aac48219ddbf149814f2284 : SUCCESS in 2h 15m 50s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/640e0c216d7f46cb88a95e9f60ac6b1e : SUCCESS in 55m 43s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/38d248df8c1b440dbeb414b9734daf8d : FAILURE in 1h 00m 35s\n- glance-grenade-centralized-cache https://zuul.opendev.org/t/openstack/build/c4bde9e3c0364011948fcaa40aafb870 : FAILURE in 27m 21s (non-voting)\n- glance-s3-multistore https://zuul.opendev.org/t/openstack/build/4f8d63d35436437c85a69c512a31cb84 : FAILURE in 25m 10s (non-voting)\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/10f3677b4b874f24ad6ccb107b27ac4b : SUCCESS in 34m 11s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/3ba0fb62c0f049f69ae1b239dd160e9f : SUCCESS in 1h 09m 05s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"007486230e50a279fb946dc75453b9d9d7c7d49c","author":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"date":"2026-02-13 17:11:04.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"70095b42de1e7ccad06ed6784b9d9d410001715d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-13 19:23:51.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/874e5f20e9d5479e87c04daa527a8f51\n\n- grenade https://zuul.opendev.org/t/openstack/build/d0b4b9b035934623a06170f715217663 : SUCCESS in 1h 01m 03s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/d9035c3028dd413fa6182445b640e64b : SUCCESS in 1h 01m 29s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/618a4ac19e2e429b963163ea5c9b0e63 : SUCCESS in 1h 31m 12s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/cc08e8dca3084ede8232cd785f833018 : SUCCESS in 1h 05m 22s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/204f871a897643b5a3e77de48b28a26f : SUCCESS in 6m 21s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/87c18e2281a547dea971b03f91eb5ad8 : SUCCESS in 4m 42s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/61a4ce1319a14cfabb73d125b83f11aa : SUCCESS in 5m 32s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/94f5f5f5fee643d8afaa0f9b32b76f4a : SUCCESS in 7m 53s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/995814ac7a914d309301d9aed83cdb2e : SUCCESS in 5m 35s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/0f63d80d69574734b3af5485a2c8820c : SUCCESS in 7m 12s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/f240c00d67b54e8da51d578c803f7269 : SUCCESS in 5m 05s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/00da74e1b896488eb9fae2f46d54e8f7 : SUCCESS in 2h 02m 35s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/9ed5d2d57ae94e33a26b5f1f6d68c92c : SUCCESS in 1h 07m 18s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/aecc8f675d774e46bc25af550d5c5d79 : SUCCESS in 1h 15m 14s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/58bc0964ec674f68be3f9bb566ae8f26 : SUCCESS in 1h 51m 40s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/7d14fe4fe061450d8d140f69ffb8cef1 : SUCCESS in 55m 01s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/5681d687ccb2431fa6f068488c1cff5e : SUCCESS in 1h 32m 00s\n- glance-grenade-centralized-cache https://zuul.opendev.org/t/openstack/build/08a1a814c1134241b00a20f2197c38f7 : FAILURE in 29m 51s (non-voting)\n- glance-s3-multistore https://zuul.opendev.org/t/openstack/build/cdf24993dc2548adbe58966f1465eab0 : FAILURE in 27m 54s (non-voting)\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/c1fe0c938270495995b03fb654a4c185 : SUCCESS in 17m 01s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/957490fe5cb24e2e93fd6fedee471d46 : SUCCESS in 1h 13m 18s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"e2363e55f21a57a31522cf89a12e14582eb348e1","author":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"date":"2026-02-24 08:19:07.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"5f877bc05a8cc6a20a2991bdf063e2101db82270","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2026-04-01 14:17:55.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"ce63bd702532d8eb8b3084fd2599267abff62f3a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-01 14:18:27.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"f319a541c0f50b7dacbe5052fe685ad8a58fc31d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-01 16:44:30.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/fcab8d4f35344b36b49fa470f058e869\n\n- grenade https://zuul.opendev.org/t/openstack/build/023f6efa6722472da5b3f2d43f3d45bf : SUCCESS in 56m 54s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/0b986ce75c3d43989357bab6e45f5925 : SUCCESS in 1h 02m 16s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/d841d1aad6dd4551ba62334121764798 : SUCCESS in 1h 09m 49s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/b6a52186d66d442eac542e5abc00ac5e : SUCCESS in 1h 01m 59s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4c3aeb15a8f443f59d1cdb082319ee56 : SUCCESS in 5m 16s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/b6254723a2c44025bec042aeab8fc37c : SUCCESS in 6m 16s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/cc39a61b374c4a25b75460fd6c395184 : SUCCESS in 3m 13s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1fc4d9a096ae495c90118b65b3561df6 : SUCCESS in 3m 35s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/ba8dd50d21a64b75b41b148f544b0dd6 : SUCCESS in 7m 30s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/b7839c5374da468c904f7e8cd0de3891 : SUCCESS in 7m 24s\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/6591c0187ee246afaf7b53515e4a368f : SUCCESS in 1h 55m 53s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/ba80ce9820bb4a26a251196653dfdc01 : SUCCESS in 2h 10m 37s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/92d844ee6d884f04895e673bffbb7692 : SUCCESS in 1h 12m 58s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/949b00d3b2f64c28b630a8e794d11df9 : SUCCESS in 2h 20m 27s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/b126887f93b24e0280edf47432c1a25f : SUCCESS in 18m 10s","accounts_in_message":[],"_revision_number":1},{"id":"85acb5388534a924a2a194f68d12f24301abd0f6","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-01 16:44:30.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"7ef048eb7c41e46f5a430e8b72227c4450493bdc","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-01 16:46:25.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8b3447b97c9145cf96bd755dd68d4094\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/946be3e03cf044c8b302c087ffe5e7ca : SUCCESS in 54s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"698959790fc0eb0c3a10c7d85ab0f9282e609370","revisions":{"698959790fc0eb0c3a10c7d85ab0f9282e609370":{"kind":"REWORK","_number":1,"created":"2026-02-04 05:00:10.000000000","uploader":{"_account_id":38762,"name":"Sahil Kumbhar","display_name":"sakumbha","email":"sakumbha@redhat.com","username":"sakumbha"},"ref":"refs/changes/90/975590/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/90/975590/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/90/975590/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/90/975590/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/90/975590/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/90/975590/1"}}},"commit":{"parents":[{"commit":"435a5dd2a6b2a3ff12572836feaf08ce89f03e32","subject":"Add glance/wsgi/__init__.py","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/435a5dd2a6b2a3ff12572836feaf08ce89f03e32"}]}],"author":{"name":"sakumbha","email":"sakumbha@redhat.com","date":"2026-01-23 06:57:41.000000000","tz":0},"committer":{"name":"Sahil Kumbhar","email":"sakumbha@redhat.com","date":"2026-02-04 05:00:10.000000000","tz":0},"subject":"Raised NotImplementedError instead of returning","message":"Raised NotImplementedError instead of returning\n\nCloses-Bug: #2138919\nChange-Id: I4770a86152607e0fa030e3acfb4dddbf18dc5abd\nSigned-off-by: sakumbha \u003csakumbha@redhat.com\u003e\n(cherry picked from commit 1ade72a5d2e7dea5b6958b269267c316caf6c503)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/698959790fc0eb0c3a10c7d85ab0f9282e609370"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/698959790fc0eb0c3a10c7d85ab0f9282e609370"}]},"branch":"refs/heads/stable/2025.2"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
