)]}'
{"id":"openstack%2Fglance~976627","triplet_id":"openstack%2Fglance~master~Ief14a61f6a1d9327acfd6ada61c2b5976d153555","project":"openstack/glance","branch":"master","attention_set":{},"removed_from_attention_set":{"19138":{"account":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"last_update":"2026-02-24 08:34:02.000000000","reason":"Change was submitted"},"9303":{"account":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"last_update":"2026-02-24 06:35:41.000000000","reason":"\u003cGERRIT_ACCOUNT_9303\u003e replied on the change","reason_account":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"}},"8122":{"account":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"last_update":"2026-02-24 08:34:02.000000000","reason":"Change was submitted"},"35676":{"account":{"_account_id":35676,"name":"Max","email":"max.lamprecht@digits.schwarz","username":"lamprechtm"},"last_update":"2026-02-24 08:34:02.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ief14a61f6a1d9327acfd6ada61c2b5976d153555","subject":"feat: return size in get_image_data_iter","status":"MERGED","created":"2026-02-12 14:14:13.000000000","updated":"2026-02-24 08:35:19.000000000","submitted":"2026-02-24 08:34:02.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":14,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"976627","meta_rev_id":"224880f95f4eb5209f00c9c8eab8b742d8b99b68","_number":976627,"virtual_id_number":976627,"owner":{"_account_id":35676,"name":"Max","email":"max.lamprecht@digits.schwarz","username":"lamprechtm"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-02-24 08:34:01.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"all":[{"value":0,"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},{"value":2,"date":"2026-02-24 06:35:41.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":2,"date":"2026-02-23 15:26:38.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"all":[{"value":0,"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},{"value":1,"date":"2026-02-24 06:35:41.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Top Priority / Holds Gate"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-02-12 15:05:19.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2026-02-12 16:52:08.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-02-12 17:09:43.000000000","updated_by":{"_account_id":35676,"name":"Max","email":"max.lamprecht@digits.schwarz","username":"lamprechtm"},"reviewer":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"state":"REVIEWER"},{"updated":"2026-02-23 11:29:25.000000000","updated_by":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"reviewer":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"state":"REVIEWER"},{"updated":"2026-02-23 11:29:25.000000000","updated_by":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"reviewer":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"state":"REVIEWER"}],"messages":[{"id":"8545f771018939d2d6911a6099589c8f110204ce","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35676,"name":"Max","email":"max.lamprecht@digits.schwarz","username":"lamprechtm"},"date":"2026-02-12 14:14:13.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"9155e328b5297ee8a795424ede817ff72f10415f","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-12 15:05:19.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3cc35982443e43a59d47ab15b4a54c46\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/456ed2b16887430eb2fef2a284432d5b : SUCCESS in 9m 34s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/3df8e93d40314d929228bf1c8716b4ad : SUCCESS in 10m 55s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/af865ebc55ca404e8b772196418cd7e5 : RETRY_LIMIT in 4m 31s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"f33ab87b334b0ca5459a43abe90a636dbb446bea","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-12 16:52:08.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4c77b9ad8a054323a3d07428fba713bc\n\n- grenade https://zuul.opendev.org/t/openstack/build/f219b09d246e45ad9f6868f75bd15896 : SUCCESS in 1h 01m 48s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/35843587734248acb90e21f2d67b37e8 : SUCCESS in 52m 40s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/5973cfe6d6864a65b951f4fa3db45e04 : SUCCESS in 1h 38m 53s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/b335f3a97fad405597f83a72d3058a50 : SUCCESS in 56m 27s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ddf59d77af154fc3ba8d986e31f135f3 : SUCCESS in 4m 42s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/8ad79481a9b14be4a3e5d2233bd1347f : SUCCESS in 4m 47s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/980939da2e9e4f839b970da87c0f7388 : SUCCESS in 3m 15s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/06f9a36f54bb42d39e8cfb3bf8afa462 : SUCCESS in 3m 15s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/d4fbfcb3d8d241acad81f2b3f3deeaf6 : SUCCESS in 6m 59s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/b1077f6bd3a54f9c8bb44f140a5e29dc : SUCCESS in 7m 38s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/076a09fc89cb4f47b4e7f400085fd5eb : FAILURE in 1h 07m 23s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/eb584d3a4a2e4e65bc7c9549e711071a : SUCCESS in 1h 37m 17s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/c8ed77b2a2074aa18b153e1f597fa3be : SUCCESS in 1h 43m 23s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/42be19e537d748eea08de6fa3114095b : SUCCESS in 2h 29m 47s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/3c3634cbb4ec4da8916a12d22dd576c8 : SUCCESS in 1h 00m 31s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/1a354646f3104b3da6716de724a07fc7 : SUCCESS in 1h 14m 00s\n- glance-grenade-centralized-cache https://zuul.opendev.org/t/openstack/build/5241924aedec4783a91f241225cfe33b : FAILURE in 22m 00s (non-voting)\n- glance-s3-multistore https://zuul.opendev.org/t/openstack/build/c32c624330d44dee9ede26eee86bc6f4 : SUCCESS in 2h 00m 11s (non-voting)\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/e56cc0281fdc4ebfb12fd4ad903b58de : SUCCESS in 18m 44s\n- ironic-tempest-bios-ipmi-direct https://zuul.opendev.org/t/openstack/build/6c57e034af91441fba6487dff524cfe1 : SUCCESS in 1h 10m 20s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"3fb2b84cee02bea8d8486f4fbba832217fbc6e37","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35676,"name":"Max","email":"max.lamprecht@digits.schwarz","username":"lamprechtm"},"date":"2026-02-16 10:33:19.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":2},{"id":"dca962b3ee48ec52b0f6aff6733ed960ea66a5b4","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-16 11:10:20.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/df2917f214ae4f1ebe650ce8e1be3725\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/eb008d2c654e4b29960874f324202363 : FAILURE in 9m 54s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/d2c24d7c3aa64b158062589db91c7683 : FAILURE in 9m 47s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/5e12daf168804c7ca076e48a668e5819 : RETRY_LIMIT in 6m 24s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"fd96b9f62286c6406a1852485a2541dbbd7b19ce","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-16 12:31:06.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/a51dc6763edb42378df69f957523723e\n\n- grenade https://zuul.opendev.org/t/openstack/build/7fa835b9d4a84ceaa5cd70e310243762 : SUCCESS in 46m 10s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/285ba2f803a84c85a2ef97d149ec083e : SUCCESS in 1h 06m 11s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/14e7f63f031a48de8c41c252053948b0 : SUCCESS in 1h 16m 22s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/25c90955982d410796b64ad377328d82 : SUCCESS in 59m 33s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9dad5fdb90294fedbf59bfd7f1b12f13 : SUCCESS in 4m 56s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/feafc6e72c744cb2be2a7a5bca5ca7bd : FAILURE in 4m 55s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/d7cb33b8eadc408a966fc0719afc2e51 : FAILURE in 6m 46s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/71c388b7e74e45f5b49e19c28ba4dfa6 : SUCCESS in 6m 01s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/0b459538b0ad4958a7cc869813c654dd : SUCCESS in 6m 04s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/c6f29b091e4d422daa06c60f35ce5a8f : SUCCESS in 7m 06s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/d351291143144d248258f2c97e8f047b : SUCCESS in 1h 15m 19s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/b36790da193049a1acda4aea0ae60950 : SUCCESS in 1h 14m 56s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/95b5f3725274470fa6428c1535bbd62c : FAILURE in 1h 50m 00s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/d4f709033b534ec5a09085598fac26b9 : SUCCESS in 1h 18m 21s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/9d1fbe83d8494800b6e9e85d577592ad : SUCCESS in 1h 03m 07s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/5fb59ca125ae4b21aa313d440f8c7b77 : SUCCESS in 1h 33m 27s\n- glance-grenade-centralized-cache https://zuul.opendev.org/t/openstack/build/89e9bad2901f4592a618cd82b426556b : FAILURE in 21m 25s (non-voting)\n- glance-s3-multistore https://zuul.opendev.org/t/openstack/build/c9377b2a87d34329b7beb2b3d58ee462 : SUCCESS in 1h 49m 34s (non-voting)\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/42e534d64be2486c9f41237cb1b037b9 : SUCCESS in 37m 13s\n- ironic-tempest-bios-ipmi-direct https://zuul.opendev.org/t/openstack/build/424f0bf7eee4489c98022453948f3604 : SUCCESS in 1h 10m 37s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"ee5fe43ecaf7db23e7dc5cc72f86a10befa4c5a2","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2026-02-17 05:55:31.000000000","message":"Patch Set 2: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"9787cfe3a9726b028121b262b79faf5d181957d3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35676,"name":"Max","email":"max.lamprecht@digits.schwarz","username":"lamprechtm"},"date":"2026-02-17 14:03:45.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified-1\n","accounts_in_message":[],"_revision_number":3},{"id":"309ed5d071cb02aa05e927517a9bd10bb47d8981","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-17 15:49:13.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/157abf4e95bf45fbbb7d936392759ba3\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/cf0e232d5f1747e4b7bbc64f5d61ca06 : SUCCESS in 6m 37s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/1bf74a8b5f2941a2ba76b7830dda17e6 : SUCCESS in 8m 30s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/688a1dc5c8fe4becb8670a7d972f19ab : RETRY_LIMIT in 3m 37s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"09715a38af59f771b954af863cc4ea318f1829fa","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-17 16:44:08.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/eb979b29b94442e08c94f751dcfbac45\n\n- grenade https://zuul.opendev.org/t/openstack/build/8b978c700b424cf28fde5e72fe5ac0ab : SUCCESS in 1h 01m 17s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/f48f289196a7424bbd5a00fe2baecb3e : SUCCESS in 58m 57s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/8a41b9825ace46feb332af337648b6f2 : SUCCESS in 1h 30m 23s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/d8c5e02ff6b74cf1babd540c57df3e90 : SUCCESS in 51m 13s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/40cf127662094e0baf396f7c7bebe6ff : SUCCESS in 5m 25s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/8f30b8daf9894deca0c9051307008147 : SUCCESS in 4m 35s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/4166c36a7dff41f598c778dd974a39d1 : SUCCESS in 5m 13s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ceace1f381b548bcb6deb1056b7f895d : SUCCESS in 3m 41s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/46ad56e884bc40b591093a7f5b566444 : SUCCESS in 6m 24s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/8c05c398f65e4bab8e26b7b8f555e862 : SUCCESS in 6m 04s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/57c7404bf0f74eab8da821e599ac6aaa : SUCCESS in 1h 44m 15s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/d0c1e1b89021471a9d77ae5578a381b4 : SUCCESS in 1h 04m 22s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/def889ae8632462d952d6ad62d741a3c : FAILURE in 1h 41m 18s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/2ae5377eb27241228b9d665706ca8adb : SUCCESS in 2h 34m 59s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/780e7b63b5724f74a58b9767b6440615 : SUCCESS in 50m 40s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/585b4376d55e489aa110f9ea6b9441f0 : SUCCESS in 1h 53m 51s\n- glance-grenade-centralized-cache https://zuul.opendev.org/t/openstack/build/6a894d35283f49a6a7cf800c58fd760e : FAILURE in 15m 26s (non-voting)\n- glance-s3-multistore https://zuul.opendev.org/t/openstack/build/fc1486e33eeb4392abbe3d14ef5ea7d5 : SUCCESS in 1h 49m 40s (non-voting)\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/b76def3de27345108be0d07bbf04432b : SUCCESS in 28m 44s","accounts_in_message":[],"_revision_number":3},{"id":"4131602d5443df09a57688c787a02515292bcc05","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2026-02-17 17:19:32.000000000","message":"Patch Set 3: Code-Review-1\n\n(3 comments)","accounts_in_message":[],"_revision_number":3},{"id":"3d280668172bccc37ed6abbef2fc0da5732b39b1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35676,"name":"Max","email":"max.lamprecht@digits.schwarz","username":"lamprechtm"},"date":"2026-02-18 13:16:50.000000000","message":"Uploaded patch set 4.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified-1\n","accounts_in_message":[],"_revision_number":4},{"id":"6edcb5ca708eac650fcc4b7f06710585ed109fc2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-18 15:27:15.000000000","message":"Patch Set 4: Verified-1\n\n(2 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/99606d1f62c84ab7a729876085f63430\n\n- grenade https://zuul.opendev.org/t/openstack/build/8587b739277e4a20aed3e8462d4700e1 : SUCCESS in 34m 54s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/f9b3cc823f8945a4b397df444d49882e : SUCCESS in 52m 06s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/808ae79509e043ec8a0a5d47c8839603 : SUCCESS in 1h 33m 15s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/c4305a7f3d7f47aaa5fd43423a6f886c : SUCCESS in 37m 43s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e9c7f7fdfe9a46a3814f7e1e995d3c61 : FAILURE in 3m 19s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/f0f8eb7bb35e44e8a83e0704b62af856 : SUCCESS in 3m 35s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/144b0153633e4e50af6db2fb2b4f11d6 : SUCCESS in 3m 52s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d775f6a49e88437387ee38367eb1cdcc : SUCCESS in 3m 38s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/d9fee8bfe3cc4c8390aa60626746dc0d : SUCCESS in 7m 03s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/21569975a7eb459eb1285d426af8f2a6 : SUCCESS in 5m 26s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/aad568d06682441cb1ff5e18972a5e8d : SUCCESS in 1h 03m 39s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/ba6b7849b1bd47438fc67b56f50b2c82 : SUCCESS in 1h 10m 33s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/69912045e66b48459933120ff70a9e3c : SUCCESS in 1h 57m 17s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/523455e2bf984475a01b9284eacd625d : SUCCESS in 2h 06m 42s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/8afd2cd5649642b1927dbe45d66a17b4 : SUCCESS in 53m 35s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/19b83b69b9844570ba3d36c477f7767b : SUCCESS in 1h 53m 17s\n- glance-grenade-centralized-cache https://zuul.opendev.org/t/openstack/build/b11ff841969a4b4783b789c51304831c : FAILURE in 16m 25s (non-voting)\n- glance-s3-multistore https://zuul.opendev.org/t/openstack/build/48a1656b28e940f3a3d8e02f00f1e886 : SUCCESS in 1h 01m 11s (non-voting)\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/13b49765b3e74bff907ee3eb7bc6aef6 : SUCCESS in 34m 24s","accounts_in_message":[],"_revision_number":4},{"id":"f24130c1f6b2526faf92f0dfd4b3473edf672902","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35676,"name":"Max","email":"max.lamprecht@digits.schwarz","username":"lamprechtm"},"date":"2026-02-18 15:56:18.000000000","message":"Uploaded patch set 5.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":5},{"id":"9550618c9e5137b9d50fb49a44e57270b206f005","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-18 18:26:10.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e10d8e60798b4cc5a0e47465956282ef\n\n- grenade https://zuul.opendev.org/t/openstack/build/0f229780a3b14eca841261bfeb8ee39b : SUCCESS in 32m 39s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/7729824626054b6fbe478e5f215d7b46 : SUCCESS in 33m 06s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/f14c3c3e869b476783beda392c06a3d1 : SUCCESS in 1h 32m 12s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/7fbab2428710442a8253e05fca64e430 : SUCCESS in 47m 57s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f12e66bf576b40a9bc983a1d954de3b8 : SUCCESS in 3m 18s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/5ae368bda5b54fe78615f26d6e1f1fec : SUCCESS in 3m 31s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/f8dd2fa888c44cafb771dd2344e9803d : SUCCESS in 3m 21s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c6a05fede0744e90a601103bdb06702e : SUCCESS in 4m 03s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/00406a009ec449aaa2a561281b39e33d : SUCCESS in 4m 31s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/8502fccf59e94348aed6e457bc284f90 : SUCCESS in 6m 30s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/6250b5663fa74b05879251ff1433f17a : SUCCESS in 1h 21m 33s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/a132030fc1fb45c7bcebe18f8af0be02 : SUCCESS in 2h 10m 30s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/2a739f37a12c47709fd627f9d39fb784 : SUCCESS in 1h 52m 12s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/2ffa08357c2e4ccf99d34012f76c644d : SUCCESS in 2h 24m 17s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/0fdd2ecafe7a4659a1d00a7a6e96ab50 : SUCCESS in 45m 06s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/06713bc4c5b146e7b1e9b0c6638564a4 : SUCCESS in 2h 02m 50s\n- glance-grenade-centralized-cache https://zuul.opendev.org/t/openstack/build/9067307c3bb94d1598f470871c154f08 : FAILURE in 11m 30s (non-voting)\n- glance-s3-multistore https://zuul.opendev.org/t/openstack/build/14944e4912884b14a4bbf94dbdd5ad5f : SUCCESS in 1h 27m 14s (non-voting)\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/db7ce43d35b54cf4bc2ca307bd9f6c95 : SUCCESS in 34m 42s","accounts_in_message":[],"_revision_number":5},{"id":"99ad650d8041f067d5256c12fbe26f8b88409613","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-18 21:46:35.000000000","message":"Patch Set 5:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/dbeb0277ff774e48a903afe933f6d83f\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/72e051ed600c4aa595118c9ec46e51bd : SUCCESS in 7m 31s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/b41bcb5685fa45478bbb3b33dfb3ca4a : SUCCESS in 8m 55s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/14c8e9d9d53d41bcba5144f44d9dd6a9 : RETRY_LIMIT in 4m 41s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"8b5e2e92ea6604a0458465fd40faf461eee69882","author":{"_account_id":35676,"name":"Max","email":"max.lamprecht@digits.schwarz","username":"lamprechtm"},"date":"2026-02-19 10:04:34.000000000","message":"Patch Set 5:\n\n(4 comments)","accounts_in_message":[],"_revision_number":5},{"id":"4a77f0e2cb6531285758e2b599002f452444cae1","author":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"date":"2026-02-23 11:29:25.000000000","message":"Patch Set 5: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"9d0d2f18787b657a063505faf5198e2c5cfff356","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"date":"2026-02-23 15:26:09.000000000","message":"Uploaded patch set 6.\n\nOutdated Votes:\n* Code-Review+1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":6},{"id":"0dcf0c63608021b57b62d48b4b11218ec6309e01","author":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"date":"2026-02-23 15:26:38.000000000","message":"Patch Set 6: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"f696ea9aba28f1242314a61e3aedaa3b2681decf","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-23 15:54:06.000000000","message":"Patch Set 6:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9d1bc145c3c748f1b95ba76c59608fcd\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/7de8e8ada6e8464f9849e55e4c923970 : SUCCESS in 6m 59s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/cd00969c0a934fc1ab9453d9c766ce3c : SUCCESS in 6m 29s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/006d40e29eae4a939f26b1cd9d8f4162 : RETRY_LIMIT in 4m 09s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"8aa8abeecec67054e02cf5d11e8e97d6c143934a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-23 17:36:37.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/298298a48abc4aaea299ca1be5face25\n\n- grenade https://zuul.opendev.org/t/openstack/build/c8ca36239e74429094887aafbbd03570 : SUCCESS in 1h 05m 06s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/72b7789d24404409ad918011cca5fa72 : SUCCESS in 1h 02m 27s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/d0681567e34c4f57b689b86ccf3295e4 : SUCCESS in 1h 22m 28s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/9763ec4bb5d84fd489a8dd29ab6fd563 : SUCCESS in 36m 07s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7667ccfd666941508cb86facb2efe04e : SUCCESS in 7m 32s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/c632da7b97ea49518cae00d79e53d58f : SUCCESS in 4m 53s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/40381d16ac464f94a8296f1843566606 : SUCCESS in 7m 10s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/a08edc6818474cd192126322684044d6 : SUCCESS in 7m 00s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dca3977025f645f0a6f61f1605aaff12 : SUCCESS in 8m 07s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/50d7eb0208b64e7c85a8829ff1b81fb0 : SUCCESS in 5m 29s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/eacadc2504c34eea8f3ffc2fdae16599 : SUCCESS in 6m 03s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/efcd27b8e4724384ac9e833387e599bc : SUCCESS in 6m 45s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/4db4f2bf6828431297657e7efdb23a2f : SUCCESS in 1h 53m 12s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/70bfdc470b7a49ebaca1670e23bffe92 : SUCCESS in 1h 05m 54s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/bc03acca32a048c0af28bbb8e71ebc43 : SUCCESS in 1h 10m 47s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/14520e80f2f44f3aa55af161618f0848 : SUCCESS in 1h 20m 47s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/f58450e1549b4b609c0d970f774d8d3c : SUCCESS in 1h 08m 03s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/5b9bd68398884cc0ab9810c2d6568977 : SUCCESS in 1h 35m 37s\n- glance-grenade-centralized-cache https://zuul.opendev.org/t/openstack/build/5f57fa3a20ab4acf8a8d83e6fdddb9d8 : FAILURE in 16m 32s (non-voting)\n- glance-s3-multistore https://zuul.opendev.org/t/openstack/build/2feb5ed17bf24ef39c32d44d6e1ccaf6 : SUCCESS in 1h 06m 30s (non-voting)\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/8f9bd434def140698fba7955d93dd31f : SUCCESS in 38m 48s","accounts_in_message":[],"_revision_number":6},{"id":"f7f8367edc6df0d005a97105a45f364104784b82","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2026-02-24 06:35:41.000000000","message":"Patch Set 6: Code-Review+2 Workflow+1\n\n(4 comments)","accounts_in_message":[],"_revision_number":6},{"id":"b8b0ce7bc8b60e25735d06461471fbdcc3efa854","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-24 06:36:06.000000000","message":"Patch Set 6: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":6},{"id":"b5a0813f9aeefec11812da41b168b6fd6fa53f10","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-24 08:34:01.000000000","message":"Patch Set 6: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1cf681896819410e93f50d73b73ddb3c\n\n- grenade https://zuul.opendev.org/t/openstack/build/6a536e4a893d4e93a1f026db9cd2b9bc : SUCCESS in 32m 39s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/5a7d4d17f424466f8e0f1ba8176f12d0 : SUCCESS in 33m 28s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/642a2f369e704ddbabf8f5d18ecd80f5 : SUCCESS in 1h 26m 41s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/bb91d2cf534d46258580e4a046065ba6 : SUCCESS in 46m 53s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/de557afcb9b1473d89e04d79b82c5a68 : SUCCESS in 4m 43s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/5157b91a1655423da7715370ddd7d797 : SUCCESS in 2m 38s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/c8b9d60d713d49478ba3630db4b2490b : SUCCESS in 2m 37s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/51f03eb6fd5741c2962dea5d541004b6 : SUCCESS in 5m 15s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/7753fd4993874ae29a4e77fb82de020e : SUCCESS in 7m 27s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/fb2fdf2bbb1a411a9c7dd31fd7011719 : SUCCESS in 7m 35s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/ca0c0cf6a2344a929d08e35758bc22ca : SUCCESS in 6m 25s\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/e2f07345b9f347f482a849a48f35be73 : SUCCESS in 1h 49m 26s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/172d7f30fd474820b08a3d4a21977c47 : SUCCESS in 1h 45m 08s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/5b63a529444c45adb229ce19d105efcf : SUCCESS in 43m 28s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/292070e0231c4f86af44e5af7168444e : SUCCESS in 1h 53m 34s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/8df71d660e6140cb84ed0cdd46eeeae3 : SUCCESS in 28m 07s","accounts_in_message":[],"_revision_number":6},{"id":"caf96236a381ff6bd4d9846a2db7383add83ce86","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-24 08:34:02.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":6},{"id":"224880f95f4eb5209f00c9c8eab8b742d8b99b68","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-24 08:35:19.000000000","message":"Patch Set 6:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f5f965eecfe44e4ca45ab6cf2463aadc\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/536f5ec057ca4dd183cfbf16da7b831e : SUCCESS in 48s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/6340ca4b28654a1d97f307915ac6e9c3 : SUCCESS in 54s","accounts_in_message":[],"_revision_number":6}],"current_revision_number":6,"current_revision":"8d6333b46277f48fa41fb11026fd83b4bf2efe4c","revisions":{"454cf7d0f67f8283ee4b6f081ae2d13c0df66d4b":{"kind":"REWORK","_number":1,"created":"2026-02-12 14:14:13.000000000","uploader":{"_account_id":35676,"name":"Max","email":"max.lamprecht@digits.schwarz","username":"lamprechtm"},"ref":"refs/changes/27/976627/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/27/976627/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/27/976627/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/27/976627/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/27/976627/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/27/976627/1"}}},"commit":{"parents":[{"commit":"96040c00a65c3c73ca11e771663b0d7f01415fe0","subject":"Merge \"Deprecate the glance-cache-{cleaner,prefetcher,pruner} commands\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/96040c00a65c3c73ca11e771663b0d7f01415fe0"}]}],"author":{"name":"Max Lamprecht","email":"max.lamprecht@stackit.cloud","date":"2026-02-11 09:37:37.000000000","tz":60},"committer":{"name":"Max Lamprecht","email":"max.lamprecht@stackit.cloud","date":"2026-02-12 14:13:34.000000000","tz":60},"subject":"feat: return size in get_image_data_iter","message":"feat: return size in get_image_data_iter\n\nReturn the size with the data iter in get_image_data_iter(uri).\nWe will either fetch it from the local filesize or from the http\n\u0027content-length\u0027 header.\n\nFor example the s3 store driver needs to know the image size in order\nto decide between singlepart and multipart upload.\nWithout the size the s3 driver will always attempt to use singlepart\nwhich can fail for larger objects. e.g. aws part size of 5GB\n\nCloses-Bug: #2141673\nChange-Id: Ief14a61f6a1d9327acfd6ada61c2b5976d153555\nSigned-off-by: Max Lamprecht \u003cmax.lamprecht@stackit.cloud\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/454cf7d0f67f8283ee4b6f081ae2d13c0df66d4b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/454cf7d0f67f8283ee4b6f081ae2d13c0df66d4b"}]},"branch":"refs/heads/master"},"f6531750729961671986ff5433c60de3d6c8271b":{"kind":"REWORK","_number":2,"created":"2026-02-16 10:33:19.000000000","uploader":{"_account_id":35676,"name":"Max","email":"max.lamprecht@digits.schwarz","username":"lamprechtm"},"ref":"refs/changes/27/976627/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/27/976627/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/27/976627/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/27/976627/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/27/976627/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/27/976627/2"}}},"commit":{"parents":[{"commit":"9987754ba4a29ff6a811cbae354dd19df4d7cf83","subject":"Merge \"Update unit tests to use native threading\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/9987754ba4a29ff6a811cbae354dd19df4d7cf83"}]}],"author":{"name":"Max Lamprecht","email":"max.lamprecht@stackit.cloud","date":"2026-02-11 09:37:37.000000000","tz":60},"committer":{"name":"Max Lamprecht","email":"max.lamprecht@stackit.cloud","date":"2026-02-16 10:32:43.000000000","tz":60},"subject":"feat: return size in get_image_data_iter","message":"feat: return size in get_image_data_iter\n\nReturn the size with the data iter in get_image_data_iter(uri).\nWe will either fetch it from the local filesize or from the http\n\u0027content-length\u0027 header.\n\nFor example the s3 store driver needs to know the image size in order\nto decide between singlepart and multipart upload.\nWithout the size the s3 driver will always attempt to use singlepart\nwhich can fail for larger objects. e.g. aws part size of 5GB\n\nCloses-Bug: #2141673\nChange-Id: Ief14a61f6a1d9327acfd6ada61c2b5976d153555\nSigned-off-by: Max Lamprecht \u003cmax.lamprecht@stackit.cloud\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/f6531750729961671986ff5433c60de3d6c8271b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/f6531750729961671986ff5433c60de3d6c8271b"}]},"branch":"refs/heads/master"},"3e386fd4338a893d3a9fce85c16da14789225f07":{"kind":"REWORK","_number":3,"created":"2026-02-17 14:03:45.000000000","uploader":{"_account_id":35676,"name":"Max","email":"max.lamprecht@digits.schwarz","username":"lamprechtm"},"ref":"refs/changes/27/976627/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/27/976627/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/27/976627/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/27/976627/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/27/976627/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/27/976627/3"}}},"commit":{"parents":[{"commit":"9987754ba4a29ff6a811cbae354dd19df4d7cf83","subject":"Merge \"Update unit tests to use native threading\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/9987754ba4a29ff6a811cbae354dd19df4d7cf83"}]}],"author":{"name":"Max Lamprecht","email":"max.lamprecht@stackit.cloud","date":"2026-02-11 09:37:37.000000000","tz":60},"committer":{"name":"Max Lamprecht","email":"max.lamprecht@stackit.cloud","date":"2026-02-17 14:03:30.000000000","tz":60},"subject":"feat: return size in get_image_data_iter","message":"feat: return size in get_image_data_iter\n\nReturn the size with the data iter in get_image_data_iter(uri).\nWe will either fetch it from the local filesize or from the http\n\u0027content-length\u0027 header.\n\nFor example the s3 store driver needs to know the image size in order\nto decide between singlepart and multipart upload.\nWithout the size the s3 driver will always attempt to use singlepart\nwhich can fail for larger objects. e.g. aws part size of 5GB\n\nCloses-Bug: #2141673\nChange-Id: Ief14a61f6a1d9327acfd6ada61c2b5976d153555\nSigned-off-by: Max Lamprecht \u003cmax.lamprecht@stackit.cloud\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/3e386fd4338a893d3a9fce85c16da14789225f07"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/3e386fd4338a893d3a9fce85c16da14789225f07"}]},"branch":"refs/heads/master"},"3810f2215af3db0e37705abeff13cd71f88872f1":{"kind":"REWORK","_number":4,"created":"2026-02-18 13:16:50.000000000","uploader":{"_account_id":35676,"name":"Max","email":"max.lamprecht@digits.schwarz","username":"lamprechtm"},"ref":"refs/changes/27/976627/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/27/976627/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/27/976627/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/27/976627/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/27/976627/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/27/976627/4"}}},"commit":{"parents":[{"commit":"9987754ba4a29ff6a811cbae354dd19df4d7cf83","subject":"Merge \"Update unit tests to use native threading\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/9987754ba4a29ff6a811cbae354dd19df4d7cf83"}]}],"author":{"name":"Max Lamprecht","email":"max.lamprecht@stackit.cloud","date":"2026-02-11 09:37:37.000000000","tz":60},"committer":{"name":"Max Lamprecht","email":"max.lamprecht@stackit.cloud","date":"2026-02-18 13:16:03.000000000","tz":60},"subject":"feat: return size in get_image_data_iter","message":"feat: return size in get_image_data_iter\n\nReturn the size with the data iter in get_image_data_iter(uri).\nWe will either fetch it from the local filesize or from the http\n\u0027content-length\u0027 header.\n\nFor example the s3 store driver needs to know the image size in order\nto decide between singlepart and multipart upload.\nWithout the size the s3 driver will always attempt to use singlepart\nwhich can fail for larger objects. e.g. aws part size of 5GB\n\nCloses-Bug: #2141673\nChange-Id: Ief14a61f6a1d9327acfd6ada61c2b5976d153555\nSigned-off-by: Max Lamprecht \u003cmax.lamprecht@stackit.cloud\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/3810f2215af3db0e37705abeff13cd71f88872f1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/3810f2215af3db0e37705abeff13cd71f88872f1"}]},"branch":"refs/heads/master"},"40202b08b4f7c0a9a63e58af29cf41686b6e71e1":{"kind":"REWORK","_number":5,"created":"2026-02-18 15:56:18.000000000","uploader":{"_account_id":35676,"name":"Max","email":"max.lamprecht@digits.schwarz","username":"lamprechtm"},"ref":"refs/changes/27/976627/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/27/976627/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/27/976627/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/27/976627/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/27/976627/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/27/976627/5"}}},"commit":{"parents":[{"commit":"9987754ba4a29ff6a811cbae354dd19df4d7cf83","subject":"Merge \"Update unit tests to use native threading\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/9987754ba4a29ff6a811cbae354dd19df4d7cf83"}]}],"author":{"name":"Max Lamprecht","email":"max.lamprecht@stackit.cloud","date":"2026-02-11 09:37:37.000000000","tz":60},"committer":{"name":"Max Lamprecht","email":"max.lamprecht@stackit.cloud","date":"2026-02-18 15:56:02.000000000","tz":60},"subject":"feat: return size in get_image_data_iter","message":"feat: return size in get_image_data_iter\n\nReturn the size with the data iter in get_image_data_iter(uri).\nWe will either fetch it from the local filesize or from the http\n\u0027content-length\u0027 header.\n\nFor example the s3 store driver needs to know the image size in order\nto decide between singlepart and multipart upload.\nWithout the size the s3 driver will always attempt to use singlepart\nwhich can fail for larger objects. e.g. aws part size of 5GB\n\nCloses-Bug: #2141673\nChange-Id: Ief14a61f6a1d9327acfd6ada61c2b5976d153555\nSigned-off-by: Max Lamprecht \u003cmax.lamprecht@stackit.cloud\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/40202b08b4f7c0a9a63e58af29cf41686b6e71e1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/40202b08b4f7c0a9a63e58af29cf41686b6e71e1"}]},"branch":"refs/heads/master"},"8d6333b46277f48fa41fb11026fd83b4bf2efe4c":{"kind":"REWORK","_number":6,"created":"2026-02-23 15:26:09.000000000","uploader":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"ref":"refs/changes/27/976627/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/27/976627/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/27/976627/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/27/976627/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/27/976627/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/27/976627/6"}}},"commit":{"parents":[{"commit":"9987754ba4a29ff6a811cbae354dd19df4d7cf83","subject":"Merge \"Update unit tests to use native threading\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/9987754ba4a29ff6a811cbae354dd19df4d7cf83"}]}],"author":{"name":"Max Lamprecht","email":"max.lamprecht@stackit.cloud","date":"2026-02-11 09:37:37.000000000","tz":60},"committer":{"name":"Cyril Roelandt","email":"cyril@redhat.com","date":"2026-02-23 15:26:02.000000000","tz":60},"subject":"feat: return size in get_image_data_iter","message":"feat: return size in get_image_data_iter\n\nReturn the size with the data iter in get_image_data_iter(uri).\nWe will either fetch it from the local filesize or from the http\n\u0027content-length\u0027 header.\n\nFor example the s3 store driver needs to know the image size in order\nto decide between singlepart and multipart upload.\nWithout the size the s3 driver will always attempt to use singlepart\nwhich can fail for larger objects. e.g. aws part size of 5GB\n\nCloses-Bug: #2141673\nChange-Id: Ief14a61f6a1d9327acfd6ada61c2b5976d153555\nSigned-off-by: Max Lamprecht \u003cmax.lamprecht@stackit.cloud\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/8d6333b46277f48fa41fb11026fd83b4bf2efe4c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/8d6333b46277f48fa41fb11026fd83b4bf2efe4c"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
