)]}'
{"id":"openstack%2Fglance~978088","triplet_id":"openstack%2Fglance~master~I7bdf1b7a91de29cf254ed7ef4e368fc3b16c18f5","project":"openstack/glance","branch":"master","topic":"LUKS-image-encryption","attention_set":{"4393":{"account":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"last_update":"2026-02-26 18:04:53.000000000","reason":"A robot voted negatively on a label"}},"removed_from_attention_set":{},"hashtags":[],"change_id":"I7bdf1b7a91de29cf254ed7ef4e368fc3b16c18f5","subject":"DNM: Test glance against LUKS images","status":"NEW","created":"2026-02-26 15:42:15.000000000","updated":"2026-02-27 22:51:44.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":false,"submittable":false,"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"dcbcd8025b221f52e8fb7b12372ccb07bc41c8a8","_number":978088,"virtual_id_number":978088,"owner":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2026-02-27 22:51:44.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Top Priority / Holds Gate"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-02-26 15:58:34.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2026-02-26 18:04:53.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"ecc864d9ede9717733a6fb44c2843399f8783cea","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2026-02-26 15:42:15.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"fbfa7c82853eebd8e12cbb20cee8198fbe9d5ae1","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-26 15:58:34.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b56b8cb45d87420882f913294fea61b8\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/a790e4905805452bae3de04273bd06df : SUCCESS in 5m 39s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/3dfa1d6955254c08a945a8843366478d : RETRY_LIMIT in 3m 51s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/fc065047232a40edb3a19a10f42aa34f : SUCCESS in 9m 30s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"e75738a9ac51e583d58b6997eb8c65917959d351","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-26 18:04:53.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/7f2273c2499d4d84acaaf67fbdeab3aa\n\n- grenade https://zuul.opendev.org/t/openstack/build/6249ea53283b42689bc68fc21813ce28 : SUCCESS in 1h 03m 59s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/c31af6b134eb4f2aadce57b22a447d14 : SUCCESS in 51m 36s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/6b29d49fb26545af87d96aa499d3e562 : FAILURE in 1h 22m 36s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/158b5ab98e334244ac459ffbf2b1a428 : SUCCESS in 1h 01m 00s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ce9a392005b144ecaac3ba7f75eda469 : SUCCESS in 4m 18s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/e5ded4dcb6e847e69a931ab544f5ba51 : SUCCESS in 6m 13s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/2624145f81dd49cd965ec52b74902b7f : SUCCESS in 4m 46s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/559abb2eff264c9c9f05556a3edaa10c : SUCCESS in 10m 20s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/216d82b52f334ad7952535c5ceb50d70 : SUCCESS in 3m 56s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/ded1cf2119e349e9a33762ce000036a9 : SUCCESS in 6m 49s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/d5bf8204d9a84c9bad446f572c764a56 : SUCCESS in 5m 04s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/29be15964e0748e1b76a42ef6d41b370 : SUCCESS in 1h 43m 51s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/4eefe1fc944c499abd992c4f60d2260c : FAILURE in 58m 38s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/2d04615032c146d4a9039ee624d98408 : FAILURE in 2h 10m 24s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/363c3960abda45a7ad37c51397ae6278 : FAILURE in 1h 40m 42s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/7d182baaecc14d3497f1cb00a91fb6d0 : SUCCESS in 1h 02m 07s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/af4235729e504384801a8ccc289a6bff : SUCCESS in 56m 49s\n- glance-grenade-centralized-cache https://zuul.opendev.org/t/openstack/build/2d5913f690ad4c81931922fc82390e45 : FAILURE in 13m 11s (non-voting)\n- glance-s3-multistore https://zuul.opendev.org/t/openstack/build/284313fc2b8e44c88ac6f3443663e0e6 : FAILURE in 1h 51m 33s (non-voting)\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/2413c67d61b8468590fd5833487f4440 : SUCCESS in 33m 52s","accounts_in_message":[],"_revision_number":1},{"id":"4655453c4c0aeb83770009da26ca9fdc45fedff9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2026-02-26 18:05:31.000000000","message":"Uploaded patch set 2: Commit message was updated.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"4e1e914c7470452110f4cb28d85c652db779738b","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-26 19:03:48.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4ebcbdf00ebb4ad18b18dcc5b91e12fe\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/652c0b93d4ad479381f9f9d1161f4632 : SUCCESS in 8m 23s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/0a6f3dff86fd4c159ce4dfa8fd4850b6 : RETRY_LIMIT in 4m 29s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/1346bee4e2684cef857a2f0a81b6a256 : SUCCESS in 11m 31s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"fe1bed79e82f6e1a18bec31f48a99d5d5364fa10","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-26 19:59:31.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/ca88be73af18465a99b3d4de0a88fa47\n\n- grenade https://zuul.opendev.org/t/openstack/build/b7c768c088ff48bc93be952b93cbfe3c : SUCCESS in 36m 37s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/eea07c8012994579a49fe763b3e6f2f2 : SUCCESS in 33m 20s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/f5c48253718d406d9879ba4cadf3a871 : FAILURE in 1h 16m 12s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/68b4517201fc4a9c9d30e2273b4b160a : SUCCESS in 34m 47s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8c6f92a17c4d4f85bf06f8569dca2aae : SUCCESS in 6m 17s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/9f288fce57bd4983b43b67c8997e202b : SUCCESS in 4m 09s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/13f56efca8ce4ec9a7f6a892b2989a6f : SUCCESS in 3m 14s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/af175a2f82d141cba9b53967cb3f7c3f : SUCCESS in 5m 13s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3be329ab5c56420ba6c3da33b68d9c42 : SUCCESS in 4m 15s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/d7e12299d03548bbbf5253f154f84380 : SUCCESS in 7m 52s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/c48b50ab55524267aa946d36417d4fca : SUCCESS in 6m 23s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/ae1dec143bf64c2e9a1d29bf17bc55e5 : SUCCESS in 1h 05m 52s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/51fa95018b7a4378ac387379f01fcd58 : FAILURE in 1h 08m 25s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/659674b78a3c44ddb25abdab3db995e6 : FAILURE in 1h 42m 33s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/66d48a58071a4ad2914e36b82dafa10b : FAILURE in 1h 11m 30s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/9df399737e244b5db7b683c4ba135fd0 : SUCCESS in 49m 21s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/a08595a639844500bf200f6182a87151 : SUCCESS in 1h 02m 32s\n- glance-grenade-centralized-cache https://zuul.opendev.org/t/openstack/build/131819e920694c86a435c39ea4443d5e : FAILURE in 15m 19s (non-voting)\n- glance-s3-multistore https://zuul.opendev.org/t/openstack/build/f456ce7a49a046d398544dbcb720dc9a : FAILURE in 1h 45m 22s (non-voting)\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/1521e5b2189e472ea0f464096a49170a : SUCCESS in 30m 47s","accounts_in_message":[],"_revision_number":2},{"id":"5c27eceff217ab62d3ed6695c1700faf721e69fe","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2026-02-27 15:26:43.000000000","message":"Uploaded patch set 3: Commit message was updated.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":3},{"id":"27de04de0c0c9d2c03dd79da56d4739a460f065f","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-27 17:40:18.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a38cd28f99f74f28a8c72a538f785cc0\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/8abab24ac717483a8b23536f0beeb2ef : SUCCESS in 7m 25s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/e8cd77ddf5ce4ecc9151b1921e292ff0 : RETRY_LIMIT in 4m 02s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/009d2f32cf964b80bc50457f7f21036a : SUCCESS in 8m 14s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"576f4cdf200ab63abb6243de984d04996a7aa8ee","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-27 17:56:10.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/0290aab2775b4de088fd77c2c796df34\n\n- grenade https://zuul.opendev.org/t/openstack/build/907dda4bd4e24032a0d0b8ecc50485c8 : SUCCESS in 1h 16m 51s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/9d85ad50e35c48078d2280b3dbb5a746 : SUCCESS in 33m 17s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/db99c52070c64c9cbd7bf6490cdce80d : FAILURE in 1h 46m 55s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/d7617b76d626420a9753a6ef5f9300f8 : SUCCESS in 1h 03m 43s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/07a0060164fc40a58e7dd10b8c2c92a2 : SUCCESS in 5m 41s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/3e5ca06a0ac04c18a8416e9f22514634 : SUCCESS in 3m 14s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/e647d9efe677415caa11eb3d966d6a3b : SUCCESS in 5m 43s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/7a44c90b866240ffb6d2a52f1ca8c2b7 : SUCCESS in 8m 01s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/47e1a2305cd148b3ac53a75a4bda4d29 : SUCCESS in 6m 30s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/0cf94158306c4dcd983a0aa07db3d386 : SUCCESS in 8m 34s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/4901f552c344482f8719f03eb793b0e2 : SUCCESS in 4m 55s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/5285c18cb4e945169ef684d0be13507e : SUCCESS in 2h 02m 01s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/8560a65042e4435eb3cc723571818731 : FAILURE in 1h 45m 54s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/e2076a17db1f4c6097ffdac54c9a9539 : FAILURE in 2h 10m 05s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/b35d75e18d7b46e6a6f2e7b361a2f14e : FAILURE in 2h 18m 15s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/7b91282913424c6d8c5addf5c86ae4ed : SUCCESS in 1h 08m 24s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/bf81d5b614a047fd818112ade8abfbb9 : SUCCESS in 1h 47m 52s\n- glance-grenade-centralized-cache https://zuul.opendev.org/t/openstack/build/57291001080f44f7b749f7d0f3e5f394 : FAILURE in 25m 57s (non-voting)\n- glance-s3-multistore https://zuul.opendev.org/t/openstack/build/2b6c9ff63c9943cb8b94ce22cc642562 : FAILURE in 1h 47m 36s (non-voting)\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/eb3de5d3df1b4626aef8a85ca2c34c8f : SUCCESS in 38m 23s","accounts_in_message":[],"_revision_number":3},{"id":"9b5fe5192b8fb470fa258d8c20285cc8db265f01","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2026-02-27 18:02:33.000000000","message":"Uploaded patch set 4.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":4},{"id":"5145e36761cfa885f1b42b2ec37543168a832b40","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2026-02-27 18:03:20.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"eec17f0bd3fc4496ad51bc519878e1156d3eb9e0","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-27 18:04:41.000000000","message":"Patch Set 5:\n\nZuul encountered a syntax error while parsing its\nconfiguration in the repo openstack/glance on branch master.  The\nproblem was:\n\n  found undefined alias\n    in \"openstack/glance/.zuul.yaml@master\", line 347, column 31","accounts_in_message":[],"_revision_number":5},{"id":"f350a8558e7e1cc85e6b23eb5436b49952f6df05","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-27 18:05:03.000000000","message":"Patch Set 5: Verified-1\n\nZuul encountered a syntax error while parsing its\nconfiguration in the repo openstack/glance on branch master.  The\nproblem was:\n\n  found undefined alias\n    in \"openstack/glance/.zuul.yaml@master\", line 347, column 31","accounts_in_message":[],"_revision_number":5},{"id":"2f1d976b7f2935f8bb60c4676fe05e1514ca0f59","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2026-02-27 18:08:13.000000000","message":"Uploaded patch set 6.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":6},{"id":"0352144adc4424a5525a292d3a40b0a8821280dc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-27 20:31:58.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/92bce20edc4044b590d4c5b78171d10a\n\n- grenade https://zuul.opendev.org/t/openstack/build/c19b09bff39643ad86edd1641a85adf4 : SUCCESS in 1h 02m 12s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/4794189c98dd4865b64baef2274a40a9 : SUCCESS in 56m 47s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/a0755e6f48284554bda7a44f6d404316 : SUCCESS in 1h 26m 04s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/369da72e7a844859b7f442b7e49be046 : SUCCESS in 1h 03m 04s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cb98c473171e47678d1958ffe3d4d497 : SUCCESS in 4m 41s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/c63abd94024448d59bbe3b35a1836406 : SUCCESS in 4m 37s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/36984b9ae1a743e6989331fa9c9b3015 : SUCCESS in 7m 01s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/72019516f4cd4029bc58de4dddd32339 : SUCCESS in 4m 14s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/32a8cf32a6e843328d60bcd5c68c1a26 : SUCCESS in 4m 54s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/e0a9067814ee4a198f401fe42f0d5def : SUCCESS in 6m 32s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/aa511955937c4374824cebd9ee44eaeb : SUCCESS in 8m 25s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/e4187555949c49929d5e93a1238bf2ec : SUCCESS in 2h 17m 13s (non-voting)\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/4947cae05822418e96e78185eeb18c6e : FAILURE in 1h 05m 38s","accounts_in_message":[],"_revision_number":6},{"id":"f432226284fc22c89b5de75130c728dd20d693a3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2026-02-27 20:38:26.000000000","message":"Uploaded patch set 7.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":7},{"id":"71016cab4aa2b1afdc27c2965a5b98dbc0f6f2fb","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2026-02-27 20:49:01.000000000","message":"Uploaded patch set 8.","accounts_in_message":[],"_revision_number":8},{"id":"8699fb62cfc96515dba065c605020def2aa125c0","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-27 22:04:18.000000000","message":"Patch Set 8:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/54531e3d1f914d78a45936c8a1fa8d7e\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/68f18b994422459f827a264866a5a9f0 : SUCCESS in 6m 13s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/84be28be7c384ca6808149d22a4c82ba : RETRY_LIMIT in 4m 01s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/f893c09b333d4a82b7c2eafca4ca7357 : SUCCESS in 8m 18s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"dcbcd8025b221f52e8fb7b12372ccb07bc41c8a8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-27 22:51:44.000000000","message":"Patch Set 8: Verified-1\n\n(3 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/75830c86d64c4f969ee9621d62a6e4b5\n\n- grenade https://zuul.opendev.org/t/openstack/build/417cbfe5e4c6484292e305aa76f41052 : SUCCESS in 31m 57s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/4587f4ce80244c90be7bb044b167e0ff : SUCCESS in 1h 03m 51s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/07dfeb0d60bf4237baf7be1c538b7005 : SUCCESS in 1h 30m 06s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/88b5e911c1cc4766a2beab8d54e28d96 : SUCCESS in 1h 02m 34s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/04381682e17f45ee97603127f436849a : FAILURE in 6m 15s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/c86df230ea144feb9b36f68adb90a054 : SUCCESS in 5m 07s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/8cf9ab706009496a9ba2d65bab5874ff : SUCCESS in 5m 25s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/907b7c9a78e74147a81b45d22759f54a : SUCCESS in 7m 36s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2e42817230944dce8d1351031b434a30 : SUCCESS in 5m 47s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/1d38210918694c3a91f0b3ae895c4652 : SUCCESS in 6m 54s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/df2b603900ed499cb5febe0f601d35da : SUCCESS in 7m 13s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/c8658778e5024d37958500c283be7e10 : SUCCESS in 1h 50m 46s (non-voting)\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/84303a77ce5941e0a8c3125b3eeed4af : FAILURE in 1h 54m 39s","accounts_in_message":[],"_revision_number":8}],"current_revision_number":8,"current_revision":"e7f9bedfcc0b00d6e6507b563edc0c45017182dd","revisions":{"7e7e2f0d78d9c8051cbd7ab6dd2241223ad80df3":{"kind":"REWORK","_number":1,"created":"2026-02-26 15:42:15.000000000","uploader":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"ref":"refs/changes/88/978088/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/88/978088/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/88/978088/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/88/978088/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/88/978088/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/88/978088/1"}}},"commit":{"parents":[{"commit":"8070fae771718eda7f626844ca84a949f77b2dd5","subject":"Merge \"feat: return size in get_image_data_iter\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/8070fae771718eda7f626844ca84a949f77b2dd5"}]}],"author":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2026-02-26 15:41:31.000000000","tz":-480},"committer":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2026-02-26 15:42:13.000000000","tz":-480},"subject":"DNM: Test glance against LUKS images","message":"DNM: Test glance against LUKS images\n\nDepends-On: https://review.opendev.org/c/openstack/os-test-images/+/926942\nChange-Id: I7bdf1b7a91de29cf254ed7ef4e368fc3b16c18f5\nSigned-off-by: Dan Smith \u003cdansmith@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/7e7e2f0d78d9c8051cbd7ab6dd2241223ad80df3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/7e7e2f0d78d9c8051cbd7ab6dd2241223ad80df3"}]},"branch":"refs/heads/master"},"50ef4b12b365226d97089b83f1ed11c206337614":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2026-02-26 18:05:31.000000000","uploader":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"ref":"refs/changes/88/978088/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/88/978088/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/88/978088/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/88/978088/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/88/978088/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/88/978088/2"}}},"commit":{"parents":[{"commit":"8070fae771718eda7f626844ca84a949f77b2dd5","subject":"Merge \"feat: return size in get_image_data_iter\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/8070fae771718eda7f626844ca84a949f77b2dd5"}]}],"author":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2026-02-26 15:41:31.000000000","tz":-480},"committer":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2026-02-26 17:02:11.000000000","tz":-480},"subject":"DNM: Test glance against LUKS images","message":"DNM: Test glance against LUKS images\n\nDepends-On: https://review.opendev.org/c/openstack/os-test-images/+/926942\nDepends-On: https://review.opendev.org/c/openstack/oslo.utils/+/978097\nDepends-On: https://review.opendev.org/c/openstack/glance/+/926295\nChange-Id: I7bdf1b7a91de29cf254ed7ef4e368fc3b16c18f5\nSigned-off-by: Dan Smith \u003cdansmith@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/50ef4b12b365226d97089b83f1ed11c206337614"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/50ef4b12b365226d97089b83f1ed11c206337614"}]},"branch":"refs/heads/master"},"28c4d5e5d7bb2c003373674645df4db792e5eeed":{"kind":"NO_CODE_CHANGE","_number":3,"created":"2026-02-27 15:26:43.000000000","uploader":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"ref":"refs/changes/88/978088/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/88/978088/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/88/978088/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/88/978088/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/88/978088/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/88/978088/3"}}},"commit":{"parents":[{"commit":"8070fae771718eda7f626844ca84a949f77b2dd5","subject":"Merge \"feat: return size in get_image_data_iter\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/8070fae771718eda7f626844ca84a949f77b2dd5"}]}],"author":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2026-02-26 15:41:31.000000000","tz":-480},"committer":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2026-02-26 19:11:58.000000000","tz":-480},"subject":"DNM: Test glance against LUKS images","message":"DNM: Test glance against LUKS images\n\nDepends-On: https://review.opendev.org/c/openstack/os-test-images/+/926942\nDepends-On: https://review.opendev.org/c/openstack/oslo.utils/+/978097\nDepends-On: https://review.opendev.org/c/openstack/glance/+/926295\nDepends-On: https://review.opendev.org/c/openstack/tempest/+/978215\nChange-Id: I7bdf1b7a91de29cf254ed7ef4e368fc3b16c18f5\nSigned-off-by: Dan Smith \u003cdansmith@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/28c4d5e5d7bb2c003373674645df4db792e5eeed"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/28c4d5e5d7bb2c003373674645df4db792e5eeed"}]},"branch":"refs/heads/master"},"ffb714fe592abd1b319186f839df22e246f1d521":{"kind":"REWORK","_number":4,"created":"2026-02-27 18:02:33.000000000","uploader":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"ref":"refs/changes/88/978088/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/88/978088/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/88/978088/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/88/978088/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/88/978088/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/88/978088/4"}}},"commit":{"parents":[{"commit":"8070fae771718eda7f626844ca84a949f77b2dd5","subject":"Merge \"feat: return size in get_image_data_iter\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/8070fae771718eda7f626844ca84a949f77b2dd5"}]}],"author":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2026-02-26 15:41:31.000000000","tz":-480},"committer":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2026-02-26 22:01:31.000000000","tz":-480},"subject":"DNM: Test glance against LUKS images","message":"DNM: Test glance against LUKS images\n\nDepends-On: https://review.opendev.org/c/openstack/os-test-images/+/926942\nDepends-On: https://review.opendev.org/c/openstack/oslo.utils/+/978097\nDepends-On: https://review.opendev.org/c/openstack/glance/+/926295\nDepends-On: https://review.opendev.org/c/openstack/tempest/+/978215\nChange-Id: I7bdf1b7a91de29cf254ed7ef4e368fc3b16c18f5\nSigned-off-by: Dan Smith \u003cdansmith@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/ffb714fe592abd1b319186f839df22e246f1d521"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/ffb714fe592abd1b319186f839df22e246f1d521"}]},"branch":"refs/heads/master"},"1b00f960c6065068a80706d056ff61b806631c5f":{"kind":"REWORK","_number":5,"created":"2026-02-27 18:03:20.000000000","uploader":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"ref":"refs/changes/88/978088/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/88/978088/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/88/978088/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/88/978088/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/88/978088/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/88/978088/5"}}},"commit":{"parents":[{"commit":"8070fae771718eda7f626844ca84a949f77b2dd5","subject":"Merge \"feat: return size in get_image_data_iter\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/8070fae771718eda7f626844ca84a949f77b2dd5"}]}],"author":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2026-02-26 15:41:31.000000000","tz":-480},"committer":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2026-02-26 22:02:55.000000000","tz":-480},"subject":"DNM: Test glance against LUKS images","message":"DNM: Test glance against LUKS images\n\nDepends-On: https://review.opendev.org/c/openstack/os-test-images/+/926942\nDepends-On: https://review.opendev.org/c/openstack/oslo.utils/+/978097\nDepends-On: https://review.opendev.org/c/openstack/glance/+/926295\nDepends-On: https://review.opendev.org/c/openstack/tempest/+/978215\nChange-Id: I7bdf1b7a91de29cf254ed7ef4e368fc3b16c18f5\nSigned-off-by: Dan Smith \u003cdansmith@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/1b00f960c6065068a80706d056ff61b806631c5f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/1b00f960c6065068a80706d056ff61b806631c5f"}]},"branch":"refs/heads/master"},"0cc081b43307abae1d2aa6da4ce84a6342787d50":{"kind":"REWORK","_number":6,"created":"2026-02-27 18:08:13.000000000","uploader":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"ref":"refs/changes/88/978088/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/88/978088/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/88/978088/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/88/978088/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/88/978088/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/88/978088/6"}}},"commit":{"parents":[{"commit":"8070fae771718eda7f626844ca84a949f77b2dd5","subject":"Merge \"feat: return size in get_image_data_iter\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/8070fae771718eda7f626844ca84a949f77b2dd5"}]}],"author":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2026-02-26 15:41:31.000000000","tz":-480},"committer":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2026-02-26 22:07:13.000000000","tz":-480},"subject":"DNM: Test glance against LUKS images","message":"DNM: Test glance against LUKS images\n\nDepends-On: https://review.opendev.org/c/openstack/os-test-images/+/926942\nDepends-On: https://review.opendev.org/c/openstack/oslo.utils/+/978097\nDepends-On: https://review.opendev.org/c/openstack/glance/+/926295\nDepends-On: https://review.opendev.org/c/openstack/tempest/+/978215\nChange-Id: I7bdf1b7a91de29cf254ed7ef4e368fc3b16c18f5\nSigned-off-by: Dan Smith \u003cdansmith@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/0cc081b43307abae1d2aa6da4ce84a6342787d50"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/0cc081b43307abae1d2aa6da4ce84a6342787d50"}]},"branch":"refs/heads/master"},"d1bdd47447ec8a33f7ac4d2f232db8ed158dbe53":{"kind":"REWORK","_number":7,"created":"2026-02-27 20:38:26.000000000","uploader":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"ref":"refs/changes/88/978088/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/88/978088/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/88/978088/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/88/978088/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/88/978088/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/88/978088/7"}}},"commit":{"parents":[{"commit":"8070fae771718eda7f626844ca84a949f77b2dd5","subject":"Merge \"feat: return size in get_image_data_iter\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/8070fae771718eda7f626844ca84a949f77b2dd5"}]}],"author":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2026-02-26 15:41:31.000000000","tz":-480},"committer":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2026-02-27 20:38:21.000000000","tz":-480},"subject":"DNM: Test glance against LUKS images","message":"DNM: Test glance against LUKS images\n\nDepends-On: https://review.opendev.org/c/openstack/os-test-images/+/926942\nDepends-On: https://review.opendev.org/c/openstack/oslo.utils/+/978097\nDepends-On: https://review.opendev.org/c/openstack/glance/+/926295\nDepends-On: https://review.opendev.org/c/openstack/tempest/+/978215\nChange-Id: I7bdf1b7a91de29cf254ed7ef4e368fc3b16c18f5\nSigned-off-by: Dan Smith \u003cdansmith@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/d1bdd47447ec8a33f7ac4d2f232db8ed158dbe53"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/d1bdd47447ec8a33f7ac4d2f232db8ed158dbe53"}]},"branch":"refs/heads/master"},"e7f9bedfcc0b00d6e6507b563edc0c45017182dd":{"kind":"REWORK","_number":8,"created":"2026-02-27 20:49:01.000000000","uploader":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"ref":"refs/changes/88/978088/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/88/978088/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/88/978088/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/88/978088/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/88/978088/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/88/978088/8"}}},"commit":{"parents":[{"commit":"8070fae771718eda7f626844ca84a949f77b2dd5","subject":"Merge \"feat: return size in get_image_data_iter\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/8070fae771718eda7f626844ca84a949f77b2dd5"}]}],"author":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2026-02-26 15:41:31.000000000","tz":-480},"committer":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2026-02-27 01:02:19.000000000","tz":-480},"subject":"DNM: Test glance against LUKS images","message":"DNM: Test glance against LUKS images\n\nDepends-On: https://review.opendev.org/c/openstack/os-test-images/+/926942\nDepends-On: https://review.opendev.org/c/openstack/oslo.utils/+/978097\nDepends-On: https://review.opendev.org/c/openstack/glance/+/926295\nDepends-On: https://review.opendev.org/c/openstack/tempest/+/978215\nChange-Id: I7bdf1b7a91de29cf254ed7ef4e368fc3b16c18f5\nSigned-off-by: Dan Smith \u003cdansmith@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/e7f9bedfcc0b00d6e6507b563edc0c45017182dd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/e7f9bedfcc0b00d6e6507b563edc0c45017182dd"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY"},{"label":"Workflow","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{"label:Review-Priority\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
