)]}'
{"id":"openstack%2Fglance~980777","triplet_id":"openstack%2Fglance~master~Ib579158a21366488515471abe82e22d1405ffb6c","project":"openstack/glance","branch":"master","topic":"bug/2098514","attention_set":{},"removed_from_attention_set":{"9303":{"account":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"last_update":"2026-04-02 15:31:25.000000000","reason":"\u003cGERRIT_ACCOUNT_9303\u003e replied on the change","reason_account":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"}},"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2026-04-02 17:43:41.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ib579158a21366488515471abe82e22d1405ffb6c","subject":"Use consistent program name","status":"MERGED","created":"2026-03-16 15:41:33.000000000","updated":"2026-04-02 17:44:54.000000000","submitted":"2026-04-02 17:43:41.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"980777-bug/2098514","meta_rev_id":"91ae710da596f4f4e5ecdbceb53cba76851376bb","_number":980777,"virtual_id_number":980777,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-04-02 17:43:41.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"all":[{"value":2,"date":"2026-04-02 15:31:25.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":2,"date":"2026-03-17 15:12:20.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2026-03-30 21:20:38.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"all":[{"value":1,"date":"2026-04-02 15:31:25.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Top Priority / Holds Gate"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-03-16 18:13:11.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-03-17 15:12:20.000000000","updated_by":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"reviewer":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"state":"REVIEWER"},{"updated":"2026-03-30 21:20:38.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"state":"REVIEWER"},{"updated":"2026-03-30 21:20:38.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"state":"REVIEWER"}],"messages":[{"id":"80bed791990cdc3b3f44ee863f8ea262f5c427c7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2026-03-16 15:41:33.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"f11f3a64d44900a32d2ab912dfffbd27e37e131c","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2026-03-16 15:42:06.000000000","message":"Topic set to bug/2098514","accounts_in_message":[],"_revision_number":1},{"id":"117f73cbb94e29b71344b79f493dc3ded77fc365","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-16 18:13:11.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/52f835aa721041d1a613a68f4c361f81\n\n- grenade https://zuul.opendev.org/t/openstack/build/e32a7cb85344424a9970f563ca152b10 : SUCCESS in 1h 02m 32s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/de5c10f493854a089f18cde3dd8d619a : SUCCESS in 33m 17s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/29e7af0f653748abb619bd0c29bcd77f : SUCCESS in 2h 00m 07s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/a6a309d230504a66b5babe5aa3118573 : SUCCESS in 1h 05m 29s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0bb62c6050384227892c15e569305240 : SUCCESS in 4m 20s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/e496a60022764452b269f01e9bf1d68d : SUCCESS in 4m 01s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/300ce76c9f794d069cfc9601840716fa : SUCCESS in 5m 13s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/84a3139ba421491ea1f0a22f45b46c6c : SUCCESS in 4m 25s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9725ef610e114263bd309f15696b290c : SUCCESS in 5m 46s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/379aa8abdbd94f54a92f39061a93f550 : SUCCESS in 6m 39s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/f532d052fd994ccf909cd89a6b9e90e8 : SUCCESS in 5m 26s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/034241b763f04efa8832231ca87334f9 : SUCCESS in 56m 28s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/8fc63560199e49db9dc57aece7c4cbde : SUCCESS in 1h 55m 00s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/0a49bcc197c1402e87e8c7649b897742 : SUCCESS in 2h 08m 10s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/e92af52039bc4481a49f329851e14f0e : SUCCESS in 1h 55m 16s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/482f79afb7c9426aa38257a6c7d6b51f : SUCCESS in 32m 40s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/11b9fcb78b2248e4ab6ac836c65a68f2 : SUCCESS in 1h 39m 39s\n- glance-grenade-centralized-cache https://zuul.opendev.org/t/openstack/build/429ef78d2db0487b8c8a4534d36c274f : FAILURE in 23m 26s (non-voting)\n- glance-s3-multistore https://zuul.opendev.org/t/openstack/build/f578c74c85224eca872a80f32b5a5e79 : SUCCESS in 2h 25m 36s (non-voting)\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/81fa0f42ea79404daad09fb3d98c8f2b : SUCCESS in 35m 52s","accounts_in_message":[],"_revision_number":1},{"id":"517747d3855e7bf84a71f058372425fbe906e4bb","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-16 19:20:56.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c359b02f07ae45c992f1e41e576f141b\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/6519f7ea99f842829dffaa0558808b06 : SUCCESS in 8m 13s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/bc1df26682104f16b2ce4f97c961fad1 : SUCCESS in 7m 51s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/e48871f2f3864e80a80710df8154d05f : SUCCESS in 8m 58s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"7745f8d5b14c84d7cc8946b497bf8f0c25a5d2c7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2026-03-17 09:05:47.000000000","message":"Uploaded patch set 2: Commit message was updated.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":2},{"id":"208f0baa830b9d19dd5ab3549be9a6ef229d8313","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-17 09:31:28.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a67ecf4916db4bb392b6abf26d84f571\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/e5e32e31db90422d9451e85877949766 : SUCCESS in 8m 52s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/8626f508314a4a29a6c18bd7af4b1355 : SUCCESS in 8m 11s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/05d957ddf5ca4aeba6f84d1bdeadab21 : SUCCESS in 16m 25s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"15fb9eb30b99aa92ee10139d8388ece2aea2ddbb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-17 11:32:48.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/719abc12e9b9484fb207cd9a387304d5\n\n- grenade https://zuul.opendev.org/t/openstack/build/43b2194152124ad7809293c9f4271479 : SUCCESS in 42m 18s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/f96d66be89184707ad3f011f3e8ca738 : SUCCESS in 51m 58s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/cec6c1022b094e608f36056a256a464a : SUCCESS in 1h 33m 15s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/4b4ade7e815449a19afa5277e453c0aa : SUCCESS in 1h 03m 03s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/000712e5e290476597a4e017f5adefee : SUCCESS in 4m 32s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/d3bc02b4411842bebbf7ce1d467761e8 : SUCCESS in 4m 36s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/20fb4236b5de48638ad5b8dfc7ef6c9a : SUCCESS in 4m 34s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/fbe07b0e9a2249e283ed13ba7fb3efe6 : SUCCESS in 7m 17s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2d3bc0f27022416a92f8703b9e18106e : SUCCESS in 3m 16s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/41e49680086a48bd90d237113592ab33 : SUCCESS in 6m 29s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/4e8d7a1ab15847459c1ed5d7d1526e52 : SUCCESS in 6m 24s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/87bb7408c0c74c86902459fe799bd923 : SUCCESS in 1h 34m 41s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/bb23188c87e947af85f6f7ac0f44b348 : SUCCESS in 2h 01m 56s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/3e112816acd54db99f84f0545a99d97b : SUCCESS in 1h 58m 44s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/59670e5146cc42698f2c1b44a634230d : SUCCESS in 2h 21m 45s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/f641f718e83948f4ba42f8638956e6ba : SUCCESS in 54m 41s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/96c63288a26e4e7fbea76fbbf5a72ef1 : SUCCESS in 1h 56m 58s\n- glance-grenade-centralized-cache https://zuul.opendev.org/t/openstack/build/2e2aada792574148bd3b0c5922ed73c2 : FAILURE in 21m 51s (non-voting)\n- glance-s3-multistore https://zuul.opendev.org/t/openstack/build/81ce30bb0f3d41678d93e9851d5f2dc7 : SUCCESS in 1h 59m 35s (non-voting)\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/f212055428ae4134852db57ccc3546fc : SUCCESS in 35m 22s","accounts_in_message":[],"_revision_number":2},{"id":"8498f4848e36a308fc3e9d8debf685479831400e","author":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"date":"2026-03-17 15:12:20.000000000","message":"Patch Set 2: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"4f1bdc07dc8dadd98f18038d446990decc0fa1b2","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-03-30 21:20:38.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"ba81cb63d3c48e712d326a52912313a7313cef8e","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2026-04-02 15:31:25.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"55dc423eacda058375270cdd211a5e985b2f0936","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-02 15:31:52.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"ff7a6d1609af7f489ed2470380aa02951110e8fc","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-02 17:43:41.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7c5c57e8127e4d26967bc06ad2689647\n\n- grenade https://zuul.opendev.org/t/openstack/build/1588130287f14f3c9b736f019344a604 : SUCCESS in 1h 07m 18s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/e9aee7f53c154c71ba165571d6356860 : SUCCESS in 46m 27s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/fa0ba48675f74f6ebf2d56ead914b8a1 : SUCCESS in 1h 17m 44s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/f610d7bfb1a9487b941568d0f2460e01 : SUCCESS in 36m 14s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/716414e5daf3412fb00c596e2e4d8487 : SUCCESS in 4m 45s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/8aada7a5d17e45ff958f8a4d8308be33 : SUCCESS in 4m 44s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/750ffc67f77046928f011140fbd09217 : SUCCESS in 4m 55s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a2e829ca2fe448a9a00ec31eeee06370 : SUCCESS in 3m 58s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/8deb736e1a9d45d7882a0512a71efaf3 : SUCCESS in 4m 33s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/60ac09dd3ee443b6848b5d5785cd66bb : SUCCESS in 8m 05s\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/2d543915bd0249b9b6bc88840c020555 : SUCCESS in 2h 04m 50s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/18820e4a52ca45de85f528fe5d56101a : SUCCESS in 2h 06m 58s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/eb40ebe88bb142b8806f8188ccf6f32e : SUCCESS in 41m 39s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/63a5472a66f84695ba81526d203008a5 : SUCCESS in 1h 04m 45s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/b0fa372dead7429fae627244f2d74199 : SUCCESS in 18m 57s","accounts_in_message":[],"_revision_number":2},{"id":"c2c061e7af09b2576c6cc1ac5084348b50772a1e","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-02 17:43:41.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"91ae710da596f4f4e5ecdbceb53cba76851376bb","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-02 17:44:54.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/edb9d2fe4dc64f4abe02473372171284\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/729303f0574a4c5f8c8a458f580e1391 : SUCCESS in 1m 02s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"2e8204e16f4e54180640c4a534e86c7b107eace6","revisions":{"10d6ab08c8b6cbb9aff5a826e80401969b148f27":{"kind":"REWORK","_number":1,"created":"2026-03-16 15:41:33.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/77/980777/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/77/980777/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/77/980777/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/77/980777/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/77/980777/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/77/980777/1"}}},"commit":{"parents":[{"commit":"8070fae771718eda7f626844ca84a949f77b2dd5","subject":"Merge \"feat: return size in get_image_data_iter\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/8070fae771718eda7f626844ca84a949f77b2dd5"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-03-16 15:40:34.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-03-16 15:41:25.000000000","tz":540},"subject":"Use consistent program name for wsgi scripts and entry points","message":"Use consistent program name for wsgi scripts and entry points\n\nMake sure that the consistent program name is always set,so that\nthe same config sub-directory ( /etc/{project}/{prog}.conf.d ) is used\nafter migration to entry point.\n\nCloses-Bug: #2098514\nChange-Id: Ib579158a21366488515471abe82e22d1405ffb6c\nSigned-off-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/10d6ab08c8b6cbb9aff5a826e80401969b148f27"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/10d6ab08c8b6cbb9aff5a826e80401969b148f27"}]},"branch":"refs/heads/master"},"2e8204e16f4e54180640c4a534e86c7b107eace6":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2026-03-17 09:05:47.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/77/980777/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/77/980777/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/77/980777/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/77/980777/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/77/980777/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/77/980777/2"}}},"commit":{"parents":[{"commit":"8070fae771718eda7f626844ca84a949f77b2dd5","subject":"Merge \"feat: return size in get_image_data_iter\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/8070fae771718eda7f626844ca84a949f77b2dd5"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-03-16 15:40:34.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-03-17 09:05:23.000000000","tz":540},"subject":"Use consistent program name","message":"Use consistent program name\n\nMake sure that the consistent program name is always set,so that\nthe same config sub-directory ( /etc/{project}/{prog}.conf.d ) is used\nafter migration to WSGI server + entry point.\n\nCloses-Bug: #2098514\nChange-Id: Ib579158a21366488515471abe82e22d1405ffb6c\nSigned-off-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/2e8204e16f4e54180640c4a534e86c7b107eace6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/2e8204e16f4e54180640c4a534e86c7b107eace6"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
