)]}'
{"id":"openstack%2Fglance~981297","triplet_id":"openstack%2Fglance~stable%2F2025.2~Ib8d337dc68411d18c70d5712cc4f0986ef6205f4","project":"openstack/glance","branch":"stable/2025.2","topic":"bug-2138602-ssrf","attention_set":{},"removed_from_attention_set":{"9303":{"account":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"last_update":"2026-03-21 04:58:09.000000000","reason":"Change was submitted"},"8122":{"account":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"last_update":"2026-03-21 04:58:09.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ib8d337dc68411d18c70d5712cc4f0986ef6205f4","subject":"Fix SSRF vulnerabilities in image import API","status":"MERGED","created":"2026-03-19 14:37:47.000000000","updated":"2026-03-21 04:59:05.000000000","submitted":"2026-03-21 04:58:09.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"981297-bug-2138602-ssrf","meta_rev_id":"594c9fac1f0e7e8f6573bbd4ea373beb33167b57","_number":981297,"virtual_id_number":981297,"owner":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-03-21 04:58:09.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"all":[{"value":2,"date":"2026-03-20 14:26:36.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"all":[{"value":1,"date":"2026-03-20 14:26:36.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Top Priority / Holds Gate"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-03-19 14:40:54.000000000","updated_by":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"reviewer":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"state":"REVIEWER"},{"updated":"2026-03-19 19:26:08.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-03-20 14:26:36.000000000","updated_by":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"reviewer":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"state":"REVIEWER"}],"messages":[{"id":"4402044bc8dff467a4c872c0bb90f8aedc9fc1d6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2026-03-19 14:37:47.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"8447d56bc867386e6f088014e06edcc2f7be444e","author":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"date":"2026-03-19 14:40:54.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"4da69373ac5080f30450ad68b685e38a0736ea01","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2026-03-19 15:28:21.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Workflow+1\n","accounts_in_message":[],"_revision_number":2},{"id":"8b2f9e40adbd9dcf9b00ba15280de43fce511835","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2026-03-19 17:16:01.000000000","message":"Patch Set 3: Commit message was updated.","accounts_in_message":[],"_revision_number":3},{"id":"1659b5a4e787e1e8f5bc486e8b300b2ec650e272","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-19 19:26:08.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/decc668418734ba9a975b5cd3b9c6b85\n\n- grenade https://zuul.opendev.org/t/openstack/build/734093729a4746cd9a54cca75089dabe : SUCCESS in 1h 11m 37s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/3a675b83d3a241b8a830feddabab0bf9 : SUCCESS in 31m 50s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/7b8d920506c34c0a8a750aff1fe3af80 : SUCCESS in 1h 34m 11s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/a7159803b2d84ed08011bf374d52cac9 : SUCCESS in 44m 05s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b93b1f03bfa3494b85316fd894613f0d : SUCCESS in 6m 35s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/61ae43608b7648708b88357273695b71 : SUCCESS in 5m 00s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/14a0b98422ea4f409e6120578a6b27ad : SUCCESS in 3m 38s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/7e333dd8ad48485bbc558a754b08319c : SUCCESS in 9m 51s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dc585d446ae94d6bbc14574244a62439 : SUCCESS in 5m 54s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ae5e64e4155748f2ad3b96d48d495d7a : SUCCESS in 3m 52s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/898ac61c0e73455f9e0a2e83b76f6c25 : SUCCESS in 7m 01s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/7d0a4ec216524743a78fa85286f9a699 : SUCCESS in 7m 25s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/da1a1e04c5194603affcc91a7cf00b3f : FAILURE in 1h 51m 23s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/267e8ee32d644944a1bfb9b3a55cea86 : SUCCESS in 2h 03m 22s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/aeb8e0d840494e4ea17de3c937e5e9a2 : SUCCESS in 2h 02m 16s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/7e3b4970e93e425896583ad5ef23079d : SUCCESS in 1h 12m 20s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/3508878103744dfebfc1514737603fae : SUCCESS in 56m 35s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/f61964643aa7406785033df70c0029ce : SUCCESS in 1h 46m 42s\n- glance-grenade-centralized-cache https://zuul.opendev.org/t/openstack/build/598d216dcef445eba570cf74a7188f59 : FAILURE in 32m 27s (non-voting)\n- glance-s3-multistore https://zuul.opendev.org/t/openstack/build/8f9987647a934e848f283c1347c8fbb0 : FAILURE in 29m 56s (non-voting)\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/b9e9137a199446f591eb3775f472573e : SUCCESS in 42m 42s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/245338379430477596e08f2a132741e1 : SUCCESS in 1h 09m 14s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"c7a81a1db607c581b81c7b22b46e74fe7c6ce5aa","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2026-03-20 13:07:14.000000000","message":"Topic set to bug-2138602-ssrf","accounts_in_message":[],"_revision_number":3},{"id":"b8ee048f64f52e96ced79a1103fc17ee356379b2","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2026-03-20 14:26:36.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"472d81b23c14d62864a4af089bbf36c562aac8d3","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-20 14:27:14.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"dbd45a2c65592a2519e3993728bcf292b32a3b28","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-20 18:58:20.000000000","message":"Patch Set 3: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/36c51e45c8834bc8bc17a70e7f68939e\n\n- grenade https://zuul.opendev.org/t/openstack/build/27019ff697f94068be9c7d33ae522c66 : SUCCESS in 59m 34s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/cf15a8f16b2f4f2db29a24e83cd54b42 : SUCCESS in 1h 11m 43s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/ace91f6dd08b4d28a9f89904657fc89c : SUCCESS in 1h 54m 17s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/328d9589b5714626a22625466a7e0762 : SUCCESS in 35m 48s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/42d7b84b6965469db2cb6125245d3db0 : SUCCESS in 2m 52s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/3b319bd1a0e14dc5af78768386c6c2d2 : SUCCESS in 5m 00s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/cb40d31b23564b3c9645cec7ab1cde5e : SUCCESS in 7m 04s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e677170516a2447b9cb6c7ef30d7ddd9 : SUCCESS in 4m 07s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/22c0fa42361e4ba2bc52037a4f6effaa : SUCCESS in 7m 39s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/af1c779310f843a8931472f27247554c : SUCCESS in 6m 14s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/9dfbf10181294e49a4f399fa4328e1f5 : SUCCESS in 5m 33s\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/8c9eeb8517954144b2e61b86c5db45a3 : SUCCESS in 1h 50m 39s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/1cfc2fb7dad44db7867803b5672c84f8 : FAILURE in 1h 51m 48s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/35ebeaf14a2b485f9a1627acbd6dff22 : SUCCESS in 32m 16s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/fd410889d91a4f29b8e7af5a6e63017b : SUCCESS in 2h 01m 30s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/0faf7414bd994bb2b3e42436c129d86b : SUCCESS in 35m 23s","accounts_in_message":[],"_revision_number":3},{"id":"f358e4806b60fb80ede8c75875d3c2ae03fdf6fa","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2026-03-20 19:31:21.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"436e256ca9f1d50e15c267f1278331ec8fc09a2f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-20 22:02:26.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4d4d941088854f63b7896876fe25b1c3\n\n- grenade https://zuul.opendev.org/t/openstack/build/e114cdefe83b4870ab4cf5da8c65e45d : SUCCESS in 1h 02m 07s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/a2fdc70e09db45bebe2963ea8d181565 : SUCCESS in 32m 09s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/38d04818774741278f1f0bbab2a57f44 : SUCCESS in 1h 33m 29s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/6cbc1cd3076d45d69ef3333f09b299ee : SUCCESS in 46m 20s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2769b41be5534f76b1f40eae8fb3dc22 : SUCCESS in 5m 56s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/b8623c2cc2d4424f9109aeb3e9c2febf : SUCCESS in 8m 37s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/d154a4fb291a4820883176a6bda7ad74 : SUCCESS in 5m 35s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/60506c042a6a47ae81d0106fc2213f8f : SUCCESS in 4m 19s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/59ce929bda6e45a299d1b399d3f828f8 : SUCCESS in 3m 06s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/764ddf7b5f71407b958267e1432ff6bf : SUCCESS in 6m 58s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/2fc2bc7d73604200b9c1259fb0cac708 : SUCCESS in 7m 49s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/863227f3b83b4cd4b8a1d2540ff3ab22 : SUCCESS in 7m 20s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/b15d2b5c698948cc884359bc9bbda38f : SUCCESS in 1h 34m 24s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/72705ae0406344cf9183735ace9fc7db : SUCCESS in 1h 06m 27s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/126a2af3bb18422ea66002861ba21308 : SUCCESS in 1h 44m 02s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/62ece417cc93465281a078c35d86bacd : SUCCESS in 2h 24m 43s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/10d665a876434801a437040288e8edd2 : SUCCESS in 59m 46s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/a2f7a8424a8b49f5b356ee375e8df8b4 : SUCCESS in 56m 39s\n- glance-grenade-centralized-cache https://zuul.opendev.org/t/openstack/build/2746be99326442da9322687ba3c08d44 : FAILURE in 29m 00s (non-voting)\n- glance-s3-multistore https://zuul.opendev.org/t/openstack/build/b83d0327ba8e4aa7924de58a576bea14 : FAILURE in 11m 23s (non-voting)\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/8579e9cc589c4d7ea39f77165abaaa6c : SUCCESS in 34m 45s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/ca97393298f5400f9ab92dcb90a874ab : SUCCESS in 1h 12m 38s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"b5cf13d232b294cee6df17c1fd424d62bfd9e19e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-20 22:02:54.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"33eb980550915694bc0a905b9c723c51f3491fcf","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-21 00:04:41.000000000","message":"Patch Set 3: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/1a9247ceaabd41f4889f8732e416992f\n\n- grenade https://zuul.opendev.org/t/openstack/build/622666fa9a0a45ed9cb3f18b44753cc2 : SUCCESS in 1h 01m 41s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/e488abfb41f347cfb4fc2848a4d7838b : FAILURE in 27m 15s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/110a945c39084cd2921504cb6a1d8268 : SUCCESS in 1h 31m 18s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/4f1d758e54d544eb93db8719b48ea438 : SUCCESS in 58m 52s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/569ed0a55e854196acbfbccb9c526c07 : SUCCESS in 5m 02s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/7024e2b21dd14a808afd6c528eca06ef : SUCCESS in 4m 49s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/f2fdd41d3cd24fe2be1961b91a650642 : SUCCESS in 5m 09s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dceef80ca7904166ba46c2093f2cff38 : SUCCESS in 5m 16s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/410e2c28137b49fa9561dcfe8be94f4c : SUCCESS in 6m 43s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/65a674c5482a460e80083b02d85c0c70 : SUCCESS in 6m 36s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/ee4baf3cec89472a9d50ebee5766f49b : SUCCESS in 6m 58s\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/fca527c7686d47019b733f41ef007515 : SUCCESS in 1h 57m 03s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/69a35161b87f45919bcf295d740dfc22 : SUCCESS in 1h 58m 48s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/055a53118fb04f969e76df441a478a9b : SUCCESS in 57m 29s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/ab1bea1789ac41d7b260e0a76db88907 : SUCCESS in 1h 35m 06s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/ff8d3402aa624a4a949bc6b30181768b : SUCCESS in 33m 40s","accounts_in_message":[],"_revision_number":3},{"id":"ebb81bd07004d8a64dcbf50d72a7dd23ce6980ab","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2026-03-21 00:17:55.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"4fd948f1ac63a2ad869bf75284671af82cff2dab","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-21 02:53:23.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/47e89296ddb34df0902e3a1714aa4af8\n\n- grenade https://zuul.opendev.org/t/openstack/build/6eb6e59025fd4a22a4d747d85db5a6fe : SUCCESS in 32m 12s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/7db53395246144c88badf8c15f816b1d : SUCCESS in 34m 31s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/8c3cd9a4cda946b48a2c1f017cfde06e : SUCCESS in 1h 21m 39s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/09f2cedcebfd41cf96e0700e857676c3 : SUCCESS in 35m 12s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/279ac429afd14ba9a8c9dbc9a86efa78 : SUCCESS in 3m 05s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/9f2b86e255424b619ddfe3c2daf6e2a4 : SUCCESS in 3m 17s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/0c81c78e789c474ebb9054663e7ec3de : SUCCESS in 3m 40s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/5cc127296f9c4d149b990a5fda4578e9 : SUCCESS in 9m 19s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6ed404d1b75f493c8e9fc7b36dff87cb : SUCCESS in 5m 25s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e96c6fc5a871451492c02247a67dc6cd : SUCCESS in 4m 53s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/4f1a4680624c43408d8780e7fce61316 : SUCCESS in 6m 10s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/f66481c7cafe45d98f407dbad7a0e932 : SUCCESS in 5m 06s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/3c74231acb4d49e3899732f8f30dc36c : SUCCESS in 1h 46m 49s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/cc3b8b39a2af47bfb5c59223565f8fc8 : SUCCESS in 2h 13m 06s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/821139d278264332997780d7453a29c9 : SUCCESS in 1h 56m 16s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/347e82c6d4124e378edd9aff7447ee02 : SUCCESS in 2h 29m 55s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/cb9eb8f44b324efba14f110c5e088afe : SUCCESS in 59m 43s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/1a54d3c2e8444f20bb3428531d5b950f : SUCCESS in 1h 56m 44s\n- glance-grenade-centralized-cache https://zuul.opendev.org/t/openstack/build/e5b187ea875b4b60b55ee70dde4e21b2 : FAILURE in 14m 00s (non-voting)\n- glance-s3-multistore https://zuul.opendev.org/t/openstack/build/f16b2368bdad414e81b1546db7264bae : FAILURE in 11m 06s (non-voting)\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/b6ce84e83f7347b4bc2c26af4cd0b56c : SUCCESS in 25m 28s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/6d35e2410d00486eb36097e46384ad1c : SUCCESS in 51m 00s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"62a4eb5cfa8fa703d9f62f96e763f23da77edb28","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-21 02:53:52.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"2da610ae1acbeab7aaf0652cd4b066d9e9a9b459","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-21 04:58:09.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/388e4638f9f84a57b0e50fb9d3f93d10\n\n- grenade https://zuul.opendev.org/t/openstack/build/4b5a610df3c843c789740baccf984ff2 : SUCCESS in 1h 02m 47s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/0d5aa0bdcd424d8d8dd6c76901d31358 : SUCCESS in 1h 02m 06s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/4384f65a00ba4e91a54208cf8754ad04 : SUCCESS in 1h 21m 50s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/7507373cba0340aa853f3adb87d415a2 : SUCCESS in 1h 05m 21s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2f83a00b4d3a49709bf08c49c2aadf04 : SUCCESS in 6m 54s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/7a8c74712bf94f079263e008a201294e : SUCCESS in 6m 40s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/822d6f7488ad4f83a45dcebe41137570 : SUCCESS in 7m 27s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6e4f46cd9f304ba1832c3c705189f0d7 : SUCCESS in 7m 31s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/551ef4e9e3934e0fab514f49518a7aa5 : SUCCESS in 7m 03s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/c03db91c8bcd49c490797865cfad62da : SUCCESS in 8m 47s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/caa0aecaa8594ed8ba245ca3525422b9 : SUCCESS in 9m 41s\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/8cef9f9214fb44a59c99f5b1cc24a215 : SUCCESS in 1h 07m 38s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/f3d611d0107e49c896f6888da1185dd9 : SUCCESS in 1h 59m 15s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/79acdb1c1e0c4541a1ff83985d527ef1 : SUCCESS in 34m 32s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/a49b7945af1a48c885c13fcd950cbc52 : SUCCESS in 1h 01m 21s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/73dadb7d9cdb43c0acc0596e4d30dd16 : SUCCESS in 32m 28s","accounts_in_message":[],"_revision_number":3},{"id":"4c0bbad6ab2c9ac01d15fd8cdca4ba144245be33","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-21 04:58:09.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3},{"id":"594c9fac1f0e7e8f6573bbd4ea373beb33167b57","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-21 04:59:05.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/31813b55b01e40f8ab60ddd3ded91480\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f50e6ea90ad242169df2e48cbdbe6ad3 : SUCCESS in 47s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b9dd5115042a4e82b1b55d86e97b6175 : SUCCESS in 42s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"fdfb958b6e948e8d7fd183999c026e7c71a050f9","revisions":{"8c6305babb3bf577c74ccad2a6e97a46a1da7af3":{"kind":"REWORK","_number":1,"created":"2026-03-19 14:37:47.000000000","uploader":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"ref":"refs/changes/97/981297/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/97/981297/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/97/981297/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/97/981297/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/97/981297/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/97/981297/1"}}},"commit":{"parents":[{"commit":"435a5dd2a6b2a3ff12572836feaf08ce89f03e32","subject":"Add glance/wsgi/__init__.py","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/435a5dd2a6b2a3ff12572836feaf08ce89f03e32"}]}],"author":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2026-01-20 19:02:08.000000000","tz":0},"committer":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2026-03-19 14:36:27.000000000","tz":0},"subject":"Fix SSRF vulnerabilities in image import API","message":"Fix SSRF vulnerabilities in image import API\n\nFixed Server-Side Request Forgery (SSRF) vulnerabilities in Glance\u0027s image\nimport functionality that could allow attackers to bypass URL validation\nand access internal resources.\n\nThe fix includes:\n- IP address validation using Python\u0027s ipaddress module to reject encoded\n  IP formats (decimal, hexadecimal, octal) that could bypass blacklist checks\n- HTTP redirect validation for web-download, glance-download, and OVF\n  processing to prevent redirect-based SSRF attacks\n- URI validation for OVF processing which previously had no protection\n\nThe implementation uses Python\u0027s built-in ipaddress module which inherently\nrejects all non-standard IP encodings and only accepts standard formats,\nproviding robust protection against IP encoding bypass attacks.\n\nAssisted-by: Used Cursor (Auto) for unit tests.\n\nCloses-Bug: #2138602\nCloses-Bug: #2138672\nCloses-Bug: #2138675\nSecurityImpact\n\nChange-Id: Ib8d337dc68411d18c70d5712cc4f0986ef6205f4\nSigned-off-by: Abhishek Kekane \u003cakekane@redhat.com\u003e\n(cherry picked from commit 38040d086bf6c159f45d201f361a71e7a32e62af)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/8c6305babb3bf577c74ccad2a6e97a46a1da7af3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/8c6305babb3bf577c74ccad2a6e97a46a1da7af3"}]},"branch":"refs/heads/stable/2025.2"},"ad26b95ba007d6d9f6618666e48f0688be1de691":{"kind":"REWORK","_number":2,"created":"2026-03-19 15:28:21.000000000","uploader":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"ref":"refs/changes/97/981297/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/97/981297/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/97/981297/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/97/981297/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/97/981297/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/97/981297/2"}}},"commit":{"parents":[{"commit":"435a5dd2a6b2a3ff12572836feaf08ce89f03e32","subject":"Add glance/wsgi/__init__.py","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/435a5dd2a6b2a3ff12572836feaf08ce89f03e32"}]}],"author":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2026-01-20 19:02:08.000000000","tz":0},"committer":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2026-03-19 15:26:55.000000000","tz":0},"subject":"Fix SSRF vulnerabilities in image import API","message":"Fix SSRF vulnerabilities in image import API\n\nFixed Server-Side Request Forgery (SSRF) vulnerabilities in Glance\u0027s image\nimport functionality that could allow attackers to bypass URL validation\nand access internal resources.\n\nThe fix includes:\n- IP address validation using Python\u0027s ipaddress module to reject encoded\n  IP formats (decimal, hexadecimal, octal) that could bypass blacklist checks\n- HTTP redirect validation for web-download, glance-download, and OVF\n  processing to prevent redirect-based SSRF attacks\n- URI validation for OVF processing which previously had no protection\n\nThe implementation uses Python\u0027s built-in ipaddress module which inherently\nrejects all non-standard IP encodings and only accepts standard formats,\nproviding robust protection against IP encoding bypass attacks.\n\nAssisted-by: Used Cursor (Auto) for unit tests.\n\nCloses-Bug: #2138602\nCloses-Bug: #2138672\nCloses-Bug: #2138675\nSecurityImpact\n\nChange-Id: Ib8d337dc68411d18c70d5712cc4f0986ef6205f4\nSigned-off-by: Abhishek Kekane \u003cakekane@redhat.com\u003e\n(cherry picked from commit 38040d086bf6c159f45d201f361a71e7a32e62af)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/ad26b95ba007d6d9f6618666e48f0688be1de691"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/ad26b95ba007d6d9f6618666e48f0688be1de691"}]},"branch":"refs/heads/stable/2025.2"},"fdfb958b6e948e8d7fd183999c026e7c71a050f9":{"kind":"NO_CODE_CHANGE","_number":3,"created":"2026-03-19 17:16:01.000000000","uploader":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"ref":"refs/changes/97/981297/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/97/981297/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/97/981297/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/97/981297/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/97/981297/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/97/981297/3"}}},"commit":{"parents":[{"commit":"435a5dd2a6b2a3ff12572836feaf08ce89f03e32","subject":"Add glance/wsgi/__init__.py","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/435a5dd2a6b2a3ff12572836feaf08ce89f03e32"}]}],"author":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2026-01-20 19:02:08.000000000","tz":0},"committer":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2026-03-19 17:16:01.000000000","tz":0},"subject":"Fix SSRF vulnerabilities in image import API","message":"Fix SSRF vulnerabilities in image import API\n\nFixed Server-Side Request Forgery (SSRF) vulnerabilities in Glance\u0027s image\nimport functionality that could allow attackers to bypass URL validation\nand access internal resources.\n\nThe fix includes:\n- IP address validation using Python\u0027s ipaddress module to reject encoded\n  IP formats (decimal, hexadecimal, octal) that could bypass blacklist checks\n- HTTP redirect validation for web-download, glance-download, and OVF\n  processing to prevent redirect-based SSRF attacks\n- URI validation for OVF processing which previously had no protection\n\nThe implementation uses Python\u0027s built-in ipaddress module which inherently\nrejects all non-standard IP encodings and only accepts standard formats,\nproviding robust protection against IP encoding bypass attacks.\n\nDepends-On: https://review.opendev.org/c/openstack/tempest/+/981329\n\nAssisted-by: Used Cursor (Auto) for unit tests.\n\nCloses-Bug: #2138602\nCloses-Bug: #2138672\nCloses-Bug: #2138675\nSecurityImpact\n\nChange-Id: Ib8d337dc68411d18c70d5712cc4f0986ef6205f4\nSigned-off-by: Abhishek Kekane \u003cakekane@redhat.com\u003e\n(cherry picked from commit 38040d086bf6c159f45d201f361a71e7a32e62af)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/fdfb958b6e948e8d7fd183999c026e7c71a050f9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/fdfb958b6e948e8d7fd183999c026e7c71a050f9"}]},"branch":"refs/heads/stable/2025.2","description":"Edit commit message"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
