)]}'
{"id":"openstack%2Fglance~981510","triplet_id":"openstack%2Fglance~stable%2F2025.1~If79a129288019c2f5e369b797e7806910c198d79","project":"openstack/glance","branch":"stable/2025.1","topic":"ssrf","attention_set":{},"removed_from_attention_set":{"5314":{"account":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"last_update":"2026-03-20 20:57:48.000000000","reason":"Change was submitted"},"9303":{"account":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"last_update":"2026-03-20 20:57:48.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"If79a129288019c2f5e369b797e7806910c198d79","subject":"[stable-only] fix stable/2025.1 gate jobs","status":"MERGED","created":"2026-03-20 12:53:38.000000000","updated":"2026-03-20 21:01:40.000000000","submitted":"2026-03-20 20:57:48.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"981510-ssrf","meta_rev_id":"862eb754498e0359bdaa55f6980f197136159613","_number":981510,"virtual_id_number":981510,"owner":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-03-20 20:57:47.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"all":[{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":2,"date":"2026-03-20 16:36:03.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"all":[{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":1,"date":"2026-03-20 16:36:03.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Top Priority / Holds Gate"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-03-20 12:54:02.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2026-03-20 12:54:24.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-03-20 13:12:37.000000000","updated_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"reviewer":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"state":"REVIEWER"},{"updated":"2026-03-20 16:36:03.000000000","updated_by":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"reviewer":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"state":"REVIEWER"}],"messages":[{"id":"519abc8e5183525d62891f2021e5ff8ee44a36bf","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2026-03-20 12:53:38.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"8d2285179e3c2e2ae279632afc2eccb5d578585d","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-20 12:54:02.000000000","message":"Patch Set 1:\n\nZuul encountered a syntax error while parsing its\nconfiguration in the repo openstack/glance on branch stable/2025.1.  The\nproblem was:\n\n  found undefined alias\n    in \"openstack/glance/.zuul.yaml@stable/2025.1\", line 470, column 31","accounts_in_message":[],"_revision_number":1},{"id":"19239db4ac6a39bf8f52387b436e2b4d051720bd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-20 12:54:24.000000000","message":"Patch Set 1: Verified-1\n\nZuul encountered a syntax error while parsing its\nconfiguration in the repo openstack/glance on branch stable/2025.1.  The\nproblem was:\n\n  found undefined alias\n    in \"openstack/glance/.zuul.yaml@stable/2025.1\", line 470, column 31","accounts_in_message":[],"_revision_number":1},{"id":"c6dfbd4f0db70e5a3373d05e4ab8caae35a72a29","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2026-03-20 12:57:50.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"aec8a4a6e41fc69ba5a24635b8077f0dae7a0047","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2026-03-20 13:12:37.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"22b873e9f64a94604ec50c9e2b4bd889565184a4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2026-03-20 13:34:09.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n","accounts_in_message":[],"_revision_number":3},{"id":"e2855945f1eec8befc155d5ebdd33c4ed610bac3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-20 15:49:56.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/42cb1731f34f436787a6e9aab8713269\n\n- grenade https://zuul.opendev.org/t/openstack/build/af7bc7ab02c942eeb94aa9931f0955d5 : SUCCESS in 1h 01m 54s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/122daed98ac24170aab07082348829ad : FAILURE in 43m 41s (non-voting)\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/f1ef53f545e445f18c6ea79a4c315ca6 : SUCCESS in 1h 52m 11s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/43e1b123417b47089a3722eff53c6862 : SUCCESS in 1h 03m 28s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/06d4d3bd75e1450583e317a4097828de : SUCCESS in 4m 28s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/0a4878a603e3424c9eecbcb0e578cfdd : SUCCESS in 5m 04s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/d4a26376da5e4ee39b6c8659bbcc51e8 : SUCCESS in 5m 50s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1eaeed16b9b44ae1b82e84abf98b448c : SUCCESS in 8m 05s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/931c0b9d03b64e25a206c73b162cedf9 : SUCCESS in 6m 21s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/e0dfdcf5ea554eb297f5913ba5d36cdc : SUCCESS in 8m 57s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/716d87607ca54f759df0280a7679fc96 : SUCCESS in 7m 47s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/7aefccb33b7040adb74faca1e714ea20 : SUCCESS in 8m 36s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/f000ceb30281435c94f05e85d6a43614 : SUCCESS in 1h 49m 13s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/cb9f190bb3964454a410c4bef2fed672 : SUCCESS in 1h 48m 23s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/04673fe3574944e9836299c5e4a35319 : SUCCESS in 2h 05m 19s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/5561a0d87427434e831dc9258b3ccde4 : SUCCESS in 1h 37m 46s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/a0a30e4d1a904264aba8ff908868caf3 : SUCCESS in 2h 05m 48s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/a70f588000ad40e4a831777b4a51dec4 : SUCCESS in 41m 54s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/71a016f235a04f76996ba615ddc7e432 : SUCCESS in 2h 01m 44s\n- glance-centralized-cache https://zuul.opendev.org/t/openstack/build/6441aa748f604fd39e2f23f3ed468a95 : SUCCESS in 1h 44m 57s (non-voting)\n- glance-grenade-centralized-cache https://zuul.opendev.org/t/openstack/build/39e66ee280624334a6d531f5f8bd1abe : FAILURE in 28m 37s (non-voting)\n- glance-s3-multistore https://zuul.opendev.org/t/openstack/build/360a6b23df7341a4841f3f9eb3013ced : FAILURE in 25m 27s (non-voting)\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/31b275d5e9304eafa34d8e6bdd389e96 : SUCCESS in 35m 08s","accounts_in_message":[],"_revision_number":3},{"id":"a97fe2c8a0304dc26c2c5f8923c980bfab333c63","author":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"date":"2026-03-20 16:36:03.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"9b027322c8bc2df5ff94132be25e392baed69842","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-20 16:37:06.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"2d9df07c839540fee7df3e4bbf7809462e377af5","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-20 20:57:47.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c0ed37ec46f04c478d427e2d728f9ecd\n\n- grenade https://zuul.opendev.org/t/openstack/build/04e87580405440109b374923dbd3866e : SUCCESS in 29m 39s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/73368bd0d4be43e0b29f6c7640565e9b : FAILURE in 37m 30s (non-voting)\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/78c8dd156b334f8ebf9154f99be4f6eb : SUCCESS in 1h 21m 07s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/d126259ff576481bba07ee768cbe5d1d : SUCCESS in 35m 12s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e48c8bdbe19c4cda947aff0165712a7f : SUCCESS in 8m 02s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/adefa896b2bc4a78850c67cadf58b46f : SUCCESS in 9m 36s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/53da0b97b564493ea340e868c351a01b : SUCCESS in 4m 31s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d76f05088ac54d8ea02cc1e446bb84fc : SUCCESS in 5m 16s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/837e7231a6ea44c0b17b6e39687a8999 : SUCCESS in 12m 26s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/d466b8c875384716b7a624d53ff189f7 : SUCCESS in 8m 41s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/437b81e470914d068d212e55ee579f02 : SUCCESS in 8m 54s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/fc47648ec5654cf399ae3c08489d4478 : SUCCESS in 8m 22s\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/717fda4d8d214e859a79e81992705f93 : SUCCESS in 1h 08m 21s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/d796e5de98dc408a82e93a44f9312977 : SUCCESS in 1h 04m 39s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/83781c566ead4de2a7605be0b7b7b83a : SUCCESS in 1h 21m 47s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/404ffd6b7eb44974b99250891b00148e : SUCCESS in 35m 27s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/0d2844170f3844ee9ba32858b435441a : SUCCESS in 1h 51m 24s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/6d0ff166f89341f9bf2328682a387cac : SUCCESS in 39m 10s","accounts_in_message":[],"_revision_number":3},{"id":"d2ba702896c5be7f1d40e396651158a4be1abcb5","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-20 20:57:48.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3},{"id":"862eb754498e0359bdaa55f6980f197136159613","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-20 21:01:40.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/dc3b1f07ed5841d6998c4f10d438df55\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c712e9605b8e4ff8a5db83256b07d719 : SUCCESS in 57s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ca51fffe7206480e8281287aafdf7fde : SUCCESS in 1m 07s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"d88f696b76156e7ce192f8ff798c783be9f1b4fe","revisions":{"3d82b0088a96dec10cd95386bc07f7d72247317e":{"kind":"REWORK","_number":1,"created":"2026-03-20 12:53:38.000000000","uploader":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"ref":"refs/changes/10/981510/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/10/981510/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/10/981510/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/10/981510/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/10/981510/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/10/981510/1"}}},"commit":{"parents":[{"commit":"9b0af563193ef03a1b0f5e1e80030d96d723590b","subject":"Add glance/wsgi/__init__.py","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/9b0af563193ef03a1b0f5e1e80030d96d723590b"}]}],"author":{"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","date":"2026-03-20 12:51:32.000000000","tz":-240},"committer":{"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","date":"2026-03-20 12:51:32.000000000","tz":-240},"subject":"[stable-only] non-voting grenade-skip-level-always","message":"[stable-only] non-voting grenade-skip-level-always\n\nThe grenade-skip-level-always job is broken in the epoxy branch,\nfailing during the \"upgrade nova\" phase. The nova project addressed\nthis with Change-Id: I12e03b8aa1330d6396017e2dcbface798b6be7b1\nby making that job non-voting in stable/2025.1, so we do that also as\na temporary change to unblock the gate.\n\nChange-Id: If79a129288019c2f5e369b797e7806910c198d79\nSigned-off-by: Brian Rosmaita \u003crosmaita.fossdev@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/3d82b0088a96dec10cd95386bc07f7d72247317e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/3d82b0088a96dec10cd95386bc07f7d72247317e"}]},"branch":"refs/heads/stable/2025.1"},"a6399abe722bd0fcb919f56e24361a22c63670f7":{"kind":"REWORK","_number":2,"created":"2026-03-20 12:57:50.000000000","uploader":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"ref":"refs/changes/10/981510/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/10/981510/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/10/981510/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/10/981510/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/10/981510/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/10/981510/2"}}},"commit":{"parents":[{"commit":"9b0af563193ef03a1b0f5e1e80030d96d723590b","subject":"Add glance/wsgi/__init__.py","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/9b0af563193ef03a1b0f5e1e80030d96d723590b"}]}],"author":{"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","date":"2026-03-20 12:51:32.000000000","tz":-240},"committer":{"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","date":"2026-03-20 12:57:42.000000000","tz":-240},"subject":"[stable-only] non-voting grenade-skip-level-always","message":"[stable-only] non-voting grenade-skip-level-always\n\nThe grenade-skip-level-always job is broken in the epoxy branch,\nfailing during the \"upgrade nova\" phase. The nova project addressed\nthis with Change-Id: I12e03b8aa1330d6396017e2dcbface798b6be7b1\nby making that job non-voting in stable/2025.1, so we do that also as\na temporary change to unblock the gate.\n\nChange-Id: If79a129288019c2f5e369b797e7806910c198d79\nSigned-off-by: Brian Rosmaita \u003crosmaita.fossdev@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/a6399abe722bd0fcb919f56e24361a22c63670f7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/a6399abe722bd0fcb919f56e24361a22c63670f7"}]},"branch":"refs/heads/stable/2025.1"},"d88f696b76156e7ce192f8ff798c783be9f1b4fe":{"kind":"REWORK","_number":3,"created":"2026-03-20 13:34:09.000000000","uploader":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"ref":"refs/changes/10/981510/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/10/981510/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/10/981510/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/10/981510/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/10/981510/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/10/981510/3"}}},"commit":{"parents":[{"commit":"9b0af563193ef03a1b0f5e1e80030d96d723590b","subject":"Add glance/wsgi/__init__.py","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/9b0af563193ef03a1b0f5e1e80030d96d723590b"}]}],"author":{"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","date":"2026-03-20 12:51:32.000000000","tz":-240},"committer":{"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","date":"2026-03-20 13:29:14.000000000","tz":-240},"subject":"[stable-only] fix stable/2025.1 gate jobs","message":"[stable-only] fix stable/2025.1 gate jobs\n\nTwo things:\n1. need to constrain setuptools because the versions of dependencies\n   constrained by upper-constraints still use pkg_resources\n2. The grenade-skip-level-always job is broken in the epoxy branch,\n   failing during the \"upgrade nova\" phase. The nova project addressed\n   this with Change-Id: I12e03b8aa1330d6396017e2dcbface798b6be7b1\n   by making that job non-voting in stable/2025.1, so we do that also\n   as a temporary change to unblock the gate.\n\n(Ideally these would be separate commits, but neither change can pass\nwithout the other, so we do them both at the same time.)\n\nCo-authored-by: Abhishek Kekane \u003cakekane@redhat.com\u003e\nCo-authored-by: Brian Rosmaita \u003crosmaita.fossdev@gmail.com\u003e\n\nChange-Id: If79a129288019c2f5e369b797e7806910c198d79\nSigned-off-by: Brian Rosmaita \u003crosmaita.fossdev@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/d88f696b76156e7ce192f8ff798c783be9f1b4fe"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/d88f696b76156e7ce192f8ff798c783be9f1b4fe"}]},"branch":"refs/heads/stable/2025.1"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
