)]}'
{"id":"openstack%2Fglance~981520","triplet_id":"openstack%2Fglance~master~I99103c7846c05df38f8477357f7f7693b28fe8c7","project":"openstack/glance","branch":"master","attention_set":{},"removed_from_attention_set":{"9303":{"account":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"last_update":"2026-03-26 09:55:24.000000000","reason":"\u003cGERRIT_ACCOUNT_9303\u003e replied on the change","reason_account":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"}},"6476":{"account":{"_account_id":6476,"name":"Thomas Goirand","email":"thomas@goirand.fr","username":"thomas-goirand"},"last_update":"2026-03-26 15:55:00.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I99103c7846c05df38f8477357f7f7693b28fe8c7","subject":"No DNS resolution in test","status":"MERGED","created":"2026-03-20 14:20:39.000000000","updated":"2026-03-26 18:07:47.000000000","submitted":"2026-03-26 15:55:00.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":6,"unresolved_comment_count":2,"has_review_started":true,"submission_id":"981520","meta_rev_id":"bae368b8a9ac185cda60249266bef3a58b2841c1","_number":981520,"virtual_id_number":981520,"owner":{"_account_id":6476,"name":"Thomas Goirand","email":"thomas@goirand.fr","username":"thomas-goirand"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-03-26 15:55:00.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"all":[{"value":2,"date":"2026-03-26 09:55:24.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"all":[{"value":1,"date":"2026-03-26 09:55:24.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Top Priority / Holds Gate"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-03-20 15:02:05.000000000","updated_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"reviewer":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"state":"REVIEWER"},{"updated":"2026-03-20 15:44:09.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"40891ec048374f3863c47bfb4a73be3c2adea18a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":6476,"name":"Thomas Goirand","email":"thomas@goirand.fr","username":"thomas-goirand"},"date":"2026-03-20 14:20:39.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"1d6dae423286a94ba3f0a5a3c6e1ef62521b3184","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":6476,"name":"Thomas Goirand","email":"thomas@goirand.fr","username":"thomas-goirand"},"date":"2026-03-20 14:24:25.000000000","message":"Uploaded patch set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"4a6b448829e93f34e206420a4ff8b7a15afaff67","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":6476,"name":"Thomas Goirand","email":"thomas@goirand.fr","username":"thomas-goirand"},"date":"2026-03-20 14:29:53.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"894ee22a11339da0d50fed08d3b0a5d4f6b48c74","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2026-03-20 15:02:05.000000000","message":"Patch Set 3: Code-Review-1\n\n(2 comments)","accounts_in_message":[],"_revision_number":3},{"id":"7af8fc1b99322f926efb194196401392449108b7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-20 15:44:09.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/44d67189103e4cc4a452bbeaafdf0d71\n\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/781638db7a344a959ebd026f0a6b4ffb : SUCCESS in 1h 00m 54s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/f0f9ef01b8a34749b1fab7a5c25693df : SUCCESS in 1h 06m 46s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/218371d873c74053b8ff5731ae0378f8 : SUCCESS in 6m 48s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/14ff0869599949deade4c86474497d77 : SUCCESS in 5m 24s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/f2f1ab54b069477cb4ccbef7cc75cef1 : SUCCESS in 2m 49s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/c1ee13f2538249dbaf5b085021cd1b80 : SUCCESS in 4m 17s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b145ccf046c04ad9a2953e1c0d58d6b9 : SUCCESS in 6m 05s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/da1038ed81844e108e43dbde84980a79 : SUCCESS in 9m 37s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/eeb0321e2e3f46888a6ecff3052261c0 : SUCCESS in 4m 19s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/84803d4a45274c11b25cbe8bd2d84c5b : SUCCESS in 34m 13s","accounts_in_message":[],"_revision_number":3},{"id":"cc7af3087d40a77c92f8224fd5fb0bf03b0fa07a","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-20 16:00:21.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/540c31bb969a4888b2eb6e52f485ca49\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/ba462982de484744b2b8917c3430752f : SUCCESS in 10m 11s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/e5852cc31f114cbe83d722044ff57a75 : SUCCESS in 6m 11s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/f94461d6d72843e6b3a522e82c5df112 : SUCCESS in 9m 21s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"2ba6674e6d2ce394b12674d6d163655a43e43382","author":{"_account_id":6476,"name":"Thomas Goirand","email":"thomas@goirand.fr","username":"thomas-goirand"},"date":"2026-03-25 15:07:25.000000000","message":"Patch Set 3:\n\n(2 comments)","accounts_in_message":[],"_revision_number":3},{"id":"4db84340c7e779462d3cfee1475db02df919b8d6","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2026-03-26 09:55:24.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"82eb243374ddd1cc87116c8d6454a720c3654aa6","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-26 09:55:49.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"5e2b8f631353458bece3f6f8d6ce609620e5466d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-26 10:58:54.000000000","message":"Patch Set 3: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/bf9e49aac6b842159b2800300d762a0b\n\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/18b168f3ae12483088b2230f807d3a29 : FAILURE in 17m 59s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/962fe1984ff14457b6cd9ff468282435 : SUCCESS in 58m 41s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/deb9b170da8f49c5929935f9c9b20ead : SUCCESS in 4m 36s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/2b9ef906a598435cb537212dc383df48 : SUCCESS in 4m 58s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/1453765333bb4789bb66df32c777c947 : SUCCESS in 5m 07s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ec3913d1dd5f47de829346b017e01d49 : SUCCESS in 5m 43s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/827ee8acb6d0441283a15daaabb9d848 : SUCCESS in 7m 27s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/b0288208a42748d49c47a43436949560 : SUCCESS in 5m 11s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/dced2c20dc7d4edaaf95a41c343fc2be : SUCCESS in 17m 20s","accounts_in_message":[],"_revision_number":3},{"id":"a068bcf747930f9827cb2c65f791b2383d20f829","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2026-03-26 13:29:21.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"bce40693ca625cbb69afa33d1bdac5c1418e594b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-26 14:09:27.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/86ff4c4e7dcd43d89553c2904337c952\n\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/d3c37fdc744e45b7b0497ab2a92b29f2 : SUCCESS in 31m 51s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/27dc4d39e38c4d0a9ffc5de3c75a3c3a : SUCCESS in 34m 40s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/045c02a20f06423f80541fb7c98dbdce : SUCCESS in 2m 55s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/1ff70155693f4376a632b4163f54209a : SUCCESS in 3m 02s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/8cbeba8dfaef47e88293a126d14f94f8 : SUCCESS in 5m 13s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/3e392cadc8bc4c0f8648cab09222722d : SUCCESS in 5m 06s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9cdbf2fbf84840d9a5d52322d72205c7 : SUCCESS in 11m 12s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/bd9a72c8ca0344ef9ffc60968d811fe0 : SUCCESS in 6m 21s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/079adba5e6494a4399df5c3ed2ce6346 : SUCCESS in 6m 32s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/0dd39e5f679447a1bbb2d80e89481fac : SUCCESS in 34m 58s","accounts_in_message":[],"_revision_number":3},{"id":"2d1851da9252c1b2ab854360e0522b25ee7edc9a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-26 14:10:07.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"60a55087c2fe3e4a75499bb508d1087ea5970dee","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-26 15:55:00.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b9983ddae67242009372c6a6f32511ce\n\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/4735c486251a4faf9b6a96d6cd750815 : SUCCESS in 57m 24s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/1897bbcab88747fa9f70c941a0e78188 : SUCCESS in 1h 04m 43s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e77149dd294143f1a9b435cafb0d0722 : SUCCESS in 5m 45s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/95fcfb7cafc445daa8e785fbde1e2808 : SUCCESS in 5m 22s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/a7bca9b7e2d2466cbb95fd5a44b3a87f : SUCCESS in 3m 10s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8ae672304ce147a3ba072f59e4bb74c8 : SUCCESS in 6m 06s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/ec88b9c6a542437a9a8e9d0a5ebd9a2c : SUCCESS in 5m 38s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/256d68d7390f44648addd6dfd4a78790 : SUCCESS in 6m 57s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/95bc9de013f046ad89795ad45b37a8b6 : SUCCESS in 36m 13s","accounts_in_message":[],"_revision_number":3},{"id":"04179e3e4c143ccf0c99fd60f8cc0116861a1d50","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-26 15:55:00.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3},{"id":"13f18f6b96dc407256abacc00d652d86812a1c47","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-26 15:56:44.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b1d16da842844762a92939d0897667f3\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/514be2e5569d475ebd2ff884a2907183 : SUCCESS in 53s","accounts_in_message":[],"_revision_number":3},{"id":"bae368b8a9ac185cda60249266bef3a58b2841c1","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-26 18:07:47.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/013643b6042c4b25964fbdf70f4c617f\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/dde08c852ad541ad9c3a8d1cfae87d5c : SUCCESS in 8m 13s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/87eaf4df739e40aa9b0dc04ab18f344c : SUCCESS in 7m 28s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/6274088e22fd4aaf8dcab0fd6efb151b : SUCCESS in 11m 09s (non-voting)","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"ee97a78e03bc297f10443a3bd912f2ad26583bd0","revisions":{"90829899f086927cd4b8ad10bb747f43ad89b940":{"kind":"REWORK","_number":1,"created":"2026-03-20 14:20:39.000000000","uploader":{"_account_id":6476,"name":"Thomas Goirand","email":"thomas@goirand.fr","username":"thomas-goirand"},"ref":"refs/changes/20/981520/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/20/981520/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/20/981520/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/20/981520/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/20/981520/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/20/981520/1"}}},"commit":{"parents":[{"commit":"8d6f68d7de89c1e2891f530f99609fc0b15e7dee","subject":"Merge \"Fix missing context when replacing image location\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/8d6f68d7de89c1e2891f530f99609fc0b15e7dee"}]}],"author":{"name":"Thomas Goirand","email":"zigo@debian.org","date":"2026-03-20 14:19:24.000000000","tz":60},"committer":{"name":"Thomas Goirand","email":"zigo@debian.org","date":"2026-03-20 14:20:23.000000000","tz":60},"subject":"No DNS resolution in test","message":"No DNS resolution in test\n\nThis avoids DNS resolution during unit tests:\ntest_utils.ImportURITestCase.test_ignored_filtering_options\nand\ntest_utils.ImportURITestCase.test_validate_import_uri\n\nwhich are failing in downstream distributions that do not have\nInternet access when running tests.\n\nSigned-off-by: Thomas Goirand \u003czigo@debian.org\u003e\nChange-Id: I99103c7846c05df38f8477357f7f7693b28fe8c7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/90829899f086927cd4b8ad10bb747f43ad89b940"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/90829899f086927cd4b8ad10bb747f43ad89b940"}]},"branch":"refs/heads/master"},"c04c78ea0e9ada69868049cf3eb359d0656fe6d4":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2026-03-20 14:24:25.000000000","uploader":{"_account_id":6476,"name":"Thomas Goirand","email":"thomas@goirand.fr","username":"thomas-goirand"},"ref":"refs/changes/20/981520/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/20/981520/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/20/981520/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/20/981520/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/20/981520/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/20/981520/2"}}},"commit":{"parents":[{"commit":"8d6f68d7de89c1e2891f530f99609fc0b15e7dee","subject":"Merge \"Fix missing context when replacing image location\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/8d6f68d7de89c1e2891f530f99609fc0b15e7dee"}]}],"author":{"name":"Thomas Goirand","email":"zigo@debian.org","date":"2026-03-20 14:19:24.000000000","tz":60},"committer":{"name":"Thomas Goirand","email":"zigo@debian.org","date":"2026-03-20 14:23:57.000000000","tz":60},"subject":"No DNS resolution in test","message":"No DNS resolution in test\n\nThis avoids DNS resolution during unit tests:\ntest_utils.ImportURITestCase.test_ignored_filtering_options\nand\ntest_utils.ImportURITestCase.test_validate_import_uri\n\nwhich are failing in downstream distributions that do not have\nInternet access when running tests.\n\nNote this is a regression introduced by the OSSA-2026-004 patch:\nIb8d337dc68411d18c70d5712cc4f0986ef6205f4\n\nSigned-off-by: Thomas Goirand \u003czigo@debian.org\u003e\nChange-Id: I99103c7846c05df38f8477357f7f7693b28fe8c7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/c04c78ea0e9ada69868049cf3eb359d0656fe6d4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/c04c78ea0e9ada69868049cf3eb359d0656fe6d4"}]},"branch":"refs/heads/master"},"ee97a78e03bc297f10443a3bd912f2ad26583bd0":{"kind":"REWORK","_number":3,"created":"2026-03-20 14:29:53.000000000","uploader":{"_account_id":6476,"name":"Thomas Goirand","email":"thomas@goirand.fr","username":"thomas-goirand"},"ref":"refs/changes/20/981520/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/20/981520/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/20/981520/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/20/981520/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/20/981520/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/20/981520/3"}}},"commit":{"parents":[{"commit":"8d6f68d7de89c1e2891f530f99609fc0b15e7dee","subject":"Merge \"Fix missing context when replacing image location\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/8d6f68d7de89c1e2891f530f99609fc0b15e7dee"}]}],"author":{"name":"Thomas Goirand","email":"zigo@debian.org","date":"2026-03-20 14:19:24.000000000","tz":60},"committer":{"name":"Thomas Goirand","email":"zigo@debian.org","date":"2026-03-20 14:29:44.000000000","tz":60},"subject":"No DNS resolution in test","message":"No DNS resolution in test\n\nThis avoids DNS resolution during unit tests:\ntest_utils.ImportURITestCase.test_ignored_filtering_options\nand\ntest_utils.ImportURITestCase.test_validate_import_uri\n\nwhich are failing in downstream distributions that do not have\nInternet access when running tests.\n\nNote this is a regression introduced by the OSSA-2026-004 patch:\nIb8d337dc68411d18c70d5712cc4f0986ef6205f4\n\nSigned-off-by: Thomas Goirand \u003czigo@debian.org\u003e\nChange-Id: I99103c7846c05df38f8477357f7f7693b28fe8c7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/ee97a78e03bc297f10443a3bd912f2ad26583bd0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/ee97a78e03bc297f10443a3bd912f2ad26583bd0"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
