)]}'
{"id":"openstack%2Fglance~982433","triplet_id":"openstack%2Fglance~master~I525a2ddea7bdd72a12fa16482dd5fd7faddf3fd2","project":"openstack/glance","branch":"master","topic":"eventlet-removal","attention_set":{},"removed_from_attention_set":{"9303":{"account":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"last_update":"2026-03-31 14:36:17.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I525a2ddea7bdd72a12fa16482dd5fd7faddf3fd2","subject":"Remove usage of eventlet form utils","status":"MERGED","created":"2026-03-27 14:37:38.000000000","updated":"2026-03-31 14:37:37.000000000","submitted":"2026-03-31 14:36:17.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"982433-eventlet-removal","meta_rev_id":"fbc00f1cec15759d19ec488aeddcff2ae4ddb602","_number":982433,"virtual_id_number":982433,"owner":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-03-31 14:36:17.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"all":[{"value":2,"date":"2026-03-30 00:17:29.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"all":[{"value":1,"date":"2026-03-31 12:12:40.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Top Priority / Holds Gate"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-03-27 16:58:22.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-03-30 00:17:29.000000000","updated_by":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"reviewer":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"state":"REVIEWER"}],"messages":[{"id":"0a2d4a9c794f6be796f398f7d153f659285e0b9b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2026-03-27 14:37:38.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"1d8e89312ff3ae7f9aefa11810489dfabf621573","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2026-03-27 14:37:58.000000000","message":"Topic set to eventlet-removal","accounts_in_message":[],"_revision_number":1},{"id":"104e3ab432d190f797e628db88ff8fb33b08af58","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-27 16:58:22.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/b5bb2b861d8e422aa8b916c82c13c9f5\n\n- grenade https://zuul.opendev.org/t/openstack/build/f9d6a704a42b481891bf12582c0c443b : SUCCESS in 36m 26s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/b692847375c940728e9a62e3a662d8f1 : SUCCESS in 1h 09m 57s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/bd2139c9003f4e4aa1e494f785cf72f7 : SUCCESS in 1h 54m 30s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/0928234d4e744242abde164683c65075 : POST_FAILURE in 11m 12s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/58bc54eeb7294078960d49a2c10fff02 : SUCCESS in 4m 44s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/d467e10ee7b348afa441c65cf1f23f9f : SUCCESS in 4m 44s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/8cfa6007a40941c8acab50a6555edc2f : SUCCESS in 3m 20s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/562068a0654f47b8a82dae05ecaa99e7 : SUCCESS in 7m 00s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d2d4fd9440624e5bb3f3878d6312c157 : SUCCESS in 4m 48s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/8549a27e46544cfe8969c320335e4c14 : SUCCESS in 6m 22s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/4a910c9ba23e431d8e8b74b9134ee0ba : SUCCESS in 6m 07s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/4eaca6c11f2c4cc688df68b2e23d4a6b : SUCCESS in 1h 32m 41s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/a89f5bec64ba4e199f488f21aaaee2e2 : SUCCESS in 1h 25m 36s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/7224ddf3ae784ec7b33f9c464689e6d0 : SUCCESS in 2h 14m 59s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/847a161e09684dbf82bdb574a4eb21f8 : SUCCESS in 1h 10m 13s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/bbafd391f8e140de82f84d735b791042 : SUCCESS in 1h 02m 32s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/c38ae7935d9e4f16915a5118fa7b4e0e : SUCCESS in 1h 48m 58s\n- glance-grenade-centralized-cache https://zuul.opendev.org/t/openstack/build/572219488636478796f8cbc2f5346733 : FAILURE in 15m 14s (non-voting)\n- glance-s3-multistore https://zuul.opendev.org/t/openstack/build/b4aac5707cd24c5092ece904299b4d0a : SUCCESS in 2h 03m 46s (non-voting)\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/0f4b0afe0a364623a3bc9fcc99cf721a : SUCCESS in 36m 19s","accounts_in_message":[],"_revision_number":1},{"id":"465bafa49f94d323221ee3de837e185ee7b7eb99","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-27 17:44:06.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/02dc9ae5457d4c6bb54195562750cf52\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/7bea8cc359cd470b9dceca385bf1efda : SUCCESS in 12m 11s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/8b59ac3d6b7247a6bb32a562810085ff : SUCCESS in 9m 47s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/3b4690599a9243799c48e66190624da6 : SUCCESS in 14m 00s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"a23113660f27e2f33e0e6fe37355627dba67a702","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2026-03-27 17:52:44.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"e771646c36f13b5ce46f70ed7238dde714955cc5","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-27 19:11:08.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/875870fa90704c5083b50629bbb1f538\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/b5a110010a1a45ca8629f4d979f059a5 : SUCCESS in 7m 42s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/197c64f4a44c40caaec500ad6c5dc2b1 : SUCCESS in 7m 02s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/f16ab9b8b6d047f39b300bb60728e5bc : SUCCESS in 11m 54s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"58f186b745fed5d5a48f316c243c5ee046d1a5de","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-27 20:20:03.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/fb08d705009e4572ac943f44a49da5af\n\n- grenade https://zuul.opendev.org/t/openstack/build/60b8ace780a1406281fbed3194efa0ae : SUCCESS in 1h 03m 22s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/c47e11203c8546648c0089175f2688d2 : SUCCESS in 1h 00m 11s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/35ad47d2f2d14db18bcc4111b3917810 : SUCCESS in 1h 41m 09s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/9489010c167140458f88f86a923f9360 : SUCCESS in 55m 38s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/09375de406c54018ad000d09e914c609 : SUCCESS in 6m 17s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/e59f3bbd3ee749ffa1639d4a8cd1dd83 : SUCCESS in 4m 32s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/37d5310837204963a0803df944f533c8 : SUCCESS in 5m 06s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/2ebf997ce0a042ea84703a7bef44c66e : SUCCESS in 7m 48s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/378fdb9d0e3841a4a07b1f53c1168ad4 : SUCCESS in 8m 43s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/11af494e7af848b3a992fcc770d2b8c1 : SUCCESS in 6m 42s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/f4d444727769497ebea3a6fe5bf58827 : SUCCESS in 8m 01s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/eb7c8036b7944e878922d5ebc197126c : SUCCESS in 1h 26m 05s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/6d20692c85704aabaa3c5a4aaf4e07d3 : SUCCESS in 1h 42m 49s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/e2ec1f1bfc3b404d8c909436a22ce993 : SUCCESS in 1h 52m 00s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/a8ba7531418b4ada8ec77637e01bf718 : SUCCESS in 2h 24m 48s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/a6dd561fc98b42198227e6a3b668aa08 : SUCCESS in 1h 10m 28s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/c36ba3c8f6a9404598b6cde00cc10571 : SUCCESS in 1h 43m 38s\n- glance-grenade-centralized-cache https://zuul.opendev.org/t/openstack/build/f14f1c9e244e4d9b8f129664e2154ff1 : FAILURE in 15m 49s (non-voting)\n- glance-s3-multistore https://zuul.opendev.org/t/openstack/build/4e59b177c3b843e88f2a1c795359eedd : SUCCESS in 2h 01m 16s (non-voting)\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/004f50e37b3444bd9c5872d904b3aa71 : SUCCESS in 34m 44s","accounts_in_message":[],"_revision_number":1},{"id":"41c6f47bfd924b5c97e7fcacb693d7860a92f0fd","author":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"date":"2026-03-30 00:17:29.000000000","message":"Patch Set 1: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"7964d5559778e5f531319af49e66fbe7f64e6863","author":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"date":"2026-03-31 12:12:40.000000000","message":"Patch Set 1: Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"1377637971349e9b0cfd8c14fea6aee0c2f6b7d0","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-31 12:13:13.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"d4101ccea12f8e01c26a36956aa2b32d34cef769","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-31 14:36:17.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d71d48bdd3824f22bb9f951d09ef97ed\n\n- grenade https://zuul.opendev.org/t/openstack/build/d4b6d3549874433280be004b50ecc41c : SUCCESS in 57m 24s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/6a7aea7b5e954935a39d34a7d18ec77b : SUCCESS in 1h 07m 56s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/44b444529bfd4fb286c2ee32ae03173b : SUCCESS in 1h 34m 16s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/071c3ebcc3cb4ab4806f9ac2b533c20e : SUCCESS in 1h 07m 45s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/00ee08be252e49468cb0f5b345d28924 : SUCCESS in 7m 02s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/a59e6fd5b2874b91879739eba938d791 : SUCCESS in 5m 25s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/9f304229f7394419a12563944196c3b6 : SUCCESS in 6m 53s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/11f745e34cf2493db273af21e16ae05b : SUCCESS in 6m 05s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/97ab6f2cbc384026b2e6ec28a9712b86 : SUCCESS in 7m 38s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/654bf985944f497f9ba8267af8857358 : SUCCESS in 6m 46s\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/b72f043636d64852b0b1c4a879a7fec4 : SUCCESS in 2h 17m 03s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/41dbc05873c2411aae9e5059c0510e06 : SUCCESS in 2h 00m 27s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/392d8e7ebb2344df9d6e0284d82ada15 : SUCCESS in 57m 06s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/a31441a4da964b7dac9feb525daae956 : SUCCESS in 1h 45m 48s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/ada187f63ba34809a8c5aa550e401a21 : SUCCESS in 34m 48s","accounts_in_message":[],"_revision_number":1},{"id":"57b820167e38c51b16cba1aa7dcbe00273bdc1a8","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-31 14:36:17.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"fbc00f1cec15759d19ec488aeddcff2ae4ddb602","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-31 14:37:37.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/62676a3b6e5f4ae4ad2a06a1f6111695\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/41b2ef9208d5495fb9d299e493955c73 : SUCCESS in 1m 04s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"95d769e3cc82b03c9ec2f32d5a8bf726705579ac","revisions":{"95d769e3cc82b03c9ec2f32d5a8bf726705579ac":{"kind":"REWORK","_number":1,"created":"2026-03-27 14:37:38.000000000","uploader":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"ref":"refs/changes/33/982433/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/33/982433/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/33/982433/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/33/982433/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/33/982433/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/33/982433/1"}}},"commit":{"parents":[{"commit":"5192128670c67b0d1d4209e88eef97c01c638185","subject":"Merge \"No DNS resolution in test\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/5192128670c67b0d1d4209e88eef97c01c638185"}]}],"author":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2026-03-27 14:35:16.000000000","tz":0},"committer":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2026-03-27 14:35:16.000000000","tz":0},"subject":"Remove usage of eventlet form utils","message":"Remove usage of eventlet form utils\n\nRemoved use of eventlet socket from utils modules as\nwell as some mentioned in the comments.\n\nChange-Id: I525a2ddea7bdd72a12fa16482dd5fd7faddf3fd2\nSigned-off-by: Abhishek Kekane \u003cakekane@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/95d769e3cc82b03c9ec2f32d5a8bf726705579ac"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/95d769e3cc82b03c9ec2f32d5a8bf726705579ac"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
