)]}'
{"id":"openstack%2Fglance~983415","triplet_id":"openstack%2Fglance~master~Ibfff59c2726cb8fcb8ef7672d4f174dadeebe28e","project":"openstack/glance","branch":"master","attention_set":{},"removed_from_attention_set":{"19138":{"account":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"last_update":"2026-08-19 07:07:25.000000000","reason":"Change was submitted"},"9303":{"account":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"last_update":"2026-08-17 06:42:45.000000000","reason":"\u003cGERRIT_ACCOUNT_9303\u003e replied on the change","reason_account":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"}},"8122":{"account":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"last_update":"2026-07-28 16:37:56.000000000","reason":"\u003cGERRIT_ACCOUNT_8122\u003e replied on the change","reason_account":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"}},"8556":{"account":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"last_update":"2026-08-19 07:07:25.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ibfff59c2726cb8fcb8ef7672d4f174dadeebe28e","subject":"Add glance-multistore-add-location-do-secure-hash-false job","status":"MERGED","created":"2026-04-06 10:16:25.000000000","updated":"2026-08-19 07:09:31.000000000","submitted":"2026-08-19 07:07:25.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":15,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"983415","meta_rev_id":"9cffc4f8e8290d21377df5d6ee0513688ce4d111","_number":983415,"virtual_id_number":983415,"owner":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-08-19 07:07:25.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"all":[{"value":2,"date":"2026-08-19 05:36:32.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":2,"date":"2026-07-28 16:37:56.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"all":[{"value":1,"date":"2026-08-19 05:36:32.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Top Priority / Holds Gate"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-04-06 10:28:23.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2026-04-06 12:37:06.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-04-13 10:59:07.000000000","updated_by":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"reviewer":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"state":"REVIEWER"},{"updated":"2026-04-13 10:59:07.000000000","updated_by":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"reviewer":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"state":"REVIEWER"},{"updated":"2026-04-13 10:59:07.000000000","updated_by":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"reviewer":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"state":"REVIEWER"}],"messages":[{"id":"ac061ac330fe23f31db5ae3e365abbd8c8c0f9e5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"date":"2026-04-06 10:16:25.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"faf97e62fff3161743f1db50d65ebd4caeca7509","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-06 10:28:23.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/239a7bd044fd4263bcffbb3121d80d35\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/e5d1552c3b3f4aadbe7dbe2c53b0cff8 : SUCCESS in 6m 24s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/b3aea423108f4564a762af05aeef25c4 : SUCCESS in 6m 31s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/fd28ba01cb5e4a3a9a5b7192f1057d00 : SUCCESS in 8m 55s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"e48c995c271a04bb833419050974537b00147d03","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-06 12:37:06.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/28674ef3aee34a2e8bff5bb5f1432a33\n\n- grenade https://zuul.opendev.org/t/openstack/build/a5f73bd6410346659762361c5d673866 : SUCCESS in 51m 59s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/86fb5367cf714688bee6805c4b598aa0 : SUCCESS in 1h 11m 47s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/fcedbfa417c84e0292e9e09b347ffd7b : SUCCESS in 1h 42m 22s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/7ebcc4c2d19542ae970c86310b7bc7ad : SUCCESS in 1h 03m 10s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/370f921c768e4619b0836795c3771866 : SUCCESS in 4m 29s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/629829c3a28b4b73b5889da4b4d12340 : SUCCESS in 4m 57s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/8697aea34ac54c76aa9bf172e1ff92c5 : SUCCESS in 4m 57s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/7cf2e58141294364b8dc58fc3d0756c1 : SUCCESS in 7m 34s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/729554842c394b99b856555214f57ccf : SUCCESS in 5m 33s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/cd33b9564f3449009674bf5d2ba203b0 : SUCCESS in 6m 22s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/37719c9afaff4a7bba83095fba4e724b : SUCCESS in 6m 10s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/3e62e9b4d7614506a2dd5f037827d94c : SUCCESS in 1h 57m 10s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/90d6f3dc0cb74d37ac81f6ab68304e28 : SUCCESS in 1h 57m 42s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/1237a4f3dc2f4f92a8d8d98608c3af6c : SUCCESS in 2h 14m 41s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/98fb6165ab6348b3827a1b137e991837 : SUCCESS in 1h 50m 23s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/26930b3b74d54aa89fb8048996b813d3 : SUCCESS in 1h 03m 00s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/b32c0ebfdfd4492cb453b570bd1d7da0 : SUCCESS in 1h 53m 04s\n- glance-grenade-centralized-cache https://zuul.opendev.org/t/openstack/build/2d12b68bfcaa4cb8b40a197fc08e4b33 : FAILURE in 15m 52s (non-voting)\n- glance-s3-multistore https://zuul.opendev.org/t/openstack/build/067d5a0d595042d3b6f4314333e2f5da : FAILURE in 2h 00m 26s (non-voting)\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/90ad8cbdae4a42909bb08d79cc8a13de : SUCCESS in 31m 58s","accounts_in_message":[],"_revision_number":1},{"id":"02febf536cc140739c7fc4bfe18dc437c8da6f49","author":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"date":"2026-04-28 14:27:44.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"692ba7feb6267019602957fb202e3ef677431200","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2026-04-29 00:00:02.000000000","message":"Patch Set 1: Code-Review-1\n\n(2 comments)","accounts_in_message":[],"_revision_number":1},{"id":"c02b5f40a47a52dfb3537d9e9f4dbddb86ff4470","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2026-04-29 00:12:23.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"962c0e32595d4b682194e8d6f6890f8a53023786","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2026-04-29 00:14:18.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"1a2ab0f5096e9e587cdeb561746f100376add73a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"date":"2026-07-14 07:46:43.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Code-Review+2, Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":2},{"id":"452af70f3c8b5e1351e44c75a9e39af6d5396c1f","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-14 08:12:06.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c032f9cad06d4a338d8247fc0625e46a\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/6e6f9917a4124f9e910418a2f7147646 : SUCCESS in 7m 57s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/c660795a08fc4491820b45bcd1985915 : SUCCESS in 7m 00s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/4334c4c808fd42f79bc44c53ca515ab7 : SUCCESS in 12m 07s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"16fb4be9d638a7c97903237c0e94c0cfb781735e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-14 09:54:41.000000000","message":"Patch Set 2: Verified-1\n\nThis change depends on a change that failed to merge.\n\nChange https://review.opendev.org/c/openstack/tempest/+/891026 is needed.","accounts_in_message":[],"_revision_number":2},{"id":"4e75279953eea051ac0cb3febf4ab33a9266f69f","author":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"date":"2026-07-14 09:57:02.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"717f71eafdb734891255833706d1a911e7bf31f0","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-14 10:27:01.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8665db37ec2e46388f9d24c1fe7fb1ce\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/7c33ffe23c694a4bb71b27b4e77f9b50 : SUCCESS in 7m 24s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/191f73c60472491d9ef2cfa8edc8e58a : SUCCESS in 8m 43s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/faf98d6a8b90441da75a87eaa923d0c6 : SUCCESS in 12m 13s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"4d1e7c7eef308fea601b1a46221e089e59af9f59","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-14 11:57:00.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/3014ce76f65c4fcd890bab51d451fb87\n\n- grenade https://zuul.opendev.org/t/openstack/build/4efb4c3202b64334842189f15c9c408a : SUCCESS in 1h 16m 26s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/56b42c51968547fc862ef90b1b3e5712 : SUCCESS in 1h 00m 11s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/382179f684354498b1e8fc67907f5dd6 : SUCCESS in 1h 25m 44s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/d78ade1d990b48c59b264fd95a686890 : SUCCESS in 59m 32s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7f70d7e6ae944d5382caad3af6b52d24 : SUCCESS in 4m 58s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/3f3c161b49e64a6cb52936b6dd2d99a4 : SUCCESS in 4m 35s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/b04edb684b934b478990dfe7bde7bd38 : SUCCESS in 7m 08s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/60560d23261b4bb2a862c35f604b04b3 : SUCCESS in 7m 51s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/50e3e1d7618640b19f95a21de690f14a : SUCCESS in 5m 40s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e2ee28807e654e83bf50c74597ee5c94 : SUCCESS in 4m 59s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/6681040451534df6a136556633529f9d : SUCCESS in 6m 48s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/59e4747457814c9d9216c96c44022af4 : SUCCESS in 6m 24s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/877a67580f264492b920e46f2d5b9c22 : SUCCESS in 1h 27m 53s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/fb77724e214d4f3e9617826eae2254d3 : SUCCESS in 1h 13m 24s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/bb72d6b2c19d481f894ee77f2c3d169f : FAILURE in 1h 37m 36s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/147a0938846c4b96a9b76064ffcdcb1d : SUCCESS in 1h 27m 33s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/fd0ae8ee7ac744039dc474a1576fc826 : SUCCESS in 1h 03m 48s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/9a135d1ee9fa4385a4f83c065dc730e4 : SUCCESS in 41m 55s\n- glance-s3-multistore https://zuul.opendev.org/t/openstack/build/458ce274ffbd45de97aca7587fe644db : SUCCESS in 1h 45m 01s (non-voting)\n- glance-s3-multistore-do-secure-hash-false https://zuul.opendev.org/t/openstack/build/6f99c44a8a8245aab1a6b7456dc10dfb : SUCCESS in 1h 28m 12s (non-voting)\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/e769071272e74f868d6f16eaebe60fdc : SUCCESS in 17m 31s\n- glance-rally-task-uwsgi https://zuul.opendev.org/t/openstack/build/24c85dab1cdf4f8eb4ca668099be3ce3 : SUCCESS in 40m 12s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"67e0eb0dc5d600ccd5e909ecd16563e863373124","author":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"date":"2026-07-14 13:36:37.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"fa4f2bdc9e172a7c22fa70f3513f28adcc326666","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-14 15:14:55.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/dd0cf751b411466cb21aa5c8bf6d1a5e\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/14c395e7759248e0b3c3d31758e5f000 : SUCCESS in 6m 11s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/957a81dd823a4eaea014868e8151f9a9 : NODE_FAILURE Node(set) request b11292866576437197084018ee4d119f failed in 0s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/150f45b698f54a83bd9a3fd6b605143d : SUCCESS in 15m 16s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"14ca55caaa3cdfaff3b8a02e270d4ed95686aa60","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-14 15:37:06.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3c23369c2313420e8d239bb66d74e795\n\n- grenade https://zuul.opendev.org/t/openstack/build/5349b46c52bc4b7fbc03ce169f737ad3 : SUCCESS in 1h 04m 35s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/19819f51c30f4955bbe48a6a9f65af95 : SUCCESS in 43m 29s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/ff6a67c20b8248ffb23866956aba5b0f : SUCCESS in 1h 30m 26s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/0a8b4d24b19247b6ba109d2a659bee8c : SUCCESS in 59m 02s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9f7008dbd7624020b0c0aacbd5a57aea : SUCCESS in 4m 51s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/958034fec662437eb7595e6098fe4e3d : SUCCESS in 5m 09s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/a4ee260e013c4c5d9fa5824f91c24afb : SUCCESS in 6m 57s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/9da88520e9814e9a97e03d3f8dad7c47 : SUCCESS in 7m 06s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c3a4cfc7dfd14695a20d09d70f354051 : SUCCESS in 5m 53s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d3454217e58b4cd5b7ebf57fad8141d6 : SUCCESS in 8m 33s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/f4f3eacdfe894bba8066d960f66fedfa : SUCCESS in 7m 09s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/45bbb1a9c79e434da8b9324c03288891 : SUCCESS in 7m 08s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/771061e16c8149379771679d38f03598 : SUCCESS in 1h 31m 25s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/2e699bd1cbda4859ab658e19361da144 : SUCCESS in 1h 43m 08s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/367b3326e8bb443eb6e28461a473d98a : SUCCESS in 1h 43m 47s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/3191ebedc7bb4197a26bbfcf9108d2ec : SUCCESS in 1h 43m 25s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/46f9ee6fdbaf4fe4a7ccecfa62836214 : SUCCESS in 52m 58s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/acc0355b49aa403a9cf6babd5ad06e0b : SUCCESS in 1h 23m 47s\n- glance-s3-multistore https://zuul.opendev.org/t/openstack/build/fd389a35533e48458d4321359d33358c : SUCCESS in 1h 27m 16s (non-voting)\n- glance-s3-multistore-do-secure-hash-false https://zuul.opendev.org/t/openstack/build/216da3e34a154ca5b44707da26dfe3b8 : SUCCESS in 1h 48m 14s (non-voting)\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/1c5826e65f80460e934ed290bf8e3909 : SUCCESS in 34m 31s\n- glance-rally-task-uwsgi https://zuul.opendev.org/t/openstack/build/0a87ca7bbb2849908e3e03fe5f874ce0 : SUCCESS in 1h 00m 18s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"1dd52cbc85406bafcfd27e62f223ff20aebb83f1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"date":"2026-07-28 05:52:32.000000000","message":"Patch Set 3: Published edit on patch set 2.\n\nOutdated Votes:\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":3},{"id":"f046a1b002756d984f4793c98bc33dbae94f2197","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-28 06:34:04.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/cfe151a8b2e74f9e90b844dc4dda3c89\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/6180e332cb004cdb8ce4ca42d0da3ec4 : NODE_FAILURE Node(set) request d61487c122a84379aacd9535325f9f52 failed in 0s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/21106f1d6db645458faad5267b17498a : SUCCESS in 9m 05s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/a3e8587222d34a0c99bd93c3fe52ef83 : SUCCESS in 12m 49s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"b4dbd53c94870b0498c7926b1925b4c324683e1b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-28 08:04:13.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1532cf99080c4cbe94da0306f86e6390\n\n- grenade https://zuul.opendev.org/t/openstack/build/4ebdfa1b9f604921b26e800a6771f75f : SUCCESS in 1h 17m 06s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/db1388cb12004288b86412767b64354b : SUCCESS in 34m 04s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/c3776c9b70694b22996e6c30f954afeb : SUCCESS in 1h 14m 53s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/24bedf73a6574a4abdfb885a54034612 : SUCCESS in 49m 35s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/330c968e8b1543f4836a569ea17cbef5 : SUCCESS in 4m 56s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/7b0929057abd417dafb145a8f76f05e7 : SUCCESS in 3m 42s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/bb8be3318af14f2e9ba0f882c93d1908 : SUCCESS in 6m 38s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/8cd309b1cc55404f8bba0c4ae06c69aa : SUCCESS in 9m 24s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cfb0012342f8431eab5fcf77cf5c72aa : SUCCESS in 5m 44s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/004bec3cc4104587a567d80fa86b8972 : SUCCESS in 7m 55s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/63a3f3e1c78c487ba757b34288a3c784 : SUCCESS in 6m 09s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/ac3e23c5661b4213b74925a7c36f022e : SUCCESS in 6m 15s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/716713e70dce4dd3945439f20fb02f0b : SUCCESS in 56m 33s (non-voting)\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/3d632ca06ab4437b9f8b86aee4b39ada : SUCCESS in 59m 57s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/240ed3897ec04e928fd202abe2729c4c : SUCCESS in 1h 56m 14s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/8b3d0a71ee6846d798a5459b41c770b4 : SUCCESS in 31m 50s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/747cc1a2513942c2a97587f508d44fcc : SUCCESS in 1h 25m 27s\n- glance-s3-multistore https://zuul.opendev.org/t/openstack/build/108cc1744bfb4e8eb1c5425d2bb1b07b : SUCCESS in 51m 48s (non-voting)\n- glance-multistore-add-location-do-secure-hash-false https://zuul.opendev.org/t/openstack/build/9a5a0c8749bc4b2cb56363656382c7c0 : FAILURE in 1h 32m 33s (non-voting)\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/8bb4fe7fb3fc4bbba783304a49f7a7de : SUCCESS in 23m 15s\n- glance-rally-task-uwsgi https://zuul.opendev.org/t/openstack/build/dbe5b4214af4499182822451436dd6e8 : SUCCESS in 54m 21s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"11a25cc088266be8c04fdaa93e21641f9495943b","author":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"date":"2026-07-28 16:37:56.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"5b821f749816acbf51cb3accc3ec94ca3a0ebe0a","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2026-08-04 13:36:54.000000000","message":"Patch Set 3: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"f20f6b03e48b08865abcde649945014114698970","author":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"date":"2026-08-12 06:03:10.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"1311497dd5c4d63ae1144457dea4a4a6c3f11042","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-12 06:24:05.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/abf61ecfd80d42a6a2e16f5aab4e47e2\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/e1b0b3d927354d35956e58615f006fdd : SUCCESS in 8m 06s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/8700c4e84d3d487082d14e0ee1a5c0b6 : SUCCESS in 9m 34s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/7eb70eaf3fa6437997282ed0f9cce1e4 : SUCCESS in 17m 55s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"8f56f71d4d40e46796226e5aa0f3a7b872fd3a0b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-12 08:17:32.000000000","message":"Patch Set 3:\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/305286a0c50c4c7586e6d801fca28c7b\n\n- grenade https://zuul.opendev.org/t/openstack/build/80f41d9045584b18bb115523cafa59c6 : SUCCESS in 34m 22s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/f63aa7de5ada4ffd8d931cb1cfc60dbb : SUCCESS in 30m 32s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/03a3de2a8cca4ec69f58640d7a6cf041 : SUCCESS in 1h 36m 16s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/10eb9bb7faad4d66b04778c20d29fae1 : SUCCESS in 1h 01m 04s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e54cb61a4c7440a7b542f95457f40012 : SUCCESS in 4m 35s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/dbd101f6d8a3446e995591bb8753af18 : SUCCESS in 3m 16s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/27fe74468181438495e16c68cd41d65b : SUCCESS in 4m 21s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/43edf424d1f545ac88a76c3a35a0e319 : SUCCESS in 6m 30s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/04fff15cf9f24767b5a6cdc2f4bc89e7 : SUCCESS in 5m 29s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/2058440d6845431bad346fc457183b39 : SUCCESS in 7m 58s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/e1fa1af0391945089259db0f8cc43e68 : SUCCESS in 4m 48s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/d0cf57903f514a4cb6c36a8f8b2b5dc3 : SUCCESS in 6m 03s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/d5e50e38257a4720ad0d0f222bc4ce28 : SUCCESS in 1h 09m 50s (non-voting)\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/796e63eb3ff24e10a6d7b5d29abebc79 : SUCCESS in 1h 45m 01s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/96fb883efa134c498866be5c4cf0b806 : SUCCESS in 2h 07m 12s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/0e6e1995bfcf40e19c5a85e3c9e988f6 : SUCCESS in 56m 56s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/7fc4458a910743b38493454531e6aa7b : SUCCESS in 1h 12m 58s\n- glance-s3-multistore https://zuul.opendev.org/t/openstack/build/09f6ec26a063458b9974b9e4c719ed61 : SUCCESS in 1h 44m 12s (non-voting)\n- glance-multistore-add-location-do-secure-hash-false https://zuul.opendev.org/t/openstack/build/88260647c29741de9606309b9e0463fa : SUCCESS in 1h 38m 28s (non-voting)\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/aeb911a14f524446b099ac9930cf85cb : SUCCESS in 35m 06s\n- glance-rally-task-uwsgi https://zuul.opendev.org/t/openstack/build/2906e53b55ce4c37b932e17eb0b74ca9 : SUCCESS in 32m 39s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"2884d8cd2e6d37baf27573b0442ba678bb0465d1","author":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"date":"2026-08-17 06:40:23.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"17bf5bdac1240179ebe566d1f925720da065156f","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2026-08-17 06:42:45.000000000","message":"Patch Set 3:\n\n(2 comments)","accounts_in_message":[],"_revision_number":3},{"id":"203fff1d165aa46a4a2239139b7f2849d4613d62","author":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"date":"2026-08-18 05:32:14.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"789abf81fc17b1150be67aeb6fd14e45ed9076fb","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-18 05:52:15.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9511a409241342569f59c4fc692f6ecf\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/b0bb87d7995f4ac0875473cb5315105f : SUCCESS in 11m 03s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/2ac1208d46904530a40415d7853f4247 : SUCCESS in 10m 11s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/4d9f530375d4488f90b7c69152030cea : SUCCESS in 16m 40s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"71facfd055be7798645897150dadee76f320bdb4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-18 06:59:11.000000000","message":"Patch Set 3:\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/fd88db1b86724d80b7bffc4fadff48dd\n\n- grenade https://zuul.opendev.org/t/openstack/build/3102248eb7dd42599859b3fbd1cdae9c : SUCCESS in 1h 02m 27s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/e80fc73171c7425cbe6f1c0f2c843f7d : SUCCESS in 1h 07m 53s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/2dee49c947ef4e34a5838d8cfd389641 : SUCCESS in 1h 23m 59s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/3fb40f82e0ee45a9b12f07797d1236b7 : SUCCESS in 47m 18s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/566fd7d25cbc4984aef27d3f4d1fb51a : SUCCESS in 4m 53s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/88b318944241428d83844ff21cdaf6d1 : SUCCESS in 4m 44s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/d8b23e4233f747608ba2e13bd1ec31be : SUCCESS in 3m 21s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/f1151369a89c4aad8f83f6a6e4a4a224 : SUCCESS in 5m 02s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/44e0a9deabb94d559361d2e3171dbd32 : SUCCESS in 5m 35s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/587fb41dff17430fa5af316a689f9b6a : SUCCESS in 11m 07s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/4f5ddaa565824b5789e01384384d058c : SUCCESS in 6m 04s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/203b09110746441b9b00f50a679bbb50 : SUCCESS in 6m 13s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/3024afa451bb4f818e3be51ba69c4f34 : SUCCESS in 50m 46s (non-voting)\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/f88cfc3462604c58b5625130fa2cd091 : SUCCESS in 1h 16m 05s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/53c1461faead4ce78bbeaa64916637b9 : SUCCESS in 1h 07m 41s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/693f2851917648f8afda61df29b70af0 : SUCCESS in 32m 07s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/da59eca464b54c09bd909da5a1f75f25 : SUCCESS in 1h 15m 08s\n- glance-s3-multistore https://zuul.opendev.org/t/openstack/build/192215c87d524d799630803c7a128f09 : SUCCESS in 1h 09m 03s (non-voting)\n- glance-multistore-add-location-do-secure-hash-false https://zuul.opendev.org/t/openstack/build/844b6179c4f544daa554e0110d11764b : SUCCESS in 45m 32s (non-voting)\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/ff09ec317c784667b44d2eaa39514251 : SUCCESS in 34m 11s\n- glance-rally-task-uwsgi https://zuul.opendev.org/t/openstack/build/01e0d0e0b9d34b9db948a428be755637 : SUCCESS in 48m 35s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"9e4f395ad7e094a8b3acb4978b6bc38cd1e761fe","author":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"date":"2026-08-18 07:10:43.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"cb99a01540e80288f81e9c4e1de9360d4033fe52","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-18 07:27:56.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e5c98f69a025413ca710cb7b240f845a\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/13b221457a9c40c5bce7192537d12234 : SUCCESS in 7m 49s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/bbdc89eb28004c1499392e5d50987e82 : SUCCESS in 7m 38s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/7424263da897412ba15b0881be658f08 : SUCCESS in 12m 31s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"7286bcb2a725104e1cffa6d87b0d2e123882d49a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-18 08:56:32.000000000","message":"Patch Set 3:\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0fd8eb5e98ab41e98214335b51e86121\n\n- grenade https://zuul.opendev.org/t/openstack/build/46bba1be386644b8aa07e23504a0203f : SUCCESS in 1h 08m 17s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/373670784a0f4e27aa281a9bd654f867 : SUCCESS in 1h 00m 15s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/bed4407363194c2c94fe91444d123c81 : SUCCESS in 1h 26m 50s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/a1837bb7373f479896d1c37c0ea03485 : SUCCESS in 58m 56s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/797e51db1f834872ab5e4ead16669beb : SUCCESS in 4m 30s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/bd113560f26b48b782e559bf58dec7de : SUCCESS in 7m 19s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/d4da35b13a494022bd843d2d4a2554d3 : SUCCESS in 4m 49s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/b2040eb7c3f44a7e870a0d90e9ce66ff : SUCCESS in 6m 54s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/559e33d1672e487fae28a5660ae01d36 : SUCCESS in 3m 09s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/5d0250b8417d47da9f13e487a543d8fc : SUCCESS in 8m 23s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/b7e85a6b1ba04e729f00a7165ef711f8 : SUCCESS in 6m 17s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/0b67a6ef4961469d8f36c49851cfa8df : SUCCESS in 6m 22s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/fdec3be6221240478c784422cacce388 : SUCCESS in 1h 34m 47s (non-voting)\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/90a390b011fd402eae616f026306601a : SUCCESS in 59m 57s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/c8140246f1a047f692fb14882bcad533 : SUCCESS in 1h 43m 53s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/96b35117f5e04039996c8cb9c40fd96d : SUCCESS in 51m 42s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/1b658f9d74b34c12a37fca813366d48f : SUCCESS in 1h 13m 32s\n- glance-s3-multistore https://zuul.opendev.org/t/openstack/build/88d8b0ea86694ce7b67fd06b4d1fbd75 : SUCCESS in 49m 03s (non-voting)\n- glance-multistore-add-location-do-secure-hash-false https://zuul.opendev.org/t/openstack/build/5f7f0f78a4f44fc18224b443164414ff : SUCCESS in 1h 37m 34s (non-voting)\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/5fc40582624e4169b3b77890d7daca37 : SUCCESS in 32m 49s\n- glance-rally-task-uwsgi https://zuul.opendev.org/t/openstack/build/c12e4531098a48e9845020e55b7f5e24 : SUCCESS in 41m 33s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"961a5ec3a1b2359db32ddbb82522462d8d0ff885","author":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"date":"2026-08-18 10:08:15.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"f10e066ce44c52cc43b54e10e931accce998a48f","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-18 10:29:51.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3a25f3d62ebf4acca4d87a2d9501f13a\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/ba93133a283f405893f147dcfb9101a0 : FAILURE in 4m 13s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/7ceedd794a664182ab06ced5eb144e07 : NODE_FAILURE Node(set) request 0a310cac92f84e9c8ab62ffaea181487 failed in 0s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/ca8cc9f8b590410eba50ce7d19b5b029 : FAILURE in 6m 40s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"f469ab69653cf764323778bad43bd4c24a0e59f0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-18 12:13:59.000000000","message":"Patch Set 3:\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/74467092e48e4a83935cd1b2bf644958\n\n- grenade https://zuul.opendev.org/t/openstack/build/71f41da9aaf440c3894655ccb3b90e34 : SUCCESS in 56m 14s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/7ee83488f6264bd8b06eafb753c720f5 : SUCCESS in 1h 02m 53s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/062aadd979144386acd7e7498342b3e4 : SUCCESS in 1h 31m 59s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/92aba549873a463cbc6d30707443d6fb : SUCCESS in 41m 18s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5ee9d2e53aea40ca9aaa20efd6425426 : SUCCESS in 4m 29s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/5fbb6a38d16841c2a86ae10826741358 : SUCCESS in 4m 16s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/d0e3854f0d844678bc4d87f12373e945 : SUCCESS in 4m 23s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/6c2116f8e7624309b0fc51d7a9ba36b1 : SUCCESS in 4m 19s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bf8bf65109964767a9cbd060fe7f9d4e : SUCCESS in 5m 15s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/4155f2364813439c90963e95c1e17c23 : SUCCESS in 8m 25s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/bce030d1073c43139b3abc85e36ae2a3 : SUCCESS in 6m 04s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/30ba47ac4eb245619879f4c8c3bea805 : SUCCESS in 6m 05s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/af246d36bcfb4e0296157b7f9aac4be4 : SUCCESS in 1h 43m 44s (non-voting)\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/3cd8c8e51d8041f69c9dd0e9f5179f46 : SUCCESS in 50m 20s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/3227718e58004ec094cdb6c85180a061 : SUCCESS in 1h 51m 05s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/7d0fa831a5fe4c39a92d72a3861d830c : SUCCESS in 55m 38s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/c25221103aaa474fb51841122b07ca46 : SUCCESS in 1h 00m 55s\n- glance-s3-multistore https://zuul.opendev.org/t/openstack/build/15572b1bf623497ca5786394c9fd09ae : SUCCESS in 45m 04s (non-voting)\n- glance-multistore-add-location-do-secure-hash-false https://zuul.opendev.org/t/openstack/build/aa837273369d46c9b55f7ed00ad8d4a4 : SUCCESS in 1h 20m 57s (non-voting)\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/85a667a02e964a1088d32ce3d361b388 : SUCCESS in 32m 11s\n- glance-rally-task-uwsgi https://zuul.opendev.org/t/openstack/build/d0d3732eb04d43eeb7998c986d33a7f6 : SUCCESS in 24m 45s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"a3fd4f8c106d8fb3cbb7c286886084b2cc6ef79b","author":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"date":"2026-08-19 05:01:19.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"10ac33880e1e18816bed2ff346b6f8e524ec7009","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2026-08-19 05:36:32.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"e5b1813b9898313383787dc8c953e184c796909f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-19 05:37:20.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"ae56873f268cf2c030e5bb41a5e20c7708cea189","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-19 07:07:25.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/dc38d474293f473a84d067fc3c150faa\n\n- grenade https://zuul.opendev.org/t/openstack/build/b22fe3ba8bc846658fe7724abdecf1d8 : SUCCESS in 29m 11s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/71ffe39045a149d58bd56ed7ab8184fc : SUCCESS in 1h 00m 45s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/15290ec12f494b5598d909c55e323ea8 : SUCCESS in 1h 24m 29s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/62f746585f1c43278db2cdf9f968eecd : SUCCESS in 1h 00m 38s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6ba414d40bfa4fe998c5f6175bbeaf53 : SUCCESS in 6m 00s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/6e69070ec9df44efb62af1139c3afcb0 : SUCCESS in 4m 00s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/94d96d7765584aa5b557ce137fbffac1 : SUCCESS in 2m 38s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/62801d708d7d4ee9acd22aaac7cba690 : SUCCESS in 5m 38s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/88fac4b0b3634eccadf5e0a53678c50e : SUCCESS in 3m 21s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/5452cbe98f5e453bb918160e3b7f9bcc : SUCCESS in 5m 14s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/6c4a484101f8485bb820110072fd60f2 : SUCCESS in 5m 32s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/7611e629c87046da9c59c979df0f9af6 : SUCCESS in 1h 15m 15s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/3125848db741407b8f096da7339522f8 : SUCCESS in 24m 15s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/efaa8b37c51640a890dea249202ac880 : SUCCESS in 45m 49s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/1b06f3d3e96d4e6e9c01ef66d0d5c2ed : SUCCESS in 15m 41s","accounts_in_message":[],"_revision_number":3},{"id":"a7ea622f9113c03696d9e81fb3a5406ebb64c9bd","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-19 07:07:25.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3},{"id":"9cffc4f8e8290d21377df5d6ee0513688ce4d111","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-19 07:09:31.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e318f6226590426d8dae77c1bd0ab25d\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1f5811b5991b419885f74c8ffad97ee9 : SUCCESS in 56s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"80c117809cbe10eb0eec41a2e3a3bae778316630","revisions":{"72fbe5d43e76b467e617ecff0a2aa6aa21ba030d":{"kind":"REWORK","_number":1,"created":"2026-04-06 10:16:25.000000000","uploader":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"ref":"refs/changes/15/983415/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/15/983415/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/15/983415/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/15/983415/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/15/983415/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/15/983415/1"}}},"commit":{"parents":[{"commit":"aa92eba418209da48849bc22fa2a8b5f84ecdbde","subject":"Add store weight config to glance-s3-multistore job","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/aa92eba418209da48849bc22fa2a8b5f84ecdbde"}]}],"author":{"name":"Pranali Deore","email":"pdeore@redhat.com","date":"2026-04-06 10:09:33.000000000","tz":0},"committer":{"name":"Pranali Deore","email":"pdeore@redhat.com","date":"2026-04-06 10:14:07.000000000","tz":0},"subject":"Set do_secure_hash\u003dFalse config to glance-s3-multistore job","message":"Set do_secure_hash\u003dFalse config to glance-s3-multistore job\n\nThe location API tests have two paths: one for do_secure_hash\u003dTrue\nand one for do_secure_hash\u003dFalse. The do_secure_hash\u003dTrue path is\nalready covered by tempest-integrated-storage-import, so configure\nglance-s3-multistore with do_secure_hash\u003dFalse to cover the\nremaining path without adding a new CI job.\n\nDepends-On: https://review.opendev.org/c/openstack/tempest/+/891026\nChange-Id: Ibfff59c2726cb8fcb8ef7672d4f174dadeebe28e\nSigned-off-by: Pranali Deore \u003cpdeore@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/72fbe5d43e76b467e617ecff0a2aa6aa21ba030d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/72fbe5d43e76b467e617ecff0a2aa6aa21ba030d"}]},"branch":"refs/heads/master"},"478faa739216fd28cf13c280282a897235f5f8ac":{"kind":"REWORK","_number":2,"created":"2026-07-14 07:46:43.000000000","uploader":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"ref":"refs/changes/15/983415/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/15/983415/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/15/983415/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/15/983415/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/15/983415/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/15/983415/2"}}},"commit":{"parents":[{"commit":"4d3aa425eaca1818ce9168c21fbed71ae971468b","subject":"Merge \"Remove tempest-pg-full\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/4d3aa425eaca1818ce9168c21fbed71ae971468b"}]}],"author":{"name":"Pranali Deore","email":"pdeore@redhat.com","date":"2026-04-06 10:09:33.000000000","tz":0},"committer":{"name":"Pranali Deore","email":"pdeore@redhat.com","date":"2026-07-14 07:45:40.000000000","tz":0},"subject":"Add glance-s3-multistore-do-secure-hash-false job","message":"Add glance-s3-multistore-do-secure-hash-false job\n\nAdd a dedicated non-voting multistore job with do_secure_hash\ndisabled in Glance and Tempest so Tempest add-location False\nAPI tests run:\nLocationImportTest.test_add_location_with_do_secure_hash_false and\nMultistoreLocationImportTest.test_add_location_with_do_secure_hash_false.\n\nDepends-On: https://review.opendev.org/c/openstack/tempest/+/891026\nChange-Id: Ibfff59c2726cb8fcb8ef7672d4f174dadeebe28e\nSigned-off-by: Pranali Deore \u003cpdeore@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/478faa739216fd28cf13c280282a897235f5f8ac"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/478faa739216fd28cf13c280282a897235f5f8ac"}]},"branch":"refs/heads/master"},"80c117809cbe10eb0eec41a2e3a3bae778316630":{"kind":"REWORK","_number":3,"created":"2026-07-28 05:52:32.000000000","uploader":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"ref":"refs/changes/15/983415/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/15/983415/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/15/983415/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/15/983415/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/15/983415/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/15/983415/3"}}},"commit":{"parents":[{"commit":"4d3aa425eaca1818ce9168c21fbed71ae971468b","subject":"Merge \"Remove tempest-pg-full\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/4d3aa425eaca1818ce9168c21fbed71ae971468b"}]}],"author":{"name":"Pranali Deore","email":"pdeore@redhat.com","date":"2026-04-06 10:09:33.000000000","tz":0},"committer":{"name":"Pranali Deore","email":"pdeore@redhat.com","date":"2026-07-28 05:52:32.000000000","tz":0},"subject":"Add glance-multistore-add-location-do-secure-hash-false job","message":"Add glance-multistore-add-location-do-secure-hash-false job\n\nAdd a dedicated non-voting multistore job with do_secure_hash\ndisabled in Glance and Tempest so Tempest add-location False\nAPI tests run:\nLocationImportTest.test_add_location_with_do_secure_hash_false and\nMultistoreLocationImportTest.test_add_location_with_do_secure_hash_false.\n\nDepends-On: https://review.opendev.org/c/openstack/tempest/+/891026\nChange-Id: Ibfff59c2726cb8fcb8ef7672d4f174dadeebe28e\nSigned-off-by: Pranali Deore \u003cpdeore@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/80c117809cbe10eb0eec41a2e3a3bae778316630"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/80c117809cbe10eb0eec41a2e3a3bae778316630"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{"label:Review-Priority\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
