)]}'
{"id":"openstack%2Fglance~983688","triplet_id":"openstack%2Fglance~stable%2F2025.1~I0eee75a07c2b48e003bce1be43731f5c4a863846","project":"openstack/glance","branch":"stable/2025.1","attention_set":{},"removed_from_attention_set":{"38762":{"account":{"_account_id":38762,"name":"Sahil Kumbhar","display_name":"sakumbha","email":"sakumbha@redhat.com","username":"sakumbha"},"last_update":"2026-04-30 01:47:29.000000000","reason":"Change was submitted"},"8122":{"account":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"last_update":"2026-04-30 01:35:54.000000000","reason":"\u003cGERRIT_ACCOUNT_8122\u003e replied on the change","reason_account":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"}},"37598":{"account":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"last_update":"2026-04-10 07:52:11.000000000","reason":"\u003cGERRIT_ACCOUNT_37598\u003e replied on the change","reason_account":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}}},"hashtags":[],"change_id":"I0eee75a07c2b48e003bce1be43731f5c4a863846","subject":"doc: Fix incorrect config section name in interoperable image import docs","status":"MERGED","created":"2026-04-08 09:36:39.000000000","updated":"2026-04-30 01:48:23.000000000","submitted":"2026-04-30 01:47:29.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"983688","cherry_pick_of_change":982234,"cherry_pick_of_patch_set":2,"meta_rev_id":"85293ac5eb6abc3dd429fd56df2df59a06f2cba6","_number":983688,"virtual_id_number":983688,"owner":{"_account_id":38762,"name":"Sahil Kumbhar","display_name":"sakumbha","email":"sakumbha@redhat.com","username":"sakumbha"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-04-30 01:47:29.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"all":[{"value":2,"date":"2026-04-30 01:35:54.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2026-04-10 07:52:11.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"all":[{"value":1,"date":"2026-04-30 01:35:54.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Top Priority / Holds Gate"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-04-08 09:49:10.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-04-09 14:13:15.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"state":"REVIEWER"},{"updated":"2026-04-29 23:05:05.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"state":"REVIEWER"}],"messages":[{"id":"82840ac8227bcb32849736f98e963484361b24d8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":38762,"name":"Sahil Kumbhar","display_name":"sakumbha","email":"sakumbha@redhat.com","username":"sakumbha"},"date":"2026-04-08 09:36:39.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"02d71a418754ef87ee880077466288a5d3057dd3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-08 09:49:10.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3487918a0fcd4e3f93a32cc59ae081ae\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6425eb67a4754703af0af415a3ac771b : SUCCESS in 5m 07s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b2c74897719c4fdeae10ddc2276f78f3 : SUCCESS in 3m 40s","accounts_in_message":[],"_revision_number":1},{"id":"6a4e3586ec12c8184927f3b1f0f973fb71ddd9e8","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-04-09 14:13:15.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"de1d5737313f23debbd8cb7fbfece13ed75ea9da","tag":"autogenerated:gerrit:abandon","author":{"_account_id":38762,"name":"Sahil Kumbhar","display_name":"sakumbha","email":"sakumbha@redhat.com","username":"sakumbha"},"date":"2026-04-10 06:51:15.000000000","message":"Abandoned\n\nShould be cherry picked form 2025.2","accounts_in_message":[],"_revision_number":1},{"id":"cc8cab9e941dc5a63957795f07828bf89916c602","tag":"autogenerated:gerrit:restore","author":{"_account_id":38762,"name":"Sahil Kumbhar","display_name":"sakumbha","email":"sakumbha@redhat.com","username":"sakumbha"},"date":"2026-04-10 07:24:13.000000000","message":"Restored","accounts_in_message":[],"_revision_number":1},{"id":"c73db754e714b1e6d04627c47c4315dfc55af18d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":38762,"name":"Sahil Kumbhar","display_name":"sakumbha","email":"sakumbha@redhat.com","username":"sakumbha"},"date":"2026-04-10 07:27:53.000000000","message":"Uploaded patch set 2: Patch Set 1 was rebased. Commit message was updated.\n\nOutdated Votes:\n* Code-Review+1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":2},{"id":"e91ffbbdb83e72187e259a35f447dd7b807f8465","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":38762,"name":"Sahil Kumbhar","display_name":"sakumbha","email":"sakumbha@redhat.com","username":"sakumbha"},"date":"2026-04-10 07:32:08.000000000","message":"Patch Set 3: Commit message was updated.","accounts_in_message":[],"_revision_number":3},{"id":"72f0de375253f1d23a9d91578540871501253db4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-10 07:38:18.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3d0186664eb444dfa91c6c2a05fb8a44\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/55353cce66a04ddcae5aeaba0bcff0be : SUCCESS in 4m 59s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a4cc25dfd502410bbf5f7da33d977001 : SUCCESS in 5m 29s","accounts_in_message":[],"_revision_number":3},{"id":"49a5b532ee2117ed6b0d9be0bcf59512b132c408","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-04-10 07:52:11.000000000","message":"Patch Set 3: Code-Review+1","accounts_in_message":[],"_revision_number":3},{"id":"fb48eaead6d8892abd8771f29133285b1d987935","author":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"date":"2026-04-30 01:35:54.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"7eb3a874e8eddb283223bce4eec6e066439eeb4f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-30 01:36:08.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"a4e6cb6a286169a0ff9329dc862cd4b1c8953ab6","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-30 01:47:29.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c4894299a3d44e70abe8fe5080ccd533\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/47b2d78ced4c4b6fb06b88cdce596d30 : SUCCESS in 4m 35s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fa2677b6a0424af1a2a4f5f4495d10af : SUCCESS in 5m 49s","accounts_in_message":[],"_revision_number":3},{"id":"8f7445706164e9c225f492a95a8a7b5187434b34","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-30 01:47:29.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3},{"id":"85293ac5eb6abc3dd429fd56df2df59a06f2cba6","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-30 01:48:23.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/96b51404f7a0481893f6d7a2e268db87\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/482a11cfb15d4831b753d0a6cc47fda2 : SUCCESS in 44s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"98849cb604bf6f354e5d7f1a69d591a60faef7e4","revisions":{"cb7d5541fe55e28aeb8bb40cb79b8714e6bc513f":{"kind":"REWORK","_number":1,"created":"2026-04-08 09:36:39.000000000","uploader":{"_account_id":38762,"name":"Sahil Kumbhar","display_name":"sakumbha","email":"sakumbha@redhat.com","username":"sakumbha"},"ref":"refs/changes/88/983688/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/88/983688/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/88/983688/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/88/983688/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/88/983688/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/88/983688/1"}}},"commit":{"parents":[{"commit":"bb0d59fff722dbccb5a3e7696eada3d8d372ba8d","subject":"Fix SSRF vulnerabilities in image import API","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/bb0d59fff722dbccb5a3e7696eada3d8d372ba8d"}]}],"author":{"name":"sakumbha","email":"sakumbha@redhat.com","date":"2026-03-26 12:21:15.000000000","tz":0},"committer":{"name":"Sahil Kumbhar","email":"sakumbha@redhat.com","date":"2026-04-08 09:36:39.000000000","tz":0},"subject":"doc: Fix incorrect config section name in interoperable image import docs","message":"doc: Fix incorrect config section name in interoperable image import docs\n\nThe documentation incorrectly refers to [DEFAULTS] instead of\n[DEFAULT] in glance-api.conf examples. This patch corrects the\nsection name in all occurrences.\n\nCloses-Bug: #2144985\nChange-Id: I0eee75a07c2b48e003bce1be43731f5c4a863846\nSigned-off-by: sakumbha \u003csakumbha@redhat.com\u003e\n(cherry picked from commit b5abd90e84bdfdc34c398568dffcdf26d035a50d)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/cb7d5541fe55e28aeb8bb40cb79b8714e6bc513f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/cb7d5541fe55e28aeb8bb40cb79b8714e6bc513f"}]},"branch":"refs/heads/stable/2025.1"},"7e14625784d10fba75c005dba247f39d483e7ac0":{"kind":"TRIVIAL_REBASE_WITH_MESSAGE_UPDATE","_number":2,"created":"2026-04-10 07:27:53.000000000","uploader":{"_account_id":38762,"name":"Sahil Kumbhar","display_name":"sakumbha","email":"sakumbha@redhat.com","username":"sakumbha"},"ref":"refs/changes/88/983688/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/88/983688/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/88/983688/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/88/983688/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/88/983688/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/88/983688/2"}}},"commit":{"parents":[{"commit":"2699710692feb745b4a2fd7fd94b63b41abef8e5","subject":"Merge \"Handle images detected as ISO+GPT\" into stable/2025.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/2699710692feb745b4a2fd7fd94b63b41abef8e5"}]}],"author":{"name":"sakumbha","email":"sakumbha@redhat.com","date":"2026-03-26 12:21:15.000000000","tz":0},"committer":{"name":"sakumbha","email":"sakumbha@redhat.com","date":"2026-04-10 07:24:28.000000000","tz":0},"subject":"doc: Fix incorrect config section","message":"doc: Fix incorrect config section\n\nFix incorrect config section name in interoperable image import docs\n\nThe documentation incorrectly refers to [DEFAULTS] instead of\n[DEFAULT] in glance-api.conf examples. This patch corrects the\nsection name in all occurrences.\n\nNote: fix the commit subject line\n\nCloses-Bug: #2144985\nChange-Id: I0eee75a07c2b48e003bce1be43731f5c4a863846\nSigned-off-by: sakumbha \u003csakumbha@redhat.com\u003e\n(cherry picked from commit b5abd90e84bdfdc34c398568dffcdf26d035a50d)\n(cherry picked from commit 3bc1e946ef799c61b9ff6523580cdc8bb5daca30)\n(cherry picked from commit 4e138a740c59b96151025bcef254182bd4d681b4)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/7e14625784d10fba75c005dba247f39d483e7ac0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/7e14625784d10fba75c005dba247f39d483e7ac0"}]},"branch":"refs/heads/stable/2025.1"},"98849cb604bf6f354e5d7f1a69d591a60faef7e4":{"kind":"NO_CODE_CHANGE","_number":3,"created":"2026-04-10 07:32:08.000000000","uploader":{"_account_id":38762,"name":"Sahil Kumbhar","display_name":"sakumbha","email":"sakumbha@redhat.com","username":"sakumbha"},"ref":"refs/changes/88/983688/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/88/983688/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/88/983688/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/88/983688/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/88/983688/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/88/983688/3"}}},"commit":{"parents":[{"commit":"2699710692feb745b4a2fd7fd94b63b41abef8e5","subject":"Merge \"Handle images detected as ISO+GPT\" into stable/2025.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/2699710692feb745b4a2fd7fd94b63b41abef8e5"}]}],"author":{"name":"sakumbha","email":"sakumbha@redhat.com","date":"2026-03-26 12:21:15.000000000","tz":0},"committer":{"name":"Sahil Kumbhar","email":"sakumbha@redhat.com","date":"2026-04-10 07:32:08.000000000","tz":0},"subject":"doc: Fix incorrect config section name in interoperable image import docs","message":"doc: Fix incorrect config section name in interoperable image import docs\n\nThe documentation incorrectly refers to [DEFAULTS] instead of\n[DEFAULT] in glance-api.conf examples. This patch corrects the\nsection name in all occurrences.\n\nCloses-Bug: #2144985\nChange-Id: I0eee75a07c2b48e003bce1be43731f5c4a863846\nSigned-off-by: sakumbha \u003csakumbha@redhat.com\u003e\n(cherry picked from commit b5abd90e84bdfdc34c398568dffcdf26d035a50d)\n(cherry picked from commit 4e138a740c59b96151025bcef254182bd4d681b4)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/98849cb604bf6f354e5d7f1a69d591a60faef7e4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/98849cb604bf6f354e5d7f1a69d591a60faef7e4"}]},"branch":"refs/heads/stable/2025.1","description":"Edit commit message"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{"label:Review-Priority\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
