)]}'
{"id":"openstack%2Fglance~984218","triplet_id":"openstack%2Fglance~master~I21f548117cfab5a84a58e644fb8a358b1c0c9815","project":"openstack/glance","branch":"master","attention_set":{},"removed_from_attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2026-08-04 16:37:14.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I21f548117cfab5a84a58e644fb8a358b1c0c9815","subject":"Drop redundant overrides","status":"MERGED","created":"2026-04-12 07:03:05.000000000","updated":"2026-08-04 16:38:34.000000000","submitted":"2026-08-04 16:37:14.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":2,"has_review_started":true,"submission_id":"984218","meta_rev_id":"603e21d564692cfe25790679304bb68bc6fbf5e6","_number":984218,"virtual_id_number":984218,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-08-04 16:37:14.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"all":[{"value":2,"date":"2026-08-04 14:10:39.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":2,"date":"2026-06-24 16:32:37.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"all":[{"value":1,"date":"2026-08-04 14:10:39.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Top Priority / Holds Gate"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-04-12 07:14:44.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2026-04-12 09:25:25.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-06-24 16:32:37.000000000","updated_by":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"reviewer":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"state":"REVIEWER"},{"updated":"2026-08-04 14:10:39.000000000","updated_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"reviewer":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"state":"REVIEWER"}],"messages":[{"id":"b5956bfbd5de0521763e7db851c37eb1536b2c10","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2026-04-12 07:03:05.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"106ba010c364f8dec1c6a7ace48babb5860503b2","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-12 07:14:44.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4749a109de124d809c25801e11d394c3\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/5208c31a41db4b6eb91a3274dd620352 : SUCCESS in 6m 27s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/c936e54c6c444e588d918f4f4cf1399c : SUCCESS in 5m 38s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/cb7177b013fa48308acafdaacbca578e : SUCCESS in 10m 01s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"1895027e32d227d50dab4ed7f501c6b2309e1cd3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-12 09:25:25.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/db84d25e24794eefbda24e203f152b05\n\n- grenade https://zuul.opendev.org/t/openstack/build/46d2125e2fd243078c866613ef63c693 : SUCCESS in 33m 15s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/b61942a9c1264355a8ad24830df4f8d0 : SUCCESS in 1h 03m 15s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/188c954bdd7840f2811ad86d34f3a32c : SUCCESS in 1h 25m 08s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/0831a1e35bfb4fb49afca0c5a0710715 : SUCCESS in 1h 02m 28s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f06e61c2f90c407baef5c66d66f8cb47 : SUCCESS in 3m 52s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/8de817cf778e4448a3c432a4ffa1dc33 : SUCCESS in 5m 37s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/ef128d798081428e9d8309193467d91c : SUCCESS in 3m 07s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/06f40baebef04396921dac3c2dccd8fb : SUCCESS in 4m 36s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a33a716f1f9641089075523602289726 : SUCCESS in 3m 43s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/70d75bb6729d4d93917ff0fc70faf16b : SUCCESS in 4m 27s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/75106a259de04990ada923870e664dc3 : SUCCESS in 5m 05s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/c61bb24dffd8439bbb8ffa26552b73cd : SUCCESS in 1h 41m 38s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/184b9bf7dfea462e8d2d2f0f455967c9 : SUCCESS in 2h 20m 22s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/d07860d2c35c40b9b6661ac3d61a1a63 : SUCCESS in 1h 04m 12s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/993eaf4e9c404af8b65f138d0db6033e : SUCCESS in 1h 46m 56s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/cd12a888a643423a8f5fc64127560d3e : SUCCESS in 54m 41s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/7dea2d9946434af9a0d149086051ed8d : SUCCESS in 1h 04m 14s\n- glance-grenade-centralized-cache https://zuul.opendev.org/t/openstack/build/367be89989354887ae1ef8e1c8c6cde7 : FAILURE in 17m 31s (non-voting)\n- glance-s3-multistore https://zuul.opendev.org/t/openstack/build/d1df435beb844886a824d7bb4b02bbe9 : SUCCESS in 1h 40m 21s (non-voting)\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/c8e37063713d4bfe94aeddc1232dc45c : SUCCESS in 36m 12s","accounts_in_message":[],"_revision_number":1},{"id":"ccb8f89dadea633f2c5d8407d3b5c73ba43459d8","author":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"date":"2026-06-24 16:32:37.000000000","message":"Patch Set 1: Code-Review+2\n\n(2 comments)","accounts_in_message":[],"_revision_number":1},{"id":"48a48bc417029914bb464001e380933435e26387","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2026-08-04 14:10:39.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"2cb9c9cde6140b3a4d8b224e44ca299643b26af8","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-04 14:11:18.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"f609cc467ee43eab8107e520e600e3a9c410fa11","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-04 16:37:14.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ca8288011d9a4bae9c0dc2a8143fceea\n\n- grenade https://zuul.opendev.org/t/openstack/build/c55a010c4b174aab911779e34aaa5a22 : SUCCESS in 1h 04m 11s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/3bacc801df7d440a914f5285ef6055f9 : SUCCESS in 1h 08m 01s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/6151cf836d82402a84e74a0b0504464c : SUCCESS in 1h 39m 15s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/b50dd1e7b58642029c71736ecda4aa03 : SUCCESS in 1h 05m 30s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b2765f894c274f4a94feb24eb6b24429 : SUCCESS in 5m 23s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/9bcb7de1b4f74ccda9199358bd2ff7e9 : SUCCESS in 5m 16s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/2a67a0af79404c9ab36f8ac00c6b33e2 : SUCCESS in 5m 21s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b1a0a981db7442928b0597dad5a67d1c : SUCCESS in 6m 55s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/410d17fdd607459db6318fb8f9c09e2f : SUCCESS in 6m 52s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/ab0ecfe08bda481b9d0a87f3f981d932 : SUCCESS in 8m 44s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/14b72b10a3bc4b119a8f8249ec3a1a10 : SUCCESS in 6m 47s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/22d9e3fbf1ec40f48257c294e18f7bb7 : SUCCESS in 1h 03m 35s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/d744beae73aa4ceebc7c904e844b5f72 : SUCCESS in 56m 11s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/2bb67b102e1b4bfaa9ac8215cbaeff91 : SUCCESS in 1h 26m 48s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/0581de079d5447d099b3c8e12569bf30 : SUCCESS in 33m 09s","accounts_in_message":[],"_revision_number":1},{"id":"f23ecc289461fc5e4cc7f8b7b080cccd965764fe","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-04 16:37:14.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"603e21d564692cfe25790679304bb68bc6fbf5e6","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-04 16:38:34.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5bb50169959f4d72befbe201c16a70d6\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ab145ddcb0f447bd8f38ed94d0372263 : SUCCESS in 59s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"95efea51bae50019ec2251f39a34e6ea5946b6b7","revisions":{"95efea51bae50019ec2251f39a34e6ea5946b6b7":{"kind":"REWORK","_number":1,"created":"2026-04-12 07:03:05.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/18/984218/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/18/984218/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/18/984218/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/18/984218/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/18/984218/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/18/984218/1"}}},"commit":{"parents":[{"commit":"3f8cce1d7415d8358a9679c585ce3db32c3fde8e","subject":"Merge \"Use consistent program name\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/3f8cce1d7415d8358a9679c585ce3db32c3fde8e"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-04-12 07:00:03.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-04-12 07:01:32.000000000","tz":540},"subject":"Drop redundant overrides","message":"Drop redundant overrides\n\nThe role argument is added by the original to_dict by default.\n\nAlso use the native interface for from_dict, with FROM_DICT_EXTRA_KEYS,\nto support the usage with kwargs.\n\nChange-Id: I21f548117cfab5a84a58e644fb8a358b1c0c9815\nSigned-off-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/95efea51bae50019ec2251f39a34e6ea5946b6b7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/95efea51bae50019ec2251f39a34e6ea5946b6b7"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{"label:Review-Priority\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
