)]}'
{"id":"openstack%2Fglance~984363","triplet_id":"openstack%2Fglance~master~Id1efa02e648c7d53bf699482670b56e1458e813a","project":"openstack/glance","branch":"master","topic":"eventlet-removal","attention_set":{},"removed_from_attention_set":{"19138":{"account":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"last_update":"2026-05-12 14:21:56.000000000","reason":"\u003cGERRIT_ACCOUNT_19138\u003e replied on the change","reason_account":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"}},"9303":{"account":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"last_update":"2026-05-12 14:22:01.000000000","reason":"Change was abandoned"},"8122":{"account":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"last_update":"2026-05-12 14:22:01.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"Id1efa02e648c7d53bf699482670b56e1458e813a","subject":"Use native threads in glance-cache-prefetcher","status":"ABANDONED","created":"2026-04-13 10:58:13.000000000","updated":"2026-05-12 14:22:01.000000000","total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"42ddbd14b5e0d3a74775c055294d86d38455bd47","_number":984363,"virtual_id_number":984363,"owner":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"tag":"autogenerated:zuul:check","value":1,"date":"2026-04-13 14:12:15.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"disliked":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"all":[{"value":-1,"date":"2026-05-08 06:27:54.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":-1,"default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Top Priority / Holds Gate"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-04-13 14:12:15.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-05-08 05:59:48.000000000","updated_by":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"reviewer":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"state":"REVIEWER"},{"updated":"2026-05-08 05:59:48.000000000","updated_by":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"reviewer":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"state":"REVIEWER"}],"messages":[{"id":"6583e07b8cffc76ad4b0bd38b7cade0685550d37","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"date":"2026-04-13 10:58:13.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"b94dcadb839a18f3c1599c28cf2839ce529bc1bb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-13 14:12:15.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9dc0b2d17fdb48019cbf15f0dd1bf9f0\n\n- grenade https://zuul.opendev.org/t/openstack/build/08d2fb1597d7421fbf277e3e1903fc7f : SUCCESS in 1h 06m 06s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/c46d24e6460e474a876b1aa52d2171ed : SUCCESS in 55m 36s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/2e318db8755447228ceba115ae9105a7 : SUCCESS in 1h 40m 24s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/b51a4b57a0e14826847a22fcbf8a9e0d : SUCCESS in 55m 25s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/da5635da2fa240858cfe532837a2f99f : SUCCESS in 5m 07s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/7ce43126e9714ae1b90b320bd6662e11 : SUCCESS in 5m 14s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/7efd65a2e214450690182f186828d569 : SUCCESS in 6m 05s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/39266835e9434a47b754497ffaacbffb : SUCCESS in 8m 09s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f4006f360cb846899e34408cfcfe2582 : SUCCESS in 6m 45s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/a479b5fe78584a3081d5d66a7a91c3ee : SUCCESS in 7m 32s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/d87d2a281c2b4f429e9c2384e23f193c : SUCCESS in 8m 48s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/c64ba758cceb45c1b5d2d9fd739bcf80 : SUCCESS in 2h 00m 02s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/6d0db51f23a04836bbc0f024773eb3fc : SUCCESS in 2h 12m 40s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/7b09f4966dbf4e91ac7c5d0c612bee33 : SUCCESS in 2h 14m 23s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/4bafa5822b5a4d6face427c69eaf4877 : SUCCESS in 2h 20m 19s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/dfc15c4b916243c7885eeb6aaef4fa6e : SUCCESS in 1h 31m 21s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/0e217100a56c4834ae255a51f08f2f64 : SUCCESS in 1h 32m 44s\n- glance-grenade-centralized-cache https://zuul.opendev.org/t/openstack/build/76a0338bf43346f78886474e8cfb409e : FAILURE in 10m 22s (non-voting)\n- glance-s3-multistore https://zuul.opendev.org/t/openstack/build/c3c4cd25810543028e1a8b388a05bdd4 : FAILURE in 1h 59m 27s (non-voting)\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/eb16705a01cc42c186721e32bff8f400 : SUCCESS in 36m 06s","accounts_in_message":[],"_revision_number":1},{"id":"a3d3a85e53004a2c178b1556d45f6c90286861a1","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-13 16:45:48.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7c4e8d7e37da43899bcf843cb1abec04\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/6cb1b45931274ce68af271412aa711c6 : SUCCESS in 6m 38s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/6cf90053d3b14c39af901acdebe39fc1 : SUCCESS in 6m 35s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/4dbdbf607f01462aacdc990651bf78e5 : SUCCESS in 11m 10s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"5baec85552cb57f5521e06fccf87a2fac4a5bb3c","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"date":"2026-04-24 14:41:46.000000000","message":"Topic set to eventlet-removal","accounts_in_message":[],"_revision_number":1},{"id":"9dcbbb2335a1364a52c8e7ac440d04e07412cedd","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2026-05-08 06:27:54.000000000","message":"Patch Set 1: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"510361543bc8b5af1f1308716c224d0ada331900","author":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"date":"2026-05-12 14:21:56.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"42ddbd14b5e0d3a74775c055294d86d38455bd47","tag":"autogenerated:gerrit:abandon","author":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"date":"2026-05-12 14:22:01.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"d5d95f0ed4c4fe677bb46d0c54d30d6094f8c2e9","revisions":{"d5d95f0ed4c4fe677bb46d0c54d30d6094f8c2e9":{"kind":"REWORK","_number":1,"created":"2026-04-13 10:58:13.000000000","uploader":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"ref":"refs/changes/63/984363/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/63/984363/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/63/984363/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/63/984363/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/63/984363/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/63/984363/1"}}},"commit":{"parents":[{"commit":"3f8cce1d7415d8358a9679c585ce3db32c3fde8e","subject":"Merge \"Use consistent program name\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/3f8cce1d7415d8358a9679c585ce3db32c3fde8e"}]}],"author":{"name":"Pranali Deore","email":"pdeore@redhat.com","date":"2026-04-13 10:48:29.000000000","tz":0},"committer":{"name":"Pranali Deore","email":"pdeore@redhat.com","date":"2026-04-13 10:56:54.000000000","tz":0},"subject":"Use native threads in glance-cache-prefetcher","message":"Use native threads in glance-cache-prefetcher\n\nThe prefetcher CLI does not run under eventlet, setting\nthe threadpool model to native and added a unit test\nfor the same.\n\nChange-Id: Id1efa02e648c7d53bf699482670b56e1458e813a\nSigned-off-by: Pranali Deore \u003cpdeore@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/d5d95f0ed4c4fe677bb46d0c54d30d6094f8c2e9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/d5d95f0ed4c4fe677bb46d0c54d30d6094f8c2e9"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{"label:Review-Priority\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
