)]}'
{"id":"openstack%2Fglance~984938","triplet_id":"openstack%2Fglance~master~Iaea848462fe550cff995286e24aa11b86702ee88","project":"openstack/glance","branch":"master","attention_set":{},"removed_from_attention_set":{"19138":{"account":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"last_update":"2026-04-16 17:09:07.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Iaea848462fe550cff995286e24aa11b86702ee88","subject":"Replace eventlet.sleep with time.sleep in tasks API test","status":"MERGED","created":"2026-04-16 13:36:17.000000000","updated":"2026-04-16 17:10:30.000000000","submitted":"2026-04-16 17:09:07.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"984938","meta_rev_id":"cc3c6862188f504d26d3989463b4ca2c5f43a856","_number":984938,"virtual_id_number":984938,"owner":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-04-16 17:09:07.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"all":[{"value":2,"date":"2026-04-16 14:05:02.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":2,"date":"2026-04-16 13:55:02.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"all":[{"value":1,"date":"2026-04-16 14:05:02.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Top Priority / Holds Gate"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-04-16 13:55:02.000000000","updated_by":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"reviewer":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"state":"REVIEWER"},{"updated":"2026-04-16 14:02:04.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2026-04-16 14:05:02.000000000","updated_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"reviewer":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"state":"REVIEWER"},{"updated":"2026-04-16 14:40:47.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"31867955ba765702e010287b090da44336b6ff87","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"date":"2026-04-16 13:36:17.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"c9b82efa9cf147c34dbf54d346218c7a7fd1d141","author":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"date":"2026-04-16 13:55:02.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"b11b1a005ba904013ee8c73ebbbe4d19bfd33297","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-16 14:02:04.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/47cdf7edc176463e867662a36b912579\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/04e324bd5d074da486678a59dfdcb322 : SUCCESS in 6m 35s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/303a9b0b520e4cbebfb4b062c5659e80 : SUCCESS in 5m 47s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/4532db3092e04f9384232a514d7a876e : SUCCESS in 8m 14s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"c28042079d7ff595a599897b158121bf2b9bdd33","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2026-04-16 14:05:02.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"355aa20b721a1ed1126e877f32c29567055a46e7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-16 14:40:47.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/7cd58001b7774dd7bbcab5a2de623bc0\n\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/2d72fc503d534cb38348c9f74464361f : SUCCESS in 31m 00s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/629dac0a98444574a3ff510e935d13f2 : SUCCESS in 58m 29s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/26ab5e258cf046b99d29b0d6d48cf93d : SUCCESS in 4m 59s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/e0fba0c4c25046b78b22b1feb1930e66 : SUCCESS in 4m 48s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/7a90ce9e5d65492c927550d78284f883 : SUCCESS in 2m 48s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/b6c8973f720045d883076186a7540548 : SUCCESS in 8m 47s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4a89241ebd46490b81aa5e904c1e6442 : SUCCESS in 6m 40s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/111c219eda954614bb2abd7477c33fad : SUCCESS in 7m 28s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/f364a62ad93a44328fb377e0cde722d4 : SUCCESS in 7m 03s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/e05ab73476284d4db59b22c9556fb38d : FAILURE in 28m 58s","accounts_in_message":[],"_revision_number":1},{"id":"afd63609a191b1124823e3d20eac3f24724290eb","author":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"date":"2026-04-16 15:04:44.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"b85cda910237985d9455436164c3efd9d36120e6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-16 16:22:44.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/fb87da308cd3469c83695d019e41423e\n\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/6a993f9dea4742e48701f0d473f08ba9 : SUCCESS in 1h 05m 39s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/7fa1c643613b40b99598123c8407ae98 : SUCCESS in 1h 11m 03s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b1a9af18f05b4cf58bcc32865568388f : SUCCESS in 3m 29s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/febe3fe2b6844b3cb1b43140902872db : SUCCESS in 4m 54s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/a019a6e9c7cb4c0faffa0ece249bb9c5 : SUCCESS in 3m 38s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/d26157c302534a0d9adc80e35c1fb2a9 : SUCCESS in 5m 49s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f224368e81e94098b041a16d87245e4b : SUCCESS in 7m 31s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/2c50e696dde242aebf3e780366d1df2a : SUCCESS in 6m 27s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/fa04aba58319474c9b2ec87fab96379d : SUCCESS in 6m 49s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/d9ceb254ee6f449396a6bc07f6f5d7a8 : SUCCESS in 38m 46s","accounts_in_message":[],"_revision_number":1},{"id":"674af2faaca4346853ff692af24bf28379cdf9e5","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-16 16:23:14.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"0ece615970696ff49ed414eacb20396088530c43","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-16 16:53:58.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c1b43f7a8870450c84aac86c5c08af52\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/40d6040583364be4bd44138d22919a99 : SUCCESS in 5m 05s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/01bfa6d67da04f21a672f0f4fed8dcfd : SUCCESS in 5m 17s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/699cf24cf61941d4a9c77e856694106e : SUCCESS in 8m 12s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"e2f33c1a8209ed6a69002265fab0dbf1c84daae8","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-16 17:09:07.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/72d223c76a77492eb33d1142c85f499d\n\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/5c0108455b1a49ef9b06470a2475552d : SUCCESS in 34m 09s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/2cad16c6b57044acbf117122b53b5d11 : SUCCESS in 45m 30s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/be0e0c64c5034727b669531fcc3425dd : SUCCESS in 4m 17s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/e3eca4c669204f568e68f98dfe6817e1 : SUCCESS in 4m 11s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/2fa08cc2bc7449739fe1ede41b35d682 : SUCCESS in 5m 44s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dab1e6ec60224409b82f5867481f4230 : SUCCESS in 5m 53s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/916a2439657e4d92874f44449f71d1fb : SUCCESS in 7m 49s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/c6b2bbb5a49d4330909386f1704803a2 : SUCCESS in 7m 32s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/e7510237d9b84803bd3a57ee3a20c702 : SUCCESS in 37m 05s","accounts_in_message":[],"_revision_number":1},{"id":"cf2f123a833685b8188bad726a0199b53dcfde8f","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-16 17:09:07.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"cc3c6862188f504d26d3989463b4ca2c5f43a856","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-16 17:10:30.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c9bf0d7ce55b4a4d9e4e56f7c60c1fff\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2689c7139d714b238009115f7bf2a5b4 : SUCCESS in 48s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"03dbf8eac63a35aa0207eba21a3c33e0aa51bb18","revisions":{"03dbf8eac63a35aa0207eba21a3c33e0aa51bb18":{"kind":"REWORK","_number":1,"created":"2026-04-16 13:36:17.000000000","uploader":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"ref":"refs/changes/38/984938/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/38/984938/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/38/984938/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/38/984938/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/38/984938/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/38/984938/1"}}},"commit":{"parents":[{"commit":"8b6f60528642daf1cd796f334d2d9df71c029645","subject":"Merge \"doc: update RBD storage backend auth caps to use Ceph profiles\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/8b6f60528642daf1cd796f334d2d9df71c029645"}]}],"author":{"name":"Pranali Deore","email":"pdeore@redhat.com","date":"2026-04-16 13:32:08.000000000","tz":0},"committer":{"name":"Pranali Deore","email":"pdeore@redhat.com","date":"2026-04-16 13:32:36.000000000","tz":0},"subject":"Replace eventlet.sleep with time.sleep in tasks API test","message":"Replace eventlet.sleep with time.sleep in tasks API test\n\nChange-Id: Iaea848462fe550cff995286e24aa11b86702ee88\nSigned-off-by: Pranali Deore \u003cpdeore@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/03dbf8eac63a35aa0207eba21a3c33e0aa51bb18"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/03dbf8eac63a35aa0207eba21a3c33e0aa51bb18"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{"label:Review-Priority\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
