)]}'
{"id":"openstack%2Fglance~986093","triplet_id":"openstack%2Fglance~master~Iae258108284b84412c6e52a3fbade1213547c480","project":"openstack/glance","branch":"master","topic":"eventlet-removal","attention_set":{},"removed_from_attention_set":{"19138":{"account":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"last_update":"2026-06-10 04:07:02.000000000","reason":"\u003cGERRIT_ACCOUNT_19138\u003e replied on the change","reason_account":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"}},"9303":{"account":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"last_update":"2026-05-20 08:34:16.000000000","reason":"\u003cGERRIT_ACCOUNT_9303\u003e replied on the change","reason_account":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"}},"8122":{"account":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"last_update":"2026-06-10 08:29:29.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Iae258108284b84412c6e52a3fbade1213547c480","subject":"Add non-voting Zuul Rally job for Glance on uWSGI","status":"MERGED","created":"2026-04-24 14:18:12.000000000","updated":"2026-06-10 08:31:19.000000000","submitted":"2026-06-10 08:29:29.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":11,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"986093-eventlet-removal","meta_rev_id":"2ec3475fa58f482f7cf6dbb3cfe5112604e04c4c","_number":986093,"virtual_id_number":986093,"owner":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-06-10 08:29:29.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"all":[{"value":2,"date":"2026-06-09 11:48:36.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":2,"date":"2026-06-08 11:02:57.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"all":[{"value":1,"date":"2026-06-09 11:48:36.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Top Priority / Holds Gate"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-04-24 16:45:20.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-05-11 14:40:45.000000000","updated_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"reviewer":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"state":"REVIEWER"},{"updated":"2026-06-02 15:30:43.000000000","updated_by":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"reviewer":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"state":"REVIEWER"}],"messages":[{"id":"b5b1f97f553e6cc0df0925911f081cf0b0f78a5b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"date":"2026-04-24 14:18:12.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"5abb041ad59169406106427e55032e2eb1fffb2c","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"date":"2026-04-24 14:41:24.000000000","message":"Topic set to eventlet-removal","accounts_in_message":[],"_revision_number":1},{"id":"7f85e6f1cfdad8a0933e94dd6cf0bde1fbe40519","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-24 16:45:20.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/0a1196d568de460f8d7cc85d9c8f1336\n\n- grenade https://zuul.opendev.org/t/openstack/build/c78f157113d84dfbb79b848654294a12 : SUCCESS in 1h 06m 09s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/05a748fbad5c4fe8a2ff0e75714396b7 : SUCCESS in 1h 06m 48s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/b5700937c0834143999eb0ba17052e2e : SUCCESS in 1h 48m 48s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/787e94dc55494727acf04c4382b66c38 : SUCCESS in 38m 38s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f99f0c3e8cc440c1aa6b0d2dcd45fc81 : SUCCESS in 4m 32s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/a0d13803ca694d6c837717e48798a86e : SUCCESS in 7m 21s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/acfeedde2fe1432f891a3d1c8735ee48 : SUCCESS in 2m 57s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/3fe2347af0c7446fb8b2aa655a1a949f : SUCCESS in 9m 39s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/920a91f5ad864b3999891420455de472 : SUCCESS in 6m 29s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/b1e02df01a1a45bbb491b0810da0619f : SUCCESS in 5m 53s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/e2332473a4c74f0d9d198f0f10ed5abc : SUCCESS in 5m 28s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/7fb081d77f3a4bdbaecfc5c588371818 : SUCCESS in 1h 56m 14s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/7b07e2c57fe54a07aea4a5d88b1784f2 : SUCCESS in 1h 41m 17s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/f46bfbedfc6849b0992e783185bc598d : SUCCESS in 2h 19m 22s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/fc34e278e74e4d67bf9cbef5b066a564 : SUCCESS in 2h 04m 32s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/4c8483d7778b45fda6ec39642600de5b : SUCCESS in 55m 59s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/4fdc20efa9c549039a9a3754a1679ea0 : FAILURE in 16m 22s\n- glance-grenade-centralized-cache https://zuul.opendev.org/t/openstack/build/9e8882d4a1924900b0597414c7e53afb : FAILURE in 14m 22s (non-voting)\n- glance-s3-multistore https://zuul.opendev.org/t/openstack/build/be9810b42be247b18cc5024a04e869fd : FAILURE in 1h 39m 26s (non-voting)\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/acd716ab04a34886a28e96bb85ec8d90 : SUCCESS in 43m 00s\n- glance-rally-task-uwsgi https://zuul.opendev.org/t/openstack/build/78c730debde849a7ad8140983684fc78 : FAILURE in 31m 10s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"dccf4c7ea6e9b3f09ac4d8d85abf7396d0d3a6c1","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-24 20:58:50.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/15de61cc4c1e4e62bcc59a3ae8f27081\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/257be8956d994dcd8dc4c3a8ed84b953 : SUCCESS in 6m 49s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/5845d3f6178b44bcb50fdf982f6ccab5 : SUCCESS in 6m 44s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/cdc8706caf1b489fadce96d3e9917bb7 : SUCCESS in 7m 56s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"b9004a077a461a4bcfb91724f9e3cd30735aedc3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"date":"2026-05-11 14:11:24.000000000","message":"Patch Set 2: Commit message was updated.\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":2},{"id":"b4bfeedf0dfd84240c81a8e10a83fe5407c5fb64","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"date":"2026-05-11 14:16:38.000000000","message":"Uploaded patch set 3: Patch Set 2 was rebased. Commit message was updated.","accounts_in_message":[],"_revision_number":3},{"id":"30505f3ea613128f862af42ac2404acb63255ec1","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-11 14:27:30.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2087b7c8bd7d44a6940b44bf8c6cce0b\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/57bb8587421749d29f0ffff6dced239b : FAILURE in 5m 36s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/478008acabac45fa82501431c219cd8f : SUCCESS in 7m 29s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/77ccfeed0dfb438ca558cf4caa7a67f2 : SUCCESS in 8m 41s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"f133b33fa4d253939647d964cc0ccd75c942739e","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2026-05-11 14:40:45.000000000","message":"Patch Set 3: Code-Review-1\n\n(2 comments)","accounts_in_message":[],"_revision_number":3},{"id":"2f64b601bc2af33c9415b121b51e513f1c6fc7b4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-11 16:54:29.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/03b1d6b554dd46e3817446499e1a0d0c\n\n- grenade https://zuul.opendev.org/t/openstack/build/c785f0ecfce74aeaac40100f91ef1a82 : SUCCESS in 1h 10m 53s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/ecef7a91585143459f2382cd0897a007 : SUCCESS in 55m 52s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/6dffe164584a4980980c5903cb4e154c : SUCCESS in 1h 38m 11s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/1fbba37001454aaeba275b26dc81a59e : SUCCESS in 1h 10m 03s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2479f18d32b146e399ea7222fd082d6f : SUCCESS in 3m 17s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/478d4225403e43218bf728ef903bea3a : SUCCESS in 5m 10s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/51ef1bd79133477c91743aabcfdcfac7 : SUCCESS in 5m 31s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/aae458784774462fb9a4692674ca636a : SUCCESS in 8m 22s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/57b1a7aa63fb4c63adf3408898f58c09 : SUCCESS in 5m 20s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/650838c280c4421c8588a531dc61d22b : SUCCESS in 9m 36s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/22fba9e18425448f9d7772762d83295f : SUCCESS in 5m 59s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/13315002b58145088bee60b2bdc7a90e : SUCCESS in 1h 56m 55s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/ca65798bc99b471e94aa30b7baad41c2 : SUCCESS in 2h 02m 13s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/df25afab62aa413c97d030917662d86c : SUCCESS in 1h 57m 01s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/ac40ca5d7db44fe6b543463ed4991f28 : SUCCESS in 2h 30m 37s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/e76a456f43964ae7979df0812de2a3f6 : SUCCESS in 55m 43s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/84063efbcc754fc79fa735fe21d8ea57 : SUCCESS in 1h 52m 47s\n- glance-grenade-centralized-cache https://zuul.opendev.org/t/openstack/build/ff9bde5d3ba94149a66f767a51a445d5 : FAILURE in 19m 12s (non-voting)\n- glance-s3-multistore https://zuul.opendev.org/t/openstack/build/7b0a04fabf724f30b8580f143ed66d39 : SUCCESS in 59m 35s (non-voting)\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/ca2b33adcb784d58bb7f44f62f6fbfc3 : SUCCESS in 23m 28s\n- glance-rally-task-uwsgi https://zuul.opendev.org/t/openstack/build/df906ade49a449d281b65ad7f1d7fa65 : FAILURE in 27m 36s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"4ae6a9cdda00227927eb96102c02b726daaf351e","author":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"date":"2026-05-12 09:37:39.000000000","message":"Patch Set 3:\n\n(2 comments)","accounts_in_message":[],"_revision_number":3},{"id":"88d4cdc7d708c8c5162d61b52ca7823fd18356dc","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"date":"2026-05-12 09:37:57.000000000","message":"Uploaded patch set 4: Patch Set 3 was rebased. Commit message was updated.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":4},{"id":"694683174b0ec06273c73e57b07e31f0eee82679","author":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"date":"2026-05-12 09:38:49.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"549655a39ac5ce56b057827775dc904c3caddadd","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-12 10:34:59.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f45a399cb0d94561835835efd6dc1a7a\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/a6e78927ce7a424b953d2467ae9bb257 : FAILURE in 5m 44s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/c1df0c3a9ac44dc09189cf83a2fad3aa : SUCCESS in 7m 04s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/231b3e9f4225410f94098dbad6f77f16 : SUCCESS in 9m 42s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"8154430d793e95aa4fa9b8703cc46981d33b1e82","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"date":"2026-05-12 10:59:35.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"277626ee38a6f6a4eec82e1c4006130f2a00dadf","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-12 11:28:19.000000000","message":"Patch Set 5:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f9868eb87b5e473481deacc07a7fa612\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/fb03d6a48a3348dd9cf86bd23af957e5 : FAILURE in 4m 53s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/034513c0fb674989b32f596ee12346b8 : SUCCESS in 7m 17s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/917992e58f394821b82310ee0b755443 : SUCCESS in 8m 07s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"593cc3fe6514cb149c88b1fb86ebebf0ff638579","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-12 13:47:52.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4bf7328702cb4d078e2aab5db2a4238f\n\n- grenade https://zuul.opendev.org/t/openstack/build/50fa5a664f464c57a2702173dc3e8b83 : SUCCESS in 1h 03m 21s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/7300d1abf58d45ae81941bb628800eea : SUCCESS in 59m 13s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/190b9c7b98b84869807fbbc313a3991c : SUCCESS in 1h 37m 44s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/a0c77fd1285c40bd992ac35848e35a20 : SUCCESS in 1h 07m 18s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/89eaf8f5c9bb4ee49a9407bba11d6073 : SUCCESS in 5m 11s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/c289e855494e40fd9772923d3edf4fac : SUCCESS in 5m 25s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/e621ac47ea9c4e91ac4a852fdceeb5d6 : SUCCESS in 7m 13s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/472867b75962446289c32c725a768cb1 : SUCCESS in 8m 13s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9129afcbc4414c75b75ad9320241f28b : SUCCESS in 6m 53s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/de60a1dc6623454083a9b14cf717b1e9 : SUCCESS in 7m 14s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/cad3cf11e67448128f0616496e085970 : SUCCESS in 6m 25s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/dc2f850c30584bf19f555f1b46ab38e3 : SUCCESS in 1h 39m 40s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/8391f8a3bf474b3a801f3948d84c2ad9 : SUCCESS in 2h 05m 26s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/5c005c6910fa4656bdd68b115f69e8af : SUCCESS in 2h 27m 48s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/b3d9baf9b11a472190e9f4b7c4cc9fca : SUCCESS in 2h 36m 44s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/8abd33d4a88348f0a84cec13c666bfe4 : SUCCESS in 47m 43s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/7a5ee36b636842f1a975379af1c620c8 : SUCCESS in 1h 49m 52s\n- glance-grenade-centralized-cache https://zuul.opendev.org/t/openstack/build/1a4f33fb8d5d4e829c10a562de460194 : FAILURE in 16m 13s (non-voting)\n- glance-s3-multistore https://zuul.opendev.org/t/openstack/build/dd70e1ca1b6c4d06b9793e0a16acc42b : SUCCESS in 1h 38m 57s (non-voting)\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/ad46208418ac4387a76418f8e8607cb3 : SUCCESS in 46m 32s\n- glance-rally-task-uwsgi https://zuul.opendev.org/t/openstack/build/7147483952ae4b1f8dcf62d21d2540ac : FAILURE in 36m 49s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"7737de281a319327ea8c73a1ac0204a7f4308286","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2026-05-14 14:43:13.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"508652297f0f511226d2107b2f6978f8a4038883","author":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"date":"2026-05-18 11:29:16.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"ce2834e792dc6c467f4b4440b4e977b60384871e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"date":"2026-05-18 11:49:27.000000000","message":"Patch Set 6: Commit message was updated.\n\nOutdated Votes:\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":6},{"id":"091680bb5ab2222bbeb4f23348dceebcde060546","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-18 12:04:41.000000000","message":"Patch Set 6:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f6c17325e76141dfa0eff7d6e4805091\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/67b3e541fc384762b29239c116640a15 : SUCCESS in 7m 19s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/d57369694aee430f86869327426b8d89 : SUCCESS in 7m 05s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/3aafb5f17c104e2d94a6178e68022ab5 : SUCCESS in 12m 18s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"ef37fca8f74c37514d926cf2d8de92bcc36c6b2b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-18 13:56:39.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/eccf5443288e4d10b745f3543b2421e4\n\n- grenade https://zuul.opendev.org/t/openstack/build/a2a884f7f0d64c6ba9cd2e96b5fe6112 : SUCCESS in 1h 04m 13s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/01e24e1a621e4f12a06a547bf63dba12 : SUCCESS in 30m 21s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/69d8803cb558451da38bcf9e4341dbe0 : SUCCESS in 1h 33m 50s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/f5d086703f964bda95480e4b3baa8498 : SUCCESS in 51m 25s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3bc96636ae0e4e04aafb708fa4d97749 : SUCCESS in 5m 06s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/e528424c379848f185e0c9b892480d71 : SUCCESS in 4m 49s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/7aeb0ee8b65648c6b24de0ec9948bba9 : SUCCESS in 6m 50s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/f22a31a1dfda44f88d400e253bb66a5e : SUCCESS in 7m 05s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d65a5ed8cee04f87803e0e67b7ce5c37 : SUCCESS in 6m 59s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/65c2e4d810cc489ab1afcd27f30ae4d4 : SUCCESS in 7m 02s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/a49a40471f4f4458ab5aefe633fff219 : SUCCESS in 6m 12s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/8506e7220ac448e2b37abcc561eb4fcf : SUCCESS in 8m 08s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/87fe9b97e6684c93a1bb3a187c546a21 : SUCCESS in 1h 45m 38s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/253f06599b3a49bea69e3f869dce262a : SUCCESS in 2h 04m 37s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/cf170c7885bd4936b943c15819a5b87c : SUCCESS in 1h 36m 37s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/a53ce1ecf7794deb83d36dce7f141ea3 : SUCCESS in 1h 50m 43s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/5a81aa6482b943eba80dd6839f431421 : SUCCESS in 36m 14s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/39d1abdd94274935841d0bfacd263cbf : SUCCESS in 1h 43m 47s\n- glance-grenade-centralized-cache https://zuul.opendev.org/t/openstack/build/ec8942e7ace946e0a04e00f235ac0233 : FAILURE in 7m 23s (non-voting)\n- glance-s3-multistore https://zuul.opendev.org/t/openstack/build/dccfb1a10af44caabeffa44a5809b86e : SUCCESS in 1h 59m 51s (non-voting)\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/78bbc65b3ff44ad0a2c958cbf7e40696 : SUCCESS in 28m 45s\n- glance-rally-task-uwsgi https://zuul.opendev.org/t/openstack/build/8fa67f40f9cf44c1a59e7bbd5b19b487 : SUCCESS in 59m 15s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"3540cb5117c5596ec8eb477309389fad1714dc32","author":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"date":"2026-05-18 13:58:14.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"a212fa53bc974c1798ae74342f4053a1579b17a7","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2026-05-20 08:34:16.000000000","message":"Patch Set 6: Code-Review+2","accounts_in_message":[],"_revision_number":6},{"id":"4370589a6cc69fa91e3f7c17f97d82ea15ced109","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2026-05-26 12:27:37.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"e1b01fd04b7b8720aff0883595ce2516cf20b0d8","author":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"date":"2026-06-02 15:30:43.000000000","message":"Patch Set 6: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":6},{"id":"3ec5d6b32f31620d8803947adcaa25bfc75fefac","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"date":"2026-06-08 11:02:57.000000000","message":"Patch Set 7: Patch Set 6 was rebased\n\nCopied Votes:\n* Code-Review+2 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n\nOutdated Votes:\n* Verified+1 (copy condition: \"NEVER\")\n* Workflow+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":7},{"id":"5e94f2cfffcb58d833fe182e7812d5a2c72e3081","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-08 13:13:22.000000000","message":"Patch Set 7:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/21cfec66926f436c86bc1b182a4e0c18\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/748485a0881344a992950d61b46c08d5 : SUCCESS in 6m 02s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/39a93d33ec5f4b34b7f9e990773b3ce8 : SUCCESS in 7m 46s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/4ba94a48a756433e8f8d6cbfb7d87306 : SUCCESS in 13m 34s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"e33514e2038d9d417ec9ec1aecee4a6623cfc8db","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-08 13:44:09.000000000","message":"Patch Set 7: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/d529e25fec4e4ff59846d6bb501c10ed\n\n- grenade https://zuul.opendev.org/t/openstack/build/f10a69fe766542cc810e05cc419719f5 : SUCCESS in 1h 03m 22s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/adacd196dbdc40e089d1af43a6594af5 : SUCCESS in 1h 08m 49s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/109a0a0c1d034f90a24feb5413fb3a25 : FAILURE in 1h 34m 17s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/4d4fd4c616ae479b860e6dba1b3c68f9 : SUCCESS in 1h 05m 13s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9e957825f1df409091c23c25cede2e4c : SUCCESS in 5m 46s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/af73b9b9f7af47e996f62c0c2ee8d1c9 : SUCCESS in 5m 03s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/f6da59c222504973946f917012848735 : SUCCESS in 5m 06s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/0e471cbca1264f85b0d30f5702600a50 : SUCCESS in 8m 27s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e50cffd749084ced89abba30343db382 : SUCCESS in 6m 14s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ce59c26deb7145488ea701a13aff108d : SUCCESS in 9m 29s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/8d218b389b0a48c987fb256365727173 : SUCCESS in 7m 01s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/48f59eb341884b0f9c4707613cb259f4 : SUCCESS in 6m 38s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/fbc99dd84238460c855eaa44db2229ab : SUCCESS in 2h 10m 33s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/1fcd9d74b75b4dc0b7ea68169752ee84 : SUCCESS in 2h 03m 08s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/86022a53f89f425ab05e9ef92a18e99a : SUCCESS in 2h 28m 14s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/a748448451e84420ae733da6e4f8679f : SUCCESS in 1h 55m 06s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/6dcb19d3f7b14f7b9ce5bfa21c4a4e9f : SUCCESS in 54m 43s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/1f67f7a75d804cbab2c1e8ac49f98cdb : SUCCESS in 1h 54m 07s\n- glance-grenade-centralized-cache https://zuul.opendev.org/t/openstack/build/c99aef69854a4b9bb2fb6dad7081d631 : FAILURE in 16m 56s (non-voting)\n- glance-s3-multistore https://zuul.opendev.org/t/openstack/build/a7eeb0c618ae45fbb5fa2d41c9fd7e08 : SUCCESS in 1h 11m 05s (non-voting)\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/a6a0e80adf6d494b883cc57df7c0067d : SUCCESS in 35m 38s\n- glance-rally-task-uwsgi https://zuul.opendev.org/t/openstack/build/baf507df5a0843beadd4590e60699449 : SUCCESS in 53m 26s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"792ff07b49179cb86025e4fb351a745c859646bb","author":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"date":"2026-06-09 07:24:45.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"20271084be0cb1938301555a126f3c01cf3d4d73","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-09 09:59:24.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/01ac26530d6940228fb0b440a9089785\n\n- grenade https://zuul.opendev.org/t/openstack/build/319da3f8c1f0480fab32440b617e5531 : SUCCESS in 1h 07m 36s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/c3a91139e7f744148cb84205354a8494 : SUCCESS in 1h 09m 03s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/ea9b0246f2214188974855619a6c305a : SUCCESS in 1h 23m 30s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/e8a99f9c3a1443bfa7dd0eaeaee06bd0 : SUCCESS in 39m 49s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/91d17394cfd042fb8616737d86518b18 : SUCCESS in 4m 55s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/0b223d246bdf4258a218d731298a20a5 : SUCCESS in 3m 52s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/61b5868ec08a45d9936aa50bf2888dc4 : SUCCESS in 3m 11s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/ac96f0fabe324b8683ad14a7165bb37c : SUCCESS in 6m 30s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1abe99c950b64d868c0cf2c00730becf : SUCCESS in 3m 27s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f557d32fc0a64d6d89d371f0b42b3d1a : SUCCESS in 3m 58s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/a18eb8afb3034d74b34f9f08811fa756 : SUCCESS in 6m 38s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/897b68acd89949719dc802b2c90971fd : SUCCESS in 6m 47s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/790dce2e4e08441f86da290067e65cea : SUCCESS in 2h 02m 36s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/43a2b94d613145348934f81bfd23b3b6 : SUCCESS in 1h 44m 29s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/a7e98185df694a57bd779c6cd7419c8f : SUCCESS in 1h 20m 43s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/f3ac6a7ce9464dd5a20981464fcf9364 : SUCCESS in 2h 24m 25s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/fcc9f1a0d89146d9adc544fedc5e0492 : SUCCESS in 41m 17s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/b88d27d0b2264564b3d63d43664ea1ad : SUCCESS in 1h 07m 45s\n- glance-grenade-centralized-cache https://zuul.opendev.org/t/openstack/build/10848a2cd481453d94a69ca32697e132 : FAILURE in 17m 16s (non-voting)\n- glance-s3-multistore https://zuul.opendev.org/t/openstack/build/abce61902a614fef864d3382c7ec577e : SUCCESS in 1h 17m 46s (non-voting)\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/ead804d8bd0c4845867995a7a1bfe211 : SUCCESS in 22m 53s\n- glance-rally-task-uwsgi https://zuul.opendev.org/t/openstack/build/747150ed38bb4064983500ec59d9b3c0 : SUCCESS in 32m 50s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"2684468f5236d19fed6e2255469626a8aab67276","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-09 11:03:42.000000000","message":"Patch Set 7:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e3a1c39e60064fb1804473b9d5a55597\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/aaef28af9eb240b9ba1b0d5fcf256eba : SUCCESS in 6m 10s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/8cbec4a5b2c642c29939e26d0ebef078 : SUCCESS in 6m 10s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/4b5758ed5a21496d8f034a4cc1ef2472 : SUCCESS in 14m 15s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"e388f241c80ddebde5d03f1f68d6fd04de297103","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2026-06-09 11:48:36.000000000","message":"Patch Set 7: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":7},{"id":"be0baf9c41a039b22643e54998e5c0786f55a9ad","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-09 11:51:10.000000000","message":"Patch Set 7: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":7},{"id":"68512249d7d1d69a61f04f69022228c47f311e8c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-09 13:57:03.000000000","message":"Patch Set 7: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/93e5386d57f248a0b86d18dbff2c2ff5\n\n- grenade https://zuul.opendev.org/t/openstack/build/c18a95a811744ca9b8b459563f5a53ec : SUCCESS in 1h 01m 24s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/ffebe2fa8ffa4be19a07bafbdaed41c4 : SUCCESS in 1h 03m 41s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/9079d38cd819425e9340d04aed86d50d : SUCCESS in 2h 00m 17s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/5c2d21b2d4b04535a6abb47d4222a76a : SUCCESS in 58m 08s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d67fe33b4a1649d7a02c9362ecaf41f7 : SUCCESS in 6m 51s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/9bba12691f7d46f8a63ba2ff1658cb2d : SUCCESS in 5m 28s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/3b7519247cae40a9b51b20af8bf9813a : SUCCESS in 3m 11s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/82f8e4f594294f83ac9c572d61aaf38c : SUCCESS in 5m 50s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/353abe63563a4ff8ae3003d0da7dbf1a : SUCCESS in 8m 33s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/c256bc24ad5c4b9ea040bc92145e9d50 : SUCCESS in 6m 31s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/eabb5df7777b4c0cb5c9a671c2c5675c : SUCCESS in 7m 05s\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/727910e967844947ac1c4ebaa956b960 : SUCCESS in 1h 13m 08s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/217839e2375040019a20051cb0037c67 : FAILURE in 1h 43m 19s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/7e2d545320ab4bbbb3ae1b8afa37ee12 : SUCCESS in 34m 06s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/b8c276609a784fb9a401c43ef33ef0e6 : SUCCESS in 1h 38m 41s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/b1499a1e518d485aaf9845bce7713b98 : SUCCESS in 35m 36s","accounts_in_message":[],"_revision_number":7},{"id":"d0f7b35d7b7c07f91a0370d4e8fc98c0b613ecbf","author":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"date":"2026-06-10 04:07:02.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"03d531d27972f47695fbf231e408239554a41ee0","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-10 04:43:31.000000000","message":"Patch Set 7:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9bb0ad9274224fd6a0e66cae4d4f1b0c\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/4a116cd683a442468bc18a623f1b96b1 : SUCCESS in 8m 20s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/0202a52ad68b4901b8d805e9ae47fe29 : SUCCESS in 8m 35s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/0a091b2acfcd44a68afd86f5c8a62915 : SUCCESS in 13m 49s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"dd05f2d2a75c2297caf13d00c6fe28821ae0294e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-10 06:12:44.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/fc7d9913b9c94ed9b4872a8eb0d91e7d\n\n- grenade https://zuul.opendev.org/t/openstack/build/1357de09d42645f7b6e3af7e9c8ea704 : SUCCESS in 50m 22s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/3488b78449bf4ee9879732c51c48a611 : SUCCESS in 33m 18s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/352c6ad4585040ebb8c660a0c44b4a2e : SUCCESS in 1h 25m 35s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/a0025f7819644e77be96c34c082eea47 : SUCCESS in 37m 28s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a9931d234cad46efb81fb59b16d829bb : SUCCESS in 4m 34s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/fd96270b85374ea4857e994aed65e000 : SUCCESS in 4m 45s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/9389e7cfd2e249008c7652c75faac320 : SUCCESS in 4m 09s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/1dd301f846dd4e409d025f9b46b41882 : SUCCESS in 9m 07s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/aa07c828a94b42da8242382164419ba0 : SUCCESS in 6m 40s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c6e8be42765c42ce8e70c8eca0b049d7 : SUCCESS in 3m 51s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/5806b7e057704cadb652a6418257c18e : SUCCESS in 4m 44s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/cdfb5b3e6c96404f99a39ba6a06c52cd : SUCCESS in 6m 17s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/5f1c4c6fa82d4c318b3e97d4154a2688 : SUCCESS in 1h 46m 43s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/1aee5f0410464339b7a33ef5a9042105 : SUCCESS in 2h 03m 37s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/c35ba83dec774c7691fd54978a1d2574 : SUCCESS in 1h 05m 51s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/679ee5ca2743492c9cf02d1d51cd9160 : SUCCESS in 1h 57m 36s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/d123c3e7a49d4599826d822791480d52 : SUCCESS in 1h 23m 18s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/162ecda4364e46889efb6a5d5a9ce78d : SUCCESS in 1h 02m 35s\n- glance-grenade-centralized-cache https://zuul.opendev.org/t/openstack/build/edd822db8af84548b3c2e64649274a98 : FAILURE in 15m 18s (non-voting)\n- glance-s3-multistore https://zuul.opendev.org/t/openstack/build/c7d4c0502cf14f088a7be51dd63708ff : SUCCESS in 1h 55m 23s (non-voting)\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/fa884b0868f948b8afc92077a56d6448 : SUCCESS in 33m 55s\n- glance-rally-task-uwsgi https://zuul.opendev.org/t/openstack/build/6f84aefe9f524ae2a3e30cb9eae789c4 : SUCCESS in 52m 11s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"9b63ee99bf965ca32b5fcb8a1a5a112a18935398","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-10 06:13:30.000000000","message":"Patch Set 7: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":7},{"id":"cf8755184071586a1e0d06ed3363f10519bd4906","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-10 08:29:29.000000000","message":"Patch Set 7: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e446b75f328f41f086d60154a9c8568d\n\n- grenade https://zuul.opendev.org/t/openstack/build/26820d610e8a494ca02855481a8432ff : SUCCESS in 1h 01m 55s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/497427369569471ba845e9873d4ef2aa : SUCCESS in 34m 09s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/3e212c9ec53640cb882773f36b9f2292 : SUCCESS in 1h 28m 26s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/8b4679a86bb34c2c81aaaf54e6b95f5e : SUCCESS in 49m 22s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/329691b3ab604a109e7dbc74f7504700 : SUCCESS in 2m 37s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/fecb9fd0d8e3415cbc4cb7d06a0da1ff : SUCCESS in 3m 08s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/d02a40e5467849f4a3d08e101ccd29a6 : SUCCESS in 4m 51s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/95639ed29d514ff3aa9a5cd0bf40cbb5 : SUCCESS in 5m 06s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/19ac87f9aa28440bb36c67f05fe4d7fa : SUCCESS in 7m 47s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/9b028ede53544f4da506568543d44c4f : SUCCESS in 4m 51s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/6df33f776e234a7ba136ed01ad1ef2c5 : SUCCESS in 6m 09s\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/021fb716f7f74969bde2e734808f9341 : SUCCESS in 2h 11m 11s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/ec21a040500f4ceab075595d2beb482d : SUCCESS in 1h 54m 43s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/48c7566ed2b24926ae833e280f5dc376 : SUCCESS in 31m 02s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/4fe6c3a51c6d4b79a28aed9216785316 : SUCCESS in 1h 12m 25s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/9083b9a02bea43859a40253a66ea9b36 : SUCCESS in 36m 39s","accounts_in_message":[],"_revision_number":7},{"id":"e6572b831cda2beb2e92f60d040a65f794e85d84","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-10 08:29:29.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":7},{"id":"2ec3475fa58f482f7cf6dbb3cfe5112604e04c4c","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-10 08:31:19.000000000","message":"Patch Set 7:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/eca232465d9f48e4a22767b9fbdbcf73\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/555fcdd3fdc64da2a52a8493d2ba1c93 : SUCCESS in 46s","accounts_in_message":[],"_revision_number":7}],"current_revision_number":7,"current_revision":"66dea869ec16795729776b8061444da56fadd23a","revisions":{"26aa6a715b6125109537134464898a01cad91f42":{"kind":"REWORK","_number":1,"created":"2026-04-24 14:18:12.000000000","uploader":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"ref":"refs/changes/93/986093/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/93/986093/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/93/986093/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/93/986093/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/93/986093/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/93/986093/1"}}},"commit":{"parents":[{"commit":"03dbf8eac63a35aa0207eba21a3c33e0aa51bb18","subject":"Replace eventlet.sleep with time.sleep in tasks API test","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/03dbf8eac63a35aa0207eba21a3c33e0aa51bb18"}]}],"author":{"name":"Pranali Deore","email":"pdeore@redhat.com","date":"2026-04-24 13:31:36.000000000","tz":0},"committer":{"name":"Pranali Deore","email":"pdeore@redhat.com","date":"2026-04-24 13:32:28.000000000","tz":0},"subject":"Add non-voting Zuul Rally job for Glance on uWSGI","message":"Add non-voting Zuul Rally job for Glance on uWSGI\n\nAdding glance-rally-task-uwsgi (parent rally-task-at-devstack)\nso CI runs rally-jobs/glance.yaml against uwsgi devstack\ndeployment.\n\nChange-Id: Iae258108284b84412c6e52a3fbade1213547c480\nSigned-off-by: Pranali Deore \u003cpdeore@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/26aa6a715b6125109537134464898a01cad91f42"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/26aa6a715b6125109537134464898a01cad91f42"}]},"branch":"refs/heads/master"},"83cea8449a6c485fd8b885e0d5bb5df2186c9490":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2026-05-11 14:11:24.000000000","uploader":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"ref":"refs/changes/93/986093/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/93/986093/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/93/986093/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/93/986093/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/93/986093/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/93/986093/2"}}},"commit":{"parents":[{"commit":"03dbf8eac63a35aa0207eba21a3c33e0aa51bb18","subject":"Replace eventlet.sleep with time.sleep in tasks API test","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/03dbf8eac63a35aa0207eba21a3c33e0aa51bb18"}]}],"author":{"name":"Pranali Deore","email":"pdeore@redhat.com","date":"2026-04-24 13:31:36.000000000","tz":0},"committer":{"name":"Pranali Deore","email":"pdeore@redhat.com","date":"2026-05-11 14:11:23.000000000","tz":0},"subject":"Add non-voting Zuul Rally job for Glance on uWSGI","message":"Add non-voting Zuul Rally job for Glance on uWSGI\n\nAdding glance-rally-task-uwsgi (parent rally-task-at-devstack)\nso CI runs rally-jobs/glance.yaml against uwsgi devstack\ndeployment.\n\nDepends-On: https://review.opendev.org/c/openstack/glance/+/988112\n\nChange-Id: Iae258108284b84412c6e52a3fbade1213547c480\nSigned-off-by: Pranali Deore \u003cpdeore@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/83cea8449a6c485fd8b885e0d5bb5df2186c9490"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/83cea8449a6c485fd8b885e0d5bb5df2186c9490"}]},"branch":"refs/heads/master","description":"Edit commit message"},"20afd025c7614f915eeea170a6810d4eaabfdf1e":{"kind":"TRIVIAL_REBASE_WITH_MESSAGE_UPDATE","_number":3,"created":"2026-05-11 14:16:38.000000000","uploader":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"ref":"refs/changes/93/986093/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/93/986093/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/93/986093/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/93/986093/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/93/986093/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/93/986093/3"}}},"commit":{"parents":[{"commit":"d69e477f6804402891071e48281d56259d06924c","subject":"Fix 415 bodies for unsupported media type responses","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/d69e477f6804402891071e48281d56259d06924c"}]}],"author":{"name":"Pranali Deore","email":"pdeore@redhat.com","date":"2026-04-24 13:31:36.000000000","tz":0},"committer":{"name":"PranaliD","email":"pdeore@redhat.com","date":"2026-05-11 14:14:32.000000000","tz":0},"subject":"Add non-voting Zuul Rally job for Glance on uWSGI","message":"Add non-voting Zuul Rally job for Glance on uWSGI\n\nAdding glance-rally-task-uwsgi (parent rally-task-at-devstack)\nso CI runs rally-jobs/glance.yaml against uwsgi devstack\ndeployment.\n\nChange-Id: Iae258108284b84412c6e52a3fbade1213547c480\nSigned-off-by: Pranali Deore \u003cpdeore@redhat.com\u003e\nSigned-off-by: PranaliD \u003cpdeore@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/20afd025c7614f915eeea170a6810d4eaabfdf1e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/20afd025c7614f915eeea170a6810d4eaabfdf1e"}]},"branch":"refs/heads/master"},"9a6cc3d7b4c65cb754ed99aa0492c86f0049f42a":{"kind":"TRIVIAL_REBASE_WITH_MESSAGE_UPDATE","_number":4,"created":"2026-05-12 09:37:57.000000000","uploader":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"ref":"refs/changes/93/986093/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/93/986093/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/93/986093/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/93/986093/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/93/986093/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/93/986093/4"}}},"commit":{"parents":[{"commit":"0847ff925307c6cf80658955e60526a2d110b682","subject":"Rally: use raw disk_format for synthetic fake.img","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/0847ff925307c6cf80658955e60526a2d110b682"}]}],"author":{"name":"Pranali Deore","email":"pdeore@redhat.com","date":"2026-04-24 13:31:36.000000000","tz":0},"committer":{"name":"PranaliD","email":"pdeore@redhat.com","date":"2026-05-12 09:19:02.000000000","tz":0},"subject":"Add non-voting Zuul Rally job for Glance on uWSGI","message":"Add non-voting Zuul Rally job for Glance on uWSGI\n\nAdding glance-rally-task-uwsgi (parent rally-task-at-devstack)\nso CI runs rally-jobs/glance.yaml against uwsgi devstack\ndeployment.\n\nChange-Id: Iae258108284b84412c6e52a3fbade1213547c480\nSigned-off-by: PranaliD \u003cpdeore@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/9a6cc3d7b4c65cb754ed99aa0492c86f0049f42a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/9a6cc3d7b4c65cb754ed99aa0492c86f0049f42a"}]},"branch":"refs/heads/master"},"7ce5bb0987e47b0f8761e308452320aae31497a2":{"kind":"REWORK","_number":5,"created":"2026-05-12 10:59:35.000000000","uploader":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"ref":"refs/changes/93/986093/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/93/986093/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/93/986093/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/93/986093/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/93/986093/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/93/986093/5"}}},"commit":{"parents":[{"commit":"0847ff925307c6cf80658955e60526a2d110b682","subject":"Rally: use raw disk_format for synthetic fake.img","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/0847ff925307c6cf80658955e60526a2d110b682"}]}],"author":{"name":"Pranali Deore","email":"pdeore@redhat.com","date":"2026-04-24 13:31:36.000000000","tz":0},"committer":{"name":"PranaliD","email":"pdeore@redhat.com","date":"2026-05-12 10:56:45.000000000","tz":0},"subject":"Add non-voting Zuul Rally job for Glance on uWSGI","message":"Add non-voting Zuul Rally job for Glance on uWSGI\n\nAdding glance-rally-task-uwsgi (parent rally-task-at-devstack)\nso CI runs rally-jobs/glance.yaml against uwsgi devstack\ndeployment.\n\nSet GLANCE_LIMIT_IMAGE_COUNT_TOTAL and\nGLANCE_LIMIT_IMAGE_COUNT_UPLOADING in devstack_localrc for\nthis job so the 700-iteration Rally subtask does not\nhit the default Keystone image_count_total limit (HTTP 413).\n\nChange-Id: Iae258108284b84412c6e52a3fbade1213547c480\nSigned-off-by: PranaliD \u003cpdeore@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/7ce5bb0987e47b0f8761e308452320aae31497a2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/7ce5bb0987e47b0f8761e308452320aae31497a2"}]},"branch":"refs/heads/master"},"9363d90fd908c73acdd5d0e9f790ceb28942a9a3":{"kind":"NO_CODE_CHANGE","_number":6,"created":"2026-05-18 11:49:27.000000000","uploader":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"ref":"refs/changes/93/986093/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/93/986093/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/93/986093/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/93/986093/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/93/986093/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/93/986093/6"}}},"commit":{"parents":[{"commit":"0847ff925307c6cf80658955e60526a2d110b682","subject":"Rally: use raw disk_format for synthetic fake.img","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/0847ff925307c6cf80658955e60526a2d110b682"}]}],"author":{"name":"Pranali Deore","email":"pdeore@redhat.com","date":"2026-04-24 13:31:36.000000000","tz":0},"committer":{"name":"Pranali Deore","email":"pdeore@redhat.com","date":"2026-05-18 11:49:26.000000000","tz":0},"subject":"Add non-voting Zuul Rally job for Glance on uWSGI","message":"Add non-voting Zuul Rally job for Glance on uWSGI\n\nAdding glance-rally-task-uwsgi (parent rally-task-at-devstack)\nso CI runs rally-jobs/glance.yaml against uwsgi devstack\ndeployment.\n\nSet GLANCE_LIMIT_IMAGE_COUNT_TOTAL and\nGLANCE_LIMIT_IMAGE_COUNT_UPLOADING in devstack_localrc for\nthis job so the 700-iteration Rally subtask does not\nhit the default Keystone image_count_total limit (HTTP 413).\n\nDepends-On: https://review.opendev.org/c/openstack/devstack/+/988953\n\nChange-Id: Iae258108284b84412c6e52a3fbade1213547c480\nSigned-off-by: PranaliD \u003cpdeore@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/9363d90fd908c73acdd5d0e9f790ceb28942a9a3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/9363d90fd908c73acdd5d0e9f790ceb28942a9a3"}]},"branch":"refs/heads/master","description":"Edit commit message"},"66dea869ec16795729776b8061444da56fadd23a":{"kind":"NO_CHANGE","_number":7,"created":"2026-06-08 11:02:57.000000000","uploader":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"ref":"refs/changes/93/986093/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/93/986093/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/93/986093/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/93/986093/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/93/986093/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/93/986093/7"}}},"commit":{"parents":[{"commit":"6892eabdd9b494bfd7dbf410ef2d23c5325bb174","subject":"Rally: use raw disk_format for synthetic fake.img","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/6892eabdd9b494bfd7dbf410ef2d23c5325bb174"}]}],"author":{"name":"Pranali Deore","email":"pdeore@redhat.com","date":"2026-04-24 13:31:36.000000000","tz":0},"committer":{"name":"Pranali Deore","email":"pdeore@redhat.com","date":"2026-06-08 11:02:57.000000000","tz":0},"subject":"Add non-voting Zuul Rally job for Glance on uWSGI","message":"Add non-voting Zuul Rally job for Glance on uWSGI\n\nAdding glance-rally-task-uwsgi (parent rally-task-at-devstack)\nso CI runs rally-jobs/glance.yaml against uwsgi devstack\ndeployment.\n\nSet GLANCE_LIMIT_IMAGE_COUNT_TOTAL and\nGLANCE_LIMIT_IMAGE_COUNT_UPLOADING in devstack_localrc for\nthis job so the 700-iteration Rally subtask does not\nhit the default Keystone image_count_total limit (HTTP 413).\n\nDepends-On: https://review.opendev.org/c/openstack/devstack/+/988953\n\nChange-Id: Iae258108284b84412c6e52a3fbade1213547c480\nSigned-off-by: PranaliD \u003cpdeore@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/66dea869ec16795729776b8061444da56fadd23a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/66dea869ec16795729776b8061444da56fadd23a"}]},"branch":"refs/heads/master","description":"Rebase","conflicts":{"base":"0847ff925307c6cf80658955e60526a2d110b682","ours":"9363d90fd908c73acdd5d0e9f790ceb28942a9a3","theirs":"6892eabdd9b494bfd7dbf410ef2d23c5325bb174","merge_strategy":"recursive","contains_conflicts":false}}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{"label:Review-Priority\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
