)]}'
{"id":"openstack%2Fglance~986467","triplet_id":"openstack%2Fglance~master~I9604420fb4c2df7c1a9c7113eecf982f199f0770","project":"openstack/glance","branch":"master","topic":"eventlet-removal-TestImagesIPv6","attention_set":{},"removed_from_attention_set":{"8122":{"account":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"last_update":"2026-05-11 08:13:18.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I9604420fb4c2df7c1a9c7113eecf982f199f0770","subject":"eventlet removal: Remove TestImagesIPv6","status":"MERGED","created":"2026-04-27 22:37:45.000000000","updated":"2026-05-11 08:14:22.000000000","submitted":"2026-05-11 08:13:18.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"986467-eventlet-removal-TestImagesIPv6","meta_rev_id":"5c9455b229ef3ae8d7ed42bd3cb2751f24fb8b68","_number":986467,"virtual_id_number":986467,"owner":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-05-11 08:13:18.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"all":[{"value":2,"date":"2026-05-11 06:43:41.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":2,"date":"2026-05-07 14:17:33.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"all":[{"value":1,"date":"2026-05-11 06:43:41.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Top Priority / Holds Gate"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-04-27 23:02:46.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2026-04-27 23:44:26.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-05-07 14:17:33.000000000","updated_by":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"reviewer":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"state":"REVIEWER"},{"updated":"2026-05-11 06:43:41.000000000","updated_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"reviewer":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"state":"REVIEWER"}],"messages":[{"id":"55e44a3a79c92034b8670bd02d386b894491429c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"date":"2026-04-27 22:37:45.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"8edb23de66ca6a45fb442e52fdc023b49bdfdf03","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-27 23:02:46.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e67e797e4d6a4c8f949008cc1137e5f6\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/c50244c800f8448a9f1ceec2c8cecf68 : SUCCESS in 5m 46s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/29d76f9845eb4ab48b0f710d2bbd80ae : SUCCESS in 8m 49s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/bc0494255abb4d8480675e5e45e3a093 : SUCCESS in 12m 59s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"f065b21dd049a626e1d4ba2e33f703848fe729e4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-27 23:44:26.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/71f7a3e80cec4ce090c4d22370f5a227\n\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/78cc2a797aa74316aefc002af4ef53cc : SUCCESS in 37m 42s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/f88f5bb8ebd94951a14c26690740a0a4 : SUCCESS in 1h 01m 46s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fbb4c7b716e6411b8e2b5a4dc215f565 : SUCCESS in 3m 02s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/a6290ce7da01429183370f8d7a3a0b13 : SUCCESS in 2m 34s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/9e559441425c41999aeae5d7684d4f23 : SUCCESS in 4m 19s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/2014f437995d473abf6a3025d77e84fa : SUCCESS in 4m 05s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dcabbad12afb49ccb53475ebc1c0d34f : SUCCESS in 4m 23s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/698ec0e8185d4e1eaf1f144aa935c9a7 : SUCCESS in 6m 58s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/f3c5048b6a1e426ebe51ad6440dc4efc : SUCCESS in 8m 56s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/64a2a8c5d39a488ebb07835b3a888f58 : SUCCESS in 32m 13s","accounts_in_message":[],"_revision_number":1},{"id":"4bd16cf5a9316d16966a493b4983a321818568a7","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2026-05-07 14:17:33.000000000","message":"Patch Set 1: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"81932b3ae44c9c9cbecffaf51c4157a018e2e81a","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2026-05-11 06:43:41.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"d0afaa6c119ed410fee25cbcd20989d8d9578750","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-11 06:44:16.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"d136251f2428cf4821588b1f18fa845ceb5f0f36","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-11 08:13:18.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/cd8baa107067482dabaa3ed909efded2\n\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/825b6f9380f64976a00cee410e33091a : SUCCESS in 1h 03m 28s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/71e3f58a83fb44429798c5abdeb41eeb : SUCCESS in 1h 04m 59s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/46bba74d5f1649389b52c644340a2b75 : SUCCESS in 4m 08s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/f2e355df89e84f0a8f9ff960c13776e7 : SUCCESS in 4m 27s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/f628854c81134115b47536aa7aee7129 : SUCCESS in 4m 49s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f6bb7bf5ecfa400b826fc43480795d77 : SUCCESS in 5m 30s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/2518995a55a14bcea2ee32f9c5be40cb : SUCCESS in 6m 15s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/58cb8d3dccce44eeb2eebd704df7ba56 : SUCCESS in 5m 54s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/193edd37a3654ba99557843c4650880b : SUCCESS in 34m 13s","accounts_in_message":[],"_revision_number":1},{"id":"129e67bf2a5d22f7af53a752ceea60c1042b8e8f","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-11 08:13:18.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"5c9455b229ef3ae8d7ed42bd3cb2751f24fb8b68","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-11 08:14:22.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a6a37d190900440f8348b3a00f93741f\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e51eaee74415499c8f1ac474e49978f7 : SUCCESS in 51s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"42fe30907dd32cc44952bbd023cd273394c18a75","revisions":{"42fe30907dd32cc44952bbd023cd273394c18a75":{"kind":"REWORK","_number":1,"created":"2026-04-27 22:37:45.000000000","uploader":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"ref":"refs/changes/67/986467/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/67/986467/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/67/986467/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/67/986467/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/67/986467/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/67/986467/1"}}},"commit":{"parents":[{"commit":"03dbf8eac63a35aa0207eba21a3c33e0aa51bb18","subject":"Replace eventlet.sleep with time.sleep in tasks API test","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/03dbf8eac63a35aa0207eba21a3c33e0aa51bb18"}]}],"author":{"name":"Cyril Roelandt","email":"cyril@redhat.com","date":"2026-04-27 21:46:39.000000000","tz":120},"committer":{"name":"Cyril Roelandt","email":"cyril@redhat.com","date":"2026-04-27 22:36:22.000000000","tz":120},"subject":"eventlet removal: Remove TestImagesIPv6","message":"eventlet removal: Remove TestImagesIPv6\n\nWithout standalone mode, it is no longer up to us to test IPv4/IPv6\nsettings. TestImagesIPv6 still runs glance.cmd.api in a separate\nprocess, so it cannot easily be ported to SynchronousAPIBase anyway.\n\nAlso remove tracecmd_osmap, which could have been removed in\n865fbf6a31b431aab0c9a5f7b80d5b01189dbc00 .\n\nChange-Id: I9604420fb4c2df7c1a9c7113eecf982f199f0770\nSigned-off-by: Cyril Roelandt \u003ccyril@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/42fe30907dd32cc44952bbd023cd273394c18a75"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/42fe30907dd32cc44952bbd023cd273394c18a75"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{"label:Review-Priority\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
