)]}'
{"id":"openstack%2Fglance~986887","triplet_id":"openstack%2Fglance~master~I4fd96b255790c6a21d9dec99aa84a16555298a9c","project":"openstack/glance","branch":"master","topic":"drop-python310","attention_set":{},"removed_from_attention_set":{"15334":{"account":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"last_update":"2026-05-05 09:31:08.000000000","reason":"\u003cGERRIT_ACCOUNT_15334\u003e replied on the change","reason_account":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"}},"8122":{"account":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"last_update":"2026-05-04 15:02:07.000000000","reason":"\u003cGERRIT_ACCOUNT_8122\u003e replied on the change","reason_account":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"}}},"hashtags":[],"change_id":"I4fd96b255790c6a21d9dec99aa84a16555298a9c","subject":"Drop Python 3.10","status":"MERGED","created":"2026-05-01 10:32:09.000000000","updated":"2026-05-05 13:57:50.000000000","submitted":"2026-05-05 13:55:01.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"986887-drop-python310","meta_rev_id":"80d50001bde39a03ade70528877d6ae03cc8779c","_number":986887,"virtual_id_number":986887,"owner":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-05-05 13:55:01.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"all":[{"value":2,"date":"2026-05-04 15:02:07.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2026-05-03 18:37:05.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"all":[{"value":1,"date":"2026-05-04 15:02:07.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Top Priority / Holds Gate"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-05-01 10:47:22.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2026-05-01 13:01:36.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-05-03 18:37:05.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"state":"REVIEWER"},{"updated":"2026-05-03 18:37:05.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"state":"REVIEWER"}],"messages":[{"id":"6f730df86f59e767c0a5759da947b8ffa42d9534","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2026-05-01 10:32:09.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"1b295c8409371f8c37f7742fcbb88a48192ff761","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-01 10:47:22.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/fc328bb622de4cc6902e2c45ad276c79\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/c232563ded0a41eb91778dda9694a696 : FAILURE in 5m 50s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/ed2455626eea4191b3f70841a23c665a : SUCCESS in 8m 10s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/0848a9410656480f9cffb7f640bb3ab3 : SUCCESS in 13m 10s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"b43020d1fa305c003020f2b783032c5f91e55143","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-01 13:01:36.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/2a97870cbc534462865461634170bc5b\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/ed150690d814452c9887ac7b4bef0705 : SUCCESS in 2m 22s\n- requirements-check https://zuul.opendev.org/t/openstack/build/fc2ccebb69644392840b7171d5033ae4 : SUCCESS in 2m 52s\n- grenade https://zuul.opendev.org/t/openstack/build/6c85348185314f43bbdde39a5e51edbc : SUCCESS in 1h 14m 52s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/6480e24833374668a0cf2641f79b57e8 : SUCCESS in 1h 01m 29s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/c355560cb55d419799e687e63cbc4ee5 : SUCCESS in 1h 12m 06s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/9ff92183af4044f6a2d8ef94823cefd7 : SUCCESS in 1h 08m 49s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/71016f2ac87c4710a2d965883e13b80f : SUCCESS in 4m 32s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/45d0089506974c958a4300fa25896057 : SUCCESS in 4m 34s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/67c49fae804b4299a6511ff4aefa3812 : SUCCESS in 4m 49s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/5c3db6126b3941d78b9f163bcab9b58b : SUCCESS in 8m 44s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e57fa2f918ba42659a5190403a9ff8f0 : SUCCESS in 5m 18s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ba70dbfa0d934821bc1ac7044e10533e : SUCCESS in 6m 41s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/b071a02470554be999af7506d9737eea : FAILURE in 4m 13s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/d1b40ec5cefd48d09de3b9a868edfbed : SUCCESS in 6m 11s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/0b2d1f8266ff45488320aa1890826a75 : SUCCESS in 1h 49m 17s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/34a82b75b80c42d1bcca819634d41466 : SUCCESS in 2h 02m 26s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/aa6d40609f1d42158efd2c18d5346ea2 : SUCCESS in 1h 59m 48s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/cc3b1414cc874d60a1e5bba7eae204f4 : SUCCESS in 2h 23m 59s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/1ad44393c04444a6808a645bb2b97e08 : SUCCESS in 1h 01m 12s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/00279776010c4dd0adbe18f669a86dee : SUCCESS in 1h 33m 38s\n- glance-grenade-centralized-cache https://zuul.opendev.org/t/openstack/build/959c5b9616fc412d9568aa6a33b3c319 : FAILURE in 14m 31s (non-voting)\n- glance-s3-multistore https://zuul.opendev.org/t/openstack/build/a4d0b284fc39415984106449035713f3 : SUCCESS in 1h 59m 46s (non-voting)\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/762a34282e6549c0aa94eb458cc12682 : SUCCESS in 38m 42s","accounts_in_message":[],"_revision_number":1},{"id":"781a23a4b096a3a778a0978bbb071a599d3649b2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2026-05-01 15:55:38.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":2},{"id":"1768705d1ff1bb627623a196599e2ecbe2097056","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-01 16:07:35.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/36eed4acd1b8426b93bcab0a87a9d1a6\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/1664bb80d57d492494e7b80d12462242 : FAILURE in 6m 17s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/2860230c8d9949af9e2fc7713e1f9ae6 : SUCCESS in 7m 21s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/fecfeaa379c94d91b41fe3d94caf60e3 : SUCCESS in 8m 59s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"940695caac50ecc08353f0587b5eb637eea28fe7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-01 18:10:24.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7b0f20c39ce1453690fd5cc2e3ce032e\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/bff39e9db0734bdfba4d1c570861b6bf : SUCCESS in 2m 28s\n- requirements-check https://zuul.opendev.org/t/openstack/build/495b714d3b6e4651a289cb084eb4c5a9 : SUCCESS in 3m 27s\n- grenade https://zuul.opendev.org/t/openstack/build/c4129a384af64f8aa3fb013516e0ea7c : SUCCESS in 1h 02m 51s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/697ab4d360544b288dcae69ba6dfc336 : SUCCESS in 1h 09m 54s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/0279cd7e8d554a62a42ce986e10c8fc8 : SUCCESS in 1h 20m 23s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/a49571a006ae4d8ca571bd63653f524f : SUCCESS in 1h 07m 52s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8020bc12c5fa40cdb7e6bdb8d9fe5e17 : SUCCESS in 5m 26s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/1f74efe7abc84e3fa999b5544b586483 : SUCCESS in 4m 32s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/95d39a325bd54f889dcbc3f4ec960757 : SUCCESS in 5m 01s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/5f0f3a31ee134624812e5a0577987af1 : SUCCESS in 7m 30s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9d162d93791d4887bdd58fee49dcaf7f : SUCCESS in 6m 04s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/671583726b9d436b964d3b785c769db7 : SUCCESS in 7m 33s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/939a0ea9b5574411a7873605501f209e : SUCCESS in 6m 26s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/f80034958ce24be4a89aedb4dc44f18c : SUCCESS in 6m 02s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/b7cffdb7ae6e47cfb1e0dce8a82d083e : SUCCESS in 1h 38m 44s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/7afe3afdb44247f9992883e1eb60364d : SUCCESS in 2h 05m 05s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/0024c26dabe0432db9c7cf54c47228ec : SUCCESS in 1h 58m 24s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/bd90e162581b49d09cd9637b98da40fb : SUCCESS in 1h 58m 17s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/b1bba025ce5a4c3fb97ab4b3e5df9968 : SUCCESS in 1h 02m 44s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/43483a0c608b4b8d925a3c72791e5369 : SUCCESS in 1h 45m 05s\n- glance-grenade-centralized-cache https://zuul.opendev.org/t/openstack/build/8ff93fc2ab7f4f39911c603fccf563aa : FAILURE in 7m 40s (non-voting)\n- glance-s3-multistore https://zuul.opendev.org/t/openstack/build/0bc8e3031c924f99ad30ff2b35eb67b9 : SUCCESS in 1h 08m 02s (non-voting)\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/9063d1e22a1a4241bde2dc9511cf0774 : SUCCESS in 40m 17s","accounts_in_message":[],"_revision_number":2},{"id":"11ce690d0e52f470e64606dcbd38d6a439dca018","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-05-03 18:37:05.000000000","message":"Patch Set 2: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"3b3655b92caecdea47dc93f05a6cb64d7d82e98d","author":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"date":"2026-05-04 15:02:07.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"77d902bc0cb1971827961a0a72e20da8e0f7b010","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-04 15:03:31.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"9ce0d623cc036eb1b34a6053bb3ec77e4c77b69e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-04 17:29:02.000000000","message":"Patch Set 2: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/7ec6afec957247bd9555db2e9a5e7e6a\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/ab6bae69aee9438297d0f8a81c8bf513 : SUCCESS in 1m 49s\n- requirements-check https://zuul.opendev.org/t/openstack/build/14991b7fdb0d44a2b00d9a1351abb2a4 : SUCCESS in 3m 05s\n- grenade https://zuul.opendev.org/t/openstack/build/bf473303bc204f5090afa3da020a2636 : SUCCESS in 1h 31m 32s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/2504c7d466f2495b8446ec96361ed773 : SUCCESS in 56m 35s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/64a6d99085714d218b28dcef0763e3e9 : POST_FAILURE in 1h 32m 49s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/e9bb6f9273b149809e35e05be6e7ac78 : SUCCESS in 1h 07m 22s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bd2995d563444739b32e6d175a219b9a : SUCCESS in 5m 35s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/626839f57f3c49e8ac17cccc89a892a4 : SUCCESS in 3m 40s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/7d487ee0e22b441594d172dca4c41980 : SUCCESS in 5m 47s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5a2a06bc0c1c4a549ea8b6f7766c14d9 : SUCCESS in 6m 04s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/174936967c8c41f2bb72a027cec21ac9 : SUCCESS in 6m 32s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/706dd1ab61c24f8789d079fef841bd73 : SUCCESS in 6m 29s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/1206738845d74a72a847f8dd2465019b : SUCCESS in 7m 57s\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/6032cf008b2e472a9004d239bad552e8 : SUCCESS in 2h 18m 59s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/e4f0f52cf6144c7182349a5219c76fcc : SUCCESS in 1h 06m 14s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/9277bb2f933d42099671292f3680a3f3 : SUCCESS in 1h 00m 52s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/eb60d32664614a788d1d57f8059770d4 : SUCCESS in 1h 49m 30s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/c61a537c8b674757afb4287ba6d199ec : SUCCESS in 29m 09s","accounts_in_message":[],"_revision_number":2},{"id":"42dd4abca3e4939914b41c94ae5788aef9124938","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-05-04 17:34:44.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"b93421830877a2674f0788bc361251de2e70fa6d","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-04 18:20:55.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e3252591539c4605bfee54c4ef585ba1\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/da979cde255142d1b3e5bf3a7b2ab428 : FAILURE in 5m 19s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/2f449d7d39394b7aaf4c716acf98b155 : SUCCESS in 6m 09s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/3f4f3765853a4ad09f2a7a4413e2ed2b : SUCCESS in 11m 12s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"9103f9f8ec1eb631dee0d243eaae7640cf6a1de4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-04 19:59:43.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/e3f06e94f6674989b0a0a708581ddaec\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/c69d984730c445b196df331e8e3c152a : SUCCESS in 2m 31s\n- requirements-check https://zuul.opendev.org/t/openstack/build/108e3c9790d04b5d86ca93e0a1da3b26 : SUCCESS in 3m 08s\n- grenade https://zuul.opendev.org/t/openstack/build/4f3fdaadb42f4b93acdec3ddfa3c6b26 : SUCCESS in 1h 12m 37s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/0a671e8d93a241369fd42dc66a852180 : SUCCESS in 1h 07m 17s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/358dc86dfc404df285099a15d2b0fb37 : SUCCESS in 1h 45m 51s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/7095b9aac37c4d9ba8a70d1a3f81f2e7 : SUCCESS in 1h 04m 40s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8a4c5853167c49e2900e6c6573190481 : SUCCESS in 3m 29s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/85048c8ad4c949cdb2afe2acfc6eed4a : SUCCESS in 7m 33s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/aa41a823418c482e9bd1c90def83e614 : SUCCESS in 3m 30s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/1af0ebae22c648fd9c4569a430627788 : SUCCESS in 8m 37s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3c75593432434bc2b38f26553fa38a96 : SUCCESS in 5m 42s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e8259961fa2248fcbd303f3ca957583c : SUCCESS in 9m 11s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/fd12fbc5b9394569968215eb2a14e801 : SUCCESS in 7m 59s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/ef5b4dba050d4e959dd87aee51d9e78a : SUCCESS in 6m 15s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/112dc0a910f34dd3833636e1b7a51722 : SUCCESS in 2h 02m 48s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/0ae169bb6cc241afa7ca4075db1e0f3a : SUCCESS in 2h 05m 53s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/dd1dfdb8317e432d8f190f127d36cde3 : SUCCESS in 1h 23m 53s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/2bc4eb465d9242a098642dde33ffb8b2 : SUCCESS in 2h 22m 18s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/0d65ef62aea74d8483129378ca12a7fc : SUCCESS in 1h 06m 18s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/ffcbe0462b004d84b1950fe07d7380d0 : SUCCESS in 1h 47m 12s\n- glance-grenade-centralized-cache https://zuul.opendev.org/t/openstack/build/db4541dc3f764a0db8bb59b6bf25ed49 : FAILURE in 12m 04s (non-voting)\n- glance-s3-multistore https://zuul.opendev.org/t/openstack/build/aae6858c75fc483eaeffd95063fb15fa : SUCCESS in 1h 50m 26s (non-voting)\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/c8b23d8d6b504f3f9258a1fe9ba63f6f : FAILURE in 40m 49s","accounts_in_message":[],"_revision_number":2},{"id":"e328b7e5665d2c2cbe0025d9e57e2dc71622e787","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2026-05-05 09:31:08.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"aa501820aebe7ee7cdbe17b8afa76628b7417433","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-05 09:49:46.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2cd7e32aacbc4d98aea3b296bf205cfe\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/c1fdb4b21a514a969b15836050ea0661 : FAILURE in 4m 54s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/61d10aaaff0c43b49b8cbae8a7eb4836 : SUCCESS in 7m 05s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/7cdc3c91cfe44fd2b175590d0bc64dae : SUCCESS in 8m 21s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"2e49f80f5aad18f4e00dee4a7595c189fc141fcd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-05 11:38:46.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/627c7e0490a94b31a6635ebde5299af3\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/060bbb2fe4424b2098baa8c368e4923f : SUCCESS in 2m 38s\n- requirements-check https://zuul.opendev.org/t/openstack/build/8a057a3ed3da49be8d8bf0c2386da0f4 : SUCCESS in 1m 55s\n- grenade https://zuul.opendev.org/t/openstack/build/99588a8e0eaf43e19e8f3e0c96b1001f : SUCCESS in 1h 02m 25s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/a9d982dbfd9b4cb480f6dd846bef6f71 : SUCCESS in 46m 36s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/332e853eb8514fc2be6c569aaeddb0f9 : SUCCESS in 1h 22m 39s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/87fb4b1007604699b134bdc75d2cddd3 : SUCCESS in 1h 01m 55s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cdbb7f0bb30842d09aa43130bd904cc7 : SUCCESS in 6m 13s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/2846861bd942458bb647dc87a3ef52b6 : SUCCESS in 4m 47s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/afe31b61dac94eba99ace205c693e093 : SUCCESS in 3m 19s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/e8b119ae350b4dfb9ad362aace673aba : SUCCESS in 7m 13s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0891eaee771b470a858d827aca0f0971 : SUCCESS in 5m 47s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/fb111e82084a4496b173924d3478c4ec : SUCCESS in 7m 40s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/f815073eb3f7489fb8acb45dbe6ff0e0 : SUCCESS in 6m 48s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/9752dd9fb70845ce8badbe70662ccce2 : SUCCESS in 5m 59s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/3920da3aa04d49078623beec42727a92 : SUCCESS in 1h 37m 54s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/654814d5cb684889b8c96033a42a5972 : SUCCESS in 1h 57m 26s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/7f1abd04a3584f6da166b30fed114110 : SUCCESS in 2h 01m 03s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/479cd3e191a7427d8c1805a2aa0aba55 : SUCCESS in 1h 19m 21s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/b2635e2e23744dbb8c7f89b86a1d4aec : SUCCESS in 1h 10m 32s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/aa9d0d861ccf4868821e62e9ef93eda5 : SUCCESS in 1h 18m 02s\n- glance-grenade-centralized-cache https://zuul.opendev.org/t/openstack/build/f1df043fe085456db7407664a81e7e13 : FAILURE in 13m 49s (non-voting)\n- glance-s3-multistore https://zuul.opendev.org/t/openstack/build/d7cd284c90964d3a9b2c94c6087a6310 : SUCCESS in 1h 49m 59s (non-voting)\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/6fdb7cb9eeed4980a2981b1f935413de : SUCCESS in 17m 50s","accounts_in_message":[],"_revision_number":2},{"id":"db0ebc9d555a628557436bea5b4773d97cd23022","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-05 11:39:35.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"e9b17ff2647c5eef217be709e5b31db665436a68","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-05 13:55:01.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/04056bae2bf8474aa845c909f94280a4\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/6385c8b77a8642f0b95db2265577a004 : SUCCESS in 2m 41s\n- requirements-check https://zuul.opendev.org/t/openstack/build/52dcfbf0dfe640ba8fe862feee63674d : SUCCESS in 2m 47s\n- grenade https://zuul.opendev.org/t/openstack/build/7618eb3d76c642e6bfa8c6fe0934cbd3 : SUCCESS in 1h 06m 37s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/195eb515dab2446fb8451ee609279d27 : SUCCESS in 1h 07m 59s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/a62ea8271cde4ed7947ab70c540b0b87 : SUCCESS in 1h 17m 42s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/82987906a752418db1bd40eed337c569 : SUCCESS in 51m 52s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8281236cf6bd4a1091143ce03097f32c : SUCCESS in 4m 43s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/adc2888ef3914bc6aea533d98bdebaf5 : SUCCESS in 4m 57s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/1adf276db6444aea89208f0029638df7 : SUCCESS in 5m 06s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bbb766998a8d406c99598d61092613b3 : SUCCESS in 5m 59s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/eda3b311d6094119b12ab2732cb58fda : SUCCESS in 4m 00s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/1b13c78dbea246c69f3370eecde9e09a : SUCCESS in 4m 24s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/4129007b1cc24a2ab54b38bd59a641be : SUCCESS in 5m 51s\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/2c9e94a93d004abcb280ed0adf17bcc8 : SUCCESS in 2h 09m 37s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/55b2205cb88c49abb731e806e87d9c7b : SUCCESS in 2h 02m 39s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/8aed4d584ac54ce891f133a69658a287 : SUCCESS in 56m 31s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/9c85eb7b14ca4358ab0a32abbc1dd52e : SUCCESS in 1h 19m 18s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/3051429b059947138a8975b3ed9ad275 : SUCCESS in 17m 08s","accounts_in_message":[],"_revision_number":2},{"id":"14a0f8f7fd2b63e045087db1fd7f336dac3735ca","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-05 13:55:01.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"80d50001bde39a03ade70528877d6ae03cc8779c","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-05 13:57:50.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/de3a96af04e74c1694b1d8ad0b827808\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c5da8596adea4791afc0ca138aacabea : SUCCESS in 1m 39s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a361f85e56974e3bab4109fa0370bacb : SUCCESS in 1m 05s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"6d48d707a27f8b58eeded2925fd055888eca4f45","revisions":{"d2f93d1d1b55173037fa1e9b8c366d52a8a07ecc":{"kind":"REWORK","_number":1,"created":"2026-05-01 10:32:09.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/87/986887/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/87/986887/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/87/986887/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/87/986887/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/87/986887/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/87/986887/1"}}},"commit":{"parents":[{"commit":"ad10cfd88ea416ffeaea67daf21d03242f9befb5","subject":"Merge \"Remove default to relaxed MBR safety checks\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/ad10cfd88ea416ffeaea67daf21d03242f9befb5"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2026-05-01 10:30:56.000000000","tz":60},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2026-05-01 10:32:01.000000000","tz":60},"subject":"Drop Python 3.10","message":"Drop Python 3.10\n\nPython 3.11 is the minimum runtime for 2026.2 (Hibiscus). We also add a\nmarker for Python 3.14, since this is tested in the gate. We may\neventually wish to drop these markers since they duplicate information\nprovided by `requires_python`.\n\nhttps://governance.openstack.org/tc/reference/runtimes/2026.2.html\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\n\nChange-Id: I4fd96b255790c6a21d9dec99aa84a16555298a9c\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/d2f93d1d1b55173037fa1e9b8c366d52a8a07ecc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/d2f93d1d1b55173037fa1e9b8c366d52a8a07ecc"}]},"branch":"refs/heads/master"},"6d48d707a27f8b58eeded2925fd055888eca4f45":{"kind":"REWORK","_number":2,"created":"2026-05-01 15:55:38.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/87/986887/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/87/986887/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/87/986887/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/87/986887/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/87/986887/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/87/986887/2"}}},"commit":{"parents":[{"commit":"ad10cfd88ea416ffeaea67daf21d03242f9befb5","subject":"Merge \"Remove default to relaxed MBR safety checks\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/ad10cfd88ea416ffeaea67daf21d03242f9befb5"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2026-05-01 10:30:56.000000000","tz":60},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2026-05-01 10:30:56.000000000","tz":60},"subject":"Drop Python 3.10","message":"Drop Python 3.10\n\nPython 3.11 is the minimum runtime for 2026.2 (Hibiscus). We also add a\nmarker for Python 3.14, since this is tested in the gate. We may\neventually wish to drop these markers since they duplicate information\nprovided by `requires_python`.\n\nhttps://governance.openstack.org/tc/reference/runtimes/2026.2.html\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\n\nChange-Id: I4fd96b255790c6a21d9dec99aa84a16555298a9c\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/6d48d707a27f8b58eeded2925fd055888eca4f45"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/6d48d707a27f8b58eeded2925fd055888eca4f45"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{"label:Review-Priority\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
