)]}'
{"id":"openstack%2Fglance~987475","triplet_id":"openstack%2Fglance~master~Id4fbf232915ba2df2e8177968dbac991e10842be","project":"openstack/glance","branch":"master","topic":"eventlet-removal","attention_set":{},"removed_from_attention_set":{"19138":{"account":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"last_update":"2026-05-12 09:58:27.000000000","reason":"\u003cGERRIT_ACCOUNT_19138\u003e replied on the change","reason_account":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"}},"9303":{"account":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"last_update":"2026-05-08 06:34:43.000000000","reason":"\u003cGERRIT_ACCOUNT_9303\u003e replied on the change","reason_account":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"}},"8122":{"account":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"last_update":"2026-05-12 02:13:47.000000000","reason":"\u003cGERRIT_ACCOUNT_8122\u003e replied on the change","reason_account":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"}}},"hashtags":[],"change_id":"Id4fbf232915ba2df2e8177968dbac991e10842be","subject":"Use stdlib socket and ssl in glance.common.client","status":"MERGED","created":"2026-05-06 10:52:00.000000000","updated":"2026-05-12 15:43:51.000000000","submitted":"2026-05-12 15:41:31.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"987475-eventlet-removal","meta_rev_id":"3ba0a017fce550a485743133309c6e9b6ac931ff","_number":987475,"virtual_id_number":987475,"owner":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-05-12 15:41:31.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"all":[{"value":2,"date":"2026-05-08 06:34:43.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":2,"date":"2026-05-12 02:13:47.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"all":[{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":1,"date":"2026-05-12 02:13:47.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Top Priority / Holds Gate"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-05-06 11:39:16.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2026-05-06 13:19:26.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-05-08 05:58:35.000000000","updated_by":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"reviewer":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"state":"REVIEWER"},{"updated":"2026-05-08 05:58:35.000000000","updated_by":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"reviewer":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"state":"REVIEWER"}],"messages":[{"id":"6bcc983af3aeb36a662dc3cc0fd6adea8d518eae","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"date":"2026-05-06 10:52:00.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"ce5b0faf3d2faa7b6b2684311352793114e5ce5b","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-06 11:39:16.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/44fc47cb847440278271b8c5bc7d8704\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/0e6c810046c1434c8e9044b19f232c77 : FAILURE in 5m 57s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/0480177b423e4c9c851f165772ea6ef1 : SUCCESS in 7m 18s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/ed26e66d9036464c8d840a1ec98a16f5 : SUCCESS in 9m 17s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"1f7e8a33349f4fa111d0f31b436aee228012dc78","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-06 13:19:26.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9ba18a17625346bc828de3a3d972f5cd\n\n- grenade https://zuul.opendev.org/t/openstack/build/b4fb39e020714d399ba362e6cea3ace8 : SUCCESS in 1h 10m 03s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/662f062aa3344b4e892d0a0be5f3d852 : SUCCESS in 1h 03m 36s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/b56e43f5617149d9af0c9abd307e826a : SUCCESS in 1h 19m 11s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/0080356a9545434bba4b1e1415adf323 : SUCCESS in 1h 07m 38s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/28af38d727a24a558ec9bcea38646c52 : SUCCESS in 5m 07s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/e930c41bbfc84a4cad7940fd52276e08 : SUCCESS in 5m 00s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/7a9674f2a06a4df49310b527646889c5 : SUCCESS in 5m 17s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/08078d9b6e0c4831a9e18cadb48a1c74 : SUCCESS in 7m 48s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c0baec969c6541ac809a1e961b692912 : SUCCESS in 3m 36s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/73f94359a3d346dca62b9d69316106f1 : SUCCESS in 6m 32s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/4e05738ab98e4619828a1d8f6be73932 : SUCCESS in 7m 35s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/75a45cce2ab84a849a28e84f5268efc5 : SUCCESS in 1h 46m 16s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/2fc8db1ff84c4753b8f1ccc775abd4b3 : SUCCESS in 2h 04m 29s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/0238e00e50654821bf9243381665f79c : SUCCESS in 2h 04m 05s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/127f6d402103453a97dad182f689c297 : SUCCESS in 2h 14m 17s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/322396979db240839ba2c720a142510d : SUCCESS in 59m 39s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/aa02730c55ac42899a190053095506fc : SUCCESS in 1h 52m 14s\n- glance-grenade-centralized-cache https://zuul.opendev.org/t/openstack/build/bf1ea22a306a41c097e5513227912c47 : FAILURE in 12m 27s (non-voting)\n- glance-s3-multistore https://zuul.opendev.org/t/openstack/build/a8e06c38741f4016ad90549b7ba5d171 : SUCCESS in 2h 03m 35s (non-voting)\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/17e82d92254b45e8a5554a0eb50fce38 : SUCCESS in 16m 47s","accounts_in_message":[],"_revision_number":1},{"id":"9e75bed21eb88af0deb8f92a861f24e4a7db2f88","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2026-05-08 06:34:43.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"88fc8548ef7812f1db6d06080b501f04c456dd93","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2026-05-08 06:35:08.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"b9afa799d2dfd2a91ec84b75ad4992b8110ef387","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"date":"2026-05-08 08:34:26.000000000","message":"Topic set to eventlet-removal","accounts_in_message":[],"_revision_number":1},{"id":"df01919d5a1123ca7f96907fd304bc355d30ca2d","author":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"date":"2026-05-12 02:13:47.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"0d66f4222d06b02e1eaa00981c16fc3da0fdc7b0","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-12 02:21:22.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"244dd147d113358b28aa6b4715fd6749eb01c651","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-12 04:47:43.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/92e621056d92496aa83e869577432dc5\n\n- grenade https://zuul.opendev.org/t/openstack/build/b0eb06a0ab114862b34c33a6c08d534e : SUCCESS in 1h 06m 28s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/9729bf092d8046dcb47ba85a9252e7c8 : SUCCESS in 56m 53s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/484d552e3727404bb2fac8e2ee4ec85f : SUCCESS in 1h 29m 52s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/ecb65ba80a35408db872122cc9adcf53 : FAILURE in 1h 06m 17s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4c09346728be48ad8893e1c5f49acc07 : SUCCESS in 4m 59s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/ebeefaae1177485ca972de7955e11310 : SUCCESS in 5m 12s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/0cb3644f6f664a1789d2c73fa782c8ab : SUCCESS in 5m 16s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8197655f5321425984a93cb77079a718 : SUCCESS in 5m 20s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/5902b3bf72bd4c969fa5da7709657b76 : SUCCESS in 6m 35s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/af816abca0df4baba6bd4a341268f011 : SUCCESS in 7m 47s\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/9e4d30b52e03477f80b1a3f8aa0bfb64 : SUCCESS in 2h 07m 49s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/eaabccb403aa4deba4c795eec2ecb64d : SUCCESS in 1h 52m 08s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/b3f38f218f764d9fa77de620d6824ee2 : SUCCESS in 36m 24s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/a0a9b09b026d49b687819b016b709bf8 : SUCCESS in 1h 11m 02s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/7bdcc96140384974bae0db2fbb0819d4 : SUCCESS in 36m 30s","accounts_in_message":[],"_revision_number":1},{"id":"8ede81899ef180e79d694f910ea59ec149f4f05f","author":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"date":"2026-05-12 09:58:27.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"82fe012dd6510c06ec3ed6aa22dd2f6f2d6b7dfd","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-12 10:56:30.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/54502daa08c94f34b6099f65e8a3310f\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/7c0a1dc43eaf4136b79b2babdcb042c6 : FAILURE in 6m 36s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/b4ebff78b86b41e99aed3830c1d86819 : SUCCESS in 7m 01s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/49587285d6e348148244595bc62d32c4 : SUCCESS in 12m 00s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"73fd6105153e171fd6c9d9eefe1afaf8c5b76f1f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-12 12:12:20.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f3a687d47bb649b3937a2c007dd083f0\n\n- grenade https://zuul.opendev.org/t/openstack/build/ce1cfe48ecc14998bef1d9233c95b9a9 : SUCCESS in 1h 07m 18s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/1798f7ac1c7e4b10a18ea0072f5e86a2 : SUCCESS in 1h 02m 14s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/92f49d7a109148c687082a5129b30a63 : SUCCESS in 1h 40m 45s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/c451bbb6b2d842f4aa7889f4a5d05944 : SUCCESS in 1h 06m 04s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/14a37555e90844338895468334ea0eab : SUCCESS in 5m 07s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/84c83a6f1a3845018d3a7e285b7eea13 : SUCCESS in 3m 37s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/11a1fef691ac4aef90dff993cc84e0ab : SUCCESS in 5m 07s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/cc6ae69834074b6e9da0a79692d2451c : SUCCESS in 5m 06s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4d9e311a496f445698c4e6d6c576ed62 : SUCCESS in 5m 45s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/3753898d2c7847189942c5a71e4a0b84 : SUCCESS in 7m 40s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/f79ce67cc0454db8bf0ab3ad811067c1 : SUCCESS in 4m 11s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/4fd847abf7004bd296855c899298f129 : SUCCESS in 1h 49m 59s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/458d352b66794890bd70e6f5d3a9aaa8 : SUCCESS in 2h 02m 25s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/0d62af6b88d74c6e858c96a680588fd9 : SUCCESS in 2h 09m 29s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/645de83601354cdaac7cc61e3d3a124c : SUCCESS in 1h 58m 38s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/123c0ec42b384d949205397b2d1ca147 : SUCCESS in 1h 10m 43s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/43f0ef2af5fc40e6acb224d54e8b238b : SUCCESS in 1h 52m 31s\n- glance-grenade-centralized-cache https://zuul.opendev.org/t/openstack/build/661f6dc3047242d880b569063a73471f : FAILURE in 12m 23s (non-voting)\n- glance-s3-multistore https://zuul.opendev.org/t/openstack/build/d32fe4b6810e4a848284fd869106afe6 : SUCCESS in 1h 56m 25s (non-voting)\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/61b3a302fb79407184637068b580f2f6 : SUCCESS in 21m 11s","accounts_in_message":[],"_revision_number":1},{"id":"168491103eacdb2d83f1fb8e9dd450a50bb9e7d3","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-12 12:12:55.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"25de1186ea21e74ae3481d5cbfaa910c1c8b34e7","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-12 15:41:31.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/635e9532567a45029fc4dddbe8731b7c\n\n- grenade https://zuul.opendev.org/t/openstack/build/1ea96d2331a84e34b28f8f6d04e35c25 : SUCCESS in 1h 04m 23s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/934ee23b416d4c1f9656cd633b6067f3 : SUCCESS in 45m 27s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/e45a9cbf60834334a5144fb59bd2e907 : SUCCESS in 1h 48m 42s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/6789bb2ef2e04205842e93520bb6fbdb : SUCCESS in 41m 57s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2ee3ff23466144f19e4d2639aaf8d026 : SUCCESS in 4m 47s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/4b123f953d814717a4306966bf809bfc : SUCCESS in 5m 04s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/52b400480da141a79a399fad8a3e4e16 : SUCCESS in 4m 32s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4f1b506788f94708b3823092fc4e9e41 : SUCCESS in 6m 58s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/c2b6f3dcc5264c8cb03fad8d8901f970 : SUCCESS in 6m 45s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/07aa619a89d14f97b360a43f695d916c : SUCCESS in 8m 09s\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/c99f76e5d81b4b65883a409679af79de : SUCCESS in 2h 25m 39s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/1a258b4f94774b4b984e31544c17b1ad : SUCCESS in 2h 00m 37s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/7e35a8cffc204c458a0a354463a41e0d : SUCCESS in 1h 05m 44s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/c0d6597c2534435f9dea357cabb5738b : SUCCESS in 1h 04m 39s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/93d0c9fad05443a2ab7ede80a3a1312e : SUCCESS in 26m 10s","accounts_in_message":[],"_revision_number":1},{"id":"9a45a9303675f6ec583c23fd4073af53280a4608","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-12 15:41:31.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"3ba0a017fce550a485743133309c6e9b6ac931ff","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-12 15:43:51.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c857a9b3f4ae431daf842327a6ad7785\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c25f06bbcd9f4902b7207c55fc084301 : SUCCESS in 1m 49s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"e22c20bf79abcc510adfd0fe353f6b727fbc1ffa","revisions":{"e22c20bf79abcc510adfd0fe353f6b727fbc1ffa":{"kind":"REWORK","_number":1,"created":"2026-05-06 10:52:00.000000000","uploader":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"ref":"refs/changes/75/987475/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/75/987475/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/75/987475/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/75/987475/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/75/987475/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/75/987475/1"}}},"commit":{"parents":[{"commit":"6d48d707a27f8b58eeded2925fd055888eca4f45","subject":"Drop Python 3.10","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/6d48d707a27f8b58eeded2925fd055888eca4f45"}]}],"author":{"name":"Pranali Deore","email":"pdeore@redhat.com","date":"2026-05-06 10:50:55.000000000","tz":0},"committer":{"name":"Pranali Deore","email":"pdeore@redhat.com","date":"2026-05-06 10:51:41.000000000","tz":0},"subject":"Use stdlib socket and ssl in glance.common.client","message":"Use stdlib socket and ssl in glance.common.client\n\nDrop the eventlet.green socket/ssl import path and always use the\nstandard library. The HTTP client no longer switches to greenthread\nwrappers when eventlet is present, which matches WSGI/native-thread\ndeployments and the eventlet-removal effort.\n\nAssisted-By: Cursor (claude-4.5-sonnet)\nChange-Id: Id4fbf232915ba2df2e8177968dbac991e10842be\nSigned-off-by: Pranali Deore \u003cpdeore@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/e22c20bf79abcc510adfd0fe353f6b727fbc1ffa"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/e22c20bf79abcc510adfd0fe353f6b727fbc1ffa"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{"label:Review-Priority\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
