)]}'
{"id":"openstack%2Fglance~988853","triplet_id":"openstack%2Fglance~master~Ib0e13e6ad5021612da065db793775b0422f15115","project":"openstack/glance","branch":"master","attention_set":{},"removed_from_attention_set":{"19138":{"account":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"last_update":"2026-07-01 16:49:58.000000000","reason":"Change was submitted"},"9303":{"account":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"last_update":"2026-07-01 14:36:42.000000000","reason":"\u003cGERRIT_ACCOUNT_9303\u003e replied on the change","reason_account":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"}},"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2026-07-01 16:49:58.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ib0e13e6ad5021612da065db793775b0422f15115","subject":"Validate enabled_import_methods","status":"MERGED","created":"2026-05-16 16:53:21.000000000","updated":"2026-07-01 16:51:16.000000000","submitted":"2026-07-01 16:49:58.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":4,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"988853","meta_rev_id":"242c9e0aab33bd18099fab9e970c406d5145a970","_number":988853,"virtual_id_number":988853,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-07-01 16:49:58.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"all":[{"value":0,"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},{"value":2,"date":"2026-07-01 14:36:42.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":2,"date":"2026-07-01 13:28:13.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"all":[{"value":0,"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},{"value":1,"date":"2026-07-01 14:36:42.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Top Priority / Holds Gate"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-05-16 17:37:41.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2026-05-16 19:12:25.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-05-18 14:38:11.000000000","updated_by":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"reviewer":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"state":"REVIEWER"},{"updated":"2026-05-29 07:19:29.000000000","updated_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"reviewer":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"state":"CC"},{"updated":"2026-07-01 13:28:13.000000000","updated_by":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"reviewer":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"state":"REVIEWER"},{"updated":"2026-07-01 14:36:42.000000000","updated_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"reviewer":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"state":"REVIEWER"}],"messages":[{"id":"3d52191d63cb50434c9c88aa9a290ac7bc987657","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2026-05-16 16:53:21.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"d0b7c2b92e843d5865f7622672588743c7a871c2","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-16 17:37:41.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ab8c67f1366746bc830cf90fb1829236\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/a106d5bf793d4d0591b6b1413ace29d4 : FAILURE in 7m 13s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/24bbeb7536664968bc318ecb3fa34ed9 : FAILURE in 8m 57s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/80ed3e2524cf41a0a019111b03ed0ecf : FAILURE in 11m 52s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"226a8aae2dff166c8955c7a90959606e0df5ed0f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-16 19:12:25.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/728fe2ec24b340fab20396bcb74af4d0\n\n- grenade https://zuul.opendev.org/t/openstack/build/1719ec9fa4d84fafa6a9b1f62f564d44 : SUCCESS in 1h 02m 22s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/018e23494b41464e83f27e6d1fc1c0a1 : SUCCESS in 48m 18s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/ea81967354684c95bc6d16bebd6db13d : SUCCESS in 1h 36m 03s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/a055c41b8c2a481a85d2bd4e62322bf5 : SUCCESS in 51m 26s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1e04623f495d4fcb9ad947d1f7fe5db5 : SUCCESS in 5m 28s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/0781deda4a0343ebbb76a616556b2e3e : FAILURE in 3m 16s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/fe7004273f0f46cb807adf394208f41f : FAILURE in 3m 37s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/3c5e0fd02f604a8bb3c9d300ecbca3ce : FAILURE in 7m 27s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/081382a461974dd88ab3edca566f539d : SUCCESS in 5m 28s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b2de401a1d644efd997630c48f1bede1 : SUCCESS in 7m 42s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/d9d569c9298042bf9299d29949ad6e34 : SUCCESS in 8m 15s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/0f20efa6d38448058a0859ce4d735d02 : SUCCESS in 6m 14s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/efdf5ad373254555b74c4e95e34989f7 : SUCCESS in 1h 47m 24s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/e1ca9b7907b14106a5d4479a632081b5 : SUCCESS in 2h 10m 49s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/b345c9faeb544d38941cad6da35cea34 : SUCCESS in 1h 37m 44s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/d06b04649a4a4282a8fd907981c63e7a : FAILURE in 2h 12m 48s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/7fac3f8bbcf94121af9ea535def26647 : SUCCESS in 55m 53s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/c5422e9d370c41aebd6791fd6c7c70ea : SUCCESS in 56m 47s\n- glance-grenade-centralized-cache https://zuul.opendev.org/t/openstack/build/0079e0e7af164c0c95ff2ecad2ef6197 : FAILURE in 6m 58s (non-voting)\n- glance-s3-multistore https://zuul.opendev.org/t/openstack/build/f95cd5eddb674f4d882138fe603c86e9 : SUCCESS in 1h 11m 56s (non-voting)\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/c3cbde17e3184c189161e6063583fede : SUCCESS in 35m 36s","accounts_in_message":[],"_revision_number":1},{"id":"6e75191d95bdf4995b9f3ea0d04b9d374acc2739","author":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"date":"2026-05-18 14:38:11.000000000","message":"Patch Set 1: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"377739a761593f07670f97fbd5f1198d4fcadb86","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2026-05-18 15:01:43.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":2},{"id":"012ea9a77f7b6fe31680938d5dd24c5c1accc875","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2026-05-18 15:06:36.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"961b8bde6e0661355efc5bba537b9607e20dd71a","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-18 15:21:44.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/36a6d942737f488eaad137c260fefe1f\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/e391903fbf0144dfacf182c2de4c01d3 : SUCCESS in 10m 24s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/d3dbfcf72a444bc59b92ffb0b9cb43b4 : SUCCESS in 7m 11s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/15451719069d4c858ba475a59aad1abb : SUCCESS in 17m 12s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"45fc745f1d552df0a0644217b9b2c46143ff30f3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-18 17:40:59.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5efe1df638504af2aa02b885bf155b55\n\n- grenade https://zuul.opendev.org/t/openstack/build/c236efc1aec144dc9c136367942e12cd : SUCCESS in 45m 19s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/0124ecea91e9495593700162c71b9ea8 : SUCCESS in 1h 03m 22s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/b94396cc36294757b5083a77372419a3 : SUCCESS in 1h 19m 40s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/c3a4547d8b05414d8c53700eedbeb700 : SUCCESS in 1h 06m 18s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8c5d160c01de4f8f9dd588fd5ddd3102 : SUCCESS in 4m 48s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/5fbc07313e344113a29ba003b8ca0d73 : SUCCESS in 5m 12s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/786324ee9c6245629ae5d1662aa888b8 : SUCCESS in 4m 40s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/1b2cdd5356cb435db1b25686b9784b5a : SUCCESS in 7m 38s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/79e9dd3037864090914e5fc6318c51bf : SUCCESS in 6m 20s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/28fdc45690f84a3dbce9176327428f80 : SUCCESS in 10m 20s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/d55be8b53a2448569c46c2eebb046009 : SUCCESS in 6m 38s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/6e8538b0a9ac4c47b1f9319e8052c026 : SUCCESS in 8m 08s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/73f59bb0726b4d7aa9f0cadecd6a31d9 : SUCCESS in 1h 42m 07s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/c421b92fbde148718297650ffd365f53 : SUCCESS in 1h 46m 32s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/789fc667f00242ee99d875e357acbc3f : SUCCESS in 2h 28m 33s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/a04441820d6843249ab34ab9221ede4f : SUCCESS in 1h 28m 47s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/4ead72c764cf4522a31c48ca00c31787 : SUCCESS in 46m 54s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/ba470bcc2d604e3bbe1d0ed4630c65ad : SUCCESS in 2h 04m 21s\n- glance-grenade-centralized-cache https://zuul.opendev.org/t/openstack/build/45fef91740d940aba4aeea751ccfa21e : FAILURE in 15m 20s (non-voting)\n- glance-s3-multistore https://zuul.opendev.org/t/openstack/build/4aef9bcc6472414d95f9d64be0d1d211 : SUCCESS in 2h 04m 18s (non-voting)\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/cb1c38f4445c42f69d441d3ce1fe1b04 : SUCCESS in 33m 54s","accounts_in_message":[],"_revision_number":2},{"id":"91ae389ff18df8bfcb8345739571f597f64d5d88","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2026-05-29 07:19:29.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"3ad749d56693626f9b7f08825b97adf89f740cb8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2026-06-25 09:07:11.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":3},{"id":"04082dfad97e441646f617bbfaefe2a1e7f8f2b7","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2026-06-25 09:07:38.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"56f0d8ed2bcfce9066d894dbb0be12d0444a68cb","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-25 09:22:03.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8271640581f94de4a8cd62fbe366112e\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/73182775c5ac4f17a107e45768d75756 : SUCCESS in 7m 00s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/7fe0dd7942ff49d685408a7620ce977f : SUCCESS in 7m 54s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/4e2f83bfdfb44a8ab86cabd857aacd2c : SUCCESS in 12m 41s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"c5b5df83a270d6b6d110aafbc96e55ac3cf1ae64","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-25 12:03:55.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/aaa075e241d444918ec65cd13a46ba82\n\n- grenade https://zuul.opendev.org/t/openstack/build/a2da023ede284c24a58a16eccdfc5c25 : SUCCESS in 35m 50s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/d4b64f29c36349208fce7a0b59372438 : SUCCESS in 54m 15s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/b5eb750a92bc4088828715ea23a38b54 : SUCCESS in 1h 49m 25s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/ec999a198962460dabc591033377da4e : SUCCESS in 1h 12m 00s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a59bde85dc5141d480a384e61b8e5929 : SUCCESS in 5m 00s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/383d67e17c7a4d4aa77d9dd53d68926f : SUCCESS in 7m 01s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/29e64d917f894c8a91f6e6b6646896d8 : SUCCESS in 4m 58s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/4975fd35893b43a084f83157aae74868 : SUCCESS in 8m 21s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/38810d0bcbca48398022ba5dc3ff53b5 : SUCCESS in 7m 54s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/af283c5c665b4a0eb24d2e28f53bd2a6 : SUCCESS in 10m 28s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/b5784df7a2aa478da65f8a8efd302719 : SUCCESS in 9m 09s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/e8ea4e89b9614210913be294108b81e6 : SUCCESS in 6m 05s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/2daec537ddc04f7c803b6293dd7ad970 : SUCCESS in 1h 54m 35s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/a61cca37186b4cf3b0bc44d7cfd409c3 : SUCCESS in 1h 58m 54s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/4c6935c7005c4a66976e59e5d46834d2 : SUCCESS in 1h 15m 24s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/1bf51d35033f4830beacbf3302b3f32e : SUCCESS in 2h 50m 40s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/c2a247fc8bd44552bcba7c708e4a1b16 : SUCCESS in 33m 57s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/15159e78f8494ade89c8decad99bee72 : SUCCESS in 1h 45m 12s\n- glance-grenade-centralized-cache https://zuul.opendev.org/t/openstack/build/c1ff82b6772c4d18a839426afb3fa4d1 : FAILURE in 8m 04s (non-voting)\n- glance-s3-multistore https://zuul.opendev.org/t/openstack/build/16d3bf6b58f9466bba2b08879d880166 : SUCCESS in 1h 09m 40s (non-voting)\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/5237dd48a884495192782a165a2a0acb : SUCCESS in 19m 09s\n- glance-rally-task-uwsgi https://zuul.opendev.org/t/openstack/build/7d6f603a891f430daa2035ee575b137e : SUCCESS in 51m 45s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"77406d4c0121593a1f78d51b5c2ced3bb25d789b","author":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"date":"2026-07-01 13:28:13.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"f8a4ce92e5e2258efce9c7736dbd65fbb9799af2","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2026-07-01 14:36:42.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"cf8573fc5d21d0fcdfb9da54048fa1f24deeb8e8","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-01 14:37:39.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"15d07046de2391e4895a423ad02621c20e1a187a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-01 16:49:58.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ebe8d600273c40c68fe5ff023481265f\n\n- grenade https://zuul.opendev.org/t/openstack/build/6b0155fdbe45472f915825f96ffd19dc : SUCCESS in 1h 09m 18s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/61f4d2650e9f4ab3be007ed93f7e21de : SUCCESS in 34m 12s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/45605942906b408c977a305153c00c26 : SUCCESS in 1h 50m 46s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/b71af0f8c4b74aa89876e4fa1d180126 : SUCCESS in 1h 07m 39s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/67d604b5d69a4b1abc4443be003aec58 : SUCCESS in 3m 15s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/49246a1c97b74915952f2d20c40ba95e : SUCCESS in 4m 50s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/86ed6f9e9fff48ecad26f1207ad3230e : SUCCESS in 5m 39s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e2f6b9b83ffc48089ea343f337137139 : SUCCESS in 3m 37s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/42c8e5a32ad848d39be24a3fb514679f : SUCCESS in 8m 47s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/08c5e137fbc7483ba5cfbab1001c2fee : SUCCESS in 5m 56s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/f4359d993481458eadd96725b95d843c : SUCCESS in 6m 59s\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/c5e56cd4e2cc41f880f1e0dd3c6ba3cb : SUCCESS in 2h 07m 49s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/6721b3d93aef4177af43dea5596d51ed : SUCCESS in 2h 10m 34s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/f3dc34538efc4d5b95d0f127d22d5465 : SUCCESS in 51m 08s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/85d6a9656ac6418ca4ef89e7f84a3082 : SUCCESS in 1h 06m 28s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/c52fb3b03b064c4797843f5d14c2aa55 : SUCCESS in 41m 30s","accounts_in_message":[],"_revision_number":3},{"id":"ddb6ec647e019f68926009a4936970ef3ed1aeea","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-01 16:49:58.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3},{"id":"242c9e0aab33bd18099fab9e970c406d5145a970","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-01 16:51:16.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/bb650df0d99140a794f6590829e8f4d0\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/710c352f32864945a10a0d210c6dc972 : SUCCESS in 1m 02s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"b2648b3a6efcfb6a9ae816604904477c1b026389","revisions":{"13cb03c4a866e8f8735d0c913ed19ffd4640319a":{"kind":"REWORK","_number":1,"created":"2026-05-16 16:53:21.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/53/988853/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/53/988853/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/53/988853/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/53/988853/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/53/988853/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/53/988853/1"}}},"commit":{"parents":[{"commit":"03dbf8eac63a35aa0207eba21a3c33e0aa51bb18","subject":"Replace eventlet.sleep with time.sleep in tasks API test","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/03dbf8eac63a35aa0207eba21a3c33e0aa51bb18"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-05-16 16:51:21.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-05-16 16:53:14.000000000","tz":540},"subject":"Validate enabled_import_methods","message":"Validate enabled_import_methods\n\n... to reject unsupported values actually.\n\nChange-Id: Ib0e13e6ad5021612da065db793775b0422f15115\nSigned-off-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/13cb03c4a866e8f8735d0c913ed19ffd4640319a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/13cb03c4a866e8f8735d0c913ed19ffd4640319a"}]},"branch":"refs/heads/master"},"cbcb230d35e0fd647e5dd70266d07c2a0a416999":{"kind":"REWORK","_number":2,"created":"2026-05-18 15:01:43.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/53/988853/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/53/988853/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/53/988853/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/53/988853/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/53/988853/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/53/988853/2"}}},"commit":{"parents":[{"commit":"03dbf8eac63a35aa0207eba21a3c33e0aa51bb18","subject":"Replace eventlet.sleep with time.sleep in tasks API test","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/03dbf8eac63a35aa0207eba21a3c33e0aa51bb18"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-05-16 16:51:21.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-05-18 15:01:14.000000000","tz":540},"subject":"Validate enabled_import_methods","message":"Validate enabled_import_methods\n\n... to reject unsupported values actually.\n\nChange-Id: Ib0e13e6ad5021612da065db793775b0422f15115\nSigned-off-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/cbcb230d35e0fd647e5dd70266d07c2a0a416999"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/cbcb230d35e0fd647e5dd70266d07c2a0a416999"}]},"branch":"refs/heads/master"},"b2648b3a6efcfb6a9ae816604904477c1b026389":{"kind":"REWORK","_number":3,"created":"2026-06-25 09:07:11.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/53/988853/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/53/988853/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/53/988853/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/53/988853/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/53/988853/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/53/988853/3"}}},"commit":{"parents":[{"commit":"03dbf8eac63a35aa0207eba21a3c33e0aa51bb18","subject":"Replace eventlet.sleep with time.sleep in tasks API test","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/03dbf8eac63a35aa0207eba21a3c33e0aa51bb18"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-05-16 16:51:21.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-06-25 09:07:02.000000000","tz":540},"subject":"Validate enabled_import_methods","message":"Validate enabled_import_methods\n\n... to reject unsupported values actually.\n\nChange-Id: Ib0e13e6ad5021612da065db793775b0422f15115\nSigned-off-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/b2648b3a6efcfb6a9ae816604904477c1b026389"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/b2648b3a6efcfb6a9ae816604904477c1b026389"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{"label:Review-Priority\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
