)]}'
{"id":"openstack%2Fglance~988890","triplet_id":"openstack%2Fglance~master~I56cad23b91707608edd3b3083729ea9a653c9e04","project":"openstack/glance","branch":"master","attention_set":{"8122":{"account":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"last_update":"2026-10-01 17:55:08.000000000","reason":"\u003cGERRIT_ACCOUNT_10910\u003e replied on the change","reason_account":{"_account_id":10910,"name":"Stig Telfer","email":"stig.openstack@telfer.org","username":"stigtelfer"}}},"removed_from_attention_set":{"10910":{"account":{"_account_id":10910,"name":"Stig Telfer","email":"stig.openstack@telfer.org","username":"stigtelfer"},"last_update":"2026-10-01 17:55:08.000000000","reason":"\u003cGERRIT_ACCOUNT_10910\u003e replied on the change","reason_account":{"_account_id":10910,"name":"Stig Telfer","email":"stig.openstack@telfer.org","username":"stigtelfer"}}},"hashtags":[],"change_id":"I56cad23b91707608edd3b3083729ea9a653c9e04","subject":"Add guide on image store migration","status":"NEW","created":"2026-05-17 19:53:18.000000000","updated":"2026-10-01 18:05:46.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":true,"submittable":false,"total_comment_count":16,"unresolved_comment_count":1,"has_review_started":true,"meta_rev_id":"3097075caf6d1f18eee4080a7a4c688ffd7f8847","_number":988890,"virtual_id_number":988890,"owner":{"_account_id":10910,"name":"Stig Telfer","email":"stig.openstack@telfer.org","username":"stigtelfer"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"tag":"autogenerated:zuul:check","value":1,"date":"2026-10-01 18:05:46.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Top Priority / Holds Gate"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-05-17 20:07:11.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-06-24 14:55:40.000000000","updated_by":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"reviewer":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"state":"REVIEWER"}],"messages":[{"id":"1b0dea351d476e0b9b162595cd635634b07c6f93","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":10910,"name":"Stig Telfer","email":"stig.openstack@telfer.org","username":"stigtelfer"},"date":"2026-05-17 19:53:18.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"83bf4044ef6651b04e81d112a452e995f0ed949e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-17 20:07:11.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4cb22291393b461a9c27e18907d6a5ef\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/abc61c19cba445ab9952a5ed2fef326c : SUCCESS in 5m 50s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b6cdbbd05ea741899541b5dac06cb937 : SUCCESS in 4m 46s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/bb6e0cd1159c4ad5a4f7e76c8e6c9e59 : SUCCESS in 7m 48s","accounts_in_message":[],"_revision_number":1},{"id":"6106b94fd5e8c283fbb05b6f0297a81a8d3d11dc","author":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"date":"2026-06-24 14:55:40.000000000","message":"Patch Set 1: Code-Review-1\n\n(8 comments)","accounts_in_message":[],"_revision_number":1},{"id":"bd23306ccfe4b846469743b22017fbefc7d1f0c1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":10910,"name":"Stig Telfer","email":"stig.openstack@telfer.org","username":"stigtelfer"},"date":"2026-10-01 17:54:23.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":2},{"id":"8c0903f1ad1f7b26d68784d79ae7990b15ba7433","author":{"_account_id":10910,"name":"Stig Telfer","email":"stig.openstack@telfer.org","username":"stigtelfer"},"date":"2026-10-01 17:55:08.000000000","message":"Patch Set 1:\n\n(8 comments)","accounts_in_message":[],"_revision_number":1},{"id":"3097075caf6d1f18eee4080a7a4c688ffd7f8847","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-10-01 18:05:46.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3ad1bb4ac02f4e4abfcf1b093064e469\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3920c1f8b59e402fb9e80e2ac656d18c : SUCCESS in 4m 35s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d922e0625f7b498bb76242cb0de562dd : SUCCESS in 6m 03s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e1324949947b4ae485903dd0f23c72fb : SUCCESS in 4m 39s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"12ae9ea9295fad1b6b82d4aae756a4008142bdc5","revisions":{"0aca067ba4596ce113ca94909fd5c954c0bc9caa":{"kind":"REWORK","_number":1,"created":"2026-05-17 19:53:18.000000000","uploader":{"_account_id":10910,"name":"Stig Telfer","email":"stig.openstack@telfer.org","username":"stigtelfer"},"ref":"refs/changes/90/988890/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/90/988890/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/90/988890/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/90/988890/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/90/988890/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/90/988890/1"}}},"commit":{"parents":[{"commit":"eac2fa47f26da3515c7a1e8c91226750517c52d4","subject":"Merge \"Remove usage of eventlet for image cache SQLite timeouts\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/eac2fa47f26da3515c7a1e8c91226750517c52d4"}]}],"author":{"name":"Stig Telfer","email":"stig@stackhpc.com","date":"2026-05-17 19:51:10.000000000","tz":60},"committer":{"name":"Stig Telfer","email":"stig@stackhpc.com","date":"2026-05-17 19:51:10.000000000","tz":60},"subject":"Add guide on image store migration","message":"Add guide on image store migration\n\nAn process is described here which should be helpful for admins\nlooking to move images between backend stores, for example as part\nof a storage migration process.\n\nChange-Id: I56cad23b91707608edd3b3083729ea9a653c9e04\nSigned-off-by: Stig Telfer \u003cstig@stackhpc.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/0aca067ba4596ce113ca94909fd5c954c0bc9caa"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/0aca067ba4596ce113ca94909fd5c954c0bc9caa"}]},"branch":"refs/heads/master"},"12ae9ea9295fad1b6b82d4aae756a4008142bdc5":{"kind":"REWORK","_number":2,"created":"2026-10-01 17:54:23.000000000","uploader":{"_account_id":10910,"name":"Stig Telfer","email":"stig.openstack@telfer.org","username":"stigtelfer"},"ref":"refs/changes/90/988890/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/90/988890/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/90/988890/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/90/988890/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/90/988890/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/90/988890/2"}}},"commit":{"parents":[{"commit":"eac2fa47f26da3515c7a1e8c91226750517c52d4","subject":"Merge \"Remove usage of eventlet for image cache SQLite timeouts\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/eac2fa47f26da3515c7a1e8c91226750517c52d4"}]}],"author":{"name":"Stig Telfer","email":"stig@stackhpc.com","date":"2026-05-17 19:51:10.000000000","tz":60},"committer":{"name":"Stig Telfer","email":"stig@stackhpc.com","date":"2026-10-01 17:54:01.000000000","tz":60},"subject":"Add guide on image store migration","message":"Add guide on image store migration\n\nAn process is described here which should be helpful for admins\nlooking to move images between backend stores, for example as part\nof a storage migration process.\n\nChange-Id: I56cad23b91707608edd3b3083729ea9a653c9e04\nSigned-off-by: Stig Telfer \u003cstig@stackhpc.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/12ae9ea9295fad1b6b82d4aae756a4008142bdc5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/12ae9ea9295fad1b6b82d4aae756a4008142bdc5"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY"},{"label":"Workflow","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{"label:Review-Priority\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
