)]}'
{"id":"openstack%2Fglance~993644","triplet_id":"openstack%2Fglance~master~I3ae9a5364b91c719922c78157f9825afcdfad222","project":"openstack/glance","branch":"master","attention_set":{},"removed_from_attention_set":{"9303":{"account":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"last_update":"2026-06-17 15:53:05.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I3ae9a5364b91c719922c78157f9825afcdfad222","subject":"Fix flaky import task failure waits in functional tests","status":"MERGED","created":"2026-06-16 17:04:23.000000000","updated":"2026-06-17 16:25:58.000000000","submitted":"2026-06-17 15:53:05.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":7,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"993644","meta_rev_id":"5de9fa3661841fa3a946a19e01dbe0079c459159","_number":993644,"virtual_id_number":993644,"owner":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-06-17 15:53:05.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"all":[{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":2,"date":"2026-06-17 14:14:02.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"all":[{"value":0,"date":"2026-06-17 14:05:09.000000000","permitted_voting_range":{"min":0,"max":1},"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":1,"date":"2026-06-17 14:14:02.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Top Priority / Holds Gate"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-06-16 18:22:05.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-06-17 14:14:02.000000000","updated_by":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"reviewer":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"state":"REVIEWER"}],"messages":[{"id":"49a6aeb3369d36befd36bf8fb1b9a662f3b00093","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2026-06-16 17:04:23.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"3e1cbed65eac2009be18c5f5e8bf6ad012d20075","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2026-06-16 17:05:03.000000000","message":"Patch Set 1: Workflow-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"e6e3299e3719b0618f5cdcf70c9f9d5139ad8f54","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-16 18:22:05.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4b55d59d32474968a461c9a30db30e57\n\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/5e56a50a6a094936b65414a4d34f384c : SUCCESS in 1h 07m 20s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/6162a1a701b548bb9bb12b67ea2da519 : SUCCESS in 1h 05m 20s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/90bdf54039044de6a2080423a1f9b6b9 : SUCCESS in 3m 04s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/c36de6b627b74af49fed9d81a7568585 : SUCCESS in 3m 20s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/534716b550664dfbbf115471f716663d : SUCCESS in 4m 53s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/bc054b37464c4c68a6f085fa92246d18 : SUCCESS in 8m 07s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6b9f3438dd6b425e9abe5be81f467dcf : SUCCESS in 3m 31s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e29db9e6d2244ef1a4c2730761c2d834 : SUCCESS in 8m 22s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/0342917e54ad4b438acfe5a4260340dc : SUCCESS in 4m 44s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/946581da317848aea6f3649b9dcc2c7a : SUCCESS in 5m 00s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/673c052be7d546cc91e2719cf8c86741 : SUCCESS in 40m 21s","accounts_in_message":[],"_revision_number":1},{"id":"a871d62b7d4f8751c795f79d71cb6824dbb70a75","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2026-06-16 18:28:09.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"3c8b93f04a5ca7dc00f1a43e0765ddd7a7903161","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-16 19:49:08.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/a510007a30964ec8b0afb3a733d4cad1\n\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/6b39369ab8754a889787c4c24a7048b6 : SUCCESS in 1h 14m 37s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/446d6033f47e4ee39f9450fe3ed6b3a6 : SUCCESS in 49m 03s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/eb1fe92a65d644c88c469a7b27a121df : SUCCESS in 4m 45s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/9192ca17936b4491ad26e6921dddbb81 : SUCCESS in 5m 02s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/a5c0e692c75441719e1150ef611fd2af : SUCCESS in 5m 28s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/ec41105e3a3244b9a30c745deeea715a : SUCCESS in 9m 33s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fb5e41dd89cd488fa4468f421e7676ab : SUCCESS in 5m 55s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ec6591d0daf54fda8dee88aa3465e87b : FAILURE in 7m 55s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/ba58f29ce8b540a48b37a8eb89320a67 : SUCCESS in 6m 44s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/a1f9d01fae084f8b835e9549de74889b : SUCCESS in 6m 52s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/ff3cfebea5ba40eaa77afc22e25912c9 : SUCCESS in 35m 21s","accounts_in_message":[],"_revision_number":1},{"id":"fe0c6571ac07485a54ae881dea0f792fb038f6b1","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2026-06-16 19:54:03.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"d92ce206f15b33ee9fbeeb7abf93501f0f7f8eef","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-16 21:17:15.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/e431eb689b9d42358092521903f88a11\n\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/ad3061a076914a918d77a3e33ca02901 : SUCCESS in 1h 16m 46s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/d7cf9c0f08fe419fba357d662afa9aba : SUCCESS in 1h 08m 12s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1dfad351747c425aa62d10ab2705fd67 : SUCCESS in 4m 18s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/0a0ffd62315941c48dd179c32b1cfa83 : SUCCESS in 5m 17s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/5156ac58399a4251ad0fb08749c9aba2 : SUCCESS in 4m 58s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/3fca341faa404268bd03c5d5bbea1dcb : SUCCESS in 9m 29s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9a4847e10fcf467fba49c3e2a60cc2a3 : SUCCESS in 6m 31s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/fcea2727ff424d9e88d77aef86006a8f : FAILURE in 5m 30s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/49dcd4aa2ef44a4b8ee67d17b8ea579f : SUCCESS in 6m 13s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/2d6f1dfffbbe4d4ca25cf781f9d0c04a : SUCCESS in 6m 40s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/74922bd3b151463a985a76e657b1b0ba : SUCCESS in 34m 57s","accounts_in_message":[],"_revision_number":1},{"id":"8c9a7bd76f0e57d1d80701e0c7d8fefc9ff9fd07","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2026-06-17 06:23:57.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"c3c418b4f359c8ef83cc1005466f758b41c5af3d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-17 07:39:40.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/fff86a78f98046e48ca9c1290365fc71\n\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/6c7d0d974af647a2a490184440ca5862 : SUCCESS in 1h 10m 23s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/e846148efcdb4612a7800342788d9eab : SUCCESS in 1h 00m 21s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8cfb9ff0c4374110a743613255f9124c : SUCCESS in 4m 50s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/4e96f1a7d6904d6186df3a3c591480c5 : SUCCESS in 4m 47s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/c50ad46001ba48e69e681ca0a04ba5bc : SUCCESS in 4m 30s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/92c788196b844167853692f6aac4be47 : SUCCESS in 7m 36s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ba8d2f437f1c49b18a33dfdeb3db1c85 : SUCCESS in 5m 10s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d0645678de4b4038ae3d87b627e5bcb0 : SUCCESS in 7m 36s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/704e818fde07433e81114b744d7c5ecf : SUCCESS in 6m 14s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/406c411023b948a48928a55829d1b10f : SUCCESS in 6m 03s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/f9db19f61cf844dfbecc9af6db65fa7d : SUCCESS in 27m 30s","accounts_in_message":[],"_revision_number":1},{"id":"1f04693cff12d58323138d5f001038b6089fb5eb","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2026-06-17 07:48:49.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"b95a849c44e5d18910a308457c6f8e33880f8500","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-17 09:00:14.000000000","message":"Patch Set 1:\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f3ff3800a1c14ae99cf41e2230db933d\n\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/4b8c1a24a7514bada3eca7b7a8f2064e : SUCCESS in 1h 05m 45s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/627f80f951e24668a47d2378af38c6ee : SUCCESS in 1h 04m 35s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/effb6ffd57b645f681f99baa59e93a01 : SUCCESS in 5m 22s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/c66a34ecbde04f329c559b7531a83d75 : SUCCESS in 4m 57s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/73324db13da645fbac464e2be77fb4d2 : SUCCESS in 4m 49s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/c78a4c2bb8494ac9b69556baca20a6d3 : SUCCESS in 8m 22s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0dfd4eeda47d4020a6fc7fc5206fc7c3 : SUCCESS in 6m 19s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/dbca1f857d514229ac3f05f985ac4c10 : SUCCESS in 7m 53s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/f4cd88c5e9794aa08c974424b633600f : SUCCESS in 6m 37s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/b0f1c39207834e609ab62ddd6622e123 : SUCCESS in 6m 35s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/6c44ac9ed04740709e5c168042467e30 : SUCCESS in 34m 23s","accounts_in_message":[],"_revision_number":1},{"id":"f1f5a3ae5b105d633097c4602a36017c16129134","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2026-06-17 10:21:44.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"348e2f13bb88aa11066419d8e8732f9743ba1be3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-17 11:33:39.000000000","message":"Patch Set 1:\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ac35b6bff9134b41afa9b4427a6ed00a\n\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/361d36ef72844b30b1748764a2354115 : SUCCESS in 1h 03m 02s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/78339c6489814829bf405e73f2ab1ba4 : SUCCESS in 1h 05m 58s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/40d2e2d058eb43ca9bc582d40e9017a6 : SUCCESS in 5m 53s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/16b998a5bfd34d13b2d0361cf540e167 : SUCCESS in 4m 51s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/64e194c7a2a640bc97670cd1bff7da5d : SUCCESS in 3m 47s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/896a388ff93b46e5ae18036fd548529e : SUCCESS in 4m 20s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b818f65c38544ed6aa590f746e5b29c8 : SUCCESS in 7m 04s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a581bf0a1bac40e5969b38be9cf44b68 : SUCCESS in 3m 43s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/818f3444fc3c42c3b3a44a23523e194f : SUCCESS in 4m 07s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/610ab6aa00ce4954919542fdc600f49e : SUCCESS in 6m 28s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/b1a75a53b06441abae4ef0a6d2078d1e : SUCCESS in 36m 24s","accounts_in_message":[],"_revision_number":1},{"id":"33dcbf5b1c91accb7ba9c61bcf6596d3b81146ba","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2026-06-17 11:58:57.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"747c06e5be555bfe1d742e61bd251d6d218b4845","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-17 13:05:32.000000000","message":"Patch Set 1:\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5d1ebf8a47024930bf93702e86b62db7\n\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/31d7da4f81b5472eb9bcfd7de0b14889 : SUCCESS in 1h 05m 43s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/78c7c196b7264c8398efed424e042474 : SUCCESS in 51m 42s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/038043163e264a2f8c4969fb584cd0a4 : SUCCESS in 3m 06s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/11d4b3d354354fa1a85b837b3d919e05 : SUCCESS in 4m 51s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/504fbee23d164fd3b797213933037e74 : SUCCESS in 3m 15s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/8263cc32676c47af915d6c2d859d8ec4 : SUCCESS in 7m 37s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fa3fd9a657504645b164bea56306d748 : SUCCESS in 5m 19s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f8b7d27b34bc4d08ba619c7063aaad8d : SUCCESS in 4m 38s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/d11a8c0518fa45eabddcc18bfa6b351b : SUCCESS in 5m 07s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/c215e15b462144d19269aa4b153c5fe0 : SUCCESS in 6m 10s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/f2e31231531646e6b901a5f24ce21c02 : SUCCESS in 44m 16s","accounts_in_message":[],"_revision_number":1},{"id":"40b92726bd685ac24f03cd3c195b47060d8519a1","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2026-06-17 14:05:09.000000000","message":"Patch Set 1: -Workflow","accounts_in_message":[],"_revision_number":1},{"id":"206a3a8cf4026a6787e0b28ec0bef6cd59cd2439","author":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"date":"2026-06-17 14:14:02.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"149e022be00605c2d359b08df248507170d9b7f9","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-17 14:14:40.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"e855edc1362316c5efbef98b6cf1725f36001f62","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-17 15:53:05.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/08ce9c8519634e1bb73e4809e62161ab\n\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/9aa60144ebad4cad938147f123bc989a : SUCCESS in 1h 06m 45s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/a79b5192fede40edad0b79e013da3299 : SUCCESS in 1h 02m 55s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6dcc10138207403999c5caa4831d29b3 : SUCCESS in 5m 04s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/19a3753d44074926acb616b09872a540 : SUCCESS in 3m 06s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/8b69272ac70d40d6a3a9eef40072d0b6 : SUCCESS in 5m 15s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2c342dbf4ef64006b851eb506a55b054 : SUCCESS in 6m 23s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/0cd721da1fc94446b90147332f42e804 : SUCCESS in 5m 42s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/aa05370e297540d2a0004b2103d86d1d : SUCCESS in 6m 36s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/fe0a38a74e954f85ba680434de2644dc : SUCCESS in 6m 36s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/ff9bc457154242e09438ea69f04b7afc : SUCCESS in 27m 32s","accounts_in_message":[],"_revision_number":1},{"id":"61f4f0825f2f1ef823fbc437565f43a29759a255","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-17 15:53:05.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"52b15ecace6560180bc44dfaeba1dd8b2d111468","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-17 15:55:18.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b19de666bb8a4fec9a17993ed59859a4\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9a62ae63b41649dfac74a4577783428d : SUCCESS in 54s","accounts_in_message":[],"_revision_number":1},{"id":"5de9fa3661841fa3a946a19e01dbe0079c459159","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-17 16:25:58.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/949cddb6820f4e67a9e1ace90879d252\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/a94ca57cc0c54b0a9e86bdf984ef1f30 : SUCCESS in 6m 03s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/9852cd78b4c6416f8d983d459741eba0 : SUCCESS in 8m 33s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/f3e5ef8d35a14dfd9dbba74d2fdebd20 : SUCCESS in 8m 20s (non-voting)","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"e396d60cfbda85bd6c3e7df4865d44494ada19a7","revisions":{"e396d60cfbda85bd6c3e7df4865d44494ada19a7":{"kind":"REWORK","_number":1,"created":"2026-06-16 17:04:23.000000000","uploader":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"ref":"refs/changes/44/993644/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/44/993644/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/44/993644/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/44/993644/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/44/993644/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/44/993644/1"}}},"commit":{"parents":[{"commit":"edf97a1b5b6790f7e79315a11c3f70cc92553ef8","subject":"Merge \"Fix flaky import task waits in functional tests\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/edf97a1b5b6790f7e79315a11c3f70cc92553ef8"}]}],"author":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2026-06-16 16:59:22.000000000","tz":0},"committer":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2026-06-16 16:59:56.000000000","tz":0},"subject":"Fix flaky import task failure waits in functional tests","message":"Fix flaky import task failure waits in functional tests\n\nPoll /v2/tasks/{id} with admin credentials when waiting for import\nfailures, since /v2/images/{id}/tasks can report stale processing\nstatus under CI load even after the task has failed.\n\nChange-Id: I3ae9a5364b91c719922c78157f9825afcdfad222\nSigned-off-by: Abhishek Kekane \u003cakekane@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/e396d60cfbda85bd6c3e7df4865d44494ada19a7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/e396d60cfbda85bd6c3e7df4865d44494ada19a7"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{"label:Review-Priority\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
