)]}'
{"id":"openstack%2Fglance~993942","triplet_id":"openstack%2Fglance~master~I957d50703177ff3642d7f4937993e188ccdcd88c","project":"openstack/glance","branch":"master","attention_set":{},"removed_from_attention_set":{"13252":{"account":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"last_update":"2026-06-19 10:16:56.000000000","reason":"\u003cGERRIT_ACCOUNT_13252\u003e replied on the change","reason_account":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}},"9303":{"account":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"last_update":"2026-06-25 16:41:26.000000000","reason":"Change was submitted"},"8122":{"account":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"last_update":"2026-06-25 14:45:49.000000000","reason":"\u003cGERRIT_ACCOUNT_8122\u003e replied on the change","reason_account":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"}}},"hashtags":[],"change_id":"I957d50703177ff3642d7f4937993e188ccdcd88c","subject":"Remove glance-grenade-centralized-cache zuul job","status":"MERGED","created":"2026-06-18 14:44:56.000000000","updated":"2026-06-25 16:46:02.000000000","submitted":"2026-06-25 16:41:26.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":6,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"993942","meta_rev_id":"b85e6346ae96fd73d42e3006163bc55728b2b604","_number":993942,"virtual_id_number":993942,"owner":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-06-25 16:41:25.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"recommended":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":1,"date":"2026-06-24 05:30:37.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},{"value":1,"date":"2026-06-19 10:16:56.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":2,"date":"2026-06-25 14:45:49.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"all":[{"value":0,"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":1,"date":"2026-06-25 14:45:49.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Top Priority / Holds Gate"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-06-18 16:05:39.000000000","updated_by":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"reviewer":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"state":"REVIEWER"},{"updated":"2026-06-18 17:32:47.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-06-19 06:56:54.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"CC"},{"updated":"2026-06-19 10:16:56.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"REVIEWER"},{"updated":"2026-06-24 05:30:37.000000000","updated_by":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"reviewer":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"state":"REVIEWER"}],"messages":[{"id":"9e1550f358d2c1639615edadc1162016c248c648","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2026-06-18 14:44:56.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"7226c9325062e5c724d144599d85a129ce450c52","author":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"date":"2026-06-18 16:05:39.000000000","message":"Patch Set 1: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"853c1e843a6968a2e0c1b44f525df9774aab3226","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-18 17:32:47.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ced803fa45bc43a292ade87810616b19\n\n- grenade https://zuul.opendev.org/t/openstack/build/60c2e1e1b18f4a3f9897b76032db5d6e : SUCCESS in 1h 10m 28s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/0a8943c6fa1a484b912469103c875ae6 : SUCCESS in 1h 04m 11s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/60e87e779bb24af5adc3f18156307518 : SUCCESS in 1h 24m 43s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/eda5ceede1bc4c8aa62438f04744ff8f : SUCCESS in 1h 09m 37s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/69f321b92e57474192d2a4ae54167390 : SUCCESS in 5m 31s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/4e01a9f832224cb7825f8f86858737a7 : SUCCESS in 3m 15s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/46e6358a8ea64d2dba5959ec6286bf9b : SUCCESS in 6m 20s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/ca5fdbb8976f459791dabf91690a9c75 : SUCCESS in 9m 00s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/76b148bbaea242ebac4cb205a5d4e3f8 : SUCCESS in 7m 11s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/5a039cdcf7e042e1999e414799886eb7 : SUCCESS in 8m 45s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/b3ff1b2e86344d8eacf30d646d92526f : SUCCESS in 5m 24s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/a6ed2cb9935147af924773625aca7510 : SUCCESS in 9m 35s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/e4203e0263a04e87b50f836b2f70db4d : SUCCESS in 1h 52m 20s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/f9ffa306441647d0afe087df1482a326 : SUCCESS in 2h 11m 38s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/31d3a346548c4acf8da12ac72dd3ad0f : SUCCESS in 1h 09m 49s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/92d2c6be23e24330a154a433536b7f47 : SUCCESS in 2h 37m 59s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/8edc01f714b04388bf37424d090bd0d2 : SUCCESS in 1h 03m 09s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/b0d5d981f6b14e9db5651dd323087b07 : SUCCESS in 1h 58m 18s\n- glance-grenade-centralized-cache https://zuul.opendev.org/t/openstack/build/673ef69d98f34e1389f832dda9e51139 : FAILURE in 25m 34s (non-voting)\n- glance-s3-multistore https://zuul.opendev.org/t/openstack/build/b8204eae566e4f2d80c6f0e94349dcef : SUCCESS in 2h 01m 08s (non-voting)\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/d97e8ee4aa944ca3af3ef24fe8c4c4e1 : SUCCESS in 40m 26s\n- glance-rally-task-uwsgi https://zuul.opendev.org/t/openstack/build/65589a3b7f9c4db3af2e62220a6d589e : SUCCESS in 26m 52s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"feff404a5c114f13fd98fba8697ac31f49dabbeb","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-18 17:56:53.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/645d9db56dec49deabb150ef49443046\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/0e4eafc1b9be47239fb936669e095eab : SUCCESS in 5m 33s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/0e47dec5e1cc430b8fc2f46eab012bba : SUCCESS in 7m 48s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/9e14b92beb8e49969c8168601899a692 : SUCCESS in 13m 10s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"78231429343085545b9b81d84517028fbfeed2cd","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2026-06-19 06:09:59.000000000","message":"Patch Set 2: Published edit on patch set 1.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":2},{"id":"2a4dfe95769d5d8ee7137eac1f8bef86628a2a66","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2026-06-19 06:10:17.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"150267b39bc07b4f0d7cf7d17c3bee9498d9bdb4","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-19 06:38:36.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/69008cfba04c4dca9b2bf383cc9ede38\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/fc0ce95bcdc143e7947cca3808a308f8 : SUCCESS in 8m 17s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/c8e5ce30a9354064a4305b54abf618e4 : SUCCESS in 7m 14s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/682096077da14e59993d60a3a6e379e5 : SUCCESS in 13m 56s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"112ecba96639d05e7dbd992a22b21d5628ba0b63","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2026-06-19 06:56:54.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"69664a9b7a23c27091a616ed10354a5dc62b8604","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2026-06-19 07:02:04.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"2d19a0c21d4cb57d806cf86640fab97ba1600b63","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-19 08:46:54.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/6c1369e448f546f59ad46c6b63e8232f\n\n- grenade https://zuul.opendev.org/t/openstack/build/e88b22f46790469fbcc8eba2290701df : SUCCESS in 31m 24s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/0f3c7087843f4ee58ed10e1d0b379679 : SUCCESS in 1h 07m 15s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/b4c286f9d07545108836efd53966782e : SUCCESS in 1h 24m 19s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/ad581cee3dee4606b670f22a906e5df7 : SUCCESS in 33m 07s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ec5c864613dd44b7930d8ef228e73c26 : SUCCESS in 2m 38s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/5211561ddae445939bf27746ddb3814c : SUCCESS in 4m 51s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/5c5c517691a84de5a6c26c202f7ea97f : SUCCESS in 2m 39s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/95b6ce693f0f446da73fe9ea1e2a0a0e : SUCCESS in 8m 30s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1d7af530c35b47258026a61f29c59cf7 : SUCCESS in 5m 24s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/93720a7c78294fd49c746f8e97275b2f : SUCCESS in 7m 55s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/56003b3b4ad24852af6c0d5721e6a579 : SUCCESS in 6m 26s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/a3adf785d3434feea3e8bbc05c8c3b87 : SUCCESS in 7m 34s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/da6ceb12adbc4e41897af764c7acc8df : SUCCESS in 1h 48m 49s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/2a20badbcbf84e5385f14341952dece9 : SUCCESS in 2h 04m 11s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/97f936f9c92f47199894baa00a43503c : SUCCESS in 1h 58m 23s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/eab4675c560141f988f61a2773c8a0d6 : SUCCESS in 2h 30m 11s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/f20f60725998419c9e05d0ba0cc6fef4 : SUCCESS in 33m 06s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/a4347a955fe5432ab99edb275a6f55d4 : FAILURE in 1h 32m 18s\n- glance-grenade-centralized-cache https://zuul.opendev.org/t/openstack/build/438fc6f87b8f4938b7bfd3ac5851a88c : FAILURE in 13m 59s (non-voting)\n- glance-s3-multistore https://zuul.opendev.org/t/openstack/build/0aec9e816a414a379726dd061bda33e0 : SUCCESS in 1h 59m 53s (non-voting)\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/0df93c08a2b64885abede73c5161ea5e : SUCCESS in 18m 28s\n- glance-rally-task-uwsgi https://zuul.opendev.org/t/openstack/build/c2abfcd1b6574bf691036237efda2f07 : SUCCESS in 42m 49s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"ffa02469fc6fedeb3e3c19fe13efe381170e830d","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2026-06-19 08:50:20.000000000","message":"Patch Set 2: Workflow-1","accounts_in_message":[],"_revision_number":2},{"id":"3139ee46f049d69c1d5254ead96d2bcce6bc59b0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2026-06-19 09:31:38.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n* Workflow-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":3},{"id":"9320189e612835532ea76768fd874d166a0e99e2","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-19 09:59:56.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/92e65ee13e0c4dff99b6603389189720\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/85a4cafaa3da4fa3a01772a2a8e1e9f2 : SUCCESS in 8m 10s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/1e74d57e861c4dc09cfe1c299f159a0c : SUCCESS in 9m 26s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/68bbf6971ede439ba7dd167b1617673b : SUCCESS in 17m 21s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"d87a48644ff1a97bb47aeeae8a5e1ae547b75905","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2026-06-19 10:16:56.000000000","message":"Patch Set 3: Code-Review+1","accounts_in_message":[],"_revision_number":3},{"id":"f5d78ef53b4af2ddc6e28ed373aa42d29b1fae82","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-19 11:57:38.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/e4fb591b230e4a70b5abeb91d39ef96c\n\n- grenade https://zuul.opendev.org/t/openstack/build/d041a044cd064effa1f55685cde748dd : SUCCESS in 1h 06m 43s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/6d14ece9eed2422099cca99de0f8b50f : SUCCESS in 31m 05s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/01c227e3c04e4eea932afecba23fc24c : SUCCESS in 1h 21m 42s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/cdfd2885597844ef93b6bee9e344a948 : SUCCESS in 1h 04m 06s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2d31e002ea82440abd955c3b322a2e35 : SUCCESS in 5m 15s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/fa8d8f58b4194143aaa2e7a3ef0aae3e : SUCCESS in 4m 30s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/d720a3000f154625b4695f4765337773 : SUCCESS in 4m 42s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/b9e0ccba7c2042f4aa8134955d699238 : SUCCESS in 7m 31s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a5b5e57f910f423aac26e711c24643b2 : SUCCESS in 6m 17s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/880e4cc622d6424288bfa44901b41729 : SUCCESS in 7m 59s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/a25a8c69e1324717b72cb97506838dfc : FAILURE in 6m 25s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/ccc7c2716ceb44539df30f4d978d5755 : FAILURE in 5m 43s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/cdaef7534bba4a3b8956ce53bf525c10 : SUCCESS in 1h 48m 08s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/6e645f0d1ffa43b39a41256159ba596e : SUCCESS in 1h 54m 20s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/652a6a0a33564669901c0e4ac1389d07 : SUCCESS in 1h 58m 37s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/944ecbaa08e94f52a0d29e186aca0e9c : SUCCESS in 2h 20m 01s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/9b87d95b53e74cc083273c40f87deed8 : SUCCESS in 1h 02m 58s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/30f0cfb2878e4bc3882eb2785ccb4c39 : SUCCESS in 1h 48m 03s\n- glance-s3-multistore https://zuul.opendev.org/t/openstack/build/5e39a948880e479aa57ca9499b7f3689 : FAILURE in 1h 52m 04s (non-voting)\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/f9fa761a7ae345cda60ed5b3dde598b0 : SUCCESS in 34m 16s\n- glance-rally-task-uwsgi https://zuul.opendev.org/t/openstack/build/8d942bb3dffa408d9205ee789f80632f : SUCCESS in 51m 31s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"f74d87d4d269a624e349e96479983b9be8cc07d1","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2026-06-19 14:02:58.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"159a8205122bb77e5436838705552b03e7b09008","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-19 14:36:39.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/067ac5a9015d4860a877484251bd8818\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/be9bad664da0453986812e984a92989e : SUCCESS in 8m 51s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/bd112c402a52408682ab7ad3e2ed4fbd : SUCCESS in 7m 16s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/abffd2f1f12848bfa872939ffb75ffcd : SUCCESS in 11m 19s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"3b5bf9447b816ad75e9f891501b8cdcdbaae1881","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-19 16:33:26.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2eaf49a782184821b687e3de1374aaa2\n\n- grenade https://zuul.opendev.org/t/openstack/build/c326ed18091c43f6897eb3344c384959 : SUCCESS in 1h 04m 29s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/057d888487a240668877cb9eb4536325 : SUCCESS in 1h 02m 48s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/70603dc3df254aa99027ac1cdb1e93e7 : SUCCESS in 1h 34m 45s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/add8db8ef5354bc4889e62dd2e30cc57 : SUCCESS in 1h 09m 31s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c1ffac9c356a4afc8a46974e52fb3e89 : SUCCESS in 4m 44s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/61e102c0dbf440c89aadb5ab8e3c6451 : SUCCESS in 6m 53s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/e31ce0e76b0241e08222a75604ffde14 : SUCCESS in 5m 13s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/4b21108c9ea34b52a548b4e29d02bac7 : SUCCESS in 7m 25s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5f4fa5b2495d45edac11b2a2a23b81b4 : SUCCESS in 7m 49s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/fd8e49aeff3f4b4dac309ddda1922e0c : SUCCESS in 7m 55s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/9cab65053c34431caf4ee3d2b1379582 : SUCCESS in 6m 57s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/2bd90a855b04482cb570957e0cab6383 : SUCCESS in 4m 58s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/45cc9e10de9146f9bfaf0ad8e6432bf5 : SUCCESS in 1h 53m 45s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/88da36613bd443538d0ba5f4d18959d1 : SUCCESS in 1h 10m 58s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/82578e7e0e694a6a98e6df6a421b1f66 : SUCCESS in 1h 43m 25s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/d6cac3080b7f4dc7acac4658e037a8af : SUCCESS in 2h 18m 35s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/89ee2dbc656a404fa0093a9f2ba76383 : SUCCESS in 52m 00s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/0abbaba765e84cfe86e346431c03234d : SUCCESS in 1h 27m 08s\n- glance-s3-multistore https://zuul.opendev.org/t/openstack/build/4a91285521164effb1dee239bd597cc1 : SUCCESS in 1h 03m 37s (non-voting)\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/57d4f4d6544c47748b3745315e95895a : SUCCESS in 27m 11s\n- glance-rally-task-uwsgi https://zuul.opendev.org/t/openstack/build/c84101e4ca174b88865bda34109218da : SUCCESS in 28m 11s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"a95e3223abdb44334aeaacb1571b1b8725c5c0f9","author":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"date":"2026-06-24 05:30:37.000000000","message":"Patch Set 3: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"c32daea447ebe31e2e3b72f73eeb2c20090ead7f","author":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"date":"2026-06-25 14:45:49.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"ac7755e7c403e3d89eb870609db025d2fc52e2f1","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-25 14:47:13.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"b04c9f1d58543b535656f9ac3665d1ef584342bc","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-25 16:41:25.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/70ec8f6ec0a44b03b1b269cf2151c531\n\n- grenade https://zuul.opendev.org/t/openstack/build/82980cbb59dd4582ae9d474e85e86109 : SUCCESS in 1h 08m 41s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/edcb69c116ce4c06a8e9cfbf2e283ccf : SUCCESS in 46m 47s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/3ac62a11aa9542f7b2993bf88130c675 : SUCCESS in 1h 52m 11s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/db60373df93f4cfc898d34aae13efa80 : SUCCESS in 1h 13m 09s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/901a6fa1de5f4217b73e1aa1529acd09 : SUCCESS in 4m 13s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/7dd0c5024c2144e498068df0f35efb1d : SUCCESS in 6m 45s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/6673d85c90ce4ca4aba7a74030d70b82 : SUCCESS in 7m 05s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/84772d0df4e84243bbd9097908d6d39f : SUCCESS in 7m 02s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/97cc41dd0fbf43f5b2affdd7ffec89c4 : SUCCESS in 7m 29s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/32343aceb8cc42dea621211f8c556271 : SUCCESS in 5m 14s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/c163198027f94838b8553a72d33aaa66 : SUCCESS in 7m 17s\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/cd3dd17917ce4e62b13c5e78ab778174 : SUCCESS in 1h 51m 28s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/3d359fae919b4555971a84eeb8b2eca3 : SUCCESS in 1h 12m 10s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/6a0d7662ddbf4b43be865931cd5da742 : SUCCESS in 48m 45s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/8f83f8292bd64ffab880e55a84322242 : SUCCESS in 1h 49m 10s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/80c4a1a6e7144fffb2c4a5cbb84d2b5e : SUCCESS in 33m 01s","accounts_in_message":[],"_revision_number":3},{"id":"86c20e85aaf3c5d3df6e16a9306a2b4aea72e98b","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-25 16:41:26.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3},{"id":"b85e6346ae96fd73d42e3006163bc55728b2b604","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-25 16:46:02.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ed458fd5faa340a1a0088964fe1db4c7\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/47fa93a6900647a0962e46a3503ef1a8 : SUCCESS in 1m 04s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"928a056f35d37ab35bb6baba4fe34e95d3ddcbb4","revisions":{"f2742cf7a030c72a1aab787aa350fc06c5f2ff00":{"kind":"REWORK","_number":1,"created":"2026-06-18 14:44:56.000000000","uploader":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"ref":"refs/changes/42/993942/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/42/993942/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/42/993942/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/42/993942/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/42/993942/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/42/993942/1"}}},"commit":{"parents":[{"commit":"9734292f7c4da958b4e29afa0a7f40b8a2003efc","subject":"Merge \"Fix flaky test_import_copy_bust_lock functional test\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/9734292f7c4da958b4e29afa0a7f40b8a2003efc"}]}],"author":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2026-06-18 14:40:09.000000000","tz":0},"committer":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2026-06-18 14:41:05.000000000","tz":0},"subject":"Use Jammy nodeset for glance-grenade-centralized-cache job","message":"Use Jammy nodeset for glance-grenade-centralized-cache job\n\nThe stable/2023.2 devstack baseline does not support Ubuntu Noble,\nso pin this grenade job to openstack-two-node-jammy until it is removed.\n\nChange-Id: I957d50703177ff3642d7f4937993e188ccdcd88c\nSigned-off-by: Abhishek Kekane \u003cakekane@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/f2742cf7a030c72a1aab787aa350fc06c5f2ff00"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/f2742cf7a030c72a1aab787aa350fc06c5f2ff00"}]},"branch":"refs/heads/master"},"611fb60f9e4f4514a6b7a5e399475b14a63cdd24":{"kind":"REWORK","_number":2,"created":"2026-06-19 06:09:59.000000000","uploader":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"ref":"refs/changes/42/993942/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/42/993942/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/42/993942/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/42/993942/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/42/993942/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/42/993942/2"}}},"commit":{"parents":[{"commit":"9734292f7c4da958b4e29afa0a7f40b8a2003efc","subject":"Merge \"Fix flaky test_import_copy_bust_lock functional test\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/9734292f7c4da958b4e29afa0a7f40b8a2003efc"}]}],"author":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2026-06-18 14:40:09.000000000","tz":0},"committer":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2026-06-19 06:09:59.000000000","tz":0},"subject":"Use Jammy nodeset for glance-grenade-centralized-cache job","message":"Use Jammy nodeset for glance-grenade-centralized-cache job\n\nThe stable/2023.2 devstack baseline does not support Ubuntu Noble,\nso pin this grenade job to openstack-two-node-jammy until it is removed.\n\nChange-Id: I957d50703177ff3642d7f4937993e188ccdcd88c\nSigned-off-by: Abhishek Kekane \u003cakekane@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/611fb60f9e4f4514a6b7a5e399475b14a63cdd24"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/611fb60f9e4f4514a6b7a5e399475b14a63cdd24"}]},"branch":"refs/heads/master"},"928a056f35d37ab35bb6baba4fe34e95d3ddcbb4":{"kind":"REWORK","_number":3,"created":"2026-06-19 09:31:38.000000000","uploader":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"ref":"refs/changes/42/993942/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/42/993942/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/42/993942/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/42/993942/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/42/993942/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/42/993942/3"}}},"commit":{"parents":[{"commit":"c5ad2013898f2deb151ef0070a60004fadee2b6d","subject":"Switch tests to native threading","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/c5ad2013898f2deb151ef0070a60004fadee2b6d"}]}],"author":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2026-06-18 14:40:09.000000000","tz":0},"committer":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2026-06-19 09:27:47.000000000","tz":0},"subject":"Remove glance-grenade-centralized-cache zuul job","message":"Remove glance-grenade-centralized-cache zuul job\n\nThe glance-grenade-centralized-cache grenade job verified migration of\nimage cache data from the sqlite driver to centralized_db when upgrading\nfrom stable/2023.2. That upgrade path has been validated and the sqlite\ncache driver is being removed, so this job no longer serves a purpose.\n\nChange-Id: I957d50703177ff3642d7f4937993e188ccdcd88c\nSigned-off-by: Abhishek Kekane \u003cakekane@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/928a056f35d37ab35bb6baba4fe34e95d3ddcbb4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/928a056f35d37ab35bb6baba4fe34e95d3ddcbb4"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{"label:Review-Priority\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
