)]}'
{"id":"openstack%2Fglance~995179","triplet_id":"openstack%2Fglance~master~I40da42195057404de9088830885057419b42df01","project":"openstack/glance","branch":"master","topic":"bp/amd-sev-snp-libvirt-support","attention_set":{},"removed_from_attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2026-07-18 06:52:44.000000000","reason":"\u003cGERRIT_ACCOUNT_9816\u003e replied on the change","reason_account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"}}},"hashtags":[],"change_id":"I40da42195057404de9088830885057419b42df01","subject":"Update metadata definition for AMD SEV-SNP support in nova","status":"MERGED","created":"2026-06-27 17:14:55.000000000","updated":"2026-07-18 09:58:49.000000000","submitted":"2026-07-18 09:57:53.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"995179-bp/amd-sev-snp-libvirt-support","meta_rev_id":"726659c6f19c06990869d5cfdf9b373ab9196aa0","_number":995179,"virtual_id_number":995179,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-07-18 09:57:53.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"all":[{"value":2,"date":"2026-07-01 13:22:13.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"all":[{"value":1,"date":"2026-07-01 13:22:13.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Top Priority / Holds Gate"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-06-27 17:37:26.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2026-06-27 19:55:26.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-07-01 13:22:13.000000000","updated_by":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"reviewer":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"state":"REVIEWER"}],"messages":[{"id":"492a2712cbdbaf9702fc0593a9852f3ec5a90bfd","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2026-06-27 17:14:55.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"cec5dd2d4f3ddd642ba83693b3acdbc8c97845ee","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2026-06-27 17:23:11.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"cb583b3ebb7669d5c07bf36fb99e54e158504e09","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-27 17:37:26.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1395d6a9fbd2496b8c4bd30175bde49b\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/eec55c66e2624274b1598703a6d4271e : SUCCESS in 9m 12s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/d7057f2381f241329a85b7331735f5d9 : SUCCESS in 9m 28s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/f240efbc85124fe0a0a813e0e63bfbbd : SUCCESS in 11m 56s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"646d62fb82c8fde369110fd7388839d12220e016","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-27 19:55:26.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/198b32ede27c4ed69f1830424cdd686c\n\n- grenade https://zuul.opendev.org/t/openstack/build/1f5b276200424c7e8bf693b144b1d80f : SUCCESS in 1h 07m 29s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/a081e0114cd34c03aab4767aa4cc0503 : SUCCESS in 1h 05m 34s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/9a9286add51940d99250b1b650162052 : SUCCESS in 1h 29m 42s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/6a4b978f6d7f41539f13a1baeea4fd33 : SUCCESS in 1h 02m 16s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/764171277e284df1b589643545cbe035 : SUCCESS in 5m 27s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/74065cc5d805407f827f07bf9783443e : SUCCESS in 5m 11s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/8581fe5ec9304a89be0cf2ec90d95813 : SUCCESS in 4m 57s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/1fe22d7aeae548b3a19c7580ec9c2b93 : SUCCESS in 4m 34s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/664afb70b61e4f6bafab9809e7c345a5 : SUCCESS in 6m 50s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/10bb857a438641519d24e2a6b9d256a0 : SUCCESS in 8m 16s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/c2077c87043846b09f2b85854a5c9129 : SUCCESS in 6m 24s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/c42f9acf62fa409f8f18834d06e86071 : SUCCESS in 5m 23s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/c1b9794636194509bfd19b7dd02c479e : SUCCESS in 1h 49m 17s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/214c20e1d5e34fed921e2763c03e7aae : SUCCESS in 2h 12m 45s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/d3f9418e317849e785c65178b2d39d91 : SUCCESS in 1h 53m 34s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/e542324a7afb4ac1b16c952bfcce8e2f : SUCCESS in 2h 26m 13s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/aa75e71aeaae47dc9c6b591a9abf7fe3 : SUCCESS in 31m 12s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/a6f379bbb344438288a2c53bf1337a64 : SUCCESS in 1h 05m 48s\n- glance-s3-multistore https://zuul.opendev.org/t/openstack/build/669e8aff4b1b45058ffacb1580d803ea : FAILURE in 1h 56m 56s (non-voting)\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/8a256cfa11cb45f1a537b21091db769a : SUCCESS in 37m 09s\n- glance-rally-task-uwsgi https://zuul.opendev.org/t/openstack/build/745171fb283f437885b979e88a3390a5 : SUCCESS in 56m 16s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"0bdbfc77271e4bf79ef7b9fef3e6d8f4deba12a4","author":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"date":"2026-07-01 13:22:13.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"a236580697b703f2b64ffc3a50b1aed3e3a32598","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-16 17:36:26.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"6be3990a6c4d009f1207e8d798a1f521461b89ea","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-16 20:06:00.000000000","message":"Patch Set 2: Verified-2\n\nThis change depends on a change that failed to merge.\n\nChange https://review.opendev.org/c/openstack/nova/+/994930 is needed.","accounts_in_message":[],"_revision_number":2},{"id":"c95685f4f3eaa24c487d63dae6a71105fd3e5c47","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2026-07-18 04:42:24.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"d7cf9b3dc0a418a24bb804a8241ad5594d39f0d0","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-18 04:56:19.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6f3f2a78f9bd44158684256db220aeb1\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/6bba16c9b3f94145a1121de828dbacb9 : SUCCESS in 6m 49s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/1a23caae01c24f4bb9f8b7071c5379a7 : SUCCESS in 6m 35s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/3af6f85df1bc4faeb633a546d2c4acd7 : SUCCESS in 12m 16s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"8f39cab91082871869d6ec9fa30120ff290bc855","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-18 06:45:56.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/57aef74290aa4498a7e33f42f46e3bab\n\n- grenade https://zuul.opendev.org/t/openstack/build/1b601f438ed647e782a835231fd0ea84 : SUCCESS in 37m 09s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/40112ec43e8e45378784e956058f5fd9 : SUCCESS in 1h 08m 02s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/f4d1377ce70a45f19438e17f5065ca43 : FAILURE in 1h 17m 10s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/5fe43dda45f643f28931ef97cbc7ceab : SUCCESS in 27m 42s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a14ad43d575c4e46b4697e3ea6b8a2df : SUCCESS in 2m 33s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/fc63144b5fba40c8968d3f2b41255fae : SUCCESS in 4m 21s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/666ee335101544a39091d7a6aec2b76d : SUCCESS in 4m 30s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/c52d5cb5fe994a78b9ef0deb4d56ecc5 : SUCCESS in 8m 11s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/522a3ea59fb64769ba539b95cbfa08be : SUCCESS in 3m 58s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b5abbff09c774273bba49cbe8720a92c : SUCCESS in 4m 17s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/e770199e98c54412b1724e7a5908beec : SUCCESS in 5m 16s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/d914a84445da427c994de8fabc270762 : SUCCESS in 4m 22s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/2ce80474aa164f2b81faf01289aefe4a : SUCCESS in 45m 54s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/7d8177fc0d7245d19eea8ea0597d9a4f : SUCCESS in 1h 26m 04s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/06082ee7148248e2bd8a3e89d53759a4 : SUCCESS in 48m 30s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/03317e3e9e3e4a108bb9ca04069d13d9 : SUCCESS in 2h 00m 56s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/112d89b8ac8d463ba16201c27e039108 : SUCCESS in 26m 31s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/63589e55a28b42a18239d8083018e94b : SUCCESS in 1h 29m 55s\n- glance-s3-multistore https://zuul.opendev.org/t/openstack/build/c80372e96835432e8d0762f5a465786e : SUCCESS in 1h 47m 40s (non-voting)\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/6be08ff5fca2447f9a485c6092b156b5 : SUCCESS in 36m 30s\n- glance-rally-task-uwsgi https://zuul.opendev.org/t/openstack/build/d316ee562b07410499a974b04f306139 : SUCCESS in 22m 50s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"88fd12544761a885c19d9864630058a16e7adfac","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2026-07-18 06:52:44.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"339cbed813607fb5c390df4ae9d496b997ca7f21","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-18 07:07:13.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d839ee72643b405e88176145ab72e403\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/e74939e88f9f4583a28499f3768c9719 : SUCCESS in 5m 50s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/8c76eee2dda841599b9cb5f6968002a1 : SUCCESS in 7m 51s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/5ac8f874ffec4d64a16edbdec319771a : SUCCESS in 12m 32s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"e741da4ec6604c0d26b64177126bf88df5027fc6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-18 08:52:54.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/08ba785a3fd54019ae20e6fec1ba30a7\n\n- grenade https://zuul.opendev.org/t/openstack/build/672454762a0c4327b16290d610ff3ceb : SUCCESS in 1h 01m 49s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/5f8d4f300cd04d4d8d02408ba23d8e43 : SUCCESS in 59m 58s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/56e13487df35444aa53dbaa2fe78e105 : SUCCESS in 1h 18m 47s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/e6ee8a0e8acd403aa1378fd24c4b23e1 : SUCCESS in 57m 00s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8f696381ab3b4c4db30d03556117c2d7 : SUCCESS in 4m 27s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/5fafa77421484a19852f231001d1ef6d : SUCCESS in 2m 40s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/8a819b66c1864e33bedfe605524b65e6 : SUCCESS in 2m 37s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/021454a3918c44aea4b764ee280551ad : SUCCESS in 7m 20s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d7e46993282a4892bfc5902a4c89ba9f : SUCCESS in 5m 01s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/6dd3371cf3344032bb0229e471b0283e : SUCCESS in 8m 10s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/41a63ea37bce4d868c0e971300bcb390 : SUCCESS in 7m 50s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/ce377cba7ed64254a13814c7989aba32 : SUCCESS in 5m 21s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/6e56d489c7274d7087a4340129f1376d : SUCCESS in 46m 50s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/8a2fc84a34104b5f83f07fed62c08fd3 : SUCCESS in 1h 43m 15s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/3ed71f4866134c05acab4e59f502f75e : SUCCESS in 1h 33m 56s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/ee64f494079948d6a06afc5f21f5548f : SUCCESS in 1h 54m 48s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/89db0bb3c9424779bc40044b4a1c5ec7 : SUCCESS in 53m 43s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/0019fae5611c46a6a4267e4152b7f71a : SUCCESS in 1h 28m 06s\n- glance-s3-multistore https://zuul.opendev.org/t/openstack/build/3200e0756b534c95b0c8a6ac754256ef : SUCCESS in 1h 01m 34s (non-voting)\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/034d944eedc04b8db0cf8f9ea609b561 : SUCCESS in 35m 04s\n- glance-rally-task-uwsgi https://zuul.opendev.org/t/openstack/build/ad587823129749eda0c1f4a589653ad7 : SUCCESS in 27m 14s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"6df9f68c435fc10a25e2e7066176a67a6aa8502a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-18 08:53:26.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"6e167114d3eb1ff6df3ca04730fc8d0c1259239f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-18 09:57:53.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/334ab0e0e63b4427bf2a52ec685b5ec1\n\n- grenade https://zuul.opendev.org/t/openstack/build/98419ea8109040f2b06b5ff371a15b5c : SUCCESS in 34m 47s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/422a282db0534ae0ae0ac18462547625 : SUCCESS in 34m 22s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/49d554f7defc449e860e9b143f20d736 : SUCCESS in 1h 04m 18s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/65fb0578274a4a8196fddcb45da3e320 : SUCCESS in 32m 11s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ebd672776c71433181addde7a52b1d08 : SUCCESS in 4m 37s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/7b3039065e2a4e77bc829fa8d4574046 : SUCCESS in 3m 54s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/0b31a3a113324af3abe2dd2d59014988 : SUCCESS in 4m 09s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3ae6bd67f5ca46349bf18e9b0f73987e : SUCCESS in 5m 42s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/da8a9442614f4c1ebcd5ef92d5d72527 : SUCCESS in 4m 34s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/94e33a4cfdb048499a7b335dc03eb812 : SUCCESS in 5m 49s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/024ceeaeb68540c89564367f5d4fbdd2 : SUCCESS in 4m 57s\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/7f1e06d1033b43b88b9f89bdacf2e9e9 : SUCCESS in 45m 40s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/5cfa2ad5579d46da8c6d3d0146094469 : SUCCESS in 46m 58s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/6d77a6a4f1d14a628ce6c68eef57abda : SUCCESS in 33m 10s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/7b71af55be8c48e6bd09abf856b24d96 : SUCCESS in 43m 20s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/bd83702eb66b434ca33d2fc4a6bec2bb : SUCCESS in 21m 09s","accounts_in_message":[],"_revision_number":2},{"id":"4683884681a492ac3865952101c779e6a9991cd2","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-18 09:57:53.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"726659c6f19c06990869d5cfdf9b373ab9196aa0","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-18 09:58:49.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/09e63e8c87294c4ab43c887e73a8a093\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1cd8fae6553e413c9095bfb1642416a9 : SUCCESS in 46s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/cf774407f92e4fbdbb05b7700aeef291 : SUCCESS in 41s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"6b3bae62ea8af950ae1107d02d5f1a7d8e833b8f","revisions":{"47ce38dee5e056c87fbfdeab4ef9c6a4a2d96ff7":{"kind":"REWORK","_number":1,"created":"2026-06-27 17:14:55.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/79/995179/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/79/995179/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/79/995179/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/79/995179/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/79/995179/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/79/995179/1"}}},"commit":{"parents":[{"commit":"fa9a03f0a5fdda81607376fde6a542168049c2e0","subject":"Merge \"Remove unnecessary direct dependency on Pygments\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/fa9a03f0a5fdda81607376fde6a542168049c2e0"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-06-27 17:14:00.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-06-27 17:14:39.000000000","tz":540},"subject":"Add amd-sev-snp memory encrtption model","message":"Add amd-sev-snp memory encrtption model\n\nDepends-on: https://review.opendev.org/c/openstack/nova/+/994930\nChange-Id: I40da42195057404de9088830885057419b42df01\nSigned-off-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/47ce38dee5e056c87fbfdeab4ef9c6a4a2d96ff7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/47ce38dee5e056c87fbfdeab4ef9c6a4a2d96ff7"}]},"branch":"refs/heads/master"},"6b3bae62ea8af950ae1107d02d5f1a7d8e833b8f":{"kind":"REWORK","_number":2,"created":"2026-06-27 17:23:11.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/79/995179/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/79/995179/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/79/995179/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/79/995179/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/79/995179/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/79/995179/2"}}},"commit":{"parents":[{"commit":"fa9a03f0a5fdda81607376fde6a542168049c2e0","subject":"Merge \"Remove unnecessary direct dependency on Pygments\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/fa9a03f0a5fdda81607376fde6a542168049c2e0"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-06-27 17:14:00.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-06-27 17:21:33.000000000","tz":540},"subject":"Update metadata definition for AMD SEV-SNP support in nova","message":"Update metadata definition for AMD SEV-SNP support in nova\n\nNova introduced launching instances with AMD SEV-SNP memory encryption,\nwhich can be enabled by the new \u0027amd-sev-snp\u0027 memory encryption model\nin image properties.\n\nUpdate the enum to support this new value.\n\nDepends-on: https://review.opendev.org/c/openstack/nova/+/994930\nChange-Id: I40da42195057404de9088830885057419b42df01\nSigned-off-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/6b3bae62ea8af950ae1107d02d5f1a7d8e833b8f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/6b3bae62ea8af950ae1107d02d5f1a7d8e833b8f"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{"label:Review-Priority\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
