)]}'
{"id":"openstack%2Fglance~995180","triplet_id":"openstack%2Fglance~master~Ie4506d1130be7af32c898169338ad5e32466a8b7","project":"openstack/glance","branch":"master","topic":"eventlet-removal-add-gmr-back","attention_set":{},"removed_from_attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2026-07-30 16:16:35.000000000","reason":"Change was submitted"},"8122":{"account":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"last_update":"2026-07-30 13:48:25.000000000","reason":"removed on reply"}},"hashtags":[],"change_id":"Ie4506d1130be7af32c898169338ad5e32466a8b7","subject":"Add support for GMR report generation to glance-scrubber","status":"MERGED","created":"2026-06-27 17:41:43.000000000","updated":"2026-07-30 16:17:44.000000000","submitted":"2026-07-30 16:16:35.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":5,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"995180-eventlet-removal-add-gmr-back","meta_rev_id":"6858b030a817c8ea18a5c7416da9708120a792a6","_number":995180,"virtual_id_number":995180,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-07-30 16:16:34.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"all":[{"value":2,"date":"2026-07-30 14:17:22.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":2,"date":"2026-07-30 13:48:25.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"all":[{"value":1,"date":"2026-07-30 14:17:22.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Top Priority / Holds Gate"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-06-27 17:56:46.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2026-06-27 18:47:01.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-07-01 13:22:59.000000000","updated_by":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"reviewer":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"state":"REVIEWER"},{"updated":"2026-07-07 12:10:47.000000000","updated_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"reviewer":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"state":"CC"},{"updated":"2026-07-30 14:17:22.000000000","updated_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"reviewer":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"state":"REVIEWER"}],"messages":[{"id":"d015a9555b88fb69409b73b094d086d5468d6d72","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2026-06-27 17:41:43.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"24f60a5e593387be179ed52df155f70f72cdea71","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-27 17:56:46.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/af1f5ac84ff947a6863ffd6a16d1ce23\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/85f3a9c6b2f44a0692dc79b5fce8eda1 : FAILURE in 6m 54s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/40f6e31f95fb4509a5932aa9a6fcaaf8 : FAILURE in 6m 49s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/1ef9ffc4756d4636bbd242d936a664d6 : FAILURE in 12m 33s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"6e2ddf417921c96ddbff10d75b8d958b65470169","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-27 18:47:01.000000000","message":"Patch Set 1: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/9827d8b73fd94ad0bd5ea7ade41bf012\n\n- grenade https://zuul.opendev.org/t/openstack/build/ff099de962144003b0c3b250e093af96 : FAILURE in 1h 02m 48s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/9a43f1b4cfc645adaab718adc9a919aa : FAILURE in 29m 41s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/c4c84d541dd0458e80aee71622eef033 : FAILURE in 24m 03s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/bf0f8af403b8486ab54a4ecbe9a0debf : POST_FAILURE in 13m 18s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2475a27dd8ac4465b78d42ad8901979b : FAILURE in 5m 27s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/a6e0d2505ca8433480184d31442d5c3c : FAILURE in 3m 45s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/efc49ce0c7854d22baa783a8924bd39f : FAILURE in 4m 38s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/41443546f7454ca895176479a9e05fb7 : FAILURE in 4m 38s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cc96ddd8bcef43ffa1e21cc65101e748 : SUCCESS in 3m 26s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/389a9ee62314447da53e50aa5cbbea12 : SUCCESS in 3m 34s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/a0dc520cab9a430d8befa057d17da9d7 : FAILURE in 10m 19s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/65710a09a428431d97b40b584eb16b49 : FAILURE in 10m 09s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/0cdb54dbce8447ffbb2e8e6c5fc48297 : FAILURE in 14m 43s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/38edbf3c1b014097ae81032380710387 : FAILURE in 25m 05s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/05f6e10fe2c74098b22800f567f0b9b7 : FAILURE in 25m 42s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/82a35f2f0fba44f68655787c031975f8 : FAILURE in 12m 58s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/38d3f8172a374f03856b084bba09d0a6 : FAILURE in 13m 14s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/568894ed11c440c8897781db74b7d059 : FAILURE in 29m 03s\n- glance-s3-multistore https://zuul.opendev.org/t/openstack/build/76c91fb7306f4fee9b9a4e4e1d4f1a5c : FAILURE in 30m 00s (non-voting)\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/82337cd200ab42ff9940797d17685a7f : FAILURE in 25m 35s\n- glance-rally-task-uwsgi https://zuul.opendev.org/t/openstack/build/5bb8d51018354e22b4797383f9d593ca : FAILURE in 16m 44s (non-voting)\n\nWarning:\n  Comments left for invalid file glance/common/wsgi_app.py","accounts_in_message":[],"_revision_number":1},{"id":"3d548bf8f35754d86eb8b528de48a6dc1a0fcdf0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2026-06-28 11:55:20.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":2},{"id":"6e9439a489118e2cabcc7c7820abec0b541fb5c4","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-28 12:11:23.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/dcc7dc2e98ab48dea4f15d56dda55031\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/68864e643a154929909b3c8433134a06 : SUCCESS in 8m 37s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/6a101eaef36f4099a935c6cc86853273 : SUCCESS in 8m 57s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/7f2b8186557a4cfc91b90dc1b010ed79 : SUCCESS in 14m 03s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"70733c44244d1bd6ded05f3c5ac552efe1363ed0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-28 14:35:31.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1f515f4b31144d87846ffe392ed69cf1\n\n- grenade https://zuul.opendev.org/t/openstack/build/a8a250b0c92849a088e5424a22036ef7 : SUCCESS in 32m 31s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/ed3d5b2e5a2a4ec191e985b659fa3c24 : SUCCESS in 1h 04m 00s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/7158de7f6b7c440b9fd480674aec4d4e : SUCCESS in 1h 40m 57s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/1592143232a049c0bb5b210d5743129f : SUCCESS in 47m 04s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b553d236bfe74762a78d56d97cb6ba31 : SUCCESS in 3m 17s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/19d344b2d1534fb49a229e9f6e7581ca : SUCCESS in 4m 49s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/11796cd2c7224b6393e4ca7311cc0c07 : SUCCESS in 3m 01s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/8699eec0dcee4c599cebafc990965cb2 : SUCCESS in 6m 49s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fc49b3248efc476689b87d13ebdbccd6 : SUCCESS in 5m 28s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/1e50bd8be4bf44fd8429d3a6fcf8c0f3 : SUCCESS in 7m 33s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/8cada29fbd504e4997d80a431052ed70 : SUCCESS in 6m 35s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/1ef964f4448c45019df81b39a0c098f4 : SUCCESS in 4m 28s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/75175b36f7704f48a65c4bb6b852f66a : SUCCESS in 1h 05m 38s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/2b96fbc537d34adb937854282c76c5de : SUCCESS in 1h 59m 35s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/ca561fc89c174679ba383388e2ea8e18 : SUCCESS in 1h 55m 31s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/1fa4c948fa10493b89f78f29cbf72ddd : SUCCESS in 2h 34m 34s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/9143bb442d184da88a9b1fc587b34f43 : SUCCESS in 58m 05s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/8e7c7c28816e4ff781d9f22753d3533d : SUCCESS in 1h 45m 22s\n- glance-s3-multistore https://zuul.opendev.org/t/openstack/build/649fb5047bf94b619f3340960b7a1d11 : SUCCESS in 1h 51m 30s (non-voting)\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/58e199d6a9fa4cb8985f9d7ba327e24a : SUCCESS in 34m 23s\n- glance-rally-task-uwsgi https://zuul.opendev.org/t/openstack/build/15a91cb2580d495399727356e12ca3d0 : SUCCESS in 53m 04s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"187d95630bc0042b9a1d83537b3a441ea70fcd03","author":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"date":"2026-07-01 13:22:59.000000000","message":"Patch Set 2: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"27c52d2a0ac123c08d761ca918288c4c245ce160","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2026-07-01 13:47:20.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":3},{"id":"8e8d732a7a3c2ec5b73aa9b7d81591cc4dc989ee","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2026-07-01 13:48:58.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"2cc069843a42addb97733df3e7a91eac3b4b951f","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-01 14:35:23.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5a142e8512b24137aa49347770b41740\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/041a25f4587440908e005d4b94398561 : SUCCESS in 5m 39s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/ed9ab8235e164b95bd4d0ae4ad8aa616 : SUCCESS in 5m 45s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/9a80c3b2ee9e494fa619feeda9c39459 : SUCCESS in 9m 55s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"ca86aa7efc90f1a97d4acbb349d348cace7ee6a9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-01 16:06:29.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/abea0f165c9d4578a73956ffc41cc638\n\n- grenade https://zuul.opendev.org/t/openstack/build/11e3e096cc5943b7b82d9dba9c719890 : SUCCESS in 1h 07m 06s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/ecabbc5a28b94dff89e5f70e309bd646 : SUCCESS in 34m 25s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/26bf9ee55c9a44be8ddcac892da126b8 : SUCCESS in 1h 44m 47s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/b5c1535f93cd43c8a5aeb613ebe2f9c9 : SUCCESS in 1h 07m 18s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/926b542ac9f34c2899aadf2082774a24 : SUCCESS in 4m 54s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/0d57d94b25a246cba77e05b22e6c5377 : SUCCESS in 5m 15s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/68021c3f841f48c097cdbd0c43ebe174 : SUCCESS in 3m 24s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/e4835e79af9649ed8c4e1732ab454c30 : SUCCESS in 7m 38s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/00a60a1c7fde404bbc06ed9dd343a871 : SUCCESS in 5m 42s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a8e9b836be1441e58782d6b6f9b8bd38 : SUCCESS in 7m 49s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/f86dcf5aa19d44389c39a29e04b68537 : SUCCESS in 7m 00s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/d055d61d29f349a188d2fff11f21df98 : SUCCESS in 8m 38s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/a5d8241adab7457fb4e8c35d094cac80 : SUCCESS in 1h 48m 24s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/87604416527e40f3b4dee2aa9fc2de99 : SUCCESS in 1h 54m 30s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/1f7202539eaf4a099acecfb3dd5180a4 : SUCCESS in 1h 55m 20s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/0d13bf8ff929492994126f39e9acb108 : FAILURE in 2h 07m 39s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/ab7d3d2c3be04da2ae64ceb9e66a05d1 : SUCCESS in 1h 03m 31s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/851a725afa174db6a2a23afbf2c7e332 : SUCCESS in 1h 48m 04s\n- glance-s3-multistore https://zuul.opendev.org/t/openstack/build/a0e47ecf1cbd4e3b94f59c69aeb21323 : SUCCESS in 2h 05m 27s (non-voting)\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/5462a8f2c51d4f839e99cc02b0639815 : SUCCESS in 41m 21s\n- glance-rally-task-uwsgi https://zuul.opendev.org/t/openstack/build/7996a9d408164226945e6815930630bb : FAILURE in 58m 15s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"1523a5878ad35219cf3169d6f26e8a0080f4d22d","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2026-07-07 12:10:47.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"ec8a51fd22108a4be1cab5b71a0531d935901236","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-07 14:24:55.000000000","message":"Patch Set 3:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/fde2588433654447b9f72e6b91559069\n\n- grenade https://zuul.opendev.org/t/openstack/build/331f8418cf2b4959abd86e0d95c04c0e : SUCCESS in 1h 08m 01s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/6f52d29d5b7b4d67a381bc570cbcb278 : SUCCESS in 1h 03m 09s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/9ff096b6b3ac4342a99991d8b7d9d7bd : SUCCESS in 1h 48m 41s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/2cc4deb076184a4f8532baa9068b44a9 : SUCCESS in 55m 09s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4837ab9c2ef049febc3c6ff75a965f57 : SUCCESS in 4m 40s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/f75a647243c4441c9e68551f2a061b79 : SUCCESS in 4m 47s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/59a67f4e95664d74a62785bcc5ca9d97 : SUCCESS in 5m 24s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/5839ec13246e455aa030e7732f66aadc : SUCCESS in 8m 32s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b51d92cc57144b69a27d4b65a0314dbd : SUCCESS in 5m 44s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/da10182fdf784fe3b491295fa74697a7 : SUCCESS in 8m 29s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/318d22a8199c48dbbecef8aa093bca17 : SUCCESS in 7m 04s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/d15c983c2ef14d218e08ac1b40edb4f7 : FAILURE in 6m 45s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/7854f1b9e47d4082879e0812e6ab21a8 : SUCCESS in 1h 38m 32s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/ca4f53681b0d47d0bca708719551218a : SUCCESS in 2h 08m 02s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/9fe7571cf1734d89bcb3ea0e5fb626e3 : SUCCESS in 2h 00m 08s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/b402e8aa6ec2492fb6e3f1cbc1eabfe5 : SUCCESS in 1h 53m 22s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/bf261b3917ed47fbbc2888a7ebf82f2f : SUCCESS in 47m 08s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/931b9a07d1f9494ea97296821e56e4e8 : SUCCESS in 1h 54m 07s\n- glance-s3-multistore https://zuul.opendev.org/t/openstack/build/d62a62fe2d1e468788c33bb86c638808 : SUCCESS in 2h 05m 28s (non-voting)\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/7e4e1ed47b57400aa8cb3808b770d2cc : SUCCESS in 35m 31s\n- glance-rally-task-uwsgi https://zuul.opendev.org/t/openstack/build/a36f79aec2b340978c5bf37a3215d732 : SUCCESS in 58m 38s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"d8b71d71cfdebdcf2a44bfc96d4f54aa2c46f707","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-07 16:02:31.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/145e77bacfa143cab5c31cc2728fe57e\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/3ad65dec2eb54254bde2bbdee4d468d2 : SUCCESS in 6m 31s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/b1276240125c4874bb6b74430ec11cee : SUCCESS in 9m 34s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/55a112f4feae487581ff298da889e789 : SUCCESS in 15m 41s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"1ca255a0d823b392c7523e4433b7f90ae0797f33","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2026-07-09 16:49:04.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"b56f5217d03e212f21d9d1ab7877079f23bc2752","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-09 19:16:34.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/820ee5471525476b92338a96a88f7f7e\n\n- grenade https://zuul.opendev.org/t/openstack/build/067961ab16c4494d900934ec805356a9 : SUCCESS in 1h 12m 53s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/6d0c82c04055490bab376ac75465db98 : SUCCESS in 1h 03m 19s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/01ed0ff7ca6a47dcb5174d76d0867024 : SUCCESS in 1h 39m 51s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/80545a13824e47c99420a1f23e94be8f : SUCCESS in 1h 03m 47s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7ef855cc41ca45fdb279be2f30be87ae : SUCCESS in 5m 00s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/89b5b7c59c184147a5b03a7943480112 : SUCCESS in 5m 49s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/b6b7248f1b8345b7918b8f22a9f75dc4 : SUCCESS in 4m 46s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/6cc4198342aa4396a12029bd1969fd1c : SUCCESS in 8m 00s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a2d3050747b84ac0b3714ee411674768 : SUCCESS in 7m 10s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/2b13924fc6e048da88615cddf69e7c03 : SUCCESS in 9m 17s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/3aa7d191483847ae88b13119a36eee66 : SUCCESS in 8m 15s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/e46895d959b3462c9ca2f335f9c26ed2 : SUCCESS in 6m 40s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/6558e82dc6a84cd3adfbd1d2e3e3b8c4 : SUCCESS in 1h 36m 14s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/24c9a6878a2c4b8b9f85a201a7212b18 : SUCCESS in 2h 03m 53s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/9e43ef2958464273a7a89f1744970c37 : SUCCESS in 2h 09m 10s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/2af20254f9264276b07f7df7f3eba320 : SUCCESS in 2h 22m 07s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/1945a7b556c04055a073e74ed6cda8f2 : SUCCESS in 43m 44s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/acf0669e98ce454589f0a41c2b3310ec : SUCCESS in 1h 53m 32s\n- glance-s3-multistore https://zuul.opendev.org/t/openstack/build/49ea419eb16348f19353ea76cc617be0 : SUCCESS in 1h 59m 45s (non-voting)\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/737d811971b54c38b0c1fd45d973d14b : SUCCESS in 32m 21s\n- glance-rally-task-uwsgi https://zuul.opendev.org/t/openstack/build/3d3fc641cbe445d4b523d031438f8745 : SUCCESS in 52m 32s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"fb15f5693247d19478c15726a14449093a894812","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-09 20:24:42.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5620fbdd533344fea79613f467ac000e\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/37260e89082f4b8a9a78e9d8f06eebcc : SUCCESS in 4m 55s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/f03bbb79e46d4fd38c6d05ba58104a2f : SUCCESS in 5m 08s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/8ae5eb0c777144439d00450b33877077 : SUCCESS in 17m 14s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"44e27a23d16251f9f4b62846d28a9890bc9067e3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2026-07-28 15:05:00.000000000","message":"Patch Set 4: Patch Set 3 was rebased\n\nOutdated Votes:\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":4},{"id":"1d5e4956eb64b8462010d64be68e7aa99f538ea3","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-28 15:50:48.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/829ed84acad949289a415f1a430c36a3\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/d2bdaf1b131f40b8aeb1093cda9a1be3 : SUCCESS in 8m 31s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/60eb4ca381fd4dcdbe23b39f7b7a3fa0 : SUCCESS in 5m 28s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/a8fc98728d5149e5af4c815d4fba90b3 : SUCCESS in 15m 54s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"9a5f4f45e99313ee20a0a93fbbcc1ad019530071","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-28 17:07:42.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d8a815f6117d4d5d9332c5d543bf2043\n\n- grenade https://zuul.opendev.org/t/openstack/build/763ef08aaba8433e82df230cda02687e : SUCCESS in 56m 12s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/7c11b1008d4543458802e04db68561a0 : SUCCESS in 1h 03m 45s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/f17823c35cdf454f87c4709f8106cc1c : SUCCESS in 1h 24m 18s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/01fe91f34a2e4c50b45fadefc7461293 : SUCCESS in 29m 09s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/05334bf7fb2b4ed2b830135db0a2ac1f : SUCCESS in 6m 19s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/f34073e2b5d2466980267c830fe226fd : SUCCESS in 5m 23s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/beddbfd2f1914bc2989b9e621686376b : SUCCESS in 5m 12s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/cf413440b6a24e548194f6b61402cfd3 : SUCCESS in 8m 11s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3033a8466b6648b4bb9ea83224c14fd4 : SUCCESS in 4m 40s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/7417f65854764297ac2da258af4ce01d : SUCCESS in 8m 14s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/921d19f5514641b094944e46c1260073 : SUCCESS in 6m 47s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/66b9f9aeb18148d48f0bd1b417ad6641 : SUCCESS in 6m 55s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/235db6bc8f554d4a9b4a980fc484bb32 : SUCCESS in 1h 43m 17s (non-voting)\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/f1351c193c46422c94b2421310a97f78 : SUCCESS in 1h 45m 04s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/0ff59178e2114dc3ba6364ec649ac468 : SUCCESS in 1h 56m 39s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/dde2fc57a3a54141b115dec1c9719201 : SUCCESS in 1h 06m 21s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/6e82594dd94449d7bd7805036505483a : SUCCESS in 1h 26m 42s\n- glance-s3-multistore https://zuul.opendev.org/t/openstack/build/b7979f88532c4e3ba793e506ea26f86a : SUCCESS in 1h 44m 58s (non-voting)\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/04143f732aaa45c8bda52e2925728533 : SUCCESS in 32m 55s\n- glance-rally-task-uwsgi https://zuul.opendev.org/t/openstack/build/fd00358a62694eb39283673ef8930672 : SUCCESS in 53m 44s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"c39ac2d8c7922524e284125c3f649c625b2f6c52","author":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"date":"2026-07-30 13:48:25.000000000","message":"Patch Set 4: Code-Review+2\n\nthanks!","accounts_in_message":[],"_revision_number":4},{"id":"0014cc5021d10e469a6ec3384fb39af35979a2e3","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2026-07-30 14:17:22.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":4},{"id":"8e6d52d6b85982f5c2c25eade6b5bdef809a0f8a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-30 14:18:08.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"b321f8f5c180a11bd52c05af105a719b6091dfa0","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-30 16:16:34.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/756cd64cd9fc48f4ab77a5511c20c235\n\n- grenade https://zuul.opendev.org/t/openstack/build/2799ae9179d44da1b0a344e401dfa838 : SUCCESS in 52m 31s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/fac70c83fd3040489cfc640ac5f00901 : SUCCESS in 37m 32s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/b756e6fb9593470bb830aaa41bc47adb : SUCCESS in 1h 00m 17s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/d291383434c041dd9ee43155c638aa29 : SUCCESS in 57m 47s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7e550b77c2424d17884a77b5e8ccc9f9 : SUCCESS in 3m 34s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/896b33271dff4638974ce4d878d3380e : SUCCESS in 3m 42s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/faae88c989794ae7b13a02d9d4782467 : SUCCESS in 3m 44s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3d078db120524329bb39917870f0819d : SUCCESS in 4m 15s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ebd1eaa957a74fffafa143b8e9ff922d : SUCCESS in 7m 51s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/293198cb874140f98fda1c79ed005a77 : SUCCESS in 6m 13s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/4243b3c6d8544d0d9365b15e20c6e7d9 : SUCCESS in 6m 30s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/c1a0211ef4b74b1abdc2ca877f48e032 : SUCCESS in 1h 53m 27s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/7c83dc760b1f498399d3b6b019c40e31 : SUCCESS in 55m 44s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/7415b45378c84a81b2b33ad9964e16ca : SUCCESS in 1h 25m 01s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/7b1eb6d614d44931a81f937eaf9867a6 : SUCCESS in 35m 04s","accounts_in_message":[],"_revision_number":4},{"id":"da3ee1ef319b626924b09f15bd45de6327beea7e","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-30 16:16:35.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":4},{"id":"6858b030a817c8ea18a5c7416da9708120a792a6","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-30 16:17:44.000000000","message":"Patch Set 4:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9d83e0098f2042dbb7c6eaeea05b0d84\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/40e1068758d34a7790b3f255bd0dc9c5 : SUCCESS in 57s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b2d8b3a8745b44a189f3035c49db5fc9 : SUCCESS in 49s","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"41f12c876ab6e2e77c427bb17e9279cbac424357","revisions":{"71d957e29e9fbe384c95e829aea2f31d2396f513":{"kind":"REWORK","_number":1,"created":"2026-06-27 17:41:43.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/80/995180/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/80/995180/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/80/995180/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/80/995180/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/80/995180/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/80/995180/1"}}},"commit":{"parents":[{"commit":"baec4b982c01dc2d13fa85e703ebe946e843fa2d","subject":"Restore GMR report","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/baec4b982c01dc2d13fa85e703ebe946e843fa2d"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-06-27 17:39:31.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-06-27 17:40:52.000000000","tz":540},"subject":"Add support for GMR report generation to glance-scrubber","message":"Add support for GMR report generation to glance-scrubber\n\nThe glance-scrubber command has the daemon mode. Supporting GMR allows\nusers to inspect its internal status in detail while keeping\nthe process running.\n\nChange-Id: Ie4506d1130be7af32c898169338ad5e32466a8b7\nSigned-off-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/71d957e29e9fbe384c95e829aea2f31d2396f513"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/71d957e29e9fbe384c95e829aea2f31d2396f513"}]},"branch":"refs/heads/master"},"a67c697ea060be5e9c396c91428a518a3ae9fa2d":{"kind":"REWORK","_number":2,"created":"2026-06-28 11:55:20.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/80/995180/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/80/995180/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/80/995180/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/80/995180/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/80/995180/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/80/995180/2"}}},"commit":{"parents":[{"commit":"47e9f3bc5b623e70930522aa4eef81c431287ff2","subject":"Restore GMR report","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/47e9f3bc5b623e70930522aa4eef81c431287ff2"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-06-27 17:39:31.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-06-28 11:55:10.000000000","tz":540},"subject":"Add support for GMR report generation to glance-scrubber","message":"Add support for GMR report generation to glance-scrubber\n\nThe glance-scrubber command has the daemon mode. Supporting GMR allows\nusers to inspect its internal status in detail while keeping\nthe process running.\n\nChange-Id: Ie4506d1130be7af32c898169338ad5e32466a8b7\nSigned-off-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/a67c697ea060be5e9c396c91428a518a3ae9fa2d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/a67c697ea060be5e9c396c91428a518a3ae9fa2d"}]},"branch":"refs/heads/master"},"8a3b87d523e400e02ad3b20839d918c60b08df47":{"kind":"REWORK","_number":3,"created":"2026-07-01 13:47:20.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/80/995180/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/80/995180/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/80/995180/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/80/995180/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/80/995180/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/80/995180/3"}}},"commit":{"parents":[{"commit":"02d0f9f94c71480f9a14dfd51c5223a03127172d","subject":"Restore GMR report","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/02d0f9f94c71480f9a14dfd51c5223a03127172d"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-06-27 17:39:31.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-07-01 13:46:20.000000000","tz":540},"subject":"Add support for GMR report generation to glance-scrubber","message":"Add support for GMR report generation to glance-scrubber\n\nThe glance-scrubber command has the daemon mode. Supporting GMR allows\nusers to inspect its internal status in detail while keeping\nthe process running.\n\nChange-Id: Ie4506d1130be7af32c898169338ad5e32466a8b7\nSigned-off-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/8a3b87d523e400e02ad3b20839d918c60b08df47"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/8a3b87d523e400e02ad3b20839d918c60b08df47"}]},"branch":"refs/heads/master"},"41f12c876ab6e2e77c427bb17e9279cbac424357":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2026-07-28 15:05:00.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/80/995180/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/80/995180/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/80/995180/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/80/995180/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/80/995180/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/80/995180/4"}}},"commit":{"parents":[{"commit":"0e9bdd326f3a0dccc051fd4400c84cb6e9019c3b","subject":"Restore GMR report","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/0e9bdd326f3a0dccc051fd4400c84cb6e9019c3b"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-06-27 17:39:31.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-07-28 15:05:00.000000000","tz":0},"subject":"Add support for GMR report generation to glance-scrubber","message":"Add support for GMR report generation to glance-scrubber\n\nThe glance-scrubber command has the daemon mode. Supporting GMR allows\nusers to inspect its internal status in detail while keeping\nthe process running.\n\nChange-Id: Ie4506d1130be7af32c898169338ad5e32466a8b7\nSigned-off-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/41f12c876ab6e2e77c427bb17e9279cbac424357"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/41f12c876ab6e2e77c427bb17e9279cbac424357"}]},"branch":"refs/heads/master","description":"Rebase","conflicts":{"base":"02d0f9f94c71480f9a14dfd51c5223a03127172d","ours":"8a3b87d523e400e02ad3b20839d918c60b08df47","theirs":"0e9bdd326f3a0dccc051fd4400c84cb6e9019c3b","merge_strategy":"recursive","contains_conflicts":false}}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{"label:Review-Priority\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
