)]}'
{"id":"openstack%2Fglance~995412","triplet_id":"openstack%2Fglance~master~Icebc4af8b6be834db4473520af63e3ff1e50b4f9","project":"openstack/glance","branch":"master","topic":"secure-rbac","attention_set":{},"removed_from_attention_set":{"4393":{"account":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"last_update":"2026-07-22 20:48:09.000000000","reason":"\u003cGERRIT_ACCOUNT_4393\u003e replied on the change","reason_account":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"}},"8556":{"account":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"last_update":"2026-07-22 21:54:05.000000000","reason":"Change was submitted"}},"hashtags":["glance","remove-enforce-scope-flag"],"change_id":"Icebc4af8b6be834db4473520af63e3ff1e50b4f9","subject":"Remove disabling the scope enforcement","status":"MERGED","created":"2026-06-29 19:57:44.000000000","updated":"2026-07-22 21:55:03.000000000","submitted":"2026-07-22 21:54:05.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":12,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"995412-secure-rbac","meta_rev_id":"537c8b8903f04adbdc9b1f88936ea675cba0766a","_number":995412,"virtual_id_number":995412,"owner":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"value":0,"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-07-22 21:54:04.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"recommended":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"all":[{"value":1,"date":"2026-07-20 08:09:02.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},{"value":2,"date":"2026-07-20 08:23:54.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":2,"date":"2026-07-22 20:48:09.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"value":0,"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"all":[{"value":0,"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":1,"date":"2026-07-22 20:48:09.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"value":0,"date":"2026-07-17 18:35:21.000000000","_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"value":0,"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Top Priority / Holds Gate"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-06-29 20:10:06.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2026-06-29 20:27:28.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-06-30 06:02:28.000000000","updated_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"reviewer":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"state":"CC"},{"updated":"2026-06-30 16:13:57.000000000","updated_by":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"reviewer":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"state":"CC"},{"updated":"2026-07-20 08:09:02.000000000","updated_by":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"reviewer":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"state":"REVIEWER"},{"updated":"2026-07-20 08:23:54.000000000","updated_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"reviewer":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"state":"REVIEWER"},{"updated":"2026-07-22 20:48:09.000000000","updated_by":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"reviewer":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"state":"REVIEWER"}],"messages":[{"id":"c09026a9ee30c14fdfa029960a71f614b3bce623","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2026-06-29 19:57:44.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"7fefb35833e7be72555f7e29a26823f99f418223","tag":"autogenerated:gerrit:setHashtag","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2026-06-29 20:02:46.000000000","message":"Hashtag added: remove-enforce-scope-flag","accounts_in_message":[],"_revision_number":1},{"id":"2d0296a35d4d817cde2d863e46216b0266bd831b","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-29 20:10:06.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4f3989f595d646f4ac87ae954464be91\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/256462fd0cac4e2b82e456272673a9ef : FAILURE in 3m 42s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/d854cd343c0841728ab64095c6084df2 : FAILURE in 4m 05s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/55a2b6fcd7674c9eaa13b42e3b47d0e7 : FAILURE in 9m 08s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"64afcc4b86473c8765115f079f5e9dbf94252b6b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-29 20:27:28.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/734be6ab00bc46638dfd5569ad500842\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/fa14777389c44b29935dbd1136ec2e60 : SUCCESS in 2m 47s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/b45ae4ef015a4ee79e256eeef65cf3d6 : FAILURE in 27m 13s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/22063fc858784abfa0b8d13fc695761c : POST_FAILURE in 8m 47s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6843446490b64240bcbfb7e9bc3a5c30 : FAILURE in 3m 20s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/97ae5ea9489843a08c3b3000b0f93214 : FAILURE in 2m 06s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/7ec25b7392dc4193ba0922b3f7199ef1 : FAILURE in 3m 02s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/e7f60ce13cf7462da1bb4fdd0089f6e4 : FAILURE in 7m 05s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f058b19e60c340b5828f160a50be55d0 : FAILURE in 3m 25s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/7e242279bd2043c7a77e5a79bc4b0d3c : FAILURE in 3m 34s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/3481676271d843b69b3394123f0463f5 : FAILURE in 2m 22s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/33119aa3b66b4b7f94403efbd5cbb26b : FAILURE in 4m 32s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/3cd77e98272f4bec887f7c2ef29d0059 : FAILURE in 18m 31s","accounts_in_message":[],"_revision_number":1},{"id":"6f5a0bd3983cc670f63267cac2df11c45c36908c","tag":"autogenerated:gerrit:setHashtag","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2026-06-30 02:31:25.000000000","message":"Hashtag added: glance","accounts_in_message":[],"_revision_number":1},{"id":"5dcc5ebbcdc16dc2bded003ce92fac96e4d1fef7","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2026-06-30 06:02:28.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"efc99e44b5bb643b34b12bd78d3269ae33049cb6","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-30 06:13:18.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/cfca9e410203432fa951650dd3076330\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/d7e464c4add94d33872a4298d00376ca : FAILURE in 4m 23s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/3a63bbcc9d8b4b118208b158a2da731d : FAILURE in 4m 22s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/fda05ce5c61149dcb2c1caa15beaf48a : FAILURE in 9m 07s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"50687ed9867bbc03407cb520e0f44b7d714a0ef3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-30 07:00:16.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/116478fee84042c19bdc8c82e1fb43c9\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/2ecc897c80664e76bda3eef9df7cc317 : SUCCESS in 2m 44s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/f38acf166e0b4d9bb084ad7f4410ba6a : FAILURE in 51m 41s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/695a4a4481ee4111bfa00a685fe381b7 : POST_FAILURE in 8m 45s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a4e799554e6741b998ce7acfbd335d52 : FAILURE in 2m 21s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/a74a81bfe6a14fc2a8e97c042cbbc8e1 : FAILURE in 4m 16s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/b34e95e614bc4a9f97296ab94cf641f5 : FAILURE in 2m 56s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/7a75195781854e24b1a5cb390d1dc23d : FAILURE in 5m 36s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4150b3d73982472fb216baee522d26d3 : FAILURE in 3m 32s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c2272a0dd89f45009b45624fc93074b6 : FAILURE in 3m 24s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/9a8352f567834c5c87f7d5a94c009258 : FAILURE in 3m 17s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/85271d9652a44eb3b6a6ac6135d9e041 : FAILURE in 2m 12s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/89fdf1fe50304e5fad34f36dcfeb7775 : FAILURE in 7m 46s","accounts_in_message":[],"_revision_number":1},{"id":"826e21eaf87288098746476e7cc4f5adfc970690","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2026-06-30 15:24:17.000000000","message":"Patch Set 1: Workflow-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"80bd30dd6604e51338704686b04559a9e5363758","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2026-06-30 16:13:57.000000000","message":"Patch Set 1:\n\n(3 comments)","accounts_in_message":[],"_revision_number":1},{"id":"42cff8efd801baa0e46738392dea568b1a0d8ebd","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2026-06-30 16:30:12.000000000","message":"Patch Set 1:\n\n(3 comments)","accounts_in_message":[],"_revision_number":1},{"id":"12c7809c6d93f65aa2ec3cb0f44dc10a77b8ad4c","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2026-06-30 19:10:15.000000000","message":"Patch Set 1:\n\n(2 comments)","accounts_in_message":[],"_revision_number":1},{"id":"85c5df0a42636a5e95481912cb4c5b8f65c1b95c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2026-06-30 19:10:21.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n* Workflow-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":2},{"id":"dd3e132d62a802764b7ccea710099f1358c8c073","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2026-06-30 19:10:30.000000000","message":"Patch Set 2: Workflow-1","accounts_in_message":[],"_revision_number":2},{"id":"3ae4dabebf508882e0024e43a8125bc3fb7e0325","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-30 19:21:10.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a2d9a1d4d3f04824b713498b4da359d5\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/75ce1e8362b5443290b390e7d46335a7 : FAILURE in 5m 28s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/32d1e5373479418b81fa11481a322139 : FAILURE in 4m 58s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/f5c3359d405a4c8786ee98fa7253bf3c : FAILURE in 7m 55s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"0147ec414582c89937e5abb0f74021b968320337","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-30 19:48:43.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/d0995a8075b144f187bda44dea935f97\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/d067bdaff55f4d8b8b65d753b3669a26 : SUCCESS in 2m 49s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/efc851e1c1564fada592021c1ac2c38b : FAILURE in 36m 16s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/61ba996aaf0c44f4bba3fe0e1455b5ce : POST_FAILURE in 15m 51s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0d89c1689a144235b817410c1eb6f413 : FAILURE in 2m 14s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/6de269aeb28b431cb5b30eb5f4d0e38d : FAILURE in 3m 12s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/edb8cb07db2f4e26a8a1887a9dcdd8f0 : FAILURE in 2m 23s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/6304adf3f78e45fa878bcb1777c8bc79 : FAILURE in 5m 21s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a6a0c984842c4a3bbca237d0ab9a67ca : FAILURE in 2m 11s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/6ad84d81751948b6b40471ea5560f42d : FAILURE in 3m 25s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/3eb10354f193498e80ee3acbe3868ac2 : FAILURE in 2m 50s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/312da37639f044c2a19118f515c2406e : FAILURE in 2m 52s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/a2d2c6ff1c4a491b800a0811a300901f : FAILURE in 17m 05s","accounts_in_message":[],"_revision_number":2},{"id":"974406efa0fc10fcdac9c92222fc82f3ea006202","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2026-07-17 18:35:21.000000000","message":"Patch Set 2: -Workflow\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"159577212ab8f910a24d9371080247ebd5f381ed","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-17 18:49:57.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3666be847d8843598cb6b8956cb6811f\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/59af8ca0bd004e09a59f749959a55a03 : SUCCESS in 6m 03s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/642929d727be4e0dbf9e0fda74fae7ff : SUCCESS in 7m 44s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/7dfe376aa2694792993d02bc2ffe9c60 : SUCCESS in 12m 46s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"992c7c916c55ca085216388a545a3d662e51335d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-17 19:39:08.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/06ada328ddcd44ef988109fca5c32fab\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/a7cf24cb76514533b497e3c77a35e051 : SUCCESS in 2m 52s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/db4c058451744024bfc2f809eca4d0c1 : SUCCESS in 32m 59s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/918610e5c7e3404dac67d595dd78d179 : SUCCESS in 57m 27s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5086868dfea44825b5bb24ca49022662 : SUCCESS in 4m 53s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/8a4801ef23cd4221831d96071f425f65 : SUCCESS in 3m 03s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/8d3924cf38144b3b8e1e3c620cd05bae : SUCCESS in 5m 08s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/69c3a0a3ad2a4db08e2cd28e4e0fe00f : SUCCESS in 5m 05s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cb0aef015136402ba9dc0048c2e0f793 : SUCCESS in 6m 37s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a5ea82c923e6433593f626d76fa3b3a9 : SUCCESS in 9m 00s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/c0083771e3bb435a8edf2fa6b443a36d : SUCCESS in 7m 11s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/19018d94590b4fd4be1a8a18af61b421 : SUCCESS in 7m 50s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/ddf50d776cff4d82a067cf5de541bf80 : SUCCESS in 18m 18s","accounts_in_message":[],"_revision_number":2},{"id":"1023fb63e6ff5be33a70fe7dd745f63b6b09985c","author":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"date":"2026-07-20 08:09:02.000000000","message":"Patch Set 2: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"69d15ca20400a1f3aea6075e1445c1a7889ca40d","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2026-07-20 08:23:54.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"06e2c9df5e778db9b1bfa645156bcd7922b93dd6","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2026-07-22 20:48:09.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"e0758add75539ca432c242ce928c814178019812","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-22 20:48:48.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"747b5b9c0cc4b0d2a298350d72125cc12d2eb9c3","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-22 21:54:04.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/405357c06af84e879e3a484fc7ba9b8d\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/264e197da6874a1fbdf8ed0be732012c : SUCCESS in 2m 37s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/9860f9c6188d4972b41f8d58f4c41d94 : SUCCESS in 1h 00m 30s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/9c59af77900c431d9c3615bfcb4900d1 : SUCCESS in 58m 53s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/efcfadd922d447088c560b6fc96d1258 : SUCCESS in 4m 44s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/6a55004529f64e29be3ada17e64655e3 : SUCCESS in 4m 04s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/ce088f4ad7874c019a7f544044ae7a0e : SUCCESS in 7m 56s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c2d69a3f1dad4bef801669857ba21af0 : SUCCESS in 6m 41s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/dd3fb64f94784c679ac0af562d38e2c6 : SUCCESS in 7m 46s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/abd6dccb56884a6b825af2ff28ccb8a0 : SUCCESS in 6m 08s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/a3c00a4723c843e0b9daa8bd60f45e0b : SUCCESS in 7m 54s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/82e48afae44746b0963acd20f9731437 : SUCCESS in 40m 12s","accounts_in_message":[],"_revision_number":2},{"id":"cfe288f635e08259c486f2e9a1647bf81761ed17","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-22 21:54:05.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"537c8b8903f04adbdc9b1f88936ea675cba0766a","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-22 21:55:03.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b54e2aaa55634b47bc7b43b3088c9800\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3853267a286e4408b2aea67317f2fb52 : SUCCESS in 48s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/3def615c6c614fb09a16532167b6f090 : SUCCESS in 43s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"8716d6d775ad715d17d4c63db082937f0ce01a55","revisions":{"806a177becce250e272248a92a8a4cee11a86d9c":{"kind":"REWORK","_number":1,"created":"2026-06-29 19:57:44.000000000","uploader":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"ref":"refs/changes/12/995412/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/12/995412/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/12/995412/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/12/995412/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/12/995412/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/12/995412/1"}}},"commit":{"parents":[{"commit":"489bd23e2c115035caf788bd9b950d4ed4bde6e8","subject":"Update policy scope enforcement checks and testing","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/489bd23e2c115035caf788bd9b950d4ed4bde6e8"}]}],"author":{"name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","date":"2026-06-29 19:43:51.000000000","tz":0},"committer":{"name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","date":"2026-06-29 19:43:51.000000000","tz":0},"subject":"Remove disabling the scope enforecment","message":"Remove disabling the scope enforecment\n\nRBAC scope enforcement was controlled by the oslo.policy\ndeprecated config option enforce_scope which has been removed\nin oslo.policy 6.0.0. That removes the ability to disable\nthe scope enforcement which is the end goal of SRBAC.\n\n- https://governance.openstack.org/tc/goals/selected/consistent-and-secure-rbac.html#id5\n\nAdding upgrade impact in releasenotes with olso.policy version\nbump in requirements.txt.\n\nDepends-On: https://review.opendev.org/c/openstack/oslo.policy/+/986475\nDepends-On: https://review.opendev.org/c/openstack/releases/+/994789\n\nChange-Id: Icebc4af8b6be834db4473520af63e3ff1e50b4f9\nSigned-off-by: Ghanshyam Maan \u003cgmaan.os14@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/806a177becce250e272248a92a8a4cee11a86d9c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/806a177becce250e272248a92a8a4cee11a86d9c"}]},"branch":"refs/heads/master"},"8716d6d775ad715d17d4c63db082937f0ce01a55":{"kind":"REWORK","_number":2,"created":"2026-06-30 19:10:21.000000000","uploader":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"ref":"refs/changes/12/995412/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/12/995412/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/12/995412/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/12/995412/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/12/995412/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/12/995412/2"}}},"commit":{"parents":[{"commit":"489bd23e2c115035caf788bd9b950d4ed4bde6e8","subject":"Update policy scope enforcement checks and testing","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/489bd23e2c115035caf788bd9b950d4ed4bde6e8"}]}],"author":{"name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","date":"2026-06-29 19:43:51.000000000","tz":0},"committer":{"name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","date":"2026-06-30 19:09:41.000000000","tz":0},"subject":"Remove disabling the scope enforcement","message":"Remove disabling the scope enforcement\n\nRBAC scope enforcement was controlled by the oslo.policy\ndeprecated config option enforce_scope which has been removed\nin oslo.policy 6.0.0. That removes the ability to disable\nthe scope enforcement which is the end goal of SRBAC.\n\n- https://governance.openstack.org/tc/goals/selected/consistent-and-secure-rbac.html#id5\n\nAdding upgrade impact in releasenotes with olso.policy version\nbump in requirements.txt.\n\nDepends-On: https://review.opendev.org/c/openstack/oslo.policy/+/986475\nDepends-On: https://review.opendev.org/c/openstack/releases/+/994789\n\nChange-Id: Icebc4af8b6be834db4473520af63e3ff1e50b4f9\nSigned-off-by: Ghanshyam Maan \u003cgmaan.os14@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/8716d6d775ad715d17d4c63db082937f0ce01a55"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/8716d6d775ad715d17d4c63db082937f0ce01a55"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{"label:Review-Priority\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
