)]}'
{"id":"openstack%2Fglance~995444","triplet_id":"openstack%2Fglance~master~Ieda9cacdfec092422173ec645d0e21d2d26e02d9","project":"openstack/glance","branch":"master","topic":"eventlet-removal","attention_set":{},"removed_from_attention_set":{"19138":{"account":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"last_update":"2026-07-06 07:48:22.000000000","reason":"Change was submitted"},"8122":{"account":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"last_update":"2026-07-01 13:06:30.000000000","reason":"\u003cGERRIT_ACCOUNT_8122\u003e replied on the change","reason_account":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"}}},"hashtags":[],"change_id":"Ieda9cacdfec092422173ec645d0e21d2d26e02d9","subject":"Remove eventlet from requirements","status":"MERGED","created":"2026-06-30 06:51:37.000000000","updated":"2026-07-06 07:50:38.000000000","submitted":"2026-07-06 07:48:22.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":4,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"995444-eventlet-removal","meta_rev_id":"66aa863e5c2b2282a627e061ac4f0eed672b201b","_number":995444,"virtual_id_number":995444,"owner":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-07-06 07:48:21.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"all":[{"value":2,"date":"2026-07-06 05:40:42.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":2,"date":"2026-07-01 13:06:30.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"all":[{"value":1,"date":"2026-07-06 05:40:42.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Top Priority / Holds Gate"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-06-30 07:06:34.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2026-06-30 08:09:07.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-06-30 13:58:18.000000000","updated_by":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"reviewer":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"state":"REVIEWER"},{"updated":"2026-07-06 05:40:42.000000000","updated_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"reviewer":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"state":"REVIEWER"}],"messages":[{"id":"6d7b8c03db766540b94f94fac6e25435dbd633da","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"date":"2026-06-30 06:51:37.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"bb2806dad972539141bed716ec4f35d7b6d93605","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"date":"2026-06-30 06:52:26.000000000","message":"Topic set to eventlet-removal","accounts_in_message":[],"_revision_number":1},{"id":"c0a7c8dedf1ed43c2d8585832b7cbbb81629f8d8","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-30 07:06:34.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/cb9b4cda98254961825a5e60616e0c7e\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/58881b4142d849feb8b16769d0843bc3 : SUCCESS in 5m 22s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/e29f02963e2346a083a3d6236eb3f3f6 : SUCCESS in 7m 23s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/da2b8c14b8154f2c9efbf1bf74cb75b7 : SUCCESS in 13m 15s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"6702e937681513b957001b2d1a7cac62679ffbcc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-30 08:09:07.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6a063c2c48264136a4f7bf46a5fca6c1\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/810eff3acc2b43b0b059aad4807176f3 : SUCCESS in 2m 42s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/cde85222221b4b80a28d8d727bd72b19 : SUCCESS in 1h 08m 47s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/00f298fc2ebf42fda569eda1ed76b871 : SUCCESS in 33m 36s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1e428a660c5d433e8df01eb63f74ebbd : SUCCESS in 2m 48s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/3c173315f6824bb5ba057994ba5b9930 : SUCCESS in 4m 54s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/de5904f4f9e8431e8feae137283a83f3 : SUCCESS in 6m 44s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/18826978c24e46db89f0d382479bb32d : SUCCESS in 9m 04s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/165fc8b080634371a2ccc818d8b03b08 : SUCCESS in 5m 58s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/724f1652d8d44fcda07a642b03526e72 : SUCCESS in 8m 21s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/ea34b73162894ceaa1444c6741a5bbd0 : SUCCESS in 5m 09s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/2374775f9f644ccdb946ba5c36e4600b : SUCCESS in 6m 21s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/96be36bdc6d449c49e15a4eee3abf4ee : SUCCESS in 18m 27s","accounts_in_message":[],"_revision_number":1},{"id":"4acbdf5390dcc1eb9ad4d286f1349a7c71cdc5f5","author":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"date":"2026-06-30 13:58:18.000000000","message":"Patch Set 1: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"ef94c1300b79bec53476ac973ce0889fee9f1e21","author":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"date":"2026-07-01 05:27:13.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"88bdea16d9881f5cdff27608c3d04bd0c38f5cbf","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"date":"2026-07-01 06:02:38.000000000","message":"Uploaded patch set 2: Patch Set 1 was rebased.\n\nCopied Votes:\n* Code-Review-1 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n\nOutdated Votes:\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":2},{"id":"a7d44ddbe74a0c845161d29b0f98ca08d821a33a","author":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"date":"2026-07-01 06:03:29.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"2d547993b0858dda850bbefecbc964d6acacd6e8","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-01 06:19:19.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a4ff6d0ea3e44519ba6af30fe08e7284\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/ba5fa7e85c314cf1906626feca97a915 : SUCCESS in 6m 31s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/c95b6bfa253d480391a36d22bc4dfd3c : SUCCESS in 8m 21s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/962a406d969341598f8e9f1464b95a38 : SUCCESS in 13m 46s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"58e74a8902fab7c8ccb8aa46aee98a7062803e01","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-01 06:43:12.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8a3c40876f3445199b37306437fab120\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/ef05ae9899c34faeba43575a092ba924 : SUCCESS in 2m 53s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/6f760882cb0549fc83bcae32f412bb58 : SUCCESS in 37m 41s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/494f0f1da9f345d78766a0d39c9197f1 : SUCCESS in 37m 38s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/63404ce7f23f442dab223fdd52bf086b : SUCCESS in 5m 11s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/4d030305929f450898acbf68323d0eea : SUCCESS in 4m 36s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/5cac4136b2194e568bbde03b6fc66fec : SUCCESS in 5m 55s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/e110de5a2ded47cf95871683a8c8f7ce : SUCCESS in 7m 44s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1361aea32935448a89abc43c2a44b3e2 : SUCCESS in 3m 30s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/6e426136e8594999aa05f7620533b48d : SUCCESS in 6m 03s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/f3be19cadfca4988aa7e933221baa132 : SUCCESS in 5m 22s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/1b91b247c355471f99abae9f2e829487 : SUCCESS in 6m 39s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/7c8f9a9497c045ecb8f8e31d76e5457d : SUCCESS in 35m 05s","accounts_in_message":[],"_revision_number":2},{"id":"e325c72f72d1cd2860cb30e225a31918098af891","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"date":"2026-07-01 10:20:03.000000000","message":"Uploaded patch set 3: Patch Set 2 was rebased.\n\nCopied Votes:\n* Code-Review-1 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n\nOutdated Votes:\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":3},{"id":"787f3803febb51465a261810f6a28eccc383e036","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-01 10:34:55.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8b90170aee2545ee9e93c02a68fffa92\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/fb457ed0d11a4923b37a99e33f071f5b : SUCCESS in 5m 16s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/25ae9cb11b1645ef859fd5b9d89cd995 : SUCCESS in 6m 04s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/06749b3a52bd4d27b9896c3633353b79 : SUCCESS in 10m 51s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"538e9143a7e4816f990dc344f1b19c9b233a8990","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-01 11:30:32.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/dfc455660b7d4c61a41772b2d7945576\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/d2272864b6284c7fa5f38407be8cb873 : SUCCESS in 2m 59s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/652a626bcfa54e519bfb53687902fb1e : SUCCESS in 1h 04m 17s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/26a6b376de24453e86f4f21ba0537836 : SUCCESS in 51m 57s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/92a99f8be328421bbb1484182d7a43e2 : SUCCESS in 5m 40s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/5e147a4cf58a445688637151f07fefd9 : SUCCESS in 4m 38s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/6eee6ea774f74417aef74290cda514cc : SUCCESS in 5m 11s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/bb2e41e717e348cd9c04038daf2c3afb : SUCCESS in 7m 26s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bc2b7257f342484c952dbcf50e71cab4 : SUCCESS in 6m 45s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/049aba4171f6421ab54e1b8679090bb4 : SUCCESS in 4m 29s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/4df4a2997d4c430ebb0447e40739ffb6 : SUCCESS in 6m 04s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/ab47447495ad488ba9bdd3b051d867ca : SUCCESS in 6m 38s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/39b1d255e4cc49a7918d3f49d641811e : SUCCESS in 34m 17s","accounts_in_message":[],"_revision_number":3},{"id":"014570a97186cd717f7da39234e6aae412fe6d3c","author":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"date":"2026-07-01 13:06:30.000000000","message":"Patch Set 3: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"4a41baa941beee7d64fd5fa6ca98ce34a00a155a","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2026-07-06 05:40:42.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"19371ba23226fe27bec42a2ceb4b370af91f3fac","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-06 05:41:15.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"2a8ad24c0e427aec284058e5f9ec4f2ee3e61db2","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-06 07:48:21.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b7a95f8776404bd08680a47513b620ee\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/a6ede3d129994bea8dbbc82bcb437b3f : SUCCESS in 3m 02s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/fce4e510141b46baa8bd27ed0b26961a : SUCCESS in 31m 39s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/290527688ac64b7881ff93261329606f : SUCCESS in 51m 27s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7e48a7600b444fc9bb742a7b508eaa32 : SUCCESS in 2m 57s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/794de266f0d34d99b20e58dbb48674ca : SUCCESS in 4m 38s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/a50bec03b10a4f65b3599d9b5f00e335 : SUCCESS in 6m 22s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d40a856d09c94ac992cd4cf0464e1f0f : SUCCESS in 5m 22s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/52a29ad5844c450ba8850a4b09dff8c2 : SUCCESS in 7m 35s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/98e23cc14711429dae8f6f3d59f8aee1 : SUCCESS in 4m 27s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/0f64d9572fd3441e8854babda28ce081 : SUCCESS in 4m 18s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/1d3bb4d491d845548df2a6c21d59b6ee : SUCCESS in 33m 13s","accounts_in_message":[],"_revision_number":3},{"id":"5e6b55c0ae1e3f433c946d3263297e8888580beb","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-06 07:48:22.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3},{"id":"66aa863e5c2b2282a627e061ac4f0eed672b201b","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-06 07:50:38.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/70dc4dfd08fd47a99faa085c04cc33e9\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cd278c017bf543ec8a0b447a72a4513b : SUCCESS in 1m 06s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/5eff9014f6644dd4bbf80e443c21f497 : SUCCESS in 1m 05s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"e3af46ba33900c1ef725432f0c5ec560c2578d31","revisions":{"fa68ea2450babf07723335f40474123cf21d2469":{"kind":"REWORK","_number":1,"created":"2026-06-30 06:51:37.000000000","uploader":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"ref":"refs/changes/44/995444/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/44/995444/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/44/995444/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/44/995444/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/44/995444/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/44/995444/1"}}},"commit":{"parents":[{"commit":"7f6f19fdbbcb33a94e446cdcd1ddc541b559d09a","subject":"Replace futurist thread pool with concurrent.futures","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/7f6f19fdbbcb33a94e446cdcd1ddc541b559d09a"}]}],"author":{"name":"Pranali Deore","email":"pdeore@redhat.com","date":"2026-06-30 06:43:36.000000000","tz":0},"committer":{"name":"Pranali Deore","email":"pdeore@redhat.com","date":"2026-06-30 06:44:33.000000000","tz":0},"subject":"Remove eventlet from requirements","message":"Remove eventlet from requirements\n\nGlance no longer uses eventlet at runtime after the\neventlet-removal work.\n\nDrop the direct dependency and document the change in release\nnote.\n\nChange-Id: Ieda9cacdfec092422173ec645d0e21d2d26e02d9\nSigned-off-by: Pranali Deore \u003cpdeore@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/fa68ea2450babf07723335f40474123cf21d2469"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/fa68ea2450babf07723335f40474123cf21d2469"}]},"branch":"refs/heads/master"},"0d49dffd1383e7749fa9e8911680127aac764e33":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2026-07-01 06:02:38.000000000","uploader":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"ref":"refs/changes/44/995444/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/44/995444/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/44/995444/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/44/995444/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/44/995444/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/44/995444/2"}}},"commit":{"parents":[{"commit":"8095efb7e4cb84c6abaef53d2f4b778988cd18cb","subject":"Merge \"Replace futurist thread pool with concurrent.futures\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/8095efb7e4cb84c6abaef53d2f4b778988cd18cb"}]}],"author":{"name":"Pranali Deore","email":"pdeore@redhat.com","date":"2026-06-30 06:43:36.000000000","tz":0},"committer":{"name":"Pranali Deore","email":"pdeore@redhat.com","date":"2026-07-01 05:27:01.000000000","tz":0},"subject":"Remove eventlet from requirements","message":"Remove eventlet from requirements\n\nGlance no longer uses eventlet at runtime after the\neventlet-removal work.\n\nDrop the direct dependency and document the change in release\nnote.\n\nChange-Id: Ieda9cacdfec092422173ec645d0e21d2d26e02d9\nSigned-off-by: Pranali Deore \u003cpdeore@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/0d49dffd1383e7749fa9e8911680127aac764e33"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/0d49dffd1383e7749fa9e8911680127aac764e33"}]},"branch":"refs/heads/master"},"e3af46ba33900c1ef725432f0c5ec560c2578d31":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2026-07-01 10:20:03.000000000","uploader":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"ref":"refs/changes/44/995444/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/44/995444/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/44/995444/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/44/995444/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/44/995444/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/44/995444/3"}}},"commit":{"parents":[{"commit":"01b828cbf371c3bfbbc5ce8bb7d04d64a194183c","subject":"Drop deprecated standalone server config options","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/01b828cbf371c3bfbbc5ce8bb7d04d64a194183c"}]}],"author":{"name":"Pranali Deore","email":"pdeore@redhat.com","date":"2026-06-30 06:43:36.000000000","tz":0},"committer":{"name":"Pranali Deore","email":"pdeore@redhat.com","date":"2026-07-01 10:16:49.000000000","tz":0},"subject":"Remove eventlet from requirements","message":"Remove eventlet from requirements\n\nGlance no longer uses eventlet at runtime after the\neventlet-removal work.\n\nDrop the direct dependency and document the change in release\nnote.\n\nChange-Id: Ieda9cacdfec092422173ec645d0e21d2d26e02d9\nSigned-off-by: Pranali Deore \u003cpdeore@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/e3af46ba33900c1ef725432f0c5ec560c2578d31"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/e3af46ba33900c1ef725432f0c5ec560c2578d31"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{"label:Review-Priority\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
