)]}'
{"id":"openstack%2Fglance~995723","triplet_id":"openstack%2Fglance~master~Id1238e1557657e525dc232e7fd917ac8779f38d0","project":"openstack/glance","branch":"master","attention_set":{},"removed_from_attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2026-07-03 10:47:56.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Id1238e1557657e525dc232e7fd917ac8779f38d0","subject":"Drop redundant space in an exception message","status":"MERGED","created":"2026-07-01 17:07:09.000000000","updated":"2026-07-03 10:50:00.000000000","submitted":"2026-07-03 10:47:56.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"995723","meta_rev_id":"c32203305cbbc37b958313dc7f9b364ac2413877","_number":995723,"virtual_id_number":995723,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-07-03 10:47:56.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"all":[{"value":2,"date":"2026-07-01 17:13:37.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"all":[{"value":1,"date":"2026-07-01 17:13:37.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Top Priority / Holds Gate"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-07-01 17:13:37.000000000","updated_by":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"reviewer":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"state":"REVIEWER"},{"updated":"2026-07-01 17:22:30.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2026-07-01 19:28:31.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-07-03 05:52:08.000000000","updated_by":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"reviewer":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"state":"CC"}],"messages":[{"id":"491a8c2c84df2fb13233a1d52492ce3070a837d1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2026-07-01 17:07:09.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"59d7d3c251fa0b1e1ee6f503f9175bcc38c9f3b1","author":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"date":"2026-07-01 17:13:37.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"8f23d9ade830c15dfefb1d82c9e0d41f34b9dc43","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-01 17:22:30.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/17695c608140489d8541dd7dba147d11\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/49e29ab57eea4bfabda88669936dcf96 : SUCCESS in 7m 26s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/b080d0e988ce4e67a1dc9b7ab2e80022 : SUCCESS in 7m 36s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/bfbd6c309ec84c79a1a091da64082f4e : SUCCESS in 12m 41s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"35ad15eb7982e3f7868f6ab96ca3e8f57b27396c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-01 19:28:31.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/239a0884dcf84264a2fca3db016a60cc\n\n- grenade https://zuul.opendev.org/t/openstack/build/1c4c6b920a0c44bf9efcffedc2807318 : SUCCESS in 34m 42s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/f144fbaa7e9148639b0a6c47fe50b2ed : SUCCESS in 50m 19s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/cae3528082c54473a7b2e63d21d2b850 : SUCCESS in 1h 36m 44s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/38ffb2ed35d14cc692e40301f76f964c : SUCCESS in 34m 18s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cf511e0c874243d3841d792d50f18a21 : SUCCESS in 2m 49s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/d521250b58404b2da8a7912db909193b : SUCCESS in 4m 58s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/9eb2a330e30047d8bed2d0b347426edb : SUCCESS in 5m 41s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/a3e7ce8be449409980eeaa04d937739d : SUCCESS in 4m 16s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7c7eb09e507b4cf992b3aeb304067296 : SUCCESS in 3m 50s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a15f85262ec2407bbe9ea87a0e3eb1b8 : SUCCESS in 3m 31s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/a52c4d29a46e4315b919eb5a0cd1f1f8 : SUCCESS in 6m 41s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/2320e2cd4f8141eeb8f679b4f2a6ef93 : SUCCESS in 6m 03s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/059a9267b9414c9ebc3e271d72c9bbd3 : SUCCESS in 1h 49m 04s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/6489c166a4754d21b28501fe81e245ba : SUCCESS in 1h 57m 56s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/5d2b1dad764a41d0bfe219328863ea53 : SUCCESS in 2h 15m 48s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/baf191c1975a4312ae7810e7b2b32484 : FAILURE in 2h 13m 05s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/a278c2f668494db6aab6bd813c2adce0 : SUCCESS in 56m 38s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/0196a901873244f0be3a435e0e99b988 : SUCCESS in 1h 04m 06s\n- glance-s3-multistore https://zuul.opendev.org/t/openstack/build/19e979a7426447e7a9dd906f298ac974 : SUCCESS in 2h 01m 36s (non-voting)\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/2bcd5a1ee795439b92485d618343f1da : SUCCESS in 35m 46s\n- glance-rally-task-uwsgi https://zuul.opendev.org/t/openstack/build/6a9206a9b8f340539325f749c7bb5b5e : FAILURE in 1h 07m 22s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"d4cc71b17fec11989a08c9de97a08fc4202da873","author":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"date":"2026-07-03 05:52:08.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"1ec8ae7bdcc7c1ef395225b52dd4f9f9a35e8709","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-03 06:07:08.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/494ae162caf34f5ba9bcba9f61faff1a\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/c1a14ee1cf7c4da2803147ecf4d2a77d : SUCCESS in 7m 28s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/f18c1faf65f942f2915b42e43ee9199a : SUCCESS in 7m 17s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/bd9023c716ac45f7a24ac3a3d6b6a98c : SUCCESS in 12m 44s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"82345d5bbaba95073108938100a1c2d9f28d9aba","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-03 08:29:58.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/24e3c18320d7465a8cb100a2ea4097d7\n\n- grenade https://zuul.opendev.org/t/openstack/build/1ddf9951be7d436d85dce101a2e36d0e : SUCCESS in 1h 07m 12s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/e7cae758394546ecb5c0d6bd917e50b3 : SUCCESS in 50m 15s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/d276c455278c4addbaa212a133d26eb9 : SUCCESS in 1h 47m 40s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/fc62815fff464879a723a5398c7ca9e8 : SUCCESS in 54m 42s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1e250b18749f4fd0bccf2ac9fbb19b6e : SUCCESS in 3m 03s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/6243cc85c4ce448e9c7c8919692ef7bd : SUCCESS in 4m 48s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/12ed641cc22d44c6b64d06ded117e5fd : SUCCESS in 2m 34s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/fb895a8e5ff84fc381295d67326e4259 : SUCCESS in 8m 16s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8ae40a961a894bc481fa3e6575f85cdf : SUCCESS in 5m 09s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/326d5843d55e4ff58199a25a465fb933 : SUCCESS in 7m 02s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/62acfccc6f504b37b8baac7375a54e08 : SUCCESS in 6m 46s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/b3b98b4b9d3b456e8c74ef71dcc5ce6b : SUCCESS in 6m 05s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/edcc19b9c5884ec490ccd7386bd3bd08 : SUCCESS in 1h 56m 56s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/f4b1c3a3912e44e1b9a13f52fd8c917e : SUCCESS in 1h 05m 10s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/5a2e9392cdcb4e3eab8eebbbc4cfa9bd : SUCCESS in 2h 11m 07s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/6097e482199e4455a6e7fae63b19a800 : SUCCESS in 2h 31m 58s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/8a7de3f03f844e8c82e6d382f10ba185 : SUCCESS in 1h 09m 16s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/29d5291a58dc4298b6f9fa6f8dbc941c : SUCCESS in 2h 01m 27s\n- glance-s3-multistore https://zuul.opendev.org/t/openstack/build/24ea6b4c4648466ca91bd543ca760cab : SUCCESS in 1h 46m 54s (non-voting)\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/2a995c5f83e7481d8eb3525995fd0572 : SUCCESS in 35m 10s\n- glance-rally-task-uwsgi https://zuul.opendev.org/t/openstack/build/8022ed0eeac04f3e98cef40ec5683234 : FAILURE in 1h 01m 28s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"14608f746027934d5e4e4dd39cafd9a2d1ebdfa1","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-03 08:30:45.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"768d6e2618a82c9fcc6c2663c7733851d76a6b2a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-03 10:47:56.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f50f2eed71394ef6bef156d65771964b\n\n- grenade https://zuul.opendev.org/t/openstack/build/908449bfe2a44c6da4bc0a20c299d42b : SUCCESS in 1h 03m 22s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/136612eaf12e4d34958e76de5e7d97f5 : SUCCESS in 1h 13m 50s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/b7502372c67f465ebeae244878279492 : SUCCESS in 1h 50m 38s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/8f75ddd9a3aa43f185d30844e43d3b10 : SUCCESS in 1h 14m 01s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3069e94214a74daf82a9d23536351139 : SUCCESS in 5m 00s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/712cf95ae2874a79b6eb916c8d541e1d : SUCCESS in 5m 14s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/2ec24ded1bd94eddaff94c065b84d588 : SUCCESS in 3m 10s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5e5932a1b2c24028b82c34a8ba0fc86a : SUCCESS in 5m 18s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e0a296fad3dd45978748a47f96bd645c : SUCCESS in 4m 15s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/275217091eb647b3b546bad2397d0e30 : SUCCESS in 4m 55s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/7119353556cc4670ae421dbd506edf90 : SUCCESS in 6m 56s\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/0e363bb3df464674b30d27a37879d1e0 : SUCCESS in 2h 00m 49s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/351702169b054a1395e72cb6e2a8be00 : SUCCESS in 2h 12m 10s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/a066fa5666b94a038d1e5b066eb50f5a : SUCCESS in 1h 02m 34s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/3809d62bba904cc98123c518a4f0400f : SUCCESS in 1h 57m 06s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/38e36fe164b84bdba516035b597184bf : SUCCESS in 23m 36s","accounts_in_message":[],"_revision_number":1},{"id":"a55e410151a24c641d34638d728c74ebe333328a","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-03 10:47:56.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"c32203305cbbc37b958313dc7f9b364ac2413877","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-03 10:50:00.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b8610c3ab9924676bf3b29dcf1e1f0c2\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c1184c6e60474049a1ebee0461b6627c : SUCCESS in 59s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"0f4b293db11f299baaf6b9221940a2a23d3ba799","revisions":{"0f4b293db11f299baaf6b9221940a2a23d3ba799":{"kind":"REWORK","_number":1,"created":"2026-07-01 17:07:09.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/23/995723/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/23/995723/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/23/995723/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/23/995723/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/23/995723/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/23/995723/1"}}},"commit":{"parents":[{"commit":"e20488b1a6cfd59573130f56443dd4075f157989","subject":"Merge \"Validate enabled_import_methods\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/e20488b1a6cfd59573130f56443dd4075f157989"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-07-01 17:06:16.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-07-01 17:07:01.000000000","tz":540},"subject":"Drop redundant space in an exception message","message":"Drop redundant space in an exception message\n\nReplace two spaces by a single space.\n\nTrivial-Fix\n\nChange-Id: Id1238e1557657e525dc232e7fd917ac8779f38d0\nSigned-off-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/0f4b293db11f299baaf6b9221940a2a23d3ba799"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/0f4b293db11f299baaf6b9221940a2a23d3ba799"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{"label:Review-Priority\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
