)]}'
{"id":"openstack%2Fglance~996220","triplet_id":"openstack%2Fglance~stable%2F2025.2~Ic6de91db83c9dda1d5d735566876b90c1ac6b2e3","project":"openstack/glance","branch":"stable/2025.2","attention_set":{},"removed_from_attention_set":{"9303":{"account":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"last_update":"2026-07-08 11:17:57.000000000","reason":"\u003cGERRIT_ACCOUNT_9303\u003e replied on the change","reason_account":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"}}},"hashtags":[],"change_id":"Ic6de91db83c9dda1d5d735566876b90c1ac6b2e3","subject":"Drop Jammy nodeset from enforce-scope-new-defaults job","status":"MERGED","created":"2026-07-07 08:11:40.000000000","updated":"2026-07-08 15:13:18.000000000","submitted":"2026-07-08 15:11:02.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":4,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"996220","meta_rev_id":"d93bcc675abde3656cb9996df6d517a3a5f7ecea","_number":996220,"virtual_id_number":996220,"owner":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-07-08 15:11:02.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"all":[{"value":2,"date":"2026-07-07 13:07:33.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"all":[{"value":1,"date":"2026-07-07 13:07:33.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Top Priority / Holds Gate"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-07-07 10:31:57.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-07-07 13:07:33.000000000","updated_by":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"reviewer":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"state":"REVIEWER"}],"messages":[{"id":"46697b8da5b963f1f60e28b73fe34accfad2be26","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2026-07-07 08:11:40.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"705c2739c806b3b9b927d7ebabbf9759d2e7f894","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-07 10:31:57.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1b177c17c30d4778acb584620bc75647\n\n- grenade https://zuul.opendev.org/t/openstack/build/f5dc0af62e784ce4a4b7b1a79e03df1c : SUCCESS in 1h 03m 08s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/701d9e3a071a4c769ecc80f9caf60aeb : SUCCESS in 1h 28m 01s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/7b9d8c933f774f49ab522852d4224286 : SUCCESS in 1h 06m 09s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9681a5ffe7484118806ef6822a28698d : SUCCESS in 4m 32s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/2c8fc2ca4e5c483ba40e6392f2f344b4 : SUCCESS in 4m 48s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/2530763930424b01aa2cf63ae021d2f1 : SUCCESS in 5m 27s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/628507ba6fe34605a355d4cbedccaa9d : SUCCESS in 8m 17s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f9ea602fbe16419cb2e1916e364b15fe : SUCCESS in 6m 23s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d2f44fff6a604a7ba03ac0034efe17e7 : SUCCESS in 8m 35s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/e4228bb6508c419e868ef9dcb21507c5 : SUCCESS in 7m 28s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/2d907616be27432a800963f57eb6d132 : SUCCESS in 7m 27s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/10cb70936922409d9e9f94bdc8ea0e7a : SUCCESS in 2h 06m 20s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/f881ddee68bf4e84aff74662157e6bda : SUCCESS in 2h 14m 08s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/8a86ea48bc0c426fbc108fa58979858d : SUCCESS in 1h 57m 48s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/aaaf4ec6239a4d329fa98b02b71a5a15 : SUCCESS in 1h 48m 58s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/2bf0bfc60c394d178d44308205b73ebb : SUCCESS in 58m 54s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/9e5da3c8e7a9424f8c1254d0b037ccf4 : SUCCESS in 1h 55m 03s\n- glance-grenade-centralized-cache https://zuul.opendev.org/t/openstack/build/8c547700c5d24920879d98cf8dfc88e6 : FAILURE in 12m 13s (non-voting)\n- glance-s3-multistore https://zuul.opendev.org/t/openstack/build/1cfcf4a5fd9543cfa8545b50b5ad1b7a : FAILURE in 31m 11s (non-voting)\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/c4750fb2f06449b881c7db0e99028bb9 : SUCCESS in 32m 01s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/cc22a12611564d76a446073c04de2f99 : SUCCESS in 1h 15m 45s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"0fedd1c589378ab6c134fcb3a9d413c8d6cf0a99","author":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"date":"2026-07-07 13:07:33.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"5ab1f3db18be12e8ce035b1385869c03493e2457","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-07 13:08:33.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"4cb95158620ebaf76f924ae05b339f7d5e9260cb","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-07 15:30:01.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/9dba77b4c53241caa486b7ebcd270a9c\n\n- grenade https://zuul.opendev.org/t/openstack/build/4dd63f85a348451d8b05e4a0c47bc75e : SUCCESS in 59m 49s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/844a3fe477dc46b1b5ca80cd8b19aacb : SUCCESS in 1h 41m 03s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/0ef3177551f749bfb4c9d058543e49c5 : SUCCESS in 1h 03m 41s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f491c46e164448e29f1880423ea60399 : SUCCESS in 5m 10s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/8e33bab255764ff5bd3863f965f33738 : SUCCESS in 3m 10s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/1a3e6f473fdf410a9af0e2e3bc8b8479 : SUCCESS in 6m 15s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5a5beadc1a964e53b4d0271cbf379178 : SUCCESS in 7m 00s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/393f8cf192c945149b5cb8427526dcf5 : SUCCESS in 3m 59s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/d976f380aae8445f8b51eeb19c47aa43 : SUCCESS in 6m 51s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/7120f75cf24b40268c82a55226f2f612 : FAILURE in 7m 27s\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/de7e81b3313e4e38925dd94a70167a17 : SUCCESS in 2h 03m 12s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/118ae04c76f842ab80385a7b785ae240 : SUCCESS in 2h 16m 26s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/936e8118fcc84d828d34ed9b9913243f : SUCCESS in 59m 19s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/26dc4c9686e94d13bf44775aa0870c9b : SUCCESS in 1h 41m 48s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/f3cec093019e49b5b7e070ca9dc6a543 : SUCCESS in 37m 15s","accounts_in_message":[],"_revision_number":1},{"id":"22de915e27f9efa932aafa285f1d038d1ff1b303","author":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"date":"2026-07-07 15:48:46.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"04063ef5fc5c054766bc7bcc35486428b291c097","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-07 18:07:09.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/54205d29a9ad47ceb7ab4b44ed9a849f\n\n- grenade https://zuul.opendev.org/t/openstack/build/5cedc03b1761415dbf5b157d4efec32c : SUCCESS in 1h 08m 00s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/f1d6ebaa3d62401e8a6a89d95e549afc : SUCCESS in 1h 27m 42s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/ccc50fedaedc4de193d748b5fc7735a3 : SUCCESS in 1h 06m 46s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a2757ee2b386429d940ae2b9d1ddfb3a : SUCCESS in 6m 54s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/ec24715395574d0eabe4374c7877fdd9 : SUCCESS in 7m 02s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/ef66311817394aca8449569db773b21e : SUCCESS in 4m 21s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/7f3d91b241f74ac99204dceb7d20cec0 : SUCCESS in 10m 40s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/49ee5c4fbca94005ae32d7941b58822d : SUCCESS in 8m 05s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/4f00017a1f784367b76c88c7e75c003a : SUCCESS in 8m 56s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/778907fa911746b693c05a6d4103db6d : SUCCESS in 9m 16s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/88f6a0ec42bc4313845fe9a5176d28a2 : SUCCESS in 5m 14s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/c4c76bac37a648c7955f32a07053e1ae : SUCCESS in 1h 11m 55s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/9bc1bad14df247afb24a6be08b13a08f : SUCCESS in 2h 09m 21s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/f579b67779504efcbb44e72b35c766e4 : SUCCESS in 1h 56m 35s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/3443dd0ccbc24cc8afb1816c30f0c6ea : FAILURE in 2h 12m 41s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/28b59e9eb8ed48229d730a057ccdcf53 : SUCCESS in 56m 43s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/7f07a5aa704e49cfad0eba3779e11547 : FAILURE in 1h 04m 54s\n- glance-grenade-centralized-cache https://zuul.opendev.org/t/openstack/build/1b0ef8baad074bff855715eb7211a9ab : FAILURE in 13m 16s (non-voting)\n- glance-s3-multistore https://zuul.opendev.org/t/openstack/build/9c2a408810ff4fb886c1406f7c808afb : FAILURE in 25m 19s (non-voting)\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/0ae6612eca6b4ec1b7cf7b46914fdf51 : SUCCESS in 32m 55s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/c3258faff4514c91b6707c403eb52a78 : SUCCESS in 1h 05m 12s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"eb45f1b07c69609605d4c95f3b1088dc6e95f121","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2026-07-07 18:11:22.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"9fa8868e5d7386340f93da9b540a21cdfee84c9f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-07 20:43:27.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/c3607e8e8c984fc2b16f29453626f6b9\n\n- grenade https://zuul.opendev.org/t/openstack/build/f9002a9853a1477b9f7446efcd5ca5c6 : SUCCESS in 1h 05m 29s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/f7d11e2183ad4c958ca25563cb559ba0 : SUCCESS in 1h 20m 45s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/408c39bf2efa4deba62354a4a9549222 : SUCCESS in 57m 55s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7fcb794141e447f8bde37128506f6abf : SUCCESS in 4m 52s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/b8bcdfba94fc47edbe811a0d3b6960fe : SUCCESS in 5m 10s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/2eb9f87a91384adfa51516307c480624 : SUCCESS in 6m 02s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/d60e217bd25d430eb34e688f4af4d814 : SUCCESS in 9m 16s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/217adddedf6541da9b04cb616f85188b : SUCCESS in 7m 38s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f33e62b3e65846279ca7ff6f240e3455 : SUCCESS in 8m 27s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/16656a6dbc9840d8bd1c36de73ad303c : SUCCESS in 7m 54s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/8fdd129e3dd5429dad168f12f2f05211 : SUCCESS in 9m 44s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/9cd8b4ff0bda428f8d445a70b4b11514 : SUCCESS in 1h 47m 10s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/2a0e2505fe2f4f37aa14e23fb636b844 : SUCCESS in 1h 50m 27s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/d058c922b8d247a38411af9534c20585 : SUCCESS in 2h 24m 40s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/8b18b788496b40f4a75d2fb2fa577722 : SUCCESS in 2h 13m 33s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/08449eb267f546b790902f3bfd4e4d84 : SUCCESS in 1h 02m 55s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/c960081398024f4c94e1a53fdb6d3b86 : FAILURE in 1h 48m 07s\n- glance-grenade-centralized-cache https://zuul.opendev.org/t/openstack/build/d4a6c3e7bd1e44dabecbdcc835720669 : FAILURE in 18m 30s (non-voting)\n- glance-s3-multistore https://zuul.opendev.org/t/openstack/build/6c60c2d218bf42de9e50c9278a5b4453 : FAILURE in 28m 31s (non-voting)\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/b4671aafb6aa40a1873373ff6647a2b3 : SUCCESS in 37m 07s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/268b8d53163f4f568fdf86a580f2d7cd : SUCCESS in 1h 14m 13s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"a0fa214f6a2601463ca591c3ead38ab2f91dfa5c","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2026-07-08 00:55:29.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"6bb6a618e656ce91dc1912522daedc036a49ded5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-08 03:20:37.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/741d1f28d8d143a1928fc9663fb0d1ac\n\n- grenade https://zuul.opendev.org/t/openstack/build/acfc2df3d81740e5b91074be81a76902 : SUCCESS in 1h 01m 44s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/8ecfa376e6b34df7a403e909977f2bbd : FAILURE in 32m 12s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/91b87ad4dd0443abb05b3178afa9d285 : SUCCESS in 32m 48s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/42f009a6993c44f68ee294a086f44346 : SUCCESS in 3m 05s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/b1fe6c254ac540b2a898a4d7690f7f5b : SUCCESS in 5m 23s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/499c7d0def454b4fa64e794e531321b1 : SUCCESS in 3m 34s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/dacf0c880a2f439b842a407a674ad3f1 : SUCCESS in 7m 45s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e4eec42ae222443fa4b6a96e87aa8bd0 : FAILURE in 16m 47s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/0ca0dc03f639414b84a5aeca48f4e6e3 : FAILURE in 16m 49s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/c257ccc155c040cb80578bac570cb57e : SUCCESS in 7m 04s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/f6c44b3f83a94cb4abb308cb00b6bc67 : SUCCESS in 7m 22s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/125a2dfe184e4c00bb27d625bef79e63 : FAILURE in 26m 13s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/42238ad1300c4d3db8b3eb5b3ff28df3 : SUCCESS in 1h 10m 04s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/754b00fd8f8b46138506698a645a9de5 : SUCCESS in 2h 01m 41s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/80bec524ad114b30bbddcf099e5520e1 : SUCCESS in 2h 19m 23s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/96518f292a424a70af2fca9a9ecd648f : SUCCESS in 56m 58s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/acd2610f9d6f4d4faa89431e2963e5b3 : SUCCESS in 2h 00m 24s\n- glance-grenade-centralized-cache https://zuul.opendev.org/t/openstack/build/a865243d41c84bf894091c6aa4cb2533 : FAILURE in 9m 29s (non-voting)\n- glance-s3-multistore https://zuul.opendev.org/t/openstack/build/4280d9daf046478ab7ebafb7e70718e6 : FAILURE in 16m 29s (non-voting)\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/986ef18a63cd47cc8a0c36d41fff1942 : SUCCESS in 36m 34s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/3b8829385a5542bb90c743b5bf54c1dc : SUCCESS in 1h 19m 39s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"bfe1394b9219a0fb7c2ce1eeb56166ab30d638a1","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2026-07-08 06:30:04.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"0912860aef5ff553ea89fd2f3cc9968f0f55bd1c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-08 08:58:53.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b746834c9e164a4c9054d985be76146e\n\n- grenade https://zuul.opendev.org/t/openstack/build/e06b27b212dc458bbd47359aad72f058 : SUCCESS in 1h 02m 34s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/ddcd26fdb03f44ada54fbfc54ed221bf : SUCCESS in 1h 32m 18s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/9a7594c56981420d848c24e3d420f5f6 : SUCCESS in 1h 06m 31s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5a7818620d3d4d16929e7707eeb7efac : SUCCESS in 4m 44s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/a308d378ca024ab396712ee2481f4f28 : SUCCESS in 5m 43s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/b50b49822e534a30af2c4dd694e002f3 : SUCCESS in 3m 49s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/6b728566f6264d84af1913da36329d37 : SUCCESS in 9m 37s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/be2c538d96374e9bb4691a8f113764d0 : SUCCESS in 5m 37s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/af124182874e48e286c63dbce9dddc34 : SUCCESS in 8m 22s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/3d52208173b041c693170c3bab3dff29 : SUCCESS in 8m 31s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/0d543440e33448f7a49e7d68b2520167 : SUCCESS in 7m 51s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/38dd733a59a44c7e83cf87b134e8ca70 : SUCCESS in 2h 08m 45s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/13108ba64a304dd78a38517ea98fb308 : SUCCESS in 1h 25m 47s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/8b2ae38781b746c6b065e14917d11130 : SUCCESS in 2h 12m 31s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/d62e91fd631f463fb4443b2eb816fb2e : SUCCESS in 2h 09m 49s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/5ab047cde34240ac89ba0608d927a551 : SUCCESS in 32m 07s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/40617d137d8f4c688bc53fe53be0cf6e : SUCCESS in 2h 01m 49s\n- glance-grenade-centralized-cache https://zuul.opendev.org/t/openstack/build/c80870c73dc3437a9c8b9bf16aaf29b1 : FAILURE in 17m 27s (non-voting)\n- glance-s3-multistore https://zuul.opendev.org/t/openstack/build/d942af50358848e1b04e03e3585bee61 : FAILURE in 12m 48s (non-voting)\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/c453c6362d124d1786a6199cfcc50e15 : SUCCESS in 35m 47s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/5c2dbd42b0044b88a4c49f0516177e12 : SUCCESS in 1h 08m 44s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"f8f5e8aec5048e712e8fecc4632a1cc7a9c199dc","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-08 09:01:01.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"dba1894d2858d2d37f35b3920641f4f2888fea22","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-08 11:02:50.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/348dde39fa514ffa9f60ccdec5594edc\n\n- grenade https://zuul.opendev.org/t/openstack/build/5d062503e84745909f812041ae30027c : SUCCESS in 58m 36s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/48537e38432a42b3b542919a6d906727 : SUCCESS in 1h 39m 04s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/49f210d5e12c496fb25949da8cd9aa23 : SUCCESS in 1h 07m 33s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e546b3fa001e41a4b0921c96c1f9a50c : SUCCESS in 5m 15s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/235820e743cb493d9219f8b85440940b : SUCCESS in 5m 35s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/5c6ae7536c6e4eef80549876dbde5bb2 : SUCCESS in 6m 34s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/552d43fe6cda48018b2f2ecff031f07d : SUCCESS in 6m 27s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/9ee1c1f11f4142688412de0f3680023d : SUCCESS in 10m 23s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/2d27ccf3abc6466c95b5794296af31bc : FAILURE in 6m 29s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/1e9cc4569e75420da4bb616bbbd55967 : SUCCESS in 8m 07s\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/fd33321ec64e4ce981d467870541cbb4 : SUCCESS in 1h 36m 41s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/24126af73a114593a128fc1d451f7696 : SUCCESS in 1h 54m 32s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/fca3c5a518734a06ab65bad626787ffe : SUCCESS in 51m 14s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/aabb95280e7b4c79bb720e316f350941 : SUCCESS in 1h 54m 14s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/787fa8de7828468ca257545f9d7af38d : SUCCESS in 33m 33s","accounts_in_message":[],"_revision_number":1},{"id":"6a44eedad87e5d5b442944ce3e8682ac5eb0f1a5","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2026-07-08 11:17:57.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"d317923edc9e480fb9bbd2331e2405788ac87e2e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-08 13:10:19.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/073fa150e3164f3e9180729baacc1b7c\n\n- grenade https://zuul.opendev.org/t/openstack/build/ae7dbc6aa6d8476cb3eae609d8c772dc : SUCCESS in 1h 06m 01s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/0929cc8924794355883e17ce2b242626 : SUCCESS in 1h 45m 03s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/d7faca94f5dc4c839c3208e1c54c7596 : SUCCESS in 50m 14s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/33ebb85fbfc1491ba49e2f67fad509bd : SUCCESS in 5m 13s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/28d34dfe3a374dd4b18ac26ac31c3667 : SUCCESS in 5m 03s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/1b57756c6a3b4b1fbd1e85854923e465 : SUCCESS in 5m 32s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/c214a31c124a4ce08e5c0feb20194aa9 : SUCCESS in 8m 03s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6d76f511e3634b7ca269a51b2d26103c : SUCCESS in 3m 06s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/38edce544efa4b10b6a8165bc15683c0 : SUCCESS in 6m 26s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/0c31e51885594fdf90723d1cb7de7b43 : SUCCESS in 6m 47s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/e290b80ad5094f528502f04d61b04456 : SUCCESS in 7m 44s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/6b80a3e2213040d38be27d37cf950426 : SUCCESS in 1h 50m 37s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/aab866d6a41949bb8ebc25d8cb00a026 : SUCCESS in 1h 01m 16s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/4204bad7294248118d91c5fde571aa62 : SUCCESS in 1h 09m 43s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/0597ff3c40154cb0903e248ec9df98a0 : SUCCESS in 1h 15m 27s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/1a4dd5275f11480a9f36e7ffb65f9185 : SUCCESS in 32m 46s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/7f3ffe54d1f547739a96c34a55fae707 : SUCCESS in 1h 46m 19s\n- glance-grenade-centralized-cache https://zuul.opendev.org/t/openstack/build/0873626459624c6c89ec0f8065033322 : FAILURE in 11m 46s (non-voting)\n- glance-s3-multistore https://zuul.opendev.org/t/openstack/build/3cdc1cc34e5842b4a0b4653e992b40b7 : FAILURE in 12m 19s (non-voting)\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/eeb642c9bb6b4d84aaefea5c5990ecb2 : SUCCESS in 26m 28s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/02e334a0f5fa46a6970b46201e76a660 : SUCCESS in 54m 37s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"5bf59f448e7346b93e2a17d98fe90cc9e5c64b27","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-08 13:11:22.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"0e2b857719a336bd3c4812809059d5739f783f18","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-08 15:11:02.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7fcafa51c6674b0c9f4aacee2e623f33\n\n- grenade https://zuul.opendev.org/t/openstack/build/f35acc5b336e4488a2fe2a45b3c27db9 : SUCCESS in 1h 07m 38s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/cb0b8c6c173e49aea8debc86a0340435 : SUCCESS in 1h 35m 16s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/c7254befcbc94eeebf590ce3e253884a : SUCCESS in 1h 00m 15s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7af6f37d23bd4acebb6674dd19de137d : SUCCESS in 4m 47s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/646b55c459d140698cf1fc6fbff19b35 : SUCCESS in 5m 07s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/a97b63f3465d4d9d869ad591d307a272 : SUCCESS in 5m 47s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9f7ac181e69a4c2bb9d62441978ad0b9 : SUCCESS in 5m 44s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/236679ff6b2c4d419e93c32669e59f4e : SUCCESS in 7m 38s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/5965d20f68d24200a2110ffe5d05ad49 : SUCCESS in 6m 36s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/963dd692519248cd8073e7fc681294d1 : SUCCESS in 9m 52s\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/4ab8050381834e6a92d706eca6f93533 : SUCCESS in 1h 57m 53s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/9a9db786a1b64e57bcf270d28f6a4305 : SUCCESS in 1h 59m 15s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/3ee6e5a5f8b24df6b6224650303c1296 : SUCCESS in 1h 21m 29s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/7a19d52060aa40608e9f2110a75d6b14 : SUCCESS in 1h 47m 21s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/98b7abc2c69b4c2e8cffa777ee00c273 : SUCCESS in 35m 14s","accounts_in_message":[],"_revision_number":1},{"id":"499a00730aa300745c2187dbbe080e4838a68948","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-08 15:11:02.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"d93bcc675abde3656cb9996df6d517a3a5f7ecea","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-08 15:13:18.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5220807ea31e4bdc9df48b44f00f30bf\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2ded04cd183341bc9e8d3dc57ab46f10 : SUCCESS in 1m 01s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"cc13b93838ced0de63538ec2acb5a665be49bc6d","revisions":{"cc13b93838ced0de63538ec2acb5a665be49bc6d":{"kind":"REWORK","_number":1,"created":"2026-07-07 08:11:40.000000000","uploader":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"ref":"refs/changes/20/996220/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/20/996220/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/20/996220/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/20/996220/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/20/996220/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/20/996220/1"}}},"commit":{"parents":[{"commit":"7664a841166e972a3547e73550f3e774cd7541d9","subject":"Merge \"doc: Fix incorrect config section name in interoperable image import docs\" into stable/2025.2","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/7664a841166e972a3547e73550f3e774cd7541d9"}]}],"author":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2026-07-07 08:08:17.000000000","tz":0},"committer":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2026-07-07 08:08:17.000000000","tz":0},"subject":"Drop Jammy nodeset from enforce-scope-new-defaults job","message":"Drop Jammy nodeset from enforce-scope-new-defaults job\n\nJammy/Python 3.10 cannot satisfy upper/master oslo.config 10.5.0\nduring tempest venv install. Use the default Noble nodeset instead.\n\nChange-Id: Ic6de91db83c9dda1d5d735566876b90c1ac6b2e3\nSigned-off-by: Abhishek Kekane \u003cakekane@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/cc13b93838ced0de63538ec2acb5a665be49bc6d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/cc13b93838ced0de63538ec2acb5a665be49bc6d"}]},"branch":"refs/heads/stable/2025.2"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{"label:Review-Priority\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
