)]}'
{"id":"openstack%2Fglance~996357","triplet_id":"openstack%2Fglance~stable%2F2026.1~I30798ede5b0d3bbdaefc1d6e1e4962c59e1703e8","project":"openstack/glance","branch":"stable/2026.1","attention_set":{"8122":{"account":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"last_update":"2026-08-07 14:35:22.000000000","reason":"Vote got outdated and was removed: Code-Review-1"},"38762":{"account":{"_account_id":38762,"name":"Sahil Kumbhar","display_name":"sakumbha","email":"sakumbha@redhat.com","username":"sakumbha"},"last_update":"2026-07-09 16:31:36.000000000","reason":"\u003cGERRIT_ACCOUNT_39209\u003e replied on the change","reason_account":{"_account_id":39209,"name":"Anatoli Barzev","email":"anatoli.barzev@digits.schwarz","username":"abarzev"}}},"removed_from_attention_set":{"39209":{"account":{"_account_id":39209,"name":"Anatoli Barzev","email":"anatoli.barzev@digits.schwarz","username":"abarzev"},"last_update":"2026-08-10 10:18:23.000000000","reason":"\u003cGERRIT_ACCOUNT_39209\u003e replied on the change","reason_account":{"_account_id":39209,"name":"Anatoli Barzev","email":"anatoli.barzev@digits.schwarz","username":"abarzev"}}},"hashtags":[],"change_id":"I30798ede5b0d3bbdaefc1d6e1e4962c59e1703e8","subject":"Fix S3 URL parsing/constructing causing constant DB updates","status":"NEW","created":"2026-07-07 20:24:54.000000000","updated":"2026-08-10 12:36:40.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":true,"submittable":false,"total_comment_count":24,"unresolved_comment_count":2,"has_review_started":true,"meta_rev_id":"1035c1530f8528baf54b078d3c1d6ec22effc407","_number":996357,"virtual_id_number":996357,"owner":{"_account_id":39209,"name":"Anatoli Barzev","email":"anatoli.barzev@digits.schwarz","username":"abarzev"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"tag":"autogenerated:zuul:check","value":1,"date":"2026-08-10 12:36:40.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Top Priority / Holds Gate"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":38762,"name":"Sahil Kumbhar","display_name":"sakumbha","email":"sakumbha@redhat.com","username":"sakumbha"}],"REVIEWER":[{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-07-07 22:39:30.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-07-08 10:06:39.000000000","updated_by":{"_account_id":38762,"name":"Sahil Kumbhar","display_name":"sakumbha","email":"sakumbha@redhat.com","username":"sakumbha"},"reviewer":{"_account_id":38762,"name":"Sahil Kumbhar","display_name":"sakumbha","email":"sakumbha@redhat.com","username":"sakumbha"},"state":"CC"},{"updated":"2026-07-09 10:54:26.000000000","updated_by":{"_account_id":38762,"name":"Sahil Kumbhar","display_name":"sakumbha","email":"sakumbha@redhat.com","username":"sakumbha"},"reviewer":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"state":"REVIEWER"}],"messages":[{"id":"5f20ab3e35057cbeef2fc17a226581d5aff5fce8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":39209,"name":"Anatoli Barzev","email":"anatoli.barzev@digits.schwarz","username":"abarzev"},"date":"2026-07-07 20:24:54.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"0fe487c31d483837b44e31847d9e8ab3bd85f578","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-07 22:39:30.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f90f02b6c79d4b3b84798683f0170928\n\n- grenade https://zuul.opendev.org/t/openstack/build/094c6b42c2164232b38625aaf6ea9d24 : SUCCESS in 1h 03m 21s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/8c49be3007de44f1ba01b49d93d339bc : SUCCESS in 1h 23m 55s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/246f76f1c20f421ea44be417c2839faf : SUCCESS in 1h 42m 56s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/3115fa81ffef41aca27c512de78af725 : SUCCESS in 1h 04m 34s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6f073493ddc14fdea4f144856fbf933e : SUCCESS in 4m 43s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/911ae83a1c874b7a91d5e3422af5ca3b : SUCCESS in 4m 51s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/b837fd13412047409349270ec0b683f8 : SUCCESS in 6m 01s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/7ee317fa8438461095663d00829e9f28 : SUCCESS in 9m 47s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3cfe50c0792447a1ab07171559f93a26 : SUCCESS in 5m 56s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/5c72c71a374d466d940c9f4a8673b6cd : SUCCESS in 9m 59s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/05d8d8d857644adfad47245775ba7355 : SUCCESS in 6m 19s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/73d7622aff9a4020bb54d75158a0a34e : SUCCESS in 6m 44s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/28346f3bb50047a58fdcc2ed300d7ceb : SUCCESS in 1h 49m 18s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/eee2deeed0c049f4b35d496c99178f06 : SUCCESS in 2h 06m 19s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/4fcef6e1e2e546e48b33d347c4f5e4dd : SUCCESS in 1h 41m 07s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/7de7e9bab8bf45fd96d0d7e9f78a1e3d : SUCCESS in 2h 08m 41s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/0453260b4f14407ca524faec92e10437 : SUCCESS in 1h 02m 56s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/fcda35e1ec1142bfaae6d9c049e4170c : SUCCESS in 1h 40m 14s\n- glance-grenade-centralized-cache https://zuul.opendev.org/t/openstack/build/ea4fec50561f4cf896f79ad23e802250 : FAILURE in 16m 50s (non-voting)\n- glance-s3-multistore https://zuul.opendev.org/t/openstack/build/508d7e4dcc6047c7ad8d63caa1a96301 : SUCCESS in 2h 05m 35s (non-voting)\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/9651cde857ca4a25aa840d2e6131c8d4 : SUCCESS in 35m 14s","accounts_in_message":[],"_revision_number":1},{"id":"02b951d1d24706e33af0abb8a38431bec835fd92","author":{"_account_id":38762,"name":"Sahil Kumbhar","display_name":"sakumbha","email":"sakumbha@redhat.com","username":"sakumbha"},"date":"2026-07-08 10:06:39.000000000","message":"Patch Set 1:\n\n(4 comments)","accounts_in_message":[],"_revision_number":1},{"id":"6120a0dad9db141d5d0e11cf93295bcb909f3b65","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":39209,"name":"Anatoli Barzev","email":"anatoli.barzev@digits.schwarz","username":"abarzev"},"date":"2026-07-08 12:08:53.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":2},{"id":"79a913eb0c8ccc27978e873874111efcbf2637ff","author":{"_account_id":39209,"name":"Anatoli Barzev","email":"anatoli.barzev@digits.schwarz","username":"abarzev"},"date":"2026-07-08 12:10:53.000000000","message":"Patch Set 2:\n\n(5 comments)","accounts_in_message":[],"_revision_number":2},{"id":"14f6d3515c31c433396dc6ccaa782e73b4b46eaf","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-08 14:19:35.000000000","message":"Patch Set 2: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/463033f2f3364a50957e71a55e0a3ddf\n\n- grenade https://zuul.opendev.org/t/openstack/build/1485f88b619749278cb1a9c2dda5a5c4 : SUCCESS in 1h 02m 51s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/ba535aa767aa47f68acc3a3b4423aa09 : SUCCESS in 1h 08m 55s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/a889167306fc47bcb6abe50bb708db85 : SUCCESS in 1h 37m 30s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/44def2e13727487bbece5b2c9c9dd2ea : SUCCESS in 1h 04m 14s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5209b6095d134d478cf3d4b6cbc88b39 : FAILURE in 4m 54s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/b6b3b66dee8249d99e7902b104f1c202 : SUCCESS in 15m 56s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/af437dd9b58945e9af00288ce58a2c3c : SUCCESS in 4m 48s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/3c37494f079140d3a16a5d66abf119a7 : SUCCESS in 7m 27s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/01f3049134354f7cbf3f7c29b6991618 : SUCCESS in 6m 04s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/459823e9c7ca469380934d9438fdbb7a : FAILURE in 11m 10s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/88835ccf2b4d422dba0896290703bb00 : SUCCESS in 6m 32s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/73a5da857a25409681eba79339141209 : FAILURE in 14m 45s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/c83985a21c164d458d6985487e9080c0 : FAILURE in 48m 35s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/730fab40612f42ce9900b54ada21e849 : FAILURE in 48m 06s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/cc7f4972951a4b7ea1142c1e97f4f79a : SUCCESS in 2h 03m 13s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/0b1c77a119864af681dff7d23eb2e63b : FAILURE in 20m 53s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/491c84c054944012bb7db392df3071ad : SUCCESS in 56m 07s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/77f904265ab643f7b2e5ac550b8e1bb5 : FAILURE in 19m 48s\n- glance-grenade-centralized-cache https://zuul.opendev.org/t/openstack/build/39bc3be785174b25b4ddd94a09774d19 : FAILURE in 15m 28s (non-voting)\n- glance-s3-multistore https://zuul.opendev.org/t/openstack/build/59b0a4d17188494e97f55e42e3a623ff : SUCCESS in 1h 55m 00s (non-voting)\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/8e62722fb4194342b0ad24a8ee3c44af : SUCCESS in 35m 28s","accounts_in_message":[],"_revision_number":2},{"id":"8053501328e72ece2fb945d0b21b954c141f95d5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":39209,"name":"Anatoli Barzev","email":"anatoli.barzev@digits.schwarz","username":"abarzev"},"date":"2026-07-08 20:40:20.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":3},{"id":"f4d51267c7239cefceb35bb4b627144d97ae1b20","author":{"_account_id":39209,"name":"Anatoli Barzev","email":"anatoli.barzev@digits.schwarz","username":"abarzev"},"date":"2026-07-08 20:40:46.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"323704359b5898d2064c095f278380b83ab1e0d6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-08 22:51:10.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b0f63ef69b2344b1b569d6cd66979821\n\n- grenade https://zuul.opendev.org/t/openstack/build/93f42bebf80b4c919eaa098fef216259 : SUCCESS in 1h 04m 57s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/e75c620a97a6466bbac7ea02ca3dc513 : SUCCESS in 52m 50s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/76a9a4db694f4fe6a6c72d86e7c31c06 : SUCCESS in 1h 33m 36s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/90b4e272ec7d4ba8b98c926c5c73b3e0 : SUCCESS in 36m 40s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3583f64a0a0c400fb0a83b5a10c4da27 : SUCCESS in 4m 23s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/de31b18e70d24aa0aa489f70a68b3323 : SUCCESS in 4m 44s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/9982fa48bee548448d800f2e3c237b51 : SUCCESS in 5m 26s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/2a229ec27f304dbb91b2ff7857307124 : SUCCESS in 7m 50s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/92313b7972ae4b698a60304c30e0f684 : SUCCESS in 7m 38s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d9b92132e07f41f88540b12bcb7500da : SUCCESS in 7m 58s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/c9b6139d846647608f5708a50835101c : SUCCESS in 6m 44s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/3ed7ad1e001a4f448e871699c5fbc291 : SUCCESS in 4m 34s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/f925c94d0b354defba73c96f9347c562 : FAILURE in 33m 49s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/7960d6e87c90494b843acf87491c6761 : SUCCESS in 2h 05m 03s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/17b63fc786924fbf8190e86ea1952cff : SUCCESS in 2h 05m 01s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/a6e2d9aa09f74ddf8510964664ffb09d : SUCCESS in 1h 14m 42s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/deae0a6559a64a9d89cccde25eedb10a : SUCCESS in 58m 47s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/441bad25ecfa4bb7be0940335159af64 : SUCCESS in 1h 50m 35s\n- glance-grenade-centralized-cache https://zuul.opendev.org/t/openstack/build/55b4abd8b59e4dafaf4f03e627a7b037 : FAILURE in 7m 03s (non-voting)\n- glance-s3-multistore https://zuul.opendev.org/t/openstack/build/6b00fa614552485799229d4f6e8986b3 : SUCCESS in 1h 50m 39s (non-voting)\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/7c64204b94f240f0841d8ae0e6cfb4a3 : SUCCESS in 33m 19s","accounts_in_message":[],"_revision_number":3},{"id":"52553913e7def04926c9eec2c621f58217bc764c","author":{"_account_id":38762,"name":"Sahil Kumbhar","display_name":"sakumbha","email":"sakumbha@redhat.com","username":"sakumbha"},"date":"2026-07-09 10:53:44.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"fc229c582fca0192f783f9d4dda983423f2b8acd","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":39209,"name":"Anatoli Barzev","email":"anatoli.barzev@digits.schwarz","username":"abarzev"},"date":"2026-07-09 16:30:21.000000000","message":"Uploaded patch set 4.\n\nOutdated Votes:\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":4},{"id":"636c06c558b4deb09758095993c9e67ad73cbf43","author":{"_account_id":39209,"name":"Anatoli Barzev","email":"anatoli.barzev@digits.schwarz","username":"abarzev"},"date":"2026-07-09 16:31:36.000000000","message":"Patch Set 4:\n\n(2 comments)","accounts_in_message":[],"_revision_number":4},{"id":"da22efb813f4e2d2cf2f08b00b166e1d874776d7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-09 18:47:21.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ead48ae79e1e4b889fa7a92dab58e8ce\n\n- grenade https://zuul.opendev.org/t/openstack/build/a9f175e2f55e4159869f0e5133bfed1a : SUCCESS in 1h 05m 12s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/6bf00f36d34c4b4cb5bad03817a3af99 : SUCCESS in 1h 13m 09s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/feac3640846e4c40bd0c141963a94534 : SUCCESS in 1h 51m 29s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/a8c63271323940c3985d1385d9af1d1c : SUCCESS in 1h 07m 09s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/aac961242462450db615b9287d640439 : SUCCESS in 5m 17s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/c5f1dc6d2760498d854f08a8dbe727c3 : SUCCESS in 4m 36s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/517080bcfca0465484625ba6bebef5cc : SUCCESS in 5m 02s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/c5b3d7e5f6154ee48277e9bc9ba2169f : SUCCESS in 7m 25s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0367ff3d33f14dea81a101624b2d3535 : SUCCESS in 6m 02s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/1010ddf7fc954cdbbf87efcd2459f91b : SUCCESS in 7m 55s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/8b7b832b1c644045ba2cc4b588ecfb57 : SUCCESS in 6m 30s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/264cfc609e754d8e835cd5144a8b3076 : SUCCESS in 6m 27s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/e20a2a765bbf4e31b12dc559151d94f8 : SUCCESS in 1h 54m 38s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/6e160244bc2e415891e53ce143e27b6a : SUCCESS in 2h 04m 13s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/85d503772c774a67898c6e8814bd9e22 : SUCCESS in 1h 51m 13s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/54fd2449df344040bde4bb97a14fab87 : SUCCESS in 2h 10m 12s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/e8db3549ece24ab7aa03aa497593ba69 : SUCCESS in 58m 56s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/2abebc164f38473c838fd43c6a37600c : SUCCESS in 1h 16m 24s\n- glance-grenade-centralized-cache https://zuul.opendev.org/t/openstack/build/24ae551af1794d8fbee8ada1f8881f6e : FAILURE in 8m 22s (non-voting)\n- glance-s3-multistore https://zuul.opendev.org/t/openstack/build/2f93f1872b094f28ab41c9e9f595c811 : SUCCESS in 1h 51m 28s (non-voting)\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/afdccff4210d4a719504a967828a063f : SUCCESS in 34m 07s","accounts_in_message":[],"_revision_number":4},{"id":"031bb512d7c3d99bb0a811c58fb304df19c57abd","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":39209,"name":"Anatoli Barzev","email":"anatoli.barzev@digits.schwarz","username":"abarzev"},"date":"2026-07-09 20:53:15.000000000","message":"Uploaded patch set 5.\n\nOutdated Votes:\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":5},{"id":"d904109cef571adf820bdbb1fc5dbcb57d123fa5","author":{"_account_id":39209,"name":"Anatoli Barzev","email":"anatoli.barzev@digits.schwarz","username":"abarzev"},"date":"2026-07-09 20:53:27.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"621fb57c2e3d43c35ad2ffdde24d6fbc5508efe6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-09 23:23:27.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ab22d5aa7b6542d292458511529e5e37\n\n- grenade https://zuul.opendev.org/t/openstack/build/13d8a099144d4e63a7a9153ff9ab54c0 : SUCCESS in 1h 04m 06s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/0841539f268b4a07bd102b170f219058 : SUCCESS in 42m 48s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/ecb4159d0a3e4b398bcbe0d90feb0cb7 : SUCCESS in 1h 46m 29s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/4ab78d2c97714d668edeb0806b08913b : SUCCESS in 1h 03m 32s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/239ce8fc4bea4b5d80e6e18f31274262 : SUCCESS in 4m 58s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/d8034bb4dc8546348097ea8cd0d97515 : SUCCESS in 5m 10s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/b29e4b08b06045419bff3fdb0d309ac2 : SUCCESS in 4m 55s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/93ffc09a515249f2a89c5387741f9a81 : SUCCESS in 4m 45s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0f5092202cdf4dd7ba394bc42324bd3c : SUCCESS in 7m 04s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/4f0c95b1529c4790855dc4ceae0de4bb : SUCCESS in 5m 29s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/e2fb1a55c0b34e2bb6c2279e37bdd5c0 : SUCCESS in 6m 33s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/ee1cbfe70eda45fa9d9c4268069d2210 : SUCCESS in 6m 31s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/9c5f45fba28d4fda88f5a2df95c524d3 : SUCCESS in 1h 46m 29s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/be6854ac333248d99243d8fab6e46b48 : SUCCESS in 2h 13m 45s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/fca64aa7cc9f42a89d439167796dc403 : SUCCESS in 2h 03m 13s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/f4ecb9398aac4e558a4790a12b77de6f : SUCCESS in 1h 19m 43s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/011c4b6110b94b53becea1e673818150 : SUCCESS in 30m 27s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/6d14af5b27f044b6919e304303c1caa5 : SUCCESS in 1h 49m 13s\n- glance-grenade-centralized-cache https://zuul.opendev.org/t/openstack/build/83ed5d4ed0284d45a28d9cbc0e30d764 : FAILURE in 8m 07s (non-voting)\n- glance-s3-multistore https://zuul.opendev.org/t/openstack/build/0b20f833fe8f46b69b70ee6d1aadd824 : SUCCESS in 1h 29m 15s (non-voting)\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/471df93ab913473fafd06e3ee91a2f33 : SUCCESS in 35m 41s","accounts_in_message":[],"_revision_number":5},{"id":"248e168fd1b324fe64378ccf7f986475902d8c6e","author":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"date":"2026-08-03 15:19:46.000000000","message":"Patch Set 5: Code-Review-1\n\n(4 comments)","accounts_in_message":[],"_revision_number":5},{"id":"fa1ab062907fd03b16939ffdba81a6f79b963372","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":39209,"name":"Anatoli Barzev","email":"anatoli.barzev@digits.schwarz","username":"abarzev"},"date":"2026-08-07 14:35:22.000000000","message":"Uploaded patch set 6.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":6},{"id":"abbd3e9b393d2fae9920494f68cd7b87c3c26e6e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-07 17:04:04.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/60a24ebfb148473892a700f9dfadb900\n\n- grenade https://zuul.opendev.org/t/openstack/build/f71e0eef47774a6e8091012daf309455 : SUCCESS in 34m 44s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/eeb465b51afb4295ab8834dbf095d5b7 : SUCCESS in 1h 00m 46s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/d22be6ccd1b7428b9b879965fd900690 : SUCCESS in 1h 55m 36s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/420cc194e2a74434827eaac8d702ec2a : SUCCESS in 1h 08m 09s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/798d4fc3dbd84b479110c19de54f4444 : SUCCESS in 3m 59s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/79a33898c8af4692b6980c94f5640551 : SUCCESS in 5m 11s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/4e3d3d6c2b9f472eb35d1abbc7c3a77e : SUCCESS in 8m 40s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/c8e6ea17991a4645b95585c58a3a3d28 : SUCCESS in 4m 14s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2a19f0c5f06f49c99187322ae5fe2f32 : SUCCESS in 5m 52s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/36c5f60dd74c485783c8d8c9dde9e9db : SUCCESS in 7m 04s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/41ecb2ccdc5f4fba939b0252ed0c43c6 : SUCCESS in 6m 04s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/8dcc0c9209174978a2e875aaf132a25a : SUCCESS in 6m 13s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/a81b31e2a193416483c6cb819d938245 : SUCCESS in 1h 50m 49s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/e01a9606ed2940c59c4a68563b38a141 : SUCCESS in 1h 07m 25s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/7df587bc9f2c40c898f7203f148d291a : FAILURE in 1h 13m 11s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/247d13d043704d7c9e66fbbef42aea6f : SUCCESS in 2h 22m 52s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/82c66a75c64c4205bbab00c32e2c3521 : SUCCESS in 1h 04m 12s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/91535dd1f99e4d4ab68defaf3c7b1ba1 : SUCCESS in 1h 47m 10s\n- glance-grenade-centralized-cache https://zuul.opendev.org/t/openstack/build/e80890360a224398ba5dc323f298e54f : FAILURE in 7m 50s (non-voting)\n- glance-s3-multistore https://zuul.opendev.org/t/openstack/build/1d4b5e8fd138481ca0770c1c5ca3fe6b : SUCCESS in 2h 01m 21s (non-voting)\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/138cd7077c814a78ae700a44f3952637 : SUCCESS in 16m 34s","accounts_in_message":[],"_revision_number":6},{"id":"aaf2413abc7cdbb91b54dd21b8548ec46ced1a4f","author":{"_account_id":39209,"name":"Anatoli Barzev","email":"anatoli.barzev@digits.schwarz","username":"abarzev"},"date":"2026-08-07 17:51:55.000000000","message":"Patch Set 6:\n\n(3 comments)","accounts_in_message":[],"_revision_number":6},{"id":"1f5afb2b71512aafdfc3ca2778280580329f09b4","author":{"_account_id":39209,"name":"Anatoli Barzev","email":"anatoli.barzev@digits.schwarz","username":"abarzev"},"date":"2026-08-07 17:52:53.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"5e252fa54338a0168ec40f4c94300bfc80d8efbd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-07 20:34:48.000000000","message":"Patch Set 6:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/4f3ba01b72144d3f81646d809c57b426\n\n- grenade https://zuul.opendev.org/t/openstack/build/03790a594e544e53a929ea5118894949 : SUCCESS in 1h 04m 31s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/a81a3fb988b842128ad2adc763129322 : SUCCESS in 31m 59s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/4dfe9e79b1db414c8d97767d7fed8213 : SUCCESS in 1h 46m 34s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/fefcba20c0374aeea077a81982986e30 : SUCCESS in 1h 01m 29s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/169bce1e3b194545899600f449977f30 : SUCCESS in 4m 29s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/138b9bb25f1f40c494bd75fcf127505d : SUCCESS in 4m 26s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/d7f60390b1914febbe6e12bb1aa82159 : SUCCESS in 4m 51s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/43b3b8b3d42e40d9aa50e1ed272de9ef : SUCCESS in 4m 21s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8ed6ea34465048f89bab00b32ee8747e : SUCCESS in 5m 43s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/0294bdca5ebf4d3bb0cfd6c479a490bf : SUCCESS in 6m 57s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/dd5d450fa0614f86bb9c50412b07b5c9 : SUCCESS in 6m 43s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/f3b48f56dd80460886b28423a0134c7c : SUCCESS in 5m 13s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/912b353bc84c44c88ac075a2a085ff34 : SUCCESS in 1h 50m 01s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/b1b8bf9e398248bcb132a8013c8fa8b4 : SUCCESS in 2h 26m 01s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/23bcc4eca4a94a658a395c6a0482fcc2 : FAILURE in 1h 47m 50s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/d0fd8c5c912d45e896b971289e3c97d5 : SUCCESS in 1h 40m 53s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/dc6a1281f64642738797d7819851a4ae : SUCCESS in 59m 41s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/95c686c22f2849188a714fedeb46ced3 : SUCCESS in 58m 39s\n- glance-grenade-centralized-cache https://zuul.opendev.org/t/openstack/build/b8d39b606ba24d8292644c85ee506489 : FAILURE in 15m 25s (non-voting)\n- glance-s3-multistore https://zuul.opendev.org/t/openstack/build/a380be286bc14d44bc3fa6845b39447f : SUCCESS in 2h 06m 10s (non-voting)\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/533fa54b3aa4406c8f5f7dd9a68c1f01 : SUCCESS in 21m 15s","accounts_in_message":[],"_revision_number":6},{"id":"1a1e0c92ea1868145ae071664384a8ff204f0cd3","author":{"_account_id":39209,"name":"Anatoli Barzev","email":"anatoli.barzev@digits.schwarz","username":"abarzev"},"date":"2026-08-10 10:18:23.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"1035c1530f8528baf54b078d3c1d6ec22effc407","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-10 12:36:40.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/161175ac49af46469d34b8d8294b228b\n\n- grenade https://zuul.opendev.org/t/openstack/build/955d1433c02f42328129e93e6625c9c8 : SUCCESS in 1h 05m 51s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/029233a31fd14c62951e8bbac8491fc0 : SUCCESS in 33m 30s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/2fdf0acfaf0e4a0baee36f199555223b : SUCCESS in 1h 24m 57s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/aa26589085a64e8eab6f2c0128e7c109 : SUCCESS in 1h 05m 22s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/829e0ac03ac947eb88396c583393cd27 : SUCCESS in 6m 19s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/e075347ce9e347c2a5f278053df4a0eb : SUCCESS in 4m 38s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/6661d10c6872407aa9cfdef57a643dd1 : SUCCESS in 3m 58s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/b2c47d1433184af6ac062f098ad86441 : SUCCESS in 9m 48s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b2e530094dc3480b85d69febb5333e35 : SUCCESS in 5m 16s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/357c1727a1444c0fb664fb0a17124868 : SUCCESS in 8m 03s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/e84f388441154db0b87ba72e68bf87dd : SUCCESS in 5m 59s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/6c39f82c17ee478dbc16ec65fd796682 : SUCCESS in 5m 45s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/bebb3fdde0c141b4bbd50c68b7882981 : SUCCESS in 1h 05m 05s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/9e57df52b4134971af96fbe2579d181b : SUCCESS in 1h 30m 50s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/67eb4f41de73465c9e50396093f45ef4 : SUCCESS in 2h 07m 51s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/9e35c85f87a04db4b4ef7bca2497eb52 : SUCCESS in 2h 12m 55s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/fc50da3c054a4430be3c6d28a5c099ae : SUCCESS in 1h 09m 01s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/a8e6b91493834f219d87e48da5c4c91c : SUCCESS in 1h 45m 05s\n- glance-grenade-centralized-cache https://zuul.opendev.org/t/openstack/build/67dc57367ed244139648159cd5131594 : FAILURE in 15m 49s (non-voting)\n- glance-s3-multistore https://zuul.opendev.org/t/openstack/build/dd00b331011747859bf7b27d2d4833f2 : SUCCESS in 1h 47m 04s (non-voting)\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/0e3144140bec4f46a3f855847f2d1ffb : SUCCESS in 35m 43s","accounts_in_message":[],"_revision_number":6}],"current_revision_number":6,"current_revision":"ac0c589524e094901eadd29fec3d328a9807fb43","revisions":{"10e539cd868ad75421b8c64e3b57acf360c9ab2e":{"kind":"REWORK","_number":1,"created":"2026-07-07 20:24:54.000000000","uploader":{"_account_id":39209,"name":"Anatoli Barzev","email":"anatoli.barzev@digits.schwarz","username":"abarzev"},"ref":"refs/changes/57/996357/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/57/996357/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/57/996357/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/57/996357/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/57/996357/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/57/996357/1"}}},"commit":{"parents":[{"commit":"98cf6945b4eb0fbead2c2257346006c0594eba04","subject":"Fix store info API response samples","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/98cf6945b4eb0fbead2c2257346006c0594eba04"}]}],"author":{"name":"Anatoli Barzev","email":"anatoli.barzev@digits.schwarz","date":"2026-06-25 06:27:27.000000000","tz":180},"committer":{"name":"Anatoli Barzev","email":"anatoli.barzev@digits.schwarz","date":"2026-07-07 20:24:52.000000000","tz":180},"subject":"Fix and refactor S3 URL parsing/constructing causing constant DB updates","message":"Fix and refactor S3 URL parsing/constructing causing constant DB updates\n\nA previous change (review.opendev.org/c/openstack/glance/+/975113)\nintroduced incorrect S3 location parsing and false-positive\nDB record updates+corruptions. Subsequent patch\n(review.opendev.org/c/openstack/glance/+/970048) attempted\nto address this, but it was incomplete, and constant,\nredundant DB record updates are still occurring.\n\nFurthermore, the fix was not backported\nto versions 2024.2, 2025.1, and 2025.2.\nIf the patch gets accepted I will backport it\nfor the specified branches.\n\nWhat is causing constant db updates:\n\n1. If `s3_host` is configured with a scheme in glance.conf (e.g.,\n`s3_host \u003d https://domain`), `_construct_s3_url()` handles it\nincorrectly, resulting in false-positive URL diffs.\n\n2. `_update_location_for_legacy_store()` bypasses `_construct_s3_url()`,\nyet still triggers constant updates if the S3 secret contains a \u0027/\u0027.\n\nCloses-Bug: #2156925\nRelated-Bug: #2156432\nChange-Id: I30798ede5b0d3bbdaefc1d6e1e4962c59e1703e8\nSigned-off-by: Anatoli Barzev \u003canatoli.barzev@digits.schwarz\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/10e539cd868ad75421b8c64e3b57acf360c9ab2e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/10e539cd868ad75421b8c64e3b57acf360c9ab2e"}]},"branch":"refs/heads/stable/2026.1"},"8dece4eb1698e1fd69d25df78f4c4d1adabc2f33":{"kind":"REWORK","_number":2,"created":"2026-07-08 12:08:53.000000000","uploader":{"_account_id":39209,"name":"Anatoli Barzev","email":"anatoli.barzev@digits.schwarz","username":"abarzev"},"ref":"refs/changes/57/996357/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/57/996357/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/57/996357/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/57/996357/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/57/996357/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/57/996357/2"}}},"commit":{"parents":[{"commit":"98cf6945b4eb0fbead2c2257346006c0594eba04","subject":"Fix store info API response samples","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/98cf6945b4eb0fbead2c2257346006c0594eba04"}]}],"author":{"name":"Anatoli Barzev","email":"anatoli.barzev@digits.schwarz","date":"2026-06-25 06:27:27.000000000","tz":180},"committer":{"name":"Anatoli Barzev","email":"anatoli.barzev@digits.schwarz","date":"2026-07-08 12:08:33.000000000","tz":180},"subject":"Fix and refactor S3 URL parsing/constructing causing constant DB updates","message":"Fix and refactor S3 URL parsing/constructing causing constant DB updates\n\nA previous change (review.opendev.org/c/openstack/glance/+/975113)\nintroduced incorrect S3 location parsing and false-positive\nDB record updates+corruptions. Subsequent patch\n(review.opendev.org/c/openstack/glance/+/970048) attempted\nto address this, but it was incomplete, and constant,\nredundant DB record updates are still occurring.\n\nFurthermore, the fix was not backported\nto versions 2024.2, 2025.1, and 2025.2.\nIf the patch gets accepted I will backport it\nfor the specified branches.\n\nWhat is causing constant db updates:\n\n1. If `s3_host` is configured with a scheme in glance.conf (e.g.,\n`s3_host \u003d https://domain`), `_construct_s3_url()` handles it\nincorrectly, resulting in false-positive URL diffs.\n\n2. `_update_location_for_legacy_store()` bypasses `_construct_s3_url()`,\nyet still triggers constant updates if the S3 secret contains a \u0027/\u0027.\n\nCloses-Bug: #2156925\nRelated-Bug: #2156432\nChange-Id: I30798ede5b0d3bbdaefc1d6e1e4962c59e1703e8\nSigned-off-by: Anatoli Barzev \u003canatoli.barzev@digits.schwarz\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/8dece4eb1698e1fd69d25df78f4c4d1adabc2f33"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/8dece4eb1698e1fd69d25df78f4c4d1adabc2f33"}]},"branch":"refs/heads/stable/2026.1"},"464ec8cfe21d80db75da9f5ec0c229e006ce3693":{"kind":"REWORK","_number":3,"created":"2026-07-08 20:40:20.000000000","uploader":{"_account_id":39209,"name":"Anatoli Barzev","email":"anatoli.barzev@digits.schwarz","username":"abarzev"},"ref":"refs/changes/57/996357/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/57/996357/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/57/996357/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/57/996357/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/57/996357/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/57/996357/3"}}},"commit":{"parents":[{"commit":"98cf6945b4eb0fbead2c2257346006c0594eba04","subject":"Fix store info API response samples","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/98cf6945b4eb0fbead2c2257346006c0594eba04"}]}],"author":{"name":"Anatoli Barzev","email":"anatoli.barzev@digits.schwarz","date":"2026-06-25 06:27:27.000000000","tz":180},"committer":{"name":"Anatoli Barzev","email":"anatoli.barzev@digits.schwarz","date":"2026-07-08 20:40:09.000000000","tz":180},"subject":"Fix and refactor S3 URL parsing/constructing causing constant DB updates","message":"Fix and refactor S3 URL parsing/constructing causing constant DB updates\n\nA previous change (review.opendev.org/c/openstack/glance/+/975113)\nintroduced incorrect S3 location parsing and false-positive\nDB record updates+corruptions. Subsequent patch\n(review.opendev.org/c/openstack/glance/+/970048) attempted\nto address this, but it was incomplete, and constant,\nredundant DB record updates are still occurring.\n\nFurthermore, the fix was not backported\nto versions 2024.2, 2025.1, and 2025.2.\nIf the patch gets accepted I will backport it\nfor the specified branches.\n\nWhat is causing constant db updates:\n\n1. If `s3_host` is configured with a scheme in glance.conf (e.g.,\n`s3_host \u003d https://domain`), `_construct_s3_url()` handles it\nincorrectly, resulting in false-positive URL diffs.\n\n2. `_update_location_for_legacy_store()` bypasses `_construct_s3_url()`,\nyet still triggers constant updates if the S3 secret contains a \u0027/\u0027.\n\nCloses-Bug: #2156925\nRelated-Bug: #2156432\nChange-Id: I30798ede5b0d3bbdaefc1d6e1e4962c59e1703e8\nSigned-off-by: Anatoli Barzev \u003canatoli.barzev@digits.schwarz\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/464ec8cfe21d80db75da9f5ec0c229e006ce3693"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/464ec8cfe21d80db75da9f5ec0c229e006ce3693"}]},"branch":"refs/heads/stable/2026.1"},"bde6e0e2456bffa289c3692c30a01c27d83ada1a":{"kind":"REWORK","_number":4,"created":"2026-07-09 16:30:21.000000000","uploader":{"_account_id":39209,"name":"Anatoli Barzev","email":"anatoli.barzev@digits.schwarz","username":"abarzev"},"ref":"refs/changes/57/996357/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/57/996357/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/57/996357/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/57/996357/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/57/996357/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/57/996357/4"}}},"commit":{"parents":[{"commit":"98cf6945b4eb0fbead2c2257346006c0594eba04","subject":"Fix store info API response samples","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/98cf6945b4eb0fbead2c2257346006c0594eba04"}]}],"author":{"name":"Anatoli Barzev","email":"anatoli.barzev@digits.schwarz","date":"2026-06-25 06:27:27.000000000","tz":180},"committer":{"name":"Anatoli Barzev","email":"anatoli.barzev@digits.schwarz","date":"2026-07-09 16:30:10.000000000","tz":180},"subject":"Fix and refactor S3 URL parsing/constructing causing constant DB updates","message":"Fix and refactor S3 URL parsing/constructing causing constant DB updates\n\nA previous change (review.opendev.org/c/openstack/glance/+/975113)\nintroduced incorrect S3 location parsing and false-positive\nDB record updates+corruptions. Subsequent patch\n(review.opendev.org/c/openstack/glance/+/970048) attempted\nto address this, but it was incomplete, and constant,\nredundant DB record updates are still occurring.\n\nFurthermore, the fix was not backported\nto versions 2024.2, 2025.1, and 2025.2.\nIf the patch gets accepted I will backport it\nfor the specified branches.\n\nWhat is causing constant db updates:\n\n1. If `s3_host` is configured with a scheme in glance.conf (e.g.,\n`s3_host \u003d https://domain`), `_construct_s3_url()` handles it\nincorrectly, resulting in false-positive URL diffs.\n\n2. `_update_location_for_legacy_store()` bypasses `_construct_s3_url()`,\nyet still triggers constant updates if the S3 secret contains a \u0027/\u0027.\n\nCloses-Bug: #2156925\nRelated-Bug: #2156432\nChange-Id: I30798ede5b0d3bbdaefc1d6e1e4962c59e1703e8\nSigned-off-by: Anatoli Barzev \u003canatoli.barzev@digits.schwarz\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/bde6e0e2456bffa289c3692c30a01c27d83ada1a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/bde6e0e2456bffa289c3692c30a01c27d83ada1a"}]},"branch":"refs/heads/stable/2026.1"},"dbc37ad217720f6c1fe3ef70e4893f7955d5396c":{"kind":"REWORK","_number":5,"created":"2026-07-09 20:53:15.000000000","uploader":{"_account_id":39209,"name":"Anatoli Barzev","email":"anatoli.barzev@digits.schwarz","username":"abarzev"},"ref":"refs/changes/57/996357/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/57/996357/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/57/996357/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/57/996357/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/57/996357/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/57/996357/5"}}},"commit":{"parents":[{"commit":"98cf6945b4eb0fbead2c2257346006c0594eba04","subject":"Fix store info API response samples","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/98cf6945b4eb0fbead2c2257346006c0594eba04"}]}],"author":{"name":"Anatoli Barzev","email":"anatoli.barzev@digits.schwarz","date":"2026-06-25 06:27:27.000000000","tz":180},"committer":{"name":"Anatoli Barzev","email":"anatoli.barzev@digits.schwarz","date":"2026-07-09 20:53:08.000000000","tz":180},"subject":"Fix and refactor S3 URL parsing/constructing causing constant DB updates","message":"Fix and refactor S3 URL parsing/constructing causing constant DB updates\n\nA previous change (review.opendev.org/c/openstack/glance/+/975113)\nintroduced incorrect S3 location parsing and false-positive\nDB record updates+corruptions. Subsequent patch\n(review.opendev.org/c/openstack/glance/+/970048) attempted\nto address this, but it was incomplete, and constant,\nredundant DB record updates are still occurring.\n\nFurthermore, the fix was not backported\nto versions 2024.2, 2025.1, and 2025.2.\nIf the patch gets accepted I will backport it\nfor the specified branches.\n\nWhat is causing constant db updates:\n\n1. If `s3_host` is configured with a scheme in glance.conf (e.g.,\n`s3_host \u003d https://domain`), `_construct_s3_url()` handles it\nincorrectly, resulting in false-positive URL diffs.\n\n2. `_update_location_for_legacy_store()` bypasses `_construct_s3_url()`,\nyet still triggers constant updates if the S3 secret contains a \u0027/\u0027.\n\nCloses-Bug: #2156925\nRelated-Bug: #2156432\nChange-Id: I30798ede5b0d3bbdaefc1d6e1e4962c59e1703e8\nSigned-off-by: Anatoli Barzev \u003canatoli.barzev@digits.schwarz\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/dbc37ad217720f6c1fe3ef70e4893f7955d5396c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/dbc37ad217720f6c1fe3ef70e4893f7955d5396c"}]},"branch":"refs/heads/stable/2026.1"},"ac0c589524e094901eadd29fec3d328a9807fb43":{"kind":"REWORK","_number":6,"created":"2026-08-07 14:35:22.000000000","uploader":{"_account_id":39209,"name":"Anatoli Barzev","email":"anatoli.barzev@digits.schwarz","username":"abarzev"},"ref":"refs/changes/57/996357/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/57/996357/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/57/996357/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/57/996357/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/57/996357/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/57/996357/6"}}},"commit":{"parents":[{"commit":"98cf6945b4eb0fbead2c2257346006c0594eba04","subject":"Fix store info API response samples","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/98cf6945b4eb0fbead2c2257346006c0594eba04"}]}],"author":{"name":"Anatoli Barzev","email":"anatoli.barzev@digits.schwarz","date":"2026-06-25 06:27:27.000000000","tz":180},"committer":{"name":"Anatoli Barzev","email":"anatoli.barzev@digits.schwarz","date":"2026-08-07 14:35:13.000000000","tz":180},"subject":"Fix S3 URL parsing/constructing causing constant DB updates","message":"Fix S3 URL parsing/constructing causing constant DB updates\n\nA previous change (review.opendev.org/c/openstack/glance/+/975113)\nintroduced incorrect S3 location parsing and false-positive DB record\nupdates. Patch (review.opendev.org/c/openstack/glance/+/970048)\nattempted to address this, but it was incomplete, and constant,\nredundant DB record updates are still occurring.\n\nWhat is causing constant db updates:\n\n1. If `s3_host` is configured with a scheme in glance.conf (e.g.,\n`s3_host \u003d https://domain`), `_construct_s3_url()` handles it\nincorrectly, resulting in false-positive URL diffs.\n\n2. `_update_location_for_legacy_store()` bypasses `_construct_s3_url()`,\nyet still triggers constant updates if the S3 secret contains a \u0027/\u0027.\n\nCloses-Bug: #2156925\nRelated-Bug: #2156432\nChange-Id: I30798ede5b0d3bbdaefc1d6e1e4962c59e1703e8\nSigned-off-by: Anatoli Barzev \u003canatoli.barzev@digits.schwarz\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/ac0c589524e094901eadd29fec3d328a9807fb43"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/ac0c589524e094901eadd29fec3d328a9807fb43"}]},"branch":"refs/heads/stable/2026.1"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY"},{"label":"Workflow","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{"label:Review-Priority\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
