)]}'
{"id":"openstack%2Fglance~996681","triplet_id":"openstack%2Fglance~master~Ie3d9430d34f2bf357905e58509349ea7eede1305","project":"openstack/glance","branch":"master","topic":"bug/2160248","attention_set":{},"removed_from_attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2026-07-27 18:14:27.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ie3d9430d34f2bf357905e58509349ea7eede1305","subject":"Fix error caused by queue directory being deleted in the middle","status":"MERGED","created":"2026-07-09 16:46:23.000000000","updated":"2026-07-27 18:15:46.000000000","submitted":"2026-07-27 18:14:27.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"996681-bug/2160248","meta_rev_id":"82c75fa716d01449a59ebb7697ed38fd975d9603","_number":996681,"virtual_id_number":996681,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-07-27 18:14:27.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"all":[{"value":2,"date":"2026-07-10 07:35:53.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":2,"date":"2026-07-27 16:13:51.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"all":[{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":1,"date":"2026-07-27 16:13:51.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Top Priority / Holds Gate"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-07-09 19:23:58.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-07-10 07:35:53.000000000","updated_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"reviewer":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"state":"REVIEWER"},{"updated":"2026-07-27 16:13:51.000000000","updated_by":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"reviewer":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"state":"REVIEWER"}],"messages":[{"id":"e21c27502e719831eae7b38de83c1f77b4580197","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2026-07-09 16:46:23.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"f4d882c42ec74d8563ee041b183df95cdb09e105","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2026-07-09 16:47:13.000000000","message":"Uploaded patch set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"9444e96b6939a4a0e1b64907f6f538e6cd90e42e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-09 19:23:58.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7633e7c196dd4c978d1da48846fc4fac\n\n- grenade https://zuul.opendev.org/t/openstack/build/3eed09730fa04125995261e8df32f904 : SUCCESS in 1h 22m 27s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/160019bc7af8447e84ce2f46be8f8b5c : SUCCESS in 1h 06m 03s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/ef987d2f522d4d01a3b5a9bbbd7f74f6 : SUCCESS in 1h 24m 19s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/d13b43d9aa1345c3b8ab068017fb1b5e : SUCCESS in 1h 05m 49s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fb9fe06b48fa4a898642770a9231b671 : SUCCESS in 6m 15s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/7b918effa31b4f22abc2416299d07563 : SUCCESS in 5m 05s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/a2a573bba4c24cdc81eb2220f74ca7bf : SUCCESS in 4m 39s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/ea6860df53b34bfd981653ed27a897aa : SUCCESS in 9m 28s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ada4342cb7d543d0b5e297462d3abf6a : SUCCESS in 6m 06s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f8c0d32c034d48e28b12416ecf607ef8 : SUCCESS in 5m 26s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/4581abb3a3fe47e58a0668ba251623af : SUCCESS in 6m 29s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/6d624a50802c40c59121eb28d73e201a : SUCCESS in 5m 54s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/91c42e9616f346b2b4792f342db1b0a9 : SUCCESS in 1h 35m 40s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/2e5af793432549afb89f2453df73135d : SUCCESS in 1h 36m 28s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/3a36084995154ad2b444b534975b2a6c : SUCCESS in 2h 03m 20s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/2223b48c044e49338955c67ba4da3dd4 : SUCCESS in 2h 29m 25s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/8dae819b2326406596848bc58af03259 : SUCCESS in 55m 20s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/958b50a2f215458199810e2180826845 : SUCCESS in 1h 50m 11s\n- glance-s3-multistore https://zuul.opendev.org/t/openstack/build/ea3bfecf6b72486e9473ec46d756c84f : SUCCESS in 2h 11m 35s (non-voting)\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/cfe92704c2c543c6881a6bec5eaf66ae : SUCCESS in 33m 32s\n- glance-rally-task-uwsgi https://zuul.opendev.org/t/openstack/build/3c1342d8c41e430d87cf22546d9d6920 : SUCCESS in 51m 15s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"dcf0fc006dae2aeb0b357437f755c4b3cf63c566","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-09 21:19:36.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/98381bdc33984128af3898645b4b1b96\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/e1145cd9bfca45238a4a6c4e21b8fb2e : SUCCESS in 7m 42s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/bd332a98831d4466b205789daae61461 : SUCCESS in 8m 30s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/033af5e9789e41e28752d6b0ae34de44 : SUCCESS in 20m 33s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"6baf3f1217071cf960ed9f51bcb4f3de82f888c4","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2026-07-10 07:35:53.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"fbba20eda6971c6a1c072687fb769b7fa369d2a3","author":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"date":"2026-07-27 16:13:51.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"001edd0f24136cff20e34f0ac80cf3a96dc131f1","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-27 16:14:30.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"3d1a7caf3dcb56ce733b20a8fef40f2c4ea46f7d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-27 18:14:27.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/cb8662db37f84f6bab3f659c39f73675\n\n- grenade https://zuul.opendev.org/t/openstack/build/6df2a1ab89df4e32a74df8b3d6a5ae02 : SUCCESS in 50m 03s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/b11661bffa454d798636d533bf2a9cae : SUCCESS in 1h 01m 58s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/002af59b7cc44b77929dd2fdf46d458e : SUCCESS in 58m 45s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/e6b5280351d447d48e17f8061f282fa5 : SUCCESS in 40m 39s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4aa1e0b62126410b8c16cd56065c4abd : SUCCESS in 4m 45s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/6d5fbb2e0079445ebd643a59409f506b : SUCCESS in 5m 17s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/37a6d850d14d4a9b8b5d1cefd0511b3f : SUCCESS in 6m 44s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9bb60c6d99384d74b68af1cc9511b467 : SUCCESS in 7m 53s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d0ebe1a69464473a8f77701a96843d6a : SUCCESS in 8m 51s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/e281377e0c48475c8536290fddf34778 : SUCCESS in 6m 21s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/759e74abedf64cdc90f43a95f764c647 : SUCCESS in 6m 49s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/832b863162974d05afbfe5a997179138 : SUCCESS in 1h 47m 59s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/b1530a197f2b4888a4302d965fafe4bf : SUCCESS in 51m 44s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/07246b878538473291924e3b43dd1fa3 : SUCCESS in 1h 50m 17s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/daaa4d6e52354a6eace28992294aa0bb : SUCCESS in 35m 12s","accounts_in_message":[],"_revision_number":2},{"id":"1840f7c7cab165f4fcf97cefe84fa2ffd72aa925","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-27 18:14:27.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"82c75fa716d01449a59ebb7697ed38fd975d9603","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-27 18:15:46.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6f91cc973e194c7aa4ba2fe86d31f1e7\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f4fa9495c4f9465188b1dfcdd7e6166c : SUCCESS in 1m 07s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"cb52de2017d72c0e1a5579b87caf8f0dffbfe887","revisions":{"f208477756deebb7224bb705ccdbfda3013d96f8":{"kind":"REWORK","_number":1,"created":"2026-07-09 16:46:23.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/81/996681/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/81/996681/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/81/996681/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/81/996681/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/81/996681/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/81/996681/1"}}},"commit":{"parents":[{"commit":"e20488b1a6cfd59573130f56443dd4075f157989","subject":"Merge \"Validate enabled_import_methods\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/e20488b1a6cfd59573130f56443dd4075f157989"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-07-09 16:38:02.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-07-09 16:45:04.000000000","tz":540},"subject":"Fix error caused by queue directory being deleted in the middle","message":"Fix error caused by queue directory being deleted in the middle\n\nCloses-Bug: #2160248\nChange-Id: Ie3d9430d34f2bf357905e58509349ea7eede1305\nSigned-off-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/f208477756deebb7224bb705ccdbfda3013d96f8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/f208477756deebb7224bb705ccdbfda3013d96f8"}]},"branch":"refs/heads/master"},"cb52de2017d72c0e1a5579b87caf8f0dffbfe887":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2026-07-09 16:47:13.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/81/996681/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/81/996681/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/81/996681/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/81/996681/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/81/996681/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/81/996681/2"}}},"commit":{"parents":[{"commit":"e20488b1a6cfd59573130f56443dd4075f157989","subject":"Merge \"Validate enabled_import_methods\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/e20488b1a6cfd59573130f56443dd4075f157989"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-07-09 16:38:02.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-07-09 16:46:26.000000000","tz":540},"subject":"Fix error caused by queue directory being deleted in the middle","message":"Fix error caused by queue directory being deleted in the middle\n\nThe queue directories might be deleted right after it appears in\ndirectory listing. Make sure such deleted directories are ignored.\n\nCloses-Bug: #2160248\nChange-Id: Ie3d9430d34f2bf357905e58509349ea7eede1305\nSigned-off-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/cb52de2017d72c0e1a5579b87caf8f0dffbfe887"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/cb52de2017d72c0e1a5579b87caf8f0dffbfe887"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{"label:Review-Priority\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
