)]}'
{"id":"openstack%2Fglance~998668","triplet_id":"openstack%2Fglance~master~I5cb310ac5f0a2dc6df026c8d516a8c2586b2bfa2","project":"openstack/glance","branch":"master","topic":"bug/2161751","attention_set":{"9303":{"account":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"last_update":"2026-08-19 20:25:42.000000000","reason":"\u003cGERRIT_ACCOUNT_8878\u003e replied on the change","reason_account":{"_account_id":8878,"name":"Masahito Muroi","email":"masahito.muroi@g.softbank.co.jp","username":"masa"}},"8878":{"account":{"_account_id":8878,"name":"Masahito Muroi","email":"masahito.muroi@g.softbank.co.jp","username":"masa"},"last_update":"2026-08-28 01:50:23.000000000","reason":"Someone else replied on the change"}},"removed_from_attention_set":{},"hashtags":[],"change_id":"I5cb310ac5f0a2dc6df026c8d516a8c2586b2bfa2","subject":"Validate image update request schema check per schema item","status":"NEW","created":"2026-07-24 18:20:58.000000000","updated":"2026-08-28 04:20:32.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":true,"submittable":false,"total_comment_count":4,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"1e2094d52d58553946d70f20ff75853b8b5a3283","_number":998668,"virtual_id_number":998668,"owner":{"_account_id":8878,"name":"Masahito Muroi","email":"masahito.muroi@g.softbank.co.jp","username":"masa"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},{"tag":"autogenerated:zuul:check","value":1,"date":"2026-08-28 03:27:14.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"recommended":{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},"all":[{"value":1,"date":"2026-08-27 06:10:37.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Top Priority / Holds Gate"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"}],"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-07-24 18:36:26.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2026-07-24 20:40:30.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-08-19 13:56:40.000000000","updated_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"reviewer":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"state":"CC"},{"updated":"2026-08-27 06:10:37.000000000","updated_by":{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},"reviewer":{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},"state":"REVIEWER"}],"messages":[{"id":"39f24b1bd0d7a47482674f5ed1345ff2660d630b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8878,"name":"Masahito Muroi","email":"masahito.muroi@g.softbank.co.jp","username":"masa"},"date":"2026-07-24 18:20:58.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"69e5d754256176f448016c5458f40ded06bf1fe2","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":8878,"name":"Masahito Muroi","email":"masahito.muroi@g.softbank.co.jp","username":"masa"},"date":"2026-07-24 18:21:44.000000000","message":"Topic set to bug/2161751","accounts_in_message":[],"_revision_number":1},{"id":"f701145d9dedc7006793bf79796897adb30f1c11","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8878,"name":"Masahito Muroi","email":"masahito.muroi@g.softbank.co.jp","username":"masa"},"date":"2026-07-24 18:22:39.000000000","message":"Uploaded patch set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"1f5cb9b220601821bebf652bcb99776e19740147","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-24 18:36:26.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9eb0542a19d746629e16b9c67527ae29\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/a1271d9d463f4fa0835820e6823a6679 : SUCCESS in 8m 25s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/548b538aefcc4417a1867802e90bc248 : SUCCESS in 8m 27s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/6613f7738f6b40139735f3c844da3c71 : SUCCESS in 13m 29s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"4938c57eb9e128124b5380c83ea20cadc25bd9e0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-24 20:40:30.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9dda0b243dca4cb6ab2eb2e2695cdf05\n\n- grenade https://zuul.opendev.org/t/openstack/build/dcacafbe7ff9496b99995ded119c57c2 : SUCCESS in 1h 02m 25s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/9e3a86cfd32c41bd845f33d76d425e77 : SUCCESS in 1h 01m 38s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/d249c1e4865742c7ac3cf67db1766eb5 : SUCCESS in 1h 26m 36s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/17585e9ee94d47ec894e0ef54618a255 : SUCCESS in 1h 00m 31s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ff39addaacf14c59a17c2ce3adbb32fb : SUCCESS in 5m 22s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/d2485cc28efb445a97b472815cea83d8 : SUCCESS in 4m 19s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/cea730522de4401ea1196f387ebc3497 : SUCCESS in 6m 37s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/f13b5100941d4b2ca8e6e49e79ede070 : SUCCESS in 7m 17s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3fbba61180d74fdc90b4c73effedd1a0 : SUCCESS in 5m 23s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/bc26de3c8d7e4552a55d1b8f003581e8 : SUCCESS in 7m 23s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/2f6740f54af44399b7708c4adda1d68f : SUCCESS in 6m 51s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/d41e5e0314f149c7a262c4f151d626ab : SUCCESS in 6m 24s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/3399eef0938c4136bfae464a24e43155 : SUCCESS in 1h 18m 29s (non-voting)\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/dc70f560b4ab4d3183577bb62b1dc9cb : SUCCESS in 2h 12m 43s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/123ba964fe3e4a49a29816d48e0eb380 : SUCCESS in 1h 45m 28s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/10fadc84f54146639679d939b62ab5b6 : SUCCESS in 56m 59s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/56a827143dd24bb1b026a6779d2bb6f6 : SUCCESS in 1h 28m 43s\n- glance-s3-multistore https://zuul.opendev.org/t/openstack/build/bbbe36bd6dd547e19654fffef6a96edd : SUCCESS in 1h 41m 18s (non-voting)\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/bef8f3cf182a4f94bf83f6cb5464fe57 : SUCCESS in 34m 18s\n- glance-rally-task-uwsgi https://zuul.opendev.org/t/openstack/build/e7e74ffc35fa46cea126a28259b336f4 : SUCCESS in 59m 27s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"961d8b46421ef94139f1d2d9318c36d15ac9ad91","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2026-08-19 13:56:40.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"2b852c598f78010ad515ef9d4eee28d45ee51b4a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8878,"name":"Masahito Muroi","email":"masahito.muroi@g.softbank.co.jp","username":"masa"},"date":"2026-08-19 20:25:24.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":3},{"id":"1d1f3a21cca9718f9ba31c66ee0a562d314044df","author":{"_account_id":8878,"name":"Masahito Muroi","email":"masahito.muroi@g.softbank.co.jp","username":"masa"},"date":"2026-08-19 20:25:42.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"0be51327d1af672f7a4594e900a597bb2043257f","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-19 20:38:15.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6de26043b1954c9d9d219c11f38f0cdb\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/4baba1def99e41adae6b562557a13e2c : SUCCESS in 5m 58s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/855685bcfaf64545ae196b40208ed220 : SUCCESS in 6m 14s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/f16575cc6c234262ba1ba2496921bece : SUCCESS in 10m 53s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"ddf42285cabdc8a3b04360f20f2af1b5ed55fa8b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-19 22:55:39.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/878121a7794240feb528511203667219\n\n- grenade https://zuul.opendev.org/t/openstack/build/9e96435d87f04e9a83804cd81adc1990 : SUCCESS in 1h 04m 13s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/aec335d4d3374bb0b4c05db2cecd60d0 : SUCCESS in 1h 04m 21s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/2743e988196746928a71d270c2c33467 : SUCCESS in 55m 51s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/dc0bb62d67aa4570b3938ce0155af25a : SUCCESS in 42m 02s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6b0a18d8f37c48c7be9fce9976323ec4 : SUCCESS in 4m 37s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/b4e849c126a5426294defdbdb3975871 : SUCCESS in 4m 32s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/43f00f8dbbee43d6b007d92772ebcee9 : SUCCESS in 5m 02s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/077d0bc436864589902d0ef7c88f9769 : SUCCESS in 6m 50s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f76a825bcb2244e6b0e1c7224bd0f2a8 : SUCCESS in 6m 35s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/fa991f6b2dfc492883348fba713f83be : SUCCESS in 3m 38s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/d83b13e435414cc2a568fe5e9b8be1e0 : SUCCESS in 6m 27s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/a17387c54e5b428c8dd7685080a5f749 : SUCCESS in 4m 28s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/d3da61881baa451cb9701dcf69de8d22 : FAILURE in 14m 59s (non-voting)\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/5a9eeb7a5c57492aa9c89ef45f94188c : FAILURE in 1h 42m 22s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/882018e62b5644d8a26a999bdd5e2e42 : SUCCESS in 2h 24m 01s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/a80acd738bfa43278a9bede7700b98ef : SUCCESS in 29m 34s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/b1da396c3bdf491e94581940575ccf88 : FAILURE in 50m 56s\n- glance-s3-multistore https://zuul.opendev.org/t/openstack/build/69e8c7aabd7f4f549e6022ce4941ffed : SUCCESS in 1h 21m 47s (non-voting)\n- glance-multistore-add-location-do-secure-hash-false https://zuul.opendev.org/t/openstack/build/147a7c4f51634e89934abe09a568b40c : SUCCESS in 1h 42m 53s (non-voting)\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/91558a9fb40a4020b252ae0906f0bd11 : SUCCESS in 24m 54s\n- glance-rally-task-uwsgi https://zuul.opendev.org/t/openstack/build/688158f4331d47e8900e237c7b0cdbc3 : SUCCESS in 51m 09s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"6b0c966718c5e90e0df2b8f8b5fe4c85009d3985","author":{"_account_id":8878,"name":"Masahito Muroi","email":"masahito.muroi@g.softbank.co.jp","username":"masa"},"date":"2026-08-20 07:02:07.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"f1726a897c85910696eaea249daf3477cd78b628","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-20 07:33:01.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9b4a434747a14e59b6c22ccc3adf3481\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/9ed10fc674bb439089fd63e1b97adcbf : SUCCESS in 7m 37s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/c0de4a69e1d44f3794f01c18b93bd51b : SUCCESS in 6m 05s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/09b429d4e9604670876e5f10dc5ad1c1 : SUCCESS in 12m 35s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"3f79bf28258e91462a78d6cdb00fe27cc06584e1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-20 09:05:43.000000000","message":"Patch Set 3:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/b49d48daaef042d092959a1f46bdffb0\n\n- grenade https://zuul.opendev.org/t/openstack/build/de388b4430da46b9b83a779b7a09528f : SUCCESS in 48m 29s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/44ba9df7186c47bc877977a8f0273e7c : SUCCESS in 1h 03m 18s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/018ceb8029064e71803831149526fbe2 : SUCCESS in 1h 27m 46s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/3016adec5e1242d3a9c7335dfa400b3b : SUCCESS in 1h 00m 56s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6d08c3dfad0546038e783f9a581bf731 : SUCCESS in 2m 56s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/4c5587167e78405f8871340f5e9d711c : SUCCESS in 5m 07s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/27e913d12e254a899f722ac45da354c8 : SUCCESS in 4m 33s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/850fe7bc15b34ced86b372ceaf4151ff : SUCCESS in 6m 20s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a5f21786ea0b45608bf885a0813aff6a : SUCCESS in 5m 09s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b629e66340d747a3a27a610c828d7db9 : SUCCESS in 8m 49s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/4b46d8ca04b44bf39c92eebe59da2131 : SUCCESS in 5m 58s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/2fdbc992877a4f9cb8368aba6172be90 : SUCCESS in 6m 15s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/5dedcb8f1bbc4bd9b0cf84a62a64ddb6 : FAILURE in 33m 42s (non-voting)\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/2b420bd25b274590a588aff238deae3f : FAILURE in 1h 35m 36s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/654c83d46a694781b29b71ba88affbd3 : SUCCESS in 1h 57m 41s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/99b083fef3984e429ef08ef25e783c28 : SUCCESS in 28m 38s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/8a49588c9cc14632a63a8bb698dce46e : FAILURE in 48m 27s\n- glance-s3-multistore https://zuul.opendev.org/t/openstack/build/11e8293674974838a7f4aaf76139e5b3 : FAILURE in 1h 21m 34s (non-voting)\n- glance-multistore-add-location-do-secure-hash-false https://zuul.opendev.org/t/openstack/build/6e22fcce77ae4fe9a04cef610da09fa8 : SUCCESS in 1h 10m 45s (non-voting)\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/21b633f6d07c40468ab87b6236dd936c : SUCCESS in 23m 53s\n- glance-rally-task-uwsgi https://zuul.opendev.org/t/openstack/build/fbe29e60c72849619928a4a7b634157b : SUCCESS in 52m 21s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"472a515b8891bc7d492feecd3891433111a8c905","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8878,"name":"Masahito Muroi","email":"masahito.muroi@g.softbank.co.jp","username":"masa"},"date":"2026-08-27 06:04:35.000000000","message":"Uploaded patch set 4: Patch Set 3 was rebased.\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":4},{"id":"f51e399784a0d5574b39170aff856e2ac5565e89","author":{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},"date":"2026-08-27 06:10:37.000000000","message":"Patch Set 4: Code-Review+1\n\nBuild succeeded (check pipeline).\nhttps://gateway-cloud-softwarefactory.apps.ocp.cloud.ci.centos.org/zuul/t/rdoproject.org/buildset/c82eddd91b2b4131bb822cc311a48bc1\n\n- s2i-openstack-container-content-provider https://gateway-cloud-softwarefactory.apps.ocp.cloud.ci.centos.org/zuul/t/rdoproject.org/build/366b9668ae6040188e29fc4ed0a8a739 : SUCCESS in 5m 10s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"8dd03a1caa66769cd224a2ddb7dc07e5ad601de2","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-27 06:24:28.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a67c6e7fe20543d49ea03657918db632\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/8ac33b687f47448aa5f72993f7a15f54 : SUCCESS in 7m 20s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/b257f8697ac44b49b5ed758b6a3f90ba : SUCCESS in 8m 21s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/9d530400bdfe479ab70fa6baf8732a45 : SUCCESS in 15m 04s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"d16280c1b8c060c49fd18e2c3ddf599551f1438e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-27 08:12:21.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/4a488dc4cc7848aeb7f66a59f65a4a1f\n\n- grenade https://zuul.opendev.org/t/openstack/build/5fbc6ecdc9354cc9b7257a6f6227bfcb : SUCCESS in 35m 24s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/b62427a6b7aa4265ae29d5202515fcc1 : SUCCESS in 33m 26s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/213135c0e01a47d1b2d1053df86d4aff : SUCCESS in 54m 04s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/0180a3a0b7294ada8b65726b49446b0e : SUCCESS in 31m 35s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8c7c9b25d20f48b9a9ebf2aed9a21d39 : SUCCESS in 3m 06s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/936b90c2f7654aebadaaa47847e88257 : SUCCESS in 5m 04s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/9f3f390bd5f6453d91ba459f814e3f14 : SUCCESS in 3m 17s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/2b6a0c3c3f054e52a31c911b2ca48382 : SUCCESS in 7m 01s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/212febaceadd471fb5265d1a37a8aee9 : SUCCESS in 5m 38s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/0657df429b4d4a5caa04db516173fd2f : SUCCESS in 8m 31s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/9a1400b0d47e4bb2b47a843e8551dc01 : SUCCESS in 4m 38s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/3c822890b2f6486b837d6541a669d032 : SUCCESS in 6m 57s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/81fa7bdd3b874c729a0fb1ded01a1b6d : SUCCESS in 1h 45m 41s (non-voting)\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/b32d884bcbbf4102b27cab90f842013e : FAILURE in 1h 47m 52s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/11fdfbc092da4971bf56bcc5ef303f3b : SUCCESS in 1h 58m 40s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/4783b48ded464b9581889d2e79cf10a8 : SUCCESS in 37m 27s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/828ffc1aabec4da3acf4445b358fe78f : SUCCESS in 1h 27m 48s\n- glance-s3-multistore https://zuul.opendev.org/t/openstack/build/7db261bd301741daa7a969922b58c60e : FAILURE in 31m 09s (non-voting)\n- glance-multistore-add-location-do-secure-hash-false https://zuul.opendev.org/t/openstack/build/12bef8625ecb4cebaa9d1b01a355ce16 : FAILURE in 27m 46s (non-voting)\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/d47aae9d725c43c08792833592429bac : SUCCESS in 34m 55s\n- glance-rally-task-uwsgi https://zuul.opendev.org/t/openstack/build/a12e89e9bb5f4e5cbb540d6782393199 : SUCCESS in 26m 11s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"29eda5fa14f89dec1ce3feec3495545a5269baa2","author":{"_account_id":8878,"name":"Masahito Muroi","email":"masahito.muroi@g.softbank.co.jp","username":"masa"},"date":"2026-08-28 01:44:02.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"56378ada6954bc1defb9c32de266ecff88497a96","author":{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},"date":"2026-08-28 01:50:23.000000000","message":"Patch Set 4:\n\nBuild succeeded (check pipeline).\nhttps://gateway-cloud-softwarefactory.apps.ocp.cloud.ci.centos.org/zuul/t/rdoproject.org/buildset/9cb725b157ff4e1f87c44cfb7cee3d0a\n\n- s2i-openstack-container-content-provider https://gateway-cloud-softwarefactory.apps.ocp.cloud.ci.centos.org/zuul/t/rdoproject.org/build/43ea0d6892a7488c847f5215ec5be708 : SUCCESS in 5m 28s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"c4f76b9e9d68f041e546079decf21ef89b72ff05","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-28 03:27:14.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/aef613e200c74efdb2e4a37d51e0ac47\n\n- grenade https://zuul.opendev.org/t/openstack/build/9359db7f7d344c8cb3e0148f89014f31 : SUCCESS in 1h 22m 35s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/b263266d54774f739efeb4656f3fb93d : SUCCESS in 1h 01m 23s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/f5cf866962c249c48cc1811db499cf89 : SUCCESS in 54m 35s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/0ea10629b7ef4053acfe0eae5657333f : SUCCESS in 27m 44s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b93af088d35d4b26a54b9fa11de64c59 : SUCCESS in 2m 37s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/31dc8f790fb04abfbfeb0201cbb91154 : SUCCESS in 2m 29s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/9b08ed829d99499e8fdd789865ad30f8 : SUCCESS in 4m 23s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/c4747cea85974a698d1d41f2b42c033b : SUCCESS in 4m 06s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4ba270fb756443caa090687cb88cda58 : SUCCESS in 3m 09s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c509a54ee2a64fcf9cf72b598dd28cc0 : SUCCESS in 3m 28s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/878c8e1d22344a00aa24b6fb95a19939 : SUCCESS in 5m 43s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/85008a2cebcf405494ce525a8469653c : SUCCESS in 5m 56s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/8d2d1921d49448e8bca2fd34bcdad610 : SUCCESS in 46m 40s (non-voting)\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/3088efb232b347f0b2e5f21bdda9ac0a : SUCCESS in 1h 14m 54s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/c01e14e83fb2451d8ca2b9215c422754 : SUCCESS in 53m 44s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/7d1e4e654a3041278b47e3f225874b1f : SUCCESS in 24m 44s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/8a602af3145941568005d6d3b913f430 : SUCCESS in 1h 41m 01s\n- glance-s3-multistore https://zuul.opendev.org/t/openstack/build/6af95c7b10f94b418fb4e2536ba63efa : FAILURE in 26m 48s (non-voting)\n- glance-multistore-add-location-do-secure-hash-false https://zuul.opendev.org/t/openstack/build/0f8494670c08485f9590f9f3db83f28c : FAILURE in 26m 10s (non-voting)\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/b7b391cfd9fd45d9b4fa393e5c8ba3d9 : SUCCESS in 15m 27s\n- glance-rally-task-uwsgi https://zuul.opendev.org/t/openstack/build/2e5dad9f27de493fa82a90fdeb96e043 : SUCCESS in 48m 39s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"1e2094d52d58553946d70f20ff75853b8b5a3283","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-28 04:20:32.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/38f8c2b55c9e40e8896179ebfd035030\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/2c43195eb7cb42f292557f7989772796 : SUCCESS in 8m 16s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/8af9303a19e747f6b29c6a8215714d37 : SUCCESS in 8m 56s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/4fe843a72f43415ebbab8cebf8eff05a : SUCCESS in 8m 55s (non-voting)","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"ca46fd85a822a5c1cb1fd8ac20022f7bcef8bcf6","revisions":{"0f82d50e6eeca24334246571673af8f6ff905050":{"kind":"REWORK","_number":1,"created":"2026-07-24 18:20:58.000000000","uploader":{"_account_id":8878,"name":"Masahito Muroi","email":"masahito.muroi@g.softbank.co.jp","username":"masa"},"ref":"refs/changes/68/998668/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/68/998668/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/68/998668/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/68/998668/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/68/998668/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/68/998668/1"}}},"commit":{"parents":[{"commit":"c76abb08e2bc82582389059dcbbc234293ce13b9","subject":"Merge \"Remove disabling the scope enforcement\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/c76abb08e2bc82582389059dcbbc234293ce13b9"}]}],"author":{"name":"Masahito Muroi","email":"masahito.muroi@g.softbank.co.jp","date":"2026-07-24 18:11:28.000000000","tz":540},"committer":{"name":"Masahito Muroi","email":"masahito.muroi@g.softbank.co.jp","date":"2026-07-24 18:20:17.000000000","tz":540},"subject":"Validate image update request schema check per schema item","message":"Validate image update request schema check per schema item\n\nThe Enhance Schema loading patch[1] adds \"required\" parameter support\nin the image resource schema. The required parameter enables glance-api\nto check required parameter is added at a new image creation. The enhancement\nworks well at the image create API. However, once the \"required\" parameter\nis activated at the image schema, Image Update API starts to fail\nbecause of missing required parameter in the Update Image API request body.\nAnd all the update image API return 400 Bad Request.\n\nThis commit adds the validate_property method in the Schema class\nfor using per property definition to validate update input, and updates\nthe _validate_change method in the RequestDeserializer class to\nvalidate input value per property definition.\n\n1. I6f4bdb78809511931b160c88e9173432fd2e6d06\n\nCloses-Bugs: #2161751\nChange-Id: I5cb310ac5f0a2dc6df026c8d516a8c2586b2bfa2\nSigned-off-by: Masahito Muroi \u003cmasahito.muroi@g.softbank.co.jp\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/0f82d50e6eeca24334246571673af8f6ff905050"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/0f82d50e6eeca24334246571673af8f6ff905050"}]},"branch":"refs/heads/master"},"818b741365ea4e47f8c6aeedb048026b28031443":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2026-07-24 18:22:39.000000000","uploader":{"_account_id":8878,"name":"Masahito Muroi","email":"masahito.muroi@g.softbank.co.jp","username":"masa"},"ref":"refs/changes/68/998668/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/68/998668/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/68/998668/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/68/998668/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/68/998668/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/68/998668/2"}}},"commit":{"parents":[{"commit":"c76abb08e2bc82582389059dcbbc234293ce13b9","subject":"Merge \"Remove disabling the scope enforcement\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/c76abb08e2bc82582389059dcbbc234293ce13b9"}]}],"author":{"name":"Masahito Muroi","email":"masahito.muroi@g.softbank.co.jp","date":"2026-07-24 18:11:28.000000000","tz":540},"committer":{"name":"Masahito Muroi","email":"masahito.muroi@g.softbank.co.jp","date":"2026-07-24 18:22:19.000000000","tz":540},"subject":"Validate image update request schema check per schema item","message":"Validate image update request schema check per schema item\n\nThe Enhance Schema loading patch[1] adds \"required\" parameter support\nin the image resource schema. The required parameter enables glance-api\nto check required parameter is added at a new image creation. The enhancement\nworks well at the image create API. However, once the \"required\" parameter\nis activated at the image schema, Image Update API starts to fail\nbecause of missing required parameter in the Update Image API request body.\nAnd all the update image API return 400 Bad Request.\n\nThis commit adds the validate_property method in the Schema class\nfor using per property definition to validate update input, and updates\nthe _validate_change method in the RequestDeserializer class to\nvalidate input value per property definition.\n\n1. I6f4bdb78809511931b160c88e9173432fd2e6d06\n\nCloses-Bug: #2161751\nChange-Id: I5cb310ac5f0a2dc6df026c8d516a8c2586b2bfa2\nSigned-off-by: Masahito Muroi \u003cmasahito.muroi@g.softbank.co.jp\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/818b741365ea4e47f8c6aeedb048026b28031443"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/818b741365ea4e47f8c6aeedb048026b28031443"}]},"branch":"refs/heads/master"},"6401b4c40af6f81bf91350c1a7031ef6c7e3e9cb":{"kind":"REWORK","_number":3,"created":"2026-08-19 20:25:24.000000000","uploader":{"_account_id":8878,"name":"Masahito Muroi","email":"masahito.muroi@g.softbank.co.jp","username":"masa"},"ref":"refs/changes/68/998668/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/68/998668/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/68/998668/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/68/998668/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/68/998668/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/68/998668/3"}}},"commit":{"parents":[{"commit":"c76abb08e2bc82582389059dcbbc234293ce13b9","subject":"Merge \"Remove disabling the scope enforcement\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/c76abb08e2bc82582389059dcbbc234293ce13b9"}]}],"author":{"name":"Masahito Muroi","email":"masahito.muroi@g.softbank.co.jp","date":"2026-07-24 18:11:28.000000000","tz":540},"committer":{"name":"Masahito Muroi","email":"masahito.muroi@g.softbank.co.jp","date":"2026-08-19 20:25:07.000000000","tz":540},"subject":"Validate image update request schema check per schema item","message":"Validate image update request schema check per schema item\n\nThe Enhance Schema loading patch[1] adds \"required\" parameter support\nin the image resource schema. The required parameter enables glance-api\nto check required parameter is added at a new image creation. The enhancement\nworks well at the image create API. However, once the \"required\" parameter\nis activated at the image schema, Image Update API starts to fail\nbecause of missing required parameter in the Update Image API request body.\nAnd all the update image API return 400 Bad Request.\n\nThis commit adds the validate_property method in the Schema class\nfor using per property definition to validate update input, and updates\nthe _validate_change method in the RequestDeserializer class to\nvalidate input value per property definition.\n\n1. I6f4bdb78809511931b160c88e9173432fd2e6d06\n\nCloses-Bug: #2161751\nChange-Id: I5cb310ac5f0a2dc6df026c8d516a8c2586b2bfa2\nSigned-off-by: Masahito Muroi \u003cmasahito.muroi@g.softbank.co.jp\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/6401b4c40af6f81bf91350c1a7031ef6c7e3e9cb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/6401b4c40af6f81bf91350c1a7031ef6c7e3e9cb"}]},"branch":"refs/heads/master"},"ca46fd85a822a5c1cb1fd8ac20022f7bcef8bcf6":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2026-08-27 06:04:35.000000000","uploader":{"_account_id":8878,"name":"Masahito Muroi","email":"masahito.muroi@g.softbank.co.jp","username":"masa"},"ref":"refs/changes/68/998668/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/68/998668/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/68/998668/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/68/998668/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/68/998668/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/68/998668/4"}}},"commit":{"parents":[{"commit":"2699961fad7f17895abbcb82bb139843cebb4bac","subject":"tests: Ignore CHECK constraints in test_models_sync","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/2699961fad7f17895abbcb82bb139843cebb4bac"}]}],"author":{"name":"Masahito Muroi","email":"masahito.muroi@g.softbank.co.jp","date":"2026-07-24 18:11:28.000000000","tz":540},"committer":{"name":"Masahito Muroi","email":"masahito.muroi@g.softbank.co.jp","date":"2026-08-27 06:04:18.000000000","tz":540},"subject":"Validate image update request schema check per schema item","message":"Validate image update request schema check per schema item\n\nThe Enhance Schema loading patch[1] adds \"required\" parameter support\nin the image resource schema. The required parameter enables glance-api\nto check required parameter is added at a new image creation. The enhancement\nworks well at the image create API. However, once the \"required\" parameter\nis activated at the image schema, Image Update API starts to fail\nbecause of missing required parameter in the Update Image API request body.\nAnd all the update image API return 400 Bad Request.\n\nThis commit adds the validate_property method in the Schema class\nfor using per property definition to validate update input, and updates\nthe _validate_change method in the RequestDeserializer class to\nvalidate input value per property definition.\n\n1. I6f4bdb78809511931b160c88e9173432fd2e6d06\n\nCloses-Bug: #2161751\nChange-Id: I5cb310ac5f0a2dc6df026c8d516a8c2586b2bfa2\nSigned-off-by: Masahito Muroi \u003cmasahito.muroi@g.softbank.co.jp\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/ca46fd85a822a5c1cb1fd8ac20022f7bcef8bcf6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/ca46fd85a822a5c1cb1fd8ac20022f7bcef8bcf6"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"}},{"label":"Workflow","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{"label:Review-Priority\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
