)]}'
{"id":"openstack%2Fglance~998838","triplet_id":"openstack%2Fglance~stable%2F2026.1~Ie3d9430d34f2bf357905e58509349ea7eede1305","project":"openstack/glance","branch":"stable/2026.1","topic":"bug/2160248","attention_set":{},"removed_from_attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2026-07-28 19:39:32.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ie3d9430d34f2bf357905e58509349ea7eede1305","subject":"Fix error caused by queue directory being deleted in the middle","status":"MERGED","created":"2026-07-27 18:19:38.000000000","updated":"2026-07-28 19:40:35.000000000","submitted":"2026-07-28 19:39:32.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"998838-bug/2160248","cherry_pick_of_change":996681,"cherry_pick_of_patch_set":2,"meta_rev_id":"ae011bd6778444cf95aed3e48d27213fb17cc9f5","_number":998838,"virtual_id_number":998838,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-07-28 19:39:32.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"all":[{"value":2,"date":"2026-07-28 16:09:58.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"all":[{"value":1,"date":"2026-07-28 16:09:58.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Top Priority / Holds Gate"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-07-27 20:34:57.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-07-28 16:09:58.000000000","updated_by":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"reviewer":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"state":"REVIEWER"}],"messages":[{"id":"8d488094ab0e6e3237eb12a1b8ec9b0e3c95c144","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2026-07-27 18:19:38.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"600c7cfa3d9de7c8cf2caf65c8ddc2ee97280564","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2026-07-27 18:20:02.000000000","message":"Topic bug/2160248-stable/2026.1 removed","accounts_in_message":[],"_revision_number":1},{"id":"51c9eb4907d2c37bfb83c78de3c3a1a7563bc231","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2026-07-27 18:20:05.000000000","message":"Topic set to bug/2160248","accounts_in_message":[],"_revision_number":1},{"id":"0d5e7abfedc07ef928dabb4152f8434f7b48867f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-27 20:34:57.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b379ce8328764034ac9027d4ad09d1b9\n\n- grenade https://zuul.opendev.org/t/openstack/build/13473927e95a466c9043d3bfa97070c3 : SUCCESS in 57m 51s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/7b11441d2da846a0bd36646658f22356 : SUCCESS in 1h 06m 45s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/eeb46423eed945168fa30db725933640 : SUCCESS in 1h 41m 15s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/5271c738ae334dbe87ce0fe8f0829cff : SUCCESS in 1h 00m 45s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/43217d0b5a0947cbbf1ae1452052b63f : SUCCESS in 4m 47s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/23df8e6e63ee4cfd8e496027e9ad6ed7 : SUCCESS in 4m 43s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/c251949bb0fb42c884897b44c5dd8276 : SUCCESS in 4m 27s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/8632127837af46f09fd431473ef6fa23 : SUCCESS in 8m 30s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e350790b0a6742a78985e2e6a15609bf : SUCCESS in 5m 22s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/adc0bc865d9d4bbf92025f3e8cb7a648 : SUCCESS in 8m 01s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/c869571c426a40d4a36baa5ff0353f8c : SUCCESS in 7m 43s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/3897ae4bdfa9464aa27b2d99f42e331a : SUCCESS in 6m 26s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/fb525f90871040968c0de05e73945ca0 : SUCCESS in 1h 52m 57s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/704bbd1e77684cae89ab0f698ebcae58 : SUCCESS in 2h 07m 08s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/1caa7ff8d9ae460aa2e99afeefebf0fe : SUCCESS in 2h 01m 26s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/09b50a8b8a0246bebf0c638019269dc2 : SUCCESS in 1h 51m 27s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/904ab7cce669437b8063af8a58ecdd29 : SUCCESS in 1h 01m 05s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/418ad09847f14ec695a7a84f71afc0e5 : SUCCESS in 1h 40m 16s\n- glance-grenade-centralized-cache https://zuul.opendev.org/t/openstack/build/d2616f4b634742b0b377fe603a92d46a : FAILURE in 16m 31s (non-voting)\n- glance-s3-multistore https://zuul.opendev.org/t/openstack/build/c502d31c770d42c08000b1c954033134 : SUCCESS in 2h 09m 14s (non-voting)\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/b803caa1a2c347f0ba13ee5d9152d0ab : SUCCESS in 38m 58s","accounts_in_message":[],"_revision_number":1},{"id":"ce167121c5bb9f770657002d8f9250614ddff6e9","author":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"date":"2026-07-28 16:09:58.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"7cb990e6c627f1dbd8e2dcae85ca4813bcc87041","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-28 16:10:42.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"d44558e1d803004e93d166250bf01d308acd1f0b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-28 19:39:32.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/aeba3ebf57e74f6d8273bdfcc59b81ee\n\n- grenade https://zuul.opendev.org/t/openstack/build/40fdc0fbd7504369ba38d23ff3597da7 : SUCCESS in 38m 09s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/0362fb04b82243f29e7baf278e7a8044 : SUCCESS in 1h 07m 24s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/f3ddb732cc9f449cb3daad95c09d09e3 : SUCCESS in 1h 18m 31s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/cc93209ab1294596b1326bc436d3a5e6 : SUCCESS in 43m 58s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1b9b3a0b56114546b93c4c1bb7994f36 : SUCCESS in 2m 50s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/c3a495cb20c147d3acbe94c421c7e320 : SUCCESS in 2m 56s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/9b89ea0cbabf4ba2834f88e812a5bfa6 : SUCCESS in 3m 00s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9e0c0684120643d383bdb330d436d438 : SUCCESS in 3m 17s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/3972f790331843829f555566f673cf17 : SUCCESS in 6m 24s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/6a68a9c1c35740acbc0ba32f4beb5e56 : SUCCESS in 6m 43s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/4ec8e323cd3d4f27b4fb1f53ee3f300d : SUCCESS in 5m 23s\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/176bb36a10a74b968003d3dd8c2facee : SUCCESS in 2h 16m 58s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/315e8f32a289423cb40f7716ebf608d8 : SUCCESS in 1h 05m 43s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/43034e04cf88400e97742cbcb348a565 : SUCCESS in 1h 00m 36s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/f1205f9854bf4c99838dfa95a8102d53 : SUCCESS in 1h 38m 35s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/3f782ff3d1974dc89d31461f63ddbc48 : SUCCESS in 37m 14s","accounts_in_message":[],"_revision_number":1},{"id":"2212bf1af49f86e1d3a660cec81bea888ed4e1e5","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-28 19:39:32.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"ae011bd6778444cf95aed3e48d27213fb17cc9f5","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-28 19:40:35.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/729611c7f1b04fd2abe5ad7ad9ce0bb3\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b124c7f102394feaa837f99ff07d54e8 : SUCCESS in 52s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"83e28e79a979116d20ba0fd6f918f56ddba460f8","revisions":{"83e28e79a979116d20ba0fd6f918f56ddba460f8":{"kind":"REWORK","_number":1,"created":"2026-07-27 18:19:38.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/38/998838/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/38/998838/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/38/998838/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/38/998838/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/38/998838/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/38/998838/1"}}},"commit":{"parents":[{"commit":"f60807e789ebad0904747df361f9ec35ae5dd6a6","subject":"Update cursive repository path","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/f60807e789ebad0904747df361f9ec35ae5dd6a6"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-07-09 16:38:02.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-07-27 18:19:38.000000000","tz":0},"subject":"Fix error caused by queue directory being deleted in the middle","message":"Fix error caused by queue directory being deleted in the middle\n\nThe queue directories might be deleted right after it appears in\ndirectory listing. Make sure such deleted directories are ignored.\n\nCloses-Bug: #2160248\nChange-Id: Ie3d9430d34f2bf357905e58509349ea7eede1305\nSigned-off-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\n(cherry picked from commit cb52de2017d72c0e1a5579b87caf8f0dffbfe887)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/83e28e79a979116d20ba0fd6f918f56ddba460f8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/83e28e79a979116d20ba0fd6f918f56ddba460f8"}]},"branch":"refs/heads/stable/2026.1","conflicts":{"base":"e20488b1a6cfd59573130f56443dd4075f157989","ours":"f60807e789ebad0904747df361f9ec35ae5dd6a6","theirs":"cb52de2017d72c0e1a5579b87caf8f0dffbfe887","merge_strategy":"recursive","contains_conflicts":false}}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{"label:Review-Priority\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
