)]}'
{"id":"openstack%2Fglance~998865","triplet_id":"openstack%2Fglance~master~Ida852295360dc96f04869837ef89bcbad082f94f","project":"openstack/glance","branch":"master","topic":"bug/2160478","attention_set":{},"removed_from_attention_set":{"36712":{"account":{"_account_id":36712,"name":"Yunho Kim","display_name":"Yunho Kim","email":"yunho01.kim@samsung.com","username":"yunokim"},"last_update":"2026-08-04 15:41:58.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ida852295360dc96f04869837ef89bcbad082f94f","subject":"Prevent unit test fail due to stestr test ordering","status":"MERGED","created":"2026-07-28 02:30:29.000000000","updated":"2026-08-04 15:41:58.000000000","submitted":"2026-08-04 15:41:58.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"998865-bug/2160478","meta_rev_id":"1c7eff730e92a5a4430583cfedf1f09f51ef6aa9","_number":998865,"virtual_id_number":998865,"owner":{"_account_id":36712,"name":"Yunho Kim","display_name":"Yunho Kim","email":"yunho01.kim@samsung.com","username":"yunokim"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-08-04 15:41:58.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"recommended":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"all":[{"value":1,"date":"2026-08-03 11:12:18.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},{"value":2,"date":"2026-08-04 13:27:48.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":2,"date":"2026-08-03 09:06:19.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"all":[{"value":0,"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},{"value":1,"date":"2026-08-04 13:27:48.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Top Priority / Holds Gate"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-07-28 03:15:03.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2026-07-28 04:03:01.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-08-03 09:06:19.000000000","updated_by":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"reviewer":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"state":"REVIEWER"},{"updated":"2026-08-03 11:12:18.000000000","updated_by":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"reviewer":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"state":"REVIEWER"},{"updated":"2026-08-04 13:27:48.000000000","updated_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"reviewer":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"state":"REVIEWER"}],"messages":[{"id":"22f09bcb7a3358a3202a6f97869ea20a27e8c04f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36712,"name":"Yunho Kim","display_name":"Yunho Kim","email":"yunho01.kim@samsung.com","username":"yunokim"},"date":"2026-07-28 02:30:29.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"98ff3ea2be8b27243668b5f74ae21060b1d7fbb7","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-28 03:15:03.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4d39ba9d66cf4ba490b152785f9fc8d5\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/a0cfd1b3244e4aa1b849d75f8f9697de : SUCCESS in 8m 17s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/65a76401deda46c5a187ac0c9f8e9fa3 : NODE_FAILURE Node(set) request dfd5406dde4942b2be75d10239633f42 failed in 0s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/97d470b4f6d845f9ab7eebca99eff6a7 : SUCCESS in 10m 31s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"72ea09271c107a23c60b59aa1443d492e9c46768","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-28 04:03:01.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a2ada102d7994b659a7d12905a25a35a\n\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/3e685225492e47979cbf40166a0f9964 : SUCCESS in 1h 02m 57s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/622cb6e0bbef4dc4b26d30c47cf5fc8a : SUCCESS in 56m 11s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/99e5b7f5fe53446ca2dc17718c38aded : SUCCESS in 4m 01s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/073b891f416c432c826b0c2b8a694659 : SUCCESS in 5m 29s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/abd1fa66dfb5457aa041c6bcc64b285e : SUCCESS in 4m 20s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/4fcb7f1dc3864149b7527cba86cb2f13 : SUCCESS in 7m 28s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/62ef10fa028440fb8f1d2ea610851bea : SUCCESS in 7m 01s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/6fbfedc9b7e44074b83f17635809c964 : SUCCESS in 6m 55s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/c922668e4b4e4fd9a71bf42d98e292b2 : SUCCESS in 5m 24s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/a6a7bec083ef4f3f9f3ef4cb895466af : SUCCESS in 5m 47s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/7e7d618913d64387b6f6661d59ee109a : SUCCESS in 27m 47s","accounts_in_message":[],"_revision_number":1},{"id":"a733ea89a58d73dedb66b87b22596b74d6091c7a","author":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"date":"2026-08-03 09:06:19.000000000","message":"Patch Set 1: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"ccf2e4f33a8b7f798af173b38f319c66dc4046f9","author":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"date":"2026-08-03 11:12:18.000000000","message":"Patch Set 1: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"d455a6b951041bc1d5f1d1794f0afb1d61d40aea","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2026-08-04 13:27:48.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"023f02a618731f1c4405ad9c3247bc0b464d092e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-04 13:28:38.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"2813acc8776c92408afeb336404ed1f68c9699db","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-04 15:41:58.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a2af5452a23a4f9198ea2b9a695d94b7\n\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/53ef172a3fb344e0bb8e001f47cdd69c : SUCCESS in 43m 24s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/8eb5bd8848844a10adaf8849d35e8fd9 : SUCCESS in 39m 23s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/09ff8cd0d3664b3792b3fd2e4b3ffbe3 : SUCCESS in 4m 39s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/62c0c583ac734476b829da86609f647a : SUCCESS in 5m 48s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/401a5fa7e7fc4daa8a6bb276e476a9cc : SUCCESS in 5m 16s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ea7aa133f88f4945a07c20a83e4060f9 : SUCCESS in 5m 43s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/2f252f27adc04e5186420f9241bee858 : SUCCESS in 5m 41s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/dfd5ca93bdc34bb59682b30c16248b07 : SUCCESS in 7m 39s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/ee76934082ab445f803bb5f19f5ea143 : SUCCESS in 6m 11s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/35b30d03f79947d9b97491d40fde3a50 : SUCCESS in 37m 10s","accounts_in_message":[],"_revision_number":1},{"id":"1c7eff730e92a5a4430583cfedf1f09f51ef6aa9","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-04 15:41:58.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"e7cfcfcfd21001d08d2623261a39dc15ee7176c0","revisions":{"e7cfcfcfd21001d08d2623261a39dc15ee7176c0":{"kind":"REWORK","_number":1,"created":"2026-07-28 02:30:29.000000000","uploader":{"_account_id":36712,"name":"Yunho Kim","display_name":"Yunho Kim","email":"yunho01.kim@samsung.com","username":"yunokim"},"ref":"refs/changes/65/998865/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/65/998865/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/65/998865/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/65/998865/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/65/998865/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/65/998865/1"}}},"commit":{"parents":[{"commit":"7c993a6adf9b65a8c1cbcb3eadfaf8f3bc122d45","subject":"Merge \"Fix error caused by queue directory being deleted in the middle\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/7c993a6adf9b65a8c1cbcb3eadfaf8f3bc122d45"}]}],"author":{"name":"Yunhokim","email":"yunho01.kim@samsung.com","date":"2026-07-28 02:00:11.000000000","tz":540},"committer":{"name":"Yunhokim","email":"yunho01.kim@samsung.com","date":"2026-07-28 02:30:00.000000000","tz":540},"subject":"Prevent unit test fail due to stestr test ordering","message":"Prevent unit test fail due to stestr test ordering\n\nThree unit tests in glance/tests/unit/common/test_utils.py are order-dependent and fail when they do not happen to run in the same stestr worker process, after another test that registers the legacy glance_store options. They pass \"by accident\" of test distribution rather than being self-contained.\n\nCloses-Bug: #2160478\nChange-Id: Ida852295360dc96f04869837ef89bcbad082f94f\nSigned-off-by: yunokim \u003cyunho01.kim@samsung.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/e7cfcfcfd21001d08d2623261a39dc15ee7176c0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/e7cfcfcfd21001d08d2623261a39dc15ee7176c0"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{"label:Review-Priority\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
