)]}'
{"id":"openstack%2Fglance~999307","triplet_id":"openstack%2Fglance~master~Ib6e8235dd46811f4d841907ac29eca8ba2ada904","project":"openstack/glance","branch":"master","attention_set":{},"removed_from_attention_set":{"8122":{"account":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"last_update":"2026-08-04 15:42:36.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ib6e8235dd46811f4d841907ac29eca8ba2ada904","subject":"Make integration tests runnable again","status":"MERGED","created":"2026-07-30 15:35:26.000000000","updated":"2026-08-04 15:46:12.000000000","submitted":"2026-08-04 15:42:36.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":4,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"999307","meta_rev_id":"4ce79ef777b1d71a3a9e4b010e5288cfbe389ccb","_number":999307,"virtual_id_number":999307,"owner":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-08-04 15:42:36.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"all":[{"value":2,"date":"2026-08-04 13:29:53.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"all":[{"value":1,"date":"2026-08-04 13:29:53.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Top Priority / Holds Gate"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-07-30 16:16:23.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-08-04 13:29:53.000000000","updated_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"reviewer":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"state":"REVIEWER"}],"messages":[{"id":"289c709ca3fb3207a686b02f6491f80372d7a16e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"date":"2026-07-30 15:35:26.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"de278c47ef9ab7932fd05cf88518c3dfaae205db","author":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"date":"2026-07-30 15:36:34.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"334d9929e817f42cd8b550a7b9604a2aee671a54","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-30 16:16:23.000000000","message":"Patch Set 1: Verified-1\n\n(3 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/5ac11cd098ee46fa8b41135da3266e89\n\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/06faff9d9db14da7a5fc54c248bb8fdd : SUCCESS in 34m 49s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/8091d818287c48da92a4f342e7ce3e0e : SUCCESS in 39m 57s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/96bac62df2664d4c95fdc60a3279136d : FAILURE in 4m 42s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/878bf80e70b947e59c08e479e3a2c939 : SUCCESS in 4m 46s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/b61ed0eab7354cf697c74a8aa832dce2 : SUCCESS in 4m 44s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/d116ca6842d2448cb04dc54ce07148a7 : SUCCESS in 5m 25s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/859d50e1b2e94c8fb0f9d0df3c9c4a2c : SUCCESS in 4m 51s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/dac459852a9c4eaba68fea7cef70f8a6 : SUCCESS in 7m 36s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/198d27251287445387dc05bc85b4cee9 : SUCCESS in 6m 44s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/1b73bc12f9214a77a7376e161164a5a9 : SUCCESS in 6m 24s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/adfc727705704b79b5517a77572ea9aa : SUCCESS in 16m 34s","accounts_in_message":[],"_revision_number":1},{"id":"12fa62b504178e15ed45859075830005e60d94c1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"date":"2026-07-30 17:28:48.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":2},{"id":"6cbbc3f20594e95b1db1b589be1200ebef7c0c02","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-30 18:39:47.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4cf48933e03c4228bd8a54b7b2a1caba\n\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/2dfec5a656f54fa2a8d886204442063a : SUCCESS in 52m 19s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/5521dfbf29ee488d97d3c220823e6985 : SUCCESS in 1h 04m 29s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/77955d325e804285a41b3c44ede7c123 : SUCCESS in 5m 38s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/11cd7a9981ed46149a96d83ca8a14903 : SUCCESS in 5m 04s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/008363246e354e87994fe146d77ea05e : SUCCESS in 5m 17s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/a6a86461505945eb9e6d7bc072b9ff14 : SUCCESS in 7m 52s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/203b387ecc0b4aec971757b65db6f52f : SUCCESS in 5m 54s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/506598cd93294d298b43e9d01a480457 : SUCCESS in 9m 21s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/3e0f563dc4e74d55b5fe358b6c8f1ce5 : SUCCESS in 6m 12s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/3ce8c88f7749420d8b4ce9b7c3761ee8 : SUCCESS in 6m 48s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/74288bf7ab794d30aa6b19d763374da0 : SUCCESS in 34m 06s","accounts_in_message":[],"_revision_number":2},{"id":"9197b1f08bd79ee277f44c6f1e4566f5a17cda72","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-30 20:38:35.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7a99e37fb1044600aef50ff44c915c4a\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/db3bb91b92bb45579cb5273c05cb2cd0 : SUCCESS in 9m 46s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/b2bfaeaa198247618c113f2779fc53d1 : SUCCESS in 10m 49s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/147512e9889a4aec90955a102708c636 : SUCCESS in 20m 39s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"1ec69e21fa35f38c20fa1b2fb52e71d99008b1f3","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2026-08-04 13:29:53.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"d8de7049f23b8410d1689983327e67d3fadd4e44","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-04 13:30:26.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"31a4412112c041c47c8ec97a988a5d77ec2faa4e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-04 15:42:36.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e3633c704fd9494c89ca7f0b41474fbb\n\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/bc2113731fb94466869ad7ea1c2a6525 : SUCCESS in 45m 52s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/2283b1d48b204a0aae12cd944c3d5da8 : SUCCESS in 39m 49s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f729ae141f7d41c0bf43ddd658d69cb9 : SUCCESS in 4m 56s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/0b981ff29fec4280a273ec5b6c116405 : SUCCESS in 5m 03s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/81a1f6c881534cd6b85c0e7c5446344a : SUCCESS in 5m 38s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/62e0f34425094d6d91a82a1a439c0bb8 : SUCCESS in 6m 28s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/92fef382d4f441fca69fcda074d95f5e : SUCCESS in 7m 30s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/b09afaa4161c4f35bb817d7e78f50186 : SUCCESS in 7m 52s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/cad98b7a37004db08664f003387c364e : SUCCESS in 6m 10s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/d1bd6f7fd2ed41c599680d686e72a792 : SUCCESS in 41m 36s","accounts_in_message":[],"_revision_number":2},{"id":"778750d6d532e1e0e7b74d3a04e29e09019cce08","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-04 15:42:36.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"4ce79ef777b1d71a3a9e4b010e5288cfbe389ccb","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-04 15:46:12.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7dc0a3dddbd848008b1951d310b5cbc3\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5d0dfb94f5aa4d9889f8dff175f2139a : SUCCESS in 50s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"b8d7d51e140672ee38db7a39f39a84812b934830","revisions":{"11701cc51393714a4a6cb42ad5d281eff4135c89":{"kind":"REWORK","_number":1,"created":"2026-07-30 15:35:26.000000000","uploader":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"ref":"refs/changes/07/999307/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/07/999307/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/07/999307/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/07/999307/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/07/999307/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/07/999307/1"}}},"commit":{"parents":[{"commit":"7c993a6adf9b65a8c1cbcb3eadfaf8f3bc122d45","subject":"Merge \"Fix error caused by queue directory being deleted in the middle\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/7c993a6adf9b65a8c1cbcb3eadfaf8f3bc122d45"}]}],"author":{"name":"Cyril Roelandt","email":"cyril@redhat.com","date":"2026-07-30 14:59:33.000000000","tz":120},"committer":{"name":"Cyril Roelandt","email":"cyril@redhat.com","date":"2026-07-30 15:35:20.000000000","tz":120},"subject":"Make integration tests runnable again","message":"Make integration tests runnable again\n\nBy default, stestr only lists and runs tests from glance/tests/unit. One\ncan still run them using the following command:\n\n    OS_TEST_PATH\u003d./glance/tests/integration/ stestr run\n\nThese tests have probably not been run since\n9a55baa2c89d11a1fcc66ee97daf1ec426c2c1b7 and required some syntax change\nin the pipline config to work again.\n\nSome small changes in the tests themselves were required to have them\npass again.\n\nAssisted-By: Claude Sonnet 4.5 \u003cnoreply@anthropic.com\u003e\nChange-Id: Ib6e8235dd46811f4d841907ac29eca8ba2ada904\nSigned-off-by: Cyril Roelandt \u003ccyril@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/11701cc51393714a4a6cb42ad5d281eff4135c89"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/11701cc51393714a4a6cb42ad5d281eff4135c89"}]},"branch":"refs/heads/master"},"b8d7d51e140672ee38db7a39f39a84812b934830":{"kind":"REWORK","_number":2,"created":"2026-07-30 17:28:48.000000000","uploader":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"ref":"refs/changes/07/999307/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/07/999307/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/07/999307/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/07/999307/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/07/999307/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/07/999307/2"}}},"commit":{"parents":[{"commit":"7c993a6adf9b65a8c1cbcb3eadfaf8f3bc122d45","subject":"Merge \"Fix error caused by queue directory being deleted in the middle\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/7c993a6adf9b65a8c1cbcb3eadfaf8f3bc122d45"}]}],"author":{"name":"Cyril Roelandt","email":"cyril@redhat.com","date":"2026-07-30 14:59:33.000000000","tz":120},"committer":{"name":"Cyril Roelandt","email":"cyril@redhat.com","date":"2026-07-30 17:28:40.000000000","tz":120},"subject":"Make integration tests runnable again","message":"Make integration tests runnable again\n\nBy default, stestr only lists and runs tests from glance/tests/unit. One\ncan still run them using the following command:\n\n    OS_TEST_PATH\u003d./glance/tests/integration/ stestr run\n\nThese tests have probably not been run since\n9a55baa2c89d11a1fcc66ee97daf1ec426c2c1b7 and required some syntax change\nin the pipline config to work again.\n\nSome small changes in the tests themselves were required to have them\npass again.\n\nAssisted-By: Claude Sonnet 4.5 \u003cnoreply@anthropic.com\u003e\nChange-Id: Ib6e8235dd46811f4d841907ac29eca8ba2ada904\nSigned-off-by: Cyril Roelandt \u003ccyril@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/b8d7d51e140672ee38db7a39f39a84812b934830"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/b8d7d51e140672ee38db7a39f39a84812b934830"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{"label:Review-Priority\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
