)]}'
{"id":"openstack%2Fglance~999754","triplet_id":"openstack%2Fglance~master~Ice2045067fa92ecc8176a6ae05bbb9e20e4cbb66","project":"openstack/glance","branch":"master","topic":"policies","attention_set":{},"removed_from_attention_set":{"39279":{"account":{"_account_id":39279,"name":"Alexandre Engloner","display_name":"AEngloner","email":"alexandre.engloner@gmail.com","username":"aengloner"},"last_update":"2026-08-17 13:59:16.000000000","reason":"\u003cGERRIT_ACCOUNT_39279\u003e replied on the change","reason_account":{"_account_id":39279,"name":"Alexandre Engloner","display_name":"AEngloner","email":"alexandre.engloner@gmail.com","username":"aengloner"}}},"hashtags":[],"change_id":"Ice2045067fa92ecc8176a6ae05bbb9e20e4cbb66","subject":"fix for 2 methods in metadefs policies","status":"MERGED","created":"2026-08-04 17:57:21.000000000","updated":"2026-08-17 20:36:34.000000000","submitted":"2026-08-17 20:35:06.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":6,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"999754-policies","meta_rev_id":"c4aa20ba421dd1f37efc05ad85c0b58571125d99","_number":999754,"virtual_id_number":999754,"owner":{"_account_id":39279,"name":"Alexandre Engloner","display_name":"AEngloner","email":"alexandre.engloner@gmail.com","username":"aengloner"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-08-17 20:35:06.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"all":[{"value":2,"date":"2026-08-06 06:23:37.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":2,"date":"2026-08-05 15:36:06.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"all":[{"value":1,"date":"2026-08-06 06:23:37.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Top Priority / Holds Gate"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-08-04 18:48:35.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2026-08-04 20:02:11.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-08-05 15:36:06.000000000","updated_by":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"reviewer":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"state":"REVIEWER"},{"updated":"2026-08-06 06:23:37.000000000","updated_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"reviewer":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"state":"REVIEWER"}],"messages":[{"id":"5521e5cd13cfe073f435a78cb1734c5c22ceda31","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":39279,"name":"Alexandre Engloner","display_name":"AEngloner","email":"alexandre.engloner@gmail.com","username":"aengloner"},"date":"2026-08-04 17:57:21.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"ced147674e41ba00f505bf80359d99cb6f08aa1e","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-04 18:48:35.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/30a8415f7dc44bf2895cb4ad721abc9f\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/9a9887bfb897490c8c3792a5b373115b : SUCCESS in 7m 01s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/31c0206046684a1494af5d91b8017614 : SUCCESS in 7m 03s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/7fa93b4dedf14f7e9a16b2eb9ec97819 : SUCCESS in 11m 03s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"bc2b3526bf292e9267f9b04afe5428584b2d3e98","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-04 20:02:11.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1e492de4a8e34c139da6cbc5037717d8\n\n- grenade https://zuul.opendev.org/t/openstack/build/a5c8b311a725478682167c07be8c0c22 : SUCCESS in 1h 02m 53s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/f0d61ee43c78429e976cc7b9e5b3c22c : SUCCESS in 58m 29s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/8cd1a3c746d14a1fac80af6aaf29cbd0 : SUCCESS in 1h 03m 18s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/b3ea7c62d4014b358c2c6205897b60df : SUCCESS in 1h 01m 53s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/76190fe1807a45219bf80b93f6cb2bec : SUCCESS in 4m 38s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/cc7ab8be4b084d07bff9ac1dc3d69c98 : SUCCESS in 3m 58s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/c2b0adea708d442d9fc721621739a91b : SUCCESS in 4m 52s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/439efcf0ccba48c998c8bff71cc9f804 : SUCCESS in 9m 09s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/68628a4de9a7464381957a4e26363238 : SUCCESS in 5m 28s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/9eb1ecfd4e544c45bd87d5a73e28e2b7 : SUCCESS in 7m 17s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/ec3a4ea056754c1ab849c10533ade170 : SUCCESS in 6m 14s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/9983f34adc574b8d95714afa52b5fc26 : SUCCESS in 6m 41s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/8c49df716e1d4f40b146bd0534416469 : SUCCESS in 1h 31m 18s (non-voting)\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/82368e242944418aadc93333a1c0bf7b : SUCCESS in 1h 09m 16s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/71e95c8d9fed4c099a3d1c0028fc5f36 : SUCCESS in 2h 02m 55s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/4d9db743dc94407795ff47d8421934ac : SUCCESS in 45m 42s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/89ce873e2f4b4b8b8616df7ba6578070 : SUCCESS in 1h 41m 51s\n- glance-s3-multistore https://zuul.opendev.org/t/openstack/build/c88bfc96d6104dc4987e10e291d73772 : SUCCESS in 1h 37m 33s (non-voting)\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/7f0d70aadeb245569739e8c397226ff7 : SUCCESS in 35m 27s\n- glance-rally-task-uwsgi https://zuul.opendev.org/t/openstack/build/d593adb684424f78900b6a984bf19952 : SUCCESS in 52m 56s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"30f179db37925ae5a6f44d2098f7dba15f855b56","author":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"date":"2026-08-05 15:36:06.000000000","message":"Patch Set 1: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"6096d43c5e4f200264f8982054309b4c64bdbe06","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2026-08-06 06:23:37.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"9ec7eb26e9c7304048ed21c0a5d1cc823e196570","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-06 06:24:09.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"a7404223c7e548b35ac8752a171c89ced93844f2","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-06 08:05:52.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/e906edcd8b234889bf72252591117457\n\n- grenade https://zuul.opendev.org/t/openstack/build/457232384adc4cad9205595b8544ed35 : FAILURE in 18m 07s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/66ca0f0ee70d4fa68fb7bb3e961436ec : SUCCESS in 1h 03m 50s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/fd2a42b70cab48e5bf3ba376a75654a0 : SUCCESS in 1h 23m 21s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/9d9c1b7cbf5f45e5a4447d056ff4e793 : SUCCESS in 45m 39s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2990fa8ed60f441d895346bd489475a8 : SUCCESS in 5m 01s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/014f0a33119f42eca96f2db2580db883 : SUCCESS in 4m 57s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/98c58367f804481eb99b76d522f03149 : SUCCESS in 4m 43s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/932f214747ab4741b4b34a193942afaa : SUCCESS in 3m 50s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/331090a0a49b4e65ae6f181b9d87cbec : SUCCESS in 5m 59s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/fbdf16311fe2401f8d71f570e9ae6349 : FAILURE in 6m 57s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/9a761e9bbe794166bfce67b43e74a885 : SUCCESS in 5m 53s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/5f6118ca8b1440268b00ad1b42a87175 : FAILURE in 28m 27s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/42fcbb33c30d43bbb9584fe79d08815d : FAILURE in 13m 21s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/c4360a8de0684c5fb4627ab15ca6dcf1 : SUCCESS in 1h 36m 46s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/b4dd10fce15a42f496b5d1d7fa9533e0 : SUCCESS in 25m 37s","accounts_in_message":[],"_revision_number":1},{"id":"b6d459e6606e408abeeb8be7eb13d8a3259eeae7","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2026-08-06 08:29:05.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"2c5e962bf73993357790e808ee63b85cd3545272","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-06 08:40:14.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/90b8b3d0549e47918c6a710ea7961092\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/aa1c79a4f8eb4b468147a9462d3fa11d : SUCCESS in 7m 48s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/8173736dc0c3434b8cb1cb6d527967c4 : SUCCESS in 6m 50s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/4c1ec09be0ad49f8a4261995585eb083 : SUCCESS in 8m 38s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"06e9a7b3ef02ce2176e481fce4762ebb5184d977","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-06 09:45:36.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/bd0dbc39798e4b6fbfbc4b8cc0a64c6c\n\n- grenade https://zuul.opendev.org/t/openstack/build/63d542eb19f34e6f81dcc1fa49541491 : SUCCESS in 1h 04m 00s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/6b04eaf039f44693be17719de0159e10 : SUCCESS in 53m 58s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/3dcf5108aeec4449b76cb12ad429738e : FAILURE in 27m 23s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/00248993bea44ab8883ae36e82bce44b : FAILURE in 1h 00m 34s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ad1c1c5cc59d4c03aed4c8b4df9a64c5 : SUCCESS in 5m 52s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/5f9b9f234bc641748ac6363f20118c90 : SUCCESS in 5m 53s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/38f4b1ffde194937abe76330251fd76d : SUCCESS in 5m 17s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/0c268e70160a484cbe5811445da997be : SUCCESS in 7m 04s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/085ad6dddaf6450f9554cf7b42704af7 : SUCCESS in 5m 21s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/973ba5b1b227470298833b95641b3c59 : SUCCESS in 8m 29s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/e24de2e5e04141069930befff7c2d6ef : SUCCESS in 7m 08s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/1d1c7db58ee24dc38f795624b7c9f9ea : SUCCESS in 7m 29s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/3783ff984ae041909a75cca31f8f6c05 : FAILURE in 21m 47s (non-voting)\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/022a2b7f30834084b69b8a5f3087c3c1 : SUCCESS in 48m 30s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/db4cf2a2a64c4ab6ac000e8d9bfda879 : SUCCESS in 1h 02m 06s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/535739f4538e4ca28e42eb7db967d7ff : FAILURE in 11m 39s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/f39ce34dd1494fed8fc076b17d4e8e57 : POST_FAILURE in 47m 03s\n- glance-s3-multistore https://zuul.opendev.org/t/openstack/build/a3e917fa222d4061989461117673cdfd : SUCCESS in 1h 09m 20s (non-voting)\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/8b6c3477fade4c0fb6d9a7ca99f201e7 : SUCCESS in 25m 43s\n- glance-rally-task-uwsgi https://zuul.opendev.org/t/openstack/build/b87ef26be5e547b4b34c4ab81cc67725 : SUCCESS in 44m 31s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"e11625bd2dda56c59b8660b8514fc9d5e8e22ecb","author":{"_account_id":39279,"name":"Alexandre Engloner","display_name":"AEngloner","email":"alexandre.engloner@gmail.com","username":"aengloner"},"date":"2026-08-06 13:40:11.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"116176eba59391c17ccb4695f68e98086eabd245","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-06 14:09:30.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3a49466bb387459b92c10847ed8ada20\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/267006349c4d47c687f689320073b923 : SUCCESS in 6m 32s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/59d66ae93fe84675a98c2a38014b3602 : SUCCESS in 10m 01s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/ca4336e1c20b4dae92408b6f8a886561 : SUCCESS in 17m 02s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"b6e46e619ffabd338655bccd59b9fb86e45f419d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-06 15:38:16.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/e94f1f9cd9be4eadad3b8088de15fb8a\n\n- grenade https://zuul.opendev.org/t/openstack/build/928b1f4dda5c49bb9faaa8121b74dce0 : SUCCESS in 38m 10s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/9fc008b6a0e7448ea2bc228eea8d0832 : FAILURE in 20m 11s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/d25e8cdbc5354438974f3fb82be2385c : SUCCESS in 1h 26m 55s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/6abc60e10ddf4bc8b53d55e91f04aa4a : SUCCESS in 57m 38s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5f5ec230a4614647ad6967a2713066ba : FAILURE in 3m 43s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/33468d5b4d2145519626f4230005a6df : SUCCESS in 6m 23s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/03010be4c295461190e27960fc7ae45a : SUCCESS in 5m 16s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/302b6aa5af1247848d9477d5cdd62ba9 : SUCCESS in 5m 23s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4d96890759794541a0131b9440177133 : SUCCESS in 5m 29s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/079b9c79fca24f83885dc8043518ee7e : SUCCESS in 5m 26s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/858acc935e124807b7e00ec40b6044ef : SUCCESS in 8m 51s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/d24bd155648f414c8d039a51ba7086f6 : FAILURE in 4m 34s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/e9b6fdab6b5f4bab960cac209278bc9a : SUCCESS in 1h 37m 10s (non-voting)\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/c14d88924f094866bacc1002b5e7b38a : FAILURE in 26m 29s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/4237784ff3ad46d1991db9387b0dd5bb : SUCCESS in 1h 11m 23s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/b3fbfac3eda5473885f28545fbaf0558 : FAILURE in 31m 02s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/90cc638c989b45f1b8b04cb41cba1518 : FAILURE in 28m 57s\n- glance-s3-multistore https://zuul.opendev.org/t/openstack/build/c4e2a372a7964d3aae4b1b1296cdec6b : SUCCESS in 56m 22s (non-voting)\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/a813d5b0e4134487a49808ed434be7e8 : SUCCESS in 32m 45s\n- glance-rally-task-uwsgi https://zuul.opendev.org/t/openstack/build/9953f27a3dc74f9db36bbd5bdd3ce7a6 : SUCCESS in 50m 17s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"a182fe0e7dd4b9aa5be57bbc3c64cb603ac69015","author":{"_account_id":39279,"name":"Alexandre Engloner","display_name":"AEngloner","email":"alexandre.engloner@gmail.com","username":"aengloner"},"date":"2026-08-07 12:29:21.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"a2b070c1f3706f86e71c67c478e28a0efb1d55b1","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-07 12:47:45.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/efaaa20b38f940c6b1eec17a02bbc515\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/11780c5bd167425f8aed71a771ac8d1c : SUCCESS in 10m 43s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/c25274bf28234fe2b3566d560895d018 : SUCCESS in 8m 52s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/953f7547fcb44b9fad2061ce02431775 : SUCCESS in 16m 11s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"f66f255a2b59e463fc12a3079f12e3c2f86b79e1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-07 14:12:47.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/c3ebae2d470943008d98606fb29841c5\n\n- grenade https://zuul.opendev.org/t/openstack/build/ebdc66da49cc4fb18343338bd979a578 : SUCCESS in 51m 45s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/0db16ac80c204a1ea89689469eb72dd6 : SUCCESS in 29m 59s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/5a4cea03df2045aba59c120a1b7cd3fd : SUCCESS in 52m 33s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/bde6eddeb1e94c46a46194c698dec07b : FAILURE in 33m 04s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b2e8a6b3492a448d8f77b1459173a9db : SUCCESS in 4m 32s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/9ac0e3bdc6644773bed1c6f251041e9c : SUCCESS in 4m 54s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/f37e8d4e4d374b7ab9a1f885b41052e5 : SUCCESS in 2m 53s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/24e5ce10ea124e058b713ba9d4db814d : SUCCESS in 4m 16s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6697397206ef4af38cb753a3758b788e : SUCCESS in 4m 33s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/8592e544a26649ad8b5b3ecda755bb52 : SUCCESS in 4m 52s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/3009a72033c744b1b63f7519228891ce : SUCCESS in 5m 20s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/f178bcdc56914e4b9092495913d5c343 : SUCCESS in 5m 04s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/3cecbd2716234b8e85adbe703ed2d7db : SUCCESS in 50m 00s (non-voting)\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/abbd8ac23f8e4ea4b1661793be28999b : SUCCESS in 59m 40s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/cf26fd0e4cfa43f88d5f8a08ce232806 : SUCCESS in 1h 40m 33s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/8063c7389af04eae8d95446ad198471e : SUCCESS in 27m 01s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/94da92dee3fb4e93998f69fadba203fd : SUCCESS in 1h 32m 20s\n- glance-s3-multistore https://zuul.opendev.org/t/openstack/build/8a48b726f5e54556bf96e1eaab7c2bbb : SUCCESS in 1h 09m 51s (non-voting)\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/d8db525b629141f585d88a6bd844df6b : SUCCESS in 34m 47s\n- glance-rally-task-uwsgi https://zuul.opendev.org/t/openstack/build/1e9a26334ed94cbd8a6376c46367ee0f : SUCCESS in 40m 14s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"6b5105bb4b155d703aaa30adc07f9566139c2674","author":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"date":"2026-08-07 19:58:32.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"af98c4768415d46e4a8348d45f44021082b0552b","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-07 20:13:30.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d7615912281d492dba41c0a9d71a43df\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/786c905234a84ffaa406f6f148c91c02 : SUCCESS in 8m 36s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/aaf3ae71f4c54048a4b8dced59986b08 : SUCCESS in 9m 24s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/ad23eafc45944bf08d7a53e03a162ef5 : SUCCESS in 12m 24s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"6b3822ebfd708ea2265912af756e3e50023001b9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-07 22:02:07.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/de5a0eb3b48341469c9a03836e3e668d\n\n- grenade https://zuul.opendev.org/t/openstack/build/2a1ac0f3444448c3abf40fade6ee670c : SUCCESS in 31m 40s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/44529826122648c2ab315dac29c6616e : FAILURE in 1h 03m 42s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/1ccda4e29e8a4f3493c11fcc457eed8b : SUCCESS in 1h 35m 29s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/1086d04a5c5b48a8ac584da0cb1212c1 : FAILURE in 55m 29s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5d795951fca24ded95d683086729c355 : SUCCESS in 5m 48s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/ba1d6eaf92624ef7b875e80ca421c719 : SUCCESS in 6m 16s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/bf0487a9b9b640bea9e89efb2c825d70 : SUCCESS in 4m 57s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/8b6834612cb547beba0e14ff4b648373 : SUCCESS in 8m 15s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2546b04e955f4dc3b4ada65755460fe2 : SUCCESS in 5m 51s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/00a20552e52d400e8b31728d6b5968a4 : SUCCESS in 8m 29s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/8f1ac696876b4e44a09d1f933ac9b99f : SUCCESS in 6m 12s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/9bcfb6a39ee241fd92cc503cc8b8ea93 : SUCCESS in 5m 12s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/afc11dfdd2034a7190d94a38faf2f8aa : SUCCESS in 1h 57m 18s (non-voting)\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/79b2aab817c1415bbccc5bc2e1b7681c : SUCCESS in 1h 52m 27s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/6bf54b41a62d4ff5859cd200b898edd5 : SUCCESS in 1h 13m 56s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/65e1cb21c15c47d486e551e25d0e15f2 : SUCCESS in 37m 39s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/794f5ccbec934bdd95d639a761569038 : SUCCESS in 1h 38m 01s\n- glance-s3-multistore https://zuul.opendev.org/t/openstack/build/12491cff69e941628522a01684583a65 : FAILURE in 1h 11m 40s (non-voting)\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/d6d7c9d662694fa29c8ec446468d693a : SUCCESS in 38m 49s\n- glance-rally-task-uwsgi https://zuul.opendev.org/t/openstack/build/fdfb87e84ce54ebbbd793a1c188d563b : SUCCESS in 40m 21s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"4d6fed36286e84a5c68f9d44bb075e66cd3106e6","author":{"_account_id":39279,"name":"Alexandre Engloner","display_name":"AEngloner","email":"alexandre.engloner@gmail.com","username":"aengloner"},"date":"2026-08-10 13:37:20.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"185284e4236334a9afcf328a2a3e6fc51d4cb838","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-10 13:53:11.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/96eba1d5d0664e8cb5f2e723f6e0619d\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/48999a9253b24caa949641ec7a633afe : SUCCESS in 6m 18s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/6dcaa9eed8b949849be798895dc6f6e0 : SUCCESS in 9m 10s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/5e3f38fa6e9d4c3ab014e0cb0faa56a6 : SUCCESS in 13m 38s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"f35712e09a2fd05ad20b3ffc2051e9ed2f38295f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-10 15:28:02.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/ab7b370f7dd140c892e9dacabc735289\n\n- grenade https://zuul.opendev.org/t/openstack/build/0af8d156a610486b89cefdf0395e7cfa : SUCCESS in 52m 12s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/d413c618e9a24147937b1a913f594841 : SUCCESS in 49m 01s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/4da1d3d210694302b5d150772ada56ea : SUCCESS in 1h 21m 34s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/3ca6a7e92923447db17246addb25de1b : FAILURE in 1h 01m 10s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cfaac965435f4d73a313199270c8d40d : SUCCESS in 4m 38s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/b4758efa0af04e02ba448cdd26c37889 : SUCCESS in 6m 55s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/cb12581898f94b0eb8d735deab991c15 : SUCCESS in 5m 31s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/a920e9ff768c4fb9b33e7897e41c1a2a : SUCCESS in 4m 56s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e4c8bff1ca794ff68821fe23ffc21f91 : SUCCESS in 3m 55s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/de52a3071e4c4581bf947cc7c564c79b : SUCCESS in 7m 57s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/5c152780db5843a1bc1c428ec2272f26 : SUCCESS in 6m 13s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/a9dc19a5358d48c5a6bf581c194c4dd8 : SUCCESS in 7m 54s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/30774fcebcb94871a211e338ac4f35ed : SUCCESS in 1h 33m 18s (non-voting)\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/b345646355924747ac1e5d711812c9fd : SUCCESS in 1h 36m 27s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/9950c775667a4db291d07507f3eba988 : SUCCESS in 1h 40m 01s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/6cdc50ac96b141a2ad7e623efaccbb05 : SUCCESS in 26m 15s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/ee37767e8084471685d59fe7d77e0c1e : SUCCESS in 1h 23m 28s\n- glance-s3-multistore https://zuul.opendev.org/t/openstack/build/cc4e80fae132419eb0b74e1070595b9c : SUCCESS in 1h 33m 55s (non-voting)\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/04c8f508f52f4090ad9718296126d04a : SUCCESS in 16m 12s\n- glance-rally-task-uwsgi https://zuul.opendev.org/t/openstack/build/d2bef3f0a1b543c2b52bd44da68e2144 : SUCCESS in 47m 55s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"67f35d3138e8670818b82b97fb34dc3ce9f4fbbe","author":{"_account_id":39279,"name":"Alexandre Engloner","display_name":"AEngloner","email":"alexandre.engloner@gmail.com","username":"aengloner"},"date":"2026-08-17 13:59:16.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"db8b52bb9fd2ae25702c6888b0dd902254d3792b","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-17 14:38:18.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/00067e176a28490490e394020fac208d\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/93894bdaef4a4e409e2d4b9712b77ace : FAILURE in 4m 09s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/73229f9e7a2d4c2ba2e2dbb11dc0d699 : FAILURE in 6m 29s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/d36f82a6f7194df5be0355bf96253326 : FAILURE in 9m 34s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"3864015adeaf57c56dcc5737c4faac6c503a4f39","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-17 16:19:31.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/68708f11f88a4605afc990f8278207da\n\n- grenade https://zuul.opendev.org/t/openstack/build/09ce927acc3746be88376939391c86b3 : SUCCESS in 1h 00m 23s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/9df0bab4b6e044c4820ebac8e6ab7329 : SUCCESS in 1h 07m 24s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/d41f851e18264910abec0c2525e064bd : SUCCESS in 1h 29m 19s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/9139a16f9f8f4f80b2267fa71aeb687c : SUCCESS in 34m 34s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/455852c6558d4bb886132111bb39882a : SUCCESS in 3m 25s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/bfd2a5d466404600b1aca30e2f989257 : SUCCESS in 3m 27s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/c90172fd86484e15a81e0b4fd487b80d : SUCCESS in 6m 37s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/e373ef062a2f4a5bae364db129316365 : SUCCESS in 7m 17s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/54057417334d4cbfab7b6370e8901a98 : SUCCESS in 6m 05s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e9d95dada67f4bbf843b3c89024f5069 : SUCCESS in 8m 14s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/89f246067273481885e1b1b66dbcccb8 : SUCCESS in 6m 07s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/3a2def2bbba64c78a76c5df9066ed802 : SUCCESS in 6m 24s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/e45152a8b6944a0e8c87b71f67d418e1 : SUCCESS in 1h 36m 52s (non-voting)\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/6a76ff653ab144429606347318c8aa10 : SUCCESS in 1h 42m 26s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/a8d17c1f51f6414284609914a85aedc4 : SUCCESS in 2h 11m 00s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/a28c3945220242a1887b614c6d6bcb89 : SUCCESS in 52m 23s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/2b02615cab7f417baf4b2a9d6064f3f6 : SUCCESS in 1h 06m 04s\n- glance-s3-multistore https://zuul.opendev.org/t/openstack/build/1620ca361a814a6ca22ad02b6fe6c31d : SUCCESS in 1h 20m 07s (non-voting)\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/f60577f369ba4b12951a58cc600aba16 : SUCCESS in 39m 51s\n- glance-rally-task-uwsgi https://zuul.opendev.org/t/openstack/build/0d73fbef60ce48b7bacff17bd8407444 : SUCCESS in 49m 17s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"21a59605898ebf3c7e1e12a6a76c76a26d8b7535","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-17 16:20:09.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"e04817dd3687d0cf0b3e002611b8b31b0017dd14","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-17 20:35:06.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4b6d95297f6e44e2b6f7909943b97e66\n\n- grenade https://zuul.opendev.org/t/openstack/build/e85a772301ea4a1082b82a7eeb2c4f9a : SUCCESS in 1h 01m 19s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/48f952d714184357a799c0d6fedda3cd : SUCCESS in 59m 26s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/48eb1c102dc84d5dad3ad0df61585d0a : SUCCESS in 1h 18m 47s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/a223d0931eba41d6804b909a1d1e30e0 : SUCCESS in 1h 01m 16s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cc318eb9f8cd444f8977b8d383ea45b2 : SUCCESS in 5m 09s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/bdce064af8ba4ad1b237853f57a46cb5 : SUCCESS in 4m 55s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/c859beee5bee4b85bdf4ec2f2d5270f4 : SUCCESS in 4m 41s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/aa01929f5eff4385947ec81c7b2d1d16 : SUCCESS in 8m 43s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c5b32d1fa5224ea0aa4ff64d260f0fd4 : SUCCESS in 8m 09s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/7560d28f143249af87f455089ee9cf1d : SUCCESS in 7m 33s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/a4a2b194b6214515abc2508ff32512f7 : SUCCESS in 7m 40s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/466f00f22dc24d68b4362e3ffb3e3e6a : SUCCESS in 1h 45m 19s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/273784159a074595b6f24cbe67996255 : SUCCESS in 52m 52s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/3664d2c03eb848ae95978bce3f736b7c : SUCCESS in 1h 39m 15s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/62faae53c2c54afdb74ac695e482a404 : SUCCESS in 33m 05s","accounts_in_message":[],"_revision_number":1},{"id":"c348e758354b57d01dcdbd8535a7b9fdb5f78246","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-17 20:35:06.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"c4aa20ba421dd1f37efc05ad85c0b58571125d99","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-17 20:36:34.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/57f66d2f3a64449d96abbb892dd3e69a\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/41ffe0dad12241b490d966ea05c172b9 : SUCCESS in 50s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"f1f60c2f4f4c1315a35c2c1a696d8a4f50317b34","revisions":{"f1f60c2f4f4c1315a35c2c1a696d8a4f50317b34":{"kind":"REWORK","_number":1,"created":"2026-08-04 17:57:21.000000000","uploader":{"_account_id":39279,"name":"Alexandre Engloner","display_name":"AEngloner","email":"alexandre.engloner@gmail.com","username":"aengloner"},"ref":"refs/changes/54/999754/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/54/999754/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/54/999754/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/54/999754/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/54/999754/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/54/999754/1"}}},"commit":{"parents":[{"commit":"5ccb9bef9c7791f6c53d3265df473b8cd1abce37","subject":"Merge \"Drop redundant overrides\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/5ccb9bef9c7791f6c53d3265df473b8cd1abce37"}]}],"author":{"name":"Alexandre Engloner","email":"alexandre.engloner@ovhcloud.com","date":"2026-08-04 17:57:09.000000000","tz":-240},"committer":{"name":"Alexandre Engloner","email":"alexandre.engloner@ovhcloud.com","date":"2026-08-04 17:57:09.000000000","tz":-240},"subject":"fix for 2 methods in metadefs policies","message":"fix for 2 methods in metadefs policies\n\n\"modify_metadef_property\" was declared with a method of \u0027GET\u0027 while \"remove_metadef_resource_type_association\" was declared as a \u0027POST\u0027 in their respective DocumentedRuleDefault.\n\nChange-Id: Ice2045067fa92ecc8176a6ae05bbb9e20e4cbb66\nSigned-off-by: Alexandre Engloner \u003calexandre.engloner@ovhcloud.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/f1f60c2f4f4c1315a35c2c1a696d8a4f50317b34"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/f1f60c2f4f4c1315a35c2c1a696d8a4f50317b34"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{"label:Review-Priority\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
