)]}'
{"id":"openstack%2Fgnocchi~425939","triplet_id":"openstack%2Fgnocchi~master~I1ddc78f64f3d909b1fa917e04717f2ccbd194d1d","project":"openstack/gnocchi","branch":"master","topic":"jd/mysql-retry-new-error-lol","hashtags":[],"change_id":"I1ddc78f64f3d909b1fa917e04717f2ccbd194d1d","subject":"sqlalchemy: factorize retry on transaction issue/table def change","status":"MERGED","created":"2017-01-26 21:40:58.000000000","updated":"2017-02-18 18:40:45.000000000","submitted":"2017-02-18 18:40:45.000000000","submitter":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"e0269ac0cc9ee9da459e1f940de9d37d86f10874","_number":425939,"virtual_id_number":425939,"owner":{"_account_id":1669,"name":"Julien Danjou","display_name":"jd","email":"julien@danjou.info","username":"jdanjou"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"all":[{"value":2,"date":"2017-02-18 18:40:44.000000000","_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":0,"_account_id":1669,"name":"Julien Danjou","display_name":"jd","email":"julien@danjou.info","username":"jdanjou"},{"value":0,"_account_id":2813,"name":"Mehdi Abaakouk","email":"sileht@sileht.net","username":"sileht"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":2813,"name":"Mehdi Abaakouk","email":"sileht@sileht.net","username":"sileht"},"all":[{"value":0,"_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":0,"_account_id":1669,"name":"Julien Danjou","display_name":"jd","email":"julien@danjou.info","username":"jdanjou"},{"value":2,"date":"2017-01-30 11:20:16.000000000","_account_id":2813,"name":"Mehdi Abaakouk","email":"sileht@sileht.net","username":"sileht"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":1669,"name":"Julien Danjou","display_name":"jd","email":"julien@danjou.info","username":"jdanjou"},"all":[{"value":0,"_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":1,"date":"2017-02-18 18:10:27.000000000","_account_id":1669,"name":"Julien Danjou","display_name":"jd","email":"julien@danjou.info","username":"jdanjou"},{"value":0,"_account_id":2813,"name":"Mehdi Abaakouk","email":"sileht@sileht.net","username":"sileht"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":3,"name":"Jenkins","username":"jenkins"},{"_account_id":1669,"name":"Julien Danjou","display_name":"jd","email":"julien@danjou.info","username":"jdanjou"},{"_account_id":2813,"name":"Mehdi Abaakouk","email":"sileht@sileht.net","username":"sileht"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2017-01-30 11:20:16.000000000","updated_by":{"_account_id":2813,"name":"Mehdi Abaakouk","email":"sileht@sileht.net","username":"sileht"},"reviewer":{"_account_id":2813,"name":"Mehdi Abaakouk","email":"sileht@sileht.net","username":"sileht"},"state":"REVIEWER"},{"updated":"2017-02-18 18:40:44.000000000","updated_by":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"reviewer":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"state":"REVIEWER"}],"messages":[{"id":"f57dbe20f005d6e230e16acaa6f099dd553bb6a2","author":{"_account_id":1669,"name":"Julien Danjou","display_name":"jd","email":"julien@danjou.info","username":"jdanjou"},"date":"2017-01-26 21:40:58.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"3bad8fa65549d039d0024f0a4526be8716f73860","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2017-01-26 22:43:39.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see http://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n- gate-gnocchi-pep8-ubuntu-xenial http://logs.openstack.org/39/425939/1/check/gate-gnocchi-pep8-ubuntu-xenial/0f3bbc1/ : SUCCESS in 2m 39s\n- gate-gnocchi-docs-ubuntu-xenial http://docs-draft.openstack.org/39/425939/1/check/gate-gnocchi-docs-ubuntu-xenial/59fdf70//doc/build/html/ : SUCCESS in 2m 35s\n- gate-gnocchi-tox-db-py27-mysql-ubuntu-xenial http://logs.openstack.org/39/425939/1/check/gate-gnocchi-tox-db-py27-mysql-ubuntu-xenial/c9922b6/ : SUCCESS in 10m 01s\n- gate-gnocchi-tox-db-py35-mysql-ubuntu-xenial http://logs.openstack.org/39/425939/1/check/gate-gnocchi-tox-db-py35-mysql-ubuntu-xenial/16bec1e/ : SUCCESS in 10m 30s\n- gate-gnocchi-tox-db-py35-postgresql-ubuntu-xenial http://logs.openstack.org/39/425939/1/check/gate-gnocchi-tox-db-py35-postgresql-ubuntu-xenial/e73b762/ : SUCCESS in 9m 13s\n- gate-gnocchi-tox-db-py27-mysql-ceph-upgrade-from-2.2-ubuntu-xenial http://logs.openstack.org/39/425939/1/check/gate-gnocchi-tox-db-py27-mysql-ceph-upgrade-from-2.2-ubuntu-xenial/3efdad3/ : SUCCESS in 7m 51s\n- gate-gnocchi-tox-db-py35-postgresql-file-upgrade-from-2.2-ubuntu-xenial http://logs.openstack.org/39/425939/1/check/gate-gnocchi-tox-db-py35-postgresql-file-upgrade-from-2.2-ubuntu-xenial/57cfd87/ : SUCCESS in 4m 48s\n- gate-gnocchi-tox-db-py27-mysql-ceph-upgrade-from-3.0-ubuntu-xenial http://logs.openstack.org/39/425939/1/check/gate-gnocchi-tox-db-py27-mysql-ceph-upgrade-from-3.0-ubuntu-xenial/f92db16/ : SUCCESS in 12m 25s\n- gate-gnocchi-tox-db-py35-postgresql-file-upgrade-from-3.0-ubuntu-xenial http://logs.openstack.org/39/425939/1/check/gate-gnocchi-tox-db-py35-postgresql-file-upgrade-from-3.0-ubuntu-xenial/9669e0d/ : SUCCESS in 11m 36s\n- gate-gnocchi-bashate http://logs.openstack.org/39/425939/1/check/gate-gnocchi-bashate/bada5a8/ : SUCCESS in 2m 08s\n- gate-gnocchi-dsvm-functional-file-mysql-ubuntu-xenial http://logs.openstack.org/39/425939/1/check/gate-gnocchi-dsvm-functional-file-mysql-ubuntu-xenial/9161afd/ : FAILURE in 19m 52s\n- gate-gnocchi-dsvm-functional-swift-postgresql-ubuntu-xenial http://logs.openstack.org/39/425939/1/check/gate-gnocchi-dsvm-functional-swift-postgresql-ubuntu-xenial/0314b43/ : FAILURE in 11m 19s\n- gate-gnocchi-dsvm-functional-ceph-mysql-ubuntu-xenial http://logs.openstack.org/39/425939/1/check/gate-gnocchi-dsvm-functional-ceph-mysql-ubuntu-xenial/64e5d6f/ : FAILURE in 10m 50s\n- gate-telemetry-dsvm-integration-gnocchi-ubuntu-xenial http://logs.openstack.org/39/425939/1/check/gate-telemetry-dsvm-integration-gnocchi-ubuntu-xenial/87fb29c/ : SUCCESS in 59m 59s\n- gate-tripleo-ci-centos-7-scenario001-multinode-puppet-nv http://logs.openstack.org/39/425939/1/check/gate-tripleo-ci-centos-7-scenario001-multinode-puppet-nv/5828554/ : SUCCESS in 58m 21s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"05198f9cb2bd915eb5fe3e436b48fd0353c8652f","author":{"_account_id":1669,"name":"Julien Danjou","display_name":"jd","email":"julien@danjou.info","username":"jdanjou"},"date":"2017-01-28 14:31:33.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"f19710812c424a64a6e7de2dfcc624f54fa3c0b4","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2017-01-28 15:31:00.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- gate-gnocchi-pep8-ubuntu-xenial http://logs.openstack.org/39/425939/1/check/gate-gnocchi-pep8-ubuntu-xenial/251dc8a/ : SUCCESS in 3m 02s\n- gate-gnocchi-docs-ubuntu-xenial http://docs-draft.openstack.org/39/425939/1/check/gate-gnocchi-docs-ubuntu-xenial/7617645//doc/build/html/ : SUCCESS in 1m 59s\n- gate-gnocchi-tox-db-py27-mysql-ubuntu-xenial http://logs.openstack.org/39/425939/1/check/gate-gnocchi-tox-db-py27-mysql-ubuntu-xenial/08027d0/ : SUCCESS in 17m 52s\n- gate-gnocchi-tox-db-py35-mysql-ubuntu-xenial http://logs.openstack.org/39/425939/1/check/gate-gnocchi-tox-db-py35-mysql-ubuntu-xenial/2d65e18/ : SUCCESS in 9m 47s\n- gate-gnocchi-tox-db-py35-postgresql-ubuntu-xenial http://logs.openstack.org/39/425939/1/check/gate-gnocchi-tox-db-py35-postgresql-ubuntu-xenial/78f8059/ : SUCCESS in 13m 14s\n- gate-gnocchi-tox-db-py27-mysql-ceph-upgrade-from-2.2-ubuntu-xenial http://logs.openstack.org/39/425939/1/check/gate-gnocchi-tox-db-py27-mysql-ceph-upgrade-from-2.2-ubuntu-xenial/9116bb9/ : SUCCESS in 10m 30s\n- gate-gnocchi-tox-db-py35-postgresql-file-upgrade-from-2.2-ubuntu-xenial http://logs.openstack.org/39/425939/1/check/gate-gnocchi-tox-db-py35-postgresql-file-upgrade-from-2.2-ubuntu-xenial/79b06da/ : SUCCESS in 4m 56s\n- gate-gnocchi-tox-db-py27-mysql-ceph-upgrade-from-3.0-ubuntu-xenial http://logs.openstack.org/39/425939/1/check/gate-gnocchi-tox-db-py27-mysql-ceph-upgrade-from-3.0-ubuntu-xenial/cbf7389/ : SUCCESS in 8m 34s\n- gate-gnocchi-tox-db-py35-postgresql-file-upgrade-from-3.0-ubuntu-xenial http://logs.openstack.org/39/425939/1/check/gate-gnocchi-tox-db-py35-postgresql-file-upgrade-from-3.0-ubuntu-xenial/5b45342/ : SUCCESS in 5m 43s\n- gate-gnocchi-bashate http://logs.openstack.org/39/425939/1/check/gate-gnocchi-bashate/27da247/ : SUCCESS in 1m 20s\n- gate-gnocchi-dsvm-functional-file-mysql-ubuntu-xenial http://logs.openstack.org/39/425939/1/check/gate-gnocchi-dsvm-functional-file-mysql-ubuntu-xenial/bd5380d/ : SUCCESS in 25m 42s\n- gate-gnocchi-dsvm-functional-swift-postgresql-ubuntu-xenial http://logs.openstack.org/39/425939/1/check/gate-gnocchi-dsvm-functional-swift-postgresql-ubuntu-xenial/68435bf/ : SUCCESS in 19m 58s\n- gate-gnocchi-dsvm-functional-ceph-mysql-ubuntu-xenial http://logs.openstack.org/39/425939/1/check/gate-gnocchi-dsvm-functional-ceph-mysql-ubuntu-xenial/d0b0c3d/ : SUCCESS in 31m 58s\n- gate-telemetry-dsvm-integration-gnocchi-ubuntu-xenial http://logs.openstack.org/39/425939/1/check/gate-telemetry-dsvm-integration-gnocchi-ubuntu-xenial/1f6220c/ : SUCCESS in 55m 41s\n- gate-tripleo-ci-centos-7-scenario001-multinode-puppet-nv http://logs.openstack.org/39/425939/1/check/gate-tripleo-ci-centos-7-scenario001-multinode-puppet-nv/6a55ed2/ : SUCCESS in 55m 46s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"0bb5cf633298caac1b0533d0278a3887b8d6338c","author":{"_account_id":2813,"name":"Mehdi Abaakouk","email":"sileht@sileht.net","username":"sileht"},"date":"2017-01-30 11:20:16.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"f8f8f222cdb8f63a8585b1ff471d76d0cf8d71a4","author":{"_account_id":1669,"name":"Julien Danjou","display_name":"jd","email":"julien@danjou.info","username":"jdanjou"},"date":"2017-02-18 18:10:27.000000000","message":"Patch Set 1: Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"1828f759ab78a28a08905715b525b2f6d0c645dc","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2017-02-18 18:10:37.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.\nhttp://status.openstack.org/zuul/","accounts_in_message":[],"_revision_number":1},{"id":"fd8eefae178f4f8df2cdfa14bdfd00b2f6d19441","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2017-02-18 18:40:44.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- gate-gnocchi-pep8-ubuntu-xenial http://logs.openstack.org/39/425939/1/gate/gate-gnocchi-pep8-ubuntu-xenial/f99e8a3/ : SUCCESS in 1m 40s\n- gate-gnocchi-docs-ubuntu-xenial http://docs-draft.openstack.org/39/425939/1/gate/gate-gnocchi-docs-ubuntu-xenial/431c6ca//doc/build/html/ : SUCCESS in 4m 16s\n- gate-gnocchi-tox-py27-mysql-ubuntu-xenial http://logs.openstack.org/39/425939/1/gate/gate-gnocchi-tox-py27-mysql-ubuntu-xenial/e3fbdfd/ : SUCCESS in 8m 44s\n- gate-gnocchi-tox-py35-mysql-ubuntu-xenial http://logs.openstack.org/39/425939/1/gate/gate-gnocchi-tox-py35-mysql-ubuntu-xenial/200636a/ : SUCCESS in 13m 01s\n- gate-gnocchi-tox-py35-postgresql-ubuntu-xenial http://logs.openstack.org/39/425939/1/gate/gate-gnocchi-tox-py35-postgresql-ubuntu-xenial/e57cc4c/ : SUCCESS in 13m 42s\n- gate-gnocchi-tox-py27-mysql-ceph-upgrade-from-2.2-ubuntu-xenial http://logs.openstack.org/39/425939/1/gate/gate-gnocchi-tox-py27-mysql-ceph-upgrade-from-2.2-ubuntu-xenial/e97f76f/ : SUCCESS in 6m 23s\n- gate-gnocchi-tox-py35-postgresql-file-upgrade-from-2.2-ubuntu-xenial http://logs.openstack.org/39/425939/1/gate/gate-gnocchi-tox-py35-postgresql-file-upgrade-from-2.2-ubuntu-xenial/9f2abaf/ : SUCCESS in 3m 59s\n- gate-gnocchi-tox-py27-mysql-ceph-upgrade-from-3.0-ubuntu-xenial http://logs.openstack.org/39/425939/1/gate/gate-gnocchi-tox-py27-mysql-ceph-upgrade-from-3.0-ubuntu-xenial/be3747f/ : SUCCESS in 11m 12s\n- gate-gnocchi-tox-py35-postgresql-file-upgrade-from-3.0-ubuntu-xenial http://logs.openstack.org/39/425939/1/gate/gate-gnocchi-tox-py35-postgresql-file-upgrade-from-3.0-ubuntu-xenial/6ed301e/ : SUCCESS in 4m 40s\n- gate-gnocchi-bashate http://logs.openstack.org/39/425939/1/gate/gate-gnocchi-bashate/5fb5835/ : SUCCESS in 1m 20s\n- gate-gnocchi-dsvm-functional-file-mysql-ubuntu-xenial http://logs.openstack.org/39/425939/1/gate/gate-gnocchi-dsvm-functional-file-mysql-ubuntu-xenial/2d910c2/ : SUCCESS in 29m 16s\n- gate-gnocchi-dsvm-functional-swift-postgresql-ubuntu-xenial http://logs.openstack.org/39/425939/1/gate/gate-gnocchi-dsvm-functional-swift-postgresql-ubuntu-xenial/7bc86b9/ : SUCCESS in 20m 21s\n- gate-gnocchi-dsvm-functional-ceph-mysql-ubuntu-xenial http://logs.openstack.org/39/425939/1/gate/gate-gnocchi-dsvm-functional-ceph-mysql-ubuntu-xenial/9ec03fe/ : SUCCESS in 29m 18s\n- gate-telemetry-dsvm-integration-gnocchi-ubuntu-xenial http://logs.openstack.org/39/425939/1/gate/gate-telemetry-dsvm-integration-gnocchi-ubuntu-xenial/2eac52e/ : SUCCESS in 30m 04s","accounts_in_message":[],"_revision_number":1},{"id":"9e35ab2106f2f7993258a22b06461d63a7c9e06d","date":"2017-02-18 18:40:45.000000000","message":"Change has been successfully merged into the git repository by Jenkins","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"9ea432183772767634d61218bbba48508c960c9b","revisions":{"9ea432183772767634d61218bbba48508c960c9b":{"kind":"REWORK","_number":1,"created":"2017-01-26 21:40:58.000000000","uploader":{"_account_id":1669,"name":"Julien Danjou","display_name":"jd","email":"julien@danjou.info","username":"jdanjou"},"ref":"refs/changes/39/425939/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/gnocchi","ref":"refs/changes/39/425939/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/gnocchi refs/changes/39/425939/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/gnocchi refs/changes/39/425939/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/gnocchi refs/changes/39/425939/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/gnocchi refs/changes/39/425939/1"}}},"commit":{"parents":[{"commit":"fc0e38c1dea9894e56bb7995f44cf29c138db034","subject":"mysql: retry on table changes","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/gnocchi/commit/fc0e38c1dea9894e56bb7995f44cf29c138db034"}]}],"author":{"name":"Julien Danjou","email":"julien@danjou.info","date":"2017-01-26 21:39:56.000000000","tz":60},"committer":{"name":"Julien Danjou","email":"julien@danjou.info","date":"2017-01-26 21:39:56.000000000","tz":60},"subject":"sqlalchemy: factorize retry on transaction issue/table def change","message":"sqlalchemy: factorize retry on transaction issue/table def change\n\nThis moves the retry condition to a common code for MySQL and PostgreSQL.\n\nChange-Id: I1ddc78f64f3d909b1fa917e04717f2ccbd194d1d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/gnocchi/commit/9ea432183772767634d61218bbba48508c960c9b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/gnocchi/commit/9ea432183772767634d61218bbba48508c960c9b"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"fc0e38c1dea9894e56bb7995f44cf29c138db034","is_merged_in_target_branch":true,"change_id":"Ib8552529e73cee4ea92c151bbe3b9879e93a3a7c","change_number":425938,"patch_set_number":1,"change_status":"MERGED"}],"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
