)]}'
{"id":"openstack%2Fgovernance~281145","triplet_id":"openstack%2Fgovernance~master~Ib270b1ef09892314ef449e0cf7b288e7f2d974ef","project":"openstack/governance","branch":"master","topic":"i18n-atc","hashtags":[],"change_id":"Ib270b1ef09892314ef449e0cf7b288e7f2d974ef","subject":"Refresh I18n ATC list","status":"MERGED","created":"2016-02-17 09:43:43.000000000","updated":"2016-02-26 10:09:40.000000000","submitted":"2016-02-26 10:09:40.000000000","submitter":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"11bbe67de6bdcdda1a99622ceb51b7553e6d79fa","_number":281145,"virtual_id_number":281145,"owner":{"_account_id":3108,"name":"daisy-ycguo","email":"guoyingc@cn.ibm.com","username":"daisy-ycguo"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"all":[{"value":0,"_account_id":10497,"name":"KATO Tomoyuki","email":"kato.tomoyuki@jp.fujitsu.com","username":"to222"},{"value":0,"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},{"value":2,"date":"2016-02-26 10:09:40.000000000","post_submit":true,"_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":0,"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":7,"name":"Jay Pipes","email":"jaypipes@gmail.com","username":"jaypipes"},{"value":0,"_account_id":2472,"name":"Doug Hellmann","email":"dhellmann@redhat.com","username":"doug-hellmann"},{"value":0,"_account_id":1561,"name":"Russell Bryant","email":"rbryant@redhat.com","username":"russellb"},{"value":0,"_account_id":6159,"name":"Flavio Percoco Premoli","display_name":"flaper87","email":"flavio.percoco@flyrlabs.com","username":"flaper87"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":10497,"name":"KATO Tomoyuki","email":"kato.tomoyuki@jp.fujitsu.com","username":"to222"},"all":[{"value":1,"date":"2016-02-22 05:50:58.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":10497,"name":"KATO Tomoyuki","email":"kato.tomoyuki@jp.fujitsu.com","username":"to222"},{"value":1,"date":"2016-02-22 13:01:32.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},{"value":0,"_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":1,"date":"2016-02-22 10:08:52.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":1,"date":"2016-02-23 19:58:44.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":7,"name":"Jay Pipes","email":"jaypipes@gmail.com","username":"jaypipes"},{"value":0,"_account_id":2472,"name":"Doug Hellmann","email":"dhellmann@redhat.com","username":"doug-hellmann"},{"value":1,"date":"2016-02-24 02:35:28.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":1561,"name":"Russell Bryant","email":"rbryant@redhat.com","username":"russellb"},{"value":1,"date":"2016-02-26 01:49:19.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":6159,"name":"Flavio Percoco Premoli","display_name":"flaper87","email":"flavio.percoco@flyrlabs.com","username":"flaper87"}],"values":{"-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"all":[{"value":0,"_account_id":10497,"name":"KATO Tomoyuki","email":"kato.tomoyuki@jp.fujitsu.com","username":"to222"},{"value":0,"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},{"value":0,"_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":1,"date":"2016-02-26 10:07:34.000000000","_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":7,"name":"Jay Pipes","email":"jaypipes@gmail.com","username":"jaypipes"},{"value":0,"_account_id":2472,"name":"Doug Hellmann","email":"dhellmann@redhat.com","username":"doug-hellmann"},{"value":0,"_account_id":1561,"name":"Russell Bryant","email":"rbryant@redhat.com","username":"russellb"},{"value":0,"_account_id":6159,"name":"Flavio Percoco Premoli","display_name":"flaper87","email":"flavio.percoco@flyrlabs.com","username":"flaper87"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Rollcall-Vote":{"approved":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"all":[{"value":0,"_account_id":10497,"name":"KATO Tomoyuki","email":"kato.tomoyuki@jp.fujitsu.com","username":"to222"},{"value":0,"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},{"value":0,"_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":1,"date":"2016-02-22 10:08:52.000000000","_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":1,"date":"2016-02-23 19:58:44.000000000","_account_id":7,"name":"Jay Pipes","email":"jaypipes@gmail.com","username":"jaypipes"},{"value":0,"_account_id":2472,"name":"Doug Hellmann","email":"dhellmann@redhat.com","username":"doug-hellmann"},{"value":1,"date":"2016-02-24 02:35:28.000000000","_account_id":1561,"name":"Russell Bryant","email":"rbryant@redhat.com","username":"russellb"},{"value":1,"date":"2016-02-26 01:49:19.000000000","_account_id":6159,"name":"Flavio Percoco Premoli","display_name":"flaper87","email":"flavio.percoco@flyrlabs.com","username":"flaper87"}],"values":{"-1":"No"," 0":"Abstain","+1":"Yes"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":3,"name":"Jenkins","username":"jenkins"},{"_account_id":7,"name":"Jay Pipes","email":"jaypipes@gmail.com","username":"jaypipes"},{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"_account_id":1561,"name":"Russell Bryant","email":"rbryant@redhat.com","username":"russellb"},{"_account_id":2472,"name":"Doug Hellmann","email":"dhellmann@redhat.com","username":"doug-hellmann"},{"_account_id":6159,"name":"Flavio Percoco Premoli","display_name":"flaper87","email":"flavio.percoco@flyrlabs.com","username":"flaper87"},{"_account_id":10497,"name":"KATO Tomoyuki","email":"kato.tomoyuki@jp.fujitsu.com","username":"to222"},{"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2016-02-19 17:27:54.000000000","updated_by":{"_account_id":2472,"name":"Doug Hellmann","email":"dhellmann@redhat.com","username":"doug-hellmann"},"reviewer":{"_account_id":2472,"name":"Doug Hellmann","email":"dhellmann@redhat.com","username":"doug-hellmann"},"state":"REVIEWER"},{"updated":"2016-02-22 05:50:58.000000000","updated_by":{"_account_id":10497,"name":"KATO Tomoyuki","email":"kato.tomoyuki@jp.fujitsu.com","username":"to222"},"reviewer":{"_account_id":10497,"name":"KATO Tomoyuki","email":"kato.tomoyuki@jp.fujitsu.com","username":"to222"},"state":"REVIEWER"},{"updated":"2016-02-22 13:01:32.000000000","updated_by":{"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},"reviewer":{"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},"state":"REVIEWER"},{"updated":"2016-02-23 19:58:44.000000000","updated_by":{"_account_id":7,"name":"Jay Pipes","email":"jaypipes@gmail.com","username":"jaypipes"},"reviewer":{"_account_id":7,"name":"Jay Pipes","email":"jaypipes@gmail.com","username":"jaypipes"},"state":"REVIEWER"},{"updated":"2016-02-24 02:35:28.000000000","updated_by":{"_account_id":1561,"name":"Russell Bryant","email":"rbryant@redhat.com","username":"russellb"},"reviewer":{"_account_id":1561,"name":"Russell Bryant","email":"rbryant@redhat.com","username":"russellb"},"state":"REVIEWER"},{"updated":"2016-02-26 01:49:19.000000000","updated_by":{"_account_id":6159,"name":"Flavio Percoco Premoli","display_name":"flaper87","email":"flavio.percoco@flyrlabs.com","username":"flaper87"},"reviewer":{"_account_id":6159,"name":"Flavio Percoco Premoli","display_name":"flaper87","email":"flavio.percoco@flyrlabs.com","username":"flaper87"},"state":"REVIEWER"},{"updated":"2016-02-26 10:07:34.000000000","updated_by":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"reviewer":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"state":"REVIEWER"},{"updated":"2016-02-26 10:09:40.000000000","updated_by":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"reviewer":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"state":"REVIEWER"}],"messages":[{"id":"42d9a0cb40e351b466d61118f28be58dd0bb3288","author":{"_account_id":3108,"name":"daisy-ycguo","email":"guoyingc@cn.ibm.com","username":"daisy-ycguo"},"date":"2016-02-17 09:43:43.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"97553f39328f7af1fec247558b00f2e7f353ae29","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2016-02-17 10:04:59.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see http://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n- gate-governance-docs http://logs.openstack.org/45/281145/1/check/gate-governance-docs/ed9c982/ : FAILURE in 27s","accounts_in_message":[],"_revision_number":1},{"id":"edd74e7a71c1cf3a914b466283c07a4331aabf2a","author":{"_account_id":2472,"name":"Doug Hellmann","email":"dhellmann@redhat.com","username":"doug-hellmann"},"date":"2016-02-17 15:32:24.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"6fea713ba218497b820f11ef645c4d1eaf3ac121","author":{"_account_id":3108,"name":"daisy-ycguo","email":"guoyingc@cn.ibm.com","username":"daisy-ycguo"},"date":"2016-02-18 02:19:40.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"02256e796b4c22054c7d68aa7a553cfe30d07813","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2016-02-18 02:23:38.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- gate-governance-docs http://docs-draft.openstack.org/45/281145/2/check/gate-governance-docs/3965cfa//doc/build/html/ : SUCCESS in 1m 13s","accounts_in_message":[],"_revision_number":2},{"id":"a01b48e974ad1b909c8e29f659b379b1e5873237","author":{"_account_id":10497,"name":"KATO Tomoyuki","email":"kato.tomoyuki@jp.fujitsu.com","username":"to222"},"date":"2016-02-18 02:40:14.000000000","message":"Patch Set 2: Code-Review+1\n\nthanks","accounts_in_message":[],"_revision_number":2},{"id":"2269b6681ec3c0434dc2eaa69e4229d3772c8b5d","author":{"_account_id":2472,"name":"Doug Hellmann","email":"dhellmann@redhat.com","username":"doug-hellmann"},"date":"2016-02-19 17:27:54.000000000","message":"Patch Set 2:\n\nThese are all set to expire in July, about 6 months from now. Don\u0027t we usually set them up for a full year?","accounts_in_message":[],"_revision_number":2},{"id":"45461ba6afed6801ee27fff05ecb504e147706e6","author":{"_account_id":3108,"name":"daisy-ycguo","email":"guoyingc@cn.ibm.com","username":"daisy-ycguo"},"date":"2016-02-22 05:46:49.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"4ed1f1b528fb3e01e578033a176a03462ca0fbf0","author":{"_account_id":3108,"name":"daisy-ycguo","email":"guoyingc@cn.ibm.com","username":"daisy-ycguo"},"date":"2016-02-22 05:47:49.000000000","message":"Patch Set 3:\n\nSet for a full year. \nThanks, Doug.","accounts_in_message":[],"_revision_number":3},{"id":"08fdd310bd82c3818551913d268402a5643b186a","author":{"_account_id":10497,"name":"KATO Tomoyuki","email":"kato.tomoyuki@jp.fujitsu.com","username":"to222"},"date":"2016-02-22 05:50:58.000000000","message":"Patch Set 3: Code-Review+1","accounts_in_message":[],"_revision_number":3},{"id":"c0f44a44ee371d28b3839d51a00c81e017752d01","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2016-02-22 05:52:47.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- gate-governance-docs http://docs-draft.openstack.org/45/281145/3/check/gate-governance-docs/a5104ab//doc/build/html/ : SUCCESS in 1m 00s","accounts_in_message":[],"_revision_number":3},{"id":"b9d0bd89ce35ee729540650465fcbb9f5ebc51c8","author":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"date":"2016-02-22 10:08:52.000000000","message":"Patch Set 3: Code-Review+1 Rollcall-Vote+1","accounts_in_message":[],"_revision_number":3},{"id":"c3f862bee8e295dea0459fe98e85a3600ac6cfd3","author":{"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},"date":"2016-02-22 13:01:32.000000000","message":"Patch Set 3: Code-Review+1","accounts_in_message":[],"_revision_number":3},{"id":"00ae8ddb3a1db7d2a744e959c0079d97248f0db9","author":{"_account_id":7,"name":"Jay Pipes","email":"jaypipes@gmail.com","username":"jaypipes"},"date":"2016-02-23 19:58:44.000000000","message":"Patch Set 3: Rollcall-Vote+1 Code-Review+1\n\n++","accounts_in_message":[],"_revision_number":3},{"id":"b1185404d66e8a9fddfcc5d5b916d3d986342bc9","author":{"_account_id":1561,"name":"Russell Bryant","email":"rbryant@redhat.com","username":"russellb"},"date":"2016-02-24 02:35:28.000000000","message":"Patch Set 3: Code-Review+1 Rollcall-Vote+1","accounts_in_message":[],"_revision_number":3},{"id":"06a5a33886c6cedab27405b779147e12dc887941","author":{"_account_id":6159,"name":"Flavio Percoco Premoli","display_name":"flaper87","email":"flavio.percoco@flyrlabs.com","username":"flaper87"},"date":"2016-02-26 01:49:19.000000000","message":"Patch Set 3: Code-Review+1 Rollcall-Vote+1","accounts_in_message":[],"_revision_number":3},{"id":"6228956327ef663e1003d89695b9a0c8b056f6b2","author":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"date":"2016-02-26 10:07:34.000000000","message":"Patch Set 3: Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"3f19a56bbb547d48d61d9a8e963f54277d1f9f57","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2016-02-26 10:07:40.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.\nhttp://status.openstack.org/zuul/","accounts_in_message":[],"_revision_number":3},{"id":"9d4ac9fc1725ace0dd06d8ef391dbc850e97e54a","date":"2016-02-26 10:09:40.000000000","message":"Change has been successfully merged into the git repository by Jenkins","accounts_in_message":[],"_revision_number":3},{"id":"638be8e8dca08d8e814dada2c0fd0b972c53c564","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2016-02-26 10:09:40.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- gate-governance-docs http://docs-draft.openstack.org/45/281145/3/gate/gate-governance-docs/e5f5965//doc/build/html/ : SUCCESS in 1m 07s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"8b3c83f28102c7b47688fbaca970a52a76eb6de5","revisions":{"b0938d8b164fd279440a83a6803838e8b7321ca5":{"kind":"REWORK","_number":1,"created":"2016-02-17 09:43:43.000000000","uploader":{"_account_id":3108,"name":"daisy-ycguo","email":"guoyingc@cn.ibm.com","username":"daisy-ycguo"},"ref":"refs/changes/45/281145/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/governance","ref":"refs/changes/45/281145/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/governance refs/changes/45/281145/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/governance refs/changes/45/281145/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/governance refs/changes/45/281145/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/governance refs/changes/45/281145/1"}}},"commit":{"parents":[{"commit":"a690a0efc99ff0faf5668331a94bb8982a8bed18","subject":"Merge \"Add sahara-tests project to sahara\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/governance/commit/a690a0efc99ff0faf5668331a94bb8982a8bed18"}]}],"author":{"name":"daisy-ycguo","email":"guoyingc@cn.ibm.com","date":"2016-02-17 09:20:51.000000000","tz":480},"committer":{"name":"daisy-ycguo","email":"guoyingc@cn.ibm.com","date":"2016-02-17 09:32:43.000000000","tz":480},"subject":"Refresh I18n ATC list","message":"Refresh I18n ATC list\n\nThose people are active translators from 2015-08-01 to 2016-01-30.\nThe name list was made by querying Zanata translators statistic API.\n\nIt\u0027s necessary to grant those people ATC in order to make them voters\nfor PTL and give them the discount codes for Austin summit.\n\nChange-Id: Ib270b1ef09892314ef449e0cf7b288e7f2d974ef\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/governance/commit/b0938d8b164fd279440a83a6803838e8b7321ca5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/governance/commit/b0938d8b164fd279440a83a6803838e8b7321ca5"}]},"branch":"refs/heads/master"},"be2803f27e47a88007655483057533300106e8be":{"kind":"REWORK","_number":2,"created":"2016-02-18 02:19:40.000000000","uploader":{"_account_id":3108,"name":"daisy-ycguo","email":"guoyingc@cn.ibm.com","username":"daisy-ycguo"},"ref":"refs/changes/45/281145/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/governance","ref":"refs/changes/45/281145/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/governance refs/changes/45/281145/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/governance refs/changes/45/281145/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/governance refs/changes/45/281145/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/governance refs/changes/45/281145/2"}}},"commit":{"parents":[{"commit":"a690a0efc99ff0faf5668331a94bb8982a8bed18","subject":"Merge \"Add sahara-tests project to sahara\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/governance/commit/a690a0efc99ff0faf5668331a94bb8982a8bed18"}]}],"author":{"name":"daisy-ycguo","email":"guoyingc@cn.ibm.com","date":"2016-02-17 09:20:51.000000000","tz":480},"committer":{"name":"daisy-ycguo","email":"guoyingc@cn.ibm.com","date":"2016-02-18 02:08:57.000000000","tz":480},"subject":"Refresh I18n ATC list","message":"Refresh I18n ATC list\n\nThose people are active translators from 2015-08-01 to 2016-01-30.\nThe name list was made by querying Zanata translators statistic API.\n\nIt\u0027s necessary to grant those people ATC in order to make them voters\nfor PTL and give them the discount codes for Austin summit.\n\nChange-Id: Ib270b1ef09892314ef449e0cf7b288e7f2d974ef\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/governance/commit/be2803f27e47a88007655483057533300106e8be"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/governance/commit/be2803f27e47a88007655483057533300106e8be"}]},"branch":"refs/heads/master"},"8b3c83f28102c7b47688fbaca970a52a76eb6de5":{"kind":"REWORK","_number":3,"created":"2016-02-22 05:46:49.000000000","uploader":{"_account_id":3108,"name":"daisy-ycguo","email":"guoyingc@cn.ibm.com","username":"daisy-ycguo"},"ref":"refs/changes/45/281145/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/governance","ref":"refs/changes/45/281145/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/governance refs/changes/45/281145/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/governance refs/changes/45/281145/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/governance refs/changes/45/281145/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/governance refs/changes/45/281145/3"}}},"commit":{"parents":[{"commit":"a690a0efc99ff0faf5668331a94bb8982a8bed18","subject":"Merge \"Add sahara-tests project to sahara\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/governance/commit/a690a0efc99ff0faf5668331a94bb8982a8bed18"}]}],"author":{"name":"daisy-ycguo","email":"guoyingc@cn.ibm.com","date":"2016-02-17 09:20:51.000000000","tz":480},"committer":{"name":"daisy-ycguo","email":"guoyingc@cn.ibm.com","date":"2016-02-22 05:36:22.000000000","tz":480},"subject":"Refresh I18n ATC list","message":"Refresh I18n ATC list\n\nThose people are active translators from 2015-08-01 to 2016-01-30.\nThe name list was made by querying Zanata translators statistic API.\n\nIt\u0027s necessary to grant those people ATC in order to make them voters\nfor PTL and give them the discount codes for Austin summit.\n\nChange-Id: Ib270b1ef09892314ef449e0cf7b288e7f2d974ef\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/governance/commit/8b3c83f28102c7b47688fbaca970a52a76eb6de5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/governance/commit/8b3c83f28102c7b47688fbaca970a52a76eb6de5"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
